)]}'
{"id":"openstack%2Fneutron~979705","triplet_id":"openstack%2Fneutron~master~I69e9fc00bfc58fb86675f72f94d10d014f25cd3a","project":"openstack/neutron","branch":"master","topic":"fix-ovn-chassis-name-with-hyphen","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-03-26 14:43:11.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"30534":{"account":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"last_update":"2026-03-27 00:53:20.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-03-26 14:02:07.000000000","reason":"removed on reply"},"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2026-03-27 00:53:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I69e9fc00bfc58fb86675f72f94d10d014f25cd3a","subject":"Update OVN doc about \u0027ovn-nbctl\u0027 command usage regarding the Gateway_Chassis registers","status":"MERGED","created":"2026-03-09 17:17:35.000000000","updated":"2026-03-27 00:54:27.000000000","submitted":"2026-03-27 00:53:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"979705-fix-ovn-chassis-name-with-hyphen","meta_rev_id":"8b025bf489b8883a459a025b3bcebb5b9f8428db","_number":979705,"virtual_id_number":979705,"owner":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-27 00:53:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-03-26 14:43:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-03-26 14:02:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-03-26 14:43:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-09 17:52:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-09 19:27:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-10 06:21:42.000000000","updated_by":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-03-10 06:21:42.000000000","updated_by":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2026-03-10 06:21:42.000000000","updated_by":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"80caf480fd0e2e77478acb3cd56604dbef4067f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"date":"2026-03-09 17:17:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"004f14da5fb3741673601e9dea4912122cb3d82e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 17:52:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2792f49687084882889b2676edeb6104\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b6ce90e85e8a4d12bb8583b84420e028 : SUCCESS in 33m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e53eff28be254fd5b87273661eeaf4fd : SUCCESS in 20m 50s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a1c4f6f76b5847ae8b0df042387c110f : SUCCESS in 31m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c1346c13e577656877b23c2f65e7892143f0c039","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 19:27:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce4c68e55792444886d661e9e743cd26\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3290d363ea4748ff9349a3d61b473dd8 : SUCCESS in 1h 07m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a0428d620ada46758f0a44d3bb8c9445 : SUCCESS in 37m 58s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1bea4bd6aa624e24b5b8ee95d81658b5 : SUCCESS in 2h 08m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/67e60dbd7b0645099f387122368aaba8 : SUCCESS in 32m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08aa2d0ef26c4d00bb7e16f2390961f6 : SUCCESS in 1h 10m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/100afefcf38f40b4bbbd08532b925108 : SUCCESS in 24m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0a5be39862a444e1a88c7b89be345f92 : SUCCESS in 32m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/92667655806a44a69f5be9131400675f : SUCCESS in 16m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/54cb183f9c0d401fa3aa70157c13c336 : SUCCESS in 27m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e9915bb56a14e60bf79ede67936b489 : SUCCESS in 11m 24s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3de8534a7be24aaabab313b7caf713b5 : SUCCESS in 1h 09m 37s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/246a96a9af264da59aadecf5d301d7f1 : SUCCESS in 1h 23m 22s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/b1d5208110e749dc9430e61db2d55d6c : SUCCESS in 1h 13m 09s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0cc1155976b041308763f9e725b562e0 : SUCCESS in 1h 23m 44s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f741967161e243c4acfe8df0915fbdeb : SUCCESS in 45m 11s","accounts_in_message":[],"_revision_number":1},{"id":"1243ab59dddeb611a7521d0cbe1b564f6cde7c37","author":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"date":"2026-03-10 06:21:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f2f76c2bcd7d6cddf8ed6e2e3f1e911383dd0b29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"date":"2026-03-10 06:35:43.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"f4dac781218771273d58a0401deb39a1fb2c5872","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 07:08:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4553e804b4114924bc97da7465d1cdd8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dfca10d622b8424f9a884d419f25ac03 : SUCCESS in 31m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/dfaf4c80a98141e5be655b2205e22b02 : SUCCESS in 19m 25s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/97c56e9236bc4ff08f03daddf93338d9 : SUCCESS in 27m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"04e34daba0698c1b9ba9ea44144aa0cecbdca13e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 08:58:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2da107cdadc49dbbca1ab9c7e7a1ad1\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/91f4873a0e6e47948c4ce1f220dea215 : SUCCESS in 1h 06m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ce1f75b7f8545418f5473b0c07f35d7 : SUCCESS in 49m 21s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f6361c51605b4a56b84332fa0579ab7b : SUCCESS in 2h 20m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9f631d3518984fc78b3670a8e927664e : SUCCESS in 32m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b05fc53dbc754d78b1222ce0f6a8d5cf : SUCCESS in 45m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da46300a08404933b037eb171e392090 : SUCCESS in 11m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/78483bd30e4d40a79d33ac67e55562fd : SUCCESS in 32m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c75eabb005a244f89dc192815007d41e : SUCCESS in 16m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1c14b15d3cb347c98ef243638b01e580 : SUCCESS in 21m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e1197b38dea4dd1bdc96a4f36cdb875 : SUCCESS in 6m 19s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ab84cf832491407a994d24886e5e185e : SUCCESS in 35m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3474c7ecaee74e9baa148e154677dd63 : SUCCESS in 1h 04m 34s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/0751eedee26c4554848dfeafa7cdefb9 : SUCCESS in 1h 10m 00s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/57c30919147e485f902c3a7f5c3be5b6 : SUCCESS in 47m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d521674d5c24415dac086c006019633f : SUCCESS in 40m 33s","accounts_in_message":[],"_revision_number":2},{"id":"25b8cf46a711f33cc04ab960dae5d6e0696b4bc3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-10 15:01:18.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"85991a7f3bc3414beac73b4ff8da4298ed49e0dd","author":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"date":"2026-03-11 16:54:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"891edc42d5488e18de91584747447f52ae1b3e67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"date":"2026-03-11 17:32:33.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"06226f212500ce5ae84457e497f2fb1245ccf46e","author":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"date":"2026-03-11 17:33:55.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"eee5bd69fbacc5275ebb65f7f3242980eafbc08b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 18:29:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1affa8f1dd364b128564b6c6e8cd7e2c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c4c4eb95a9304450ad624d69d4b4137b : SUCCESS in 24m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5f4ee73355bd4aef8783c98d4da193f5 : SUCCESS in 18m 39s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/30f52cface924f3bb5d14b7e2b7a4c6c : SUCCESS in 26m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f8da6b40ae93f269211a58ca765e1513b4d47b34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 19:15:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4f5c1e1ef164c5899ee1773272328c8\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a1bfb082560e4a12953cfe417854cc58 : SUCCESS in 1h 22m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a295435fa617405fa466f473c5ccbd77 : SUCCESS in 1h 03m 07s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0cb753bab79144db96488ecd57fe7824 : SUCCESS in 1h 40m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3015ab7daf5b4e76a2f3e85ad31c9c47 : SUCCESS in 32m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48ba156bdc1d4e1da3dd6d23301e5f76 : SUCCESS in 44m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b2fc52a3a6b4f029272f6ac33b5061c : SUCCESS in 11m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ae2ec56d547a40fe811dc16b1c27462e : SUCCESS in 32m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a67bf4194000400db1a5c5729e21de59 : SUCCESS in 25m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/24f19a4daa764ac29e2c66016a04f411 : SUCCESS in 33m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5836876db97d4db4a5550b6f49ea0617 : SUCCESS in 12m 56s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f20035afbb954cd3865b40ab53fbb569 : SUCCESS in 36m 45s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8a5878b041f640c7aaa507fb382de18a : SUCCESS in 1h 23m 48s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/308c59e178cd4cdeb9f1936d3e8b54a0 : SUCCESS in 1h 14m 13s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6091bdab3ffe49f589ae30139e23b410 : SUCCESS in 1h 11m 01s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7f818f64cd494f87b9fe418ee663f2df : SUCCESS in 1h 29m 00s","accounts_in_message":[],"_revision_number":3},{"id":"b944a2fedd5b972aa3c1eac5c1f299c61ee86b08","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-13 12:19:24.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f07b65508b0a26000997ad4eeb81908c68bfe195","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"date":"2026-03-16 11:35:25.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"0b90a442317efecc3f37fd50cd87f0538b4487de","author":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"date":"2026-03-16 11:42:11.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2e26cf5c3c0686ff9b060049338a6267cb713304","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 12:39:05.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f2cf50c1e6e435d8ead8843a65e751b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/45cc5b98e9514c48b6536ab7d9482993 : SUCCESS in 1h 01m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0b019112fb114960acfa28b0bc8f726c : SUCCESS in 29m 50s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d5fa2ea5e84a400fa61dae7ee3f3673f : SUCCESS in 51m 04s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7c94b6919c2366400dc3220b64c9d294f5d5707a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 14:09:30.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa7848d8595b455d9648b35c78536eaf\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a723929a0e2648639d8e64a4086cde84 : SUCCESS in 1h 00m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/12add427d1a84283992c09a73c1f62eb : SUCCESS in 1h 02m 30s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7ab109f7cb3d4f42890720ef3784b2b0 : SUCCESS in 2h 33m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/605e19b0a501462f88571031ddaef609 : SUCCESS in 38m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10f2f739bb09445d821da39a82d34814 : SUCCESS in 25m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/afab16ae27734d69933d99ced4c2affa : SUCCESS in 20m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5e96b6a50e4546709dfd2a5321cc42e2 : SUCCESS in 24m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/831296043dca46e89e03ca4a55ea2346 : SUCCESS in 22m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a99af6e1e6d43778559fe190cfbbd5b : SUCCESS in 12m 17s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b5fe32b5a63f4867b70e3bd86e77f735 : SUCCESS in 52m 20s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cfff5a706ac54049b04b732513b31752 : SUCCESS in 42m 36s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/16bf2c7098df4428b1d902e86985ee00 : SUCCESS in 34m 21s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/abd0ffb7fcb441b490980685e741f84f : SUCCESS in 57m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fc98e76238d7416787e6e8d83fa383c4 : SUCCESS in 36m 00s","accounts_in_message":[],"_revision_number":4},{"id":"557764c42f187e96b625955c768b03dbd33bb045","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-03-17 16:59:02.000000000","message":"Patch Set 4: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"d1d189682926a4ef327a4026fbf22a896afac3a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"date":"2026-03-18 07:36:38.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"4cbcdedb27306d95e0b62045339d98f4d1419588","author":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"date":"2026-03-18 07:37:56.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"39ea8d6f1b7d59497e4c4754d71a3129aff792d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 08:59:10.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccf51c3a686b4224a658a7e4a37ba105\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d58498266ef649f8b1ef2eacfdcb9244 : SUCCESS in 31m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0837f5fe9fa546f58a49bc45a51bb059 : SUCCESS in 28m 18s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/187666ce4258484ca8e5f3c40183bf94 : SUCCESS in 26m 22s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"29b45a5ba9f336caecaf958d54369313c189cfb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 09:42:10.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1719040bd5244adba46b690352ae6589\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/448d3195dd694afdb51b3e19f9678c16 : SUCCESS in 1h 25m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/42d7a9db8b8746b9afcd8aa6a1d81c18 : SUCCESS in 1h 06m 52s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b0d21cd393734047962c4a34ecee9116 : SUCCESS in 1h 59m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1928cbc30c54044870b26e99a1d8f2b : SUCCESS in 39m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23287bc983dc4c80a17a20f9c90ddf84 : SUCCESS in 10m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2e2064253a7345b9bfde5070ac96ee43 : SUCCESS in 32m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/783c6c9ad030450ea98cf830089d25cb : SUCCESS in 24m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fec1a8940398430ab7ea94f53391ec84 : SUCCESS in 33m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d2efb337a5147ce8e9a8332d91f9a06 : SUCCESS in 12m 10s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3160457df881471ca37538fee0ffa474 : SUCCESS in 1h 10m 36s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ce76d9618f6e4239b393b327bfa0d6f8 : SUCCESS in 1h 25m 07s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/5f4cad827d0f485488173004e24247a6 : SUCCESS in 1h 15m 22s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6cd0f1d458ac4f77bcfb2c95b9735d9f : SUCCESS in 1h 10m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/41ecbb51f70d468dbd16bd5a2caa6831 : SUCCESS in 37m 03s","accounts_in_message":[],"_revision_number":5},{"id":"d6447e8f84eddcfab9dfeee83a958fed909a721b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-26 10:40:18.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"09029e81371dc7b32c0179e0f0f23f0ddf69cf35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"date":"2026-03-26 12:33:06.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"04cb56e6cef74c8f3376e216be4de0ba29cdf1b8","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-03-26 14:02:07.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"30b4a148596f7859c05825577feae2ef872085ea","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-26 14:43:11.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"8bbd325d608925cbfbb153301ad6c07a48bebc5d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-26 14:43:18.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"87edad2a0e973ff42feb24d2b5f3a6994e1bfd8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 14:54:47.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/848f0f81917d4199b889948649210e5a\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/826fcb84d4e74fdd81f2bb5a7fe17d80 : SUCCESS in 1h 25m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/22071a86fdd248869b6c1264b8e6d323 : SUCCESS in 1h 03m 53s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bfbfa57c9f4c4cc494c30649fc71d9d8 : SUCCESS in 2h 20m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb705bc686a7476d826ee41945f2d75e : SUCCESS in 38m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c19ffe8b2c346a2ae8207873614b579 : SUCCESS in 26m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/25fe1921fb9445e1806633a7fa1cd211 : SUCCESS in 41m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9b8a1a23ab0340eca9e7c284d6aba889 : SUCCESS in 19m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f296c3295e5148ed90eef09e022cf3cc : SUCCESS in 37m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef2f3ced53cb4f98b00f8f3a90fc5c5d : SUCCESS in 6m 03s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ceb164000f8448f38e03f99bf2bd1bf2 : SUCCESS in 38m 25s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/713f26212bc44c56aa980348cc3bd3f7 : SUCCESS in 1h 31m 08s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/b569de84e4e0470e8aedeb9530406066 : SUCCESS in 1h 26m 17s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/fd232e1aaff047f8ad5c2c1277f900f2 : SUCCESS in 1h 02m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2578d62ca96b4776ade0c2022bf5b483 : SUCCESS in 1h 00m 50s","accounts_in_message":[],"_revision_number":6},{"id":"c2251453d55660b860134b0b8203265bd0e2b860","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 14:55:42.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"0a79f251e0dee98de3978ee0dbdff60d39e7e5e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 16:16:54.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aea35f627b8f4d5f95f2c5dec9380110\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/44ef242daa63449281817435368378e0 : SUCCESS in 24m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1fe1785b08f5410ba7fabb65d6524acd : SUCCESS in 27m 44s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d4067d47b2a74cc5bbc57cb2609493bb : SUCCESS in 27m 20s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"65b12ff49459747d52bacce69e1cd55faf2fc65e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 17:43:28.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/25f78271e53b46bc8cedc9378ff9b626\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f96d8983dc734f00bdece90852a421b0 : SUCCESS in 1h 41m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/94ed38aae21a460fb49f07c7ae40b31b : SUCCESS in 1h 08m 39s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/85c1a0fa5a954e8090948e7f7d8c344e : FAILURE in 2h 47m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed3d91f29c104215aa72dc15a2c74e43 : SUCCESS in 28m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/91002c119c394833aa6de89b706964ce : SUCCESS in 35m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/62936173c29242f6860b82704bf7f74a : SUCCESS in 29m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba67ca96aa2b412fb199ca95cbaffcdc : SUCCESS in 13m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/dc6c9a05430a439dbb79fb08942e9d4b : SUCCESS in 1h 32m 20s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/05ba0d596b4147029a93812848feebeb : SUCCESS in 1h 18m 10s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b692e2b43b5f46d2ac961961a460ae8f : SUCCESS in 1h 14m 22s","accounts_in_message":[],"_revision_number":6},{"id":"3403e7a27627328d57a3c236e6401fd91e068b6a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-03-26 21:26:13.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3709850f9479d40c61208edf49d3c9151cef0056","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 23:10:41.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e3cc8e23d1e43d69e23eca8dcd73711\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/fb1a5c5a2cf34c3ca93fdc744250b5bd : SUCCESS in 1h 02m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c41a3b5dde743858518b39f649efdf7 : SUCCESS in 33m 15s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/65a19c5ed9a94fc6a396a84f5b7c9433 : SUCCESS in 1h 42m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f60abf5790e4285af6de2aec79b7fc5 : SUCCESS in 38m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/642f734a74a14ae696f852dfbbeebaff : SUCCESS in 10m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3ca26e77d20a462abd300f2504277a32 : SUCCESS in 21m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7a14a8260bf444ba91c6a84533a39a05 : SUCCESS in 24m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ccdb8c7a180547dea063147c96edfe24 : SUCCESS in 28m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/701a1834e6a94b959e8687e75598ed66 : SUCCESS in 14m 17s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/7591c648b79546b29e54583734d677d8 : SUCCESS in 46m 35s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/dd03af1919a445f5ab37e87d5500c5a3 : SUCCESS in 1h 23m 34s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/3619072166fb4e0f8dacd59c16e6f4dd : SUCCESS in 1h 07m 36s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/920542aae9d840afa92f875130d7c819 : SUCCESS in 1h 19m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9c5713a03ed54ba1bda07c1ced5f180a : SUCCESS in 1h 10m 56s","accounts_in_message":[],"_revision_number":6},{"id":"cfc58a51bf7b2fb22aa04812904e8acc62aedc7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 23:11:05.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"920976ad23f6707b14977a974b870150ee986f0a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 23:18:46.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e7ad0676e114caf818e0e477772f7b6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f4c58ed0b5dc40068c4c6545d14fd0d4 : SUCCESS in 28m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bcc7a25a4b694c71a4232312b4b5bd2c : SUCCESS in 18m 44s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/50aa9aefd72e4d3f894db08097033707 : SUCCESS in 26m 06s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3740ab9bbd2d33c5ae1dcf3e26f2536a28595fd5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 00:53:20.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebf66eb3622f485c9a390a72f5271af8\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/19ef761315394584a90af1d8cd80a2eb : SUCCESS in 1h 00m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9d20c0720569447890459a3216866b86 : SUCCESS in 43m 26s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/71cccb4d54fa46edb2c51165af35325f : SUCCESS in 1h 40m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d470d6a287f04b96ae488888774ca388 : SUCCESS in 10m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fa262b4250374b1a954ec5332b81853c : SUCCESS in 33m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e13c27b4de22406b8820b003038bd1b1 : SUCCESS in 15m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e48572d1d2c42aabcc929a3c75acd70 : SUCCESS in 11m 44s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fb8a8111fd1e45048e0961c609a3c959 : SUCCESS in 42m 17s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/4450bbb31b48439883a7056bf3c6cf97 : SUCCESS in 1h 24m 48s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ca75c19d1eed49eba2a97fe9f166dd93 : SUCCESS in 48m 05s","accounts_in_message":[],"_revision_number":6},{"id":"4a1cab4261598d890f48e2255682a50d55a8139b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 00:53:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"8b025bf489b8883a459a025b3bcebb5b9f8428db","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 00:54:27.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e62c3be64a344ff2a18bfdb3e0c2b2c0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f60d18633eeb4e2bb9a1111839b95bbe : SUCCESS in 54s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"d781fac878700bc7e87992d731ce4c0c59c64b43","revisions":{"d858eecdcfe95e740debd8a1b579c0580b460988":{"kind":"REWORK","_number":1,"created":"2026-03-09 17:17:35.000000000","uploader":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"ref":"refs/changes/05/979705/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/05/979705/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/05/979705/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/05/979705/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/05/979705/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/05/979705/1"}}},"commit":{"parents":[{"commit":"ed800a2ebeb588745a933e45fefc842f77a4ae97","subject":"Merge \"Use project_id key in network plugin code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ed800a2ebeb588745a933e45fefc842f77a4ae97"}]}],"author":{"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","date":"2026-03-09 17:09:28.000000000","tz":60},"committer":{"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","date":"2026-03-09 17:17:14.000000000","tz":60},"subject":"Fix GW chassis lrp names when added via \u0027ovn-nbctl\u0027 command","message":"Fix GW chassis lrp names when added via \u0027ovn-nbctl\u0027 command\n\nThe \u0027ovn-nbctl\u0027 uses the \u0027-\u0027 (hyphen) as a separator beteen the LRP\nport UUID and the chassis hostname when adding a record to the\nGatewat_Chassis table in the NB DB.\n\nOn the other hand, Neutron uses the \u0027_\u0027 (underscore) for the same.\nAs a result, Neutron can\u0027t find LRPs on chassis which were added\nwith the \u0027ovn-nbctl\u0027 command.\n\nThis patch modifies the regexp that is used for parse these names,\nto match both the \u0027-\u0027 and \u0027_\u0027. The new rexexp searches explicitly\nfor the 36 char UUID.\n\nCloses-Bug: #2143751\nChange-Id: I69e9fc00bfc58fb86675f72f94d10d014f25cd3a\nSigned-off-by: Marcin Wilk \u003cmarcin.wilk@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d858eecdcfe95e740debd8a1b579c0580b460988"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d858eecdcfe95e740debd8a1b579c0580b460988"}]},"branch":"refs/heads/master"},"a9182752921634c8523cf2b77c6f0722f539d00f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-03-10 06:35:43.000000000","uploader":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"ref":"refs/changes/05/979705/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/05/979705/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/05/979705/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/05/979705/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/05/979705/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/05/979705/2"}}},"commit":{"parents":[{"commit":"ed800a2ebeb588745a933e45fefc842f77a4ae97","subject":"Merge \"Use project_id key in network plugin code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ed800a2ebeb588745a933e45fefc842f77a4ae97"}]}],"author":{"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","date":"2026-03-09 17:09:28.000000000","tz":60},"committer":{"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","date":"2026-03-10 06:33:36.000000000","tz":60},"subject":"Fix GW chassis LRP names when added via \u0027ovn-nbctl\u0027 command","message":"Fix GW chassis LRP names when added via \u0027ovn-nbctl\u0027 command\n\nThe \u0027ovn-nbctl\u0027 uses the \u0027-\u0027 (hyphen) as a separator between the LRP\nport UUID and the chassis hostname when adding a record to the\nGatewat_Chassis table in the NB DB.\n\nOn the other hand, Neutron uses the \u0027_\u0027 (underscore) for the same.\nAs a result, Neutron can\u0027t find LRPs on chassis which were added\nwith the \u0027ovn-nbctl\u0027 command.\n\nThis patch modifies the regexp that is used for parse these names,\nto match both the \u0027-\u0027 and \u0027_\u0027. The new regexp searches explicitly\nfor the 36 char UUID.\n\nCloses-Bug: #2143751\nChange-Id: I69e9fc00bfc58fb86675f72f94d10d014f25cd3a\nSigned-off-by: Marcin Wilk \u003cmarcin.wilk@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a9182752921634c8523cf2b77c6f0722f539d00f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a9182752921634c8523cf2b77c6f0722f539d00f"}]},"branch":"refs/heads/master"},"fc037c657ed3d6ead6ad4c3b8724b789ebaff44b":{"kind":"REWORK","_number":3,"created":"2026-03-11 17:32:33.000000000","uploader":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"ref":"refs/changes/05/979705/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/05/979705/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/05/979705/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/05/979705/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/05/979705/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/05/979705/3"}}},"commit":{"parents":[{"commit":"ed800a2ebeb588745a933e45fefc842f77a4ae97","subject":"Merge \"Use project_id key in network plugin code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ed800a2ebeb588745a933e45fefc842f77a4ae97"}]}],"author":{"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","date":"2026-03-09 17:09:28.000000000","tz":60},"committer":{"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","date":"2026-03-11 17:31:49.000000000","tz":60},"subject":"Fix GW chasiss LRP name parsing if added via \u0027ovn-nbctl\u0027 command","message":"Fix GW chasiss LRP name parsing if added via \u0027ovn-nbctl\u0027 command\n\nThe \u0027ovn-nbctl\u0027 uses the \u0027-\u0027 (hyphen) as a separator between the LRP\nport UUID and the chassis hostname when adding a record to the\nGatewat_Chassis table in the NB DB.\n\nOn the other hand, Neutron uses the \u0027_\u0027 (underscore) for the same.\nAs a result, Neutron can\u0027t find LRPs on chassis which were added\nwith the \u0027ovn-nbctl\u0027 command.\n\nThis patch modifies the regexp that is used for parse these names,\nto match both the \u0027-\u0027 and \u0027_\u0027. The new regexp searches explicitly\nfor the 36 char UUID.\n\nCloses-Bug: #2143751\nChange-Id: I69e9fc00bfc58fb86675f72f94d10d014f25cd3a\nSigned-off-by: Marcin Wilk \u003cmarcin.wilk@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fc037c657ed3d6ead6ad4c3b8724b789ebaff44b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fc037c657ed3d6ead6ad4c3b8724b789ebaff44b"}]},"branch":"refs/heads/master"},"5a7308b776a69d100a058c842514f6e7bcbd236a":{"kind":"REWORK","_number":4,"created":"2026-03-16 11:35:25.000000000","uploader":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"ref":"refs/changes/05/979705/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/05/979705/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/05/979705/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/05/979705/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/05/979705/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/05/979705/4"}}},"commit":{"parents":[{"commit":"ed800a2ebeb588745a933e45fefc842f77a4ae97","subject":"Merge \"Use project_id key in network plugin code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ed800a2ebeb588745a933e45fefc842f77a4ae97"}]}],"author":{"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","date":"2026-03-16 10:48:02.000000000","tz":60},"committer":{"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","date":"2026-03-16 11:34:43.000000000","tz":60},"subject":"Update OVN doc about \u0027ovn-nbctl\u0027 command usage regarding the Gateway_Chassis registers","message":"Update OVN doc about \u0027ovn-nbctl\u0027 command usage regarding the Gateway_Chassis registers\n\nNeutron does not support adding or modyfing the Gateway_Chassis\nregisters in NB DB using the the \"ovn-nbctl lrp-set-gateway-chassis\"\nor the ovn-nbctl set\" commands. This patch adds this information\nto the Neutron OVN documentation to increase the awareness of the\ncloud administrators.\n\nThis patch also adds extra logging if the Gateway_Chassis register\nname is not parsable by Neutron.\n\nCloses-Bug: #2143751\nChange-Id: I69e9fc00bfc58fb86675f72f94d10d014f25cd3a\nSigned-off-by: Marcin Wilk \u003cmarcin.wilk@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5a7308b776a69d100a058c842514f6e7bcbd236a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5a7308b776a69d100a058c842514f6e7bcbd236a"}]},"branch":"refs/heads/master"},"1a75ad91f9a1f01f7bfa4aae4d82883aa9af0f1a":{"kind":"REWORK","_number":5,"created":"2026-03-18 07:36:38.000000000","uploader":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"ref":"refs/changes/05/979705/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/05/979705/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/05/979705/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/05/979705/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/05/979705/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/05/979705/5"}}},"commit":{"parents":[{"commit":"ed800a2ebeb588745a933e45fefc842f77a4ae97","subject":"Merge \"Use project_id key in network plugin code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ed800a2ebeb588745a933e45fefc842f77a4ae97"}]}],"author":{"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","date":"2026-03-16 10:48:02.000000000","tz":60},"committer":{"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","date":"2026-03-18 07:35:41.000000000","tz":60},"subject":"Update OVN doc about \u0027ovn-nbctl\u0027 command usage regarding the Gateway_Chassis registers","message":"Update OVN doc about \u0027ovn-nbctl\u0027 command usage regarding the Gateway_Chassis registers\n\nNeutron does not support adding or modyfing the Gateway_Chassis\nregisters in NB DB using the \"ovn-nbctl lrp-set-gateway-chassis\"\nor the ovn-nbctl set\" commands. This patch adds this information\nto the Neutron OVN documentation to increase the awareness of the\ncloud administrators.\n\nThis patch also adds extra logging if the Gateway_Chassis register\nname is not parsable by Neutron.\n\nCloses-Bug: #2143751\nChange-Id: I69e9fc00bfc58fb86675f72f94d10d014f25cd3a\nSigned-off-by: Marcin Wilk \u003cmarcin.wilk@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a75ad91f9a1f01f7bfa4aae4d82883aa9af0f1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a75ad91f9a1f01f7bfa4aae4d82883aa9af0f1a"}]},"branch":"refs/heads/master"},"d781fac878700bc7e87992d731ce4c0c59c64b43":{"kind":"REWORK","_number":6,"created":"2026-03-26 12:33:06.000000000","uploader":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"ref":"refs/changes/05/979705/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/05/979705/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/05/979705/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/05/979705/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/05/979705/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/05/979705/6"}}},"commit":{"parents":[{"commit":"ed800a2ebeb588745a933e45fefc842f77a4ae97","subject":"Merge \"Use project_id key in network plugin code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ed800a2ebeb588745a933e45fefc842f77a4ae97"}]}],"author":{"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","date":"2026-03-16 10:48:02.000000000","tz":60},"committer":{"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","date":"2026-03-26 12:14:10.000000000","tz":60},"subject":"Update OVN doc about \u0027ovn-nbctl\u0027 command usage regarding the Gateway_Chassis registers","message":"Update OVN doc about \u0027ovn-nbctl\u0027 command usage regarding the Gateway_Chassis registers\n\nNeutron does not support adding or modyfing the Gateway_Chassis\nregisters in NB DB using the \"ovn-nbctl lrp-set-gateway-chassis\"\nor the ovn-nbctl set\" commands. This patch adds this information\nto the Neutron OVN documentation to increase the awareness of the\ncloud administrators.\n\nThis patch also adds extra logging if the Gateway_Chassis register\nname is not parsable by Neutron.\n\nCloses-Bug: #2143751\nChange-Id: I69e9fc00bfc58fb86675f72f94d10d014f25cd3a\nSigned-off-by: Marcin Wilk \u003cmarcin.wilk@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d781fac878700bc7e87992d731ce4c0c59c64b43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d781fac878700bc7e87992d731ce4c0c59c64b43"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
