)]}'
{"id":"openstack%2Fneutron~980132","triplet_id":"openstack%2Fneutron~master~I40ad1e0dbc3a29c2fa864264c7de7f32b19be051","project":"openstack/neutron","branch":"master","topic":"use-neutron-lib-TYPE_PHYSICAL","attention_set":{},"removed_from_attention_set":{"31291":{"account":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"last_update":"2026-03-12 12:08:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I40ad1e0dbc3a29c2fa864264c7de7f32b19be051","subject":"Remove TYPE_PHYSICAL constant from neutron","status":"MERGED","created":"2026-03-11 17:28:58.000000000","updated":"2026-03-12 12:11:03.000000000","submitted":"2026-03-12 12:08:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"980132-use-neutron-lib-TYPE_PHYSICAL","meta_rev_id":"4baded5195189cb57c031ecde83ee183c70a89df","_number":980132,"virtual_id_number":980132,"owner":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-12 12:08:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-03-12 07:42:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-03-12 07:42:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-11 18:18:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-11 19:33:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-12 07:42:48.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"62a3a89e369d1eecb5e9df3ec4518c5036396132","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-03-11 17:28:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"37986fb6b2d112292875c49ce3807946beb6361c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 18:18:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5956b1111fee4f6f992f9ce6d1fc87f6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ca985d1a26c940fc937287424b42ab1a : SUCCESS in 24m 34s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/87ed146164c24cc4a8f2e6986b7b1fa5 : SUCCESS in 25m 11s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/6a8dfcc4c4c0400a9d295eed95a9158e : SUCCESS in 27m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d06e0acbbd5e86d70fbfc049545daa0f9c38426a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 19:33:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c277e54da2a141cda61dce53604ae8d3\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ed1e7454a4804018b51e143d37038c5b : SUCCESS in 1h 01m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e884a626b2fa4042a79002f24a13bc05 : SUCCESS in 1h 05m 10s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/241ebb24de5f45b8a247f594174c23c5 : SUCCESS in 2h 02m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d8b1abe9602425fb48bdf1670cef003 : SUCCESS in 37m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a26c5dbf6d04a629997bb132fdaf3c4 : FAILURE in 2m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/60fb17399b8c44dd98d8fbbdda71f157 : SUCCESS in 19m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/929ccfb093ba415db34ac7e0da5d8ce3 : SUCCESS in 15m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/85ddf9979fd54d36baea5d9e7efac154 : SUCCESS in 21m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e0982a211f34e41bb4bdd31ec8b8a97 : SUCCESS in 6m 57s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/316761f5cea94718a689dc45afadeae8 : SUCCESS in 35m 22s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/37a6c482875447caae3735a91fef2d1f : SUCCESS in 1h 27m 22s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/42d625e9c08c405e94caa014c3081ecf : SUCCESS in 34m 21s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/db306c22a87245a5add39ac90fa6f22a : SUCCESS in 1h 17m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a2220baa11be448d8932b6b646a5cd0b : SUCCESS in 1h 24m 26s","accounts_in_message":[],"_revision_number":1},{"id":"f8f657f70c82be40f21faec05a1f058d9a088f90","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-03-11 19:55:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"37561ded73d849371cf0d1e9b192b5d0bcc1fd47","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 20:36:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93d4e2f06d4c4775a4475ce915a58691\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d31dcc6fe70c42e585931a754839c669 : SUCCESS in 26m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2e8ddc570ded4266a8da6cd3c378ab54 : SUCCESS in 18m 24s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/cc2bd371222044ab84031b40bf6db9f4 : SUCCESS in 37m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"266b7794879fe76e35852e9068599cdee64e4246","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 21:42:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3fb0d9436b4242438ec3693d482ea8e4\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0943a8625f984eeeb432b1ef67366fad : SUCCESS in 1h 36m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aa71b72ffb374732a211ee19a7f48afc : SUCCESS in 1h 08m 50s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4f0850dc0b234935ac592f7dcfab6efb : SUCCESS in 1h 45m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05d767c9eba146df987088a09a2ec125 : SUCCESS in 1h 05m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/221e68f2a2624766946a9a945510336d : SUCCESS in 27m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/284554f4416d44969ad1af3aaeac52b5 : SUCCESS in 34m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b8382a7b9dc54efe94ba0804e34bf81c : SUCCESS in 27m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/24636d2feabc4eb4953d7c385791e6ae : SUCCESS in 23m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdc0d2729ad04e1a859ad01dad5cc48c : SUCCESS in 12m 18s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/6664f5bd59c246a69187825b80a87998 : SUCCESS in 1h 15m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3244999ff9c14e73b3e4c9203434b2f7 : FAILURE in 30m 42s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d9adeecdc2314b2ab23965edbf1cedc9 : SUCCESS in 1h 12m 33s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/bdb1fe15340d42b0a73c3c6a24b13901 : SUCCESS in 52m 53s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/50c6693bdbea49428824bc4cbd74691e : SUCCESS in 1h 21m 41s","accounts_in_message":[],"_revision_number":1},{"id":"9114d84ed41fff8baf0bd48de1f08bd036612d50","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-03-12 05:15:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c5a6f836b010efb7e7939d7c3033a15b2a04c35c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 05:52:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa60ec0b57f44e83ae332ffa436bb951\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9d8ce44c194944a39728c9dea774c201 : SUCCESS in 36m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2a4e873c1c064d0ba1debbcb6f60cd6b : SUCCESS in 18m 44s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8a78f56e24d843cda371c19c41b4331c : SUCCESS in 24m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"11a46bee792c61cf62b74191cadbb981cdac8408","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 07:07:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/54c22dd58a9d4227a52ec8c4bd5b85c1\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5e966bee979a479c95d50dcc3dd5002c : SUCCESS in 1h 03m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8b24a184512a4befae10baae0fd3b1c9 : SUCCESS in 32m 25s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b8e927681e5c49d0a8b7007ba14ce0c1 : SUCCESS in 1h 51m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d222cad810804e769738fe3e7cc16374 : SUCCESS in 1h 03m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/872fa211d12d49cb9b9c869e9010f35f : SUCCESS in 22m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/935c68c99b174c718162b32dddd5f78d : SUCCESS in 35m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/833d5743ca644f83981eff46c9700147 : SUCCESS in 15m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0c6085c26a644b74901b78871a188a23 : SUCCESS in 34m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9382bd7b352e452da88daa806a9f54c2 : SUCCESS in 6m 43s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/766a8debd61c45ee92c3e8f26d0e2179 : SUCCESS in 1h 14m 12s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/bff7948c2dc34027a99c6eadda467bf1 : SUCCESS in 1h 25m 43s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/109f4ab7e3cc44be8cbcd92b234ffe61 : SUCCESS in 1h 13m 49s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6843b97ecdf0488e9e640b52dd66a762 : SUCCESS in 1h 04m 11s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cfe97779cced481b8e6a0a5842b095ee : FAILURE in 15m 15s","accounts_in_message":[],"_revision_number":1},{"id":"e12ad274e02b9f3c35be21a1c807070886cb980f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-03-12 07:31:45.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"6d93926837092b83f13a9cc46423502a51768b03","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-12 07:42:48.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3ceeda0614700d030671c1fddcb1907ed28994ef","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-12 07:42:52.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a95fb91ec9c110103e3e85117556f8d65b29ee79","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 08:21:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e880f5456c864a83b45957f216d57396\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9b85d56e82354942b4bf87c8d6af2d6b : SUCCESS in 24m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b14d28cc807e4d4489806399543d05ec : SUCCESS in 19m 26s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0288f13c8a1b47f48eaa69de4406a857 : SUCCESS in 32m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dad887cbc88650ecb83d15441b2e9234f0a1e391","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 09:58:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e85f0fa821e345238cf7906c8f0f6f21\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/74c18f4900ac42e8b25a3fb55ecf062a : SUCCESS in 1h 21m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ea853ae32f33444687495b1755889c15 : SUCCESS in 1h 08m 23s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4a12a6f60deb458ca448a48327d09934 : SUCCESS in 2h 25m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b0fdf8592344cfbbfdb3563f4e24782 : SUCCESS in 1h 20m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d72a645605034b06b90b2efbd00878ca : SUCCESS in 24m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3d5e2be5034b4dae80cbd01df35eaa1f : SUCCESS in 34m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9ec41fea23844fdd9c1095df576d3e47 : SUCCESS in 21m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/46f5bcc1dfda40709b3e12ada7763d42 : SUCCESS in 26m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c94ccca19e684e8689765f209bd681df : SUCCESS in 7m 11s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f4045f6470b44862886dd610f5f07775 : SUCCESS in 51m 28s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/35cd7604ea1342e8bef904e95f34a2ee : SUCCESS in 1h 09m 50s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/16288a8f86a74da3838bc4cbc885d5b5 : SUCCESS in 1h 14m 55s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5308e81247d1444b996df9b518495afe : SUCCESS in 1h 25m 05s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/909121c8aaf847b4866cade0dddf4997 : SUCCESS in 1h 28m 41s","accounts_in_message":[],"_revision_number":2},{"id":"911ec25299a001d93362134d40c13bcf2f50ce73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 09:58:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"16fa7209de731aa65042756940c25e0e0c9b3cfe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 12:08:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d95d6904f4e24ec6b7787770de3cf776\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/47d8c0f7719a460a96b3062fe88b7eb4 : SUCCESS in 1h 11m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8ed581e3120e444eb3946d2a9dcb5074 : SUCCESS in 53m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a0d50bc60d494b6cb954a667f1e46875 : SUCCESS in 2h 01m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26997e6b1cc046a1b42167d84584f0fc : SUCCESS in 11m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5b3ebe9266cc4002b94f291b1bf61546 : SUCCESS in 33m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/18ed56c575044b2fa833907efdb0a863 : SUCCESS in 25m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3299b9802b5f4d92b8cb452cda7b5fdc : SUCCESS in 13m 26s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/71ff2ffba32546e2ba157c3fe19d5a2e : SUCCESS in 42m 22s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/29ee2d9be3de4e6eaab12bfd63382efd : SUCCESS in 35m 35s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/74bb17e613104429ab35cafb19bfcf69 : SUCCESS in 1h 03m 41s","accounts_in_message":[],"_revision_number":2},{"id":"80de1876462cd267a4098b871481a52d8f67b984","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 12:08:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4baded5195189cb57c031ecde83ee183c70a89df","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 12:11:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2a713beb2ae4e31b85ea67ca12c4c69\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae0c52eff0924818b89f3fce5bbe6417 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"531658e951f2c5dffed09e1af623336d0fbabc94","revisions":{"9a0524804f03e0009bbe14b19d36c381dde97d4b":{"kind":"REWORK","_number":1,"created":"2026-03-11 17:28:58.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/32/980132/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/32/980132/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/32/980132/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/32/980132/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/32/980132/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/32/980132/1"}}},"commit":{"parents":[{"commit":"59943108f5ea6b3349efdca4967ac50116c031b8","subject":"func: Workaround LSP bind command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/59943108f5ea6b3349efdca4967ac50116c031b8"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-03-11 17:28:09.000000000","tz":60},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-03-11 17:28:36.000000000","tz":60},"subject":"Remove TYPE_PHYSICAL constant from neutron","message":"Remove TYPE_PHYSICAL constant from neutron\n\nThis constant is defined in neutron-lib instead.\n\nChange-Id: I40ad1e0dbc3a29c2fa864264c7de7f32b19be051\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9a0524804f03e0009bbe14b19d36c381dde97d4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9a0524804f03e0009bbe14b19d36c381dde97d4b"}]},"branch":"refs/heads/master"},"531658e951f2c5dffed09e1af623336d0fbabc94":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-03-12 07:31:45.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/32/980132/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/32/980132/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/32/980132/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/32/980132/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/32/980132/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/32/980132/2"}}},"commit":{"parents":[{"commit":"030a5e171a591025d76eb448f1db8080ac63b02a","subject":"Merge \"Workaround(WO), cast netaddr.IPRange to a string when logging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/030a5e171a591025d76eb448f1db8080ac63b02a"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-03-11 17:28:09.000000000","tz":60},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-03-12 07:31:45.000000000","tz":0},"subject":"Remove TYPE_PHYSICAL constant from neutron","message":"Remove TYPE_PHYSICAL constant from neutron\n\nThis constant is defined in neutron-lib instead.\n\nChange-Id: I40ad1e0dbc3a29c2fa864264c7de7f32b19be051\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/531658e951f2c5dffed09e1af623336d0fbabc94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/531658e951f2c5dffed09e1af623336d0fbabc94"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
