)]}'
{"id":"openstack%2Fneutron~980303","triplet_id":"openstack%2Fneutron~master~Id82bdac4038064b0386fd0d2920320fbd60290f4","project":"openstack/neutron","branch":"master","topic":"db_branch_2026.2","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-03-19 20:04:26.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}},"hashtags":[],"change_id":"Id82bdac4038064b0386fd0d2920320fbd60290f4","subject":"Open the 2026.2 (Hibiscus) DB branch","status":"MERGED","created":"2026-03-12 15:08:11.000000000","updated":"2026-03-19 23:55:42.000000000","submitted":"2026-03-19 23:54:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"980303-db_branch_2026.2","meta_rev_id":"4817c50e1bf05ea2437a1b9ba6411be651788182","_number":980303,"virtual_id_number":980303,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-19 23:54:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2026-03-17 15:55:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2026-03-13 02:46:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":1,"date":"2026-03-17 15:55:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-12 16:04:31.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-03-12 16:04:31.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2026-03-12 16:04:31.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2026-03-12 16:37:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-12 17:19:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-13 02:46:17.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-03-17 15:55:09.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"6f74057cea226507b879f50c95fbd66eafaece31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-12 15:08:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a1a1a766315dc3f4cf42172c8a7e3f504a675bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 16:37:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/063322b24f3b49c091ab9a9ac329de0e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ee39634dcbfb4dd3aadbef50ad883cde : SUCCESS in 30m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/aea5f24ffe264569a1696f3916c7deff : SUCCESS in 29m 25s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8e699c1aa27c43b4835a207f3c293bfb : SUCCESS in 30m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cfac81c3985375fbd90f6916d03088fc434ed9f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 17:19:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f2a8391c3ab43468f3de72b7f4e20cb\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e6418c2ba2314a5caa02c66f4731b872 : SUCCESS in 1h 18m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d0f506a7869c48b69cf8e4de6742340d : SUCCESS in 1h 00m 43s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/eb98bb1f5d8d4a9e9bf3c348a31971c7 : SUCCESS in 1h 29m 12s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2d1724396340476a98f0e703f246418e : SUCCESS in 2h 10m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fb5a539f89fd4f5980645a35e73e9960 : SUCCESS in 32m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a44d859bc89548c593b3c54016a4cc6d : SUCCESS in 51m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4add7699a02436ab700ca4bc01c7808 : SUCCESS in 11m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01c2ecdb43ed46409de6341c5e4e208c : SUCCESS in 37m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fcd9d3243c4e409e932cc82feceeeca2 : SUCCESS in 19m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/90cbc2cd3bcf4b31b441331bdf147b6a : SUCCESS in 26m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e9dfa10026d4e09b5b6baa6dd6aa2bc : SUCCESS in 11m 33s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/39615756c83348b6b339976fa93470c8 : SUCCESS in 1h 01m 02s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f72219ebb29e45719b9e4857836c9840 : SUCCESS in 42m 25s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/462c43da128b407ab106f4921bc495cd : SUCCESS in 55m 01s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4fed041494d44f8e945c2c276bc30c92 : SUCCESS in 46m 06s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/3a06d6205e334b339793c4fed8866d23 : SUCCESS in 1h 08m 32s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/bcc75c24d2454b4e8a9882ca4f8c9f65 : SUCCESS in 1h 57m 24s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d2774ec8299d4310b30c85485bf8591f : SUCCESS in 1h 34m 36s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1c0fca4b1e864f33a85c14d0c5fee719 : SUCCESS in 1h 05m 02s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1730c80fe7c447778d237abbda8acfac : SUCCESS in 1h 14m 32s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4275cf87ac1c4fa190baab4e7891aab6 : SUCCESS in 1h 31m 20s","accounts_in_message":[],"_revision_number":1},{"id":"a6561d052e1754f155403a81edd1fcd34aa5ea49","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-03-13 02:46:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c1d34a7d3342b14e12a3f481aa7cdb0ad651baa6","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-03-17 15:55:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"516a1e75586bdaec43f6100477724b5f3b4a7293","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 15:55:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0ea5e44722b6b6cc97b6611b70336292a7c30e95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 18:03:05.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/84585c9ba28e4e0b925cec524ee61cc8\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/81c0ac0a7dbc44caa60790527d2e4a53 : SUCCESS in 1h 13m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/72b9d05a061b4d52bbd4bdea4127979b : SUCCESS in 37m 53s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a2f06b40225e42a39b6a31561dc18871 : SUCCESS in 1h 32m 19s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/67382c9b2c544945be06387f74aa9b22 : SUCCESS in 1h 47m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06610e7d8d03425a93ff0070585d23e7 : FAILURE in 6m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3571e27c4064433b9510ea7ad61852f8 : SUCCESS in 25m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/af3efee7009146d6b6641ec277704323 : SUCCESS in 21m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/565b8a3e004c4fba9bb23114125d2a9b : SUCCESS in 11m 30s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e58c6889cbe741ae9cf254f1916d08cc : FAILURE in 25m 59s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/09523a9546bc43e0a4f16a9e80131cc4 : SUCCESS in 2h 02m 32s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/5f6b8c4b508740b580b5534ad1122490 : SUCCESS in 1h 10m 55s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6dfffce25713428fa32a37a7e1e3b7d8 : SUCCESS in 1h 10m 23s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3ffc687ab4d64c6384e02931b289ab69 : SUCCESS in 1h 02m 52s","accounts_in_message":[],"_revision_number":1},{"id":"de928fd588ad8d3799596c5ada9fc0dfd0e3982a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-03-17 18:21:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"441a035ac38587bcea7a7d226a1974fd30ad7187","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 20:33:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/86990d5041d94048a91040808e9033c7\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/42599dfff8e346d28ca890b4be65897d : SUCCESS in 1h 25m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a8af041472bf47279370ca465dcdc65a : SUCCESS in 59m 15s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e9186caef10f4409bc7cc8461c783d5d : SUCCESS in 1h 36m 51s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/81fb856c9cc440c6a83d3ae7a4ba229d : SUCCESS in 2h 09m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c1b6d6e3df72424e8cf016b2c6e86a5b : SUCCESS in 33m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49261511c45f4b9b8771a533bd342ad3 : SUCCESS in 45m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c05378f6631849f3b05e8cc7b24ff497 : SUCCESS in 12m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eaeea371064c43fdbff707b15c559e78 : SUCCESS in 21m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2852ab0a0c654184afdf6ff706927cb4 : SUCCESS in 16m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b74947810f8748c8980899fe269dc3a3 : SUCCESS in 26m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65f2d89c5827464b816fa36671e83452 : SUCCESS in 10m 15s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/58724beb26e34f86a9abb4c839964f40 : SUCCESS in 1h 01m 56s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/38784ac23b224cf7abd29e85300a331f : SUCCESS in 1h 12m 44s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2fb24fd10c5647c59838d82182fb3012 : FAILURE in 27m 43s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3e22b53e1db14df88bb35497b1c39dbe : SUCCESS in 48m 27s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/248074062fe3461281f5596932756158 : SUCCESS in 41m 46s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/abec719e54ed4e77b76597fab79da4ee : SUCCESS in 1h 54m 02s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/719b653cdd094c508f0f023e16188114 : SUCCESS in 1h 02m 53s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e1c5edb6a7da4b1896f6717a7f66331f : SUCCESS in 1h 11m 20s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7504b50114dd47378fce42caed7ccca7 : SUCCESS in 53m 48s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c86dd1d3b44145b88d0ba9cf4bbec3af : SUCCESS in 1h 21m 58s","accounts_in_message":[],"_revision_number":1},{"id":"8e2e1f0d5f065e2afe4eed1e104fbbc569de14af","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 20:41:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f46a55cf8da4fa7bda1d7666a675dbc\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/66f09c230f274ed582ab9cab8bc3500a : SUCCESS in 26m 20s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/14c34c38a4d64cdcb82b68e88e8eaa85 : SUCCESS in 37m 04s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c840682a8da34172b22917bd698e2f80 : SUCCESS in 33m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b92f899c834930ef6ec492c42f7d9b306a1903e5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-19 07:09:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d66cf32af1c5429fab7a1fcaadd7fb84423988e4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 08:09:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae1cf0dc42a0440796ecc33d4c6537b9\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/18e0302b918c4022849b79f8ae3f24b3 : SUCCESS in 58m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/50b1393130a14805b530c52e40e42945 : SUCCESS in 19m 44s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d826c893a0b8423aa511c20f215cd8c1 : SUCCESS in 27m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d9eafa79708bbbd4682a5ddb235b8d6d51625a49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 09:48:35.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6d179ceca96a4010868a286978075a5f\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8c9ae3b866984244ae8a27e01cadf8d6 : SUCCESS in 1h 02m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/05f4babedee44e70acda1de059f72eff : SUCCESS in 1h 03m 01s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/833f1c53b9a44a2a8acfcfb71f66fac4 : FAILURE in 58m 18s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6e3f5ee2ee3f4c4590dbce7c91814c8c : SUCCESS in 2h 12m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a62a475778e04d028f3b083626c107dd : SUCCESS in 34m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/acf2b8c1d23f449eaf0658c28a0f209a : SUCCESS in 38m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4552d4a6131341228cbb66b4af5f9077 : SUCCESS in 28m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4109e98ce5d3465fbd4dd0b2e02205ab : SUCCESS in 34m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/529edefc7917463bbb5119d84332326e : SUCCESS in 15m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0cbc3ba0359f49e6bce5d540ff2bf678 : SUCCESS in 19m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/468982d8df9a47b4b0174c834e401f54 : SUCCESS in 11m 53s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ec286a906cf1488d9effb20218961a8c : SUCCESS in 1h 01m 41s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d2cb4f7efc0b4642a9867ae76f98aa12 : SUCCESS in 1h 08m 36s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a30d2bac7649467ba985c395b5c3b84c : FAILURE in 50m 16s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0a8bf6cd5bca476e9cefc93242d15af2 : SUCCESS in 55m 45s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/90e765497266436ca06063f78b3d7204 : SUCCESS in 1h 13m 45s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/249fa169416f427ba09e3e3c1bb071a4 : SUCCESS in 2h 28m 21s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/15f5589275404042b59540c538ead25a : SUCCESS in 1h 44m 54s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/21fbea38a988451da195c2eb5b5e4710 : SUCCESS in 1h 10m 10s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/60f84c8d85a24ac0a18aa99c99bfe9e5 : SUCCESS in 1h 06m 27s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/75456b97db2642a1bddb2845f14c6b24 : SUCCESS in 35m 22s","accounts_in_message":[],"_revision_number":1},{"id":"8a999f39c4f4ea2caddb0f589905a48b86d4987c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-19 20:04:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b70de014e195d6c8ce64c8c3e09b169f74396842","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 21:47:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32fe98721a4f416389361c58de1285ba\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6dc79c0e1d954f0b8d44cd957cbf45f6 : SUCCESS in 1h 29m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/75ad9248a972440bbda9676c86783c44 : SUCCESS in 1h 06m 18s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2af54dda858a4e14a38644e665905d49 : SUCCESS in 1h 28m 57s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5a1c2f934a434c6c8befa0d67be3d869 : SUCCESS in 1h 42m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/444ed3d920bb4612a4ade36c87a936a8 : SUCCESS in 31m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e88bdd78d16342778277c60c2328d6f5 : SUCCESS in 45m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6659605a24bf492ebda719aa9d96b583 : SUCCESS in 10m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2c3dc071ee9346bbb3bc31739e82f073 : SUCCESS in 32m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1213d41f1b6246a4b0724f66dc71a558 : SUCCESS in 24m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/96551f4339da438ba02df77b9305ee0d : SUCCESS in 36m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fbfb2ccbda34d838411f01a69c802a1 : SUCCESS in 5m 38s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3d90fc8e36ee40eeb6ebcd674a498ef0 : SUCCESS in 32m 22s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d02d4ab064df4a03b9ccf977a57fc63b : SUCCESS in 56m 21s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4ece141e4db0401d86efe8740b346816 : SUCCESS in 42m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/28f64dab2e9348cb97d07267f251214b : SUCCESS in 43m 54s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d94aca21e5b44e39b81b1f75c53577f9 : SUCCESS in 1h 26m 16s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/81e5c1a9801f4c9ab151178c33bdf5d2 : SUCCESS in 1h 01m 02s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/31f098cb33184d9abeeb360b807fa3aa : SUCCESS in 1h 00m 11s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c21d0202d66d48b996b02deda5b06d3f : SUCCESS in 54m 20s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7dd629d7175844a6a66fa6d36735dcbf : SUCCESS in 31m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3067ecbf6e30465c990e6f937e7238e7 : SUCCESS in 41m 47s","accounts_in_message":[],"_revision_number":1},{"id":"b916bd541537b6a675d014e2c34f4827794ba86a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 21:48:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"efcf0694d06210cd13d438cc0da1fa2e4c319589","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 22:07:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/174b51505a884fe19b404d9615eb0a87\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/faaa050241db425486a84e03e1a51505 : SUCCESS in 51m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/81f6d9b935d943d48e8795e78d50d3ec : SUCCESS in 38m 17s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e97d180b12604f1887ea248ddd87bfcb : SUCCESS in 49m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c33ca77a09257b1eac33e603020038736f4d8e03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 23:54:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd914885e21646bfbf3252cb241ea3fd\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1174fd6efd0b4195867af828b4170f65 : SUCCESS in 1h 11m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/950ef3b4db9d441ca968234a1a7252e0 : SUCCESS in 1h 03m 51s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3633658e8c96415eb052c47ac6fe1057 : SUCCESS in 1h 15m 33s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e89ceb4f58fe4765a96199080992bb08 : SUCCESS in 2h 05m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/944dfd52bc9c49c89c37112a537e7f10 : SUCCESS in 30m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ba65b202ada34159874dd7e19e18f0f8 : SUCCESS in 29m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c5b25acbd2ad49b68713f097d417f252 : SUCCESS in 32m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3126cb2ca63a4e50a243f45fae43689f : SUCCESS in 11m 47s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/db79c121a7f64cc49b87c17f3b5da724 : SUCCESS in 1h 24m 25s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/653ad3ec5c284d1ea873f8f845faf62a : SUCCESS in 1h 26m 19s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/f3179a35011440d0b1642c7a62f8e00a : SUCCESS in 36m 35s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/612787184a8b4409b16d000580f727b9 : SUCCESS in 1h 12m 44s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/28b9e00569fa43d4aa1ac28906ef4323 : SUCCESS in 1h 08m 12s","accounts_in_message":[],"_revision_number":1},{"id":"b21a25867540846660a20a06b807f1294183c39c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 23:54:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4817c50e1bf05ea2437a1b9ba6411be651788182","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 23:55:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e07b797b9ae94a7eab4cdf1122dfb1aa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90edeea533504f378a127386f7470a61 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"74104d59a74cc5abd9207c27446e4a7fa36b5c88","revisions":{"74104d59a74cc5abd9207c27446e4a7fa36b5c88":{"kind":"REWORK","_number":1,"created":"2026-03-12 15:08:11.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/03/980303/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/03/980303/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/980303/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/980303/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/980303/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/03/980303/1"}}},"commit":{"parents":[{"commit":"6eedb96017cad5ee1d63beca2577cee285ed3b29","subject":"Merge \"functional: Update compiled OVS and OVN versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6eedb96017cad5ee1d63beca2577cee285ed3b29"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-03-12 15:06:29.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-03-12 15:06:29.000000000","tz":60},"subject":"Open the 2026.2 (Hibiscus) DB branch","message":"Open the 2026.2 (Hibiscus) DB branch\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: Id82bdac4038064b0386fd0d2920320fbd60290f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/74104d59a74cc5abd9207c27446e4a7fa36b5c88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/74104d59a74cc5abd9207c27446e4a7fa36b5c88"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
