)]}'
{"id":"openstack%2Fneutron~980397","triplet_id":"openstack%2Fneutron~master~Iaa91c7b4556c701ec3a22054b35020697b0369cd","project":"openstack/neutron","branch":"master","topic":"bug/2143336","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-03-25 10:38:48.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2026-03-25 13:21:49.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-03-25 13:21:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iaa91c7b4556c701ec3a22054b35020697b0369cd","subject":"Revert \"Mark OVN L3 test unstable\"","status":"MERGED","created":"2026-03-13 08:28:17.000000000","updated":"2026-03-25 13:23:58.000000000","submitted":"2026-03-25 13:21:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"revert_of":979088,"submission_id":"980397-bug/2143336","meta_rev_id":"3307c4275cfeebab295bfca2efdb144418013a3f","_number":980397,"virtual_id_number":980397,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-25 13:21:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-03-25 10:38:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-03-25 10:38:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-13 08:28:17.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2026-03-13 08:28:17.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-03-13 08:28:17.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-16 10:00:36.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"9294973c6c06f64701189b8f8ae00520136b19ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-13 08:28:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a04ccdbc718e3e5d4cdb4bac2adff98aef38eaa8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 09:52:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/89d9da4328b547acb3f38a3861e50515\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d340536099d14ccda3df3de899fe8677 : FAILURE in 15m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2b9ed1591284902b4c5086e635066bb : SUCCESS in 20m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f042e22635404f12873906e1364b588f : SUCCESS in 11m 37s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6bfae10176a44b8ba4e9cacb31bd5079 : SUCCESS in 1h 17m 43s","accounts_in_message":[],"_revision_number":1},{"id":"f52e256bf1393f2f3cdf93fbf369d17102a1a4b8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-13 12:10:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"00744de68db3c744ad72f0fdad25bcfb593e570c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 13:27:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b0269d3e6864c93aca43d38627a5d28\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4934654856ee40668ffdb84df2fca1fa : SUCCESS in 1h 09m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0a061a8c24d4655b1109b0a6e0b6879 : SUCCESS in 26m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3613fee3fe2240d09ac3ab4da203fab5 : SUCCESS in 11m 31s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/57c873a820a34a57abb7284710e0b328 : SUCCESS in 43m 25s","accounts_in_message":[],"_revision_number":1},{"id":"80cda01c3256029d073c7c17ff505285a226a1a3","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-03-15 21:51:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"03f41c2b10384f9e97e28b987cd504fa017608f7","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-03-16 10:00:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7fd8be428b5734c0015bcc78d1b09a66966181ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 10:00:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"279806eb76efb4856735c6be5cdc7bdecfb339d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 11:17:30.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/992f1d0f895149369b2cd0788bbf6861\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bc3149ff0c624a9fbc9278c849059191 : SUCCESS in 1h 11m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0ac5a3769e34f138a3b45fce2def358 : SUCCESS in 25m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c89dbf79fc6454ba6053be360c5531a : SUCCESS in 7m 49s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e49b3389a9fc405cadd3314bd35a708b : FAILURE in 30m 14s","accounts_in_message":[],"_revision_number":1},{"id":"0d401d5f4caf8745c7220780c2b04181f33d7f1c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-18 15:16:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"61bc7944e9b129d5fb5fc1d23d1d172890a544c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 15:54:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/70bbc1d659f04b1ebd40b2c1f2d1bf69\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3f7d61055c7e4abfaa8a9a7777193f37 : SUCCESS in 33m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d529bd5cb774e8781bb6aec590f27a9 : SUCCESS in 27m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8810da2cf482443499ceaef4644bd84a : SUCCESS in 11m 43s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d9d36ccab89441659c7640307591966a : FAILURE in 25m 15s","accounts_in_message":[],"_revision_number":1},{"id":"b572f93c515592fd94fcbe982bb11fb8381d06ed","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-20 16:43:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e627243d7544b7f10b7d503c34b735775ccc3a4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 17:39:57.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b4b07b4876c84fd7ababe393fbc08fb2\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/579db71b4e204822bb8dad62661ff5b9 : SUCCESS in 52m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb0344bffd23420298a0933ef55bed3e : FAILURE in 7m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/510ff2e4d523404680b3893049f8768b : SUCCESS in 12m 17s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4d2806308c9346e4b9a01b411ded962b : FAILURE in 22m 38s","accounts_in_message":[],"_revision_number":1},{"id":"afa732948466b9e19df98c52995e6ca09f907f7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-20 21:20:01.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"02d1034f4f5a764df158899e9342a6262d0731d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-20 21:20:17.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"c754485bbe11e960088781e44ca27f187aa21d66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 22:32:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d6fde8fc5e141b78eb1aa56451d2ab8\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/45a709e989bb4938aaaa8be6f2a73c7a : SUCCESS in 33m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/858ca454a10d40439b7768e588c8e08b : SUCCESS in 26m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cc8b93e52b74ead92dc2a27b0e1cf80 : SUCCESS in 12m 28s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fd1de6c072ac4586854cd30962219e1e : SUCCESS in 1h 10m 22s","accounts_in_message":[],"_revision_number":3},{"id":"1ea7537b12add7d86a3fe05eaebd1598f5f36b78","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-03-20 22:34:59.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"fa738aed3effa13fec7a566009776eeb2ac36b3f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-03-25 08:24:54.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f24072361b44d84bcc8942f24df25efa11752cb7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 08:25:17.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1849ba76895afe48204111ec5cf800c3d92988f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 09:45:31.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fd0e747a3dad496880f08bbf332c5d1a\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/04822558ddb443a3b754c72674a3efa2 : SUCCESS in 1h 03m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/623403930c174d05a3284d09afb4b28c : FAILURE in 6m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c41a62efd0f43d18f2827a2e85243f7 : SUCCESS in 13m 36s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/dd755f9afbfa4fae8bb5efe37f57cad7 : SUCCESS in 46m 15s","accounts_in_message":[],"_revision_number":3},{"id":"663beabde305bca2bea128183bd6d1b632e808f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-25 10:37:59.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":4},{"id":"d4988794d698bcb177b57620dcd6742a693761ac","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-25 10:38:48.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1ebb0031218159f36606e5c55366a50763538710","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 12:09:29.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7d844fc760542449a69923d0af45822\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/afa001fe82a74cbc8e1452c214f1bf84 : SUCCESS in 53m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/239d6a9be3914ad5a72ae2f25115bf10 : SUCCESS in 26m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f9758a2134b448db2e7b51698a3a0a0 : SUCCESS in 11m 39s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/95d031835c5144c59e9666d915d01be7 : SUCCESS in 1h 26m 05s","accounts_in_message":[],"_revision_number":4},{"id":"cf0aabf250cab73dae8c848919eeb9c52145f419","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 12:09:55.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"64645de206a5bea6a2e89ef5d9db7218d6a78a7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 13:21:49.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4632aec185844d8c85d62f1a8011e713\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8c3885eabb794fbc9cca5f305c9a2880 : SUCCESS in 1h 05m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f60d8d969f274e67897d0fbc915ad8ba : SUCCESS in 17m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d411b523666b4ddd8f5c26cbe4d8cf75 : SUCCESS in 6m 35s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/61c05cf1facd4c2b8b597f54225f1f0e : SUCCESS in 1h 10m 35s","accounts_in_message":[],"_revision_number":4},{"id":"f9e9dc034cc4bc442875fe6fb5e934fd5c2f5766","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 13:21:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"3307c4275cfeebab295bfca2efdb144418013a3f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 13:23:58.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd19515ac56e4056a5e51dd7799b10c6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e92e0ea5b284080bde027c397d2c867 : SUCCESS in 1m 36s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"897da28052af4d60219f264ba59c03f1a52996b2","revisions":{"0325168377a810824cefcc49a7c93185272b05db":{"kind":"REWORK","_number":1,"created":"2026-03-13 08:28:17.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/97/980397/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/980397/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/980397/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/980397/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/980397/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/980397/1"}}},"commit":{"parents":[{"commit":"7157062caa4325e2ff912c39e0fea8d6497b49c5","subject":"Mark OVN L3 test unstable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7157062caa4325e2ff912c39e0fea8d6497b49c5"}]}],"author":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2026-03-13 08:28:17.000000000","tz":0},"committer":{"name":"Gerrit Code Review","email":"review@openstack.org","date":"2026-03-13 08:28:17.000000000","tz":0},"subject":"Revert \"Mark OVN L3 test unstable\"","message":"Revert \"Mark OVN L3 test unstable\"\n\nThis reverts commit 7157062caa4325e2ff912c39e0fea8d6497b49c5.\n\nReason for revert: issue fixed in [1]\n\nRelated-Bug: #2143336\nRelated-Bug: #2139271\n\n[1]https://review.opendev.org/c/openstack/neutron/+/978874\n\nChange-Id: Iaa91c7b4556c701ec3a22054b35020697b0369cd\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0325168377a810824cefcc49a7c93185272b05db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0325168377a810824cefcc49a7c93185272b05db"}]},"branch":"refs/heads/master"},"e9297cbb6cfbde7c15dad575e291c7e83ad5fc22":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-03-20 21:20:01.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/97/980397/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/980397/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/980397/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/980397/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/980397/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/980397/2"}}},"commit":{"parents":[{"commit":"8730dbd1599dc5778e7c92e9f004f770b7a58ffe","subject":"Merge \"bgp: Fix more pep8 errors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8730dbd1599dc5778e7c92e9f004f770b7a58ffe"}]}],"author":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2026-03-13 08:28:17.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2026-03-20 21:20:01.000000000","tz":0},"subject":"Revert \"Mark OVN L3 test unstable\"","message":"Revert \"Mark OVN L3 test unstable\"\n\nThis reverts commit 7157062caa4325e2ff912c39e0fea8d6497b49c5.\n\nReason for revert: issue fixed in [1]\n\nRelated-Bug: #2143336\nRelated-Bug: #2139271\n\n[1]https://review.opendev.org/c/openstack/neutron/+/978874\n\nChange-Id: Iaa91c7b4556c701ec3a22054b35020697b0369cd\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e9297cbb6cfbde7c15dad575e291c7e83ad5fc22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e9297cbb6cfbde7c15dad575e291c7e83ad5fc22"}]},"branch":"refs/heads/master","description":"Rebase"},"9280272c11bcffb9e238df4f39c04fb75321670d":{"kind":"REWORK","_number":3,"created":"2026-03-20 21:20:17.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/97/980397/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/980397/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/980397/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/980397/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/980397/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/980397/3"}}},"commit":{"parents":[{"commit":"8730dbd1599dc5778e7c92e9f004f770b7a58ffe","subject":"Merge \"bgp: Fix more pep8 errors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8730dbd1599dc5778e7c92e9f004f770b7a58ffe"}]}],"author":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2026-03-13 08:28:17.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2026-03-20 21:20:17.000000000","tz":0},"subject":"Revert \"Mark OVN L3 test unstable\"","message":"Revert \"Mark OVN L3 test unstable\"\n\nThis reverts commit 7157062caa4325e2ff912c39e0fea8d6497b49c5.\n\nReason for revert: issue fixed in [1]\n\nRelated-Bug: #2143336\nRelated-Bug: #2139271\n\n[1]https://review.opendev.org/c/openstack/neutron/+/978874\n\nChange-Id: Iaa91c7b4556c701ec3a22054b35020697b0369cd\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9280272c11bcffb9e238df4f39c04fb75321670d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9280272c11bcffb9e238df4f39c04fb75321670d"}]},"branch":"refs/heads/master"},"897da28052af4d60219f264ba59c03f1a52996b2":{"kind":"REWORK","_number":4,"created":"2026-03-25 10:37:59.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/97/980397/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/980397/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/980397/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/980397/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/980397/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/980397/4"}}},"commit":{"parents":[{"commit":"21b995177be4472797ba99b8570e9f142f11d3fe","subject":"Merge \"ruff: Enable preview rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/21b995177be4472797ba99b8570e9f142f11d3fe"}]}],"author":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2026-03-13 08:28:17.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-03-25 10:37:45.000000000","tz":60},"subject":"Revert \"Mark OVN L3 test unstable\"","message":"Revert \"Mark OVN L3 test unstable\"\n\nThis reverts commit 7157062caa4325e2ff912c39e0fea8d6497b49c5.\n\nReason for revert: issue fixed in [1]\n\nRelated-Bug: #2143336\nRelated-Bug: #2139271\n\n[1]https://review.opendev.org/c/openstack/neutron/+/978874\n\nChange-Id: Iaa91c7b4556c701ec3a22054b35020697b0369cd\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/897da28052af4d60219f264ba59c03f1a52996b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/897da28052af4d60219f264ba59c03f1a52996b2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
