)]}'
{"id":"openstack%2Fneutron~980439","triplet_id":"openstack%2Fneutron~master~I0148111a0f9576d778da4848a22ab14801e1650a","project":"openstack/neutron","branch":"master","topic":"bug/2144309","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2026-03-17 23:15:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0148111a0f9576d778da4848a22ab14801e1650a","subject":"ovn-db-sync tool can now apply config options required by plugins","status":"MERGED","created":"2026-03-13 14:33:40.000000000","updated":"2026-03-17 23:17:05.000000000","submitted":"2026-03-17 23:15:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"980439-bug/2144309","meta_rev_id":"f5977d4618f563b9df9c0e43573631a501fc2ac9","_number":980439,"virtual_id_number":980439,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-17 23:15:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2026-03-16 04:52:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2026-03-17 09:40:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2026-03-17 09:40:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2026-03-17 09:19:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}],"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-13 15:05:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-13 16:16:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-15 21:56:53.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-03-16 04:52:42.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2026-03-16 10:26:13.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2026-03-17 09:40:46.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"7fd657f896f12abba0ee761d9db3a8b411ed93ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-03-13 14:33:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"63594bcf284aa05984a1bc407f34f02918de414a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-03-13 14:49:15.000000000","message":"Topic set to bug/2144309","accounts_in_message":[],"_revision_number":1},{"id":"ca61fb98c0920ce68eca6a001ca7f3d67878e448","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 15:05:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83eb14404a6349e588300e882c4df677\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b4274b7b923b4f9dab337e6a3d7d1f7a : SUCCESS in 26m 45s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/45a8c5c3fd9848eeac372d45b6826efa : SUCCESS in 22m 06s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7849247c412e4c219ef189e62bce66eb : SUCCESS in 29m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f9a5502f1884b2485a9fb780b88013bbbf94c681","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 16:16:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af90d69bb25648bb849d71cd2a46dd96\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c841740756324e13876b749f2eb1c9ba : SUCCESS in 1h 17m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/934fe194aa6f48ef809f299beb116afa : SUCCESS in 1h 00m 39s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3d891c275c5941378747cbe600c9bca2 : SUCCESS in 1h 41m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e7cffed3a9e4a768961e9766cf199b6 : SUCCESS in 1h 03m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/192eb19d11164c53aeb764fac3e2375d : SUCCESS in 18m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8895ffb0c1c0499d855acb7515ceffaa : SUCCESS in 20m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/37fb4074112f426cae775eac70a41314 : SUCCESS in 23m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/71aa83c34a33468899ede7dadb4cbbce : SUCCESS in 31m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/808c1c52aaea4de1b42ec802cb284179 : SUCCESS in 11m 30s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/625c1892fdd34071a1141ec6fbe4b981 : SUCCESS in 1h 10m 36s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1a7d5401a086478f9b8c7b7513d69a5d : SUCCESS in 42m 13s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d4c2d5675db14a538a8832d123d640c4 : SUCCESS in 52m 47s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/98050148066d4a8ca0614518459e886b : SUCCESS in 1h 08m 10s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/662edbf02c6a4991876b2bf82e99b02f : SUCCESS in 1h 26m 45s","accounts_in_message":[],"_revision_number":1},{"id":"d3e89322b261d0cf2d6f31215f50f6ce99c38749","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-03-16 04:52:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a05e9a11ef3b6bbcd1c34ab3416864899f7b6d79","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-03-16 10:26:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d019a84442527535cc6b21a2fa6146c724ee5781","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-03-17 09:19:36.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"a3df946e92fa45046c07a4dcb05f979d9447cf19","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-03-17 09:40:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4cb756b49bced497b507676871db86d1df6ffaaa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 09:41:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"293da2af43f34e4b31aff409807dfa9e86386d98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 12:27:53.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/14fb34ba03b54025855a2ea59d473519\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/dc1735c4efc0446f81918026f3f3a99f : SUCCESS in 1h 41m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f5929338f3c34041816eccfd6c3a230d : SUCCESS in 41m 42s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4373f74fc217464193633202871b3cc6 : SUCCESS in 2h 46m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77947e63c1f847289b8f5cc1c0f1f66a : SUCCESS in 10m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/366e03465286475eae9710d26e2f014a : SUCCESS in 29m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/819a47067eb84afb9fbc5e3a2327a8d7 : SUCCESS in 23m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/771719027cfd463ea3e227d54ad82946 : SUCCESS in 12m 59s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a0bf3983cea741dbbb58c35e182fb5e9 : FAILURE in 42m 29s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/faa4fd9c0c7947f9b640436c66bb328d : SUCCESS in 1h 12m 02s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b3f1f757feb6498193be0959d473751e : SUCCESS in 1h 11m 35s","accounts_in_message":[],"_revision_number":1},{"id":"43520ac388a3713009939be3536cb6303601ef96","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-03-17 13:15:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0937b43ed7d8abdbb29e5b61e61a3d455eb9f449","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 15:13:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4db7257241d9474ab71cd64d652f6373\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3b309b58ab23412d91392f6319142af2 : SUCCESS in 40m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1f713ad4e012484881ba417e48ba2a4d : SUCCESS in 30m 39s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2e13dcd2a46744fdb5d4f79d3382130e : SUCCESS in 51m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"97a541f9db34d219901bc67ea4e41b4ac1ab9696","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 15:40:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3be9e1033bf945f18d41e1bbe78225b8\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/68f800b1e7ec4e1dab07d655b889ab8f : SUCCESS in 1h 39m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a8c66df06aa34f44aaa686ab2c6f931f : SUCCESS in 1h 08m 05s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8c4443c314d24deda0f4fe639f2d0000 : SUCCESS in 2h 21m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12200b3c073045a7ac9eb1a4a2f2b243 : SUCCESS in 45m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aec62bb6a6144ebaa917adf977357634 : SUCCESS in 26m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9451f0c9dd45451bb812afb393679d89 : SUCCESS in 34m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a17c743b478f40818bba0e471e6bec77 : SUCCESS in 24m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/16f15a8e59104ab3b3c3d9436433d424 : SUCCESS in 26m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e6217d3ab1a4f8790d9d3b48db56ce2 : SUCCESS in 12m 30s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/5b9fe3e1523f40fe9307538cd1ee23f5 : SUCCESS in 36m 34s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2cdf79d8ab864acd978d93c63dce00d8 : FAILURE in 28m 22s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/b1463993e27f46628fa0e8a0cec9a533 : SUCCESS in 1h 02m 29s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3986fa6d7e5547d29ac782038d6f4286 : SUCCESS in 59m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6607aa91354141f2a8bcd6267e951ac6 : SUCCESS in 1h 31m 38s","accounts_in_message":[],"_revision_number":1},{"id":"ac637597fe3baac98a63f98ff3f10ce643f8dfe4","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-03-17 18:23:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2d7d996aa45c19414d3609744ca9922c99f010b2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 20:33:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b43c37d30c34b6ba0e894046ebf74f6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5264e9d3d74d4fdaa0e4fab1a7ec25e3 : SUCCESS in 27m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f16c1835531f485c9a428bd6edb1510a : SUCCESS in 19m 09s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8c90d54075af45fb9c91f6a20c7f6dc8 : SUCCESS in 26m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c698caf77a027b5dc8be180d98b4bd9fc0db6b3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 21:02:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83c66c93f1a74cc9b2ab211ee5023cec\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8a5438a078e44353a0dee09bbdba5f03 : SUCCESS in 1h 05m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/03cb353407994523b9bce4c9535fd6ba : SUCCESS in 38m 20s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6bce6d3f21a54d7db05d5e7d11a39d5d : SUCCESS in 2h 34m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b7c292323a54cac864e7c61117610c3 : SUCCESS in 1h 03m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7ceb19e3931461ab11ab43e3d3235d6 : SUCCESS in 11m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2bdb73d4b4524ba5a7c21b8fe65aecd6 : SUCCESS in 40m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9acf9e4c57954fae996f02cb2094b30d : SUCCESS in 25m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2e0cadda9a7c441b858a283e32a4ddaf : SUCCESS in 29m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12da4756338f44ac914be251a871cdf7 : SUCCESS in 13m 18s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/018e104e569a413eb9bf96d895da0bdb : SUCCESS in 42m 29s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/04418431507b465d99f975db05d08955 : SUCCESS in 1h 27m 28s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/432b7734c3f240fa806026c2f20f7edf : SUCCESS in 1h 13m 03s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/dfe023f041674cbb80b0817babbca770 : SUCCESS in 1h 18m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/27ade341ed294802bd80776396b7e418 : SUCCESS in 1h 26m 37s","accounts_in_message":[],"_revision_number":1},{"id":"e1c90c9f80f533551583e9edcf7435d81b2c45b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 21:02:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d52b0db646f8f3e32f479418415d3d9b8b097592","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 23:15:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2aac71db00b34ed3a28ae127e8b30980\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/efd881253f10423e983a7c3c38ff4415 : SUCCESS in 1h 17m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a0a98e718de644d9854a91c7693ce28e : SUCCESS in 1h 07m 47s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e478fed3bcaa43479bfc8380bbbd568e : SUCCESS in 2h 12m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/505e570ae7ae43f4b446e08eae83ea07 : SUCCESS in 26m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/14b2257a12d244b8999d4acd5415afc2 : SUCCESS in 30m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/09efd28c5b244be69f13316f8e217979 : SUCCESS in 24m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4febcdf5dbe847678d9398094fa9878c : SUCCESS in 13m 23s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/676ff8e591bf4d989b2b4db73fe4c130 : SUCCESS in 1h 24m 08s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ce5d946ddbee445c9d77420789d2bef8 : SUCCESS in 41m 46s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a455570d891f4c909d3985c2f67362c9 : SUCCESS in 55m 04s","accounts_in_message":[],"_revision_number":1},{"id":"e0b568f7a9246b7d9e7a058a2a9caae88a5064da","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 23:15:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f5977d4618f563b9df9c0e43573631a501fc2ac9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 23:17:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7424beb8397e4d41a882d03752cd7454\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eba041d9824d4966a1b244a03a9b3627 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0c3e535ae13295d966f2e5b73736590477bf6c48","revisions":{"0c3e535ae13295d966f2e5b73736590477bf6c48":{"kind":"REWORK","_number":1,"created":"2026-03-13 14:33:40.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/39/980439/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/980439/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/980439/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/980439/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/980439/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/980439/1"}}},"commit":{"parents":[{"commit":"41a2a5ed8aa064a8a54ffaade108d34d9b04ba62","subject":"Merge \"Update master for stable/2026.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/41a2a5ed8aa064a8a54ffaade108d34d9b04ba62"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-03-13 13:54:44.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-03-13 14:33:28.000000000","tz":60},"subject":"ovn-db-sync tool can now apply config options required by plugins","message":"ovn-db-sync tool can now apply config options required by plugins\n\nWith this patch neutron-ovn-db-sync tool can apply additional\nconfiguration changes required by the third party db-sync plugins, like\ne.g. configuration of the service provider required by the neutron-fwaas\nservice plugin when FWaaS ovn-db-sync plugin is installed.\n\nRelated-bug: #2144309\n\nChange-Id: I0148111a0f9576d778da4848a22ab14801e1650a\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0c3e535ae13295d966f2e5b73736590477bf6c48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0c3e535ae13295d966f2e5b73736590477bf6c48"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
