)]}'
{"id":"openstack%2Fneutron~980962","triplet_id":"openstack%2Fneutron~master~Ie55523381cf6a833cc13348c7272806e5772dcb7","project":"openstack/neutron","branch":"master","topic":"bug/2111276","attention_set":{},"removed_from_attention_set":{"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2026-03-19 18:15:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie55523381cf6a833cc13348c7272806e5772dcb7","subject":"bgp: Move lrps_to_chassis_routers under commands.py","status":"MERGED","created":"2026-03-17 14:46:48.000000000","updated":"2026-03-19 18:16:23.000000000","submitted":"2026-03-19 18:15:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"980962-bug/2111276","meta_rev_id":"e841914149f3e7bd5fb0b7cf9e59af642774ff30","_number":980962,"virtual_id_number":980962,"owner":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-19 18:15:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-03-18 15:20:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-03-18 17:00:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-03-18 17:00:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-17 16:15:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-17 18:47:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-18 15:20:23.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-03-18 17:00:59.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"}],"messages":[{"id":"c7a5edbfcb0167696037940c6b492bbbaea21a4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-03-17 14:46:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"66861dfc06bbe0e2282e2c7ee11501e3be40a947","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 16:15:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70986b0fac2d46d097dc17147d50b9d6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ca9a008f12944c8cb4f97dbcfa6406e1 : SUCCESS in 31m 34s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6bf3be09111348bfaf7744b3cef14f4d : SUCCESS in 21m 43s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4a34b59991cb4765b181e84421577556 : SUCCESS in 26m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c7da3822a7f3227b99c0982f059bdbdcba9ff340","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-03-17 16:42:49.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"68d51e9d0e9748069217e9d18b61d79d9141cfc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 18:47:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c63aad485dc5423dbf4dec7caa1f1fac\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7993a593c9b74a3e93b72b8ab8a68e2a : SUCCESS in 1h 06m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a5983003be124105be78ccf4c0a64ae2 : SUCCESS in 1h 04m 56s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ed11353351d0486caf6ead94780ba5b3 : SUCCESS in 1h 26m 07s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b6ee38580b6a4818a0e97c4b93202abe : SUCCESS in 2h 00m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5e87494d5b4f422aada1c6f19e73a988 : SUCCESS in 32m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b33a3d0ff6b74978b504000412e1ac2a : SUCCESS in 1h 16m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e67a2e160408466182870ce924619cc2 : SUCCESS in 11m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b8829dec76744c68ae667eb3aac6a032 : SUCCESS in 33m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6782657fadda420c9da94e1288d61924 : SUCCESS in 24m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bf15e9a9ef1e4a3ab1a55d61337f6abf : SUCCESS in 22m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3968bd5be2be4a7cb339254bf9219a91 : SUCCESS in 9m 34s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f77341e510984d9ba5594ccacbdd4857 : SUCCESS in 47m 11s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2edf5a58ecaa4566a6a694aa29027d92 : SUCCESS in 54m 11s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/71248756bfe743968c0b1c1e4089bc5b : SUCCESS in 1h 06m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f167962337cd4d30b8612a584aabf5da : SUCCESS in 1h 07m 11s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/73e686967b2a4d00bba021e516da7447 : SUCCESS in 1h 19m 25s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/00141f9c26874125b2396e989e5ae6a5 : SUCCESS in 1h 54m 58s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/905084709b9449dfb7ade1fffc105a30 : SUCCESS in 1h 49m 29s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d5c331e02c9e4d5d8fc4b3fc9f5828ff : SUCCESS in 1h 02m 47s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b6ee1a0f7ced445699045dadb1a4e520 : SUCCESS in 1h 19m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2f0ad065b033420f9d867633e4bf6d52 : SUCCESS in 1h 26m 27s","accounts_in_message":[],"_revision_number":2},{"id":"3b2fdd19f68700f4496b4561123ba82a8b5db0f2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 20:01:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07cc0ea022ab476688bf890b40e7ce59\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/83e1d4e575f244b68de4a8a96e7a9a96 : SUCCESS in 24m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bfa2c890bb784982917250904cfd377e : SUCCESS in 18m 12s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5eeb842e47214ec2a169d9afa618993f : SUCCESS in 52m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"21240f4ddf9d4256f72329af8feb3c6b9d7d1525","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-18 15:20:23.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b81a1093b555451ebf49f385d8d8e4cbf8c0cf32","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-03-18 17:00:59.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"08554518b319789d65de7b9655ec9ad6c8581ca9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 21:47:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c07d3a96c45e4850eb5f5fae77db29796ea5ea30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 00:11:59.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9839a62e637d4bc9be043c9091af8527\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/55c7bc81b0ac45f9b74e2652be2235c8 : SUCCESS in 1h 25m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6e9a4b21eca0468cbfe655ca46843605 : SUCCESS in 1h 00m 12s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/762cd071cc0f4f48b83bad5463fc8cd1 : SUCCESS in 1h 35m 34s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bcf497de670c4986b33049cf07408a01 : SUCCESS in 2h 01m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/950a329ca5c94ca89d4e6bf84cd56425 : SUCCESS in 23m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f030908ec3e8421e9293a6d85d3ff1d2 : SUCCESS in 39m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9d670573d3d548ea9e946305439bc954 : SUCCESS in 22m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4b16959b66d4e3189c4c42ef45ba7a9 : SUCCESS in 11m 50s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/de062acb494e40239cf8bb2360875344 : FAILURE in 50m 17s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/17ee41f0923941079ed1734deb5f1a25 : FAILURE in 50m 27s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/29075d9a34914766a5325a38a040dae6 : SUCCESS in 1h 14m 16s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2104a8ae15fb4d789a0b489ef3f05e6b : SUCCESS in 1h 12m 56s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/933dbf1c61994d089c2ee0d46620e48a : SUCCESS in 1h 06m 01s","accounts_in_message":[],"_revision_number":2},{"id":"dd26adb68c7c2fd7dd73e4792a43c741353781bd","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-03-19 14:01:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1020eb294cc418ce1115ed99751eeb0e5eb95216","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 15:37:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b836e0f0b1e04e48a6a96068962ba3f4\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/94dd768eda2341958fd47d130dc399e0 : SUCCESS in 1h 03m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/85399d388cc544d4ae2b4551c5435042 : SUCCESS in 26m 47s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/fe8646f0841f49e4a9d5b67e80269c38 : SUCCESS in 26m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5b970b9acf06d75012f6375abee9c6cee64a276c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 16:02:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/186570d025ed4772a3c8a9b4ae639d16\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/01b9b76976914041a0a940c88b647bd1 : SUCCESS in 1h 03m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/515c2285e94040918be4b43b9b5e281d : SUCCESS in 55m 28s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d5e82689004e43ef82d8a23d52b74677 : SUCCESS in 1h 45m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/04ed0e598a8a4b15a4841985df71c1a1 : SUCCESS in 1h 58m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/aed6904f691f427b9c64acffdb240e99 : SUCCESS in 38m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/37c133f11b4743aeaa4a5f972d8206d7 : SUCCESS in 39m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b58b6c566f14007b6a3facf91778dc8 : SUCCESS in 27m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/00e71c226797422e940d94e469d3dcfc : SUCCESS in 26m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c04a3c6f04c34b038890c355652e9677 : SUCCESS in 24m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3df09a8ac004410ebb2c5d6fe598e4a5 : SUCCESS in 28m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b38fd5dee0584f1dabb2c288a39033b7 : SUCCESS in 5m 52s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d4490f8e93c545648bda32080bfcdd13 : SUCCESS in 1h 07m 12s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f1e381bd5b7f4a9a82dac19bf5f54266 : SUCCESS in 1h 19m 33s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d27d19982c0241d385fd40a70a6b00c1 : SUCCESS in 56m 59s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d3c93ae9b92e4441826128fe5cd89e87 : SUCCESS in 40m 57s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/b3f3f04e6044452e9dc1bf4c42b0d18c : SUCCESS in 40m 11s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5c23584fa88d400c94c66ca42dff6fee : SUCCESS in 1h 55m 59s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/15b2eeea718449cda88d01767c137e84 : SUCCESS in 1h 47m 44s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/050c24523a8c4403be4a99e8690cdf04 : SUCCESS in 54m 24s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/91d0110968bc498d8dd64a6797193613 : SUCCESS in 1h 20m 25s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/46d674147fd94effb965ec00c5c99738 : SUCCESS in 37m 10s","accounts_in_message":[],"_revision_number":2},{"id":"22039230b9b72eba3ac1d9cd138d7edf48de479c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 16:03:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4454acf75536a7ce3314a740996bf3b859f180f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 18:15:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8aec5ec3d1ae402da5e533898cd8f810\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5629ee46fe7a480ab1e2de34c7836a11 : SUCCESS in 1h 34m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e7675913f34345a6895b2cfb21f44c12 : SUCCESS in 1h 03m 37s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2c60b4b36c9d49519752b34b97481554 : SUCCESS in 1h 31m 55s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d6d2e74949f24269b37341d70b3b8445 : SUCCESS in 2h 11m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70d56f33a4c5432b88f2cf61c64735b4 : SUCCESS in 22m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6258f07fe3a441d7a0861e84f97431b6 : SUCCESS in 26m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/54e5dfa964c34c90b9679210e3d914c5 : SUCCESS in 39m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4d2262243e34a48809069f35c4fa9ec : SUCCESS in 7m 26s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7916edbe53ea4164ac643e7e94ee73c5 : SUCCESS in 41m 47s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/36e2e29291b3462a87a2d55620ebc84d : SUCCESS in 1h 11m 39s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/266aa269145745d294ad9e018573e6a2 : SUCCESS in 46m 31s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ccea4440b02b4b61a158338d868290cf : SUCCESS in 1h 07m 48s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2ead1a409cbd4fada8c187aff9a1e066 : SUCCESS in 1h 01m 01s","accounts_in_message":[],"_revision_number":2},{"id":"689eee5447191e0d79134b7c65e16b59cddf9d48","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 18:15:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e841914149f3e7bd5fb0b7cf9e59af642774ff30","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 18:16:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f664ff4722c4372bd38742c5a49b0e9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04c7b542e8744d1b9822fe1eaaae1a23 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"10af3c4d8abe4fdf7951dfd450f78caf93709ca6","revisions":{"5239b2563bc4c005e2d160aed06822580d95639d":{"kind":"REWORK","_number":1,"created":"2026-03-17 14:46:48.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/62/980962/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/980962/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/980962/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/980962/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/980962/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/980962/1"}}},"commit":{"parents":[{"commit":"587d702cee3049e2ae192af4693f1f8854915eff","subject":"bgp: Fix pep8 errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/587d702cee3049e2ae192af4693f1f8854915eff"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-03-17 14:32:23.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-03-17 14:34:23.000000000","tz":0},"subject":"bgp: Move lrps_to_chassis_routers under commands.py","message":"bgp: Move lrps_to_chassis_routers under commands.py\n\nThe lrps_to_chassis_routers() function iterates over LRPs\nthat coudl change outside of a transaction. Since the helpers module\ndoesn\u0027t guard against that, it\u0027s not a generic function and should\nbe only used in the commands module.\n\nIt also squashes the _get_lrps_by_external_id() function into\n_lrps_to_chassis_routers() because it\u0027s no longer used elsewhere.\n\nRelated-Bug: #2111276\nChange-Id: Ie55523381cf6a833cc13348c7272806e5772dcb7\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5239b2563bc4c005e2d160aed06822580d95639d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5239b2563bc4c005e2d160aed06822580d95639d"}]},"branch":"refs/heads/master"},"10af3c4d8abe4fdf7951dfd450f78caf93709ca6":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-03-17 16:42:49.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/62/980962/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/980962/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/980962/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/980962/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/980962/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/980962/2"}}},"commit":{"parents":[{"commit":"d4b7fd02a99e83b6eeccb3c95ac13b3915a4bd50","subject":"bgp: Fix pep8 errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d4b7fd02a99e83b6eeccb3c95ac13b3915a4bd50"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-03-17 14:32:23.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-03-17 16:42:30.000000000","tz":0},"subject":"bgp: Move lrps_to_chassis_routers under commands.py","message":"bgp: Move lrps_to_chassis_routers under commands.py\n\nThe lrps_to_chassis_routers() function iterates over LRPs\nthat coudl change outside of a transaction. Since the helpers module\ndoesn\u0027t guard against that, it\u0027s not a generic function and should\nbe only used in the commands module.\n\nIt also squashes the _get_lrps_by_external_id() function into\n_lrps_to_chassis_routers() because it\u0027s no longer used elsewhere.\n\nRelated-Bug: #2111276\nChange-Id: Ie55523381cf6a833cc13348c7272806e5772dcb7\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/10af3c4d8abe4fdf7951dfd450f78caf93709ca6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/10af3c4d8abe4fdf7951dfd450f78caf93709ca6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
