)]}'
{"id":"openstack%2Fneutron~981540","triplet_id":"openstack%2Fneutron~master~I4a994fa599bec6ea8c830347aca365d8f8e6dfdf","project":"openstack/neutron","branch":"master","topic":"bug/2144312","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-04-11 09:50:21.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-04-09 13:53:11.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I4a994fa599bec6ea8c830347aca365d8f8e6dfdf","subject":"[OVN] Update the parent hostname in the virtual ports","status":"MERGED","created":"2026-03-20 16:41:28.000000000","updated":"2026-04-11 13:39:13.000000000","submitted":"2026-04-11 13:37:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":23,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"981540-bug/2144312","meta_rev_id":"c30e50b8def3c3642e79942501eeb3cb810cd78a","_number":981540,"virtual_id_number":981540,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-11 13:37:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-04-09 14:24:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2026-04-09 13:46:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":2,"date":"2026-04-09 13:53:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-04-09 14:24:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-04-07 09:39:28.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-20 18:00:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-20 18:46:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-20 20:15:37.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-03-24 15:16:44.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2026-03-24 15:16:44.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2026-04-01 16:19:28.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-04-09 08:36:31.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2026-04-09 13:46:30.000000000","updated_by":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"reviewer":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"state":"REVIEWER"},{"updated":"2026-04-09 14:24:54.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"f723887dfa11d6ffd0ec518335e7d665dfb0822c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-20 16:41:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c58dd3d8cb2e02dbc64878c0b0ee7d4663a6c7db","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-20 16:41:35.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"2b192a013ab729572822d11d4c923a41584502f4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-20 16:41:38.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"4edb5ec35c5cc0a0a9543f4e296e89e302f89240","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 18:00:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6205b47c496b445382a454233eff0985\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4ada41850498431d860361216f6b3ab5 : SUCCESS in 1h 04m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/46c10269e91b48f999c2ffc16dca00c1 : SUCCESS in 42m 31s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/6a891fa4002e490dbd8de1bd166a7d97 : SUCCESS in 1h 02m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bd9a29ebb600dc814317cb198ce66864e0bffd35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 18:46:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/98ae19a8bb664609968fb99cd58e29b2\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/56e1908b0c5544b2a04f89de7d218cfd : SUCCESS in 1h 13m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/681152323c1e4feea3dc5c75a4e511ae : SUCCESS in 57m 08s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b42ce6f6373d42388b1e6b3e86f0de98 : SUCCESS in 1h 43m 43s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/40cd5a3b291d4b6faca14eb9faf7fcca : SUCCESS in 1h 53m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b54de554c0314ac7afacbf47ae1c78cb : SUCCESS in 36m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1dfdcbeca0734fd1be8c188db4e7c406 : SUCCESS in 1h 00m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb11e771cfc44ba280e79d99a91977b8 : SUCCESS in 25m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/87d41996659c40c3910d39bbd0cff642 : SUCCESS in 35m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/10492a1ad1cc4eb99cfb45f176de6f09 : SUCCESS in 18m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8b34c65fa4d743298f8bcc9ce8ab5f4a : SUCCESS in 24m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b45abbf7d4e41318d87d8054f1ecd05 : SUCCESS in 12m 05s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f33f1cffb85d4d6e85060ad94fe6b9d7 : SUCCESS in 32m 34s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2d27651487fa430f98c9881131b12266 : SUCCESS in 1h 02m 06s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/46cf482ca41b450799f1c18e5fcd6f82 : FAILURE in 1h 26m 46s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/793ddb6063f34b8296ddbb253ce4c84f : SUCCESS in 1h 11m 55s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/47b16282a63a4bdfa8bf1596654bca16 : SUCCESS in 36m 50s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d6d46206f0b941559673b6c06f1cc69d : SUCCESS in 1h 58m 11s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/173be355a7df4c858e1f71a2790069b3 : SUCCESS in 1h 50m 48s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f5c4756e237b4b9fb76128f0b9b98e90 : SUCCESS in 1h 14m 39s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/881dd81b28954e028b0e8083925c9fdf : SUCCESS in 36m 06s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ce7136a3fc3249489f02195c0736eb74 : SUCCESS in 1h 23m 29s","accounts_in_message":[],"_revision_number":1},{"id":"56424d7e5c739caac10bfbbebc5b5dc331b26a4c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-23 13:53:05.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ab53bdbbab40d956a6161b37b46a5bfcd855a24e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 15:47:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/429ffcd6ccb04eada40c5d2abc3c39f5\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/db3906c9213f455ca1c8effdd3c1394a : SUCCESS in 1h 06m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a2a73c8f42214081a5a8ad4ef37ece4f : SUCCESS in 1h 00m 48s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3cb61dd53e43435f85a8555fa6ce3f51 : SUCCESS in 1h 19m 11s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8e747705355642fa9d9e3aa63493b711 : SUCCESS in 1h 53m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/275415d0845d4d87bc2665dc906e4fc7 : SUCCESS in 39m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eef1674d04e341c0aa042bb643f3cec1 : SUCCESS in 39m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a45dc9895d4403fa03f332e6f6c5f07 : FAILURE in 7m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8355e63cc7e84b64be1bd449117055dc : SUCCESS in 37m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7522ec2e8d55486698355ee10d6ccf10 : SUCCESS in 23m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/04b1849c5c9e4b6891ee383cbba9c0a2 : SUCCESS in 33m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/821dee99d1d041819b33cb98c03d9ece : SUCCESS in 14m 37s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2637d9f27de5410ab54fb1ba6e9d3461 : SUCCESS in 1h 03m 22s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f47e207439bf4efd865e4a7a6e8c6b29 : SUCCESS in 41m 29s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/479b538c80d644d69d4e7742e0ac8416 : SUCCESS in 57m 54s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/72c02e8574044f69989818ad6a7cfb93 : SUCCESS in 1h 00m 37s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/0bb8d153bf6e47789bc18d0c84857dbb : SUCCESS in 1h 04m 14s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f2597e5d3a524ddd9d9f42e514f2c5f2 : SUCCESS in 1h 03m 24s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fbce919add824071a1e3c799e6d5cf5b : SUCCESS in 1h 04m 59s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ba63467ad801418eb73a8e2b8638864a : SUCCESS in 1h 28m 21s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/66e65f96899f440598f7a8b40b4f0665 : SUCCESS in 1h 24m 07s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/967ffe3faf1e42dbad29e8186cf74c3c : SUCCESS in 1h 21m 18s","accounts_in_message":[],"_revision_number":2},{"id":"e3e329ffb9cff69eed016663f52f7e31c25d9746","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 18:11:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b20ab187c3e94c3c95b94267e6fb618f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8e4c59219cc745ec95cf1ef744bbdeca : SUCCESS in 25m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cdb2f33c59d247fe99856c968e54e386 : SUCCESS in 35m 16s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/81c3aabc45d34148980255ffce4b620f : TIMED_OUT in 1h 21m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2658d5f5511dcea0244fb2af978e1ae0eb0c52e6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-24 11:14:17.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"72566b0423c989d7d86ead6193019d2300374baa","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-24 11:15:17.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"a33a0cfe9a7843e4dd283d4eac8dd58818fc4bf6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 11:58:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a54288f49b04bafbc5a5f33807c88ba\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/247d8d59b61f4943a5b3974b305545c0 : SUCCESS in 30m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f3f0f32d41414c499066cc065083e08f : SUCCESS in 37m 36s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/98b20bf2a28149a4b0b7645a4236f0c9 : SUCCESS in 41m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3def4e9934d423287fb37d111fd089db13fc6c69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 13:07:37.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8ecf69bd15974c91a2b3b2968bc61f49\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/755a7ee3742442ec9ebcc6d8238d330d : FAILURE in 1h 10m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3b55e2060fa2458da05d57d0fea403b9 : SUCCESS in 1h 02m 17s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5a7df1b55b3541398c0087bbcfd91228 : SUCCESS in 1h 27m 48s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a2ece838cd2641a09a56672fa7f4fd9d : SUCCESS in 1h 51m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ed7ed7801d954cd99370afb52d86dbd9 : SUCCESS in 35m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b24e46239db74b1cbaf8abb3e441ff7f : SUCCESS in 42m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed10639343e147939069e7a4005da442 : SUCCESS in 26m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5070d93ec9a746839d76545c81b9513c : SUCCESS in 34m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/863abe18557e40dbb65eebe2c6e5b4c2 : SUCCESS in 25m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/be7c94f8e33c42aab6a4eae5b6f22110 : SUCCESS in 34m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93d552ab491a410d80124f5c6afde12d : SUCCESS in 8m 50s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/9d49b8e22a374b10b47728c47b3644a6 : SUCCESS in 32m 39s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/9d9caffe15604c0fa43ad0cf5633225e : SUCCESS in 1h 21m 01s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9c75d9a112e44c28830189321220fd0b : FAILURE in 50m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d81796dc3ca34f1eaf4605e0016f5d5e : SUCCESS in 1h 11m 20s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/cd2b3a39a51e49b9b8863a0931cd72de : SUCCESS in 45m 14s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/69fd9b1a6d71416db406b6d853ac3e1a : POST_FAILURE in 1h 03m 47s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5300a55601c84ae19a0ad3e4b6918552 : SUCCESS in 1h 38m 13s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5523d0350284424ebc5b3e0ef1a7f69e : SUCCESS in 1h 12m 47s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/28146f3a964340dab4decc96263a1551 : SUCCESS in 52m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/840d78c882e24f48b58ee1638cce7a3a : SUCCESS in 1h 26m 30s","accounts_in_message":[],"_revision_number":3},{"id":"84aad48dc4b37361a393c605c5a2805d6feaf0f6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-24 15:02:17.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c96bb97168480eafd512c6594acd80f2a2032307","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-24 15:15:34.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"7050f0fd4fce50e59e90437108fb6787af2cbf96","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-24 15:15:51.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c43d8cc28ebb06b8d51c2fca8d41dedcb830c08e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-24 15:16:33.000000000","message":"Patch Set 4: Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"a1c6415e558bef8b5ffe5ecf1c544ce996dc6320","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 16:35:30.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3057406924f547da8e478c7dddd2a5d7\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b9d73d6968d34c14a5bf8fd82cdc2e47 : SUCCESS in 25m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/610a64bcb40c4fc4a0cb4dc567267fd4 : SUCCESS in 32m 22s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/88dc1f6eeb0c4b0bb8295a45fd22586d : SUCCESS in 36m 16s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2e3ec0d661c4f9c0671d97a541c0b11a4202e24b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 17:19:27.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ec795ac5bb844ac18dd69c4d6166bf1f\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/aff13f76573e456d93695faccc2f031f : SUCCESS in 1h 02m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/718213bef0794ef484e2f2a7b19bd395 : SUCCESS in 1h 07m 16s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5196789ee79b490d84fc9266b9d403ad : SUCCESS in 1h 20m 37s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2402c33352444341869b11047fa66ba9 : SUCCESS in 1h 44m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b7b65cb8f9f84b3f89701cfd636e8c8c : SUCCESS in 34m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93903bd1d55941c8b1a59c4f68e301b0 : SUCCESS in 38m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4069de64f9504dcf8988c8e5bf66487b : FAILURE in 4m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5b809abbf0b4449a9c598ab9a41054e4 : SUCCESS in 33m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/730d799b76f74ba8a197c773b98471fb : SUCCESS in 17m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2d3cd1212a3b4ad784539f4b3ec34bf7 : SUCCESS in 25m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d0a38403b464820882cfc636bf5c854 : SUCCESS in 7m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/16b1ea6009da478783bb535cbbd8b6af : SUCCESS in 11m 29s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ce70a3766402473a8144839d79b5d937 : SUCCESS in 1h 07m 21s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/281d33a44cb74ac088c668faf0ae9dff : SUCCESS in 1h 19m 33s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d0eb11f0717e4505874e941bb47892ed : SUCCESS in 1h 06m 39s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e936619e870e4ed8817bfd27be996fd2 : SUCCESS in 1h 09m 18s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/01b4d7ee34fd4636949d3c6ab6f61d42 : SUCCESS in 1h 22m 56s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9d9c1ba9687e476d9287337ad065e417 : SUCCESS in 2h 01m 33s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6897c1bfb6a04fe38038a8414a4d1e3f : SUCCESS in 1h 45m 41s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ad0c9d0ac1ed449bb511374e97617e99 : SUCCESS in 1h 09m 00s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/52d753031a7f411d9844a0481bdf366d : SUCCESS in 1h 12m 41s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/67f3d8c0fed14b9e8e456d31a01b87cd : SUCCESS in 43m 17s","accounts_in_message":[],"_revision_number":4},{"id":"b72f346cacbc72d3b181c88f7c32db67fb1f0d49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-25 10:11:18.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"4acfa512d4977918a81db88519c0d1c6d4b93bca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-25 11:21:34.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"a8565d09c8fba47a4bdfce2090738d4e007e6894","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 12:22:44.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27f307b6531646a784d83b98a1c607d1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/84bd3556339146fab51c49320f51a853 : SUCCESS in 26m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fc761ae4c7b94de1a3b0038e5c683740 : SUCCESS in 36m 57s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4b976f77f3194c0ebda6fe0a0732fb26 : SUCCESS in 30m 45s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c46863451dba3ae2665a12f53a9d55322edccb6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 13:39:32.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db231af48c5b49da8facc36694a869aa\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2e2b35c61d5347c4aa565000b4fd5338 : SUCCESS in 1h 32m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7fc5ada8e4784068a1e63b7887eff26a : SUCCESS in 33m 49s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3ddb5ecfa4004448a4da68230352806d : SUCCESS in 1h 30m 26s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a469db0d04984b2189229967b9d99f46 : SUCCESS in 2h 14m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6c9cc4ececba45f89f51e230519d5e21 : SUCCESS in 36m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c5b43e6fbdb45459850c8a0f62f6a60 : SUCCESS in 38m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f806de604524d9583833ed594de147b : SUCCESS in 11m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/320960d8864f40fd9f281fbad6e9bff1 : SUCCESS in 32m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1a2eb945197e4e5187d9b0b44faa8ad7 : SUCCESS in 24m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/47c6a1ec1edc4a9c95961411e34de3d8 : SUCCESS in 29m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de2ced6a8fbc47a7bce09df56454ad23 : SUCCESS in 12m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa56ba6f44b444ce922b5c9e68bc2b7e : SUCCESS in 18m 08s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ffe9df6fb5bb4289911284b2fa7ff083 : SUCCESS in 1h 08m 51s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c1c340b392d441b5b41481cb664549ba : SUCCESS in 40m 12s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4e0ab7cd3588497497f9443d7bc88eb0 : SUCCESS in 1h 09m 22s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/961fac9c9310461c97012c180473d778 : SUCCESS in 1h 11m 23s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/86a4224954504e2db089d6a9604a0d10 : SUCCESS in 40m 21s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8b27b658808644cfa2ffa98cf033b1b2 : SUCCESS in 1h 11m 31s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d974a427247b43ff905e8847fd911f06 : SUCCESS in 1h 50m 17s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e33c76f0095f4edb97846ba28a62a196 : SUCCESS in 1h 04m 10s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c75b650a94e8496189095b195e4e3cf8 : SUCCESS in 36m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ebf93e98b23b4843a3ce9f240f4c99e9 : SUCCESS in 1h 36m 14s","accounts_in_message":[],"_revision_number":6},{"id":"8d9f7cbc028132b936ddd87cf721e083201806c8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-31 14:56:26.000000000","message":"Patch Set 6: Review-Priority+1","accounts_in_message":[],"_revision_number":6},{"id":"5697d743096740aeb40a1b26394d8022b1760414","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-01 16:19:28.000000000","message":"Patch Set 6: Code-Review+2\n\n(5 comments)","accounts_in_message":[],"_revision_number":6},{"id":"f297812d8f813cdd20ab87f30d5b4de911c3c5a7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-07 09:16:56.000000000","message":"Patch Set 6:\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"55969cd6c73aff899379d56d883a34755e580ce0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-07 09:17:14.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"efd6f03c602dd12f4db1d747a240a6131d4a467e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-07 09:39:28.000000000","message":"Patch Set 7: Review-Priority+1","accounts_in_message":[],"_revision_number":7},{"id":"09391b672a65d0524e9f3219e2bfaaa0fd5665fe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 09:50:50.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0cca8845280431798b8e550dafaf739\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6fe62ac882c348cda359285d05e02433 : SUCCESS in 26m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/48ebc7f20eec45f4a3007cc39a1120be : SUCCESS in 31m 45s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8bd1e2f91f294d00b820eaa15ba1a4c9 : SUCCESS in 26m 14s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"cf25cefed00d3f8481c8120a1acd63c6a4c0a2dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 11:14:54.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fc6ba2bcebfe4419bc0042376dc7da42\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ca66fa7b99ee4815bdd98191a35d487d : SUCCESS in 1h 05m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b096ce3e28e14838b1f8779715212ff4 : SUCCESS in 51m 41s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/92a9d4b70668414ca2cb254d96d40ce1 : SUCCESS in 1h 23m 50s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ae6b3bfabbb94baaaf0d4770619797f5 : SUCCESS in 1h 56m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/245077d51a4f451393175d2e43cafe48 : SUCCESS in 39m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa8602d1b21e458babfe1b714e39febd : SUCCESS in 1h 11m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f71fe5a7f9745efb9d4da37fb4c7c85 : SUCCESS in 25m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6d128bbd6d194300992324926b4c952a : SUCCESS in 42m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/765232e12e9c4c86929370746a9bea24 : SUCCESS in 17m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/370409cfc9574579b4508c5f58662687 : SUCCESS in 33m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7429de9fcfc041e1add4c451ea1ec538 : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/127c79e6286742f799a6d1690b9c7c01 : SUCCESS in 22m 47s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/af47decd2bc347fc9511c9700194e3c3 : SUCCESS in 1h 26m 22s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7513fc8db9d54d069a96e55ca0bcea16 : SUCCESS in 1h 08m 43s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/70578cdc1ae841a4a8b9dce0f56c9b25 : SUCCESS in 1h 20m 08s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0e332c9ca5d24671b894bea198633747 : FAILURE in 1h 30m 49s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ccb8907a15d64119a54414a43f6a3329 : SUCCESS in 1h 09m 46s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/31c55ae94b154a7396ef3740eb6e67d3 : SUCCESS in 1h 09m 08s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/70b9da0610d743a1a6163b978b890a1b : SUCCESS in 1h 23m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d0495e1edb44491cbbae8dde3bf6ef39 : SUCCESS in 1h 30m 31s","accounts_in_message":[],"_revision_number":7},{"id":"e18c0080f7e0d53d77b12a2fa4b3ad842c168f01","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-07 13:10:53.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"084ec9e54790a9a8cb91b4b18a61abf182eb5eab","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-07 13:33:27.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"025d3a15fc213a0fc9a8e14814327f8f07f79568","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 14:31:09.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84f9bde379684499b5f73ef5ea4bb093\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b12d00dcc61a47de82044c03843f3a94 : SUCCESS in 26m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ce1c31fc27ee437690705b637669cc77 : SUCCESS in 20m 32s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2130d4b5864b48e0bd8c60526e93cb74 : SUCCESS in 37m 47s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e9ee3d4b519e13e96fd9936593212700b693fbc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 15:33:48.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7894733e54dd435499baaa8e1dfe3597\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0c7a9701944c402181883a34d54d0b23 : SUCCESS in 1h 43m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5f0e786fe5f04f7b82ac88000ddb8f02 : SUCCESS in 1h 06m 42s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1a6c70ca77bf4132858a98ef06f6fdd0 : SUCCESS in 1h 30m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4d5e814ebd314940907c200bd223383d : SUCCESS in 2h 21m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/87da3238a34643688937ea00f66da33a : SUCCESS in 34m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf10e3821a264cf0b2a0b2bc07038908 : SUCCESS in 1h 06m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0e36e71dac94cff82283a9e3fed8ce2 : FAILURE in 4m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/35abf8590a4746acb0df0269114b9cc4 : SUCCESS in 38m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0cea800588ac48038132e7d6301472c9 : SUCCESS in 26m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7c285e8eb0c6475b818c2fbf83f634c7 : SUCCESS in 35m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f5e0fe6d409434397cdd54f9154b5c6 : SUCCESS in 12m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b282f297dafe43c7839d56687b3acc33 : SUCCESS in 22m 28s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6ed4d6e74c744fc892f3e4ecaf0137c7 : FAILURE in 1h 26m 22s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/532d1c4f2f664abfb8d384e605be8160 : SUCCESS in 1h 14m 12s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/b65e727b943040eeb830e72d7f474eca : FAILURE in 42m 28s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/addc994f99ce427d8ec2d2f9451462db : SUCCESS in 1h 59m 13s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/479bae2f4e7243f683a60513f575e2a7 : FAILURE in 28m 27s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9879a074f1814f05b9a47d4918c82733 : SUCCESS in 1h 13m 30s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ae44a801cf1f49ff988c8492fb3fd721 : SUCCESS in 1h 31m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/97dc9be80ac842e8ae381223f6054ec8 : SUCCESS in 1h 34m 00s","accounts_in_message":[],"_revision_number":7},{"id":"128c757cdd10557fcc3e8828423cc939b4f8dc27","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-07 17:04:36.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"52c53561217f8cb91ae4541ce5ce68f2719743a7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 17:55:41.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f147a78bb14e4d75b437eebb0c0f2dee\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/073b37b95638496ebca279a2ceb12a6f : SUCCESS in 27m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bdbc3dc5d49b4f4db703c0913841905e : SUCCESS in 29m 18s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a0552535c9604fbcb2021976c66d70e9 : SUCCESS in 29m 09s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"0859a600b33d1fccbdfaf21d39c81bd68d2b7d7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 19:57:38.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4ea370c91349497c9fb4bd4fde881415\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1b77a38798e942d29ee8d2518bc6e388 : SUCCESS in 1h 27m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a01408e39b21481fb1b5a5f2c676e05a : SUCCESS in 1h 09m 45s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8186f4c3f5204df4a93e5b57f916d166 : TIMED_OUT in 2h 50m 02s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d7a9090bc85f488591f22ff9584ebb50 : SUCCESS in 1h 47m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7636bdbce398431ca55b700c70e49b98 : SUCCESS in 35m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fee884e1e5b7487cbade6ea0dd73c7d7 : SUCCESS in 40m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/469fd3b6c0624aafad79d4c67291e49f : SUCCESS in 26m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d3c963886c0243bfa41ab412c5d83259 : SUCCESS in 39m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/163543d6da7245c6a9afa5c347909289 : SUCCESS in 20m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9675272db48642aeb884a74a08b40b72 : SUCCESS in 33m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bd6f82e58c248dc9d126050e52c40d4 : SUCCESS in 7m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/24ffc785bfd04e17bb637ae888529f7a : SUCCESS in 14m 23s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a9b07a3b28284ab7ba602601382de6b4 : SUCCESS in 1h 52m 12s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/036de8fb57524d17aaddce5632ac7ae5 : SUCCESS in 1h 10m 31s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/b2223a411f7746ea8989a31e65e77c5d : SUCCESS in 1h 18m 05s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/68692fb0589a4eff9932b7aa66513863 : SUCCESS in 1h 11m 39s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/66f04d50263648f7910fc0b85f76486d : SUCCESS in 1h 39m 19s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1b344cb6442f48a79021e74d55cd7fe1 : SUCCESS in 1h 20m 16s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/dce92f77d03b473296acbcda6263316c : SUCCESS in 34m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/711bd7386a2c4245be2e30a4ea260e25 : SUCCESS in 1h 31m 21s","accounts_in_message":[],"_revision_number":7},{"id":"65d86a830e2e0b8b8918097aa1bbcefca716cc10","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-08 05:49:54.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ecc3a5c9c8f4cdee227e9107d78543480c3a241d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 06:17:48.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91bd18839ad04cefbbcd86a7717bcdd1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/32bbe977d2ba420c8bbf3f70819cb196 : SUCCESS in 26m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d710e481ef114bcaaa33e2cbc62751a8 : SUCCESS in 22m 22s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/acba5e1bd5434ef4b4b17b9c1c477e1d : SUCCESS in 26m 11s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"1d980677dd65a24abeb951807e1263535864cd7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 07:40:48.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2215252bdb046af93352c6f41dd71a0\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/77814ff8edba431d95ce777af4a1c71b : SUCCESS in 1h 34m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6d7c366851d847389004dfa321ca3b6e : SUCCESS in 45m 55s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6fcaee0d337a476bbfe0f101c5e907a3 : SUCCESS in 1h 33m 20s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/41627cc5e2cf4978a998d2f20fb5009a : SUCCESS in 1h 48m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cb2bc5e5f75d4d80a002bbeebc268a3c : SUCCESS in 34m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec881da1acd548b58571a7c7359aaec9 : SUCCESS in 37m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45175b2c66a14b2ebd069c634307f598 : SUCCESS in 10m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bce95d220266492aa2cb532ea90a199c : SUCCESS in 22m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c8204495b8264c39bb5961c100b2f9ce : SUCCESS in 25m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/46e0c7b8260d4d0a91ceaf359fd885e6 : SUCCESS in 35m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2903567a88fb424291c5945d3090337e : SUCCESS in 10m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a745eb9e57354681a342ae7c1ef0f899 : SUCCESS in 20m 51s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/35459d6275884022adff92a96edeedca : SUCCESS in 1h 20m 23s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/adc148d0313545e79cc00d3c3575a357 : SUCCESS in 1h 18m 45s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/13771d5be11d4798ab72bcfa42138e96 : SUCCESS in 1h 02m 36s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c7b91d99427340c1b41d444360c8ac25 : SUCCESS in 1h 27m 59s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8881f688b77e469f9d3269758bc5bb95 : SUCCESS in 1h 18m 41s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6e5be959b2144d7896ef2654937f3aed : SUCCESS in 50m 57s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e4e82c7e9faf4a278254447ceb66adcf : SUCCESS in 37m 35s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a9392ce1ca4e464e8263692dad11d3e2 : SUCCESS in 1h 24m 01s","accounts_in_message":[],"_revision_number":7},{"id":"8f96a03e8180e7140a5205a5f2cdf5fa307a68df","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2026-04-09 13:46:30.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"2b21e78aabdce49213e057e0f64df036c0f923c1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-09 13:53:11.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"d4a206ade7b75e66711528fcd3d443fcd01f8167","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-04-09 14:24:54.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"af9f20bc7263935502907b5294514c6f4b0588f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 14:25:20.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"7dfc54ec4a5e2231e957bb57e6773e7a95cfd271","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 16:39:48.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b6c501eff7c847999a5650d709ffb296\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a9815886a6d04544b2891131679d38ce : SUCCESS in 1h 05m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5a54aa002bed47ee9a649bb9b49cea8b : SUCCESS in 42m 13s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a1bf3a1bd4404d74813435d62fff4ff7 : SUCCESS in 1h 45m 12s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/919fee660126489786a242f8d131888e : SUCCESS in 2h 13m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8901f4f5efbd44aeb17d957b45f69fdc : FAILURE in 4m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1252326d6b854d0190e69504e1e2f3dc : SUCCESS in 31m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/da689963b15b4d3ab89577d8a8f2e153 : SUCCESS in 19m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f033ca31774f4a01985ba87e45498ae6 : SUCCESS in 12m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4aad30766ca240b2acff6ae7c732f7e7 : SUCCESS in 25m 22s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1d5642b99dd84e079ed8ab1170e09907 : SUCCESS in 56m 00s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5654c3ebb3ca482dbed12cbafa6ef402 : SUCCESS in 2h 00m 33s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/79ae22a0a0da463597b25513b9572276 : SUCCESS in 1h 16m 59s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d9bad83fc9584547ac1c2009ea3e978c : SUCCESS in 41m 45s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/850b0cd927d246ea8ee51096aaa2c4ac : SUCCESS in 50m 08s","accounts_in_message":[],"_revision_number":7},{"id":"67a0dc37c996b8e5be49f1222a6ba6939b872adc","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-09 16:52:30.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ce38788465266ef856f474a0c85b199b3ba6feff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 19:06:03.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3fcce0e211324877abf5ecece624314b\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9a3c33d07db84dcca9f4448438dafa36 : SUCCESS in 1h 37m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/38280250c47248b3a52246fce845a026 : SUCCESS in 1h 08m 59s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f7875155f5354dc3903bdbf4dadc2856 : SUCCESS in 1h 25m 32s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f822d4cc951e45229e5cf35206771f10 : SUCCESS in 2h 10m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/129d713c38b04090a6fd9564af1927fc : SUCCESS in 31m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3afb4c4b71ea422c848d7a2b1e9c43a7 : SUCCESS in 51m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f4e76b0de6343adae43073db3e671b1 : SUCCESS in 26m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cade96a516604b8ca9f07da3b3b01c0c : SUCCESS in 20m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a9623bbb876b4d6284b92a8118e9b18f : SUCCESS in 16m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/61bc5627af46478da26ef8259cba3d1a : SUCCESS in 29m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/848e369d57874f8ba3297a25322c0a31 : SUCCESS in 6m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab1a8e8f2a654422a786d3151be99c41 : SUCCESS in 21m 54s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1678c43a74c0482c8c79ca759aca4f7c : FAILURE in 28m 25s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2158b0d0fe2a4b18a9128dbfdec220f4 : SUCCESS in 55m 46s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/3b6bc79f33f043e1b096e1cfd2df2d8b : SUCCESS in 1h 12m 37s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/547133d77343474f9356cf989327ada4 : SUCCESS in 1h 46m 17s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ae660af8e2954e7dba7035a73ba568b9 : SUCCESS in 57m 37s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/afec24b7891840c8a6c72c5b1977b977 : SUCCESS in 1h 18m 22s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/580152999e784c8eb31041bc9a8cede6 : SUCCESS in 57m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/aded0f2c46844b3e97c8497b41d76e19 : SUCCESS in 1h 40m 59s","accounts_in_message":[],"_revision_number":7},{"id":"c02bff4a3f9c85d9647bfb4fb801169d7fa4ace4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 20:01:03.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5240a3d792dc49d5b3faec8dda796d2a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c3f0047a34904db5b173bbb740604696 : SUCCESS in 26m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c654ec1590da45389ab514739d1deda9 : SUCCESS in 22m 31s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ea1bc8e24d0d41958ba5b324db510ed7 : SUCCESS in 29m 28s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"bad1873a0cca7203ac082104666d4dfb84a408a2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-10 07:27:46.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"b874d276154b9b0c75840ae2a6f6ce33c2fb2744","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 08:16:21.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d7f09a563a84c15bc5b59e4b73e81e3\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/100e7bdfcdf1445ea15e1c004010c3b2 : SUCCESS in 26m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c6123817a9af4b0c9d013297d64dae62 : SUCCESS in 29m 09s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7c809a7b1bd341758516a4cc1d71e568 : SUCCESS in 26m 14s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"3dc9d7349e4ce3d3c7e9d9204b5906d040517c98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 09:38:07.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4410c6944ac243b988fabd821f8e4f48\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/cc3a6405ff264adfbf099b6c78f7ccae : SUCCESS in 1h 24m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8b8e5efafa56442a8b47552cd266b97a : SUCCESS in 1h 02m 34s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d9193d87101749368aa92f5724fe02d6 : SUCCESS in 1h 18m 10s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5784336ff39b41a8b856e8a1e3b47d0a : SUCCESS in 1h 46m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b5f6190fb3bb4ce180dcb1e09b9c3515 : SUCCESS in 32m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b458f729bcb4374b092ac7e18d86517 : SUCCESS in 47m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0b88158f91a4e4e8423a3bc7fb4a205 : SUCCESS in 28m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bc6035df3b614e228b65db883df568b4 : SUCCESS in 27m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/80696e694be94c5bb070a23a21993eb4 : SUCCESS in 24m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/df2c73495aa945ffb88d0ff4b47f06e7 : SUCCESS in 34m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9915e3d43cb2465abe761a531fb9c8d2 : SUCCESS in 11m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd65f2bded91405ebd1d42473fa2d0d6 : SUCCESS in 21m 11s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/394c8c725f4441359e5df46b33a539a0 : SUCCESS in 1h 15m 04s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e643e8e6bdf148b58b00356b59723521 : SUCCESS in 1h 10m 29s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/a86fedce31794a238f9e3b47c6ea3bf5 : SUCCESS in 1h 13m 35s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/77a2b2edf90545c5bf4d6177ef3b93c8 : SUCCESS in 2h 02m 13s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f960096ba4784b4f8642235fec1b9adf : SUCCESS in 1h 48m 16s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1675bb86c1dc4dc1af05cab7a874ac46 : SUCCESS in 1h 05m 40s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/6683259a6e104fc7a59a81963090aef2 : SUCCESS in 1h 25m 27s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cd94d98955b14f5488f705eb1a25daed : SUCCESS in 1h 25m 52s","accounts_in_message":[],"_revision_number":7},{"id":"14fd4f9dd5adef73547d1661a22993e2abc33989","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 09:38:30.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"44780af6046f682613c374acf33050a4c8426220","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 11:50:43.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/34bbb28030ec41218fb9d0d5556f978f\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/162943dbea954840934d4a9426df4699 : FAILURE in 1h 09m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dd9222e0a4104dada96ef87fbe4ae788 : SUCCESS in 40m 58s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/dadcef95595c477daeebd667118eef20 : SUCCESS in 1h 16m 24s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/393367dfb3f04596a473a930f26198bd : SUCCESS in 2h 11m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/382902df80b6481785b8c17311b4ce8c : SUCCESS in 26m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e273dbf151d7402fb79b038b4141eaac : SUCCESS in 35m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c2e3ecb1626541b4913f3eec1a62d30c : SUCCESS in 25m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63693074ffc2480e991c9129467bf82d : SUCCESS in 13m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7afb10b1766a451a8c2f42b5f2a7c385 : SUCCESS in 21m 00s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a6f2ea8992874d81abf9ef76bac82438 : SUCCESS in 1h 17m 30s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8e2fc8a32ecb4ee0a817fb3a5a77de69 : SUCCESS in 2h 03m 00s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/da08b5cba6484abaa3dfabf229e2314f : SUCCESS in 1h 14m 37s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/dc71308d27564d668b2b8e687c7ddc33 : SUCCESS in 1h 15m 55s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b7ec8dde0ee6411db012ecc1886d0735 : SUCCESS in 1h 11m 43s","accounts_in_message":[],"_revision_number":7},{"id":"c003bd3dee9593f92c52b5bdab96f0cd2caedecf","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-10 13:18:38.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"dfee25cce34b6eec8449e47fe8cee5bcb77ff071","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 15:03:27.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9197f935a44c4c678019a81717f5d4b3\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8f2f6e77b9254993b9635a3855f3150f : SUCCESS in 26m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4ff576669c4f4ca58dfa730749d484a5 : SUCCESS in 26m 24s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a866ee367224493a83b0e50dc9c55648 : SUCCESS in 24m 58s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a24cc155c836c36f44587924e89fd658b3cb8b30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 15:31:15.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5d123369f7bc4203b8ccb2a5861f7830\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1c2d274ccea243b88e70958240006360 : SUCCESS in 1h 32m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/acc876da26e24ecb9172b37808d65999 : SUCCESS in 1h 01m 42s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f5b0998fda8045c5abfded990684466b : SUCCESS in 1h 53m 58s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/017086134e8f472685ed20b0c984b1bd : SUCCESS in 2h 10m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4f53e9400d184f28a102ada72e32ee5c : SUCCESS in 33m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a3afb65d0fc4d20942e793e32b3faa4 : SUCCESS in 1h 12m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfec4b8096c14d389e0000719d1d5cb7 : SUCCESS in 26m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/09656dd0adff47c18d18559c6807969c : SUCCESS in 24m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7262eb7950fb4d5198d48a1f3a29b057 : SUCCESS in 19m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/15e7f2717993489cb7220a41c8997b06 : SUCCESS in 38m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fac1fab9155846acad95d6b15c12da0d : SUCCESS in 10m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9203d305727c4efb93e0920ad1039909 : SUCCESS in 25m 32s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ea7f513af264425c96fa63ef45d7a2f5 : SUCCESS in 1h 12m 55s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f36c37f8fbc4436692d4eb4aaab17563 : SUCCESS in 1h 11m 17s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/6eab94daa07b410a8f42205c0001e0e5 : SUCCESS in 59m 34s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4bc23cd1accd4bd59458ec1d1c87c9c6 : SUCCESS in 2h 09m 18s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8ac92b1e125643479ab3fbac06f08f1d : SUCCESS in 1h 35m 05s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ebef7367f3274b71aa8cb16ad299dfc6 : SUCCESS in 1h 12m 35s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ef157b831cb64840b9abd353be101112 : FAILURE in 9m 06s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/980a94f70a6f456b80531dda0e89ebfc : SUCCESS in 1h 23m 53s","accounts_in_message":[],"_revision_number":7},{"id":"8af8ea963965101ccfbbd4fbeedf892e01dc3e6a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-10 15:32:02.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"9144b442cc7b9f7d2ac6da26b6162b0b19a6b253","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 17:12:57.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86ee7d621bc34f3f87e3845215cd5183\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9c4dfa5e4da2491bba372dd9a184abbe : SUCCESS in 25m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/633e05faa5ff4f149be0fcc2f81a2519 : SUCCESS in 23m 45s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/eabf94dffaca4122879615efe6968b80 : SUCCESS in 28m 43s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a6530825f4045b2a3ff104720746d7a9f1137902","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 17:33:31.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f33ec6ab25ce46a9b673d4216ad61919\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/77f54587993842629e2093dbb51d3a8c : SUCCESS in 1h 16m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/87de1d6fddad4844bdd96bdb997d5130 : SUCCESS in 44m 08s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7491ca668ccf420fa799ec886b171d88 : SUCCESS in 1h 22m 34s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/43d93f453aad464485edbf2a1c5cac49 : SUCCESS in 1h 59m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/06179cd6461f4adf84fb152184127844 : SUCCESS in 34m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3ef951dd9e134315a4140273b8ffb514 : TIMED_OUT in 1h 21m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb8f14590e1a44529e093a22185a96f0 : SUCCESS in 25m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6eb23884819b41cfa631aa9522b1da64 : SUCCESS in 37m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ddf3450dc992454ba3f7768ef7b7f1f2 : SUCCESS in 25m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/49848df10d174aa8b55c013b635f3782 : SUCCESS in 32m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67831a75e49b43d1b4874d19bb4b9233 : SUCCESS in 14m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d0904bfe50a4dbca07019a5f43d4a39 : SUCCESS in 21m 31s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/87d870e9b7344847bab6853709bee330 : FAILURE in 1h 25m 22s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/46efec9038144be388f780f410991f71 : SUCCESS in 1h 01m 57s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/44eeb5e60c3c4d1f9abed8a6f555c762 : SUCCESS in 41m 17s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/cd22c10d6bad4034a5a926e7380f925a : SUCCESS in 1h 49m 55s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/46226dcb316c4f84ad4450a974d41005 : SUCCESS in 57m 59s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a0067f2af29e4225946a5bd4c75ef318 : SUCCESS in 51m 37s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5269d949816f4635a5e108dd72223759 : SUCCESS in 32m 31s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e3bd285057894935833840d943512d91 : SUCCESS in 1h 32m 55s","accounts_in_message":[],"_revision_number":7},{"id":"79e98c17417ca69ad13fd298ab55e9f3ec7c3419","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-10 21:05:18.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"fbc2e0afd9a9ffec05417d338147f8081535176a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 21:53:52.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f7bcc596568497e88245e349bc19631\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4439b26d65b14176b2f584ee4681cba7 : SUCCESS in 22m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e3847b71eb094974a346eb8f21579266 : SUCCESS in 18m 21s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/fbe638a137814e9c9e495305b3d015fd : SUCCESS in 32m 33s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"46d91f085950adb9988dbbe653cd5ec1ea896822","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 23:29:50.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/85e09735ac8247b3a12c2c7edad50933\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b2f0a529b5e3490087c30578d2ee8737 : SUCCESS in 1h 12m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2643e74f73174053853f4693d186d3fa : SUCCESS in 1h 05m 34s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/dc8980924c464f5caa3dd2836ff6b6ec : SUCCESS in 1h 28m 24s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/290465261956498eaa4bd43dbcece018 : SUCCESS in 2h 23m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bcfec23ded42414f8da388fbdbfb4be3 : SUCCESS in 33m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c12a93c6924455fb4cbcf3d113ada64 : TIMED_OUT in 1h 21m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/837fccc33eb64844872d14d7e8165a6e : SUCCESS in 25m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/60323ebb331548dd94423786f0ee7d0a : SUCCESS in 24m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/61c0e91ccdbc41c6a7f6f960ab2c5473 : SUCCESS in 24m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7761a6a416cc498fad83d6de83a92efb : SUCCESS in 25m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb98792aed8142c7ab31105bb42facb2 : SUCCESS in 12m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73c038f7d4cf427e8edc3d7ddd9aafef : SUCCESS in 22m 40s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/deb782ad3faa48b4bfc67a587e59116c : SUCCESS in 1h 36m 01s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4dc74ce3d6bc4fd58f2f0344b37faea2 : SUCCESS in 43m 57s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/2513e269eb1a4f9da8ad13dd7aee5631 : SUCCESS in 1h 19m 11s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2c9f395a672d4a4f902cce45657b3f39 : SUCCESS in 1h 59m 15s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/246ad2de81f84726842eed6385a01be7 : SUCCESS in 2h 12m 43s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4824a56a2cca4db68e81ed427316561a : SUCCESS in 1h 17m 48s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/508da1793383439e83b805a408e98612 : SUCCESS in 1h 18m 53s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9a26ecab664b4a2995b61dd64fdf2101 : SUCCESS in 1h 26m 33s","accounts_in_message":[],"_revision_number":7},{"id":"42fb5480492d329ec479e8673fa7f841a919a8ea","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-11 07:27:55.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"41cfa290ea94134c3e35a8c7581bc3357a3d0c3b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 07:53:56.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e6457d6c1d142d2893d368272c23acb\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/94cf1c351422427d85d62668313eb0b6 : SUCCESS in 24m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2e2eaf0ac3d74e668f5e56fcf3953de6 : SUCCESS in 21m 30s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/6bc11682831245688fb1f35a2363c531 : SUCCESS in 23m 56s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"b2cc4de50b014a4057864f5a2f9d664a83b712b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 09:43:20.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/65c8a8356e9a4259ac334542873ba24d\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/97a024f91dcc467a8c1451b9a6fae4fa : SUCCESS in 1h 15m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/60680a6ab409473898dc86c4a0f34441 : SUCCESS in 37m 42s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c395d34fe1384001bc33b2fdc58d57b3 : SUCCESS in 1h 22m 04s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/01a5a54a50ce4ed08c2d4675e39f9970 : SUCCESS in 2h 13m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fbc8220831454cd380166a73346a0bc9 : SUCCESS in 32m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4395e4c3e0e64cefa0787292496ee188 : SUCCESS in 38m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e0078e59bab4d4e9292763aa0519724 : SUCCESS in 10m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7acb2acedaab4372a2b0c7413baf05fc : SUCCESS in 35m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8b759fd377e64331a15119d4220d4fc7 : SUCCESS in 24m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3d76fa0c9e304b52ab65017ecb5f1cf2 : SUCCESS in 24m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2947bcce2d7949cb93225efe8d1ebf5f : SUCCESS in 10m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb7d691124a64d25ae9012087bd7607c : SUCCESS in 12m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/58889af9b8d1489e83cb9800f1114104 : SUCCESS in 1h 26m 27s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9efed6638cb149eeaa9c5232d8c70862 : SUCCESS in 1h 11m 52s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/f1d8d5d4667a4acb87198906861f2dfa : SUCCESS in 58m 44s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b24f08d444724343bce1c00bf905ad94 : SUCCESS in 1h 51m 33s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f129ced1676548bb8c66d8e5f7af7909 : FAILURE in 1h 13m 27s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/475dd0c61a53412cb9b729db92c66402 : SUCCESS in 49m 31s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a845339cdce545d3ad03e1910c54f7f3 : SUCCESS in 39m 31s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b6b829c426b64d0dab3ff3a47a7f5924 : SUCCESS in 1h 04m 41s","accounts_in_message":[],"_revision_number":7},{"id":"1e724aba622998feabf2769fc7d67a9a37bd8725","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-11 09:50:21.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"62c236bc059d1e019cbd670700f92811062e5ab5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 10:47:38.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72be791955224edbaee0bbdd58f28b3a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1ebb8949a5874a9b8057b2e66a91f769 : SUCCESS in 30m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/edb3ed3e90644db0a886f7643e4b07f0 : SUCCESS in 24m 06s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/827e7e7c7b9548e1a0f9198433d11d61 : SUCCESS in 27m 33s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e57ced46fc5c48965f0976d19f10866830804c6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 11:50:07.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c31f1f81d2454d36b16ad6b4e37b3397\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f2130aab87ca40d3b73083d689d86581 : SUCCESS in 1h 23m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/555e39b5ec3b4d6892f06517f4c650b2 : SUCCESS in 1h 02m 24s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/24c481feb2204414b2c2bc684cbfb4ca : SUCCESS in 1h 38m 44s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/baeebef603034384a89839bd3f71a3f7 : SUCCESS in 1h 54m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f551981836d4441aba500af36f89fd63 : SUCCESS in 31m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a6926c4854c4aceaa9f3ce6a77d8a4c : SUCCESS in 1h 09m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b3c80cb375a4617a7b737e28e49d7d5 : SUCCESS in 19m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/51b82b0cc52144199ea270ecdac0260d : SUCCESS in 38m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2ad9df0fdde34db59c93a01646d35d6b : SUCCESS in 25m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7ae51d6f68ce49729e4bc54f72c1e715 : SUCCESS in 34m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d849ec5404ee4e15830de23ab68bdc73 : SUCCESS in 12m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c273ce21df7b493f82ff6391177ebac6 : SUCCESS in 19m 52s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4871caecd8084f7e890daf6421ecc58f : SUCCESS in 1h 21m 51s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8f88e83e2c4f474ca43f56685fb709b7 : SUCCESS in 1h 08m 20s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/11770d3bccca4fde843f879d53402d63 : SUCCESS in 1h 20m 36s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/98dc09b096c74fcc9d3256d1372b28cd : SUCCESS in 1h 52m 23s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7c827101953541019ff07726d1927beb : SUCCESS in 1h 38m 15s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9d8ce5ecbc3e43238589ac1556f6fb5d : SUCCESS in 1h 05m 31s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/511834c9ee7e4e97b1b16080bdd0161e : SUCCESS in 1h 22m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/44b91f05040a461eabc88bb0057ff2ae : SUCCESS in 1h 37m 56s","accounts_in_message":[],"_revision_number":7},{"id":"0d71f2c846c8318fc8b9e0279327851df2dd94c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 11:50:41.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"3e4b49890a9d145099bf22f0eb6d169441400293","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 13:37:44.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe487155f6ed4a99a83b8e22bf5b9fe6\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d63c331f120043f8a0873611d9ab40c0 : SUCCESS in 1h 10m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0669d946702d4bca9ce49526129038bc : SUCCESS in 38m 05s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/10008ddd52314f258e89e4b1f95da0d4 : SUCCESS in 1h 33m 59s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5d5e4c133d1249b1b841f0aac222bfa4 : SUCCESS in 1h 41m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69c2e0cbda7348a6b94ca48136b3defb : SUCCESS in 29m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c6b23d9aa267498783d22749ed8a7723 : SUCCESS in 35m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/845e8cb9f38a465fbf95c19dc7e6ca14 : SUCCESS in 20m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dddbfb9160384916b483861843c3fcf8 : SUCCESS in 11m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ef6ac1eb0dd436ba1b3d7899fffe7e6 : SUCCESS in 16m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/03707b82778d4d36a615116d309bc53b : SUCCESS in 50m 58s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/45a421bf12ff4afb8b434e4a5799d5f7 : SUCCESS in 1h 35m 54s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/8d0fc782c8f74feb8e9db5834b11c038 : SUCCESS in 1h 23m 12s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/fb4650231f25431a9fdddc72a656c990 : SUCCESS in 43m 23s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/59d825c54bed4df2ab6e954c3414b511 : SUCCESS in 1h 16m 14s","accounts_in_message":[],"_revision_number":7},{"id":"38d3037c9eef7c9f4b62d97b6c0e8431fe70d07c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 13:37:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"c30e50b8def3c3642e79942501eeb3cb810cd78a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 13:39:13.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd4eb635b7074418bd9528495a87761c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a8d153023ec4a848efd1c1f471ddc34 : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/665fd2f058a64cc78794d3a5288982a2 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"c65c18a6d2259332f4f526d1dc4b9799636a1197","revisions":{"23112afcd5290a6d219ea2c9d6c5ebc0641862ff":{"kind":"REWORK","_number":1,"created":"2026-03-20 16:41:28.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/40/981540/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/981540/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/981540/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/981540/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/981540/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/981540/1"}}},"commit":{"parents":[{"commit":"8730dbd1599dc5778e7c92e9f004f770b7a58ffe","subject":"Merge \"bgp: Fix more pep8 errors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8730dbd1599dc5778e7c92e9f004f770b7a58ffe"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-03-20 16:40:49.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-03-20 16:40:49.000000000","tz":60},"subject":"[OVN] Update the parent hostname in the virtual ports","message":"[OVN] Update the parent hostname in the virtual ports\n\nTO BE DEFINED\n\nCloses-Bug: #2144312\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I4a994fa599bec6ea8c830347aca365d8f8e6dfdf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/23112afcd5290a6d219ea2c9d6c5ebc0641862ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/23112afcd5290a6d219ea2c9d6c5ebc0641862ff"}]},"branch":"refs/heads/master"},"223277e08dfbc34e98ecd64b6da7fb01b39acb41":{"kind":"REWORK","_number":2,"created":"2026-03-23 13:53:05.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/40/981540/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/981540/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/981540/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/981540/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/981540/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/981540/2"}}},"commit":{"parents":[{"commit":"5ee55180808fb41f1ce867b1313f8b2c8153a2f2","subject":"Merge \"Use ovn-db-sync-util related constants from neutron-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5ee55180808fb41f1ce867b1313f8b2c8153a2f2"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-03-20 16:40:49.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-03-21 01:21:47.000000000","tz":60},"subject":"[OVN] Update the parent hostname in the virtual ports","message":"[OVN] Update the parent hostname in the virtual ports\n\nTO BE DEFINED\n\nCloses-Bug: #2144312\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I4a994fa599bec6ea8c830347aca365d8f8e6dfdf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/223277e08dfbc34e98ecd64b6da7fb01b39acb41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/223277e08dfbc34e98ecd64b6da7fb01b39acb41"}]},"branch":"refs/heads/master"},"36780cb645e26484887f755f8f798c3eb6b0d759":{"kind":"REWORK","_number":3,"created":"2026-03-24 11:14:17.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/40/981540/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/981540/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/981540/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/981540/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/981540/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/981540/3"}}},"commit":{"parents":[{"commit":"b5f43626c619a088ace6fef8e95de58463e42be8","subject":"Merge \"[OVN] Fix an issue in the L3 OVN scheduler (3)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b5f43626c619a088ace6fef8e95de58463e42be8"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-03-20 16:40:49.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-03-24 11:14:08.000000000","tz":60},"subject":"[OVN] Update the parent hostname in the virtual ports","message":"[OVN] Update the parent hostname in the virtual ports\n\nThe OVN virtual ports should not have a port bindings host defined.\nThis is used by the mechanism driver to bind the port to a host.\nInstead, in order to inform to the user (admin user or able to read\nthe VIF details) where the parent port with the virtual IP is bound\nand transmitting data, the Neutron port will have a VIF details\nkey named \"parent_hostname\" with this information.\n\nThis new key will be also populated in the ``Logical_Switch_Port``\nexternal_ids dictionary, under the key \"neutron:parent_hostname\".\n\nThis patch also implements an OVN maintenance method to migrate the\ncurrent virtual ports information:\n* Removes the port binding host information.\n* Add the VIF details \"parent_hostname\" information.\n* Adds the ``Logical_Switch_Port`` \"neutron:parent_hostname\" key.\n* Removes the ``Logical_Switch_Port`` \"neutron:host_id\" key.\n\nCloses-Bug: #2144312\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I4a994fa599bec6ea8c830347aca365d8f8e6dfdf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/36780cb645e26484887f755f8f798c3eb6b0d759"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/36780cb645e26484887f755f8f798c3eb6b0d759"}]},"branch":"refs/heads/master"},"a025a9614ffb6ba94de0dd139fa7095fea52f286":{"kind":"REWORK","_number":4,"created":"2026-03-24 15:15:34.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/40/981540/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/981540/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/981540/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/981540/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/981540/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/981540/4"}}},"commit":{"parents":[{"commit":"b5f43626c619a088ace6fef8e95de58463e42be8","subject":"Merge \"[OVN] Fix an issue in the L3 OVN scheduler (3)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b5f43626c619a088ace6fef8e95de58463e42be8"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-03-20 16:40:49.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-03-24 15:15:18.000000000","tz":60},"subject":"[OVN] Update the parent hostname in the virtual ports","message":"[OVN] Update the parent hostname in the virtual ports\n\nThe OVN virtual ports should not have a port bindings host defined.\nThis is used by the mechanism driver to bind the port to a host.\nInstead, in order to inform to the user (admin user or able to read\nthe VIF details) where the parent port with the virtual IP is bound\nand transmitting data, the Neutron port will have a VIF details\nkey named \"parent_hostname\" with this information.\n\nThis new key will be also populated in the ``Logical_Switch_Port``\nexternal_ids dictionary, under the key \"neutron:parent_hostname\".\n\nThis patch also implements an OVN maintenance method to migrate the\ncurrent virtual ports information:\n* Removes the port binding host information.\n* Add the VIF details \"parent_hostname\" information.\n* Adds the ``Logical_Switch_Port`` \"neutron:parent_hostname\" key.\n* Removes the ``Logical_Switch_Port`` \"neutron:host_id\" key.\n\nCloses-Bug: #2144312\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I4a994fa599bec6ea8c830347aca365d8f8e6dfdf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a025a9614ffb6ba94de0dd139fa7095fea52f286"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a025a9614ffb6ba94de0dd139fa7095fea52f286"}]},"branch":"refs/heads/master"},"d080bea6371268ff9f0da853af9893d4873b9ccc":{"kind":"REWORK","_number":5,"created":"2026-03-25 10:11:18.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/40/981540/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/981540/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/981540/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/981540/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/981540/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/981540/5"}}},"commit":{"parents":[{"commit":"b5f43626c619a088ace6fef8e95de58463e42be8","subject":"Merge \"[OVN] Fix an issue in the L3 OVN scheduler (3)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b5f43626c619a088ace6fef8e95de58463e42be8"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-03-20 16:40:49.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2026-03-25 10:11:18.000000000","tz":0},"subject":"[OVN] Update the parent hostname in the virtual ports","message":"[OVN] Update the parent hostname in the virtual ports\n\nThe OVN virtual ports should not have a port bindings host defined.\nThis is used by the mechanism driver to bind the port to a host.\nInstead, in order to inform to the user (admin user or able to read\nthe VIF details) where the parent port with the virtual IP is bound\nand transmitting data, the Neutron port will have a VIF details\nkey named \"parent_hostname\" with this information.\n\nThis new key will be also populated in the ``Logical_Switch_Port``\nexternal_ids dictionary, under the key \"neutron:parent_hostname\".\n\nThis patch also implements an OVN maintenance method to migrate the\ncurrent virtual ports information:\n* Removes the port binding host information.\n* Add the VIF details \"parent_hostname\" information.\n* Adds the ``Logical_Switch_Port`` \"neutron:parent_hostname\" key.\n* Removes the ``Logical_Switch_Port`` \"neutron:host_id\" key.\n\nCloses-Bug: #2144312\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I4a994fa599bec6ea8c830347aca365d8f8e6dfdf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d080bea6371268ff9f0da853af9893d4873b9ccc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d080bea6371268ff9f0da853af9893d4873b9ccc"}]},"branch":"refs/heads/master"},"b403616e26d6130a450e0238f0d0e749c4720ec6":{"kind":"REWORK","_number":6,"created":"2026-03-25 11:21:34.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/40/981540/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/981540/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/981540/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/981540/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/981540/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/981540/6"}}},"commit":{"parents":[{"commit":"2370e0934b34d4327f11eb34afbfaad714d2f849","subject":"Merge \"Remove ``update_qos_fip_rule_priority`` OVN maintenance method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2370e0934b34d4327f11eb34afbfaad714d2f849"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-03-20 16:40:49.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-03-25 11:18:03.000000000","tz":60},"subject":"[OVN] Update the parent hostname in the virtual ports","message":"[OVN] Update the parent hostname in the virtual ports\n\nThe OVN virtual ports should not have a port bindings host defined.\nThis is used by the mechanism driver to bind the port to a host.\nInstead, in order to inform to the user (admin user or able to read\nthe VIF details) where the parent port with the virtual IP is bound\nand transmitting data, the Neutron port will have a VIF details\nkey named \"parent_hostname\" with this information.\n\nThis new key will be also populated in the ``Logical_Switch_Port``\nexternal_ids dictionary, under the key \"neutron:parent_hostname\".\n\nThis patch also implements an OVN maintenance method to migrate the\ncurrent virtual ports information:\n* Removes the port binding host information.\n* Add the VIF details \"parent_hostname\" information.\n* Adds the ``Logical_Switch_Port`` \"neutron:parent_hostname\" key.\n* Removes the ``Logical_Switch_Port`` \"neutron:host_id\" key.\n\nCloses-Bug: #2144312\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I4a994fa599bec6ea8c830347aca365d8f8e6dfdf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b403616e26d6130a450e0238f0d0e749c4720ec6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b403616e26d6130a450e0238f0d0e749c4720ec6"}]},"branch":"refs/heads/master"},"c65c18a6d2259332f4f526d1dc4b9799636a1197":{"kind":"REWORK","_number":7,"created":"2026-04-07 09:17:14.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/40/981540/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/981540/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/981540/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/981540/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/981540/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/981540/7"}}},"commit":{"parents":[{"commit":"54befe6c096275cfe0049a2f1ab1abbee3140c8c","subject":"Use unittest.skip decorator","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/54befe6c096275cfe0049a2f1ab1abbee3140c8c"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-03-20 16:40:49.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-04-07 09:17:01.000000000","tz":120},"subject":"[OVN] Update the parent hostname in the virtual ports","message":"[OVN] Update the parent hostname in the virtual ports\n\nThe OVN virtual ports should not have a port bindings host defined.\nThis is used by the mechanism driver to bind the port to a host.\nInstead, in order to inform to the user (admin user or able to read\nthe VIF details) where the parent port with the virtual IP is bound\nand transmitting data, the Neutron port will have a VIF details\nkey named \"parent_hostname\" with this information.\n\nThis new key will be also populated in the ``Logical_Switch_Port``\nexternal_ids dictionary, under the key \"neutron:parent_hostname\".\n\nThis patch also implements an OVN maintenance method to migrate the\ncurrent virtual ports information:\n* Removes the port binding host information.\n* Add the VIF details \"parent_hostname\" information.\n* Adds the ``Logical_Switch_Port`` \"neutron:parent_hostname\" key.\n* Removes the ``Logical_Switch_Port`` \"neutron:host_id\" key.\n\nCloses-Bug: #2144312\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I4a994fa599bec6ea8c830347aca365d8f8e6dfdf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c65c18a6d2259332f4f526d1dc4b9799636a1197"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c65c18a6d2259332f4f526d1dc4b9799636a1197"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
