)]}'
{"id":"openstack%2Fneutron~982005","triplet_id":"openstack%2Fneutron~master~Ib80d0e428751f69d54db21e22a51bfd373875ef9","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-04-01 16:47:29.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"Ib80d0e428751f69d54db21e22a51bfd373875ef9","subject":"Disable skip-level jobs in check queue","status":"MERGED","created":"2026-03-24 20:18:10.000000000","updated":"2026-04-01 21:53:58.000000000","submitted":"2026-04-01 21:50:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"982005","meta_rev_id":"b3231c925c9fd64ca11ed5777a0d5d277bb4da7b","_number":982005,"virtual_id_number":982005,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-01 21:50:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-03-31 13:14:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-03-25 09:35:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-03-31 13:14:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2026-03-31 04:05:49.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-24 22:27:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-25 09:35:42.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-03-31 13:14:28.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"bf78898afd9b272c827ed6799b9dbf314b9e47a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-03-24 20:18:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f0e9ec3af754148f942919db0fcb1a08ec2a3396","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 22:27:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfbac899b9ca4390a70a393ce2e60a7b\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3f806d7d4b184db5ac1326a0f43c30b4 : SUCCESS in 36m 06s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/955533cca57d4395be5ca65142ed728e : SUCCESS in 1h 23m 02s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/27e192855e2e41408699e6b742b9e00a : SUCCESS in 2h 07m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9874f420323f433bba0fd43956548a9a : SUCCESS in 34m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4da34b3448fa49b8a3389feadb515af6 : SUCCESS in 48m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f809da817344a6bb620072ceec54afc : SUCCESS in 12m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3f525d65b7234db4875581ddb5ad2d90 : SUCCESS in 34m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4fa0835f7ec346aaa96a011587ea6ee8 : SUCCESS in 24m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c2490e5076d747c695ffa30894c621ca : SUCCESS in 33m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbae42f66d8b4938b1f3d5094848d9c5 : SUCCESS in 8m 21s","accounts_in_message":[],"_revision_number":1},{"id":"3ca32521f95bf806406cac2eb59122d140742b57","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 22:32:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48bdc345b5534b759a550ec4b18f54f1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d6e9c0e1f8a248bf83acb239badd78f7 : SUCCESS in 25m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8cd8e0db3ba44177995eb7fe797c6888 : SUCCESS in 18m 49s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/aa10038a89524ac4b5d7726f3af3d2f1 : SUCCESS in 24m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9b122e11d34a5c38fb8d2493332addc5f55d6afe","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-03-25 09:35:42.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4ae37eed00938e2d648cd04711ca4d4f72324d7b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-03-31 04:05:49.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"c70a94ec56653327062efa63ea77b6e7059429e6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-31 13:14:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"80bd61d8d6744494ce78991c3bbb93ab2952cc5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 13:15:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6b1c9e095bc7279e5672848d941da671bbf5f17f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 16:07:32.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1dc5a9f2c16545ab945a58abd785eaab\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e3eb3d59a4b64ffba306dddb384cfb28 : SUCCESS in 1h 03m 32s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0b27d744911d4051a307941f609afa05 : TIMED_OUT in 2h 51m 27s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e397648a6bbc49db80e196de922a7c87 : SUCCESS in 2h 18m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/351b87296f5a43efb2e76e60882f8f88 : SUCCESS in 30m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cb9bbdd5fa2446969d5b84ee899d35e5 : SUCCESS in 32m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/15acc5ba682842f98968c137826241c6 : SUCCESS in 20m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1920cf55c1194e56bdce9acd88ccfc97 : SUCCESS in 6m 43s","accounts_in_message":[],"_revision_number":1},{"id":"abd031b72a3f6dafc8312de10fd0d3b1a35a09e8","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-01 13:39:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5e899b60a32a28d6f505f58c2e9bed00ceb6ea77","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 15:04:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/453248f75aae4ea481a10687a4dc95b9\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c108d38d5fa644e6894745b0d0914078 : SUCCESS in 24m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/02a69e83324d481e8bd1a1cd2253b72e : SUCCESS in 22m 09s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d11840cb03ea4086bfa7cb19f75d1443 : SUCCESS in 28m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"38a8980ce26aa95895ce80a7f10c64236d45e033","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 16:15:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d1943d0b2a4341798d26d7c215c5cc62\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fc5328f8b0214aed8f6866bdd6975918 : SUCCESS in 1h 09m 51s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/130a9e3b21a14b41a0957d8c0f560c64 : SUCCESS in 1h 14m 32s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/409917c9b72347dabf29bed222ef8ef0 : SUCCESS in 2h 30m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d9caa3d58f364f8481f8d5734c77ab75 : SUCCESS in 40m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/448bb5e8b6204804b14fedb111d11aa3 : SUCCESS in 1h 05m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9e7f4a20e004b918c56db1aa1f5f3d1 : FAILURE in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e9fccc00a8264ab6a18d52a502befe8e : SUCCESS in 35m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6a6982c9f74a4a1abff1655e3342c384 : SUCCESS in 27m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f3e5871e6cab4d6d9e2db3035cd813a9 : SUCCESS in 30m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cdec18d01744883a96e335ca3d55824 : SUCCESS in 11m 08s","accounts_in_message":[],"_revision_number":1},{"id":"75598715679c35ae674854185d9175805dcfa36a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-01 16:47:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cf904cf2169379c9a2ad99df33f5b1135acc9989","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 17:17:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/894b8e580fe64a7a875c158277ce8d83\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e402498222554b1ba24e47842c4571a7 : SUCCESS in 24m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/56a098796bec4415adb90b524363a718 : SUCCESS in 18m 23s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/43863371dbbb415489033602a7d7b93e : SUCCESS in 26m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d0e69e27aa286cc537a26026084b29736ad3605d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 18:57:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b625839eb1c431fb3ba9e60cae4a149\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/77f51a4144664b10bed5f9c3254b9923 : SUCCESS in 35m 34s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8c7adbb1f19d4ab0bb85109a3d556fba : SUCCESS in 1h 17m 23s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fd0fc8e6a7a64e1eb75a71236055f023 : SUCCESS in 2h 08m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8660941109f84b0fb833f7df00396463 : SUCCESS in 32m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51cde089d0cc45e9b7c377f14734f6ed : SUCCESS in 39m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53f91b85de9c43bf95ce6d33eca10778 : SUCCESS in 21m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/baac79103a824eb39d26d124240bc141 : SUCCESS in 36m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2dad2c8aff2141bb9c9192117179e235 : SUCCESS in 18m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a7c578018154448e8ff7d8208832be4f : SUCCESS in 32m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/324cb064fc964f5da1bb727f077a9e77 : SUCCESS in 12m 30s","accounts_in_message":[],"_revision_number":1},{"id":"5c4957622159c6bded6e02681008985b56299731","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 18:59:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"02070d676258217310ef74cc59e14876592e7eaa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 21:50:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e5b08ab99cf40a4858685c988cd6dd2\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/126c7d52712c45918ac03878a046b5e4 : SUCCESS in 1h 03m 46s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b67a384ef8f449ca94a935c2bd7bbfa8 : SUCCESS in 1h 42m 57s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ae423ff17fc1476fa02b2b921d8c11e5 : SUCCESS in 2h 09m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88ad1bd3f70b4f3d95916369ae3ae987 : SUCCESS in 27m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/abb7d35f5dd34b04a6d4a1ad50b3e6d0 : SUCCESS in 33m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2687059cef6d4f0cb6c28a4dd6546bca : SUCCESS in 24m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93def29246204a7eafe9e111ed4984af : SUCCESS in 13m 57s","accounts_in_message":[],"_revision_number":1},{"id":"6d02c18f0a3bde45bcb323e5dbbb5ab1262e43ad","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 21:50:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b3231c925c9fd64ca11ed5777a0d5d277bb4da7b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 21:53:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd4c156d46e749ef83b33873c823a437\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/314f90128cf547c8bd14810b6bd46c4c : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ab26446c740d63020b3d3dba96fd70c279444833","revisions":{"ab26446c740d63020b3d3dba96fd70c279444833":{"kind":"REWORK","_number":1,"created":"2026-03-24 20:18:10.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/05/982005/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/05/982005/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/05/982005/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/05/982005/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/05/982005/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/05/982005/1"}}},"commit":{"parents":[{"commit":"0fd3ac81c59da30a853ddf0712879489cd0d2ee7","subject":"Merge \"Re-enable flake8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0fd3ac81c59da30a853ddf0712879489cd0d2ee7"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-03-24 20:09:45.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-03-24 20:15:13.000000000","tz":-240},"subject":"Disable skip-level jobs in check queue","message":"Disable skip-level jobs in check queue\n\nDuring this release (non-SLURP), these jobs are now executed in the\n\"periodic\" and \"experimental\" queues, but not the \"check\" queue.\n\nChange-Id: Ib80d0e428751f69d54db21e22a51bfd373875ef9\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ab26446c740d63020b3d3dba96fd70c279444833"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ab26446c740d63020b3d3dba96fd70c279444833"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
