)]}'
{"id":"openstack%2Fneutron~982018","triplet_id":"openstack%2Fneutron~master~I8fcdf78588783f8a2bd749eb6e9bceb1550a13d6","project":"openstack/neutron","branch":"master","topic":"segment-bridges-ovn-agent-extension","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-09-09 21:09:11.000000000","reason":"Change was submitted"},"4694":{"account":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"last_update":"2026-09-09 21:09:11.000000000","reason":"Change was submitted"},"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2026-09-09 21:09:11.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2026-09-09 21:09:11.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-09-09 21:09:11.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-09-04 01:51:51.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I8fcdf78588783f8a2bd749eb6e9bceb1550a13d6","subject":"[OVN] Agent extension for segment bridges","status":"MERGED","created":"2026-03-24 23:29:11.000000000","updated":"2026-09-09 21:12:51.000000000","submitted":"2026-09-09 21:09:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":24,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"982018-segment-bridges-ovn-agent-extension","meta_rev_id":"435365497b464963c889c33a857ce5a8acfe49e3","_number":982018,"virtual_id_number":982018,"owner":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2026-09-09 00:52:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-09 21:09:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2026-09-08 21:00:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2026-09-09 13:15:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2026-09-09 13:15:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-08 23:58:46.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}],"CC":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-25 00:09:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-25 02:25:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-25 09:06:19.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2026-04-14 04:24:38.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-05-18 21:38:40.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-22 08:28:18.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-06-26 08:52:21.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2026-08-26 17:55:15.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-09-04 01:51:51.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"5efa50bbdfa6e21b1ad43d1e0b8e7e413c2f5172","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2026-03-24 23:29:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"776c39c9aaadfa86a08a90a05cb82ed2ef11fa4f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 00:09:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/867240c0e76a467490d443430a7cae10\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/69b9f6b5091d47d1b20cfdcb363f8494 : FAILURE in 38m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7f77e6954b074b74894e477d76595589 : FAILURE in 19m 13s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/6e92a9908757449398500255a2f5b91f : FAILURE in 24m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"78a3b5b93755eab44e5ca95a5407d0338ec1fd53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 02:25:48.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8afea6fb2a5d4244bff527d0bf9812a4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c01d2ea4bf804b7181e2d8e667dd5bdb : SUCCESS in 2m 42s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/85246ef4294d4458915b154c79b417ce : SUCCESS in 1h 25m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1d63a363d3314f5ebfb0aad312d3fac9 : SUCCESS in 35m 47s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b0328913b5ee4f4fbd3d07cfdd50aac7 : TIMED_OUT in 2h 52m 49s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1d760d9e59e94a7ea4a15cf45074b4b2 : SUCCESS in 1h 58m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ef472b0c343c48398291497a124ff887 : SUCCESS in 26m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84e260adc6f1429ba1c00f271427e486 : FAILURE in 35m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c60b7cc000e84dc9ad9ab3da5789fad2 : FAILURE in 3m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9adc5404c3d848fe8beac93cce7140da : FAILURE in 35m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9add381ce3ae467f86ce07f66d3c7530 : FAILURE in 24m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/158bd244c09f4623b262b339af173d64 : FAILURE in 20m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3590c83d43b34c76ab922c4007cefda5 : SUCCESS in 14m 29s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a02975349c374b549f5840c103c33c6d : SUCCESS in 3m 01s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/6f5e139f8a594175a3bd67b063201cb8 : SUCCESS in 1h 03m 23s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/09ae07158cd34be593df1de5e8a4eb0c : SUCCESS in 40m 01s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ffaa4e67c3664c018b3fd76f2ab74766 : FAILURE in 1h 28m 18s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9cf493197ab0426f9098708e8a8425de : SUCCESS in 40m 29s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/8244e60f386a456d8eb393061ce6c8aa : SUCCESS in 38m 29s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5ad5c5e28d63440faf43e6e4de3f5424 : SUCCESS in 1h 50m 44s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a9b6797fa7fc422090d25d5a7cddff5e : SUCCESS in 1h 48m 13s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d5851b2dd2a54c1eb6dddc8dc81f98a2 : SUCCESS in 1h 05m 02s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b4a2088b06a14fac978fbf86e2e01e19 : SUCCESS in 1h 12m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1000f7a514934ca8abe4b8992f596f98 : SUCCESS in 1h 38m 25s","accounts_in_message":[],"_revision_number":1},{"id":"f868ea4444481180f956753fe7106cc9ae9a4651","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-25 09:06:19.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"790ac4fc73b0b61b614ccf199c493b590d1a70a2","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2026-03-27 23:26:24.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"1dbab13e9c6b49d821ead579401261eca1d036fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2026-03-27 23:27:12.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ed58c716523136ec203c4d3ea988428cb9d705c0","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2026-03-27 23:27:44.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"164f637afb1763077232cedf079f4c0938a75168","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2026-03-27 23:28:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"df17468308d95f7ce9c564732d6c4d669af1f6cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-28 00:06:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5781007614064d6a9e45cfb33e4d0627\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4d75cf5fba244ff486806eccfc4519f8 : FAILURE in 27m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/67c5675ae3b74839891adcbb7f5d20d8 : FAILURE in 19m 07s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/576ea9322bb042d5ab12cf96006d4d56 : FAILURE in 37m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"59342ce48f3dac4006e283546a4f03d25a117b0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-28 02:19:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bb5b63b82e1b4083b0a1ccaca89bad0c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e94da8da2cc74cc0824f849ed65cefb3 : SUCCESS in 1m 43s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/25bcb1c11e234c9fa4a048d23eaa6428 : SUCCESS in 1h 28m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a14dc6d90e7d4bdea591c8c3aaa62219 : SUCCESS in 36m 35s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/687ce7fb1fcf4808bd37c427e44cf9d2 : FAILURE in 1h 01m 09s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7e5f83ebcdbd4c1eba81c131dfbb75fd : SUCCESS in 2h 45m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/94bfa502d0434517b0c4c5ad49af90da : SUCCESS in 24m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a225b4665434da9adb24efad99a4ddb : FAILURE in 38m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc10abcec83b451787ae0f20d43191eb : FAILURE in 7m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3d91d66c42a44b68a4955b861fb133e4 : FAILURE in 25m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0d3630a5d245494ab18bfa1a1aba6d14 : FAILURE in 18m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/46cc0a7d66e743ea9d7d21c44609129c : FAILURE in 33m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/455b0b9efcb0455fbd263712fb8a801a : SUCCESS in 7m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b6bd9c7c56514a5dbc32b2eac3450bdf : SUCCESS in 2m 28s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/36439329dc194784bbe8a39ed795186c : SUCCESS in 1h 01m 13s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f99f0cc95aae4737ac1a4c950418cabd : SUCCESS in 41m 40s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/12b988caa80b42a68217e11bfabea95e : FAILURE in 1h 15m 54s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/722321e2d8254547880c69d2bd783d6b : SUCCESS in 1h 06m 11s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/f78dc82e5aa6469db3db7ae9c0098e87 : SUCCESS in 1h 22m 09s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/219a3dca9fa74f00b1b404c98c925b31 : SUCCESS in 1h 24m 48s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1a06ba7069b346c08fb5dc6f45239f28 : SUCCESS in 1h 33m 40s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d06a1d04f9444865a41cc1f9e357b8ae : SUCCESS in 1h 11m 16s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d231781e5d4242d6af6cbc6be264ac8a : SUCCESS in 57m 13s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bd79165447d449ad8a6c779a0b7299b6 : SUCCESS in 1h 27m 50s","accounts_in_message":[],"_revision_number":2},{"id":"e14320b0af725c0474e8cba0ef42494c7dad921e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2026-03-29 23:16:02.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"338d255008b5d6eb6731a340fdedb656359fad62","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-29 23:48:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54f3b36de78349d0a6e2124ffbb5d1ba\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5f25d1d69ad94a0b8fb53b7b0890978a : FAILURE in 31m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cbd0c6c7c8374298b37fdc66706b9566 : FAILURE in 26m 19s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5be890084aac411195e9d1ede179a2cb : FAILURE in 27m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"afbb8908be684df55f558fc01f5c5cb4dc0de59c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2026-03-29 23:54:23.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0eb2ceef9c593a41c2e41be933c0b35733688c95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2026-03-30 00:26:36.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f07a69b675cbcf499b9c651fc27e7c646efa0021","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2026-03-30 00:27:33.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"1deee9231cd1bc00644b33468ec6ed74c71567d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 01:00:25.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e60a9ab0b2a4426823866f9ee783be0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1b48974af79e47e4a8cd145bd46fe1e4 : FAILURE in 32m 13s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/21b79dbda84942b1be99d2b71c7cbaa4 : FAILURE in 17m 58s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8071d6b2b5054f5e9dae3d571b07912c : FAILURE in 25m 28s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"fb3af690722847d9dc5ef411a3b59651a674954a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 02:27:00.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e641078f13a94e42a4e62b7ddc386bf9\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2435c1ee059e4c3ab67dbaf769c102a9 : SUCCESS in 3m 30s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/279989e54d2d492094457c93337e3e04 : SUCCESS in 1h 23m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1e3d781163aa4859a98c83d9fe9d0f2e : SUCCESS in 59m 20s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2157623b26d14ac6ae2e0a0ee5036850 : SUCCESS in 1h 21m 35s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b7b757bfabfe45699fe2b4909aa1ec74 : SUCCESS in 1h 58m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/238e8991fa3c4eef9744a63660f9ab6c : SUCCESS in 33m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/045cb52b880b49d8bc0f6374b3ffb7a8 : FAILURE in 1h 12m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/854470bb027f46f889ef346989bf0296 : FAILURE in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5e50ff8b8f5d437aaba6f7501ea83873 : FAILURE in 38m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7d27a8bd5c8d4c95b27aa9e4a6f8e500 : FAILURE in 24m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dfaaafbd824f421bbdbdb3a922933587 : FAILURE in 34m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44c22f6ec8b048418a9ca1f6d1d8f914 : SUCCESS in 6m 23s\n- requirements-check https://zuul.opendev.org/t/openstack/build/dd8b40278b1d4d2ba31bee537999b051 : SUCCESS in 2m 48s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c66d4c4bafb24ca5ac42992b14504990 : SUCCESS in 1h 04m 43s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/5f1b1e5d50294d238ce4c2e4b6175df3 : SUCCESS in 1h 03m 06s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ff223896683141d1ab25c80c6906204f : FAILURE in 1h 23m 55s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/bcf3be4336404d90b750e29dea95e60d : SUCCESS in 1h 07m 59s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/c41aa62a773d4226a3d406b38b2cde4c : SUCCESS in 1h 13m 19s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/379e05a42e294d1ba1fdaf2ead9c6e3b : SUCCESS in 1h 54m 29s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6a069aeb501f42c1bc8c0ff078641ed5 : SUCCESS in 1h 41m 33s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5c0133b537ad4ffba3364c0716a3b1c4 : SUCCESS in 1h 10m 40s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/aa6df41a213e4edda54697fcb25d0b6b : SUCCESS in 36m 28s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1893e44ca9a843c2bd0cafcb246fee30 : SUCCESS in 58m 38s","accounts_in_message":[],"_revision_number":5},{"id":"86c799be5caefa22085735c8f3c96dab1702a70e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2026-04-09 14:19:47.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":6},{"id":"ff3ddf6c9a6f6d828d075da70aefadde58a3707d","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2026-04-09 14:20:23.000000000","message":"Patch Set 6: Workflow-1","accounts_in_message":[],"_revision_number":6},{"id":"66ee1d4916f73d944d87c03443264b11adc20487","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 16:25:26.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/40f9cbab79454414b0525402a76e5963\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/61ddd9b015184e57b665b937b83ea5ae : SUCCESS in 2m 59s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/06d701ae8c0740a9a7d8f762a50b227e : SUCCESS in 1h 27m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1793f6768f8d433390c56f9675f856ba : SUCCESS in 51m 36s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/307937ab18214e1b895f3f6ce6c99353 : SUCCESS in 1h 29m 38s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a4ac9378d4a646d6b39f447762dc2a29 : SUCCESS in 1h 51m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4f0a2d6530e843cbba5d1f4d2d20d11b : SUCCESS in 37m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8efff82c16684d1094bf74073dd38a5a : SUCCESS in 46m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0e92cb1b0624540b3f393eea892973d : FAILURE in 5m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/257c197093654ea5b628a48442639acd : SUCCESS in 30m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/57e3b6890150442fa54152730e77b480 : SUCCESS in 23m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cf6c5cf782b14ca5aa2032d1df301709 : SUCCESS in 31m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca1211f7b11e4867bbc98287aedba8e4 : SUCCESS in 11m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2be1dc3413574c85be7d6f0bbc80b6cf : SUCCESS in 3m 03s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/586c886784d942219900dec32961e147 : SUCCESS in 1h 01m 27s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9bc8af95f8814a689bf85405a80e224a : SUCCESS in 1h 12m 05s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d0c8f9e3827244b19788aef58c95181e : SUCCESS in 1h 23m 43s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9e993334afd447a88d4a1df4d345eb78 : SUCCESS in 1h 55m 59s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e0113016dbb646f3acb601142056cf7e : SUCCESS in 1h 36m 25s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d1a594268f9247308b319fa615e09ab9 : SUCCESS in 55m 54s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a4861cf8ac79448eb58ef1490efff286 : SUCCESS in 45m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/62b341bc081b4d51963e81e9ebe96207 : SUCCESS in 1h 12m 26s","accounts_in_message":[],"_revision_number":6},{"id":"73ad6fcb953c48957f0b3b204a862da6a1b546e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 16:49:17.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/768b643b4c46429ebdc3008a1a432096\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5444f74874f54329b8c157bf4edd2b76 : SUCCESS in 23m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/63850be0c7de49009e8a9fcaf241f96a : SUCCESS in 20m 18s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e2289296f363445b974e57a9e7652ea1 : SUCCESS in 23m 51s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"077e1d551363e51e74213cefe1d3c1999857001b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2026-04-13 23:35:40.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"6b030a612c5012ceaf2160213430b90b4534f056","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2026-04-13 23:36:28.000000000","message":"Patch Set 7: Workflow-1","accounts_in_message":[],"_revision_number":7},{"id":"0bbe478e0074af112d3e4ea8d9360e9b902c3105","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 01:20:41.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/176e5e79eacf441b9d43538d5d0cf4a7\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c28ba8b5d0bd4e8fb0507ac616c7121c : SUCCESS in 31m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/714975014d9c433786f2a845e97eeda6 : SUCCESS in 19m 27s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/fb991db53f634fe1a352329903f8a16a : SUCCESS in 26m 53s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"fa86f34617147028c17d8f35dac7f395196211fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 01:36:06.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c2633805308a4fcba30cf0d27f8e6ba3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7b69f37a47ab414fb96c06ed8e9ee02f : SUCCESS in 2m 30s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/638dac7d7de745669b9fd18ed7a8bed0 : SUCCESS in 1h 07m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8eecab7769864aa2b77e24dc012c361f : SUCCESS in 37m 00s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b1a3676e781a41cd9efdad2a7c0d8209 : SUCCESS in 1h 21m 10s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/cc877de882a5465fae78972efbc67f53 : SUCCESS in 1h 58m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/97a30c636f6c4a889b1ef6c76097ebdd : SUCCESS in 33m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e082d944c8a4a18b2bf2b89a8fe1089 : SUCCESS in 45m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f639ef4a19cb440884e9f371477efdd2 : FAILURE in 5m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/737f691323a44070aed7f027511ceaab : SUCCESS in 28m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7cbfdda1e5d84188a5fc537171a5ea38 : SUCCESS in 18m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/97af067b37a744838511b94659b69097 : SUCCESS in 26m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9736bb6f930e41d78004d941f207ca03 : SUCCESS in 14m 29s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5c260eaf61d8491baff859fe015aa32a : SUCCESS in 2m 39s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a8d43d85762b48e9b38eecd1c5d962db : SUCCESS in 1h 18m 57s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8c449f625f97452f9ec6e8f8d4ebd97b : SUCCESS in 1h 11m 18s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d5827a513e8c4efbbce347a453ccadec : SUCCESS in 38m 26s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8a997b5ec4a14ee183c63e0f6919d1f9 : SUCCESS in 1h 07m 43s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/847767ba83624b8c9a42aa5fc8eeacb3 : SUCCESS in 1h 18m 33s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e181ec7c2c6f49e8b3361c78861f1c52 : SUCCESS in 1h 02m 27s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1689335df7dd4c488ec0acc3fb43f56c : SUCCESS in 1h 12m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/633a8b3da3c54ec09f6e21f1ac437816 : SUCCESS in 1h 07m 26s","accounts_in_message":[],"_revision_number":7},{"id":"e9df1377d6a88f46a3d0752f2668fd74101cecb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2026-04-14 21:32:47.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"90ac77ce94400465e8aebcb3a432f09c1021244a","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2026-04-14 21:33:24.000000000","message":"Patch Set 8: Workflow-1","accounts_in_message":[],"_revision_number":8},{"id":"71afa66284682cce2cb3bc7ade55c8ed1bec7c67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 23:43:17.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5797462114c141e99c2287d8a755a0e6\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2b7e650ebf914f8991b406baa8a48ad7 : SUCCESS in 1m 54s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2a216a8050474e4989a227d751d0e6dc : SUCCESS in 1h 29m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a2e66dfeef654c8082c72cb4577c8d5b : SUCCESS in 1h 05m 02s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/91caca52c0464b9c81db3857472ce6f6 : SUCCESS in 1h 19m 15s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/53fa54723ace4f749912803c263dc348 : SUCCESS in 1h 56m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/128ad7a122e5401fb62a7107b8af72e9 : SUCCESS in 32m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/592b3813b24441e187d575e48b6ab297 : SUCCESS in 1h 00m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ce5d3c00f164901b3a1ff013eb461bf : SUCCESS in 13m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ab49240b1d224d0a99744ace70412f52 : SUCCESS in 33m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/43833fae6a1a47619b1e0cea8f98b510 : SUCCESS in 30m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/271ccefdc0f74b7187c3d7a46282b3dd : SUCCESS in 31m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e805510bd134c7893a2ec5402adc2ae : SUCCESS in 12m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a92ed71ccccb470fa60bee6c2ebf22a8 : SUCCESS in 2m 06s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/81e688fdf2e547c5b90cc8ab5f482e7c : SUCCESS in 1h 28m 26s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/cde9d39c8b834149818a48f672064cbe : SUCCESS in 1h 15m 13s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/6a22b590575e468982de51d8b0967075 : SUCCESS in 1h 19m 07s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/eb8e71bc669e422c8ad1ba32aca3a46c : SUCCESS in 2h 02m 36s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1a57535227604c45b8587b204bd2cda2 : FAILURE in 1h 23m 57s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1d2af7d9a9c84a588a2010c9bb39c32c : SUCCESS in 1h 11m 08s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7d39457b3831470bad82051d1aace8cf : SUCCESS in 1h 34m 28s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/452042cc61ae4a098685f79a3660c42c : SUCCESS in 1h 31m 38s","accounts_in_message":[],"_revision_number":8},{"id":"5e83afe82c74b7a17ef270851610416e64bf9b0b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 02:23:48.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/844d8fa7f6fb434d9ac19087a0a40212\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/60368e8d2c8e4f48a4997e1a33511a44 : SUCCESS in 24m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/307ec338e8b64005bfed5f44f486d543 : SUCCESS in 21m 50s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4b76b2e80bd5437ea627b867c8237ccd : SUCCESS in 33m 13s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"8f339db110fafc865ef6a5af77f307afc4d9158a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2026-04-15 23:25:02.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"a5b96e4eae3177a185b7cdf2eb9234b450832916","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 00:06:00.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4fbf99b75fc4f6eb5b2a5356c8b5130\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fb35195a6ff64303a6080e74bbd25d40 : SUCCESS in 26m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/affd65c0f2044cefa620673dab11110e : SUCCESS in 18m 47s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/69513718b5084b14a2030426fd9f7b52 : SUCCESS in 23m 59s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"e41d48ff5a859537d1a635c5d8906a8f10c9b996","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 01:30:00.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a6f3fbda3fd430db1ec08bb78f3b32d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6bef71ca04b64658b0a4562dbdc2a075 : SUCCESS in 2m 02s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1641593482474c9aa53ea19741fe627f : SUCCESS in 1h 09m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/da0a3be0f63d41e8adb1b3f785b6a415 : SUCCESS in 1h 12m 31s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/03ec91c8176840cfaa14ed443b670d3c : SUCCESS in 1h 23m 19s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/dc2b395c20f646ab9db026df98fd23ab : SUCCESS in 2h 03m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/20f895b3c61145b39edc276b8dad21da : SUCCESS in 32m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99fc77c930ba419d93be1ad31e90a349 : SUCCESS in 1h 14m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/824332dce0514f62ab0542f7d22aee15 : SUCCESS in 20m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cb27f432b3a74f1da581b815a1ae636f : SUCCESS in 38m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1dc7404cf6c94056a664a155fc00be93 : SUCCESS in 30m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/25ca4244d1fc47c8b84d25cd39c8cf58 : SUCCESS in 26m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fbc6a70029c34ad4bfbcedbe239af246 : SUCCESS in 23m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22f802b05c724c13bb2500273eb1dd31 : SUCCESS in 11m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/714764eff7be418aa77b2f1f452c02a1 : SUCCESS in 2m 36s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3ba16916d6304c4fbd2c1295eabeb47c : SUCCESS in 1h 20m 03s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1ff8ea01886b42b1a45bbe2846a9b0c5 : SUCCESS in 1h 14m 10s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/cfc7bead7ad3495a9be62504e3d2c776 : SUCCESS in 1h 13m 44s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3ef4c8b4bb564fbebce87bf24c3726c0 : SUCCESS in 1h 57m 40s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2ccbd7f3081f4071af07e2af334c0253 : SUCCESS in 1h 18m 12s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/dedd0ccdc8514ee689091960a31c0e15 : SUCCESS in 1h 11m 32s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f823c93f4fe04fcead8c64b11238ee2d : SUCCESS in 1h 18m 34s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/525e24d5e32149fda96de570105d541b : SUCCESS in 1h 25m 24s","accounts_in_message":[],"_revision_number":9},{"id":"18149e7accf6249eafb12adb949ff0726026b06d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2026-04-26 14:55:20.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"a6492c1d69859208cb00776c07fa0789ad0dd679","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-26 15:23:46.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b897078c763a41d18535ba9a5b49a887\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/82d9c0d108f74b72a05764d66588ca2e : SUCCESS in 25m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f76987d937494587a4c1852f4fe1353c : SUCCESS in 16m 38s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4f2c34549b1e41bda1154da8801db3f4 : SUCCESS in 26m 46s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"2125d7d92460f2715092019e2c4ef76ef5948880","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-26 17:18:04.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/176a03e533cd457c9c8607635c2b3e6d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4e208c5846d649d7a06fecd6dbcd520e : SUCCESS in 2m 54s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/24470746c63c4b36b1d9b224d3afb240 : SUCCESS in 1h 36m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/93e95ee699a24a8980cc411c0d070571 : SUCCESS in 1h 09m 55s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/00c3396a57834b3daa9380597ddeda88 : SUCCESS in 1h 18m 55s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a913cf1a775e46b3a12b30eaa3e3e9b4 : SUCCESS in 1h 38m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/51171cee1c1d4d18ba83926deeb71ad6 : SUCCESS in 30m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f07ea4b721f7480298a7e6ffeea178c1 : SUCCESS in 1h 02m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/812ca9e94011424d90491142688305a6 : SUCCESS in 27m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f940f43650d341e592e4e071a37408cb : SUCCESS in 37m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6f42e6c22898488e910b5469f92f978e : SUCCESS in 27m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f15a248f0b2d4fb0ac1fd998830df412 : SUCCESS in 24m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c41a3178450c4270aa7b048304b73432 : SUCCESS in 34m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63873aca0f324b919e82b6417f18ac3d : SUCCESS in 11m 33s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a22e771d579f4144836418385b1f6099 : SUCCESS in 2m 52s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/25dfa9f4163d4df09d6014ec7ced2f27 : SUCCESS in 1h 25m 36s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/126d02d8b9c24dd7a04703e08bec55c2 : SUCCESS in 1h 05m 57s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/e74f0e29a65d403ba97bbe0342306099 : SUCCESS in 1h 20m 40s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/67edaa489e544169a74a33852d8512bf : SUCCESS in 1h 52m 33s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/38cde754c3024b859c0481fc97dce0ad : SUCCESS in 1h 54m 35s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2cefd2d1915e4d10a32cfae1b79a4508 : SUCCESS in 1h 07m 31s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ae9649fc9b984ed5954c64d6acf09c3a : SUCCESS in 1h 12m 44s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e5df450c97db4a4ca7e0880dabc7b141 : SUCCESS in 1h 21m 15s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/8a28b48a42144a6f8c9351ecaea1f284 : SUCCESS in 2h 16m 07s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"1dccea0c0a82e07af643540e79ded01032cc8629","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2026-05-06 23:04:37.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"2099bb8d91a12ec176a72bf404bdda197251fbb4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 23:34:26.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5528cdc8e5db4be3ab19e065da1a0081\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8c515f83eb284c068a0485f2acc2fef0 : SUCCESS in 26m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/42a408e45dec42699e48c1aeb9b9dcdf : SUCCESS in 18m 07s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a97ea6212f824b2fa883917c01f485c3 : SUCCESS in 27m 24s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"88ac6ea94ccb71c5a75f5a2d180e429eb63071f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 01:20:41.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a19585a823a04dfb9af120d98d099a6d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/923c5392d0154205b986459bd94a11c5 : SUCCESS in 2m 32s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4bd45b6b044045239406e96db573e32d : SUCCESS in 1h 27m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0f0c5a6183254fec99344d6909cadf66 : SUCCESS in 1h 08m 05s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a81b6cde1dd04210b436fe149d9baf7e : SUCCESS in 1h 41m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/89be30565980432fb1a3bc52fd048b82 : SUCCESS in 1h 42m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/da5e83d118ec4f06883b20f2d9c9988b : SUCCESS in 32m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a7354f561da94f61b1ce1c33dd75ded9 : SUCCESS in 42m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e4ae31767774589a01b0404233e2856 : SUCCESS in 10m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6bcfb38fd19c4e588530062178ad57f7 : SUCCESS in 20m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9a5a3533a6c243df89b8f894201abeac : SUCCESS in 29m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b8c11ebc94b84c8c9ab5de70fef99346 : SUCCESS in 26m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1f47e6e265d7438181073852abf37cfa : SUCCESS in 22m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27f064a96af142d2ad4941f5b5660557 : SUCCESS in 11m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e0af64608e524402bafa1d50749c0a7d : SUCCESS in 1m 45s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4dc15e30827a427a92c51e1837b5eb02 : SUCCESS in 42m 53s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/bf0fd5206427408ebd058ce92e34d22a : SUCCESS in 1h 17m 50s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/040abd48c86f408fb30ea990f674a405 : SUCCESS in 1h 18m 45s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/dd372cb4780e45babf0fd283f48926bf : SUCCESS in 1h 58m 20s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/041f65c0a7f54b18bf34e1f5d01261f6 : SUCCESS in 1h 16m 27s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/99fc32ac712543f48e1d098f690704af : SUCCESS in 1h 18m 40s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/218587e3008a4e358439f89038ea998d : SUCCESS in 1h 08m 05s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/849cf255945643fab7417c7533ff84f6 : SUCCESS in 57m 49s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/751440badcc7477ba56da5a1a7af6378 : SUCCESS in 1h 19m 56s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":11},{"id":"ac9742c89f1d001d710328433e8bab6830c08ed6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2026-05-18 15:37:08.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"00c80df13735bcb9d41f0e9cd43c2860dbba63bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 16:41:03.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73504ca9b9b6451bb07bd3a0809b0bfe\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/eae6d21c4c8848849ca754cced9f7b4c : SUCCESS in 56m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/78a65885fb444526b70522aad9326ba6 : SUCCESS in 46m 32s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/17fdc11691af473096c57646e588ef62 : SUCCESS in 1h 01m 42s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"03e78a78f0606c08762cd2d3ada82a45acda8d0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 18:07:30.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b79e0ad9d18452695c4c40b1bbe7088\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/efd816b43306464db2a0a4f9b593fcea : SUCCESS in 3m 17s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/286f8b8ed95440578eb82e22384b9107 : SUCCESS in 1h 32m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b8b8d4c3cece4cf3a2ce3243da1744f2 : SUCCESS in 1h 06m 12s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e6bc1321184b4deaba77a7b94b17d4da : SUCCESS in 1h 58m 16s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8e579d973a054fe192094038e2cdd200 : SUCCESS in 2h 27m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e9c6e6b177d94e6b9d283cec28f2f2ac : SUCCESS in 33m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/243fa6bd5c184482a85e57e6d938a50e : SUCCESS in 1h 06m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef29d0bf46ff4970ab5bccd048b6b927 : SUCCESS in 12m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d642bc0e1f3b4b7e89cf508f331dcc09 : SUCCESS in 31m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9053a5376b98499a8ada15e866f81bbd : SUCCESS in 33m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/32d76689c115493c95cb5e555745e26a : SUCCESS in 36m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a0c9636b5334d5fa401608de9755597 : SUCCESS in 11m 36s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5d018c7bdc514ea18fb72cc9768de612 : SUCCESS in 2m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/861f7bc92bf04191b09cdebaa6badc8b : SUCCESS in 19m 23s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b1bea9498b5b4b76af87ee427d188d00 : SUCCESS in 1h 29m 06s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2f209494958b4de39ed02e3b0350cb60 : SUCCESS in 1h 15m 56s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/2c508cb7300b42e1850d6be281c0f115 : SUCCESS in 1h 27m 32s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e87fa57e9ccf40c6a235c7b566941bcf : SUCCESS in 2h 08m 37s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/122492903a7047c58e66ed75aacd2d69 : SUCCESS in 1h 39m 52s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a0e80dcf731f48bb81888327980308d2 : SUCCESS in 53m 39s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b8ea707058044387970422bf3e16d220 : SUCCESS in 1h 21m 17s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/506511eb64f34799a08bb4b7384f5a36 : SUCCESS in 1h 34m 08s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/0e89ab30f0f04f1893431d34b51e989e : RETRY_LIMIT Host unreachable in 50m 35s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":12},{"id":"0a1d029598b3c4480f90a4551e54ed1a912dd956","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-18 21:38:40.000000000","message":"Patch Set 12: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/18/982018/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"0538441b0d229d1b9d593b60b01a70ccaee00c76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2026-05-28 23:52:10.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"d6c57d1b6c7b54d8defc42a6cb0f7e30c862e341","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 01:59:40.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5935129147da4f74b9ef9ad56760a612\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5785f81571d04925b8093387bb7c1c5a : FAILURE in 18m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b2e19d05320942448d202f62ef1b8747 : FAILURE in 23m 44s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3b874d909f664ffbb2870af1552183e8 : FAILURE in 1h 08m 02s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"06cf0c0fbc008a060d3ee7d2f3cd87b7a720f9c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 02:59:12.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/54000b0a74724a169989d20ef8cef9ba\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7601d863eb0e448cbcdc4a1ea4618c77 : SUCCESS in 2m 53s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/eac2a991e4634db5bf299b3b2a2b7bc5 : SUCCESS in 1h 47m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1ca9b6099dd04c40960970da4a5a76cf : SUCCESS in 36m 51s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/16a6595ef9b34816af806b71e71cb355 : SUCCESS in 1h 32m 16s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4a5f2647c10d47a388c93b8b695e072b : SUCCESS in 1h 42m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/904fcf6d992b4a91b78371b3c1c150c9 : SUCCESS in 32m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f52e6a4df7f04b1bbef1b32050ee0bee : FAILURE in 1h 01m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c4692c54d2f4cf3ab0d90afa28f40f2 : FAILURE in 5m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ed6614926b6d4da68328ce2ce98213ec : FAILURE in 30m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6f65a929b3fa460bb5bf841bdc2ce025 : FAILURE in 24m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/73f8f29edc644e0589590ce0c7028cf1 : FAILURE in 26m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d41317a989034bbb8fac22055f2535e0 : SUCCESS in 12m 57s\n- requirements-check https://zuul.opendev.org/t/openstack/build/907dcc66fc08414e83ffef69127c1aae : SUCCESS in 2m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c43e57b9fe047439cb7557e6ed0f4db : SUCCESS in 9m 30s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4a6e4de5649a4f129f6d846f226525e1 : SUCCESS in 1h 18m 03s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/34f9a325ab7248c985b35833df04b45e : SUCCESS in 51m 30s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ef857cac2b424ef9b7f62b649572d85f : SUCCESS in 1h 12m 22s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e1d86fa6fbe947dea006d774329b4471 : SUCCESS in 1h 51m 39s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d2b0d2e1635f41f2b5e78205cf132bfc : SUCCESS in 1h 41m 07s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/152524755a754ed2ade2d4b435a27ec2 : SUCCESS in 49m 06s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/341a35de6a374784a8165d631939507d : SUCCESS in 1h 07m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5380a56215224355bc764d93b4786701 : SUCCESS in 1h 29m 33s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/0c53c78991dc4e1787e2504c57d04003 : SUCCESS in 2h 57m 06s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/tests/unit/common/ovn/test_utils.py","accounts_in_message":[],"_revision_number":13},{"id":"1ab7101fd70864d31d3604f3d6133dcc6d4c5f45","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-29 04:55:33.000000000","message":"Patch Set 13: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/18/982018/13 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":13},{"id":"00725cd4836ac6ec9c8bb31c8292585e2abb2427","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2026-05-29 21:55:27.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.\n\nOutdated Votes:\n* Verified+1, Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"09a9e66bc2acb42429884e63613145e393b18e5f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 23:16:00.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6b536ab17694d059b6edb1b967c77a9\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/420b760c51514d2dab672c01ee7ec054 : SUCCESS in 29m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/44e404e35349499eb88c4fe1b6f5ed85 : SUCCESS in 24m 13s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ab284e955956475cbd6eff0929630950 : SUCCESS in 1h 08m 30s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"0e60c93816b2c0306f0360098a6da31a68e1b68a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-30 00:16:31.000000000","message":"Patch Set 14: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/18/982018/14 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":14},{"id":"eb612e71242918c6eba40062b3b9f2d6d6ad552f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 00:20:16.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d7f9f67302a74708aaaa4a04e4f12121\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/85f7db84b0b24e14b7beab1e43b40c76 : SUCCESS in 3m 25s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/024232a8020e490892b306472866a55a : SUCCESS in 1h 16m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0e6f374049794f5e9cdac4204a5eb373 : SUCCESS in 48m 31s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/96c3373a95404b7aa5ebf60a121f3a05 : SUCCESS in 1h 33m 27s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d60909c200a74a54a1a07cfaca201968 : SUCCESS in 1h 53m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/635c22b1e3924f199dc5c45c2424d2f4 : SUCCESS in 31m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f48decca70248ac837955136e7e48f2 : FAILURE in 1h 09m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10c43c4cb71e44978025126492505d36 : SUCCESS in 7m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/83c6fc8472fc4d8dac1daf33ce167ee0 : SUCCESS in 25m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2d06f80fd69042f3a0dd76b6f8cddd00 : SUCCESS in 25m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f21318ae660645a29ab6db952ab551f1 : SUCCESS in 24m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79fb73c093954719803ee0f2e6490cd9 : SUCCESS in 11m 32s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8df8b3eceb6d410b893d0e45973ecb5b : SUCCESS in 2m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0d254bcc103435891debfb352444d08 : SUCCESS in 6m 50s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fed75be3c724460bb02756da5de36bb6 : SUCCESS in 1h 10m 03s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/125518dc087f4f1bb59dde10632841a1 : SUCCESS in 1h 15m 53s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/81b47c82c37d497c92385c97edb28462 : SUCCESS in 1h 30m 01s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/20bb95f4341b44939f7237d48099a5a1 : SUCCESS in 1h 54m 49s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/79c6b6c929bb4e60bfd6e160f33f9313 : SUCCESS in 1h 22m 33s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0480f66d4880465eb548cbe9ef31edbc : SUCCESS in 53m 30s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/fe354d19429344c793240e136884fab5 : SUCCESS in 1h 10m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/72743a48f4c3401eaac150f957e85f2b : SUCCESS in 1h 37m 46s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/7e38bb6107744d568a784a541f11ee01 : SUCCESS in 2h 19m 08s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":14},{"id":"6f31fab65babad478049457607995a57077d0eff","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2026-06-01 21:12:21.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"9fea6988694b730eec791d649e935f8bfa660255","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 22:39:44.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f121ef8337c5433481c80a68b91ed15a\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a71a2d97c00247c39e9ec6e614fc0cf3 : SUCCESS in 21m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7f3c84c7c6534e7e8693761c6f356852 : SUCCESS in 18m 37s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e6b1e28c00514d61a7bb1f1183631302 : SUCCESS in 39m 08s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"8d9bf1fcb96847f22f22e7b928144a8274d2664b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 00:11:52.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b26ab88c8402467789fe5c19abe27fb7\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/28df9adeb8ab4abebb02c8a3efb32dd4 : SUCCESS in 3m 24s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e2b16c805a664874b2f4b4053da31343 : SUCCESS in 1h 22m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f5877dcbd90e4d448e01ed017a537937 : SUCCESS in 1h 12m 29s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/330d7caeacd64e8c9173a487c3f0611d : SUCCESS in 1h 20m 27s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e1d1960d58e548ca9bdd126380b325e0 : SUCCESS in 2h 23m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1832fc462a464e45871ecb474459aafc : SUCCESS in 33m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/82943cbebf534dd5b4401694e70adbc6 : SUCCESS in 1h 10m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebca19e7c1d64fd8a94a8dccefd98d95 : SUCCESS in 12m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f87a732a1a0b454a9913f9259a426c0a : SUCCESS in 29m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/702f634d4575417baa6d3def82bd5ef9 : SUCCESS in 17m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/592ae56d5b43405683321a26b70a5dde : SUCCESS in 35m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17344920cfff4b169675c485fc5c9292 : SUCCESS in 10m 42s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b0cc0321e8b145b5bbeab9bff006d95a : SUCCESS in 4m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/546a588c599e4425bade8cd8f6d581e8 : SUCCESS in 14m 24s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fc0c9f0cf68a4e4d892016f874d7afc0 : SUCCESS in 1h 24m 41s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d49d845e7fe442498d1871cbb44c4d55 : SUCCESS in 51m 39s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/cf762fbd55984c7394a51862fb11ef19 : SUCCESS in 1h 14m 51s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f1862b3d1af94fe59963c55103354ee2 : SUCCESS in 1h 49m 24s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d2d4a54d8495455086c32b78fd2f89ca : SUCCESS in 1h 42m 56s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ca37433210064d80a60ae727f46d7110 : SUCCESS in 1h 20m 29s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ead7b7ccc8194641810ff15cffbc1db8 : SUCCESS in 33m 35s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/06d2d6a7873641abaac8090e35234663 : SUCCESS in 1h 13m 55s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/ca090d5696d64d3db752b9a2f4091f12 : SUCCESS in 2h 51m 17s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":14},{"id":"0685b6a9e8bb949ce115a2c4d030be42b92611f5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-02 01:28:16.000000000","message":"Patch Set 14:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/18/982018/14 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":14},{"id":"002ff947a927cb87bfb8a5b9a67225905cf7c07e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2026-06-02 14:05:31.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":15},{"id":"f0297c4592997e38a807e8e78848dda5f0cb35f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 16:55:04.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/deb17e899bc6419b8c9c8b199d3eeab8\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ce5ae84018264cb792629115f4ff4589 : SUCCESS in 3m 23s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/55614a40975a4d4bad2936a4ac0c4492 : SUCCESS in 1h 31m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fa22af15c55c4b5b928d6d1d5b1f230a : SUCCESS in 1h 09m 00s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/125b208370e74b9c8fc211d45a14135c : SUCCESS in 1h 35m 55s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1dcc1164b1094d0486f7140bf743b394 : SUCCESS in 1h 50m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1903445bc1bd4881926d933f047c772f : SUCCESS in 32m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f2e4fc9e69341988e8902ab61dea376 : SUCCESS in 1h 06m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/736bb70649384b1bbfe188d6009b7fce : SUCCESS in 8m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/64eac9ad74f64446be77a51b78bbe8fb : SUCCESS in 31m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/36df53bd90e94fae95a421a1ac61270c : SUCCESS in 20m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8b84bbfb800d43b993484464f60a448f : SUCCESS in 34m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f39c35e2b68461f8d9d9038c69ec4f7 : SUCCESS in 15m 16s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a5f68a5e3aba4fcf9e55f018d81e22d5 : SUCCESS in 4m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a023531154524f69b0a7c860d9411896 : SUCCESS in 20m 17s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/81f1fcf4d20449dc90f8f0b7dc23bb81 : SUCCESS in 1h 24m 50s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ea0e3648e8e54e2e923ff9f2b0a891c6 : SUCCESS in 1h 06m 29s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/beecde981bb34ba6aa0e271ac3f66670 : SUCCESS in 1h 19m 34s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/620606938c5944d5a91d0d115914f4fc : SUCCESS in 1h 37m 31s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2890c635644f44a88a4ef40b4c15c062 : SUCCESS in 1h 50m 36s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a2c0c88005054263a18a164d4bded295 : SUCCESS in 50m 57s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b9d9421643b04cb5aeb3c41849a63b8b : SUCCESS in 1h 02m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1503f2d6e89b4825bd5ae1ae4778d193 : SUCCESS in 54m 30s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/e44467df65604edaac39052adfd6e86a : SUCCESS in 2h 27m 35s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":15},{"id":"c6c68e5967305295db8c3b5c0a55bc32243573ee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 20:07:01.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d3161571c8b4171b4e23f3cae4ba741\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/732f632e11c24e0d9c26de49386def01 : SUCCESS in 52m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/67c33392739d4d498296f43ce3b111c0 : SUCCESS in 36m 14s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a52b970213384f2fb5a91a7bca3a0f81 : SUCCESS in 30m 05s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"4e29b9b1062b788501eddcea1189dee476d126ac","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-02 20:18:50.000000000","message":"Patch Set 15: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/18/982018/15 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":15},{"id":"a8e6a668adaa95507828a277678c328d5ef3ad9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2026-06-05 21:13:04.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":16},{"id":"7d33fde2a71d560a948d540028ba3567f01944d9","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2026-06-05 21:13:55.000000000","message":"Patch Set 16: Review-Priority+1","accounts_in_message":[],"_revision_number":16},{"id":"ca0622196fd5b51956a187a20699bfaee0ed043c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 22:39:06.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5be1974b2df40489a7ae71803b2b96d\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/30438d21dff0483ca9163211e36f28c0 : SUCCESS in 1h 19m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7f147ef04d3742eba808f34b058bd756 : SUCCESS in 1h 09m 01s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/fa4a231000fd466ea14c2bcd86d89a37 : TIMED_OUT in 1h 22m 31s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"b0143215fe54720b5143698186b40742e4399976","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 23:32:54.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/52c81ee6d6944cd49b4fe348d4fbd5c0\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3ab96861d6644f18a7fc9bf63c2106ca : SUCCESS in 4m 25s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1772126d059843ce908ea841d8afb38c : SUCCESS in 1h 04m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/098ee45bc4c043babdbeb7a21ef0e362 : SUCCESS in 1h 09m 32s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c705afbdbd9644f498a783fc6752aa51 : SUCCESS in 1h 33m 02s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/41a893579b88485ca2c617b0a1df4c09 : SUCCESS in 1h 51m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a6b3a3c6c6b14361aabd9c57f530d51b : SUCCESS in 34m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dccdb73d9a284bf29c5146237601e260 : SUCCESS in 1h 05m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af124362633646f7be10347fdf588175 : SUCCESS in 10m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3e0aa28aee364c88903e09469e26b97a : SUCCESS in 33m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4c64134202884eb0ade4802fd0d3a664 : SUCCESS in 17m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/512785fdf9a049d88c28b24bfcbf4d20 : SUCCESS in 21m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d20515379c244c2ea18fa08f5ec4567d : SUCCESS in 12m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0c81ade56f334686b001e1b1e82d7513 : SUCCESS in 3m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4deba4c79b644dc7b8f159669525c34c : SUCCESS in 9m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/946cae8c4b1b431fa73a67964e2530e9 : FAILURE in 52m 50s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7f49bf0762694c229e4757d45e03fa30 : SUCCESS in 1h 09m 43s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/9e9cf87d6ddd464bb7c073756583e64f : SUCCESS in 1h 18m 18s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/628f83b074da457a95bcbfa3fae162fe : SUCCESS in 1h 43m 05s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/47845cb3fcb5433cbb0dd346c6c30de8 : SUCCESS in 1h 08m 49s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2bd0c09faa5549db8a45cfacac422dc0 : SUCCESS in 1h 06m 57s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3be2bf94be0740c8932571c9c2a34821 : SUCCESS in 1h 19m 32s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1f9bca2435be4d0a8346bba8f1cc3cfe : SUCCESS in 1h 26m 22s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/4483360787ad4ae0b73c8a189e548c91 : SUCCESS in 2h 12m 31s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":16},{"id":"16819361f7c412008ce7ffc4ab789e0e8b6c63af","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-06 08:26:25.000000000","message":"Patch Set 16: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/18/982018/16 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":16},{"id":"063c1fac87bdb20e784875249080a7d8ea97b517","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2026-06-15 23:31:45.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n* Verified+1, Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":17},{"id":"10c6560134524a726f26489e2d6116ed42bece8a","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2026-06-15 23:32:59.000000000","message":"Patch Set 17: Review-Priority+1","accounts_in_message":[],"_revision_number":17},{"id":"4dd4b0ba9a48dfc24795f90f74ac269303d812ac","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-16 01:17:07.000000000","message":"Patch Set 17: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/18/982018/17 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":17},{"id":"7ef2795fe52594af83a55a339f96607777a9b771","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 01:38:38.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5af4e765b72c4cb38b21bf3f108782d0\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5a0fc899773a4e18b3fc891a85b1047c : SUCCESS in 2m 14s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/883d50bc3a07433883647297bd38b370 : SUCCESS in 1h 30m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6591ae449b15408ea18ecddd37945a89 : SUCCESS in 1h 04m 21s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3a7f315fc6c74d499a827e157d0dcedf : SUCCESS in 1h 23m 42s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9042ddb43e7949718ba8ff192b467672 : SUCCESS in 1h 49m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0e44e174d49e49088813b4d1beb6e4b7 : SUCCESS in 31m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/258bbe7e7ee14f4bb620e392a4a35f94 : FAILURE in 1h 05m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35466835aad242939963f9f76f5c28bb : FAILURE in 9m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4f2a268c38984c1d9c4c2c1ba3967dd6 : FAILURE in 29m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e0145bb7d53e422192d19261c758ea70 : FAILURE in 28m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/326a02f394cc492198ac395e6b430e71 : FAILURE in 35m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f7e67f4448f41ebaf56229e654a24bd : SUCCESS in 12m 03s\n- requirements-check https://zuul.opendev.org/t/openstack/build/492cea912f32466d938d3edaefb5ce51 : SUCCESS in 3m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3e446e21c5845f984b97653c12a526a : SUCCESS in 7m 29s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0b327139ac394f578611375509217472 : SUCCESS in 1h 32m 24s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0102a4c1e5fc4007b7125e782db801a0 : SUCCESS in 1h 12m 52s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/6fb1b2a5b7a74c77945274aeed1e095c : SUCCESS in 1h 20m 50s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/dfa819fc37a14cffb90f79e3be4c7245 : SUCCESS in 1h 57m 26s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/cf20e55cc94b4cbf86eaec48d1fba1c5 : SUCCESS in 1h 48m 05s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ca905af9374e4acb826391f4c35eae67 : SUCCESS in 1h 09m 37s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/28011c8f8ad7493b8c08f81a43f917b6 : SUCCESS in 1h 18m 42s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2e5d1990ee43419f8b92d0cf6ac66032 : SUCCESS in 1h 27m 20s\n\nWarning:\n  Comments left for invalid file neutron/tests/unit/plugins/ml2/drivers/ovn/mech_driver/ovsdb/test_ovn_client.py","accounts_in_message":[],"_revision_number":17},{"id":"01a93aa06fc59d90880380d9b613dabe102956e0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 08:11:32.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d4bb4c936ea449089870cc0d13cbfcf\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/16c3952ec2ef4629bc58057ebb14e354 : FAILURE in 20m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/43fd0bc95b454846a0787015318963d4 : FAILURE in 28m 47s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/877e087d2e514f6ca4bc68f1ec4e3124 : FAILURE in 35m 17s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"40b0b6ea469c79668d9afaec6961f6baab2cbabd","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-06-22 08:28:18.000000000","message":"Patch Set 17: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"d9b8153b3273f66508c6fb308cf47b943d481984","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2026-06-24 22:19:02.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Verified+1, Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":18},{"id":"24abad6c2023f4f0794e7837d5dc5a521f1fab18","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 23:20:24.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a32ac0e2c144e4b8b0309658576bb13\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/41876c044da04016a4b2aa7d490f9b37 : SUCCESS in 45m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ea7f23350a454e27914b219d36fb116e : SUCCESS in 41m 25s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/85dbae7e3bc745c6a881cb96f2c05c45 : SUCCESS in 59m 06s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"958525dbdeeaaea7d8848f6f672c015f8c14c6aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 00:36:03.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b20369c9d7a44a929045ae60d7fa1d81\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/630e5983df19469e828b857623349b7c : SUCCESS in 2m 15s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/25b80a8648084f2fa8dec21a9237eb91 : SUCCESS in 1h 29m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7b173bb2f41c4a63bb9d9bce1215b963 : SUCCESS in 1h 02m 36s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b80ca2e3e48a4b25978014a1396c413a : SUCCESS in 1h 23m 21s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0281590071cb4a0da5e49c67a57a6d52 : SUCCESS in 2h 12m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1b06fcc058ac4fc18fcb29f96ce9ea03 : SUCCESS in 30m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9741da198ff849508845523f1b4d138e : SUCCESS in 1h 06m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8022e339174c426085743449e0494f8d : SUCCESS in 9m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1be429905d774b4db6016844859eb962 : SUCCESS in 29m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5e3f0446123949f9bbfd842b8a418775 : SUCCESS in 20m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9b21d667c5e344c1bb8f4e78f5571476 : SUCCESS in 32m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0ace6f34bc7429aad30568d12ec82af : SUCCESS in 12m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/84a28eb2932e484bb14e9bf58701e5ae : SUCCESS in 3m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8845ff2e14184e08ac5067dfae4346b8 : SUCCESS in 14m 15s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a81f2e52a2af40e89d763d3b94aaf6a4 : SUCCESS in 1h 26m 31s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6d65cdcbd1694fc1b0a0f96b17cdc3e3 : SUCCESS in 1h 06m 54s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/8eb2511a28064494bfb7c509c615b1ce : SUCCESS in 1h 18m 38s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8dbea40bbf9f4987901faa1a0b3e8136 : SUCCESS in 1h 55m 55s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/597a50e3cd554584b564905bea250f6f : SUCCESS in 1h 26m 30s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5f8b6237dcbd4b89a1e0d7c3612ebeaa : SUCCESS in 51m 08s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/9c0530c94d8d43b4a58a3be9d193d652 : SUCCESS in 30m 31s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/545c9db8939c4d5788663465d1491233 : SUCCESS in 1h 29m 02s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":18},{"id":"3938630a5da33b75d0839a1d4162cdd6a6bd8bb3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-06-26 08:52:21.000000000","message":"Patch Set 18: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":18},{"id":"33fee0f64c9ea7e67ad51a8f4178159ee025bda8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2026-07-01 00:49:21.000000000","message":"Uploaded patch set 19: Patch Set 18 was rebased.\n\nCopied Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":19},{"id":"e8b955976bbd12bb1d94d5e81f9aed5f0fc2dd36","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2026-07-01 00:50:25.000000000","message":"Patch Set 19: Review-Priority+1 Workflow-1","accounts_in_message":[],"_revision_number":19},{"id":"4d31a47d725e7a0f82447a3eb45d6b272795df60","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-01 02:14:23.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8b2dc5df0444929a2ee7351c0034c2d\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/44ed931f27e94a22ac401404f4156d5f : SUCCESS in 1h 15m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e4c5bb2b86ca46be9425bf09ac85be1e : SUCCESS in 1h 09m 37s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0c864a89e06442728f75f04f9cbc1b30 : TIMED_OUT in 1h 22m 22s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"ff217a462db116c6fdf6ce6bee7417fe041fc273","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-01 02:56:23.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/274a156c84b847e9b498e73c48741489\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0164f5945d2c4c1ea4cf4bd5e8e3a82b : SUCCESS in 4m 21s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e460a1f7bf5a4ab5ac17e23417ab9093 : SUCCESS in 1h 36m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/502e79cc8e7f45f59145c9ea8887d2b5 : SUCCESS in 33m 40s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/acadfb0be13c4dcfa8ec5052d794af12 : SUCCESS in 1h 36m 32s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7a657435690644a9ade1dfaaad42bf25 : SUCCESS in 1h 51m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a2ecbc052ed44f03a1f8b6494da7b9f5 : SUCCESS in 31m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f4429e304cb9453293ba9421b65cf749 : SUCCESS in 1h 07m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b801d5f65a53463a820de8ec3263af50 : SUCCESS in 11m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/709245c264fc48c2907805ea182374b2 : SUCCESS in 29m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4b503f17d9734461921c63ed6d007a26 : SUCCESS in 27m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/46d531457f5c4f47b6bbac75b8297ba8 : SUCCESS in 33m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b2218ebfe014f7d9aa3334e539dbd6d : SUCCESS in 9m 22s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1f35ab66058e45a3bd75b53f8825eeab : SUCCESS in 3m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/16a4dcba28d34ad4979d11e320e330c7 : SUCCESS in 25m 18s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e610dc841ba94629ac72f1e4d8cb62f8 : SUCCESS in 1h 09m 18s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/da62c7e4e4f04bb69ce59e4572e8e7e6 : SUCCESS in 1h 14m 38s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/4a14cd05987c4d57ab8ada6dda598899 : SUCCESS in 1h 12m 55s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c14e47005c3a4e63a28d81d7070667f1 : SUCCESS in 2h 00m 41s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b9eb2c68d54f4c6588092c3caf806ea2 : FAILURE in 1h 28m 02s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2bccfd9966f24c58af93b7b928ec113d : SUCCESS in 59m 09s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/017fac80d3644a199adcde2316f6e3a9 : SUCCESS in 1h 16m 02s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c33fb3d79c4b4fcb8e212f426d015187 : SUCCESS in 1h 30m 01s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":19},{"id":"39d5eb1e5100eabb9192c5828d0cd5da2e081c4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2026-07-02 23:15:02.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":20},{"id":"7ae492b16ba36e39355d36f4d9304106977f1517","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2026-07-02 23:16:06.000000000","message":"Patch Set 20: Workflow-1","accounts_in_message":[],"_revision_number":20},{"id":"eb201cb6d2ab8a2791ded9bb9b29a226f5cb6063","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-03 00:12:54.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf9286332be24df28ac88b722ec12ea0\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/971958edc83847a0a657429e97d907da : SUCCESS in 25m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/31a7bc3e64604b0188281417067ec21c : SUCCESS in 38m 54s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3ca3c542e296443db29afc44f45998fe : SUCCESS in 55m 53s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"290646ebe594bf55260e6cf75932c23c5c175ddf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-03 03:27:20.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fc9b139cac14fe8af2ba10f66fe288b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/51e5fd01602a4f4c9f80ca68508ab9b7 : SUCCESS in 4m 28s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/62aa2c57b49b4c06a46039bc867835b0 : SUCCESS in 1h 29m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8598ea9544b2435dbe5d309e9c8c23ce : SUCCESS in 53m 08s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0464d091382c481092c95269b267d823 : SUCCESS in 1h 31m 09s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/cadf28fa1df44110af411eb9635e3075 : SUCCESS in 2h 11m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/29a94248fc3042d28f43f0465f219bee : SUCCESS in 31m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/861615fc245641a192c14d4a2b26646e : SUCCESS in 54m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/403f5b4f7d364f0a8bf7ba7594f88ee0 : SUCCESS in 9m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5db33c842422416ebc169d23e884f1d4 : SUCCESS in 44m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/56fde0512c2e4ef784ece2cfafb3bcc4 : SUCCESS in 25m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d8846a4e96604e65be78cdee2421e209 : SUCCESS in 25m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c86621a4d9b848399a758d148b2518d3 : SUCCESS in 12m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/49b8ba156faf43e78ecf765e6047a366 : SUCCESS in 3m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11637540f8f44fdf9a8b9400b0f080d2 : SUCCESS in 23m 44s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c5556acae97942a6a22d9fb2b15861e8 : SUCCESS in 57m 21s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ee90d016ba0642b289e681fdf22b3e76 : SUCCESS in 1h 11m 09s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/531498f6e40e4ac9b110dae920ba1c55 : SUCCESS in 1h 21m 48s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e561f1cc849446c9ae86ae0bcf73f8a1 : SUCCESS in 2h 05m 51s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f1ad66a7433046e2a9fad4fcd591c49b : SUCCESS in 1h 49m 27s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/da34bc41f18a474782a55785646f829f : SUCCESS in 1h 11m 11s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/94b99ced398c475ea8c5e748786d4e7d : SUCCESS in 1h 10m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d27d1afe99724fb481a1557a499df173 : SUCCESS in 1h 27m 00s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":20},{"id":"d3176d5ee13a6fbda250ffe99802b28e3e6cf443","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2026-07-08 00:04:29.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":21},{"id":"1267b6ae572d8fad418bd02f7f2368a58a42f44a","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2026-07-08 00:04:53.000000000","message":"Patch Set 21: Workflow-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":21},{"id":"4dc8695999d2b5734ec423a54aebc56a664f5fba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 01:53:11.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5296e7ac651e4e2f832bbb0755963893\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/759c581714a8436a9b62b0d1534b0490 : SUCCESS in 3m 17s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9199a0fb85594833a93caf0f19e68c61 : SUCCESS in 1h 27m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9decd891503a4ec6be867b39613d6844 : POST_FAILURE in 26m 46s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/907e3de924c14caf83904ec90a154a49 : SUCCESS in 1h 37m 11s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9b848aaca6f4458a96ca7bd16eb2a1c4 : SUCCESS in 1h 46m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/eb4db506d2f148d3b16fbb84dd4ebd44 : SUCCESS in 36m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f82046ce7a7479ab060077f26fa3973 : SUCCESS in 41m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3e3d05c812d4b0ab76a6c74b53ed928 : SUCCESS in 10m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/41c245a5165747429c0f7d8439470fa7 : SUCCESS in 20m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6537f17b4baf4301af8f3f90882d3050 : FAILURE in 17m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bd079be909f24b529c2d9cbba5d241fd : FAILURE in 19m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5a1b70fd9b04560b17971cf0f3afbbc : SUCCESS in 7m 41s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1814b7c285124c89bc7af09e7b0641ce : SUCCESS in 2m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e9e4f3d4c7f4a0dbbe2e6dcc4d8df13 : SUCCESS in 19m 31s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6ee5690243bb4be9b955a3e5f3dd7aff : SUCCESS in 1h 31m 02s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3f50bec70b7c449a84325c03837efd60 : SUCCESS in 55m 43s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/336ee5af89c443bf93809dffe725e402 : FAILURE in 32m 50s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f198a96ded624546870bb796f6f14935 : SUCCESS in 1h 11m 06s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f2f8a6130f6947329939fa2ab95a2bd7 : FAILURE in 47m 38s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/02c2c0433bd04872bf569b7acc55d08a : SUCCESS in 1h 04m 02s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/58bf56af729d442d9a5ec1d0e085dd1b : SUCCESS in 1h 10m 27s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2960b6a094cb49c4be44b4f1e1abac69 : SUCCESS in 37m 09s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":21},{"id":"3d941077069cea361ed492bebc6cfeec0e8943db","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 07:52:08.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b702bea1d63d4f5ea9fbc8f181e14397\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/de25268478b44db0905be8f2a998c5fc : SUCCESS in 21m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/30dc5e1083c54b69ae032bbfd8c100de : SUCCESS in 40m 09s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0b1da9d93ac54960b6b2393f10894c6f : SUCCESS in 46m 22s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"491e1bff2f32506cb8c16705d568060684fc1f14","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2026-07-08 14:50:27.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"9959e8c4b9b6676020084c1dd82391a96dbcc01e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 16:56:05.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/367c34b37d4a41d89c4d8797f30cffe9\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c7165ec48ad44cdb893d2944917fb566 : SUCCESS in 2m 34s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9a2ade54ddba465a851eb641b76455ab : SUCCESS in 1h 19m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aa04c73b6b72468da68393daad78abfa : SUCCESS in 1h 08m 36s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0e8692ec5620465daf2ebb92dc327404 : SUCCESS in 1h 20m 47s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1dfe1362f1ed4db8a907df5af40c65b7 : SUCCESS in 2h 02m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9cf3d1b2538349b0b1d83f3483718b94 : SUCCESS in 31m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25626ab7344f42c5aedd6854c17f4416 : SUCCESS in 1h 09m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2dc88bdd85e6452dbd0044290cc1402b : SUCCESS in 7m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e2a69cba6b4b4d27a17cef68cd19f71a : SUCCESS in 31m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1ae6fc63dc55411da7229fa17a1caf7f : SUCCESS in 26m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ccf2853e767945fea11cf58146f3966a : SUCCESS in 35m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6315733d11b14acdb553a7c267416a61 : SUCCESS in 8m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/66ce81b4e00d45a1881037e664009724 : SUCCESS in 1m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e25bc624a8c5482687d6777f885326a5 : SUCCESS in 15m 45s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f4253e876e3446618be917380b21fe03 : SUCCESS in 1h 46m 40s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2a2c0f695c00440ea864a0ebbe564188 : SUCCESS in 53m 17s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/8fe18af9307a41ba880cbdb46c76c42f : SUCCESS in 1h 04m 15s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1e5a07ce55954480a746fd38d13dbc47 : SUCCESS in 1h 16m 29s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/22feef89198841dab36d22164b095a7e : SUCCESS in 1h 43m 44s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f02dd88bcd2f4e688e37dff0636aa93e : SUCCESS in 59m 26s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/12e3005b63e148de96f1c7b01d3d953d : SUCCESS in 1h 18m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/641b0128ee934f7ab029e905be19f99b : SUCCESS in 1h 31m 25s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":21},{"id":"5f3917143d6cf37820adcbfc4826e5fdd4f1b6bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 20:00:21.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fda8d69fcebc48e4a6426761cb137712\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9cc80599cbb74e86a9e63c6090748ae4 : SUCCESS in 57m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/79fed86d72e34c038b0094dcde2dd465 : SUCCESS in 49m 24s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8189257451b84a02a015adf7b2ceb921 : SUCCESS in 53m 03s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"a2241dcc2ba6d8dc9e6c42eba358ec3116033dd8","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2026-07-09 14:31:34.000000000","message":"Patch Set 21: -Workflow Review-Priority+1","accounts_in_message":[],"_revision_number":21},{"id":"05a31f0663ddd2a081c0c7d3c08d534f25067503","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2026-08-10 23:26:49.000000000","message":"Uploaded patch set 22: Patch Set 21 was rebased.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":22},{"id":"12c3974d068e297cbd4f038a9e35ee06acc85b83","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 00:39:11.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5725de824c04aaeb46cc4318d21c5f2\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2f9173c486e245b2b1499a84c6e31ecd : SUCCESS in 1h 00m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/910c8c52b09149e39c5772d6e1f7d79d : SUCCESS in 46m 46s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3df83998c6554ec09170a06336a0b862 : SUCCESS in 1h 09m 50s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"5d31cf6355a762b88179b36caec5823fcbec8c3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 01:09:36.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad6bdb45f99d4fbdba3532e3e53a9741\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/24c5124f39864d5fb5ca182d5db7863a : SUCCESS in 2m 22s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/bc7d025ccf564e83898b8014b3929ccf : SUCCESS in 1h 22m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/744016c1f9044ad68b0e80cc75d8b2ef : SUCCESS in 56m 07s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8dca92d93b994ea1baa4ea2a925acbf7 : SUCCESS in 1h 00m 17s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/82808f4f0a7e40fe806635154fc68b06 : SUCCESS in 1h 40m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/12a999a1ea624f4d8b0d64ed80ae6c88 : SUCCESS in 32m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/034b509ea36e4bc0a2807a9288725489 : SUCCESS in 1h 14m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f26773ec1739401eb1ffcaaa708f4946 : SUCCESS in 9m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2698d1505cfe48d4b6e9fac2967bf89f : SUCCESS in 20m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9bb13e2468c04a1597ea769ed2c330d6 : SUCCESS in 21m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/752e1899fc8b4413ab97420fe81588c8 : SUCCESS in 20m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02473ce63ca944d78e6aa32585109087 : SUCCESS in 6m 48s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ee2bd6786add46c792669462a5b77694 : SUCCESS in 2m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a319f9157884159b3d8af343ef6c8dd : SUCCESS in 8m 50s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6bc5cd79c11d454780ae9e58de7504a4 : SUCCESS in 55m 12s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6ce8c423e4e94fd98250fd1baf5020e2 : SUCCESS in 1h 18m 11s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/2fc9e1f3e1ae4315b865fabfed603c3f : SUCCESS in 1h 19m 22s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/79ec0812128a4c85b8b0ba6b83f4634c : SUCCESS in 1h 13m 53s (non-voting)\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/78b4b8b6a481455fa28dbbe20267443d : SUCCESS in 41m 19s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1b2f49b0d68249788fa47a69bf91f03c : SUCCESS in 48m 34s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4665389e1c014367be954c832550332f : SUCCESS in 28m 31s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e2b109275b3f40f5aa38ab9b054523b7 : SUCCESS in 33m 45s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":22},{"id":"1f06dfacb287cc1be647f668360f7806d785eb70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2026-08-26 17:40:02.000000000","message":"Uploaded patch set 23: Patch Set 22 was rebased.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":23},{"id":"5bc3aa83b5141577e1507cd7323ca374bb5e2130","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-26 17:55:15.000000000","message":"Patch Set 23: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/a52483be2cd74547862191956e11d44c\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/59c481e35c4546efae3c8734467ceeba : SUCCESS in 7m 59s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"e4cff5bb4f2dd1a8287bdd2a226db8e22c435eab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 19:28:00.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c45222eab7d341fbb21dced34187a913\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/38dbadb7c33d4afd8f3ab0a4c5021210 : SUCCESS in 4m 11s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7d6cc09550f4419b9e843ac7e6fda0a5 : SUCCESS in 1h 17m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8e847e46020a40a0994e7da988471ea7 : SUCCESS in 56m 56s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/20a8ff02804040b9983b222d4e537779 : SUCCESS in 1h 03m 51s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/264d4a1a13954265beee13e6b839bbbb : SUCCESS in 1h 43m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/40ac7cfb60294cafa0a1809c516e7e85 : SUCCESS in 30m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7fdac402ead4747944e7a7bbbc6e0b5 : SUCCESS in 1h 05m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/401138b3e3a64303b942db668ee6068c : SUCCESS in 12m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3ba1244bc4474b918d2b1fe523a5b984 : SUCCESS in 19m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cafd39e2ea964962a70b0639448a05b5 : SUCCESS in 17m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c35bfce155614abc9a03263aae614ae0 : SUCCESS in 21m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b72a9da716cb47538a7699a8151d9950 : SUCCESS in 6m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/70112d318dd9469eaef4323f5004bc11 : SUCCESS in 2m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5cef12c654942b7bb04fe23558f91d0 : SUCCESS in 7m 36s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ae13fe3b30b84beba3cdd7d69a4332d2 : SUCCESS in 1h 35m 01s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d8d32a80bea346a68a837cb38481b75d : SUCCESS in 40m 18s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/23730e2260d4431c845b0ef1473a9822 : SUCCESS in 52m 04s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8911f0c9ba73408f8ae7abd659806d43 : SUCCESS in 52m 43s (non-voting)\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/57f45c3612a442d8822fbcc76e3e46e2 : SUCCESS in 52m 54s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f5c18f28b6754835bf5ddf139a75fc97 : SUCCESS in 45m 36s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/64237fb07fae4584867ea8114b704848 : SUCCESS in 1h 13m 46s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/968f019e62fd4ce28813d024fcbdda51 : SUCCESS in 49m 08s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":23},{"id":"2b79bc665983f87d0a7a02b0706de23373ba138c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-08-26 20:20:04.000000000","message":"Patch Set 23: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2026/Arista_Tempest_Testrefs/changes/18/982018/23 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":23},{"id":"dfc3c81aff293cbc0b1268cefad6e6848cb3b911","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 23:02:00.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5782521b736f44dd9ad1f8ee1cd30564\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7cf4022a7bcc438db38d256578fbca81 : SUCCESS in 50m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0d5946aaa4d24202b30a085fe2b25646 : SUCCESS in 1h 09m 10s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d279cf6898994bb891da6da42fce91c7 : SUCCESS in 1h 01m 04s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"c5de7de57347b231841c111f42bd1ecf6e40481d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-09-03 08:19:51.000000000","message":"Patch Set 23: Code-Review+2","accounts_in_message":[],"_revision_number":23},{"id":"3442e8ad77d289bfd51637d65259098f8e34cd32","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-09-03 15:09:07.000000000","message":"Patch Set 23:\n\n(2 comments)","accounts_in_message":[],"_revision_number":23},{"id":"73e860c031501e85c2bd82b6b257c507591971ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2026-09-03 22:40:11.000000000","message":"Patch Set 24: Published edit on patch set 23.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":24},{"id":"73e998f865f4e95dcd09250cd151e5b281f754e8","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2026-09-03 22:43:25.000000000","message":"Patch Set 24:\n\n(2 comments)","accounts_in_message":[],"_revision_number":24},{"id":"cf11f66538dd07c9042c7e1fb5d840e0467269fa","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-09-03 23:00:22.000000000","message":"Patch Set 24:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/September-2026/Arista_Tempest_Testrefs/changes/18/982018/24 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":24},{"id":"41550cb671e5025137407762e403604db810531c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 23:26:31.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d70783d655d45b4814818ee45fc1938\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/43d628fe5fcd43cb802f620ecd0d5607 : SUCCESS in 29m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/841ad362d88f4646ba2b61436d09b4f4 : SUCCESS in 43m 54s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/07528c19e28249139adf0b813e14c502 : SUCCESS in 41m 50s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"837f2f1e6b756212fbae0eda0aefd58ab4a1566f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 01:15:03.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/df0912d56e634175a4d4abb5e6b6146c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/69329367545f432bb419340551dd8f45 : SUCCESS in 2m 22s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/fc5be769760f467fb1728e3eecb4d6fd : SUCCESS in 44m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/49c3abb07e954e78acad1f68ab00508f : SUCCESS in 1h 01m 17s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/44fc9c6698c74a869523f659566e61d8 : SUCCESS in 1h 09m 22s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/01ff556d43214c30852f248f1160431d : SUCCESS in 1h 34m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d166a50384f044eba6a0dd2ad8a0fe3e : SUCCESS in 30m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e78c5ef3f8c844d6980e52d051eea8ec : SUCCESS in 56m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e218861970e4aa2adc8e154ad57a785 : SUCCESS in 10m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0876d6cd82784f44a1621d6498fd8081 : SUCCESS in 27m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/90fb01b30bd442b799533ade9b08ab4a : SUCCESS in 19m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f96df8d41f00489188399b00b3cd4b2a : SUCCESS in 22m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37a2378586644f7d8b4461cc19255e2d : SUCCESS in 7m 24s\n- requirements-check https://zuul.opendev.org/t/openstack/build/110889f50e5048eb8ddbf0aef52304be : SUCCESS in 1m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b39c78cdb644ea5a38e22c7d03d5c02 : SUCCESS in 20m 37s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2b5430bb079d43969989ebca965c0646 : TIMED_OUT in 2h 32m 05s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/13e9a1e6fa8149cb94e7069db2346132 : SUCCESS in 1h 06m 59s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/a56e568250544cccbc09b13d0c37b0ab : SUCCESS in 1h 12m 59s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/23d39b0c62d04267a575d03bffb19f7f : SUCCESS in 45m 46s (non-voting)\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3bc8a620262c437ab4559b4df99c5d5e : SUCCESS in 1h 20m 52s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/aaf19650d1f24bd68e4e361a935861d6 : SUCCESS in 33m 38s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/11e4fa1a3e8e4a32bcaa9e2a9b15d241 : SUCCESS in 50m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c3e7e9c6f942446cb0284ff303eaff3c : SUCCESS in 1h 19m 17s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":24},{"id":"6e6040462a258a702a425f73cb839ee976852b59","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-09-04 01:51:51.000000000","message":"Patch Set 24: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"c99a5b3712abdfb8d9d0b00a3f2734f1becf7730","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-09-04 02:09:53.000000000","message":"Patch Set 24:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/September-2026/Arista_Tempest_Testrefs/changes/18/982018/24 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":24},{"id":"6e3ecf08a4fa1dcf50d361b5fed20c8de3e2c1df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 02:42:48.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a701fc516a454c5e97967945e5f84e4e\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8f3e827b384942d88f4bb57c3acae901 : SUCCESS in 30m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/13428cd90e284fff85c590da147f4956 : SUCCESS in 44m 42s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7558ee57c4654da8b81e601cdd7271fd : SUCCESS in 48m 49s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"922baaf27e478c3ab1479f88d52456377fd41f80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 03:51:28.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77840be3777a45c0802a56a86fdf83f8\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7677c10afb1a48c69d12d33edfeb262d : SUCCESS in 5m 00s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/661eda983058432ba8010c5f4661aaab : SUCCESS in 49m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/28a17d1f2ed142538f60357886ca473c : SUCCESS in 59m 26s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8c746f4a82764aa796b4b37a0b0a7322 : SUCCESS in 1h 03m 37s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1f37500d2cbc41ab9bcaf10bbea97f3e : SUCCESS in 1h 44m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4215f232f32e44959ea54a2967815524 : SUCCESS in 31m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25320141c2f9416c87fcd12ab51787be : SUCCESS in 41m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf34ffd024f046a9891866f1acf9b867 : SUCCESS in 9m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4ab28b7a1ec3496db89ca5fccf5bc598 : SUCCESS in 30m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/58d7ae3213db41868f1ce9d5e386a9c4 : SUCCESS in 20m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b51eecdb8d4b4e149358afc61f33ea04 : SUCCESS in 21m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/958789c2755f48c3a84d8729494eb0f4 : SUCCESS in 11m 37s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0dbf489bcbd6481193f70abf99a31141 : SUCCESS in 2m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71ab75f31d1541a289931e31e022a0c9 : SUCCESS in 23m 39s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e11dd83fec1d461784090cb52b613f32 : SUCCESS in 1h 53m 53s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/69c818d228084cfe8cb914b28dfde6d5 : SUCCESS in 38m 16s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/05e85b2ddcf64e0aafd14df30084942e : SUCCESS in 1h 01m 32s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d5c9ece368f3436ba409aa3fde29ca71 : SUCCESS in 1h 26m 34s (non-voting)\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9780fe55cde64948b37d469a59132d19 : SUCCESS in 1h 32m 00s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0053a53d6d2b44d991a4aedcc5788967 : SUCCESS in 39m 08s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f4bb058c864c4db8855cb14b2f76a8e0 : SUCCESS in 33m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/43c9c8d16e0e4c8b9ad8c6f15b60aa46 : SUCCESS in 1h 21m 42s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":24},{"id":"c2a404ac310ac3e29b3c932365ea30e7566bd82a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-09-04 12:59:49.000000000","message":"Patch Set 24: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":24},{"id":"a723d719ca83ecce17d13253b61787dcec73ac38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 17:23:32.000000000","message":"Patch Set 24: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/neutron/+/995612 is needed.","accounts_in_message":[],"_revision_number":24},{"id":"f5e759b59c3d8b265328ca31a7df46203b36f0e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2026-09-08 18:55:55.000000000","message":"Uploaded patch set 25: Patch Set 24 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":25},{"id":"6c58f0dfccf8cb8ca5e029378d86ebea1edac91b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 19:57:16.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c5176c0c8b04555ba9f9a7f6cc62b81\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7bfdec74b4e54f34a9eded82fdce4f91 : SUCCESS in 45m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/66ce9cbb5c6243c3a9c1a70402673b49 : SUCCESS in 41m 48s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/dc8baf39d73645d3a657755ec8e7b12d : SUCCESS in 58m 18s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"34a4e9d0caf8bff1a7a5ee2e3daaceca99b32df6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-09-08 20:27:50.000000000","message":"Patch Set 25: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2026/Arista_Tempest_Testrefs/changes/18/982018/25 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":25},{"id":"924ca64c7e8c4014cfc3c7150ba1edbc64985f43","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2026-09-08 20:42:37.000000000","message":"Patch Set 25: Review-Priority+1","accounts_in_message":[],"_revision_number":25},{"id":"c2070561fad3a4e9e26d51c4fa65b3a6f74710b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 20:49:37.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d7cc80d5f9a44530b670cf423bb12fbc\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9ec056c4ba9542b9a2ff41dc168322ee : SUCCESS in 5m 31s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4759f3f31efb4d6eb6f0ce6964a75965 : SUCCESS in 1h 18m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/58622a150e7743f3a162803ba3a013a6 : SUCCESS in 48m 40s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6826bd98ad1a4b08b6dd7f1125fcec59 : SUCCESS in 1h 03m 45s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/42d5680ae1a2486a9a2a7b0d8eef7d6c : SUCCESS in 1h 50m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c70c81b17276458ab50ace7383390632 : SUCCESS in 30m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5441a4cc8deb4b66b92b83eb7960e396 : SUCCESS in 48m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e823011c4dc4e488a4f993abcea6496 : FAILURE in 10m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9131625e1ccb480eaaadff5f823e4b79 : SUCCESS in 30m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b0db90419ae24e7885883df79a8939ce : SUCCESS in 28m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d0fcf60bd10a4b68bbd904eb5f4d497c : SUCCESS in 29m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0f4f9eee0484e7fbd107b3a158b21e1 : SUCCESS in 12m 31s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b2a43cc686234ca6bee67e7769a6905a : SUCCESS in 5m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22d8f0087a48451a962dd2b67e76ae80 : SUCCESS in 24m 20s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fef581a27fc84f4cb52471a8d792e551 : SUCCESS in 1h 33m 48s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ae29cfed241845d78520b15e3e2cf6db : SUCCESS in 56m 55s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/669b12523a6a46c3b407be0f0ba312b9 : SUCCESS in 1h 25m 09s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/da823c59149f40d091b717f4eecc0e0b : SUCCESS in 1h 15m 51s (non-voting)\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9d202ac6134f4418aed533cfc1800357 : SUCCESS in 1h 09m 13s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0737f00d64f74982b9c7fa05960a3e4c : FAILURE in 30m 48s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/cbc9f77642dc4eb38bd7179c2d751e16 : SUCCESS in 1h 03m 43s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b2d6d6234c11474d967159a27d3d24c1 : SUCCESS in 1h 15m 36s\n\nWarning:\n  Comments left for invalid file neutron/tests/unit/plugins/ml2/drivers/ovn/mech_driver/ovsdb/test_ovn_client.py","accounts_in_message":[],"_revision_number":25},{"id":"b5e9a138c65ed40da390bbeed9d2303b552893b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2026-09-08 21:00:29.000000000","message":"Uploaded patch set 26: Patch Set 25 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Verified+1, Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":26},{"id":"7878d0cce5d3f55f1c4738b5b54048cd08a5ed15","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 22:51:16.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a2c3bd089a04189a4b5ac1526769e45\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c56e7d1bb60c47ada00ed77a4d5d0f2e : SUCCESS in 42m 09s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/57b7894314fb4cc49abe7b852e0f347d : SUCCESS in 41m 24s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/cdbedaaeba7349618085e37132c9168f : SUCCESS in 1h 08m 23s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"3c993c00275c2addd23adef50bd2014681cf5edd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 23:03:19.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/581415d80ef245b899cfe5db4068e82b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/11aedf5bd15c44a599dcd7bdfdab9c36 : SUCCESS in 4m 35s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4f0dec9a57b346298032ad370367a581 : SUCCESS in 50m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fd9d882efd0c483793c3bd2d85ca60bf : POST_FAILURE in 15m 45s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/02be78520b2f4d5cbaefe1fcae80e91f : SUCCESS in 1h 00m 53s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/46c9529b003e45d3b0d4d42ecec7f867 : SUCCESS in 1h 47m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4ae65e6e97264282807d37d3c9170cdb : SUCCESS in 30m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e70cf3a3fa5b492d8eb0eaf06bb35e4a : SUCCESS in 43m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/629fb2882f654447b82266b990aefd6f : SUCCESS in 8m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/903e2613c41b4a0f9a1ca1e0e8ef95bf : SUCCESS in 29m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d4c5a9ae02b041938daa421032d081b0 : SUCCESS in 17m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a2772b6643874669a328628b8109ca35 : SUCCESS in 28m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72e1b55ffda546809dfcb85d4de4ea3f : SUCCESS in 6m 13s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c0d36dd5257e4d39a5eec8ecf1e787ba : SUCCESS in 2m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33d377868cf74369b19548e69c30d3a6 : SUCCESS in 15m 08s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b7aaa7f2d5ae4ccfa454ed9d1bbee577 : SUCCESS in 1h 51m 31s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8bcac12e3d3d48afb32936ed704f936c : SUCCESS in 43m 50s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/7c58e336503749bd9fcc12715071624c : SUCCESS in 39m 07s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/39abbd859fad47ea85d28718e071552f : SUCCESS in 39m 25s (non-voting)\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/df17cc66d349425892689039c1057517 : SUCCESS in 1h 09m 10s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/13f565f7475f4500a9de7e09beb04548 : SUCCESS in 1h 02m 17s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7fca05ec5d2b48dd9e1c26758fa972f9 : SUCCESS in 35m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5c258ed790104739aed2a95f5a70edc9 : SUCCESS in 42m 02s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":26},{"id":"4489972370b5bbd9511d7504640ad63a5e748639","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2026-09-08 23:58:46.000000000","message":"Patch Set 26: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"9e225f41b34a0063b12d07504d87a7acda85df3c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-09-09 00:52:13.000000000","message":"Patch Set 26: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2026/Arista_Tempest_Testrefs/changes/18/982018/26 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":26},{"id":"756e425e1a87d2507bd27df0d9f68ba226502f1d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 01:16:28.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/193265b5f1264f3fb8a744f3f17bc335\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/65d6eb2e4be9485492fc168d401e1113 : SUCCESS in 55m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4e9ac6eb26914447a9d11a4bd2fc65f2 : SUCCESS in 43m 21s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/17070586d71241298d29f7b3e8c67fff : SUCCESS in 1h 02m 49s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"86d99676638a951c3f54feae81156bd8d0787200","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 01:48:33.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb3523a0ffce4f9e8988313b4e97f816\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4feccbf2a97c48bfaedae2932369eb5d : SUCCESS in 3m 03s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/765c259648984fa89247761bdc10b83d : SUCCESS in 50m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e1a1e250b90a46f3a8abfc97cdce9a30 : SUCCESS in 57m 21s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/95dc5a19ec1b4b4b9988b6872ecca6a3 : SUCCESS in 1h 05m 46s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/cb5be986c7ac4e07b4a97cbbc0686ee4 : SUCCESS in 1h 39m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/dabc5e5a623e45818f6a381ae96084e4 : SUCCESS in 30m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bbd8d2aee3514b6c80f41779f910535d : SUCCESS in 46m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f32d292714434f189a832abc73527f4d : SUCCESS in 5m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/30be9347350b4ca89ed2741931d8ce63 : SUCCESS in 20m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cf0443bd41eb4ddba98ad030366891f7 : SUCCESS in 26m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2e513508ec2a4922a709777fc16041df : SUCCESS in 28m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cb1cbd9a13c4b6e9c45cfa99b6b9eff : SUCCESS in 9m 24s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e7f651aabc26470eafbd83b197510475 : SUCCESS in 2m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75c3502d30b34735873aa3ef7c7fe7fb : SUCCESS in 21m 06s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4ea3ed2d4e3341b690f2b4ddd6709feb : SUCCESS in 1h 43m 00s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/46ed56a03cb84556ac84ed3e0d3344a7 : SUCCESS in 42m 52s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/3343511405144d8baa94ccdfbd1a0b4e : SUCCESS in 1h 13m 46s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/085f253358d44aa28f09dbb7745cea93 : SUCCESS in 1h 25m 25s (non-voting)\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/dab1709ed2cf4ce3831c17d94b45d695 : FAILURE in 1h 08m 51s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/25f5d80b07f74d719df9d5e9da8c505e : SUCCESS in 1h 02m 33s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/adc5bc41ed324e96b62c703f4ef64b10 : SUCCESS in 1h 06m 02s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/06847256757a4e98900de45c4d538ae0 : SUCCESS in 1h 20m 22s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":26},{"id":"375743cbf50cce964f6d09209bc6eee2621162c7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-09-09 04:05:14.000000000","message":"Patch Set 26:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2026/Arista_Tempest_Testrefs/changes/18/982018/26 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":26},{"id":"a153d635145b4f329937fd36f731bed187c99669","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-09-09 13:15:45.000000000","message":"Patch Set 26: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":26},{"id":"8d0376d8b4048617352a690d7c7e82fa0e9c2d83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 14:04:04.000000000","message":"Patch Set 26: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":26},{"id":"6b2ea4e4f5be156f37a078ebc2226a8cb30c24d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 15:51:16.000000000","message":"Patch Set 26: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/neutron/+/995612 is needed.","accounts_in_message":[],"_revision_number":26},{"id":"15b2774237f35157a96e25ae1fc59f622a80d590","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-09-09 16:44:44.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"1524641299be62e46bf1227f1778963fdfab8384","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-09-09 17:55:42.000000000","message":"Patch Set 26:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2026/Arista_Tempest_Testrefs/changes/18/982018/26 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":26},{"id":"90603f9af1f0bad75cd2a0265ada7661f4f4d8ec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 18:33:47.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb0c9460247746abbcd12258f0578444\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d79a254e0a96480ab760256af8828669 : SUCCESS in 54m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3f81d54736554d1e9ee7b4986673db62 : SUCCESS in 51m 02s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/af9b3b837be44b3bb8ad93290c3264b5 : SUCCESS in 1h 03m 18s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"1344802632211bc57910dd614604937b3fcea800","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 18:56:03.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8c9dae981244778ab70526d4d616af3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/84df846a4edb48c8bf325eae2e69c67a : SUCCESS in 2m 22s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b78010085e12422082655f4bf553cc9a : SUCCESS in 1h 21m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/316f834d1fb6482b946991be9da352a8 : SUCCESS in 1h 01m 14s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ee1ebe7039804477ac814fa4784696a1 : SUCCESS in 1h 12m 58s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8108defe160b41699be460bb6e8312d8 : SUCCESS in 1h 51m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5e1dbbea5da842eead56375f57ed218c : SUCCESS in 40m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a817b00ba1543d384e45da44c409bd2 : SUCCESS in 1h 04m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd9babcfd31c48db82b74fa3ae75cd88 : SUCCESS in 13m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4b44c207d61e4160a7a6ab4a2fa16a0c : SUCCESS in 41m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/183e6b37e4d54d0d8651f9ddaa320648 : SUCCESS in 17m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/49309cb920c14d33a8d924f78d695021 : SUCCESS in 33m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2ecc109dc2749619c8f937331b382d4 : SUCCESS in 13m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c53cc1f2883345a9b72da020e93e6a93 : SUCCESS in 2m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97c6254bd92b4097ab4da9d840660043 : SUCCESS in 25m 59s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4cdf0efb2c1c4defb372e14ad8455d23 : SUCCESS in 1h 26m 48s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2958c6f9ad724282a1c28b5d177598a3 : SUCCESS in 45m 43s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/337e6f54259242eaaafcdd3252480019 : SUCCESS in 40m 57s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/52c315a51ab145d0ab68f1c6aa5425a9 : SUCCESS in 1h 25m 05s (non-voting)\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/103bef02a08a4939bc0dfb3a3faec72b : SUCCESS in 1h 27m 29s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c8a234d4c7094754a7b7d380eb701e7e : SUCCESS in 1h 01m 35s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f9ff2a78f78c4c9bb31869b8ae1c65ee : SUCCESS in 41m 12s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/13dda60e66de4ff68c00a051cd6ae79b : SUCCESS in 1h 18m 25s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":26},{"id":"60e48c008d76fd7394fd98dadec828f17d4e00a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 19:02:27.000000000","message":"Patch Set 26: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":26},{"id":"7a230cfef03f160fa4ea5fc523d186e2e4630e0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 21:09:11.000000000","message":"Patch Set 26: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/231c2f474d6c49e9a5d26d7e06742eab\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d094c20c00524dfab15d15c711e0c63c : SUCCESS in 3m 21s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1d67594af6414f8dac82bcbeb4de68f0 : SUCCESS in 1h 33m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/859977d8170c405190179b9c535e4f75 : SUCCESS in 59m 56s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/16a58834afee4665874eb4cab410116d : SUCCESS in 1h 06m 53s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8715b226984e4e1ab4f2e5392808b823 : SUCCESS in 1h 35m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dfe1cbcffba4afa9bad7b1ba1f2c8e8 : SUCCESS in 10m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f303fe6b39c9419e8032f2fb0e6c82af : SUCCESS in 22m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c52f1ea72a934c889a5bae3f4159e2d0 : SUCCESS in 17m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8415c37bf2094932bf868bad9b2e8bdd : SUCCESS in 14m 10s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ad6f7e894eba4bfa971b491a66a64484 : SUCCESS in 3m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc6086d565d4482d8f37448b7da88940 : SUCCESS in 24m 18s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0d16340ce71c4c64abf11d8f41f94afe : SUCCESS in 1h 21m 59s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/17d170f34b22476cb13658388a68f04b : SUCCESS in 45m 13s (non-voting)\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/fb640735b984425ea42df68116fdf5b7 : SUCCESS in 38m 25s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/99458a80e94a4e40ac089f8986d9c941 : SUCCESS in 1h 23m 35s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2fc57b3ba129422f8f5833da09386f8e : SUCCESS in 59m 59s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":26},{"id":"a8e475abc158bc02d5b7c92cb658ad65e6410d59","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 21:09:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":26},{"id":"435365497b464963c889c33a857ce5a8acfe49e3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 21:12:51.000000000","message":"Patch Set 26:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a75ce8a4a534a4280084e499d805795\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86f90ed570eb488dbf78894d22563e8d : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":26}],"current_revision_number":26,"current_revision":"feb9d09fd769d118e19b4f988d95a713bfaa1b6b","revisions":{"259adb33d50a93932880855ea850bce298bacaf2":{"kind":"REWORK","_number":1,"created":"2026-03-24 23:29:11.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/18/982018/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/982018/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/982018/1"}}},"commit":{"parents":[{"commit":"e30f306882f3a27f1cb801610b841f748f91827a","subject":"[WIP][OVN] Implement multiple segments per host for routed networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e30f306882f3a27f1cb801610b841f748f91827a"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-03-24 23:21:44.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-03-24 23:21:44.000000000","tz":-300},"subject":"[WIP][OVN] OVN agent extension for segment bridges","message":"[WIP][OVN] OVN agent extension for segment bridges\n\nAn extension is defined for the OVN agent to manage the lifecycle of\nthe OVS bridges that represent network segments in routed provider\nnetworks.\n\nRelated-Bug: #2130453\nAssisted-By: claude-4-opus\nChange-Id: I8fcdf78588783f8a2bd749eb6e9bceb1550a13d6\nSigned-off-by: Miguel Lavalle \u003cmlavalle@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/259adb33d50a93932880855ea850bce298bacaf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/259adb33d50a93932880855ea850bce298bacaf2"}]},"branch":"refs/heads/master"},"14ed8733057e45bf4b9eb9837505ee5aac0bc0d6":{"kind":"REWORK","_number":2,"created":"2026-03-27 23:27:12.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/18/982018/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/982018/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/982018/2"}}},"commit":{"parents":[{"commit":"e30f306882f3a27f1cb801610b841f748f91827a","subject":"[WIP][OVN] Implement multiple segments per host for routed networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e30f306882f3a27f1cb801610b841f748f91827a"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-03-24 23:21:44.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-03-27 23:25:16.000000000","tz":-300},"subject":"[WIP][OVN] Agent extension for segment bridges","message":"[WIP][OVN] Agent extension for segment bridges\n\nAn extension is defined for the OVN agent to manage the lifecycle of\nthe OVS bridges that represent network segments in routed provider\nnetworks.\n\nRelated-Bug: #2130453\nAssisted-By: claude-4-opus\nChange-Id: I8fcdf78588783f8a2bd749eb6e9bceb1550a13d6\nSigned-off-by: Miguel Lavalle \u003cmlavalle@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/14ed8733057e45bf4b9eb9837505ee5aac0bc0d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/14ed8733057e45bf4b9eb9837505ee5aac0bc0d6"}]},"branch":"refs/heads/master"},"27ab9d9c182b55d4b6bc00a6242d6d0c7c1b5411":{"kind":"REWORK","_number":3,"created":"2026-03-29 23:16:02.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/18/982018/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/982018/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/982018/3"}}},"commit":{"parents":[{"commit":"e30f306882f3a27f1cb801610b841f748f91827a","subject":"[WIP][OVN] Implement multiple segments per host for routed networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e30f306882f3a27f1cb801610b841f748f91827a"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-03-24 23:21:44.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-03-29 23:15:17.000000000","tz":-300},"subject":"[WIP][OVN] Agent extension for segment bridges","message":"[WIP][OVN] Agent extension for segment bridges\n\nAn extension is defined for the OVN agent to manage the lifecycle of\nthe OVS bridges that represent network segments in routed provider\nnetworks.\n\nRelated-Bug: #2130453\nAssisted-By: claude-4-opus\nChange-Id: I8fcdf78588783f8a2bd749eb6e9bceb1550a13d6\nSigned-off-by: Miguel Lavalle \u003cmlavalle@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/27ab9d9c182b55d4b6bc00a6242d6d0c7c1b5411"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/27ab9d9c182b55d4b6bc00a6242d6d0c7c1b5411"}]},"branch":"refs/heads/master"},"a7f24a4ae86b5ac84ee687f5eb7e55bd84fd973f":{"kind":"REWORK","_number":4,"created":"2026-03-29 23:54:23.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/18/982018/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/982018/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/982018/4"}}},"commit":{"parents":[{"commit":"e30f306882f3a27f1cb801610b841f748f91827a","subject":"[WIP][OVN] Implement multiple segments per host for routed networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e30f306882f3a27f1cb801610b841f748f91827a"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-03-24 23:21:44.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-03-29 23:53:55.000000000","tz":-300},"subject":"[WIP][OVN] Agent extension for segment bridges","message":"[WIP][OVN] Agent extension for segment bridges\n\nAn extension is defined for the OVN agent to manage the lifecycle of\nthe OVS bridges that represent network segments in routed provider\nnetworks.\n\nRelated-Bug: #2130453\nAssisted-By: claude-4-opus\nChange-Id: I8fcdf78588783f8a2bd749eb6e9bceb1550a13d6\nSigned-off-by: Miguel Lavalle \u003cmlavalle@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a7f24a4ae86b5ac84ee687f5eb7e55bd84fd973f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a7f24a4ae86b5ac84ee687f5eb7e55bd84fd973f"}]},"branch":"refs/heads/master"},"79dfc8d5b246637dd4d1b2b94d419992f7116948":{"kind":"REWORK","_number":5,"created":"2026-03-30 00:26:36.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/18/982018/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/982018/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/982018/5"}}},"commit":{"parents":[{"commit":"e30f306882f3a27f1cb801610b841f748f91827a","subject":"[WIP][OVN] Implement multiple segments per host for routed networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e30f306882f3a27f1cb801610b841f748f91827a"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-03-24 23:21:44.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-03-30 00:25:35.000000000","tz":-300},"subject":"[WIP][OVN] Agent extension for segment bridges","message":"[WIP][OVN] Agent extension for segment bridges\n\nAn extension is defined for the OVN agent to manage the lifecycle of\nthe OVS bridges that represent network segments in routed provider\nnetworks.\n\nRelated-Bug: #2130453\nAssisted-By: claude-4-opus\nChange-Id: I8fcdf78588783f8a2bd749eb6e9bceb1550a13d6\nSigned-off-by: Miguel Lavalle \u003cmlavalle@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/79dfc8d5b246637dd4d1b2b94d419992f7116948"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/79dfc8d5b246637dd4d1b2b94d419992f7116948"}]},"branch":"refs/heads/master"},"085ec6713b4bdb4b100c59fe674e76ff8543edcf":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-04-09 14:19:47.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/18/982018/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/982018/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/982018/6"}}},"commit":{"parents":[{"commit":"a40ebd9b59ee1aa6f41241fe735da61c403e5678","subject":"[WIP][OVN] Implement multiple segments per host for routed networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a40ebd9b59ee1aa6f41241fe735da61c403e5678"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-03-24 23:21:44.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-04-09 14:18:12.000000000","tz":-300},"subject":"[WIP][OVN] Agent extension for segment bridges","message":"[WIP][OVN] Agent extension for segment bridges\n\nAn extension is defined for the OVN agent to manage the lifecycle of\nthe OVS bridges that represent network segments in routed provider\nnetworks.\n\nRelated-Bug: #2130453\nAssisted-By: claude-4-opus\nChange-Id: I8fcdf78588783f8a2bd749eb6e9bceb1550a13d6\nSigned-off-by: Miguel Lavalle \u003cmlavalle@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/085ec6713b4bdb4b100c59fe674e76ff8543edcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/085ec6713b4bdb4b100c59fe674e76ff8543edcf"}]},"branch":"refs/heads/master"},"35ef10c2c82d42d5a49d7311c333e15071d8ef52":{"kind":"REWORK","_number":7,"created":"2026-04-13 23:35:40.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/18/982018/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/982018/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/982018/7"}}},"commit":{"parents":[{"commit":"b5f51804fb5c534770363ae115e408bf76087d13","subject":"[WIP][OVN] Implement multiple segments per host for routed networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b5f51804fb5c534770363ae115e408bf76087d13"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-03-24 23:21:44.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-04-13 23:32:56.000000000","tz":-300},"subject":"[WIP][OVN] Agent extension for segment bridges","message":"[WIP][OVN] Agent extension for segment bridges\n\nAn extension is defined for the OVN agent to manage the lifecycle of\nthe OVS bridges that represent network segments in routed provider\nnetworks.\n\nRelated-Bug: #2130453\nAssisted-By: GPT-5.2-Codex\nChange-Id: I8fcdf78588783f8a2bd749eb6e9bceb1550a13d6\nSigned-off-by: Miguel Lavalle \u003cmlavalle@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/35ef10c2c82d42d5a49d7311c333e15071d8ef52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/35ef10c2c82d42d5a49d7311c333e15071d8ef52"}]},"branch":"refs/heads/master"},"a1073a043240c540164de1a5331bc9cc526c4be2":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-04-14 21:32:47.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/18/982018/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/982018/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/982018/8"}}},"commit":{"parents":[{"commit":"5dea6298001a74c1a7f1d5a6d6aa7baa880c06c7","subject":"[WIP][OVN] Implement multiple segments per host for routed networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5dea6298001a74c1a7f1d5a6d6aa7baa880c06c7"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-03-24 23:21:44.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-04-14 21:28:38.000000000","tz":-300},"subject":"[WIP][OVN] Agent extension for segment bridges","message":"[WIP][OVN] Agent extension for segment bridges\n\nAn extension is defined for the OVN agent to manage the lifecycle of\nthe OVS bridges that represent network segments in routed provider\nnetworks.\n\nRelated-Bug: #2130453\nAssisted-By: GPT-5.2-Codex\nChange-Id: I8fcdf78588783f8a2bd749eb6e9bceb1550a13d6\nSigned-off-by: Miguel Lavalle \u003cmlavalle@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a1073a043240c540164de1a5331bc9cc526c4be2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a1073a043240c540164de1a5331bc9cc526c4be2"}]},"branch":"refs/heads/master"},"b43e8606b0b117fc5b8df4ce14dc4cead9f04012":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":9,"created":"2026-04-15 23:25:02.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/18/982018/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/982018/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/982018/9"}}},"commit":{"parents":[{"commit":"3a1aab170721ed28f0f9f396865101f14047215f","subject":"[OVN] Implement multiple segments per host for routed networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3a1aab170721ed28f0f9f396865101f14047215f"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-03-24 23:21:44.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-04-15 23:24:18.000000000","tz":-300},"subject":"[OVN] Agent extension for segment bridges","message":"[OVN] Agent extension for segment bridges\n\nAn extension is defined for the OVN agent to manage the lifecycle of\nthe OVS bridges that represent network segments in routed provider\nnetworks.\n\nRelated-Bug: #2130453\nAssisted-By: GPT-5.2-Codex\nChange-Id: I8fcdf78588783f8a2bd749eb6e9bceb1550a13d6\nSigned-off-by: Miguel Lavalle \u003cmlavalle@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b43e8606b0b117fc5b8df4ce14dc4cead9f04012"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b43e8606b0b117fc5b8df4ce14dc4cead9f04012"}]},"branch":"refs/heads/master"},"8bf3c320d3c272df5d22f40796a485f2b13e146c":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2026-04-26 14:55:20.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/18/982018/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/982018/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/982018/10"}}},"commit":{"parents":[{"commit":"17de877f5aac42632815cd30d191a8c6c29db221","subject":"[OVN] Implement multiple segments per host for routed networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/17de877f5aac42632815cd30d191a8c6c29db221"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-03-24 23:21:44.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-04-26 14:54:43.000000000","tz":-300},"subject":"[OVN] Agent extension for segment bridges","message":"[OVN] Agent extension for segment bridges\n\nAn extension is defined for the OVN agent to manage the lifecycle of\nthe OVS bridges that represent network segments in routed provider\nnetworks.\n\nRelated-Bug: #2130453\nAssisted-By: GPT-5.2-Codex\nChange-Id: I8fcdf78588783f8a2bd749eb6e9bceb1550a13d6\nSigned-off-by: Miguel Lavalle \u003cmlavalle@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8bf3c320d3c272df5d22f40796a485f2b13e146c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8bf3c320d3c272df5d22f40796a485f2b13e146c"}]},"branch":"refs/heads/master"},"a215256c06f94f5cb5748fab3148d779c94f213d":{"kind":"REWORK","_number":11,"created":"2026-05-06 23:04:37.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/18/982018/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/982018/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/982018/11"}}},"commit":{"parents":[{"commit":"17de877f5aac42632815cd30d191a8c6c29db221","subject":"[OVN] Implement multiple segments per host for routed networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/17de877f5aac42632815cd30d191a8c6c29db221"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-03-24 23:21:44.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-05-06 23:03:19.000000000","tz":-300},"subject":"[OVN] Agent extension for segment bridges","message":"[OVN] Agent extension for segment bridges\n\nAn extension is defined for the OVN agent to manage the lifecycle of\nthe OVS bridges that represent network segments in routed provider\nnetworks.\n\nRelated-Bug: #2130453\nAssisted-By: GPT-5.2-Codex\nChange-Id: I8fcdf78588783f8a2bd749eb6e9bceb1550a13d6\nSigned-off-by: Miguel Lavalle \u003cmlavalle@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a215256c06f94f5cb5748fab3148d779c94f213d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a215256c06f94f5cb5748fab3148d779c94f213d"}]},"branch":"refs/heads/master"},"5f8b19b6a24673eab27838905b619e3f8d52995f":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-05-18 15:37:08.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/18/982018/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/982018/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/982018/12"}}},"commit":{"parents":[{"commit":"a35c050d2d9d96f20c8b6e60598e4868bb00e329","subject":"[OVN] Implement multiple segments per host for routed networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a35c050d2d9d96f20c8b6e60598e4868bb00e329"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-03-24 23:21:44.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-05-18 15:35:59.000000000","tz":-300},"subject":"[OVN] Agent extension for segment bridges","message":"[OVN] Agent extension for segment bridges\n\nAn extension is defined for the OVN agent to manage the lifecycle of\nthe OVS bridges that represent network segments in routed provider\nnetworks.\n\nRelated-Bug: #2130453\nAssisted-By: GPT-5.2-Codex\nChange-Id: I8fcdf78588783f8a2bd749eb6e9bceb1550a13d6\nSigned-off-by: Miguel Lavalle \u003cmlavalle@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5f8b19b6a24673eab27838905b619e3f8d52995f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5f8b19b6a24673eab27838905b619e3f8d52995f"}]},"branch":"refs/heads/master"},"aec6268ada3ce2edd86ef81432af421177f454af":{"kind":"REWORK","_number":13,"created":"2026-05-28 23:52:10.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/18/982018/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/982018/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/982018/13"}}},"commit":{"parents":[{"commit":"8678bb187607e2873623cdef638c8931808a0c80","subject":"[OVN] Implement multiple segments per host for routed networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8678bb187607e2873623cdef638c8931808a0c80"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-03-24 23:21:44.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-05-28 23:50:05.000000000","tz":-300},"subject":"[OVN] Agent extension for segment bridges","message":"[OVN] Agent extension for segment bridges\n\nAn extension is defined for the OVN agent to manage the lifecycle of\nthe OVS bridges that represent network segments in routed provider\nnetworks.\n\nRelated-Bug: #2130453\nAssisted-By: GPT-5.2-Codex\nChange-Id: I8fcdf78588783f8a2bd749eb6e9bceb1550a13d6\nSigned-off-by: Miguel Lavalle \u003cmlavalle@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aec6268ada3ce2edd86ef81432af421177f454af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aec6268ada3ce2edd86ef81432af421177f454af"}]},"branch":"refs/heads/master"},"f1d6eda5f1138bcb349a0e24ea038a69333993e7":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2026-05-29 21:55:27.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/18/982018/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/982018/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/982018/14"}}},"commit":{"parents":[{"commit":"21371e60e27532464b82536c0df5bcd31c274654","subject":"[OVN] Implement multiple segments per host for routed networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/21371e60e27532464b82536c0df5bcd31c274654"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-03-24 23:21:44.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-05-29 21:54:22.000000000","tz":-300},"subject":"[OVN] Agent extension for segment bridges","message":"[OVN] Agent extension for segment bridges\n\nAn extension is defined for the OVN agent to manage the lifecycle of\nthe OVS bridges that represent network segments in routed provider\nnetworks.\n\nRelated-Bug: #2130453\nAssisted-By: GPT-5.2-Codex\nChange-Id: I8fcdf78588783f8a2bd749eb6e9bceb1550a13d6\nSigned-off-by: Miguel Lavalle \u003cmlavalle@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f1d6eda5f1138bcb349a0e24ea038a69333993e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f1d6eda5f1138bcb349a0e24ea038a69333993e7"}]},"branch":"refs/heads/master"},"bc55a177870c87e1dde0f8db836297282cef8aaf":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2026-06-02 14:05:31.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/18/982018/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/982018/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/982018/15"}}},"commit":{"parents":[{"commit":"ede8b0e6df49937354a4276d1269b07150f6873e","subject":"[OVN] Implement multiple segments per host for routed networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ede8b0e6df49937354a4276d1269b07150f6873e"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-03-24 23:21:44.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-06-02 14:03:21.000000000","tz":-300},"subject":"[OVN] Agent extension for segment bridges","message":"[OVN] Agent extension for segment bridges\n\nAn extension is defined for the OVN agent to manage the lifecycle of\nthe OVS bridges that represent network segments in routed provider\nnetworks.\n\nRelated-Bug: #2130453\nAssisted-By: GPT-5.2-Codex\nChange-Id: I8fcdf78588783f8a2bd749eb6e9bceb1550a13d6\nSigned-off-by: Miguel Lavalle \u003cmlavalle@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bc55a177870c87e1dde0f8db836297282cef8aaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bc55a177870c87e1dde0f8db836297282cef8aaf"}]},"branch":"refs/heads/master"},"880ad0cabdaa98a5297f93185d2a7c42be9695ef":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2026-06-05 21:13:04.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/18/982018/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/982018/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/982018/16"}}},"commit":{"parents":[{"commit":"ea176019017b4672b756b0fd3304fd04dbe4ad85","subject":"[OVN] Implement multiple segments per host for routed networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea176019017b4672b756b0fd3304fd04dbe4ad85"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-03-24 23:21:44.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-06-05 21:12:04.000000000","tz":-300},"subject":"[OVN] Agent extension for segment bridges","message":"[OVN] Agent extension for segment bridges\n\nAn extension is defined for the OVN agent to manage the lifecycle of\nthe OVS bridges that represent network segments in routed provider\nnetworks.\n\nRelated-Bug: #2130453\nAssisted-By: GPT-5.2-Codex\nChange-Id: I8fcdf78588783f8a2bd749eb6e9bceb1550a13d6\nSigned-off-by: Miguel Lavalle \u003cmlavalle@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/880ad0cabdaa98a5297f93185d2a7c42be9695ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/880ad0cabdaa98a5297f93185d2a7c42be9695ef"}]},"branch":"refs/heads/master"},"b8f4d1be1ca3be3f58e2f50bba38eb6029417e12":{"kind":"REWORK","_number":17,"created":"2026-06-15 23:31:45.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/18/982018/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/982018/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/982018/17"}}},"commit":{"parents":[{"commit":"9960188061901088660e0667806d88ca6d860554","subject":"[OVN] Implement multiple segments per host for routed networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9960188061901088660e0667806d88ca6d860554"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-03-24 23:21:44.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-06-15 23:29:34.000000000","tz":-300},"subject":"[OVN] Agent extension for segment bridges","message":"[OVN] Agent extension for segment bridges\n\nAn extension is defined for the OVN agent to manage the lifecycle of\nthe OVS bridges that represent network segments in routed provider\nnetworks.\n\nRelated-Bug: #2130453\nAssisted-By: GPT-5.2-Codex\nChange-Id: I8fcdf78588783f8a2bd749eb6e9bceb1550a13d6\nSigned-off-by: Miguel Lavalle \u003cmlavalle@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b8f4d1be1ca3be3f58e2f50bba38eb6029417e12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b8f4d1be1ca3be3f58e2f50bba38eb6029417e12"}]},"branch":"refs/heads/master"},"b9b00ffd4ede4724ab7466d8714300d50216ad1e":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2026-06-24 22:19:02.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/18/982018/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/982018/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/982018/18"}}},"commit":{"parents":[{"commit":"0fcb370d8920500225325dcfa075a933e6711f0f","subject":"[OVN] Implement multiple segments per host for routed networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0fcb370d8920500225325dcfa075a933e6711f0f"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-03-24 23:21:44.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-06-24 22:17:03.000000000","tz":-300},"subject":"[OVN] Agent extension for segment bridges","message":"[OVN] Agent extension for segment bridges\n\nAn extension is defined for the OVN agent to manage the lifecycle of\nthe OVS bridges that represent network segments in routed provider\nnetworks.\n\nRelated-Bug: #2130453\nAssisted-By: GPT-5.2-Codex\nChange-Id: I8fcdf78588783f8a2bd749eb6e9bceb1550a13d6\nSigned-off-by: Miguel Lavalle \u003cmlavalle@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b9b00ffd4ede4724ab7466d8714300d50216ad1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b9b00ffd4ede4724ab7466d8714300d50216ad1e"}]},"branch":"refs/heads/master"},"7d194fce1500f3960a2052a3d1f570e7c10bbd80":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2026-07-01 00:49:21.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/18/982018/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/982018/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/982018/19"}}},"commit":{"parents":[{"commit":"194b4b70eb87c878db9a5cc26bb8cc306445d1ff","subject":"[OVN] Add segment maintenance tasks for routed networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/194b4b70eb87c878db9a5cc26bb8cc306445d1ff"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-03-24 23:21:44.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-07-01 00:47:15.000000000","tz":-300},"subject":"[OVN] Agent extension for segment bridges","message":"[OVN] Agent extension for segment bridges\n\nAn extension is defined for the OVN agent to manage the lifecycle of\nthe OVS bridges that represent network segments in routed provider\nnetworks.\n\nRelated-Bug: #2130453\nAssisted-By: GPT-5.2-Codex\nChange-Id: I8fcdf78588783f8a2bd749eb6e9bceb1550a13d6\nSigned-off-by: Miguel Lavalle \u003cmlavalle@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7d194fce1500f3960a2052a3d1f570e7c10bbd80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7d194fce1500f3960a2052a3d1f570e7c10bbd80"}]},"branch":"refs/heads/master"},"232ccec0089258a63e3692f738912f4a761d473a":{"kind":"REWORK","_number":20,"created":"2026-07-02 23:15:02.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/18/982018/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/982018/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/982018/20"}}},"commit":{"parents":[{"commit":"194b4b70eb87c878db9a5cc26bb8cc306445d1ff","subject":"[OVN] Add segment maintenance tasks for routed networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/194b4b70eb87c878db9a5cc26bb8cc306445d1ff"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-03-24 23:21:44.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-07-02 23:12:36.000000000","tz":-300},"subject":"[OVN] Agent extension for segment bridges","message":"[OVN] Agent extension for segment bridges\n\nAn extension is defined for the OVN agent to manage the lifecycle of\nthe OVS bridges that represent network segments in routed provider\nnetworks.\n\nRelated-Bug: #2130453\nAssisted-By: GPT-5.2-Codex\nChange-Id: I8fcdf78588783f8a2bd749eb6e9bceb1550a13d6\nSigned-off-by: Miguel Lavalle \u003cmlavalle@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/232ccec0089258a63e3692f738912f4a761d473a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/232ccec0089258a63e3692f738912f4a761d473a"}]},"branch":"refs/heads/master"},"a0eaa53622626f16cc0fc9342d53fb37e1ce8169":{"kind":"REWORK","_number":21,"created":"2026-07-08 00:04:29.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/18/982018/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/982018/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/982018/21"}}},"commit":{"parents":[{"commit":"6302a207d4e6d51de24bd3188e385c0281996de5","subject":"[OVN] Add segment maintenance tasks for routed networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6302a207d4e6d51de24bd3188e385c0281996de5"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-03-24 23:21:44.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-07-08 00:02:11.000000000","tz":-300},"subject":"[OVN] Agent extension for segment bridges","message":"[OVN] Agent extension for segment bridges\n\nAn extension is defined for the OVN agent to manage the lifecycle of\nthe OVS bridges that represent network segments in routed provider\nnetworks.\n\nRelated-Bug: #2130453\nAssisted-By: GPT-5.2-Codex\nChange-Id: I8fcdf78588783f8a2bd749eb6e9bceb1550a13d6\nSigned-off-by: Miguel Lavalle \u003cmlavalle@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a0eaa53622626f16cc0fc9342d53fb37e1ce8169"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a0eaa53622626f16cc0fc9342d53fb37e1ce8169"}]},"branch":"refs/heads/master"},"bf5141df93e018dca51817939241a3be88808008":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2026-08-10 23:26:49.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/18/982018/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/982018/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/982018/22"}}},"commit":{"parents":[{"commit":"5631c6e93ee5bdf5cba29c1624bf9ed967cb0d12","subject":"[OVN] Add segment maintenance tasks for routed networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5631c6e93ee5bdf5cba29c1624bf9ed967cb0d12"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-03-24 23:21:44.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-08-10 23:24:47.000000000","tz":-300},"subject":"[OVN] Agent extension for segment bridges","message":"[OVN] Agent extension for segment bridges\n\nAn extension is defined for the OVN agent to manage the lifecycle of\nthe OVS bridges that represent network segments in routed provider\nnetworks.\n\nRelated-Bug: #2130453\nAssisted-By: GPT-5.2-Codex\nChange-Id: I8fcdf78588783f8a2bd749eb6e9bceb1550a13d6\nSigned-off-by: Miguel Lavalle \u003cmlavalle@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bf5141df93e018dca51817939241a3be88808008"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bf5141df93e018dca51817939241a3be88808008"}]},"branch":"refs/heads/master"},"6a293059b369ef3adbecdee6fc955c9b9ee5febd":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2026-08-26 17:40:02.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/18/982018/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/982018/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/982018/23"}}},"commit":{"parents":[{"commit":"3ddee7e7fa8027df95c3b982668c3a6a1d299a66","subject":"[OVN] Add segment maintenance tasks for routed networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3ddee7e7fa8027df95c3b982668c3a6a1d299a66"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-03-24 23:21:44.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-08-26 17:38:53.000000000","tz":-300},"subject":"[OVN] Agent extension for segment bridges","message":"[OVN] Agent extension for segment bridges\n\nAn extension is defined for the OVN agent to manage the lifecycle of\nthe OVS bridges that represent network segments in routed provider\nnetworks.\n\nRelated-Bug: #2130453\nAssisted-By: GPT-5.2-Codex\nChange-Id: I8fcdf78588783f8a2bd749eb6e9bceb1550a13d6\nSigned-off-by: Miguel Lavalle \u003cmlavalle@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a293059b369ef3adbecdee6fc955c9b9ee5febd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a293059b369ef3adbecdee6fc955c9b9ee5febd"}]},"branch":"refs/heads/master"},"91772ae47cd4925055f7064487b27d0386241297":{"kind":"REWORK","_number":24,"created":"2026-09-03 22:40:11.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/18/982018/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/982018/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/982018/24"}}},"commit":{"parents":[{"commit":"3ddee7e7fa8027df95c3b982668c3a6a1d299a66","subject":"[OVN] Add segment maintenance tasks for routed networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3ddee7e7fa8027df95c3b982668c3a6a1d299a66"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-03-24 23:21:44.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-09-03 22:39:48.000000000","tz":0},"subject":"[OVN] Agent extension for segment bridges","message":"[OVN] Agent extension for segment bridges\n\nAn extension is defined for the OVN agent to manage the lifecycle of\nthe OVS bridges that represent network segments in routed provider\nnetworks.\n\nRelated-Bug: #2130453\nAssisted-By: GPT-5.2-Codex\nChange-Id: I8fcdf78588783f8a2bd749eb6e9bceb1550a13d6\nSigned-off-by: Miguel Lavalle \u003cmlavalle@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/91772ae47cd4925055f7064487b27d0386241297"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/91772ae47cd4925055f7064487b27d0386241297"}]},"branch":"refs/heads/master"},"2805b58ba29101cfa09e194498e1b6392b96f637":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2026-09-08 18:55:55.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/18/982018/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/982018/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/982018/25"}}},"commit":{"parents":[{"commit":"339332bfb5e87c073727968fb9de353ac4e87552","subject":"[OVN] Add segment maintenance tasks for routed networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/339332bfb5e87c073727968fb9de353ac4e87552"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-03-24 23:21:44.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-09-08 18:54:25.000000000","tz":-300},"subject":"[OVN] Agent extension for segment bridges","message":"[OVN] Agent extension for segment bridges\n\nAn extension is defined for the OVN agent to manage the lifecycle of\nthe OVS bridges that represent network segments in routed provider\nnetworks.\n\nRelated-Bug: #2130453\nAssisted-By: GPT-5.2-Codex\nChange-Id: I8fcdf78588783f8a2bd749eb6e9bceb1550a13d6\nSigned-off-by: Miguel Lavalle \u003cmlavalle@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2805b58ba29101cfa09e194498e1b6392b96f637"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2805b58ba29101cfa09e194498e1b6392b96f637"}]},"branch":"refs/heads/master"},"feb9d09fd769d118e19b4f988d95a713bfaa1b6b":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2026-09-08 21:00:29.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/18/982018/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/982018/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/982018/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/982018/26"}}},"commit":{"parents":[{"commit":"cff19cae21a4a9c5aeaf7c61f30a5c72eb1276fe","subject":"[OVN] Add segment maintenance tasks for routed networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cff19cae21a4a9c5aeaf7c61f30a5c72eb1276fe"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-03-24 23:21:44.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2026-09-08 20:59:03.000000000","tz":-300},"subject":"[OVN] Agent extension for segment bridges","message":"[OVN] Agent extension for segment bridges\n\nAn extension is defined for the OVN agent to manage the lifecycle of\nthe OVS bridges that represent network segments in routed provider\nnetworks.\n\nRelated-Bug: #2130453\nAssisted-By: GPT-5.2-Codex\nChange-Id: I8fcdf78588783f8a2bd749eb6e9bceb1550a13d6\nSigned-off-by: Miguel Lavalle \u003cmlavalle@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/feb9d09fd769d118e19b4f988d95a713bfaa1b6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/feb9d09fd769d118e19b4f988d95a713bfaa1b6b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
