)]}'
{"id":"openstack%2Fneutron~982245","triplet_id":"openstack%2Fneutron~master~Ida53af97c600f0b9a2fe8b016624908855112dff","project":"openstack/neutron","branch":"master","topic":"get_lsp_ips","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-04-01 11:09:00.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-03-30 15:57:16.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ida53af97c600f0b9a2fe8b016624908855112dff","subject":"Create a method to retrieve IPs from ``Logical_Switch_Port``","status":"MERGED","created":"2026-03-26 14:35:03.000000000","updated":"2026-04-01 11:10:40.000000000","submitted":"2026-04-01 11:09:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"982245-get_lsp_ips","meta_rev_id":"41b522ce63467701d297cabe7b046b365ad07a62","_number":982245,"virtual_id_number":982245,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-01 11:09:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2026-04-01 08:54:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2026-03-30 15:57:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":1,"date":"2026-04-01 08:54:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-26 16:35:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-26 22:22:49.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-04-01 08:54:14.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"3478da624424c7b15065c18bad20680a5e131c87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-26 14:35:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d8426ac4ab5efe4a9c85310a9667dfd5c60dc3fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 16:35:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82c1183ab8154e84b278a9f556623d94\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4c3a0074053b4117a27b4c5d84a907fc : SUCCESS in 1h 24m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6aa94f9398c8442fbbe89fb576cda2f3 : SUCCESS in 46m 14s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/aa59b89cde1540cc91513e9ff78b5d14 : SUCCESS in 1h 59m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/551f37895111475792f39e11d58589e1 : SUCCESS in 34m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0199f38efc7a40cfa00baf300eadd193 : SUCCESS in 1h 04m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e8476ab714e4239a676b2d0f2eac7da : SUCCESS in 26m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b9d73e1905834482b69b8f8b51ab3f18 : SUCCESS in 36m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bba480fa44c245709e03a6bf3af5f3cb : SUCCESS in 25m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/468382f4e19545848a8c1a98bf302a0f : SUCCESS in 28m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ff68dd0e1d944698f2ebd7795d90419 : SUCCESS in 12m 50s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/cd345f6892ca4cd59a8ffb9e12816873 : SUCCESS in 1h 20m 00s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/95cbef02cbde4e2cb46a6d0084dfdbba : SUCCESS in 1h 16m 28s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/3fe2d0c322df4ccba26062550ecfa2e9 : SUCCESS in 1h 21m 47s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/dc1ab639029642898b246396f16091eb : SUCCESS in 55m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/21b9ca8912e04e59a6ce5dad3b85b36e : SUCCESS in 1h 06m 17s","accounts_in_message":[],"_revision_number":1},{"id":"5c45105c89c1620c86c8ced0cea82b5d996783f3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 19:09:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fa4b16245ed4d1392a344a2d66f66cc\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9383e130b54a424581e81bd330b87be0 : SUCCESS in 24m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8d882c42cbc64ecaaf0dc82aa7d27cf3 : SUCCESS in 21m 38s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/169255b831b24f23858799ca62ad5bfe : SUCCESS in 27m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"712b12db148c498d76d8a7686a4fa6b83e74fbe4","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-03-26 22:22:49.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3375882096693adcce79898ee05543a4785f4b49","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-30 13:34:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e0ac69897e6f0e4f254f012146f4692930c16d0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-30 13:36:23.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"d5271f81ec03ad60a23b7b9069e6ce43eec515ba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 14:30:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f1950ca301347fe8a938fba044124d1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/58456b85143447abb0ff3ad3a282ee5b : SUCCESS in 34m 45s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2d0867e602a1434da79b2fab431c0dc9 : SUCCESS in 34m 52s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a89f248cecb34b958862fcaeff34a8ed : SUCCESS in 30m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9145d4b10435528c171403634b4d9364077575bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 15:32:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/527cfa580cab470e90ff5bbd9935a139\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3c8ab70152ed4c1cbea128003cc0f41c : SUCCESS in 1h 25m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/133cb8e42b9547aca2ce40aed1e7b382 : SUCCESS in 1h 00m 21s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2b15eb904d5b47babd739d45103d9745 : SUCCESS in 1h 54m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/608792ad4ef647e1a53f8130d7772efe : SUCCESS in 34m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d113e09567db40968d0180a3e7de815a : SUCCESS in 45m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/860081a3586b4d2ba6873bf495fffc27 : SUCCESS in 13m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/895c4c5cd7344529904228b0a6556fd3 : SUCCESS in 35m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3ef94f641bd84d4db59956459c608875 : SUCCESS in 22m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/93a3a1c08a4940dca42a679c9c3a15a5 : SUCCESS in 26m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14f0d8c0a8ae4baba7c57b128fad2f62 : SUCCESS in 7m 37s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/9aa52cc50f7646f4bc808ffd3e212735 : SUCCESS in 40m 54s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/78cfe777cef148f4bb08b531f4ead5d6 : SUCCESS in 43m 39s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/9b681779c0bf49dead010f966825e866 : SUCCESS in 58m 04s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/beb24618e21f4170a019bbc78bb4401a : SUCCESS in 1h 06m 35s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8fe38b42162e40c8800475577cd69c79 : SUCCESS in 1h 28m 32s","accounts_in_message":[],"_revision_number":2},{"id":"ec6a894af2cb0ec65f497af78199392b5443ef42","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-03-30 15:57:16.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2133daac03022e395aad1df95e67ecb6d9a480ea","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-04-01 08:54:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"379683421d480cd92077196251afae81ff448807","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 08:54:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"38302e655d16a3e131b84c33c4dde3e7988544ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 11:09:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e4c864de3c14d9889c0ce76bbb48088\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d124d0244a2d4d07967e534bd2ab5fb7 : SUCCESS in 1h 01m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/989c4fee49124b4186154f33830265fc : SUCCESS in 1h 00m 20s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/568e41608ab24f50b2d2cd312d4a6bf5 : SUCCESS in 1h 50m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39fcacc16ed84a2d9fe9b42f0bad4a3c : SUCCESS in 12m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7791286aae534310bee6e4b9b1b1c567 : SUCCESS in 36m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/47c6bb3e7c2d4985934b0f8d4e35889a : SUCCESS in 15m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/477d77ae10624b0ab72122f6cacb3900 : SUCCESS in 11m 44s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/84898b6469fa4dc6a094a3dd37529328 : SUCCESS in 1h 13m 09s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d5dc5919368a447a8b4dc2059335861c : SUCCESS in 1h 15m 18s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/dde2f03fde7147288e2c745c7e4c7da8 : SUCCESS in 1h 02m 05s","accounts_in_message":[],"_revision_number":2},{"id":"ae86b620b9200618a182d1beab26c632fba427aa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 11:09:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"41b522ce63467701d297cabe7b046b365ad07a62","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 11:10:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b527d1450314ee59b70d2065308ece9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7bd82bd108742ceaa37822150c207c1 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"420c96c31a3624550cf0d2c457cc6a6508d47e87","revisions":{"de2dc9694a693aa38ea180390d93f6a244c2684c":{"kind":"REWORK","_number":1,"created":"2026-03-26 14:35:03.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/45/982245/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/982245/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/982245/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/982245/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/982245/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/982245/1"}}},"commit":{"parents":[{"commit":"bef2d32664bcab3c72bd1af5c441fbe101bab30c","subject":"Merge \"Run mypy from tox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bef2d32664bcab3c72bd1af5c441fbe101bab30c"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-03-26 14:33:50.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-03-26 14:34:52.000000000","tz":60},"subject":"Unify ``Logical_Switch_Port`` IP retrieval","message":"Unify ``Logical_Switch_Port`` IP retrieval\n\nThis patch implements a unified method to retrieve the IP addresses\nfrom a ``Logical_Switch_Port``.\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: Ida53af97c600f0b9a2fe8b016624908855112dff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/de2dc9694a693aa38ea180390d93f6a244c2684c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/de2dc9694a693aa38ea180390d93f6a244c2684c"}]},"branch":"refs/heads/master"},"420c96c31a3624550cf0d2c457cc6a6508d47e87":{"kind":"REWORK","_number":2,"created":"2026-03-30 13:36:23.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/45/982245/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/982245/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/982245/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/982245/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/982245/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/982245/2"}}},"commit":{"parents":[{"commit":"bef2d32664bcab3c72bd1af5c441fbe101bab30c","subject":"Merge \"Run mypy from tox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bef2d32664bcab3c72bd1af5c441fbe101bab30c"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-03-26 14:33:50.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-03-30 13:35:07.000000000","tz":120},"subject":"Create a method to retrieve IPs from ``Logical_Switch_Port``","message":"Create a method to retrieve IPs from ``Logical_Switch_Port``\n\nThis patch implements a unified method to retrieve the IP addresses\nfrom a ``Logical_Switch_Port``.\n\nThis patch also removes ``get_ovn_port_addresses`` method, no longer\nused.\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: Ida53af97c600f0b9a2fe8b016624908855112dff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/420c96c31a3624550cf0d2c457cc6a6508d47e87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/420c96c31a3624550cf0d2c457cc6a6508d47e87"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
