)]}'
{"id":"openstack%2Fneutron~982364","triplet_id":"openstack%2Fneutron~stable%2F2026.1~I69e9fc00bfc58fb86675f72f94d10d014f25cd3a","project":"openstack/neutron","branch":"stable/2026.1","topic":"fix-ovn-chassis-name-with-hyphen-stable/2026.1","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-03-30 07:51:20.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"30534":{"account":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"last_update":"2026-03-31 11:53:22.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-03-31 11:53:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I69e9fc00bfc58fb86675f72f94d10d014f25cd3a","subject":"Update OVN doc about \u0027ovn-nbctl\u0027 command usage regarding the Gateway_Chassis registers","status":"MERGED","created":"2026-03-27 07:45:36.000000000","updated":"2026-03-31 11:55:43.000000000","submitted":"2026-03-31 11:53:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"982364-fix-ovn-chassis-name-with-hyphen-stable/2026.1","cherry_pick_of_change":979705,"cherry_pick_of_patch_set":6,"meta_rev_id":"e954b25884b25c1dac5b6fa367609aba4d43f22d","_number":982364,"virtual_id_number":982364,"owner":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-31 11:53:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-03-30 07:51:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-03-31 09:39:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-03-31 09:39:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-27 08:13:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-27 09:58:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-27 12:23:47.000000000","updated_by":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-03-27 12:23:47.000000000","updated_by":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-03-31 09:39:16.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"d3f05600b2f3b7b1bdc0f06c6217f22aabb3600d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"date":"2026-03-27 07:45:36.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"45a047f7b66cedf2ead7e3249b996d9d0aada91e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 08:13:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fa4a284d5a7484d893001995ffa1403\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/810a817dcd184ff08f82b0620a4f4b23 : SUCCESS in 23m 47s\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/be56eca168c84979909c580a7bfc7cbf : SUCCESS in 24m 34s\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/21052d5014134fcfa0dab13b78cab8a0 : SUCCESS in 26m 07s","accounts_in_message":[],"_revision_number":1},{"id":"66df8f9085aabb3bceb64dccdb782fa96148f3e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 09:58:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9e89155773be4028b34dc7b658e5bb52\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3031d2c63a0a4138914a35a518a021fd : SUCCESS in 1h 26m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dceb5a07fef0459e8696383ffe9a37c1 : FAILURE in 1h 11m 00s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/16f3f71d99c94ffcb58094f01d48013c : SUCCESS in 2h 12m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/69b6c897d1784ecf9e4fc800a7e5179c : SUCCESS in 1h 07m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3d8628b6fdd48f09253f2fc34fa85b0 : SUCCESS in 16m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/662a9f7363e4411d809d0fb89c983997 : SUCCESS in 24m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b58e3f3f60144f8eb7d033f8f07ecf60 : SUCCESS in 19m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c146c82c8fbd421d8d96238e734f7742 : SUCCESS in 36m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d696622dd3645b4a015196a7c509bf2 : SUCCESS in 13m 11s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ac2be6ad5377497f980d1fd7f3e72a0a : SUCCESS in 42m 20s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0c87984ee9414f3ea4137cead852e43d : SUCCESS in 1h 20m 16s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d1936b1d04014c72b5a7132fe5aa66dc : SUCCESS in 1h 05m 22s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6bee4bf3859a4ba68cbe508096ab382f : SUCCESS in 1h 12m 52s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/de8568d447d34aa9b0db2c50c1b3f68c : SUCCESS in 1h 27m 10s","accounts_in_message":[],"_revision_number":1},{"id":"f81401bf115f9128384c3ab947f2585ff6f62c65","author":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"date":"2026-03-27 10:08:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bf6fdb6c984cd824f7a25c4a3ba49705ef51255c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 11:00:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad858a135f004dc2987033d731a1cecd\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0f5bb6fc1e3945cfa78f674ac26b767d : SUCCESS in 29m 18s\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9b28d6d0d192426886a525369104caea : SUCCESS in 27m 23s\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/28a2f3e250a84b6ab097df6b458ec989 : SUCCESS in 31m 01s","accounts_in_message":[],"_revision_number":1},{"id":"15b39ec94fa5752628d4ab7f7e9d25294f895353","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 12:09:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/616329fc18cc49748ce86ed67af60e96\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/413c096992ec40b3835e830660ad1b83 : SUCCESS in 1h 30m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/32daccce5e564fbc9c5eb4d5b7306d62 : SUCCESS in 33m 49s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/fb9de500fe654e0ea372c2914e681278 : SUCCESS in 1h 59m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/796d2de0c52142c380f4b7c4fc1d55c6 : SUCCESS in 38m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbdfeccd00734396bba627b6576f011d : SUCCESS in 25m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1f70d51ff3624e57aa7ee71dc592111b : SUCCESS in 20m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bf93db340a0f40f3adfafe8274b74bd5 : SUCCESS in 15m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0e6e251f36e346f0aa1f8b1baee8f7f4 : SUCCESS in 32m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02cacbc3e5e74400a91521952e3f428e : SUCCESS in 12m 44s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/fec27310f1794ca6a72abe44110a65c8 : SUCCESS in 1h 14m 04s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/603b846757754b1d97a2775c07b0ce6d : SUCCESS in 1h 19m 58s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/4435b3ee0af34646b3798076494216a0 : SUCCESS in 1h 11m 19s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/354fa7944ddd4bb0bd874fab9d5a300c : SUCCESS in 1h 04m 43s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d9cf00808dab41ab9135ff62ec3a4262 : SUCCESS in 1h 31m 02s","accounts_in_message":[],"_revision_number":1},{"id":"a5d98a8aa63385765b066125227554413092838c","author":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"date":"2026-03-27 12:23:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"385f34130080651b377f70da0b2e583257f453d4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-30 07:51:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1c98ddc25d45aba5b690a1d9449d129ce94e4435","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2026-03-31 09:39:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8f0f037575b8f4fc65bc5d94acf3fe901e4fdcaa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 09:39:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cc2812584b9cdbe824de03280a8174f64117a37d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 11:53:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d65244814eba4d27adaf2b93efffcc41\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/fe419b6ecdfa45b291116ae0c43b0047 : SUCCESS in 1h 23m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba7935e56e344e1aa211f20e9bfd3a46 : SUCCESS in 1h 05m 28s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/91e285b49c9243ec807b12cfa1ede239 : SUCCESS in 1h 58m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f786daa002be4f9babb8c3accbac9387 : SUCCESS in 11m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e852bf8da6f949948138ae0288d181b6 : SUCCESS in 24m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e89c59e0475946a994fe15d6edc0b4ef : SUCCESS in 18m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96769f4d6fbf4ded8d0384f8c97610ec : SUCCESS in 12m 09s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/025914f9781344908d4a91d34983304d : SUCCESS in 49m 02s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/440c2e9dfbf24c648bab3aa00b345b52 : SUCCESS in 1h 21m 17s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f29e71afd90148f2b19f6537f54073da : SUCCESS in 50m 14s","accounts_in_message":[],"_revision_number":1},{"id":"d26507cdce0b46c81f783e056f3a82ce01f16666","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 11:53:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e954b25884b25c1dac5b6fa367609aba4d43f22d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 11:55:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42b46b8021a74c6aaa7d855b0b6238b3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/168f4995e4a64212939fc5d697ef7401 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3c14b7abfa6155434a471fd20bb921a15aadcdc9","revisions":{"3c14b7abfa6155434a471fd20bb921a15aadcdc9":{"kind":"REWORK","_number":1,"created":"2026-03-27 07:45:36.000000000","uploader":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"ref":"refs/changes/64/982364/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/64/982364/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/982364/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/982364/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/982364/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/64/982364/1"}}},"commit":{"parents":[{"commit":"eb9dd526a3c4ba0ffbb69d06cc387c86ca755d42","subject":"Add a release note for tenant_id deprecation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/eb9dd526a3c4ba0ffbb69d06cc387c86ca755d42"}]}],"author":{"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","date":"2026-03-16 10:48:02.000000000","tz":60},"committer":{"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","date":"2026-03-27 07:45:36.000000000","tz":0},"subject":"Update OVN doc about \u0027ovn-nbctl\u0027 command usage regarding the Gateway_Chassis registers","message":"Update OVN doc about \u0027ovn-nbctl\u0027 command usage regarding the Gateway_Chassis registers\n\nNeutron does not support adding or modyfing the Gateway_Chassis\nregisters in NB DB using the \"ovn-nbctl lrp-set-gateway-chassis\"\nor the ovn-nbctl set\" commands. This patch adds this information\nto the Neutron OVN documentation to increase the awareness of the\ncloud administrators.\n\nThis patch also adds extra logging if the Gateway_Chassis register\nname is not parsable by Neutron.\n\nCloses-Bug: #2143751\nChange-Id: I69e9fc00bfc58fb86675f72f94d10d014f25cd3a\nSigned-off-by: Marcin Wilk \u003cmarcin.wilk@canonical.com\u003e\n(cherry picked from commit d781fac878700bc7e87992d731ce4c0c59c64b43)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3c14b7abfa6155434a471fd20bb921a15aadcdc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3c14b7abfa6155434a471fd20bb921a15aadcdc9"}]},"branch":"refs/heads/stable/2026.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
