)]}'
{"id":"openstack%2Fneutron~982365","triplet_id":"openstack%2Fneutron~stable%2F2025.2~I69e9fc00bfc58fb86675f72f94d10d014f25cd3a","project":"openstack/neutron","branch":"stable/2025.2","topic":"fix-ovn-chassis-name-with-hyphen-stable/2026.1-stable/2025.2","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-03-30 07:51:16.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"30534":{"account":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"last_update":"2026-03-31 11:58:01.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-03-31 11:58:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I69e9fc00bfc58fb86675f72f94d10d014f25cd3a","subject":"Update OVN doc about \u0027ovn-nbctl\u0027 command usage regarding the Gateway_Chassis registers","status":"MERGED","created":"2026-03-27 07:46:12.000000000","updated":"2026-03-31 11:59:15.000000000","submitted":"2026-03-31 11:58:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"982365-fix-ovn-chassis-name-with-hyphen-stable/2026.1-stable/2025.2","cherry_pick_of_change":982364,"cherry_pick_of_patch_set":1,"meta_rev_id":"b01363c07f0c5f5b9b90ba565080b20ec9dbfd1e","_number":982365,"virtual_id_number":982365,"owner":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-31 11:58:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-03-30 07:51:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-03-31 09:39:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-03-31 09:39:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-27 08:07:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-27 10:07:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-27 12:24:16.000000000","updated_by":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-03-27 12:24:16.000000000","updated_by":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-03-31 09:39:27.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"2d866f463d660053c7412bf8d0655dd03c01fb7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"date":"2026-03-27 07:46:12.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2026.1.","accounts_in_message":[],"_revision_number":1},{"id":"fcb98ccccf8dbca607a2efce8555213b5098054e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 08:07:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fea44feeb1004e14914b4828c8b41490\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b5a0d9eff0b4466c90b9fa252a7f0ab4 : SUCCESS in 16m 36s\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c40689cacdc4423d9341a05b2492ba20 : SUCCESS in 18m 46s\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/31ca57ff1ace4cd2886df76a2a0ff7ea : SUCCESS in 19m 05s","accounts_in_message":[],"_revision_number":1},{"id":"498f89249ab602a97b6befeb56dd74bc92deb089","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 10:07:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3584491a5eb04973972858b1e964cb5c\n\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/6737d7f82d9c468e90ac5de61279f6e7 : SUCCESS in 2h 19m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7fcc82d22ee94843a7a7f5a7669883bf : SUCCESS in 38m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccc06e3e47f0460d9d4e87f9c37010f8 : SUCCESS in 15m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6bdf9c64472a411197376cee1cb0e3c5 : SUCCESS in 22m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fea1bd9908da4abf8048a9c48da72f1d : SUCCESS in 14m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e5603920dbeb47e78c84e697c6e3bccb : SUCCESS in 27m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e498f363d42143378e8d414b69ea03bd : SUCCESS in 10m 41s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a34286dc8c9d45d0bfb5b642cb0e8ab5 : SUCCESS in 1h 26m 10s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/94f59a9aa2a14937a4fa7abb7c6ddf73 : SUCCESS in 1h 27m 41s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0c2cead9afe54ce584a299a9c6c1c37e : SUCCESS in 1h 08m 42s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0e430902bbd34b5aaef0672d6375a343 : SUCCESS in 1h 23m 09s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7a09b736b23a43d49f8988f4a87feca0 : SUCCESS in 1h 27m 00s","accounts_in_message":[],"_revision_number":1},{"id":"4227e7fff4f08d62eae24a21a298b354ef5b3c19","author":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"date":"2026-03-27 12:24:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"15a1927e1c7c2ef19811ce669d9f291f3b517987","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-30 07:51:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ae04c191df4ac51e93fbb3779735ce352c9d3a19","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2026-03-31 09:39:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5dc58e26d14b7f007861f1ab6ff2e6a9caaecab7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 09:40:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"52b828deddf84a5d998d0a714f1fa38063aa5fcb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 11:58:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed539ce56fb944869d09e15f6079133b\n\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/2c32886f51c841f9b37721147d5d8e06 : SUCCESS in 2h 17m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bbef83500884ebe9a85a9f21e7a2dc3 : SUCCESS in 6m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cbada804cc4f4177a06c73195b7497a2 : SUCCESS in 15m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7fbbd8290f7a4f2a8e678b2eac3c64fd : SUCCESS in 22m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/377a1d3da7b14ddba6f2fb8f78ac0076 : SUCCESS in 13m 13s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/85eccbe849154a70a683328da2e9849e : SUCCESS in 1h 32m 30s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/5abfeaacb587479e804ea5cde3023cb2 : SUCCESS in 45m 55s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d0dee08b641a4c728f38416e28a64b9b : SUCCESS in 52m 39s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/457a1d0267bc4d9a823c9b44d98882f8 : SUCCESS in 1h 33m 30s","accounts_in_message":[],"_revision_number":1},{"id":"246bfa28db0556b96c1611ffdd3e93ef1a2ac3d9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 11:58:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b01363c07f0c5f5b9b90ba565080b20ec9dbfd1e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 11:59:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2b5ee01681b4c83bcf2efe7dc7f8f52\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f49a2c1ff824ec1aeaf42b6dbd24016 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"685ea9786832d4e35f59e607c0f4cf4b45e42b74","revisions":{"685ea9786832d4e35f59e607c0f4cf4b45e42b74":{"kind":"REWORK","_number":1,"created":"2026-03-27 07:46:12.000000000","uploader":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"ref":"refs/changes/65/982365/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/982365/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/982365/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/982365/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/982365/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/982365/1"}}},"commit":{"parents":[{"commit":"751f359ef3bf546192c1e8712b648bc07e66a004","subject":"Add check for null when updating router state","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/751f359ef3bf546192c1e8712b648bc07e66a004"}]}],"author":{"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","date":"2026-03-16 10:48:02.000000000","tz":60},"committer":{"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","date":"2026-03-27 07:46:12.000000000","tz":0},"subject":"Update OVN doc about \u0027ovn-nbctl\u0027 command usage regarding the Gateway_Chassis registers","message":"Update OVN doc about \u0027ovn-nbctl\u0027 command usage regarding the Gateway_Chassis registers\n\nNeutron does not support adding or modyfing the Gateway_Chassis\nregisters in NB DB using the \"ovn-nbctl lrp-set-gateway-chassis\"\nor the ovn-nbctl set\" commands. This patch adds this information\nto the Neutron OVN documentation to increase the awareness of the\ncloud administrators.\n\nThis patch also adds extra logging if the Gateway_Chassis register\nname is not parsable by Neutron.\n\nCloses-Bug: #2143751\nChange-Id: I69e9fc00bfc58fb86675f72f94d10d014f25cd3a\nSigned-off-by: Marcin Wilk \u003cmarcin.wilk@canonical.com\u003e\n(cherry picked from commit d781fac878700bc7e87992d731ce4c0c59c64b43)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/685ea9786832d4e35f59e607c0f4cf4b45e42b74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/685ea9786832d4e35f59e607c0f4cf4b45e42b74"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
