)]}'
{"id":"openstack%2Fneutron~983201","triplet_id":"openstack%2Fneutron~master~I5dab5ef758094b9bfd1f22f4939466c8a810ae6d","project":"openstack/neutron","branch":"master","topic":"aap-reject-multicast","attention_set":{},"removed_from_attention_set":{"15554":{"account":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"last_update":"2026-05-21 21:10:02.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-05-21 19:10:03.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I5dab5ef758094b9bfd1f22f4939466c8a810ae6d","subject":"Reject multicast addresses in AAP","status":"MERGED","created":"2026-04-02 13:24:17.000000000","updated":"2026-05-21 21:11:42.000000000","submitted":"2026-05-21 21:10:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"983201-aap-reject-multicast","meta_rev_id":"01c9c89222c702d13befc7d8cee10a4d82f688e8","_number":983201,"virtual_id_number":983201,"owner":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-21 21:10:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2026-05-21 07:24:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2026-05-21 19:10:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2026-05-21 19:10:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-02 13:33:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-04-02 14:38:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-02 15:10:31.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2026-04-02 16:35:02.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-04-08 13:17:24.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2026-05-21 07:24:35.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-05-21 19:10:03.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"e24928cdcd36ac52cf6597ab5bae84fa6a775846","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2026-04-02 13:24:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cc0048634c9c3ea2911b5d19f04fcccc4d35a6cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2026-04-02 13:25:59.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"eb346822536eae671444d84f56bf95963179e74a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 13:33:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/346e85fca1a546db9c4709296a72e105\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9455730bb9c041c49232de47984ac6c3 : FAILURE in 6m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/96164e3d42a3486ca5acf93f2e55f50d : FAILURE in 4m 34s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e63468ae1024429eb0acf7fe645156f9 : FAILURE in 6m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"862a0aa51e3750c7552a6e861d8fa73b68fb424f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 14:38:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3e97c7a1cba04408a36fd02978702ae3\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ee85ebe7303c47dda2cda7095bc8c306 : FAILURE in 14m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/df2c0481a59a4b15b0146dece6a4f3fb : POST_FAILURE in 27m 43s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2ff83b38548d4b0cad0bb50e5fcaf596 : FAILURE in 15m 53s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6f8184877c8940a5803b172330e3957e : FAILURE in 16m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b37ce05e51cd4f8eafb7ebc45dde5237 : FAILURE in 13m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1658737f879846a9a4ca59a041d03396 : FAILURE in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36b5a28f21bc4ba48fa6ced446d1296e : SUCCESS in 20m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/398c6444981140f89be84135db1f78e2 : FAILURE in 4m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/64793f220b9f4ca7a5d8341073a42a99 : FAILURE in 3m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3ca49f86bea34330afcb6db644f592d9 : FAILURE in 7m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f2dc77288244661a252e1c81bf9d2c4 : FAILURE in 7m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97449d3990404d9ea4212b58f54bee28 : SUCCESS in 14m 14s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6929557eba68434988a519b69adc1f6b : FAILURE in 10m 50s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e451c28e273f4e82a38552d2b565ffc0 : FAILURE in 1h 03m 12s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/bd541f445f7847e69a5b9ce4cf7ce369 : FAILURE in 1h 10m 08s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/dfefd4c369e641d28ce364c02117fd21 : FAILURE in 13m 11s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f835efc4c2c04e83b5f46964eeff56ba : FAILURE in 14m 10s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5586d065e3874262b6bec7472da79435 : FAILURE in 18m 34s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f45cf4c1537049978134d77ce32363f2 : FAILURE in 15m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a79df9f595b842e9bd55bd73c519288c : FAILURE in 18m 11s","accounts_in_message":[],"_revision_number":2},{"id":"9757d50e04536a57229faae4841c1313a8346310","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-02 16:35:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d4d46812b5113d9cd49b5cf565d26e180c049d0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2026-04-08 12:24:46.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"89d54a6508f59253c78931da449bcce4b2f05871","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 13:04:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abe266a780ab4903b220683c8403c0f6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/080115e2b28c45d08c2e21eed20ed55d : FAILURE in 5m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0b5e8aaf5f4b43b08bc10211dc58f4e1 : FAILURE in 6m 10s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/da6189f3c0d14c4983184073cb71f318 : FAILURE in 7m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0cfe5550516d10c3fe0e216ea5a62052a0da27c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 13:29:24.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/14dcd62ea14244fe9e635c7941ef9c0a\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b1a4ec0d072843ed9ef4516749b86ec4 : FAILURE in 13m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c93691c651614bc5af8467da84109201 : POST_FAILURE in 25m 48s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/bf11d472cd524b9ca72b53dbf7b1cf87 : FAILURE in 15m 26s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/925824f032db4f31b25a97bd01aab3c9 : FAILURE in 17m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6fafeb77abbd4d2d875f7e7daafee5a4 : FAILURE in 17m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf52fd6866e94451b9e9b10332ac0517 : FAILURE in 11m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f445f65ca78249bb99d819b58e333cc0 : SUCCESS in 26m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cee7356158dd458d98b0ebf6017739dc : FAILURE in 2m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d57a3b91b4ab4fc2ae7de9e67f9d1e8f : FAILURE in 2m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0c38116ef9734229bb8b89f1effa4aa2 : FAILURE in 10m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c48c3403c164ac58f0ee61b62443188 : FAILURE in 7m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/007a39898af84de892427582174396b3 : SUCCESS in 26m 48s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/169fcb6c15f24be2b07d53adf0645e4e : FAILURE in 12m 36s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/947ac4de711546b3b9c682f6b9abf5eb : FAILURE in 1h 02m 03s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/c0e9a6f33ce04299be6acd9dc5a704c9 : FAILURE in 1h 00m 43s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/17c1cc025c5c4a0aa2c9daa8c24acefd : FAILURE in 25m 00s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f021ea1682a94cd28b32168d6f9b6f03 : FAILURE in 13m 06s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6a1d63b55ff345a884fb50f7c4226d9d : FAILURE in 23m 39s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2b6e1c025d6940baa882a4e92c08741c : FAILURE in 23m 34s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0ed8d0e038204d28842bd12e4c7b97d7 : FAILURE in 21m 25s","accounts_in_message":[],"_revision_number":3},{"id":"a7beb1145942a13c953b328c82423bcc082c5cd5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2026-04-20 12:57:07.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"6d6b03668d4a1065455fac97736f93e6f5d365e4","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2026-04-20 12:58:05.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8eb93c1545b0371fc22ea79e8a8b482e45568290","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 15:06:31.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f97eb0755a9c45f0999508374afef184\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d3a1e39d4b3449c7b0205f17d4385c85 : FAILURE in 15m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d08c5bdeb8d34f94bff2f560402a9dbc : POST_FAILURE in 21m 26s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b4ea167539f64a3e946de9d47c3fd56e : FAILURE in 21m 39s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/42b0662b7e21411991bd5366cab4aca8 : FAILURE in 11m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0a370b80c4034a269a5e119f55663710 : FAILURE in 12m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec7b20d1d4094a7a81a24eb5a275febe : FAILURE in 9m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8dc2afb81434416ab8f420fb39f789a5 : FAILURE in 2m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0b5917a7bf3a493884249bbe465fcfc8 : FAILURE in 4m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/58adb3fca9ed442d9f6a40e3a31dea96 : FAILURE in 3m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/abb984d0e56542c4a2425f9cda88a074 : FAILURE in 4m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/61633e3231d1476f939c31b809d92ba0 : FAILURE in 7m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/793d1d45957b4e6b8aed5b85963b4536 : FAILURE in 6m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c2264ad0a8634110b61c40cd6173b94a : SUCCESS in 3m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06eca97159b34cddab749da41249b52b : SUCCESS in 22m 33s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1dc5253b61c14a3a9c2600dd1bccc3a4 : FAILURE in 16m 59s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/034d1bbd0a7a4c008d1d457828eb8621 : FAILURE in 1h 03m 59s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/0e12ea3019a5469ea004955903a8a2d0 : FAILURE in 37m 14s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3117ded6df3c4f4ea8654a20162de579 : FAILURE in 10m 09s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/38d2ccb0bf7e40e1965135a90b22088e : FAILURE in 20m 43s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/21c1af1b3a7f4f5280a3aa1c706bff87 : FAILURE in 19m 10s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/98ad6b4bfa724ecea8f2616e7f8d8a53 : FAILURE in 8m 41s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ce26181ee2114cdba5e766e4599fe6d0 : FAILURE in 19m 34s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/b3a6516e33e64869a0f21131c0fb7651 : FAILURE in 27m 08s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9dcac0383e36572f1e3af17cc5cb1e93e377a1c8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 15:36:29.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49998c6207ec49c08a80c2eb3b6e24a2\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/160946e7004546d0ac3ccdfb9c68243b : FAILURE in 4m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5d50abde1e9d4e88898ba25ff90dfcae : FAILURE in 4m 41s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f8d13eb4acd74db393a5a85933216278 : FAILURE in 6m 54s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e4a93612446e0f372ab51614751b05779b6e9622","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-04-21 12:01:34.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"85645cbb7d1c9071c356d1ffbc73b1da35216bd5","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-04-21 17:59:27.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b70e7ed63d9abc6ad4aac3a4c05469b5bba2b9ba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 18:46:29.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/daeee189ff9b45fc847c96f0e1eaa230\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/af257dad423049beb63b82143459bf97 : FAILURE in 27m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7ad29a7e027c4037b08197ef2ce4c6ce : FAILURE in 18m 25s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b58a876da11d401ebc5be688c1ed6ab9 : FAILURE in 39m 50s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a7d0971fb563ae40688908a184439d2a7615f593","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 21:14:02.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f4b5c446de884f479e96339ad8ed6ebf\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/28b68127012c4890a81659b47253611f : SUCCESS in 1h 13m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bc3e521e573b42c59bf488a56157bd89 : SUCCESS in 34m 33s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7b1716294dd444dba7a4d1b18da389e1 : FAILURE in 2h 22m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1d29fde507b24352aa63229df4ab0ab6 : SUCCESS in 1h 49m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b1e022dc63a54f90bcb607f1c621e0ec : SUCCESS in 41m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3ecd58de532441891b6081333d8f635 : FAILURE in 37m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0de9f9f7d634919b5611501be02a81b : SUCCESS in 17m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bd2b76b050fa4e519dca88896e938b8c : FAILURE in 33m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ce32e20092294a28a2811bf2a8c78409 : FAILURE in 24m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/017f05dc43764f0cba7e4a5e3d11bc2a : FAILURE in 27m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bc25b728fc8b47679001980318a5f279 : FAILURE in 34m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95ae3a302e95400cb0356bfeffd2c249 : SUCCESS in 11m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0c4f7621e1d6434295158ee3264f313f : SUCCESS in 2m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/81a97992d40f48e3a256f168e5772a9c : SUCCESS in 13m 33s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/906fc6d44a8c41be855b60961270e7a9 : SUCCESS in 48m 10s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f86ad1fb6e2345d28c9732fed6e24cb1 : SUCCESS in 42m 20s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/c0e3ab8a864a479fa58ee9014b634f37 : SUCCESS in 1h 17m 17s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e18a509591dd4b50b4d33bb1271c8076 : SUCCESS in 1h 09m 42s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5bf2ff6d48b446e39971ecb4ae3430ce : SUCCESS in 1h 23m 29s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/875919bc4f75423fbe378f5cb24b41ae : SUCCESS in 1h 04m 47s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a9d946820f2243129995fd5193bd8813 : SUCCESS in 1h 26m 42s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/49e5a08de2cd4b409ef01f980d84cac1 : SUCCESS in 1h 26m 18s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/55a5d7646b2843cb8d6478bcbd7c6a83 : TIMED_OUT in 3h 07m 02s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2bf8de1a376dbc5a8579c367b54ee367c419e45b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2026-04-23 12:09:39.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"b72e2e37d7016cca427ce639696fe26b2a7eeb3a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 13:35:41.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fa05d03cb13470f836cc7afc589faf7\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8fc53c2cbf4b431ba4d21032235f99bd : SUCCESS in 27m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b365c0a1277146599cff32626ce8154f : SUCCESS in 19m 21s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b9bc35c9a6474f8c91fff841318f3572 : SUCCESS in 41m 34s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d3bb3c3b944e89a94f15961913e10852533cf665","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 14:30:37.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3b8ad0f5a584909b80d8732cce52b98\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/332b56082db74c7d99dda5aa1e0e734e : SUCCESS in 1h 12m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fbb6104054874d9eb9e6d9acffda2729 : SUCCESS in 1h 07m 42s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c0743be2fc31427c8e16cafa60f9060d : SUCCESS in 1h 17m 02s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/dbfb97b7751f4c2b91ca2f6265c0a906 : SUCCESS in 1h 47m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ebca09d391ff457eb0e9faa97c49b027 : SUCCESS in 31m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a81b9691da042979cc8b81ec0a7719d : SUCCESS in 38m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfdc7d4b3b2a4da5961b8f562559f613 : SUCCESS in 30m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c5b77f136e404a55876d0b3b0b48f688 : SUCCESS in 21m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f2cbaf05c0d64853a62ed8c9a6a0cb22 : SUCCESS in 18m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eb32aa2b5ce749a2bebae728dc85c649 : SUCCESS in 21m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/80e792bbec5c467ea27fb62848c17c34 : SUCCESS in 21m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcfbbbe1617540d18a02124425cb5be9 : SUCCESS in 12m 41s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a03a7916a4b2488ea0ab4358fd8fef7c : SUCCESS in 3m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba8959009d39455584dee99bb8bcb2e3 : SUCCESS in 23m 58s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e9aeaab3a7e6419b875929ba203b102c : SUCCESS in 1h 23m 08s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e2247c44b1e24773943880c0f05ebf6b : SUCCESS in 1h 00m 42s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/7215dac444154ed4917b4a729804cb08 : SUCCESS in 1h 18m 01s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5cb3830295a64ad7afd071fdd85eae9b : SUCCESS in 1h 53m 30s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/04441cabdf6d4a5fa851e6b8efe40152 : SUCCESS in 1h 41m 30s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7394a7eb5f3047d2a9aee7a07e410ca1 : SUCCESS in 1h 02m 00s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/cf4a907e64f144b398713126e3cf6f5f : SUCCESS in 32m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e97ecc38b9d4475f832ff934de339a53 : SUCCESS in 1h 19m 54s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/51ad1e3328294a469e6a7d2b725f97bf : SUCCESS in 2h 14m 49s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d58c9cd1f9f92604b1d64e16dda896a10331afff","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-05-21 07:24:35.000000000","message":"Patch Set 5: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"2ceafade6d669d00f38c744fcbd1ad1c9d72753a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-21 19:10:03.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"6fef5764082925687e83682ed3893cc9f2567d04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-21 19:10:33.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"6cbc7d49971ba4ed240943e7ae0765bb9894dfaf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-21 21:10:02.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1dadef0937824646ba7ff664870b59a0\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/dc807807afcf4d028c8ffdf8ad2dd231 : SUCCESS in 1h 28m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/32ea90693f544770b8771a1f1f0726e0 : SUCCESS in 56m 14s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5bc72ce6a4bf4b819c6d576e6aa1c9c0 : SUCCESS in 1h 22m 58s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9a0a0c20af214df69c8be49f02d8de8f : SUCCESS in 1h 46m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50e10dcdc83a45f99306c71574bf1d1c : SUCCESS in 7m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8b49b5c062be43bc88c4047b742beee1 : SUCCESS in 29m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/deb3e925a62345e0b5d830876051f1f8 : SUCCESS in 18m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/859fe6aeae94409092a4a75a6ed97e15 : SUCCESS in 6m 22s\n- requirements-check https://zuul.opendev.org/t/openstack/build/39a07d080a3143b895f7499f1797321f : SUCCESS in 2m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0971245149e456f9cf39975a54f05d1 : SUCCESS in 21m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a382d6461f844313a34e130447f70e89 : SUCCESS in 47m 32s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/26418b186e7d42eb9b708fcbc7bc5f57 : SUCCESS in 1h 56m 50s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/4153f2d57d334182a89c62beb45737ed : SUCCESS in 1h 27m 27s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/406d92c41cff489f9797c1d1801208ff : SUCCESS in 1h 13m 10s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/bb341a8977644864b7df213aeaa6da9e : SUCCESS in 1h 06m 20s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":5},{"id":"31344cf5033773509e2b730033da96c5d9742af7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-21 21:10:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"01c9c89222c702d13befc7d8cee10a4d82f688e8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-21 21:11:42.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/161be98d1d764870bf4ac85f32c5dbe5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee2fd477e994411da24ef7fd4451269b : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d0846d718da4f95a3a2cd168c3dd920 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"58fdeda1b50965f707d952029386c94420bf302e","revisions":{"2812992a91e35dd128bb90b4f47b8c93a0a5bc5b":{"kind":"REWORK","_number":1,"created":"2026-04-02 13:24:17.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/01/983201/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/983201/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/983201/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/983201/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/983201/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/983201/1"}}},"commit":{"parents":[{"commit":"5a516696d02a054d216972d00ead01238a9c2eff","subject":"Merge \"Use project/project_id in documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5a516696d02a054d216972d00ead01238a9c2eff"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2026-03-30 12:31:50.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2026-04-02 13:23:50.000000000","tz":120},"subject":"WIP Reject multicast addresses in AAP","message":"WIP Reject multicast addresses in AAP\n\nAs discussed in [1] and tracked in [2],\nthe Allowed Address Pairs spec never intended to support the\nuse of multicast addresses in an AAP.\nHowever, when given a multicast address, the API accepted it,\nbut the backends behaved differently, most of them in buggy ways.\nAll this was never properly documented.\n\nAdditionally in a past bug report [4] we also came to the conclusion\nthat some multicast addresses (224.0.0.0/24),\nshould be handled in a different layer,\nthat is by the default behavior of our firewall drivers.\nThis was documented in [5].\n\nThis patch intends to rectify this mess, by:\n* rejecting multicast addresses in an AAP in the API and\n* documenting the current situation.\n\nAt some point in the future we may add support for multicast addresses\nin AAPs, but at the moment we don\u0027t have a use case for that.\nThis patch clears up the API and the docs that the current\nbackends do not handle multicast AAPs properly.\n\nAlso in the future different multicast ranges may need different\nhandling.\n\nWe reject IPs from 224.0.0.0/4 and ff00::/8.\nSince the IP part of an AAP allows subnets,\nwe reject subnets overlapping with the above ranges.\nBut we accept 0.0.0.0/0.\nWe reject MAC addresses with the multicast bit (01:00:00:00:00:00) set.\n\nSince this patch changes API behavior we introduce a new API extension:\naap-reject-multicast\n\nDuring an upgrade, clearing AAPs with multicast addresses is the\nresponsibility of the user.\n\nTODO bump neutron-lib requirement when the n-lib patch is merged and released\n\n[1] https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/RX4R7DNSZWJ5W7BKZ3UULSRZCDHYWALD/\n[2] https://bugs.launchpad.net/neutron/+bug/2147115\n[3] https://opendev.org/openstack/neutron-specs/src/branch/master/misc/api/allowed_address_pairs.rst\n[4] https://bugs.launchpad.net/neutron/+bug/1889631\n[5] https://opendev.org/openstack/neutron/src/branch/master/doc/source/admin/config-ovsfwdriver.rst#differences-between-ovs-and-iptables-firewall-drivers\n\nChange-Id: I5dab5ef758094b9bfd1f22f4939466c8a810ae6d\nCloses-Bug: #2147115\nDepends-On: FIXME replace with gerrit url: I3228784a12402c76c58a333e897ec2fd56e2b266\nSigned-off-by: Bence Romsics \u003cbence.romsics@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2812992a91e35dd128bb90b4f47b8c93a0a5bc5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2812992a91e35dd128bb90b4f47b8c93a0a5bc5b"}]},"branch":"refs/heads/master"},"6187682d4631c360d42cc2b633fffe0c47615934":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-04-02 13:25:59.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/01/983201/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/983201/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/983201/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/983201/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/983201/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/983201/2"}}},"commit":{"parents":[{"commit":"5a516696d02a054d216972d00ead01238a9c2eff","subject":"Merge \"Use project/project_id in documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5a516696d02a054d216972d00ead01238a9c2eff"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2026-03-30 12:31:50.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2026-04-02 13:25:31.000000000","tz":120},"subject":"WIP Reject multicast addresses in AAP","message":"WIP Reject multicast addresses in AAP\n\nAs discussed in [1] and tracked in [2],\nthe Allowed Address Pairs spec never intended to support the\nuse of multicast addresses in an AAP.\nHowever, when given a multicast address, the API accepted it,\nbut the backends behaved differently, most of them in buggy ways.\nAll this was never properly documented.\n\nAdditionally in a past bug report [4] we also came to the conclusion\nthat some multicast addresses (224.0.0.0/24),\nshould be handled in a different layer,\nthat is by the default behavior of our firewall drivers.\nThis was documented in [5].\n\nThis patch intends to rectify this mess, by:\n* rejecting multicast addresses in an AAP in the API and\n* documenting the current situation.\n\nAt some point in the future we may add support for multicast addresses\nin AAPs, but at the moment we don\u0027t have a use case for that.\nThis patch clears up the API and the docs that the current\nbackends do not handle multicast AAPs properly.\n\nAlso in the future different multicast ranges may need different\nhandling.\n\nWe reject IPs from 224.0.0.0/4 and ff00::/8.\nSince the IP part of an AAP allows subnets,\nwe reject subnets overlapping with the above ranges.\nBut we accept 0.0.0.0/0.\nWe reject MAC addresses with the multicast bit (01:00:00:00:00:00) set.\n\nSince this patch changes API behavior we introduce a new API extension:\naap-reject-multicast\n\nDuring an upgrade, clearing AAPs with multicast addresses is the\nresponsibility of the user.\n\nTODO bump neutron-lib requirement when the n-lib patch is merged and released\n\n[1] https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/RX4R7DNSZWJ5W7BKZ3UULSRZCDHYWALD/\n[2] https://bugs.launchpad.net/neutron/+bug/2147115\n[3] https://opendev.org/openstack/neutron-specs/src/branch/master/misc/api/allowed_address_pairs.rst\n[4] https://bugs.launchpad.net/neutron/+bug/1889631\n[5] https://opendev.org/openstack/neutron/src/branch/master/doc/source/admin/config-ovsfwdriver.rst#differences-between-ovs-and-iptables-firewall-drivers\n\nChange-Id: I5dab5ef758094b9bfd1f22f4939466c8a810ae6d\nCloses-Bug: #2147115\nDepends-On: https://review.opendev.org/c/openstack/neutron-lib/+/983203\nSigned-off-by: Bence Romsics \u003cbence.romsics@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6187682d4631c360d42cc2b633fffe0c47615934"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6187682d4631c360d42cc2b633fffe0c47615934"}]},"branch":"refs/heads/master"},"7706e89b4ad1b59c7be32f5085988c3ad4a1ba6a":{"kind":"REWORK","_number":3,"created":"2026-04-08 12:24:46.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/01/983201/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/983201/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/983201/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/983201/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/983201/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/983201/3"}}},"commit":{"parents":[{"commit":"52d10bfd4a9881da1aeb26d7317eed8615ec82ef","subject":"Do not explicitly mention supported values","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/52d10bfd4a9881da1aeb26d7317eed8615ec82ef"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2026-03-30 12:31:50.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2026-04-08 12:24:13.000000000","tz":120},"subject":"WIP Reject multicast addresses in AAP","message":"WIP Reject multicast addresses in AAP\n\nAs discussed in [1] and tracked in [2],\nthe Allowed Address Pairs spec never intended to support the\nuse of multicast addresses in an AAP.\nHowever, when given a multicast address, the API accepted it,\nbut the backends behaved differently, most of them in buggy ways.\nAll this was never properly documented.\n\nAdditionally in a past bug report [4] we also came to the conclusion\nthat some multicast addresses (224.0.0.0/24),\nshould be handled in a different layer,\nthat is by the default behavior of our firewall drivers.\nThis was documented in [5].\n\nThis patch intends to rectify this mess, by:\n* rejecting multicast addresses in an AAP in the API and\n* documenting the current situation.\n\nAt some point in the future we may add support for multicast addresses\nin AAPs, but at the moment we don\u0027t have a use case for that.\nThis patch clears up the API and the docs that the current\nbackends do not handle multicast AAPs properly.\n\nAlso in the future different multicast ranges may need different\nhandling.\n\nWe reject IPs from 224.0.0.0/4 and ff00::/8.\nSince the IP part of an AAP allows subnets,\nwe reject subnets overlapping with the above ranges.\nBut we accept 0.0.0.0/0.\nWe reject MAC addresses with the multicast bit (01:00:00:00:00:00) set.\n\nSince this patch changes API behavior we introduce a new API extension:\naap-reject-multicast\n\nDuring an upgrade, clearing AAPs with multicast addresses is the\nresponsibility of the user.\n\nTODO bump neutron-lib requirement when the n-lib patch is merged and released\n\n[1] https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/RX4R7DNSZWJ5W7BKZ3UULSRZCDHYWALD/\n[2] https://bugs.launchpad.net/neutron/+bug/2147115\n[3] https://opendev.org/openstack/neutron-specs/src/branch/master/misc/api/allowed_address_pairs.rst\n[4] https://bugs.launchpad.net/neutron/+bug/1889631\n[5] https://opendev.org/openstack/neutron/src/branch/master/doc/source/admin/config-ovsfwdriver.rst#differences-between-ovs-and-iptables-firewall-drivers\n\nChange-Id: I5dab5ef758094b9bfd1f22f4939466c8a810ae6d\nCloses-Bug: #2147115\nDepends-On: https://review.opendev.org/c/openstack/neutron-lib/+/983203\nSigned-off-by: Bence Romsics \u003cbence.romsics@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7706e89b4ad1b59c7be32f5085988c3ad4a1ba6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7706e89b4ad1b59c7be32f5085988c3ad4a1ba6a"}]},"branch":"refs/heads/master"},"e45ef446971742c6aad95eafa631570fd5dbbea9":{"kind":"REWORK","_number":4,"created":"2026-04-20 12:57:07.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/01/983201/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/983201/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/983201/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/983201/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/983201/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/983201/4"}}},"commit":{"parents":[{"commit":"7e423793733cea025e662fbad4b6369431bf4419","subject":"Merge \"bgp: Redistribute local only addresses\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e423793733cea025e662fbad4b6369431bf4419"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2026-03-30 12:31:50.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2026-04-20 12:55:34.000000000","tz":120},"subject":"Reject multicast addresses in AAP","message":"Reject multicast addresses in AAP\n\nAs discussed in [1] and tracked in [2],\nthe Allowed Address Pairs spec never intended to support the\nuse of multicast addresses in an AAP.\nHowever, when given a multicast address, the API accepted it,\nbut the backends behaved differently, most of them in buggy ways.\nAll this was never properly documented.\n\nAdditionally in a past bug report [4] we also came to the conclusion\nthat some multicast addresses (224.0.0.0/24),\nshould be handled in a different layer,\nthat is by the default behavior of our firewall drivers.\nThis was documented in [5].\n\nThis patch intends to rectify this mess, by:\n* rejecting multicast addresses in an AAP in the API and\n* documenting the current situation.\n\nAt some point in the future we may add support for multicast addresses\nin AAPs, but at the moment we don\u0027t have a use case for that.\nThis patch clears up the API and the docs that the current\nbackends do not handle multicast AAPs properly.\n\nAlso in the future different multicast ranges may need different\nhandling.\n\nWe reject IPs from 224.0.0.0/4 and ff00::/8.\nSince the IP part of an AAP allows subnets,\nwe reject subnets overlapping with the above ranges.\nBut we accept 0.0.0.0/0.\nWe reject MAC addresses with the multicast bit (01:00:00:00:00:00) set.\n\nSince this patch changes API behavior we introduce a new API extension:\naap-reject-multicast\n\nDuring an upgrade, clearing AAPs with multicast addresses is the\nresponsibility of the user.\n\nThis patch bumps our neutron-lib requirement from 3.24.0 to 3.25.0.\n\n[1] https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/RX4R7DNSZWJ5W7BKZ3UULSRZCDHYWALD/\n[2] https://bugs.launchpad.net/neutron/+bug/2147115\n[3] https://opendev.org/openstack/neutron-specs/src/branch/master/misc/api/allowed_address_pairs.rst\n[4] https://bugs.launchpad.net/neutron/+bug/1889631\n[5] https://opendev.org/openstack/neutron/src/branch/master/doc/source/admin/config-ovsfwdriver.rst#differences-between-ovs-and-iptables-firewall-drivers\n\nChange-Id: I5dab5ef758094b9bfd1f22f4939466c8a810ae6d\nCloses-Bug: #2147115\nDepends-On: https://review.opendev.org/c/openstack/neutron-lib/+/983203\nSigned-off-by: Bence Romsics \u003cbence.romsics@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e45ef446971742c6aad95eafa631570fd5dbbea9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e45ef446971742c6aad95eafa631570fd5dbbea9"}]},"branch":"refs/heads/master"},"58fdeda1b50965f707d952029386c94420bf302e":{"kind":"REWORK","_number":5,"created":"2026-04-23 12:09:39.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/01/983201/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/983201/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/983201/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/983201/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/983201/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/983201/5"}}},"commit":{"parents":[{"commit":"7e423793733cea025e662fbad4b6369431bf4419","subject":"Merge \"bgp: Redistribute local only addresses\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e423793733cea025e662fbad4b6369431bf4419"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2026-03-30 12:31:50.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2026-04-23 12:07:12.000000000","tz":120},"subject":"Reject multicast addresses in AAP","message":"Reject multicast addresses in AAP\n\nAs discussed in [1] and tracked in [2],\nthe Allowed Address Pairs spec never intended to support the\nuse of multicast addresses in an AAP.\nHowever, when given a multicast address, the API accepted it,\nbut the backends behaved differently, most of them in buggy ways.\nAll this was never properly documented.\n\nAdditionally in a past bug report [4] we also came to the conclusion\nthat some multicast addresses (224.0.0.0/24),\nshould be handled in a different layer,\nthat is by the default behavior of our firewall drivers.\nThis was documented in [5].\n\nThis patch intends to rectify this mess, by:\n* rejecting multicast addresses in an AAP in the API and\n* documenting the current situation.\n\nAt some point in the future we may add support for multicast addresses\nin AAPs, but at the moment we don\u0027t have a use case for that.\nThis patch clears up the API and the docs that the current\nbackends do not handle multicast AAPs properly.\n\nAlso in the future different multicast ranges may need different\nhandling.\n\nWe reject IPs from 224.0.0.0/4 and ff00::/8.\nSince the IP part of an AAP allows subnets,\nwe reject subnets overlapping with the above ranges.\nBut we accept 0.0.0.0/0.\nWe reject MAC addresses with the multicast bit (01:00:00:00:00:00) set.\n\nSince this patch changes API behavior we introduce a new API extension:\naap-reject-multicast\n\nDuring an upgrade, clearing AAPs with multicast addresses is the\nresponsibility of the user.\n\nThis patch bumps our neutron-lib requirement from 3.24.0 to 3.25.0.\n\n[1] https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/RX4R7DNSZWJ5W7BKZ3UULSRZCDHYWALD/\n[2] https://bugs.launchpad.net/neutron/+bug/2147115\n[3] https://opendev.org/openstack/neutron-specs/src/branch/master/misc/api/allowed_address_pairs.rst\n[4] https://bugs.launchpad.net/neutron/+bug/1889631\n[5] https://opendev.org/openstack/neutron/src/branch/master/doc/source/admin/config-ovsfwdriver.rst#differences-between-ovs-and-iptables-firewall-drivers\n\nChange-Id: I5dab5ef758094b9bfd1f22f4939466c8a810ae6d\nCloses-Bug: #2147115\nDepends-On: https://review.opendev.org/c/openstack/neutron-lib/+/983203\nSigned-off-by: Bence Romsics \u003cbence.romsics@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/58fdeda1b50965f707d952029386c94420bf302e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/58fdeda1b50965f707d952029386c94420bf302e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
