)]}'
{"id":"openstack%2Fneutron~983371","triplet_id":"openstack%2Fneutron~master~I822c66dea14ca05853cfc95d16ce2a9540f1570b","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-04-09 21:15:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I822c66dea14ca05853cfc95d16ce2a9540f1570b","subject":"Use constant for ha_failover_strategy value","status":"MERGED","created":"2026-04-05 07:25:05.000000000","updated":"2026-04-09 21:17:55.000000000","submitted":"2026-04-09 21:15:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"983371","meta_rev_id":"55274da9ba8d6b0ec2671f60e9135dad2026346c","_number":983371,"virtual_id_number":983371,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-09 21:15:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-04-09 06:05:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-04-08 08:33:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2026-04-08 01:44:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-04-09 06:05:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-05 08:06:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-04-05 09:42:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-08 01:44:01.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-04-08 08:33:06.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-04-09 06:05:35.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"17d5b4a3e1efa28ea7ab3aa2a4b9d5e985601ba0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-05 07:25:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c9cc83185b3527f9976385bbfe8a211f32352eb7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-05 08:06:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f560182c01f4432ea157aed8d938ecec\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/39b93eaef27a4b54aae4e7d5dd3b1de1 : SUCCESS in 22m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/61ca33aac54343868fa5b6f5b8391122 : SUCCESS in 18m 00s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/6bca6dfee235442b9158ac81d68608c3 : SUCCESS in 39m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"97ac269532dfa6ffad9e1e51d73a9c52f447127d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-05 09:42:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/16737251cbc74c06ba3c3991acc3ee05\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8ad4fccf36fe485b9b346a82afe19166 : SUCCESS in 1h 00m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1da785729562476b8278b7e8813bffd0 : SUCCESS in 1h 02m 14s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/36c6765d8c6f4e70a30fe0bbcb4cf885 : SUCCESS in 1h 25m 08s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2dfaee6e8bd349769b06c969d4c6f545 : SUCCESS in 2h 00m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/899804aed0ee448593687b27a105e1bc : SUCCESS in 33m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0473753043114a99adc615a1c2e1e27c : SUCCESS in 1h 03m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1ff52fd99b340f78e97f994510613cc : FAILURE in 6m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a3705e0594e94a5482963aba7558ef09 : SUCCESS in 59m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2fe8de33f44540c7965cbecf0a243537 : SUCCESS in 26m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7053105624494b54be4b3aa28d469ad6 : SUCCESS in 57m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90a7bc67fe934781937e2db333a1b4f6 : SUCCESS in 11m 19s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/51cf2b287df94c249b2bf6d80fc6ce16 : SUCCESS in 1h 24m 32s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e33a2902530145b5a1c2c61fa245ead0 : SUCCESS in 36m 57s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/fe4b55741bfc42cdac237d07b61fcbef : SUCCESS in 1h 07m 44s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8e0c899c927541f5850f1b5982641526 : SUCCESS in 2h 11m 45s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3a7e14f845f34e589b49123f31cdcdfd : SUCCESS in 1h 25m 19s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b8c3dae90685453a82f35819be870571 : SUCCESS in 47m 16s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2bc0556b7aff43cdb7a412311aeed924 : SUCCESS in 1h 01m 02s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bd53b6eb6808462c904fbc18c84eedba : SUCCESS in 1h 27m 11s","accounts_in_message":[],"_revision_number":1},{"id":"9234c48efb70dee828ac34624b255c565ae85e66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-05 09:56:47.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"736ee8f61894f2e086ff2c01ae571b09fd594c1b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-05 10:24:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2cd97e13c894022bf192d211b0e297d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a5214d4cf34740a6b1991f4601088c8f : SUCCESS in 23m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2204e5c4a35d4430a4248c84313f1666 : SUCCESS in 26m 04s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4617283eb57b4d778b4d768e70914d07 : SUCCESS in 24m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"573d5a30691fffe5c0c4eeea91dd59577d86095a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-05 11:47:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e43de7c956fc424280e9048be99e63eb\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/56bae6c56b734993a1172f69f3a3d2e0 : SUCCESS in 1h 09m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9b300edbaef640ab89fe31df99528946 : SUCCESS in 38m 22s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1517d62a610f47899c327b79779948f5 : SUCCESS in 1h 41m 48s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b91723a2c1664cbe8c0357a794202bba : SUCCESS in 1h 49m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/70bbb927754d453fbed65efeeb040cce : SUCCESS in 32m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da77ff0be04c476180ff50e29bd2f1ca : SUCCESS in 59m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29626559bef84d1e9f99fa96e9f98c28 : FAILURE in 24m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/937eae8ebe8b4fa1a37c72fa3ab2841c : SUCCESS in 33m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cc7ec69ff85a47d2a947c915a3f115e4 : SUCCESS in 23m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bd53bfbd25b34abcb2348394992672fa : SUCCESS in 40m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae51f0e97f074be3bd85e5698f38e85c : SUCCESS in 12m 25s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/962116f0473746f2a50e292110a37084 : SUCCESS in 1h 21m 08s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/55aabd36ba6848b8a06a9c9b4f5ebda9 : SUCCESS in 1h 09m 04s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/639481707f4e451dad0518b31a17e767 : SUCCESS in 1h 02m 15s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a88cbd0fb17b4a76a6031a43853fafd5 : SUCCESS in 1h 05m 41s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e9be716ac2fe44169d4da9ef7782f05b : SUCCESS in 1h 38m 46s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/552256ff77b34f85b7260c9306c3dc4e : SUCCESS in 1h 06m 21s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5fef3ee38a3a4b6d99fc2351fd13b270 : SUCCESS in 39m 48s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/eed1d89e31dd4b6e83f4817a460b3e24 : SUCCESS in 1h 25m 29s\n\nWarning:\n  Comments left for invalid file neutron/tests/functional/agent/l2/base.py\n  Comments left for invalid file neutron/tests/functional/agent/l3/extensions/qos/test_fip_qos_extension.py\n  Comments left for invalid file neutron/tests/functional/agent/l3/framework.py\n  Comments left for invalid file neutron/tests/functional/agent/l3/test_dvr_router.py\n  Comments left for invalid file neutron/tests/functional/agent/linux/openvswitch_firewall/test_iptables.py\n  Comments left for invalid file neutron/tests/functional/agent/linux/test_ip_lib.py\n  Comments left for invalid file neutron/tests/functional/agent/linux/test_keepalived.py\n  Comments left for invalid file neutron/tests/functional/agent/linux/test_of_monitor.py\n  Comments left for invalid file neutron/tests/functional/agent/test_dhcp_agent.py\n  Comments left for invalid file neutron/tests/functional/agent/test_ovs_flows.py\n  Comments left for invalid file neutron/tests/functional/pecan_wsgi/test_controllers.py\n  Comments left for invalid file neutron/tests/unit/agent/dhcp/test_agent.py\n  Comments left for invalid file neutron/tests/unit/agent/l2/extensions/metadata/test_host_metadata_proxy.py\n  Comments left for invalid file neutron/tests/unit/agent/l3/extensions/test_port_forwarding.py\n  Comments left for invalid file neutron/tests/unit/plugins/ml2/test_port_binding.py","accounts_in_message":[],"_revision_number":2},{"id":"e4aed4cca0a16742d6aee272dab59e8f4013401b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-07 16:15:56.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"b741a2c9cb704547dbc91fb2698c75a05c7159a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 16:43:05.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73504c502edf49aeaa992d67af4efe9f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5f4ad5c5f87a4994968f4e9e19df70ac : SUCCESS in 23m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d325abe5caec46ef81f577e23b322bbd : SUCCESS in 21m 01s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/92f8569f37464dba9c70822d783e7d79 : SUCCESS in 25m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"355c268c62738f987a4976e9ef2cb695f0044b88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 18:24:12.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bce3d1b9915846ec9b35a6370d255169\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/cb2cc99626944988885c4bec5e3996cc : SUCCESS in 1h 26m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0c26121e24be4acc90075e6255bce5ea : SUCCESS in 37m 23s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e3b96f516b554c67a8a0ed1b739d454e : SUCCESS in 1h 23m 06s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/259efe16afd444709aedf74e40f45a97 : SUCCESS in 2h 07m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/32054ffd787f4ce3930115805c11d646 : SUCCESS in 31m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a38ed64560b34bb6a814ca897108d961 : SUCCESS in 1h 03m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37a1a775213f47c78e760c9ad1bb741e : SUCCESS in 24m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4f6c97b58222401c85d378b511490032 : SUCCESS in 26m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d861e55957334166a573f783607e5c33 : SUCCESS in 16m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/baf04a236ed3400881a4cb86fe32f868 : SUCCESS in 26m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cc1662be26448f4bafeb46656bc4f6e : SUCCESS in 11m 32s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/df4a23aec8674870a38c57912c31dc04 : SUCCESS in 1h 22m 04s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d1dbe0f3ee7b4adfb1b4d21a7c171daf : SUCCESS in 42m 48s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ab7a5f57bfd0470297a6023a231d0221 : SUCCESS in 1h 17m 11s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1e54efcdd93f4046ab8fa73fd81e8d1b : SUCCESS in 1h 47m 50s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/cc6e000f9eab4d10b4c22bd6f00ffd0f : FAILURE in 1h 53m 16s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4a9bca81f1c94749ab924fe60ff3ecb2 : SUCCESS in 56m 31s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/182e37d156b24e0a8895fc988a1fa9fa : SUCCESS in 1h 14m 35s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/116dafdc32884faabb3b26dad8c457c9 : SUCCESS in 1h 41m 16s","accounts_in_message":[],"_revision_number":3},{"id":"7b252e2e125cf4495173feff4013f5426805ef94","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-08 01:44:01.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5b88553eff8ffbf7fb6938a74e1cf2c8a3e4b308","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 02:15:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5a8517ecddc4f839e6779d4941e75c8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0872a736ac2d4d7db4c0b764d3eebbb2 : SUCCESS in 24m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/de52f12d6ea7497e8896a1ba03079cad : SUCCESS in 29m 57s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/aa29ce22d4e840d7b0c6c75ebccc5abc : SUCCESS in 23m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"33964b6bd6ef0a1b08afd09a4a5b2c77e643e8e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 03:49:15.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/809fb67dfd36402bb090204497e76c08\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/85a9c89394094e49aa8a9b1d4e6b9969 : SUCCESS in 1h 12m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4edfe50970194640abd3293ab224aa6c : FAILURE in 18m 25s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d65a77cab16c447e8b2de2647f1a598f : FAILURE in 1h 39m 49s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/af0fd78fe2d843969e2157381f7eb87a : FAILURE in 29m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d9acecdece5144c28d560daffba636cd : SUCCESS in 24m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e27659c5dc534e54bf1f269120feaa0c : SUCCESS in 1h 05m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae5683ce2fe34dcfaeed8c51c2315a4f : SUCCESS in 19m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a15bac51573d4d4fbfbf307b2c424977 : SUCCESS in 25m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8136e7aa16634d96b56ca10aa71acd0d : SUCCESS in 16m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3c54e39b324645f2ad0977030ad02187 : SUCCESS in 28m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72e08ed4b8e4401a826528456b139339 : SUCCESS in 11m 23s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1936b014742d45d38228772d087c05cd : SUCCESS in 1h 36m 52s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1fd9842d8ae04dcb860994c62f50b6a8 : SUCCESS in 1h 10m 52s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ce31a5a528c746e6a31a8292fa69f57b : SUCCESS in 1h 13m 17s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/621c8de293a84c7fbd5906c4c4209e65 : SUCCESS in 1h 53m 45s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c804b7c4833246749a095f9855bf8dec : SUCCESS in 1h 22m 46s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c42899bfe0bd4b98a3b651c1873aa6a5 : SUCCESS in 1h 06m 50s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/95fc5598c2a84b489132813c1d9c1629 : SUCCESS in 1h 30m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8dcbb51760b4475ca48f3c23b789f897 : SUCCESS in 44m 50s","accounts_in_message":[],"_revision_number":3},{"id":"1e21383626a70c5551810f27b069ca1efa97f74f","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-04-08 08:33:06.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b6732e7fdcc804b832da11016b8bdf56c45c2e08","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 09:58:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebdbf40ebc64470c92a59e91e2045716\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c72a6a930fca448aa30562014900a189 : SUCCESS in 25m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d31738c5755b414ba6dd2a90a3ccfc53 : SUCCESS in 21m 07s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a9f1ed10afc844d89d1503d19965cfcf : SUCCESS in 26m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8ceba0affb8d91dfa52fabb65e665577aa809f23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 10:50:04.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/29853e4b8be144a9a764cc34730e44bc\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/76b263c26e0345c2a8679dd6c78089b9 : FAILURE in 1h 12m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6884a11e46fc496281f1770c96812c7a : SUCCESS in 1h 00m 46s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ce49d6fcf91844c8b8fda73161ea5eb5 : SUCCESS in 1h 44m 13s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b0760ab87d434df392c166d2bb2cc696 : SUCCESS in 2h 15m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/452ff1093bc641e188178f70b6cd5feb : SUCCESS in 34m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8b2744210274baeacc946d3972f8436 : SUCCESS in 1h 04m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c01f96ea4f14eaf94b58099134a06ac : SUCCESS in 10m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/63930405f8d2462e9aae3ad273feec44 : SUCCESS in 27m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ac84e7d344834b5cb94c2b9a2c0775f9 : SUCCESS in 16m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0b197994300a4ed8a6721cdb48da4bf0 : SUCCESS in 33m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cf1d2d5fe824ddaa2b62eab7cb3c262 : SUCCESS in 13m 00s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c2233d3ae97147bb90f2ed0d40cdb875 : SUCCESS in 1h 24m 15s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6ec68d2e9e43480fb5c13b424a91430a : SUCCESS in 41m 40s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/83529bfe80544690af957607d5930942 : SUCCESS in 1h 09m 59s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ec72e7a04dab42e4918eda145c7b3e9b : SUCCESS in 1h 17m 36s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7c18067132754b77b2d85fbd48fae919 : FAILURE in 1h 38m 45s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3f4c417e289e4f0b9c67cf81006e7ff0 : SUCCESS in 1h 15m 18s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ed34f3b238bd4585ae0693f01ca8e904 : SUCCESS in 1h 31m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7bc0dfef018845ddb86ac905d79ce595 : SUCCESS in 1h 22m 44s","accounts_in_message":[],"_revision_number":3},{"id":"decfad4f4a107ef0ba5abfa4eb03b675ed88b7fd","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-09 06:05:35.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"728b436d5e5b8f3fee1dc0a6e03072bf05dab1a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 06:41:07.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24a166b5c900414c91f0a0d09dfd299c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2028e55261124f57b0b9c590e86f0985 : SUCCESS in 29m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1dacccf8e31f492a8202cbf4573d00e8 : SUCCESS in 33m 39s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/90327e3d0a7c42cebbb3e5ad3783663e : SUCCESS in 26m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c3e638d02db4cfd50394c54da51b765c28d094dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 08:10:10.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e90bb5c9785548a0adf1e84896f63173\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4e1a2fad34f94db6b2eb3a161ac02ac5 : SUCCESS in 1h 57m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/22056c5d027b4dd8828f1138f9fa8e6e : SUCCESS in 58m 19s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0a450c9dd2ce4c9092605e21bfc57fba : FAILURE in 1h 54m 11s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4cdf02310c204ffba2c2744b65580118 : SUCCESS in 1h 45m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6328d1987ecf4789b5b4dc57f60a99ff : SUCCESS in 32m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cac4c97f2f2e4ce4bf0c9bb53e8f3c58 : SUCCESS in 35m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86a684279a9f4da6818b92edaa7d617e : SUCCESS in 28m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/410285431b2440b495b5566ae8b479e6 : SUCCESS in 34m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/786bb3128b3f41e797009d085fac052f : SUCCESS in 18m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3d2ae03b6de44c0a8e41341ba1cc0fe0 : SUCCESS in 20m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/005ab70009b248b58c8774e93cfea39c : SUCCESS in 6m 46s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d361d9360f954b78bd7be6f7e752b81c : SUCCESS in 1h 25m 05s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9244ca649f65460aa85a8672c09fc74a : SUCCESS in 36m 38s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/f818333707d348d9afd1a15bb8be3fca : SUCCESS in 36m 40s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f7126666373b48889821e2193cb4d14d : FAILURE in 39m 29s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/51f30481b7954e63944fbd5a2ea72435 : SUCCESS in 1h 34m 54s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2ad47f54b3654b2ca3afe3d23d0ab310 : SUCCESS in 50m 25s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/9e8915d7b82c4b3b9b7f8ce304ce66a7 : SUCCESS in 51m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a3840f523acc4ec9b15b7f704f55c301 : SUCCESS in 1h 27m 39s","accounts_in_message":[],"_revision_number":3},{"id":"32cfcd0e5314792ad74718eff19d6b7d7e25ef8b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-09 16:51:20.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8ad70dc436df079a59a41e5ba57816ca21d47ac5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 19:03:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/429777f1af044de594e9bede89f671ba\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/98d0a26220094005830ef5a48b960df2 : SUCCESS in 1h 25m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/df88b1b6e6114451bf25498efe5dd994 : SUCCESS in 1h 06m 54s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8a4735053fb34e07af19334740085a99 : SUCCESS in 1h 39m 01s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0018fb2ca07c48c8a7706491810031a6 : SUCCESS in 2h 09m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7f96a63e8a994ec4b2f644b36265ea0d : SUCCESS in 33m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/534541b0426d46dfad8f375ae41e81b1 : SUCCESS in 1h 02m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcb3bd1ffb8f417a9bf287c2cbf8e968 : SUCCESS in 25m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5f0463fc10eb4011939cf5641e03b6be : SUCCESS in 36m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bcd277174202483f86a700cc18efc96a : SUCCESS in 28m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/94d68fc952f94b9abe32705cfcf69b5c : SUCCESS in 33m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/648575910b96458d884f3109abb08697 : SUCCESS in 12m 54s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/92ce51518fe1485f8a57e22c41eb0443 : SUCCESS in 53m 04s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7df6dae784404dacb0f8efdc2993103b : SUCCESS in 1h 12m 14s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/29d60ccb39df4e0787f0e8c9177b9fec : SUCCESS in 1h 21m 46s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d5fcceb59c6f48baaf97f7a7eccc3022 : SUCCESS in 1h 15m 35s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2e07c3cd35ba4e28a5ceb9e0fd43bfc9 : SUCCESS in 1h 38m 09s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4d500b676e3748b19e10872f8410548d : SUCCESS in 50m 12s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2bcf4b90c478475d9dc75610c4fc64bd : SUCCESS in 31m 53s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7b5b72408c36439da3ebb9c39f247b76 : SUCCESS in 1h 24m 58s","accounts_in_message":[],"_revision_number":3},{"id":"9b50ae581c78d40f1343f68357c24cdabef45bdb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 19:04:28.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"93fe127cf85fd11ec7ad2463110d85777b1b0b8c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 19:42:40.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fc723ca506d4124b02a52c6e00be41e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dc1ea74c2e134ef0aa8703c33c224fc4 : SUCCESS in 27m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/409d61c55328447887819fa20811af91 : SUCCESS in 19m 11s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8a2a20d59f684888ad04c90fe04cd3d1 : SUCCESS in 26m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9c048eebb088600cfbcb6cbd4f73115e6e79b7c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 21:15:34.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68256e0ef79340b48ca399d635f3fbca\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a76757f2deba48acb6edf2c242eda373 : SUCCESS in 1h 06m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d29371d73d254287abc3c2509723a4e8 : SUCCESS in 53m 29s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/510a3043c9084d88ad8096a52c539324 : SUCCESS in 1h 28m 58s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/cb8f7d0cc98d4cf49cc5d5d0a91a2dd8 : SUCCESS in 2h 10m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f749c72f42742b0824d3325229ce10a : SUCCESS in 13m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/db546cc0cf3c44009951f779125c83c8 : SUCCESS in 21m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/65c113b7b2904c8c999d196b86418867 : SUCCESS in 19m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49659240a4374f4689c8b99e7ea855e4 : SUCCESS in 10m 54s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/456aebb960d3484c986eff05b7af22b4 : SUCCESS in 56m 27s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f1a844d7ed9648aca6389b9d99f1e921 : SUCCESS in 1h 06m 49s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/1d5c62b8ab8a4ce4905e9777ba94617e : SUCCESS in 1h 17m 15s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/98ab01dc7a5c476c881d50c140f0e3a3 : SUCCESS in 41m 02s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ef13801535084b9dbfab7f94e1c11cf0 : SUCCESS in 47m 42s","accounts_in_message":[],"_revision_number":3},{"id":"86df20f61eecc89edd59f28d11e92086b95835f0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 21:15:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"55274da9ba8d6b0ec2671f60e9135dad2026346c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 21:17:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1325ef03a057499db8bbec9eaeb97043\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/638c5f1097ca49cb88ef7599414ece89 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c741f42a424c63ef2cd652a110cd24506a75be00","revisions":{"2909ce3405fbd90c25ec98870447b241610b2ad5":{"kind":"REWORK","_number":1,"created":"2026-04-05 07:25:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/71/983371/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/71/983371/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/983371/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/983371/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/983371/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/71/983371/1"}}},"commit":{"parents":[{"commit":"5cb1eabddbd2f0b19bd3534fd5d590c8cf3b01a1","subject":"Merge \"Reduce mccabe complexity in sync_routers_and_rports()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5cb1eabddbd2f0b19bd3534fd5d590c8cf3b01a1"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-05 07:22:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-05 07:24:54.000000000","tz":540},"subject":"Use constant for ha_failover_strategy value","message":"Use constant for ha_failover_strategy value\n\nChange-Id: I822c66dea14ca05853cfc95d16ce2a9540f1570b\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2909ce3405fbd90c25ec98870447b241610b2ad5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2909ce3405fbd90c25ec98870447b241610b2ad5"}]},"branch":"refs/heads/master"},"04f6f4d591ebc5d42bff457bf9e16e967da3d908":{"kind":"REWORK","_number":2,"created":"2026-04-05 09:56:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/71/983371/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/71/983371/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/983371/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/983371/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/983371/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/71/983371/2"}}},"commit":{"parents":[{"commit":"5cb1eabddbd2f0b19bd3534fd5d590c8cf3b01a1","subject":"Merge \"Reduce mccabe complexity in sync_routers_and_rports()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5cb1eabddbd2f0b19bd3534fd5d590c8cf3b01a1"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-05 07:22:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-05 09:56:37.000000000","tz":540},"subject":"Use constant for ha_failover_strategy value","message":"Use constant for ha_failover_strategy value\n\nChange-Id: I822c66dea14ca05853cfc95d16ce2a9540f1570b\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/04f6f4d591ebc5d42bff457bf9e16e967da3d908"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/04f6f4d591ebc5d42bff457bf9e16e967da3d908"}]},"branch":"refs/heads/master"},"c741f42a424c63ef2cd652a110cd24506a75be00":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-04-07 16:15:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/71/983371/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/71/983371/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/983371/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/983371/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/983371/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/71/983371/3"}}},"commit":{"parents":[{"commit":"54befe6c096275cfe0049a2f1ab1abbee3140c8c","subject":"Use unittest.skip decorator","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/54befe6c096275cfe0049a2f1ab1abbee3140c8c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-05 07:22:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-07 16:15:56.000000000","tz":0},"subject":"Use constant for ha_failover_strategy value","message":"Use constant for ha_failover_strategy value\n\nChange-Id: I822c66dea14ca05853cfc95d16ce2a9540f1570b\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c741f42a424c63ef2cd652a110cd24506a75be00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c741f42a424c63ef2cd652a110cd24506a75be00"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
