)]}'
{"id":"openstack%2Fneutron~983372","triplet_id":"openstack%2Fneutron~master~I9adef8f5c6e49d67326bfc7b33ac5c45af031d29","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-04-08 10:23:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9adef8f5c6e49d67326bfc7b33ac5c45af031d29","subject":"Do not explicitly mention supported values","status":"MERGED","created":"2026-04-05 07:26:55.000000000","updated":"2026-04-08 10:24:44.000000000","submitted":"2026-04-08 10:23:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"983372","meta_rev_id":"1857e7d6024421fde1b543f84878a6c39f0b9c71","_number":983372,"virtual_id_number":983372,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-08 10:23:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2026-04-08 08:30:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2026-04-08 01:42:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":1,"date":"2026-04-08 08:30:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-05 07:34:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-04-05 08:38:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-08 01:42:54.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-04-08 08:30:05.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"b57f14071069af0722d8cdf5180ab22d6157a390","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-05 07:26:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7251fc1d0532f2cd104559fbc8eaa91ff7a6069","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-05 07:34:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1979d543bd3f4cdba23f22d6d75f86f5\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/548c53b6460b414595263aae1ae2e7b2 : FAILURE in 4m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0b28925f36bd4173911b379df3dbba71 : FAILURE in 4m 14s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/20af7b241d9643639decb556a57779ab : FAILURE in 6m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a01f8d1441444886d69f0add48b77e7ac9c8845a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-05 08:38:43.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/040a9791a1ea491193c9285172592e93\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2e999a0e179043b4a57ec9ae57ec9568 : FAILURE in 19m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9877ceae193d49a6a47c95c8db984934 : POST_FAILURE in 22m 29s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/05c72566d4ee4ee9bf1f17323d3aa684 : FAILURE in 18m 43s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/700592380f8543f1b86260ebf73397dd : FAILURE in 11m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/881d196ca2b746ee96e708dfe3098720 : FAILURE in 12m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d084b7cd6e564010a6c27787c69904b5 : FAILURE in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5895bf8555b7458fb82e85ab5ecaf6a5 : FAILURE in 7m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/033fe6d25be045618a702a797c2e3f9a : FAILURE in 4m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/73670397c24a426e9eead5665f619503 : FAILURE in 2m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/af88da4ee9824936b91dd3d2e9d7a688 : FAILURE in 8m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/135e843c4ae4481d81873ce62794faf6 : FAILURE in 6m 58s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f0e64d3474d74fa3a22c4049550e40f5 : FAILURE in 21m 27s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/836ea81e57c24c5da1861a3fbc10da83 : FAILURE in 30m 40s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d1e0a97ca84d44afb0f1d6c273d20142 : FAILURE in 1h 06m 50s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f527cce8469547979e44394cc7b5fc7f : FAILURE in 22m 33s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b988f62a77a94fe08a3a58e7e92141af : FAILURE in 13m 31s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d28dd55849ed4860b037116ec569922c : FAILURE in 18m 02s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/121b7b19aeca44e0bd312bc2d9629ba1 : FAILURE in 19m 31s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2738225bdb194d8da336a41942458e25 : FAILURE in 24m 44s","accounts_in_message":[],"_revision_number":1},{"id":"ac3ba0e2d84d9b74313c0ecd40868ab85de9a6de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-05 09:52:38.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7374713d5535b7ccf3c4ac22de7edf29e0424f29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-05 09:53:47.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c79a85e74027a5c08aee60d1d1871a090d419f7f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-05 10:20:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ef930e83eba421b9159570b8bfcd71a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ae120f1724b544bb9328f63e3a9b0c6f : SUCCESS in 25m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b5666a3ffc63414aaefcad6ac6eb5a4a : SUCCESS in 18m 52s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1d4e9b89103e4bf3b47d79292b968546 : SUCCESS in 25m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6b4f5464932327445c5201981d5b32694b50e150","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-05 11:48:18.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b8c0c52b490c481b91e471fbc1e70f28\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2ece4e8641dc4c76982d4c9114ef5fe1 : SUCCESS in 1h 28m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3ab46e465ba34b4680f939f14ed04394 : SUCCESS in 48m 02s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/81e91f99b9834959bebfcf78b286f412 : SUCCESS in 1h 26m 18s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7e1f1fb47d9c4cad9c21be6d31ffddc3 : SUCCESS in 1h 50m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0184d8184fa543b2bb24a7d1ef455da4 : SUCCESS in 34m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1dbe52757c24484d97c8c1cfdd478ec2 : SUCCESS in 1h 00m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2c63b0aefb943d7b62ac987250ee37d : FAILURE in 23m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/370efde46d5141fc8e80a92dce2d9815 : SUCCESS in 38m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b20d6223d62742d680a067bc35d41302 : SUCCESS in 24m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a20153613cfd4913a805ee6dc953f57b : SUCCESS in 31m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29ac28b26fa94f7fa80ff48582bf15b1 : SUCCESS in 10m 59s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fbd2ec901e7c4361ae896e19d74083a5 : SUCCESS in 1h 05m 36s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3d780a63680741968d23b9849e95fa47 : SUCCESS in 1h 15m 17s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/8ba11f3f35294f14b07c0bb1be538934 : SUCCESS in 1h 03m 04s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/68840937975f47d586cfe5c11d42f68d : SUCCESS in 1h 49m 42s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e5583738443a4c69be1a7ae721a05a30 : SUCCESS in 1h 48m 21s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/fc1bde0c626946d3864da451eb828f96 : SUCCESS in 1h 15m 15s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/9231d4ccf30043c5b67097a375278379 : SUCCESS in 56m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0df1e30b50024834bdf3bab40a5c8152 : SUCCESS in 1h 25m 29s\n\nWarning:\n  Comments left for invalid file neutron/tests/functional/agent/l2/base.py\n  Comments left for invalid file neutron/tests/functional/agent/l3/extensions/qos/test_fip_qos_extension.py\n  Comments left for invalid file neutron/tests/functional/agent/l3/framework.py\n  Comments left for invalid file neutron/tests/functional/agent/l3/test_dvr_router.py\n  Comments left for invalid file neutron/tests/functional/agent/linux/openvswitch_firewall/test_iptables.py\n  Comments left for invalid file neutron/tests/functional/agent/linux/test_ip_lib.py\n  Comments left for invalid file neutron/tests/functional/agent/linux/test_keepalived.py\n  Comments left for invalid file neutron/tests/functional/agent/linux/test_of_monitor.py\n  Comments left for invalid file neutron/tests/functional/agent/test_dhcp_agent.py\n  Comments left for invalid file neutron/tests/functional/agent/test_ovs_flows.py\n  Comments left for invalid file neutron/tests/functional/pecan_wsgi/test_controllers.py\n  Comments left for invalid file neutron/tests/unit/agent/dhcp/test_agent.py\n  Comments left for invalid file neutron/tests/unit/agent/l2/extensions/metadata/test_host_metadata_proxy.py\n  Comments left for invalid file neutron/tests/unit/agent/l3/extensions/test_port_forwarding.py\n  Comments left for invalid file neutron/tests/unit/plugins/ml2/test_port_binding.py","accounts_in_message":[],"_revision_number":3},{"id":"19631d36f728798bcb2669dda6cb22ed81555859","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-07 16:15:58.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"b836d5b83c6a9c906dbf52db6b5438ba0fd3f5c5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 16:46:38.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd07fc7ea8fb47c9841f407512a6e464\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b628760cbc2e482bb5f079ee2853cc78 : SUCCESS in 23m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1fcc3cb10a084457b24bfcea450f6545 : SUCCESS in 28m 33s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8be9210ded41494db8e4b68d1b80efc7 : SUCCESS in 23m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"df8905438d5c34e6d6524ca4e8b687eab74f1304","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 18:18:38.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e382c0ac4f34e478f62186b8a1d26e5\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8a5fa43b0b9547658a72ce43d780c44a : SUCCESS in 1h 33m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6340f83545e1476ca21bdb8289720369 : SUCCESS in 1h 05m 00s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f01e865dba93478fa3ce4b4773dde41e : SUCCESS in 1h 36m 35s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c370b1dcdd1146a7addb36c9677da034 : SUCCESS in 2h 01m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d74f8d97eef64e8cbedcce9e2ba37be0 : SUCCESS in 32m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d68f19ce20354a7d8b5c419647dc7f56 : SUCCESS in 58m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1625b21073e04aa785851caf5e89329a : SUCCESS in 25m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ec5fa5df9f3a42658a4aee762bee4012 : SUCCESS in 30m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0a974d5073434091af4b7413d6c67085 : SUCCESS in 25m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4804b39142384c34b1265d4622d23b55 : SUCCESS in 31m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/308a31c3da9f41278da91d8cff804911 : SUCCESS in 11m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3b7c8d2ebcdb4cdb81597b2367fe7e26 : SUCCESS in 1h 01m 54s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/46819e538d4f4e4cb62d19ae315b7f0b : SUCCESS in 1h 10m 03s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/7a11928c57fb4878a91d655f2f23e73b : SUCCESS in 1h 16m 31s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a6ce06de32874ea0a7891a7c49195537 : SUCCESS in 1h 40m 20s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/39f8deafc07945fe84326d738e6e266e : SUCCESS in 54m 03s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/96a5dfe0683b4610a939cf8c1ee9c01c : SUCCESS in 58m 23s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f7af1df14e954225b894aff2bed07120 : SUCCESS in 1h 16m 44s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ebd0e4102afb49fda742c7ca79db9787 : SUCCESS in 1h 28m 17s","accounts_in_message":[],"_revision_number":4},{"id":"11cb736fa9a4c4596f5016aaf0be6699e1173b07","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-08 01:42:54.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"bcfc9d5c2179cebc6b093e492f683a3c1549a071","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-04-08 08:30:05.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"5bbce3d60db99fb7bbfca455121c847830b52f7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 08:30:27.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a20839beaac445931cef44c752a5a06f5afa3fbd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 10:23:16.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ceb9d04df2db4dc5ac5b4aed307dcb59\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/01797024147f48d09a7ce5de0fd200af : SUCCESS in 1h 24m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e0244517a984459ca47e763bbdfc0fea : SUCCESS in 34m 24s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/95e078814c834a33aa444988da3d05ad : SUCCESS in 1h 36m 09s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/06db08b844f94584b84629ec1f892074 : SUCCESS in 1h 46m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb47900d9e6a4076837f465c2d1c9270 : SUCCESS in 13m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/873cfc3cc12f4508a6a2b12a1bd5ebd4 : SUCCESS in 34m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9b64fe12d9ac4807839d76109eabd021 : SUCCESS in 24m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be5a5134c36e4516858ea1fef7a12e1b : SUCCESS in 13m 24s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3e679f03e7a6484bbe7a2d425c5327a8 : SUCCESS in 49m 22s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/cf9cded534b1403497b8324e16569f73 : SUCCESS in 1h 47m 55s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ba5f1fae61224d9fa2ddeaf5da4752ba : SUCCESS in 1h 17m 50s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/95340e313bec4b64ba52137b42b6eeea : SUCCESS in 1h 11m 40s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/51aa830dc9914435b1d034213d0a9bb6 : SUCCESS in 1h 08m 02s","accounts_in_message":[],"_revision_number":4},{"id":"51105b650b7faae04b6861afd03f2791d6113dc7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 10:23:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"1857e7d6024421fde1b543f84878a6c39f0b9c71","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 10:24:44.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c29dbf17e9c249b0a41786609454af7f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9b58225aba74f3f9cc59e3d29caab6f : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"52d10bfd4a9881da1aeb26d7317eed8615ec82ef","revisions":{"049a4e111f9b5e2337437f455d9ed2e5d1406f37":{"kind":"REWORK","_number":1,"created":"2026-04-05 07:26:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/72/983372/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/983372/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/983372/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/983372/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/983372/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/983372/1"}}},"commit":{"parents":[{"commit":"5cb1eabddbd2f0b19bd3534fd5d590c8cf3b01a1","subject":"Merge \"Reduce mccabe complexity in sync_routers_and_rports()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5cb1eabddbd2f0b19bd3534fd5d590c8cf3b01a1"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-05 07:25:43.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-05 07:25:43.000000000","tz":540},"subject":"Do not explicitly mention supported values","message":"Do not explicitly mention supported values\n\nSupported values ca automatically detected by oslo-config-generator and\nadded to the auto-generated sample config files.\n\nChange-Id: I9adef8f5c6e49d67326bfc7b33ac5c45af031d29\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/049a4e111f9b5e2337437f455d9ed2e5d1406f37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/049a4e111f9b5e2337437f455d9ed2e5d1406f37"}]},"branch":"refs/heads/master"},"4228a377c92e1f96544af0aa39895c5b2b7646b6":{"kind":"REWORK","_number":2,"created":"2026-04-05 09:52:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/72/983372/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/983372/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/983372/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/983372/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/983372/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/983372/2"}}},"commit":{"parents":[{"commit":"5cb1eabddbd2f0b19bd3534fd5d590c8cf3b01a1","subject":"Merge \"Reduce mccabe complexity in sync_routers_and_rports()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5cb1eabddbd2f0b19bd3534fd5d590c8cf3b01a1"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-05 07:25:43.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-05 09:52:37.000000000","tz":0},"subject":"Do not explicitly mention supported values","message":"Do not explicitly mention supported values\n\nSupported values ca automatically detected by oslo-config-generator and\nadded to the auto-generated sample config files.\n\nChange-Id: I9adef8f5c6e49d67326bfc7b33ac5c45af031d29\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4228a377c92e1f96544af0aa39895c5b2b7646b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4228a377c92e1f96544af0aa39895c5b2b7646b6"}]},"branch":"refs/heads/master"},"a5cb9eb1f2bc9831760cc074b9808afc643913f8":{"kind":"REWORK","_number":3,"created":"2026-04-05 09:53:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/72/983372/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/983372/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/983372/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/983372/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/983372/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/983372/3"}}},"commit":{"parents":[{"commit":"5cb1eabddbd2f0b19bd3534fd5d590c8cf3b01a1","subject":"Merge \"Reduce mccabe complexity in sync_routers_and_rports()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5cb1eabddbd2f0b19bd3534fd5d590c8cf3b01a1"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-05 07:25:43.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-05 09:53:36.000000000","tz":540},"subject":"Do not explicitly mention supported values","message":"Do not explicitly mention supported values\n\nSupported values ca automatically detected by oslo-config-generator and\nadded to the auto-generated sample config files.\n\nChange-Id: I9adef8f5c6e49d67326bfc7b33ac5c45af031d29\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a5cb9eb1f2bc9831760cc074b9808afc643913f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a5cb9eb1f2bc9831760cc074b9808afc643913f8"}]},"branch":"refs/heads/master"},"52d10bfd4a9881da1aeb26d7317eed8615ec82ef":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-04-07 16:15:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/72/983372/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/983372/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/983372/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/983372/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/983372/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/983372/4"}}},"commit":{"parents":[{"commit":"54befe6c096275cfe0049a2f1ab1abbee3140c8c","subject":"Use unittest.skip decorator","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/54befe6c096275cfe0049a2f1ab1abbee3140c8c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-05 07:25:43.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-07 16:15:58.000000000","tz":0},"subject":"Do not explicitly mention supported values","message":"Do not explicitly mention supported values\n\nSupported values ca automatically detected by oslo-config-generator and\nadded to the auto-generated sample config files.\n\nChange-Id: I9adef8f5c6e49d67326bfc7b33ac5c45af031d29\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/52d10bfd4a9881da1aeb26d7317eed8615ec82ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/52d10bfd4a9881da1aeb26d7317eed8615ec82ef"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
