)]}'
{"id":"openstack%2Fneutron~983399","triplet_id":"openstack%2Fneutron~master~I0c1f121fd53edeeab198f12d250ca545aa305b87","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"36824":{"account":{"_account_id":36824,"name":"Isaac Vicente de Medeiros Silva","display_name":"Isaac Silva","email":"isaacvicentsocial@gmail.com","username":"isaacvicente","status":"Cloud Engineer working with Openstack on Laboratory of Distributed System at LSD, Brazil"},"last_update":"2026-04-06 14:45:38.000000000","reason":"Change was abandoned"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-04-06 14:45:38.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I0c1f121fd53edeeab198f12d250ca545aa305b87","subject":"Disable \u0027unreachable\u0027 pylint warning temporarily","status":"ABANDONED","created":"2026-04-06 00:07:41.000000000","updated":"2026-04-06 14:45:38.000000000","total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"5b0acd45873967728bb49b6c164d0ee64bf0be98","_number":983399,"virtual_id_number":983399,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":36824,"name":"Isaac Vicente de Medeiros Silva","display_name":"Isaac Silva","email":"isaacvicentsocial@gmail.com","username":"isaacvicente","status":"Cloud Engineer working with Openstack on Laboratory of Distributed System at LSD, Brazil"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-04-06 01:18:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":36824,"name":"Isaac Vicente de Medeiros Silva","display_name":"Isaac Silva","email":"isaacvicentsocial@gmail.com","username":"isaacvicente","status":"Cloud Engineer working with Openstack on Laboratory of Distributed System at LSD, Brazil"},"disliked":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":-1,"date":"2026-04-06 13:50:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"date":"2026-04-06 11:06:48.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2026-04-06 00:23:00.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36824,"name":"Isaac Vicente de Medeiros Silva","display_name":"Isaac Silva","email":"isaacvicentsocial@gmail.com","username":"isaacvicente","status":"Cloud Engineer working with Openstack on Laboratory of Distributed System at LSD, Brazil"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":36824,"name":"Isaac Vicente de Medeiros Silva","display_name":"Isaac Silva","email":"isaacvicentsocial@gmail.com","username":"isaacvicente","status":"Cloud Engineer working with Openstack on Laboratory of Distributed System at LSD, Brazil"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":36824,"name":"Isaac Vicente de Medeiros Silva","display_name":"Isaac Silva","email":"isaacvicentsocial@gmail.com","username":"isaacvicente","status":"Cloud Engineer working with Openstack on Laboratory of Distributed System at LSD, Brazil"},{"value":2,"date":"2026-04-06 01:03:02.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36824,"name":"Isaac Vicente de Medeiros Silva","display_name":"Isaac Silva","email":"isaacvicentsocial@gmail.com","username":"isaacvicente","status":"Cloud Engineer working with Openstack on Laboratory of Distributed System at LSD, Brazil"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-06 00:23:00.000000000","updated_by":{"_account_id":36824,"name":"Isaac Vicente de Medeiros Silva","display_name":"Isaac Silva","email":"isaacvicentsocial@gmail.com","username":"isaacvicente","status":"Cloud Engineer working with Openstack on Laboratory of Distributed System at LSD, Brazil"},"reviewer":{"_account_id":36824,"name":"Isaac Vicente de Medeiros Silva","display_name":"Isaac Silva","email":"isaacvicentsocial@gmail.com","username":"isaacvicente","status":"Cloud Engineer working with Openstack on Laboratory of Distributed System at LSD, Brazil"},"state":"REVIEWER"},{"updated":"2026-04-06 01:18:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-06 10:40:12.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2026-04-06 13:50:10.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"79c9cf2fed5316e86e8f1b9d491937f7f2bca284","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-06 00:07:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"53282a6453dc9700db5754771eef45b0f19b4dda","author":{"_account_id":36824,"name":"Isaac Vicente de Medeiros Silva","display_name":"Isaac Silva","email":"isaacvicentsocial@gmail.com","username":"isaacvicente","status":"Cloud Engineer working with Openstack on Laboratory of Distributed System at LSD, Brazil"},"date":"2026-04-06 00:23:00.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5f456d45ea82c18f5b0f32b5c804583da7de9108","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-06 01:03:02.000000000","message":"Patch Set 1: Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5df5ccb49403079c0591de66c8e542efee88ea23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-06 01:18:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ba4d55a6460489bb2d7da1ed37f8e71\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/139720231e61413e98f571515f5511ee : SUCCESS in 1h 05m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f116e0ce00f4aceaa164b10f6627741 : SUCCESS in 26m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/634fc99f82c74f61ba126266e3a5f734 : SUCCESS in 12m 19s","accounts_in_message":[],"_revision_number":1},{"id":"a9015027eefeefd9e88f07545d6287f6e571c694","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-06 10:40:12.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"80bfe84e9d87556b9eefc6a1782d23620df0a75e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-06 11:06:48.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e9b05a7a6cb5d51573e494f9d3591bb4182a67e3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-06 13:50:10.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5b0acd45873967728bb49b6c164d0ee64bf0be98","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-06 14:45:38.000000000","message":"Abandoned\n\nused https://review.opendev.org/c/openstack/neutron/+/983417 instead","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7d7397d0db3fa303789cb756f5534a6f419ce161","revisions":{"7d7397d0db3fa303789cb756f5534a6f419ce161":{"kind":"REWORK","_number":1,"created":"2026-04-06 00:07:41.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/99/983399/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/99/983399/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/983399/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/983399/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/983399/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/99/983399/1"}}},"commit":{"parents":[{"commit":"5cb1eabddbd2f0b19bd3534fd5d590c8cf3b01a1","subject":"Merge \"Reduce mccabe complexity in sync_routers_and_rports()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5cb1eabddbd2f0b19bd3534fd5d590c8cf3b01a1"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-04-06 00:05:35.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-04-06 00:05:35.000000000","tz":-240},"subject":"Disable \u0027unreachable\u0027 pylint warning temporarily","message":"Disable \u0027unreachable\u0027 pylint warning temporarily\n\nA number of skipTest() directives were added during eventlet\nremoval, which is now causing a pep8 gate failure. Until we\ncan update the tests, disable to fix the gate.\n\nChange-Id: I0c1f121fd53edeeab198f12d250ca545aa305b87\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7d7397d0db3fa303789cb756f5534a6f419ce161"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7d7397d0db3fa303789cb756f5534a6f419ce161"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
