)]}'
{"id":"openstack%2Fneutron~984307","triplet_id":"openstack%2Fneutron~master~If221129daf686b8dda2a63360deaaa8661a34cd7","project":"openstack/neutron","branch":"master","topic":"optimize_address_group","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-04-14 14:46:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If221129daf686b8dda2a63360deaaa8661a34cd7","subject":"Optimize ML2/OVN address group transactions","status":"MERGED","created":"2026-04-13 06:39:04.000000000","updated":"2026-04-14 14:48:00.000000000","submitted":"2026-04-14 14:46:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"984307-optimize_address_group","meta_rev_id":"986ad1962065684969102b7b07213bf94a262f85","_number":984307,"virtual_id_number":984307,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-14 14:46:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-04-14 09:58:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2026-04-13 20:26:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-04-14 09:58:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-04-13 14:58:33.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-13 07:12:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-04-13 08:48:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-13 14:58:02.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2026-04-13 14:58:02.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2026-04-13 14:58:02.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-04-13 20:26:12.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-04-14 09:58:11.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"aa2370d4ca4218c39fa4222c3e1736d7560bd3fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-13 06:39:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"84520b0e7998e495b63b736405ec992f2cf49157","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 07:12:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/475eaec816e640819d0d060e1162934d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d8bbe5aff58448a59415448adbab71af : SUCCESS in 23m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8b229dee20334d1f8a46521754590d64 : SUCCESS in 19m 07s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7bf47e235eb7408093de459df523c417 : SUCCESS in 28m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5222dcb2b0623b6669820909d42c917630ff5e9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 08:48:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d156a90b3c94f33ba028301881d39aa\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e3fef96c933645068084691e18cd130b : SUCCESS in 1h 25m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/150d85a62edb4752bfcf03617ca8bff7 : SUCCESS in 44m 36s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3804a8e8a5814408918907336dafdcf2 : SUCCESS in 2h 00m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f103e53b9644fbca257ac9713b3a6fc : SUCCESS in 1h 06m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3c138b8ccea4a32bc5cede5b594d352 : SUCCESS in 26m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cdc5e9025c2e470cb19dd5a05fba8e91 : SUCCESS in 23m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ae37bce5d0d74c0d9756007f572836e1 : SUCCESS in 18m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/65c9ddb2fb0b4578bfa23776eac15bf8 : SUCCESS in 24m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73e7ad5a4407467a95f9c3a77fac72a8 : SUCCESS in 9m 50s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b1c77e6ba9694fb8a812f7fce365b1a1 : SUCCESS in 48m 07s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/1e914d05fa4842d39dbc2bf073af9850 : SUCCESS in 42m 46s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/96092dc7465b4abd946140cb5528be98 : SUCCESS in 54m 12s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2b8260485e4e4bb59f259623f2c9641d : SUCCESS in 1h 03m 02s","accounts_in_message":[],"_revision_number":1},{"id":"7df5cc54f8b3015c0d8182e3b2f6dd38c6f4c0ac","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-13 14:58:33.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"c8a901678899ee4e97a82e8c52fa942da7bcf15b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-13 20:26:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"31f5922eac72d5cc4f0a9200681a972e54be2802","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-04-14 09:58:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8169f31474880840caeb37180025b89682995133","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 10:07:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3d8cc98fdf350797ce3a00c02828a04fbaecd09e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 14:46:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/721ef5de77964437a21f5805f1a68c64\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/990ed246b79442f8840733f978ebf515 : SUCCESS in 1h 18m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c765edf7e88b4d819c7f4be1e7ac6b95 : SUCCESS in 1h 07m 46s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0f17547a06fb4b5a89ab4fa211a0f306 : SUCCESS in 2h 16m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fbc537700a24ec3a6670acbb6b7289d : SUCCESS in 21m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7c2d4554a1bb4a8a95f1075843d2eed0 : SUCCESS in 30m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/63c4b9e202d34c3bb28d3ad4588ff813 : SUCCESS in 19m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a517155be3dd4cefa772538cabed3690 : SUCCESS in 13m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/18b23a68b15c44c1b3feef8a5f16b83d : SUCCESS in 45m 47s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/e1bfe16de04742259694d3494093bc0b : SUCCESS in 1h 16m 45s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/aa9bc9077644452ea73ffbbdc7cd2d71 : SUCCESS in 1h 11m 28s","accounts_in_message":[],"_revision_number":1},{"id":"5e240b1dc0f2c4403a4b33d0381ae64da44bbe1f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 14:46:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"986ad1962065684969102b7b07213bf94a262f85","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 14:48:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9b63d6c0d3d4c4188357667adcb85f5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0a21ff9a91e46bfa25e044a6e7bc143 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"25129e2f6b07aab410ccdbdc9d078a7d6c668bf5","revisions":{"25129e2f6b07aab410ccdbdc9d078a7d6c668bf5":{"kind":"REWORK","_number":1,"created":"2026-04-13 06:39:04.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/07/984307/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/07/984307/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/07/984307/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/07/984307/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/07/984307/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/07/984307/1"}}},"commit":{"parents":[{"commit":"cff9278d1cdb86b5cdacee64b664de57f20462be","subject":"Merge \"[OVN] Update the parent hostname in the virtual ports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cff9278d1cdb86b5cdacee64b664de57f20462be"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-04-12 07:32:59.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-04-12 07:32:59.000000000","tz":120},"subject":"Optimize ML2/OVN address group transactions","message":"Optimize ML2/OVN address group transactions\n\nMerge the IPv4 and IPv6 ``Address_Set`` operations into a single\nOVN NB transaction in both ``create_address_group`` and\n``update_address_group``.\n\nPreviously each IP version opened its own transaction, resulting in\ntwo OVSDB round-trips per address group operation. Now both address\nsets are created/updated in one commit.\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: If221129daf686b8dda2a63360deaaa8661a34cd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/25129e2f6b07aab410ccdbdc9d078a7d6c668bf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/25129e2f6b07aab410ccdbdc9d078a7d6c668bf5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
