)]}'
{"id":"openstack%2Fneutron~984356","triplet_id":"openstack%2Fneutron~master~I1c773a445a65414dbb3df325a3a61b94d050efc6","project":"openstack/neutron","branch":"master","topic":"bug/2146803","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-06-18 21:56:43.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2026-06-18 14:35:58.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-06-18 21:56:43.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-06-18 19:43:32.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I1c773a445a65414dbb3df325a3a61b94d050efc6","subject":"Add security-groups-default-statefulness API extension","status":"MERGED","created":"2026-04-13 10:24:49.000000000","updated":"2026-06-18 21:58:46.000000000","submitted":"2026-06-18 21:56:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":24,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"984356-bug/2146803","meta_rev_id":"f6b3e9a8c8b2d39cca90755f7c42de6704b3080a","_number":984356,"virtual_id_number":984356,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-18 21:56:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2026-06-18 14:35:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2026-06-18 19:43:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2026-06-18 19:43:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-13 13:14:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-30 04:48:12.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-05-08 17:23:02.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-05-11 10:17:24.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-29 06:52:14.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2026-06-09 20:50:18.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2026-06-17 08:07:06.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"4f53e4c0954a6616f97e4ceea5d4c129c4be1a53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-13 10:24:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0acba07c703f49539d3dd60e30a45d4cad81d617","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-13 10:24:58.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"0eac53b12a754c134c4a130c151b833fccc1d531","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 13:14:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3c7a1f8047584a3f830202ade5412c1f\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/bfb173af9cd443ceb888130506525402 : SUCCESS in 1h 14m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/134eb627e8a14005a04dd4cd48f8e993 : SUCCESS in 1h 03m 48s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/88ad26928014475580543b9b6306308e : SUCCESS in 1h 22m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bb3e4849158d409da8e7e30f2ecc774b : SUCCESS in 1h 51m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6c0780429eee4fa295f8023aa320cbad : SUCCESS in 31m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b44a123f429e4469a8a811cf2c6b703c : FAILURE in 1h 04m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3edd62e773d04ede9917791abbb58a53 : FAILURE in 6m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8bf22f40ea3b4acc96fcf7b34e9e04d3 : FAILURE in 37m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b9a9fc1264a2483b994f331f3823c3bd : FAILURE in 29m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/80f0d831cfba414fb845d7e80a9a4f17 : FAILURE in 26m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b602fd8abf64320992732449c199b54 : SUCCESS in 15m 17s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b3bf720950b646058f7b1e669950e739 : SUCCESS in 55m 44s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/38c67e43bf674479942c0f9933be7711 : SUCCESS in 57m 24s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/c8c0dc318e344346b461e4440f4651bb : SUCCESS in 1h 03m 31s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/222999ace2134f13aa4656732bd85a76 : SUCCESS in 1h 41m 18s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6a1abc2ede8549e286cd199a0ac83932 : SUCCESS in 1h 03m 49s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3226148ff93b424280a0460046fd571f : SUCCESS in 1h 16m 03s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b723fba9a1d246afaded4404e6e306e4 : SUCCESS in 1h 27m 31s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5a2b61b17cf24b14bd1e93d14110ae29 : SUCCESS in 1h 54m 53s","accounts_in_message":[],"_revision_number":1},{"id":"83d8f12f7979fa1527efe1b9e3649fd7f6a0b7e6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 16:53:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a278a26f0ec4ad0bb7ea1131d154a9f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3f2959a5d90d425383951959f99bb9e0 : FAILURE in 23m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7ec47695fbbd47fe91e1b8721328608e : FAILURE in 18m 38s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1c01f195ed3d4014a981df0895c24bfa : FAILURE in 34m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"74a8825c7a924133d9b39949ff998472d1556785","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-23 07:06:42.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"642771de99ee6cd905a696d532e0f7fee7d1cf37","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 07:50:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03ea836b4c6b4164be600d08cd5d1524\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/69cc05995d2a4a269a4156c6da76a16d : SUCCESS in 26m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d645605daf30437dafaea5dfd23d595d : SUCCESS in 17m 19s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3509653f191843d28d6fa389eb58688c : SUCCESS in 34m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e152e371ff2fb723762dc106c3425e139ff7fbcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 10:13:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4f60cd5d03a347eb87e5a006a5a15d01\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/461ae33ff6c84167892ef862ace07c07 : SUCCESS in 1h 22m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8f8a5319241f4c4a9e65d43e5145f2e1 : SUCCESS in 1h 02m 39s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d1efa9ba6be84e4da5ee941beb290ecd : SUCCESS in 1h 25m 53s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/02612405c4834aa9860deecb3dbc216c : SUCCESS in 2h 10m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ed3482fd24544ed5aff7a2e0d4395679 : SUCCESS in 32m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/024c04080c054f6281630cfa99983ede : SUCCESS in 1h 04m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df6749ce0a964802a91c3ff6175fae66 : FAILURE in 6m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b56cacc962ec487996734de27c6b6b19 : SUCCESS in 34m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/73f36f5c304740a08e052382d4e6726f : SUCCESS in 33m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4fba373d5cfb4231bbd101c40e641ea6 : SUCCESS in 19m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/19336886f68d4271996541509a0eb1e7 : SUCCESS in 34m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ee1841647044be4a1761a5e9f8f6a6c : SUCCESS in 11m 41s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2848d8e385ba43c2b24acffc955a91e2 : SUCCESS in 1h 22m 08s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d4ca18f10bb743f2acd035641382af24 : SUCCESS in 52m 00s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/57b5b4c4d136486cb36ae5d5c70309c4 : SUCCESS in 59m 14s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/16d75d61ae7a447b9110fa13c196e03d : SUCCESS in 1h 23m 24s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ed38a2b9d937475f82f5abe0e5fccbcb : SUCCESS in 1h 02m 57s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6e27fb8b4c38444f8792be2760fe82ec : SUCCESS in 1h 06m 55s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3f4c2086a1eb4ce5a44d709647dda4ce : SUCCESS in 56m 09s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3c26f74c98ca45b8af9cbb24ffa5c171 : SUCCESS in 1h 23m 18s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/d7b7e90288cb4056b1031995b7c22943 : TIMED_OUT in 3h 05m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5c70cefd2ad7f7fcc7c73385bba4e27cf7f217c8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-08 15:19:06.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"9542ce2ea0177aea23d3338b21e6b0d773d63b4d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-08 15:19:12.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"b1661fd8801fc58a4a562aef1476b9436c0c0ad6","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-08 17:23:02.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"d19cfac9ec035480604092f8b80e9a44acf70291","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 17:28:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c19a7c8e85c4c829989d25ac3b52914\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ad16d059a801445fa629c049ce60dd5e : SUCCESS in 53m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7cbf175c84d546e4990995817b69d76c : SUCCESS in 18m 36s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e4f1b900585e4814aaf81a2c0036201c : SUCCESS in 23m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6cfab881f89aeb3fe266c2f9eedfe84e2b4b0b26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 17:48:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22e5d99bb97d426ca9f519b4a6c26fd1\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f30f1ca22f1c4ffd8205a7f4235466e9 : SUCCESS in 1h 32m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/997218e60af04f4faa646fac6de0771b : SUCCESS in 1h 08m 45s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c966d3c69efc40cfb9e1827081e0805a : SUCCESS in 1h 40m 51s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e1ed1b26b3d14592981b8f3adce598f3 : SUCCESS in 1h 54m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/685c299b63ab4e19a984e45eb18a4758 : SUCCESS in 41m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3ddcbe7bc1674cb18a46f0711ad27a79 : SUCCESS in 43m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d22ce9be7bea43e49da3585251f43e2e : SUCCESS in 12m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/841a2c5bee384493a4dc255faade210b : SUCCESS in 34m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d09d9458c7c0458897dea5e46fba6ea4 : SUCCESS in 23m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/50eaf0347dd9403f895dff5e2e280e70 : SUCCESS in 26m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c0d0656e0c434423b6a917e3e2c3858c : SUCCESS in 35m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62dcbc17e3ae45f380945b64936fd235 : SUCCESS in 13m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a92a93214fd4836bf286ca0883a6311 : SUCCESS in 18m 51s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9f9840df436c4ecdab15789a4a3f291b : SUCCESS in 1h 38m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c9b8e20a21114c2a9692715783c91efe : SUCCESS in 1h 14m 27s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/5257c41d131a491d98349314e3ca51fc : SUCCESS in 1h 06m 16s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/fce06295904c489c978b4369814ca9ab : SUCCESS in 1h 58m 09s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0f123b071e4d4e1ca93313b3b3c526fa : SUCCESS in 1h 13m 35s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d9da55d779e84c069d58197ae6eaa61e : SUCCESS in 1h 02m 37s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1321fb311bab4415a45a3c2d7bf6aed3 : SUCCESS in 1h 18m 41s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2f0b701375534cc7a5982e1f22145729 : SUCCESS in 1h 02m 50s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/18321ccdee2b41a28c922f25ef2e7a68 : SUCCESS in 2h 21m 55s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":3},{"id":"5b406664ab63fdcc597e145db0d67103fc789383","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-11 07:56:44.000000000","message":"Patch Set 3:\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"b8153290a93f74f60b6458b16fa8c94f678e5f07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-11 07:57:01.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"ac6fc8cb240c8ca5c607da6a197475c315aeff4b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-11 07:57:05.000000000","message":"Patch Set 3: Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"a2cc3d08b26f6248f44d0defd3e8c552d7a0ef61","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-11 07:57:15.000000000","message":"Patch Set 4: Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"532c328b56d911e2edd336daabe9dc002102a110","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 08:34:54.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdb436058cf844ea9e058f2e4f10a099\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/afca0179abcf4d548b80c3dfd159b026 : SUCCESS in 35m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3de98166e2374cd88d56353cdbf8e384 : SUCCESS in 18m 10s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/cb05cab3b7f541b3b0973a4e6c9751d3 : SUCCESS in 25m 40s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e2fd27d5b95edec6708c2bb0a91a7a0bc69e885e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-11 10:17:24.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/56/984356/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"2c6f7b8f3f9e3ebd8c737b2eb59950bb33f83d42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 10:30:24.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b46358ac7fbe4f69a89f46d1a8b133c2\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3a9254c89c9f449a98fe04c1791060d0 : SUCCESS in 1h 27m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0ebaedf11a734c6fa8197629ad8ceebf : SUCCESS in 58m 23s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f084bd4d7f094c1eb88af30b5e069584 : SUCCESS in 1h 23m 32s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fb0f0302379b439cb676f07c5ae01746 : SUCCESS in 1h 58m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/11e1072e1a91464ca46d1d255b247e7d : SUCCESS in 33m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32aae9acc96f4e56a07c4ad207ba7877 : SUCCESS in 39m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ef0c645bd144968affaf38ab3c6353a : SUCCESS in 10m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a69d8e95afe143da95655114c976898a : SUCCESS in 35m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2830967116f243fa84b490c36f2c520d : SUCCESS in 21m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/915c8faf71c041ab821c137526b6fc6c : SUCCESS in 24m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6f40bf6c2ef84cb3b0f0ed998fc36ed7 : SUCCESS in 34m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8983de86f7004aa283c2035704060b39 : SUCCESS in 10m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a8371dc525e4bf8bcd6d4e4c8be2aa8 : SUCCESS in 11m 09s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8e98c2bf8d324fc5852409b3f44fd5da : SUCCESS in 1h 00m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9ac46d391fbe4c428ab5e575cad0edf7 : SUCCESS in 41m 11s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/66a1005f9248466ea553135a9a6569f0 : SUCCESS in 57m 02s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0a28be10af2244418e731b18488a5e0f : SUCCESS in 1h 47m 46s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1532de38951c461c8843bd1daeb032fb : SUCCESS in 1h 48m 12s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0f5c13a822fd4941a0683c9d313019a6 : SUCCESS in 53m 25s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a52f314b24d344e4addb7d89596413a3 : SUCCESS in 1h 00m 28s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1c96bb08ecc549f89ec7eadf3eb030b4 : SUCCESS in 1h 28m 15s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/803ff01bfa4f465dae0b9295ee2c8e46 : SUCCESS in 2h 20m 01s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":4},{"id":"dfbef7609e1bf6a2e02bdb13522040ea43760848","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-12 09:07:13.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"1187fd585a6b1ff32c36582700bd1790ea8661ca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 10:55:41.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06a9596bd54d4876a468a85a9a0e02c4\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a4d541a0a4d245389a78462a66d7a1a4 : FAILURE in 51m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/67828ea153e74e78906f88c9fdb53c62 : FAILURE in 20m 00s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b432a2102e4749fd935e2db1cb00df40 : FAILURE in 39m 15s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6c659dd46d5cde9b7afb1cb000ec235cef3f5d12","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-12 11:13:52.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/56/984356/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"9649d0a844cbb77d2cf6ee460bb94b5cc90ef5f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-12 11:17:13.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"eaf524f33ece8618a13e55924ba12d6783ccada3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 11:57:28.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdbd4f700a76450498cd9c97a2f62ba6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/83686697d1c747e29f04210cb6c9b841 : FAILURE in 25m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ca29691dd41d4c34bd0ff96e9d08250f : FAILURE in 26m 09s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/41f201e7e1d44c34922e793145785007 : FAILURE in 25m 49s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"791cec1a40483f79e061a78c37394f46acac109e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 13:26:23.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/425df6ff60a247ec994d3c48892ab5e3\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7d31efa7a0ca4015a730c2ef04437797 : SUCCESS in 1h 01m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c4bdc34469384be2af8256ebfc5f24e2 : SUCCESS in 47m 20s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8a08e1591da9487f92e9ad96db8e4f4c : SUCCESS in 1h 43m 58s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ff88cd69355b4b4191a965f17151d3fc : SUCCESS in 1h 56m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d636d9a1313e46039c4ff1eea712375c : SUCCESS in 32m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d50b5032a26b4b1a9d3bb6cc98d18e00 : FAILURE in 1h 08m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d518e63f24e24e159086507cbeee857e : FAILURE in 7m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4e7afebc0a9140c28bf052556a55d9ed : FAILURE in 27m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9c06b9ec875b40959406c2cdff7e2108 : FAILURE in 33m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/48b5407d6a8f429fa24b05b51e6894b6 : FAILURE in 33m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/40b2daa0bf1746d991de8dd5c1fe9e76 : FAILURE in 41m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81535720c9ec4e9981c4924be085b8f8 : SUCCESS in 10m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eafcafb76310452da38ba9477d048fb4 : SUCCESS in 10m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/10668c9b75a54a1e89db7eb5c436b597 : SUCCESS in 1h 30m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e2c607f98ac0488ab1189cb2428fb7ad : SUCCESS in 45m 23s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/07b99e48f585449eb073ee8e72684d9e : SUCCESS in 1h 14m 29s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/41caa888da0840c283666e4bd4a450b7 : SUCCESS in 1h 53m 38s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9370778a79bf4b9a80e6a61978ed9046 : SUCCESS in 1h 22m 23s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/96b440a2e89544ee8dfaf25e5373257b : SUCCESS in 52m 36s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/34462ab4d2a14f8680a65c12d127fc8e : SUCCESS in 42m 54s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/07ad6aa740dc45adbf6414db414d7401 : SUCCESS in 49m 27s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/c64a2b4b4ef74177afceb578b2107121 : SUCCESS in 2h 03m 21s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"aa28621543b51cd99f77998e503a2fa3826ad9f4","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-12 21:36:26.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/56/984356/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"35a730c8c8e5d612c1d5187f0543333741b791ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-13 07:45:51.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1, Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"2e288d24fd0b26ee5bc15d5529e6f4aa5ce3d90d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 08:43:34.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29996cb92e894b8aa6310112b0edb6ae\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/309e51e173da408688d628b30c260e9a : FAILURE in 56m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/45d98127135141ddbd1ac209931972dc : FAILURE in 40m 03s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/277a89383f9c4424b225515a5aec08a7 : FAILURE in 54m 16s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"bd91366fbd37435bf77286c76a5283704ea2154a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-13 10:15:23.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/56/984356/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"b5391dff03a8de96c1a62871907079894dfc607c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 10:35:46.000000000","message":"Patch Set 7: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d7fc347a905645339aba12981e57ef59\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/14864cfc76c04f04938da88b6d696850 : SUCCESS in 1h 28m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8be32a14bf754930b8d91dcf228207f6 : SUCCESS in 1h 01m 36s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6f1b4e6de74d4208ad292c65f8ef4517 : SUCCESS in 1h 34m 09s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/aade88c11613471bb6c3c507b7ad3c1a : SUCCESS in 2h 41m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ab6d65d4b91d4814be3dfa7cb67866a2 : SUCCESS in 31m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f097d58015a54ddfa4a6d3521ed2dcee : FAILURE in 1h 03m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f9612c73dc14baabb2554e7af675a5d : FAILURE in 6m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/427ec95d1b6c44f7ad3f0266f0a59519 : FAILURE in 28m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/de4ed6a7eff1400788423704a2327a84 : FAILURE in 25m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3be62f9825d44c53971c3df53aac4d74 : FAILURE in 28m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2498f7926e554f54b5cfe8df73f85ddf : FAILURE in 28m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/426e130de69b40a3a82656f853dc2887 : SUCCESS in 8m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec38ebd564614e9c8d1a99ea68bddab3 : SUCCESS in 18m 29s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e51ddedc938f48bcb94b65412a9c8bc3 : SUCCESS in 1h 19m 07s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ccf3d091fb03449385a1f85c370cd448 : SUCCESS in 1h 12m 10s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/40515052043f458faa9b90b3da90b6cf : SUCCESS in 55m 00s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7dfea101e21c453882a3110cc272ce10 : SUCCESS in 2h 04m 54s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/508ee4cf00574cf290a42b31ead34c92 : SUCCESS in 1h 45m 07s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b0aabccc55a54ccf8e8a31723dc721e3 : SUCCESS in 1h 14m 01s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/09c2e4fe55c345558dae2504ab7b947a : SUCCESS in 1h 26m 13s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/eeb22a1736ca4d148a4a1c8f670bbc6e : SUCCESS in 1h 22m 30s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/ea4b77018a2a42c384ba57272404f9e0 : SUCCESS in 2h 44m 01s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e3a1f759e157f6fd514b7a1adf1f7b36e9796ddc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-27 20:50:40.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1, Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"b739d5267969468af3c0686c325943adfc3d75b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 23:13:22.000000000","message":"Patch Set 8: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/844d96e81556438394cb6165bf682c1f\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d3ac6dd61e2f4c608005008f96b20cfb : SUCCESS in 1h 21m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/37bf36363ed24f9c945c82dd77bebe26 : SUCCESS in 37m 09s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/dde6f2c93f834d47bd8639d7cdb001c1 : SUCCESS in 1h 36m 23s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/98a44b54ccc44ecfb594bce9732dbf63 : SUCCESS in 2h 17m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2ab894ad0b33455d837a54901dc71ae2 : SUCCESS in 32m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/95cac49aecc74659b8b252ccddb2c00b : FAILURE in 46m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaeef063ef0040b899051f1597672b80 : FAILURE in 7m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a3bc8c0190f04f2ba2b2e3f23a2a09e9 : FAILURE in 24m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b32bc562eabc4c6ab0e5fef50083c911 : FAILURE in 20m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5c32c3ff20644fbbb85efb0560fec1c0 : FAILURE in 28m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f7bb0b3c03843e7be1a38e9cc12ac04 : SUCCESS in 7m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/43b2095edbdc4646b98994f327e3c5da : SUCCESS in 10m 16s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/56753a2ebaae4e9aa1757c9ae4d37335 : SUCCESS in 55m 04s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/eed47ab4404e40a3b13c7e725c11f248 : SUCCESS in 1h 07m 21s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ad76646169624933ac7c805d942fab0d : SUCCESS in 41m 06s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/cb54cdaf323a4feb9eb59933edef83a0 : SUCCESS in 1h 25m 11s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b1d17d72556544638487de7722cbfaf1 : SUCCESS in 1h 00m 17s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/539f7c5f39b34be8b5ee0c9dcfd17a57 : SUCCESS in 1h 01m 09s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8316574adebe440ab0feafaaadda3791 : SUCCESS in 1h 10m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7d986d40a73d410192aa0e193a3903b6 : SUCCESS in 39m 59s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/83d6c7d0258f45c19ee913e171cb832b : RETRY_LIMIT Host unreachable in 45m 51s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"717552a925f90d90c07fb36a63f576f135cc58b4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 00:38:24.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8fb678aa9534bfa910f112ce08ec736\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/04d429407e97479984102d347bd3a4a3 : FAILURE in 53m 23s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/dab0226f623f486fa7ffc82765f57f81 : FAILURE in 44m 06s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f5b578e0372a4105b955fa56a07c4144 : FAILURE in 53m 33s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"777d5e32b2051b396f5a79f234c52bb16c70aad2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-28 05:40:56.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/56/984356/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"9ea73827bbb44b6d87d3dc1bf75a54d25520e2c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-28 08:04:35.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1, Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"d85f46b5cb7ff6a86510c6755ae08104c2e2ac2c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-28 09:06:04.000000000","message":"Patch Set 9: Review-Priority+1","accounts_in_message":[],"_revision_number":9},{"id":"bf746169e598ebcbad126c4cd1ed8694a2685c49","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 10:35:02.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1f27fb35dc94255816c19c9d26e45de\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2ce49e97a47a422a84a123c50615d460 : SUCCESS in 52m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/12fb7baf5ec946f78dab9efe20831605 : SUCCESS in 42m 33s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/796fd4dcbfaf49d5a5ec4f8eb1c03b26 : SUCCESS in 37m 06s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"a46ca9251fc3a4dd6fa48bc91781de2fdfe631c1","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-28 11:02:14.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/56/984356/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"3cc044f065017e37f15d3943e3726810e1b486f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 11:29:42.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c8f6c52352141deb3facfafb6d7008e\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3a2076b3c88945d3b7ea52a59f64ff9f : SUCCESS in 1h 08m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7b105d63188c444097c1bfe56cdc53e1 : SUCCESS in 1h 06m 46s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/bbc6bde27c09488da6cf8206dbc66d02 : SUCCESS in 1h 30m 27s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3cbe7463d2c64918b2e5583ce79d0c52 : SUCCESS in 2h 33m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d170354c10184a9e89e6d29ba42b8f63 : SUCCESS in 31m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/675179dc2992423ea29860f1b0ecf7d5 : SUCCESS in 48m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/993e20deac4d4931971d0b87ad36213d : SUCCESS in 10m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6c8127ba5a62483ebfd1fd1081e9df21 : SUCCESS in 25m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9e1f40429c3e41eb9c7b4c84d84ea2fa : SUCCESS in 21m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/264978c1b9dc4e7daff3d8ae064cb182 : SUCCESS in 31m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fa921ee33ac4a5d94e954690d4919c0 : SUCCESS in 11m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ab70b33a58d4aadb4b0efb18222f454 : SUCCESS in 18m 31s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/505b683f174f4fa0a8e2255fd99d2fab : SUCCESS in 1h 09m 10s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8f0bf1feb2b249f4b8d07398b2e442fa : SUCCESS in 1h 18m 12s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/b3c6e1a2718f4a349155cb379b3c5d93 : SUCCESS in 54m 41s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/67600e3dab3b43049b5879f235583db4 : SUCCESS in 1h 52m 01s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ec58a9a295d048408cdaa08a6d68ef65 : SUCCESS in 57m 05s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ea91060ed60444a49ff44bda3fe0f8a8 : SUCCESS in 1h 14m 11s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/253370c043cb432a9e3e2c944a61fc89 : SUCCESS in 1h 11m 01s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e8bde1efc0f54b51933e97a4fd7b84f2 : SUCCESS in 1h 23m 15s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/ca273327a5d148b5b328f695f7cce596 : TIMED_OUT Host unreachable in 3h 18m 46s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":9},{"id":"281504ad122a2e38b91cd82d75d15ba5c8e625c9","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-05-29 06:52:14.000000000","message":"Patch Set 9: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":9},{"id":"0cd6c21bcc7a4776aaadd81c7b457d0e02039a16","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-29 08:05:19.000000000","message":"Patch Set 9: Workflow-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":9},{"id":"8882b158f7a0bc261078cd9aea1222b1daf6f7d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-29 08:45:41.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"3eff651cee669aa3aa8c85ae6cf3958d107f9047","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 10:00:09.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/466b4f86793a4d65bcae5f3252803ef4\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/29d76f91fd144da2accb799cfc97b894 : SUCCESS in 35m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/564170b5f18e4efebc30a49d0ac2e1ba : SUCCESS in 37m 01s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/89a3ed49bca64b0ab8d877398afb37d0 : SUCCESS in 52m 27s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"5b6b0575a1e3676bfd8377197e818400dc393182","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 10:38:53.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a581cc2dae1450c833c66ab368a0bbd\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/345da6725bad4d1c8193801922da119e : SUCCESS in 1h 29m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3a7cefd10f8a4d18afc1d46d1b481f52 : SUCCESS in 53m 09s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/618a9ff4fec24e498cebacb47aa53a49 : SUCCESS in 1h 51m 10s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/90f96cf94bca4e88bca19c4203dbc9fe : SUCCESS in 1h 44m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d9ff3ffe2f4d476ebe7c508f33bb31b6 : SUCCESS in 33m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51bb2a24d1624289b9b6fb07e3a61e2c : SUCCESS in 58m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e81f6d6a9425434bb82dc1a53950a9d2 : SUCCESS in 12m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/50ae2876fe824010a0147d7e32ea63ad : SUCCESS in 25m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3d7a3bd01e0a40abba70cf06bda1c073 : SUCCESS in 19m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8737998fbf3846989345941af46e382c : SUCCESS in 22m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f704248f46904a31996ae74c6e15f819 : SUCCESS in 12m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef0ad6292d424fc7bf21e590b4337301 : SUCCESS in 18m 42s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b5abd5effbd0429d874d6a944fb8a235 : SUCCESS in 1h 24m 58s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8225dd89ba2f419d8147280a665bb1b1 : SUCCESS in 1h 02m 23s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/987e682a3481453c81dc13c7922e9ba1 : SUCCESS in 1h 04m 46s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/20b9d45af70d45198188bcb938516224 : SUCCESS in 1h 14m 53s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/93757153e2a941118c00b8226688fb7f : SUCCESS in 1h 43m 28s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/41c2c15b3caf4c7e9cb7d170bc80ffde : SUCCESS in 1h 05m 11s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e89b548ec549477a8daf554605a15a34 : SUCCESS in 1h 00m 55s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/39741d40e9b945d28a1e102d0279828f : SUCCESS in 1h 17m 36s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/99dec071823b4ee6b3985ea29d7ca586 : SUCCESS in 1h 27m 50s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":10},{"id":"ec6daa1c09bc31575f4b27037d36aaf7cbdf8819","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-29 13:10:28.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/56/984356/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"200eff0def1685fe28fb82939bb1cc69bcbffb73","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-03 14:00:53.000000000","message":"Patch Set 10: Review-Priority+1","accounts_in_message":[],"_revision_number":10},{"id":"8fe0d68c85ad3c6343bf566fb808cb8370320204","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-08 13:20:19.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"327eeae3468afcc956915852ca995b5bf12639f0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-08 15:07:19.000000000","message":"Patch Set 11: Review-Priority+1","accounts_in_message":[],"_revision_number":11},{"id":"8b172c3382161e085254ab716bf518bed74f974e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-08 15:10:04.000000000","message":"Patch Set 11: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/56/984356/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"6178e761873398d9c37ab989d872d7b762f1bf6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 15:59:06.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1033279f38e146149434ba7f8a47bea0\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2a52bb8265df4732920bc9ec9fc89dcf : SUCCESS in 1h 13m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7af12c602eca4c44937d006797af065c : SUCCESS in 1h 17m 34s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/000008cbff7840dab4f68750a718b3f3 : SUCCESS in 1h 36m 45s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fdcbe67fb1e1470ebc368286224a1b72 : SUCCESS in 2h 22m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/822da9526f154323b90ced5370b02dab : SUCCESS in 45m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4eaef5689b1b4d47b389b9ee72bec874 : SUCCESS in 51m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cd127d9a0a54a17a29e6a66c7837bcd : SUCCESS in 12m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/984f628b0b3d46dea56ac6b583a521bc : SUCCESS in 32m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0419dbd48f044b6ab49bb16eeaf44763 : SUCCESS in 25m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d699e3a4dc2f40b5af885ee0e306fc11 : SUCCESS in 39m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b676cd5f8a24355af3b6e60790d8116 : SUCCESS in 13m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99a588c79f3c4f27bec19e25821d6469 : SUCCESS in 21m 15s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/07a8a7bcfb424937982b6273f07ac6e0 : SUCCESS in 1h 04m 27s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b2a7c877b77e46909164d573b57b163d : SUCCESS in 1h 18m 54s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/091c82ab80b8477fac2fb35bb37de32d : SUCCESS in 1h 14m 16s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/01de5c0b511847e3a255887099754432 : SUCCESS in 2h 05m 51s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ee36db5b0a2c45c1b771f8d2bf570aa6 : SUCCESS in 1h 41m 51s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d4297935ee33440aad1d4fa23bae2626 : SUCCESS in 55m 48s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/59b53694770a4eed867e198a6825fd51 : SUCCESS in 1h 13m 54s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/84c68b5341884ca8a07b81e75dee229b : SUCCESS in 1h 21m 29s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/60ea95c264ff4fca9a5a9e8fcbb97da0 : SUCCESS in 2h 27m 31s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":11},{"id":"8c0b7cc34591b344b2a7091f1f4137802f4c00b6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 16:25:36.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01b3e555740c40bcb899a1331c47bd42\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c635610995744f8192589ad0a4c88955 : SUCCESS in 47m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/932b267f1c7b4ba5bf485c36490627d9 : SUCCESS in 22m 45s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c2a530f8d7e047ac900184ca1ecd833f : SUCCESS in 33m 20s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"90cf25a4a3fbe321d09b58bd6a1630981d198a1f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-12 14:39:30.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"c4ec4f79f71f50503ae6ea6bf1cfa3e17b5aa784","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-06-16 13:15:40.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"29942c962779659cae01ca6ae2ddc4a726850028","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-06-17 08:07:06.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"4cc9b14ccfb99e826b3f1d1f80930246c92a514d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 08:07:56.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"ee6f8fb110955c4327fc95469adb955b230997e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 08:51:17.000000000","message":"Patch Set 11: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1cb1942fc5554adbac5c3da31aa58027\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4bece6bbf91c4b4ab6c070f085167bcc : FAILURE in 27m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8ad9dd37554140338744ef1dfc21be11 : POST_FAILURE in 24m 33s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3028535b36854fad9f1c79a72d2653b3 : FAILURE in 14m 47s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6efddb8a4d67465ba5b1cc55b0500aba : FAILURE in 19m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f23e73564ea47aebfd010658fcca943 : FAILURE in 7m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b0cc370e8fb24d17a031b4dc3d890973 : FAILURE in 2m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/edac83be231c4ca58285771d107e7811 : FAILURE in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc4cc459318a4928b5201aa724fb640e : FAILURE in 7m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0a111780d2b4213bf8bb25cbd20309e : SUCCESS in 19m 26s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7d4e5815e8414672ad33ae4ddb3899c9 : FAILURE in 23m 20s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/43ee5aec02a64d199223ee6c0e90404c : FAILURE in 24m 31s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/10ba80af0181456f852d207f89c63731 : FAILURE in 41m 18s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7e57304ad4f14da297bd6f75bb15f601 : FAILURE in 39m 38s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/21ca4ca8837d4e57ad87ba08e5f45960 : FAILURE in 21m 16s","accounts_in_message":[],"_revision_number":11},{"id":"221ca17da85aa714bfa78292146880ac8e88fd01","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-17 11:07:01.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"8e82cde9bc47e7bef60427bbb62bd3f8ef71e874","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-17 11:46:15.000000000","message":"Patch Set 11:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/56/984356/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"7491094e5492c4acc199eefee8c7098b461a3b6c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-17 11:51:00.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"79baaeb2ff6b9ab3630c9bc14b799e2f1ad50d5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 12:09:49.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6e568633c5e64590b43dc23322a13b68\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/011413ca5991459c8ce9da3b2d6aeaf5 : FAILURE in 11m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9ca493e14c0443f88c7497024b2c81c1 : POST_FAILURE in 24m 06s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9c4818be5bca4839b11640804c123f8e : FAILURE in 16m 07s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/329d97e9e70c4258b8567176bbbea27a : FAILURE in 18m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/81fc9a55d5454ae29e1e3afa1447f7d1 : FAILURE in 17m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6dfbba15d4eb4295a62c14dfb34a2609 : FAILURE in 12m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e215100eb844df1a4f9033a76925516 : FAILURE in 8m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/97a1d20cf3f9478e995f168a34d6bcf5 : FAILURE in 4m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b2aadeb6c9b24287a30536356dcdc62d : FAILURE in 4m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c25856739936476abab934a149d0af7d : FAILURE in 10m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad40a831bee745ef913102ea5bbd3d8b : FAILURE in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/871977bc432f455cafcac0863f68110f : SUCCESS in 22m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6e488f0bb0884e579467096930daad6a : FAILURE in 14m 33s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ddfa40bc1051447288cf4d9160437a08 : FAILURE in 34m 06s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ac6fddbc87944cdba753e987cb46ecac : FAILURE in 59m 10s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/87de505a7b614a208b3865d3c8c07967 : FAILURE in 18m 06s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d8d94cbcad3740b992722e9bcd9ac1c2 : FAILURE in 26m 42s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1f643595e9fb4e45ac90baac099a0c62 : FAILURE in 21m 59s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/08a674fdbe3b438184c0d10118eb315d : FAILURE in 14m 46s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/18d076b751c844048f62bf717927e6ea : FAILURE in 28m 29s","accounts_in_message":[],"_revision_number":11},{"id":"d40ecc6bb4da3a1d2fbae7b7e24e948020bc3001","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-17 12:26:57.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n* Verified+1, Verified-1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"34b8029d89b3abcf7fd3444b4ffd7488e7e5185c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-17 13:06:51.000000000","message":"Patch Set 12: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/56/984356/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"80181b568a10f0bd153f1c07ffbb18ec96317b6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 15:05:02.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f16af9369c146848bdcd2f9bfb6a5d3\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/27006573cf3e4c35aa5c18289d886fe3 : SUCCESS in 1h 29m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e6c519910f4141e7803e34ae9a248f4e : SUCCESS in 1h 05m 33s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a5b6d0cebda3465face99efac4120c30 : SUCCESS in 1h 25m 51s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/434272c3a1364c5385c0979971ef09ec : SUCCESS in 2h 34m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6908702d7f7d4fbcb4cca74386184b67 : SUCCESS in 31m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/14128ada822249c796775a2380ec7f51 : SUCCESS in 1h 16m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2718ba56165420b94cab0dc9c046b96 : SUCCESS in 10m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4fb86aff5a7f4a38aa13e1b5b858fbd8 : SUCCESS in 20m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/156fae4da0c741009c998a2b10659877 : SUCCESS in 21m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/710dd4d6aabb454087458044c16f7a0b : SUCCESS in 30m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1230443f211d4e4599c73ff4978ad231 : SUCCESS in 14m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/751f3fedb4884a2f8a19f84629dc2fe7 : SUCCESS in 21m 48s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/daaff5819b564c3a856ca0da839930de : SUCCESS in 1h 05m 18s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/58916d818e2f4c0c8ecd50fbd802434d : SUCCESS in 1h 09m 07s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/872bc5ae0b1d46f4aea4c389c2033924 : SUCCESS in 1h 18m 24s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f55fd76c4eb24acaa6e67c5dc47f48e0 : SUCCESS in 1h 53m 43s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/07af936c929c48469e462cf9dee6f784 : SUCCESS in 1h 32m 45s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ea4d9022bf254371942efa752b60d7e4 : SUCCESS in 1h 17m 47s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f9a945c45f814cada9113d2c21f5bb6a : SUCCESS in 57m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ab62672b30d244e39f4c2eb47d3e0e3a : SUCCESS in 38m 02s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":12},{"id":"56d808e23d8af0d171e2dd1642707615b8d34aa7","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-17 16:19:00.000000000","message":"Patch Set 12: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":12},{"id":"b04b7ec8c137c31577121e27dcf8c97d771ff78a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 23:21:34.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d114a5c4329942eaad70d7e96bae0b0f\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f929a5824e5c4ee3a087a9c35c9dba31 : SUCCESS in 55m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/965ba1d415e54fc898ea89676599dd07 : SUCCESS in 50m 58s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/242fc00ba08743679b8939ec1aefbfab : SUCCESS in 1h 03m 34s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"4c7a7d6bb0f0f836c397416b7b51316148e33305","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-18 05:58:43.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"1d9b327f7b397168927dab2338f8c5d7208c23d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-18 05:59:17.000000000","message":"Patch Set 13: Published edit on patch set 12.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"681a02e6f64c2908db9b12af680e396f3c4db529","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-18 06:55:35.000000000","message":"Patch Set 13:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/56/984356/13 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":13},{"id":"6a001202c23dc5d960f823e645690a57c7ae371e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 07:57:26.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a253cfafa10f45acbd1df2db15b2ec9d\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/83aad7238774425395c4717e6b88b32c : SUCCESS in 1h 01m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a4f0afc024ce4a85a784b6f44d2e22b9 : SUCCESS in 33m 50s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/17f96710692f4cee87abca042a84353e : SUCCESS in 1h 30m 46s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b5190abdb1364a97b99f193037e901f1 : SUCCESS in 1h 41m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/68b8deccc8474aa8833a3fca28ab08ca : SUCCESS in 31m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae4ae4e5170b416b860464ca79f8a4f6 : SUCCESS in 1h 14m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f74b645a90b24dba9c1b6c94c21d8fd9 : SUCCESS in 11m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/796d9540ab4f47219b4130210002a540 : SUCCESS in 21m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/681404da56e44570b6659cf4f46af192 : SUCCESS in 35m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c9f7dc520bea4d91b11e882b172c4f53 : SUCCESS in 28m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e54fc44aadbe4021937c047897ca763b : SUCCESS in 9m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d24b9089bb314991938a9f83ef01fb8e : SUCCESS in 8m 56s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0fc84b9dc40b4dc1ae77c8de0b8ebfd9 : SUCCESS in 57m 41s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4460eefabcb3421298da70b17ad1a1b8 : SUCCESS in 1h 12m 19s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/57478eb377d044658c69c659aca3a38d : SUCCESS in 1h 16m 11s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/515afac8b783469eb1173a8b681cf577 : SUCCESS in 1h 47m 56s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/347d31047cbb488db3c007f965279603 : SUCCESS in 1h 05m 11s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/15538f3245f743029304d3329d8cf81b : SUCCESS in 52m 30s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a63099d18d6542cd987028c3e3b50efb : SUCCESS in 35m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/da3d061afce7497bb28828ca8a821a9f : SUCCESS in 1h 27m 03s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":13},{"id":"16c722d7c17f07588c695d1c00212cb95d7059da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 10:15:04.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27cd86bfb68c43b1b782e706094ee29b\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4acb76a3215f4c64bed39d44fd9817c8 : SUCCESS in 33m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c2818d6ab296454d87e182deb2e4d0d4 : SUCCESS in 20m 26s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5bd95ec9516344e99a0919726083bf27 : SUCCESS in 1h 13m 14s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"f4d367f140092c0ad00d37f4d563d5cc426b5cb6","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-06-18 14:35:58.000000000","message":"Patch Set 13: Code-Review+2","accounts_in_message":[],"_revision_number":13},{"id":"9c848764325931d426c1ca7ebf560d24e82392be","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-18 19:43:32.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":13},{"id":"7555c59658e8116172f5d36dc643f74a1484ecfc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 19:44:01.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"967a193c882ce2f762a0e328b461dfba76753d70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 21:56:42.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98301bfd50b440528703c28ddc026808\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a0986f536fb74a46ab2ab458b1314fca : SUCCESS in 1h 26m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/190261be8fd643e8a83663234f259467 : SUCCESS in 33m 52s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6ba53cc0f5d547ed9ea7530dee0731f2 : SUCCESS in 1h 35m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a219c51b95af47d79316c37fe51e96cd : SUCCESS in 2h 12m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eff32e7fb5e44d049a953122f2923290 : SUCCESS in 11m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4d67f40fb65541c38f0c9169a45ee2c7 : SUCCESS in 33m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b8a92b52b05748968d149f7b4e260dd9 : SUCCESS in 28m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/391ecf3537394b66aca996d6c641f333 : SUCCESS in 5m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/03e0d14e09c9451f9dc322d9bb8c3924 : SUCCESS in 8m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6b25b421706746d19ac29c4aa565d447 : SUCCESS in 1h 32m 55s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8fd95a0b81214d13947cc04fe60c0c9f : SUCCESS in 2h 04m 20s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/747c76d3aa0c4fa08abb6a8382ddc979 : SUCCESS in 41m 58s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/09ec054e47584c29b3cda4752d24123c : SUCCESS in 36m 48s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/423e2c0d9c99414fa8d375450a757dfa : SUCCESS in 1h 04m 28s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":13},{"id":"29eb227986c9889ac71c083104ebca8fecd7e119","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 21:56:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":13},{"id":"f6b3e9a8c8b2d39cca90755f7c42de6704b3080a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 21:58:46.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60a535f51b3d4651815fc272043cc11f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f37929ddf47d43a194a8a1627c37c5b4 : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a37c219e0c04034b9b2e41fc8ecd2aa : SUCCESS in 40s","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"a8856b7c6f6ccab8352d3de37c0b108ae33db790","revisions":{"4f37cd887d83a610f6f0adfb7873a57c6971d019":{"kind":"REWORK","_number":1,"created":"2026-04-13 10:24:49.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/56/984356/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/56/984356/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/984356/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/984356/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/984356/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/56/984356/1"}}},"commit":{"parents":[{"commit":"7255308461ddb01e151cfd96234470ea0a7acc23","subject":"Batch SG rule ``ACL`` operations into single OVN NB transactions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7255308461ddb01e151cfd96234470ea0a7acc23"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-04-12 11:17:42.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-04-12 11:18:21.000000000","tz":120},"subject":"WIP -- Default SG statefulness","message":"WIP -- Default SG statefulness\n\nRelated-Bug: #2146803\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I1c773a445a65414dbb3df325a3a61b94d050efc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4f37cd887d83a610f6f0adfb7873a57c6971d019"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4f37cd887d83a610f6f0adfb7873a57c6971d019"}]},"branch":"refs/heads/master"},"e12bdf655e8390e62d663589523851ce2045b39c":{"kind":"REWORK","_number":2,"created":"2026-04-23 07:06:42.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/56/984356/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/56/984356/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/984356/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/984356/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/984356/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/56/984356/2"}}},"commit":{"parents":[{"commit":"c816d400b10fd7ba90fda86e81656a7e2a46638d","subject":"Merge \"tests: Handle change in hash for neutron-lib field fixes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c816d400b10fd7ba90fda86e81656a7e2a46638d"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-04-12 11:17:42.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-04-21 21:37:14.000000000","tz":120},"subject":"WIP -- Default SG statefulness","message":"WIP -- Default SG statefulness\n\nNOTE: the API definition is added temporarily with a note to remove\nit once the neutron-lib patch is released [1].\n\n[1]https://review.opendev.org/c/openstack/neutron-lib/+/984354\n\nRelated-Bug: #2146803\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I1c773a445a65414dbb3df325a3a61b94d050efc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e12bdf655e8390e62d663589523851ce2045b39c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e12bdf655e8390e62d663589523851ce2045b39c"}]},"branch":"refs/heads/master"},"c0d670e8654c251854deb0cad798793c19044e1c":{"kind":"REWORK","_number":3,"created":"2026-05-08 15:19:06.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/56/984356/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/56/984356/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/984356/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/984356/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/984356/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/56/984356/3"}}},"commit":{"parents":[{"commit":"8c322ca73ea593e6e8eb75bf88a3b15ba13f60ee","subject":"Merge \"Add --port option to remove-duplicated-port-bindings script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c322ca73ea593e6e8eb75bf88a3b15ba13f60ee"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-04-12 11:17:42.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-05-08 15:18:41.000000000","tz":120},"subject":"Add security-groups-default-statefulness API extension","message":"Add security-groups-default-statefulness API extension\n\nIntroduce a new API extension that allows configuring the default\nvalue of the ``stateful`` attribute for new security groups, on a\nper-project or system-wide basis. The new resource supports full\nCRUD operations with corresponding RBAC policy rules.\n\nThe API definition module is temporarily vendored in neutron until\nthe neutron-lib patch [1] is merged and released.\n\n[1] https://review.opendev.org/c/openstack/neutron-lib/+/984354\n\nRelated-Bug: #2146803\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I1c773a445a65414dbb3df325a3a61b94d050efc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c0d670e8654c251854deb0cad798793c19044e1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c0d670e8654c251854deb0cad798793c19044e1c"}]},"branch":"refs/heads/master"},"dbb93ba1883723167a9ca64772ba8ea4af6889a1":{"kind":"REWORK","_number":4,"created":"2026-05-11 07:57:01.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/56/984356/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/56/984356/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/984356/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/984356/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/984356/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/56/984356/4"}}},"commit":{"parents":[{"commit":"8c322ca73ea593e6e8eb75bf88a3b15ba13f60ee","subject":"Merge \"Add --port option to remove-duplicated-port-bindings script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c322ca73ea593e6e8eb75bf88a3b15ba13f60ee"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-04-12 11:17:42.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-05-11 07:56:34.000000000","tz":120},"subject":"Add security-groups-default-statefulness API extension","message":"Add security-groups-default-statefulness API extension\n\nIntroduce a new API extension that allows configuring the default\nvalue of the ``stateful`` attribute for new security groups, on a\nper-project or system-wide basis. The new resource supports full\nCRUD operations with corresponding RBAC policy rules.\n\nThe API definition module is temporarily vendored in neutron until\nthe neutron-lib patch [1] is merged and released.\n\n[1] https://review.opendev.org/c/openstack/neutron-lib/+/984354\n\nRelated-Bug: #2146803\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I1c773a445a65414dbb3df325a3a61b94d050efc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dbb93ba1883723167a9ca64772ba8ea4af6889a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dbb93ba1883723167a9ca64772ba8ea4af6889a1"}]},"branch":"refs/heads/master"},"c616831a8dc0d18f8a41d759e985a59ec9f285ba":{"kind":"REWORK","_number":5,"created":"2026-05-12 09:07:13.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/56/984356/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/56/984356/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/984356/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/984356/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/984356/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/56/984356/5"}}},"commit":{"parents":[{"commit":"99bd9c62d13616cf1fe17a4ed291b8d30c2d3757","subject":"Merge \"Set attempts\u003d1 in ``neutron-ovn-bgp-tempest-multinode`` job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/99bd9c62d13616cf1fe17a4ed291b8d30c2d3757"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-04-12 11:17:42.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-05-12 09:06:58.000000000","tz":120},"subject":"WIP \u003d\u003d Add security-groups-default-statefulness API extension","message":"WIP \u003d\u003d Add security-groups-default-statefulness API extension\n\nIntroduce a new API extension that allows configuring the default\nvalue of the ``stateful`` attribute for new security groups, on a\nper-project or system-wide basis. The new resource supports full\nCRUD operations with corresponding RBAC policy rules.\n\nThe API definition module is temporarily vendored in neutron until\nthe neutron-lib patch [1] is merged and released.\n\n[1] https://review.opendev.org/c/openstack/neutron-lib/+/984354\n\nRelated-Bug: #2146803\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I1c773a445a65414dbb3df325a3a61b94d050efc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c616831a8dc0d18f8a41d759e985a59ec9f285ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c616831a8dc0d18f8a41d759e985a59ec9f285ba"}]},"branch":"refs/heads/master"},"740cd572e920de1f7c9e8a6aa8f0667f7ca693c6":{"kind":"REWORK","_number":6,"created":"2026-05-12 11:17:13.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/56/984356/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/56/984356/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/984356/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/984356/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/984356/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/56/984356/6"}}},"commit":{"parents":[{"commit":"99bd9c62d13616cf1fe17a4ed291b8d30c2d3757","subject":"Merge \"Set attempts\u003d1 in ``neutron-ovn-bgp-tempest-multinode`` job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/99bd9c62d13616cf1fe17a4ed291b8d30c2d3757"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-04-12 11:17:42.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-05-12 11:17:02.000000000","tz":120},"subject":"WIP \u003d\u003d Add security-groups-default-statefulness API extension","message":"WIP \u003d\u003d Add security-groups-default-statefulness API extension\n\nIntroduce a new API extension that allows configuring the default\nvalue of the ``stateful`` attribute for new security groups, on a\nper-project or system-wide basis. The new resource supports full\nCRUD operations with corresponding RBAC policy rules.\n\nThe API definition module is temporarily vendored in neutron until\nthe neutron-lib patch [1] is merged and released.\n\n[1] https://review.opendev.org/c/openstack/neutron-lib/+/984354\n\nRelated-Bug: #2146803\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I1c773a445a65414dbb3df325a3a61b94d050efc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/740cd572e920de1f7c9e8a6aa8f0667f7ca693c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/740cd572e920de1f7c9e8a6aa8f0667f7ca693c6"}]},"branch":"refs/heads/master"},"cb08e7e5ba949784c02c91f69d9f594cb0b9bdb7":{"kind":"REWORK","_number":7,"created":"2026-05-13 07:45:51.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/56/984356/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/56/984356/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/984356/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/984356/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/984356/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/56/984356/7"}}},"commit":{"parents":[{"commit":"99bd9c62d13616cf1fe17a4ed291b8d30c2d3757","subject":"Merge \"Set attempts\u003d1 in ``neutron-ovn-bgp-tempest-multinode`` job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/99bd9c62d13616cf1fe17a4ed291b8d30c2d3757"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-04-12 11:17:42.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-05-13 07:45:13.000000000","tz":120},"subject":"WIP \u003d\u003d Add security-groups-default-statefulness API extension","message":"WIP \u003d\u003d Add security-groups-default-statefulness API extension\n\nIntroduce a new API extension that allows configuring the default\nvalue of the ``stateful`` attribute for new security groups, on a\nper-project or system-wide basis. The new resource supports full\nCRUD operations with corresponding RBAC policy rules.\n\nThe API definition module is temporarily vendored in neutron until\nthe neutron-lib patch [1] is merged and released.\n\n[1] https://review.opendev.org/c/openstack/neutron-lib/+/984354\n\nRelated-Bug: #2146803\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I1c773a445a65414dbb3df325a3a61b94d050efc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cb08e7e5ba949784c02c91f69d9f594cb0b9bdb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cb08e7e5ba949784c02c91f69d9f594cb0b9bdb7"}]},"branch":"refs/heads/master"},"f1872419abf95a684a38bfd33b18d95518f22125":{"kind":"REWORK","_number":8,"created":"2026-05-27 20:50:40.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/56/984356/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/56/984356/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/984356/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/984356/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/984356/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/56/984356/8"}}},"commit":{"parents":[{"commit":"78aad983c676661be1a228c6262a80ac528d2bd6","subject":"Merge \"Fix port RBAC policies to require network ownership\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/78aad983c676661be1a228c6262a80ac528d2bd6"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-04-12 11:17:42.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-05-27 20:50:25.000000000","tz":120},"subject":"WIP \u003d\u003d Add security-groups-default-statefulness API extension","message":"WIP \u003d\u003d Add security-groups-default-statefulness API extension\n\nIntroduce a new API extension that allows configuring the default\nvalue of the ``stateful`` attribute for new security groups, on a\nper-project or system-wide basis. The new resource supports full\nCRUD operations with corresponding RBAC policy rules.\n\nThe API definition module is temporarily vendored in neutron until\nthe neutron-lib patch [1] is merged and released.\n\n[1] https://review.opendev.org/c/openstack/neutron-lib/+/984354\n\nRelated-Bug: #2146803\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I1c773a445a65414dbb3df325a3a61b94d050efc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f1872419abf95a684a38bfd33b18d95518f22125"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f1872419abf95a684a38bfd33b18d95518f22125"}]},"branch":"refs/heads/master"},"302f5e7cd0762877a8d728d0898383d87745ed20":{"kind":"REWORK","_number":9,"created":"2026-05-28 08:04:35.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/56/984356/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/56/984356/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/984356/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/984356/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/984356/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/56/984356/9"}}},"commit":{"parents":[{"commit":"1b58ef38b7e83322804c16b8a223f84333107b97","subject":"Revert \"tests: Handle change in hash for neutron-lib field fixes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1b58ef38b7e83322804c16b8a223f84333107b97"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-04-12 11:17:42.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-05-28 08:04:09.000000000","tz":120},"subject":"Add security-groups-default-statefulness API extension","message":"Add security-groups-default-statefulness API extension\n\nIntroduce a new API extension that allows configuring the default\nvalue of the ``stateful`` attribute for new security groups, on a\nper-project or system-wide basis. The new resource supports full\nCRUD operations with corresponding RBAC policy rules.\n\nThe API definition ``security-groups-default-statefulness`` is\nprovided in [1] and was released in neutron-lib 4.0.0.\n\n[1]https://review.opendev.org/c/openstack/neutron-lib/+/984354\n\nCloses-Bug: #2146803\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I1c773a445a65414dbb3df325a3a61b94d050efc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/302f5e7cd0762877a8d728d0898383d87745ed20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/302f5e7cd0762877a8d728d0898383d87745ed20"}]},"branch":"refs/heads/master"},"7a786e9c2023abf70e1cc3ffee6c43e5e85533a1":{"kind":"REWORK","_number":10,"created":"2026-05-29 08:45:41.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/56/984356/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/56/984356/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/984356/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/984356/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/984356/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/56/984356/10"}}},"commit":{"parents":[{"commit":"1b58ef38b7e83322804c16b8a223f84333107b97","subject":"Revert \"tests: Handle change in hash for neutron-lib field fixes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1b58ef38b7e83322804c16b8a223f84333107b97"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-04-12 11:17:42.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-05-29 08:38:26.000000000","tz":120},"subject":"Add security-groups-default-statefulness API extension","message":"Add security-groups-default-statefulness API extension\n\nIntroduce a new API extension that allows configuring the default\nvalue of the ``stateful`` attribute for new security groups, on a\nper-project or system-wide basis. The new resource supports full\nCRUD operations with corresponding RBAC policy rules.\n\nThe API definition ``security-groups-default-statefulness`` is\nprovided in [1] and was released in neutron-lib 4.0.0.\n\n[1]https://review.opendev.org/c/openstack/neutron-lib/+/984354\n\nCloses-Bug: #2146803\nAssisted-By: Claude Opus 4.6\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I1c773a445a65414dbb3df325a3a61b94d050efc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7a786e9c2023abf70e1cc3ffee6c43e5e85533a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7a786e9c2023abf70e1cc3ffee6c43e5e85533a1"}]},"branch":"refs/heads/master"},"48ca17d946d0d12b017b07e1663e58dbe854a5b7":{"kind":"REWORK","_number":11,"created":"2026-06-08 13:20:19.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/56/984356/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/56/984356/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/984356/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/984356/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/984356/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/56/984356/11"}}},"commit":{"parents":[{"commit":"233ed6af791c7fb0df0bceea0c1fdf077f934d91","subject":"[CI] Bump tox cover timeout by 20 minutes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/233ed6af791c7fb0df0bceea0c1fdf077f934d91"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-04-12 11:17:42.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-06-08 13:20:05.000000000","tz":120},"subject":"Add security-groups-default-statefulness API extension","message":"Add security-groups-default-statefulness API extension\n\nIntroduce a new API extension that allows configuring the default\nvalue of the ``stateful`` attribute for new security groups, on a\nper-project or system-wide basis. The new resource supports full\nCRUD operations with corresponding RBAC policy rules.\n\nThe API definition ``security-groups-default-statefulness`` is\nprovided in [1] and was released in neutron-lib 4.0.0.\n\n[1]https://review.opendev.org/c/openstack/neutron-lib/+/984354\n\nCloses-Bug: #2146803\nAssisted-By: Claude Opus 4.6\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I1c773a445a65414dbb3df325a3a61b94d050efc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/48ca17d946d0d12b017b07e1663e58dbe854a5b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/48ca17d946d0d12b017b07e1663e58dbe854a5b7"}]},"branch":"refs/heads/master"},"e22d8920181cb4c5417d8880550c50878c37c4d6":{"kind":"REWORK","_number":12,"created":"2026-06-17 12:26:57.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/56/984356/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/56/984356/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/984356/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/984356/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/984356/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/56/984356/12"}}},"commit":{"parents":[{"commit":"319761f0697026227aa7b0a882bfdfddbae74f91","subject":"Merge \"Explicitly create log directories in the FrrFixture\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/319761f0697026227aa7b0a882bfdfddbae74f91"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-04-12 11:17:42.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-06-17 12:20:57.000000000","tz":120},"subject":"Add security-groups-default-statefulness API extension","message":"Add security-groups-default-statefulness API extension\n\nIntroduce a new API extension that allows configuring the default\nvalue of the ``stateful`` attribute for new security groups, on a\nper-project or system-wide basis. The new resource supports full\nCRUD operations with corresponding RBAC policy rules.\n\nThe API definition ``security-groups-default-statefulness`` is\nprovided in [1] and was released in neutron-lib 4.0.0.\n\n[1]https://review.opendev.org/c/openstack/neutron-lib/+/984354\n\nCloses-Bug: #2146803\nAssisted-By: Claude Opus 4.6\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I1c773a445a65414dbb3df325a3a61b94d050efc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e22d8920181cb4c5417d8880550c50878c37c4d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e22d8920181cb4c5417d8880550c50878c37c4d6"}]},"branch":"refs/heads/master"},"a8856b7c6f6ccab8352d3de37c0b108ae33db790":{"kind":"REWORK","_number":13,"created":"2026-06-18 05:59:17.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/56/984356/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/56/984356/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/984356/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/984356/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/984356/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/56/984356/13"}}},"commit":{"parents":[{"commit":"319761f0697026227aa7b0a882bfdfddbae74f91","subject":"Merge \"Explicitly create log directories in the FrrFixture\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/319761f0697026227aa7b0a882bfdfddbae74f91"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-04-12 11:17:42.000000000","tz":120},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2026-06-18 05:59:17.000000000","tz":0},"subject":"Add security-groups-default-statefulness API extension","message":"Add security-groups-default-statefulness API extension\n\nIntroduce a new API extension that allows configuring the default\nvalue of the ``stateful`` attribute for new security groups, on a\nper-project or system-wide basis. The new resource supports full\nCRUD operations with corresponding RBAC policy rules.\n\nThe API definition ``security-groups-default-statefulness`` is\nprovided in [1] and was released in neutron-lib 4.0.0.\n\n[1]https://review.opendev.org/c/openstack/neutron-lib/+/984354\n\nCloses-Bug: #2146803\nAssisted-By: Claude Opus 4.6\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I1c773a445a65414dbb3df325a3a61b94d050efc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a8856b7c6f6ccab8352d3de37c0b108ae33db790"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a8856b7c6f6ccab8352d3de37c0b108ae33db790"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
