)]}'
{"id":"openstack%2Fneutron~984364","triplet_id":"openstack%2Fneutron~master~Iecdf6b45c070711a3448e7bade3577c340c291cc","project":"openstack/neutron","branch":"master","topic":"fips_python312","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-04-22 18:10:47.000000000","reason":"Change was submitted"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2026-04-22 18:10:47.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2026-04-22 18:10:47.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-04-22 16:00:05.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"Iecdf6b45c070711a3448e7bade3577c340c291cc","subject":"[functional fips] Switch venv prep to python3.12","status":"MERGED","created":"2026-04-13 10:58:25.000000000","updated":"2026-04-22 18:12:54.000000000","submitted":"2026-04-22 18:10:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"984364-fips_python312","meta_rev_id":"f65ce7a9016d9d7c8573380fc2ddd9f1e69bff3b","_number":984364,"virtual_id_number":984364,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-22 18:10:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2026-04-14 09:12:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2026-04-22 16:00:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2026-04-22 16:00:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-13 12:34:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-14 09:12:18.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-04-16 05:28:04.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-04-16 05:28:04.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-04-16 05:28:04.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"cfb8a4940fe63a1ba14be1e4ae2b186b294c2340","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-04-13 10:58:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"97a541a172f23e6314c058d11a6e0b7397262cc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 12:34:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aabe66bee99d406484a71fc9a1db78e5\n\n- neutron-functional-fips https://zuul.opendev.org/t/openstack/build/d99dd70b8075454b996b35e594aa4ffd : RETRY_LIMIT in 9m 47s","accounts_in_message":[],"_revision_number":1},{"id":"d58bed1fe1dbb11b05b085d3be3c4cf235ec056d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-04-13 13:37:09.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"9e13a52d9ccdc9ed12c8d8488b063bc058766a7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 14:28:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f222f52b19294f899478140dc0c13f8c\n\n- neutron-functional-fips https://zuul.opendev.org/t/openstack/build/8cae333432f04aeba66c2f0478d32cb1 : RETRY_LIMIT in 17m 59s","accounts_in_message":[],"_revision_number":2},{"id":"a5e8fd490242f99b77a6bc05d3f13b0de1300bd7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-04-13 14:54:36.000000000","message":"Uploaded patch set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"b718d903982fe3619fa19eec0a75e86d2a596c64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 15:59:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12f58839602844008d280a64a86599a0\n\n- neutron-functional-fips https://zuul.opendev.org/t/openstack/build/55a021da25314ed095dfb19c47215b11 : SUCCESS in 57m 00s","accounts_in_message":[],"_revision_number":3},{"id":"7ec79946352d7e0e209a199ad563f6aa6afb3be8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-04-14 04:30:52.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"ae6c6a49010faff3c33d2f45dca04ffb4fa1ffe5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 06:47:47.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dd6aa5efda1f4b6bafdefb63ebb4b005\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/97ad514967704b0fa1322c9ec096dac1 : SUCCESS in 1h 24m 15s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b6201b69c20f4c8a9f58ba4b0af52177 : SUCCESS in 1h 16m 28s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3e4cc8f90e8246f8bf64dde9d43bdf61 : SUCCESS in 2h 15m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/28a6b2751e834fbf93dc3d6c6d560690 : SUCCESS in 31m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c1819ca59484f439d3f51a57720eb38 : SUCCESS in 25m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aee4e916bfc64fc2bf5a2f1fa7bd90ae : SUCCESS in 12m 06s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/873dbe7baa06441caf29cd28a2fbcf09 : FAILURE in 1h 23m 45s","accounts_in_message":[],"_revision_number":4},{"id":"f401ea7826357d4cbea89e35c18f9d90bc044176","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-04-14 07:39:10.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3c035f1da9602393aaa49de27006b227ddff379c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-04-14 09:12:18.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5efc08dba288ef6ec686b20408036d52b7fc6141","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 09:41:55.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9a0543570d94e7ea2fbb703ac9ae762\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ab08719d92984e6c8871a6240e8b0031 : SUCCESS in 1h 04m 50s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e68b4841f1bc43b296b59b0360e4b64a : SUCCESS in 1h 31m 38s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1fdad9a38de144a6824134d0d14f7444 : SUCCESS in 1h 59m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bc75e13cd08143e887d67c5f32260307 : SUCCESS in 34m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/776b5065c7f844298540956673e892b4 : SUCCESS in 24m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/126d207e2b2340978ac9fe9355cf088b : SUCCESS in 10m 54s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/eb0c88fc365447c2b21f0243159c60ea : SUCCESS in 1h 07m 56s","accounts_in_message":[],"_revision_number":4},{"id":"c737a445addcb7be448d0a1b9da7c858e94a741c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-16 14:47:45.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"932af7ce6de0b9827847fa3693ad306da70b1a06","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-04-22 15:57:45.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"34dc3c993f697e71cc77a9ac218bb7112e4b2848","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-22 16:00:05.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"fc7cd1183a6f20c8813a2a6d0ebd81fa0a1e4849","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 16:00:53.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0be76dc858f995e7102d2d39d70588bc1273d312","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 18:10:47.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff2e05adbba0414baca780771ca94496\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/48f296b92481485299728c01b47d9a15 : SUCCESS in 37m 26s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/743ddc2249674f0db7c93291d4bc4051 : SUCCESS in 1h 18m 03s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/aa6691c733e34f4f8e25751f48508ac3 : SUCCESS in 2h 09m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b03fd1f97eb84c7eaa78613ec41dcd62 : SUCCESS in 18m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ab94a067241541b6bd0e333c77ab9e60 : SUCCESS in 29m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/224195ecf56640b594e8ed108fe3dab7 : SUCCESS in 6m 10s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a994b59a57eb45128eedf78f3c924d8b : SUCCESS in 1h 39m 38s","accounts_in_message":[],"_revision_number":4},{"id":"f90eda95ce2e9787c5e4ad80f7a92081c76a0d47","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 18:10:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"f65ce7a9016d9d7c8573380fc2ddd9f1e69bff3b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 18:12:54.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b795ae2e41564572958ce607bdaf82a6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c2266663c594c39adc1c03dfce10f0a : SUCCESS in 56s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a5420f5e6fa89b77cd87ed90a239ebca77447974","revisions":{"508cc3a32b508c7784b6b3b7d2c8e134975de53a":{"kind":"REWORK","_number":1,"created":"2026-04-13 10:58:25.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/64/984364/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/64/984364/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/984364/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/984364/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/984364/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/64/984364/1"}}},"commit":{"parents":[{"commit":"720802c8038d4ba0fe85824f437ed630eb184404","subject":"Merge \"Use project_id in policy and miscellaneous plugin code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/720802c8038d4ba0fe85824f437ed630eb184404"}]}],"author":{"name":"Yatin Karel","email":"ykarel@redhat.com","date":"2026-04-13 10:57:53.000000000","tz":330},"committer":{"name":"Yatin Karel","email":"ykarel@redhat.com","date":"2026-04-13 10:57:58.000000000","tz":330},"subject":"[WIP] fips python3.12 check","message":"[WIP] fips python3.12 check\n\nSigned-off-by: Yatin Karel \u003cykarel@redhat.com\u003e\nChange-Id: Iecdf6b45c070711a3448e7bade3577c340c291cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/508cc3a32b508c7784b6b3b7d2c8e134975de53a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/508cc3a32b508c7784b6b3b7d2c8e134975de53a"}]},"branch":"refs/heads/master"},"49d7bec5bcd3b6f037796ed9e905013b5cefe720":{"kind":"REWORK","_number":2,"created":"2026-04-13 13:37:09.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/64/984364/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/64/984364/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/984364/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/984364/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/984364/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/64/984364/2"}}},"commit":{"parents":[{"commit":"720802c8038d4ba0fe85824f437ed630eb184404","subject":"Merge \"Use project_id in policy and miscellaneous plugin code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/720802c8038d4ba0fe85824f437ed630eb184404"}]}],"author":{"name":"Yatin Karel","email":"ykarel@redhat.com","date":"2026-04-13 10:57:53.000000000","tz":330},"committer":{"name":"Yatin Karel","email":"ykarel@redhat.com","date":"2026-04-13 13:36:42.000000000","tz":330},"subject":"[WIP] fips python3.12 check","message":"[WIP] fips python3.12 check\n\nSigned-off-by: Yatin Karel \u003cykarel@redhat.com\u003e\nChange-Id: Iecdf6b45c070711a3448e7bade3577c340c291cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/49d7bec5bcd3b6f037796ed9e905013b5cefe720"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/49d7bec5bcd3b6f037796ed9e905013b5cefe720"}]},"branch":"refs/heads/master"},"17af2ed776dae94995037aa635ffe293fb84d0a5":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-04-13 14:54:36.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/64/984364/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/64/984364/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/984364/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/984364/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/984364/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/64/984364/3"}}},"commit":{"parents":[{"commit":"720802c8038d4ba0fe85824f437ed630eb184404","subject":"Merge \"Use project_id in policy and miscellaneous plugin code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/720802c8038d4ba0fe85824f437ed630eb184404"}]}],"author":{"name":"Yatin Karel","email":"ykarel@redhat.com","date":"2026-04-13 10:57:53.000000000","tz":330},"committer":{"name":"Yatin Karel","email":"ykarel@redhat.com","date":"2026-04-13 14:54:08.000000000","tz":330},"subject":"[WIP] fips python3.12 check","message":"[WIP] fips python3.12 check\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/984404\nSigned-off-by: Yatin Karel \u003cykarel@redhat.com\u003e\nChange-Id: Iecdf6b45c070711a3448e7bade3577c340c291cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/17af2ed776dae94995037aa635ffe293fb84d0a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/17af2ed776dae94995037aa635ffe293fb84d0a5"}]},"branch":"refs/heads/master"},"a5420f5e6fa89b77cd87ed90a239ebca77447974":{"kind":"REWORK","_number":4,"created":"2026-04-14 04:30:52.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/64/984364/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/64/984364/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/984364/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/984364/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/984364/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/64/984364/4"}}},"commit":{"parents":[{"commit":"720802c8038d4ba0fe85824f437ed630eb184404","subject":"Merge \"Use project_id in policy and miscellaneous plugin code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/720802c8038d4ba0fe85824f437ed630eb184404"}]}],"author":{"name":"Yatin Karel","email":"ykarel@redhat.com","date":"2026-04-13 10:57:53.000000000","tz":330},"committer":{"name":"Yatin Karel","email":"ykarel@redhat.com","date":"2026-04-14 04:22:59.000000000","tz":330},"subject":"[functional fips] Switch venv prep to python3.12","message":"[functional fips] Switch venv prep to python3.12\n\nWe switched these jobs to python3.12 long back but recently\nit was noticed the job was still running with python3.9 during\ntox venv prep. The job failed after more-itertools update[1] as\nthe newer version is not compatible with python3.9.\nThis patch switches venv prep to also run with python3.12.\n\n[1] https://review.opendev.org/c/openstack/requirements/+/983340\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/984404\nSigned-off-by: Yatin Karel \u003cykarel@redhat.com\u003e\nChange-Id: Iecdf6b45c070711a3448e7bade3577c340c291cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a5420f5e6fa89b77cd87ed90a239ebca77447974"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a5420f5e6fa89b77cd87ed90a239ebca77447974"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
