)]}'
{"id":"openstack%2Fneutron~984409","triplet_id":"openstack%2Fneutron~master~I76a9f0fa27043d374be01b034af0ed1009c23ef8","project":"openstack/neutron","branch":"master","topic":"bug/2144617","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2026-05-20 06:14:37.000000000","reason":"\u003cGERRIT_ACCOUNT_13861\u003e replied on the change","reason_account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2026-05-20 15:19:43.000000000","reason":"Change was submitted"},"38298":{"account":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"last_update":"2026-05-20 15:19:43.000000000","reason":"Change was submitted"},"34271":{"account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"last_update":"2026-05-20 15:19:43.000000000","reason":"Change was submitted"},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2026-05-19 16:54:57.000000000","reason":"\u003cGERRIT_ACCOUNT_8655\u003e replied on the change","reason_account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}}},"hashtags":[],"change_id":"I76a9f0fa27043d374be01b034af0ed1009c23ef8","subject":"Created OVN Agent EVPN extension \u0026 netlink monitor","status":"MERGED","created":"2026-04-13 15:13:56.000000000","updated":"2026-05-20 15:22:58.000000000","submitted":"2026-05-20 15:19:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":94,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"984409-bug/2144617","meta_rev_id":"ad6785ce0a5bc07453eaf7872de8414c77358fd9","_number":984409,"virtual_id_number":984409,"owner":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2026-05-19 23:12:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-20 15:19:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-20 06:14:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-19 16:54:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-05-20 12:58:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"reviewers":{"REVIEWER":[{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-13 15:47:14.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"CC"},{"updated":"2026-04-13 17:15:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-30 02:04:31.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"state":"CC"},{"updated":"2026-05-05 14:48:06.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2026-05-08 02:12:08.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-14 03:14:22.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-14 19:53:39.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"state":"REVIEWER"},{"updated":"2026-05-15 07:03:40.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2026-05-18 08:17:41.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"},{"updated":"2026-05-20 06:14:37.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"38ea003eed3df04a8e25439a65b567c7040ffe07","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-04-13 15:13:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2cb3e1f04e0211d69f4183a37f30f73927e3a989","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-04-13 15:47:14.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"1fe38b329a1f7c5a423d782e1b4e9b8a94ca802a","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-04-13 15:48:28.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"7016427daa7baecdfd2ce16d2cf839c129ff60f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 17:15:29.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/69726c0ad53f49b187654ee7899157da\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/372a317351fd4f8d8558b0014549f819 : SUCCESS in 1h 31m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8c152496c4cb4a83bdeae7b87bcdf04b : SUCCESS in 1h 10m 09s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2cc08675dfab44dea3f06eff7b8f5805 : SUCCESS in 1h 22m 27s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8c5a023afc554b438dc28000e34783c4 : SUCCESS in 1h 59m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/279e5f0d1922443db2f05e2e4ff06d1a : SUCCESS in 32m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef8491cad46544b38093b94a412a4b2f : SUCCESS in 1h 00m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d00679fd2c2348928d4208a4057f5b49 : FAILURE in 7m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f43917db4622408a933f4a33d71fca02 : SUCCESS in 36m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/127dcba99bf04b7bb8ba693f32f251c6 : SUCCESS in 27m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b1f2c9c67f5745dcab35c7a35d9ed886 : SUCCESS in 32m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/149ebde94a3141c5990af2b403a3b0bb : SUCCESS in 10m 58s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/49088d8afd6a4d8d8fd768312fd83d1b : FAILURE in 42m 54s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/36b2d8e188414e28b95397af10e5dfe1 : SUCCESS in 1h 10m 26s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/76ad4e290eea40699772cce2839005e8 : SUCCESS in 1h 02m 39s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b0f1887f7d774bab9dd9342d06460253 : SUCCESS in 1h 39m 28s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8cde686e60194766b3dbb7afed79f80b : SUCCESS in 1h 37m 57s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a3b7b69fbe4b4ab0bac0ef49d41fa976 : SUCCESS in 1h 14m 11s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1d184d849ade49e9b41534ccf62566bf : SUCCESS in 1h 12m 16s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/836d25340a2247f5ac1353fb6b27ec0b : SUCCESS in 53m 44s","accounts_in_message":[],"_revision_number":1},{"id":"504e38811735ccd6712e784db642fb127683c671","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 23:07:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa4ac351ba9f4f7c88515da8419593aa\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0274bfb992c24fe9897aeaeab190cffe : SUCCESS in 27m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5b8c091337a549488ea6b05c1e97bfaf : SUCCESS in 30m 17s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/329807782b854a8aa830e4399263a007 : SUCCESS in 43m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"951a25d5b28ac86933aa4fe4da1bdb620bb8d390","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-04-14 16:00:33.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"1078d7abce03b56d4112b3b8a693a8083e3f8647","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 18:03:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0bf6215a9f16486ca1142a7b25e6faf8\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3b4ae4657ec34307a4c7edca14e6e951 : SUCCESS in 1h 15m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/13a9b5b865054232808557a9a12755aa : SUCCESS in 1h 02m 57s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3a0df60f26ac49bb90d7f614fe864592 : SUCCESS in 1h 46m 49s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b9adbbcc67eb4b87a304e7624aadb7c1 : SUCCESS in 2h 01m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4bc69d7cfb42444c93abee30d65e63f6 : SUCCESS in 35m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b7bf90554d2f4db1a5c087115f96c680 : SUCCESS in 47m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50aabc6e90d64eb9aff40827ef252c68 : SUCCESS in 27m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a889cc63253e4a49a70520048d7edcfb : SUCCESS in 35m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/52ee0c4914c547a4881f0ffc37d59e15 : SUCCESS in 21m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8bcc8964604a4ec9be03a34e73fdd56d : SUCCESS in 32m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5f9803f894c422aa384da47e4fe03b5 : SUCCESS in 12m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c5d03f9e63254ad59cc84c7d61f7e4d7 : FAILURE in 59m 54s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5dcc66efb83c4a37b333d0395ad444b0 : SUCCESS in 1h 19m 20s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/4a28014d24bd468c912519c5c48ba561 : SUCCESS in 1h 20m 09s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0dd1b69add944be1b7fb250cb45396a3 : SUCCESS in 1h 07m 19s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/21b78cb5f56f435ebcb240bc6bdf3bbe : SUCCESS in 1h 42m 27s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c90e7c0dc3a34f88afb284e231857aa9 : SUCCESS in 53m 12s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e102993180f14c9fbeed33a955f7d234 : SUCCESS in 40m 09s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6711374e130944a4a57f49214b29c906 : SUCCESS in 1h 12m 38s","accounts_in_message":[],"_revision_number":2},{"id":"0543244c7a1c710e47d46069352ae47a5e7a961b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 22:00:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2b7a12f67714f8290e006754917c0a0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8b1516ea402f4dbd97ee982f716faa0b : SUCCESS in 36m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/30c0774efe564775a792db01af5df088 : SUCCESS in 20m 09s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/736680d147bb491e882eaa65003f3d3d : SUCCESS in 23m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1531de87d23f09de284f6133c884a28d787fe62b","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-04-20 13:38:45.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"4a883958ab19dedac5508bb3ae7bf9e0d5d7ba36","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-04-29 14:44:54.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"2afc27a703ea0395e5e833299f17c5ec4229579d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 19:44:49.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f5371a4619ef4b2c9c9dfd5b199c88c1\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1162a2b7aab6423fa4b28a4cbee11f92 : SUCCESS in 2m 57s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a7444ca4b5584504b389accc03228aae : SUCCESS in 1h 22m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7271dcc2f6ba4f9aac8b6dfc4b5fb828 : SUCCESS in 1h 03m 59s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f319787ca14c48daa74454f64861e0f9 : SUCCESS in 1h 20m 14s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f07db38c531d4a02bcf1996c06a6727e : SUCCESS in 1h 55m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0a4d03eed9d6495eb5344288e468e56e : SUCCESS in 34m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57b4ee4a672e413bbca8634d23f66565 : SUCCESS in 43m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edccc7f1d56d4619974e2d4e8bfbc0a2 : FAILURE in 8m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/82438920f78e4d7090bb018a5ac51669 : SUCCESS in 26m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/635693bd972a49e7be1852e049c79d75 : SUCCESS in 21m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/27a07fd333f947d1a8eba89e5f7b2232 : SUCCESS in 26m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f6b61093f1d947e4b140933000942c2e : SUCCESS in 35m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02381de34a024e2580c931e14c0038f2 : SUCCESS in 12m 06s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ca111da2bd0143ec93367b2c59892a0c : SUCCESS in 3m 28s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/de7d4b9ff6aa44af9d92b3440a105b2c : SUCCESS in 1h 18m 55s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ad70b01a0dcd4d18b29b360cf9db0bed : SUCCESS in 1h 24m 08s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/e6bd3ddcad0c476aaeaae4cdf90cb3c7 : SUCCESS in 58m 31s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9e3efb5dfa554663b46b0850fa2a34fd : SUCCESS in 1h 08m 32s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/abbef4a1ab694918b3f02b57fe979f62 : SUCCESS in 1h 39m 14s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e0de21c22d3d44f5b8093a1d571eddbd : SUCCESS in 57m 38s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/cd3eed76109c4f98b06e3214b0a5c2eb : SUCCESS in 1h 22m 32s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d11fd5eb77d944b2b29895cdc3f97901 : SUCCESS in 1h 27m 56s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/64a175e9429447f08112f82b878eb94a : SUCCESS in 1h 18m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0c6405ebdb5d3a1da4700d9d81f16f99433d0838","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 20:14:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f129547959c4e33b56696539b9d434a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/be6869a7dcb040c9b8006b9e1a715d48 : SUCCESS in 1h 02m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e09bc120544946cd9b6de93a0e48b65c : SUCCESS in 40m 29s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/148b293e715c4a1a8c2112a24d16737e : SUCCESS in 31m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"255cf79b23c3b6d5dd6c9f18f12617c3a94ffcc6","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-04-30 02:04:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d64a3d384bf0e899f95a6b9ca0e9a265aa142e44","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-04-30 16:11:51.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"d10fb39c53fbefca68355a31ce391711aa3d2db1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 17:36:35.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06a46a9bb834480db781d600690a56c9\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2ed6705fe70040b8a951e1c4cf7faa06 : SUCCESS in 30m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cf336f2813c741ceb0f705d8410c701c : SUCCESS in 41m 13s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/26da7ba1d1204ffd98685d19f9ec187e : SUCCESS in 27m 51s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9ed4ee8fa7f4c410748f06f6ed42746ef496c5ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 19:28:22.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a52b9718e78b470c9de8cad6a77137c5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/11c948011b994792a293154f6b7ddcd4 : SUCCESS in 2m 43s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4d5d5957d159473bacff386ce5fa5ded : SUCCESS in 1h 21m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cd9be0c2e2e548ee90c401fe030bae54 : SUCCESS in 34m 19s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/12dacc0612f542d08691fbd16011218f : SUCCESS in 1h 18m 28s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e6f06f6490d149b9bcb9b619974897a9 : SUCCESS in 2h 21m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cdd239fa36f84ca5977177d7576222c7 : SUCCESS in 33m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/558d85c06e47406fac48121d17f41f1b : SUCCESS in 46m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf22c7a750f940e5a2e169ba2c088417 : FAILURE in 7m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d28912b2485b428b846c29ad17d51992 : FAILURE in 21m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/848a5f682f5d430088631e55ec3b729c : SUCCESS in 19m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/05e877848a144274b46c8a4967a31ea5 : SUCCESS in 17m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/02eccaf5141943ea99a1ea4d1b35d593 : SUCCESS in 21m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61bd4bdd83ae4854bebee433f0d333cb : SUCCESS in 11m 59s\n- requirements-check https://zuul.opendev.org/t/openstack/build/117a49a71cef41bfa09ea1a1358e10f4 : SUCCESS in 1m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/438b0412227345e2bc136a8438068b50 : FAILURE in 26m 45s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/04073f74cf984de0ab5d2103ec67c385 : SUCCESS in 54m 55s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/4235419c9c2c4feba7ce462931cdbe9d : SUCCESS in 55m 11s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1bd90f0c1c054c1e91d8cf57aa60314d : SUCCESS in 1h 52m 50s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/673200bb3f974b4e8e0d944b08acf64f : SUCCESS in 1h 38m 41s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/31f9812fce6049ccbb996ca80f358e61 : SUCCESS in 1h 02m 42s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/201ef08b96e04dad9c717cf6fbad8a3b : SUCCESS in 1h 02m 42s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4afc19d968d54f959b520fd0189a550f : SUCCESS in 36m 44s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/a80d5d9a0afa4bfbab6df3bb8ce83c68 : TIMED_OUT in 3h 13m 42s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9346f5c3332fe0875477f8991c21bd17fd5c952c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-05-01 20:54:38.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"43bfd130b2a2bc851e9524237d6490996f8df7f2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-01 21:59:22.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66728afcd12d4e17a2404199ea917b53\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/467290cb61da4d81b02ee271daaa5e8b : SUCCESS in 28m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f736e18c471c4c608242e55d2676c5d1 : SUCCESS in 45m 30s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2689f729fe5c441dabb3dcd8acbde486 : SUCCESS in 1h 02m 35s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ae537cb760832c0c491bcc70b27a42be0506ca71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-01 23:45:45.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/20c4044ec7074322af55570f94f34d72\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5cf7429abfea409e86407f9639c68ca4 : SUCCESS in 2m 13s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/55f854dbb05b4b42b7bbf66cbd1fc450 : SUCCESS in 1h 22m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/404e420b0fda40798b9fae39dafd0ff1 : SUCCESS in 1h 02m 55s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/be59d8d48fa441eaa4309ff9715b51fb : TIMED_OUT in 2h 50m 03s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c6061558003044ffa4d0082f084cb9a4 : SUCCESS in 1h 53m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d1480fee3225415b88b55c94ccbe486e : SUCCESS in 31m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8208a1cba5c64fcbb7c7c730e27ff8e6 : SUCCESS in 38m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4478493b360c4aa1b425b19e35fd58dd : FAILURE in 10m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a4ecf1ada85a4a76a9e53f39d3f243d8 : SUCCESS in 34m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/163bc8bfc4be4c1e8b790472dc0a3174 : SUCCESS in 18m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7f3ef4644fcb4e9bbb76624ae4e4b20f : SUCCESS in 20m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4ba0942171794d50bb16d7afbf7620ac : SUCCESS in 20m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4d7bbc5efe044a5ac7ce9f88dd2a3aa : SUCCESS in 11m 18s\n- requirements-check https://zuul.opendev.org/t/openstack/build/103cd4c47fda443dae0338b2011a5834 : SUCCESS in 2m 46s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d7ef784f86bb4e83bc63ef31507dee6e : SUCCESS in 1h 27m 01s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/705c33d3d13f4b419e93df151873e871 : SUCCESS in 1h 08m 41s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/630728f18e4c4a74ab4e465f8e0e2266 : SUCCESS in 38m 44s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a703d0389b3942e6b81aaa42fb653659 : SUCCESS in 1h 12m 59s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9d6757ed576f4d0cada7b77dd9847f65 : SUCCESS in 2h 06m 27s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6fd6921552ad43038209e9deab0f16a8 : SUCCESS in 1h 16m 37s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c936662777ee410b85861f706d2d4130 : SUCCESS in 1h 14m 55s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0eb8be1f3fbe49abbd582280cee6b59b : SUCCESS in 1h 36m 04s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/038bd7d4477748a088a43676772094bf : SUCCESS in 2h 23m 23s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9988fe19014430d40748fbbb3250316b8130a5fe","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-05-04 14:06:10.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"12506706bccb975eeaab8b4df41c4e24557fd949","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-05-04 14:08:58.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":6},{"id":"b98b2fcc0278312cfbb6916e79a78925276facb3","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-05-04 14:48:20.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"bd5e42adffa6863982b158b3ac919c76af4b1f98","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-05-04 14:48:36.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5c153c1be3693555a4a202ce336ce0f1b7679692","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 14:56:45.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79552126415545a5835c4c971ac0e591\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b1d2a80da2ae4161b359212a7ce63799 : SUCCESS in 26m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a3e0135b7b7d4e3b9fb18f98800204ec : SUCCESS in 18m 28s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/84644bd5249b43008235a29cfdc0b45e : SUCCESS in 24m 34s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"0a24f81b0d98dc3d9a396f03e497af40cd527583","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 17:50:01.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18d72f7d8b9d4e119268c1b71de5b6a9\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6a097438ae584bc29b799ce6ef339137 : SUCCESS in 2m 31s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/eded2ed442074659bce733a30183fddf : SUCCESS in 1h 30m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bbcad743f31d4ebc8a2c1b84d73a7b61 : SUCCESS in 1h 06m 26s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/cdfeef83f8724a8b9704f48cb56bc6c3 : SUCCESS in 1h 34m 39s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2f3fb2fe456f42d1b6c694b56e26a100 : SUCCESS in 2h 01m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8f4bc8110d844e30a4b9498641a68ee6 : SUCCESS in 34m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cdfa3f7da71c4d438ac2a313833f8906 : SUCCESS in 1h 16m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd9c451d89da4eabad2d2c2572a720e1 : SUCCESS in 12m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b1f2cf794d264bb0b92dbabf0877fd44 : SUCCESS in 29m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3d12a6b417de4b0ab59e9f4854bd9253 : SUCCESS in 18m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6da2a35878c04423856208e3772b6a4d : SUCCESS in 28m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a11d639344b449efb5223b34c615e8c9 : SUCCESS in 36m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61b8c2dcf8204667a222e04d5f007654 : SUCCESS in 10m 34s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9395487dff704ee682f2f9fef3369078 : SUCCESS in 1m 44s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b71c82cfb086429191012131b120199b : SUCCESS in 1h 35m 34s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1891fb88965a4df48429ac7074197002 : SUCCESS in 1h 07m 03s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ba97ae5fd58b4ce89c2f5ee13257955d : SUCCESS in 1h 20m 30s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/af0fa0fb3a8d417aac60fc0faea97658 : SUCCESS in 1h 53m 20s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/dadaf81fb7a24ace87bc2e0f0cefa547 : SUCCESS in 58m 49s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1c391a7c6cdc47eca8698122e3016591 : SUCCESS in 56m 08s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/da06e139e945437d891e0b1bb0e55c35 : SUCCESS in 1h 10m 46s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/34cc30edaa444c6eb0b0f36843a7025a : SUCCESS in 1h 36m 39s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/5cc5f755b2a049859d5eceb535be7122 : SUCCESS in 2h 37m 20s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":6},{"id":"e0964c0c29b46df436d33faca67a00408f9f8b98","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-05-05 14:48:06.000000000","message":"Patch Set 6: Code-Review-1\n\n(11 comments)","accounts_in_message":[],"_revision_number":6},{"id":"ee3e4ffe676c5f5e23efc7d0de551626ee142262","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-05-06 02:15:59.000000000","message":"Patch Set 6:\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"29858243e27e782f53f9f031931653eabc24a839","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-05-06 13:01:26.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"89602873c382d93ba4887e8a172b840889384766","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-05-06 19:44:19.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"72482e3f933fbde4513aa6c8b5b1689726bd9292","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 21:00:37.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9426e7855e4480ab2c257bed90a674d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/91a62cf7096041a886a9a3eac0ad737c : SUCCESS in 28m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/21c7cdc88592479ba822991b3a36f599 : SUCCESS in 37m 50s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/934abb508eac4d028eed29e8728313c0 : SUCCESS in 30m 48s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"f8118b6675cb0f32357a1288d877e2251e48c616","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 21:48:34.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c50c328765c4fc3bfe8ed0f3ec82ea6\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d8ecbd2892da474e9b0da7af1d23eee9 : SUCCESS in 2m 34s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/78e0f5fd610c45feb7477560cf44eaab : SUCCESS in 1h 25m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d2bfdfb38f34486e9d3d12f9c0706598 : SUCCESS in 1h 08m 25s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2e2f3c75644e46c3aa4361301f373269 : SUCCESS in 1h 22m 01s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0f937a7d7784426eb34757e16933182a : SUCCESS in 2h 00m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/61d9973286ff4fecbe41cd74d567e171 : SUCCESS in 31m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f57cd8df01f444969495d996e54df7fa : SUCCESS in 1h 01m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0207a69473b4b19934991f68fc8c735 : SUCCESS in 9m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e788f5e03a9e474d88a848c0592e27d4 : SUCCESS in 20m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0e424f2f40a24128bbfc14090d3c3154 : SUCCESS in 18m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/714a0400a02c4d4cb58ac58b5d729f0f : SUCCESS in 24m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cfe1681bc9b74c10b217488da1ccf57d : SUCCESS in 30m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bcf494926524793b60412ee9110f3b2 : SUCCESS in 6m 04s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2fadf72a2b9242dc99f8e73387b48639 : SUCCESS in 2m 50s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7f8d474822234b66bfe8eb7c571716c0 : SUCCESS in 1h 20m 47s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b5a14682e9094449a80db2e842e9fc7e : SUCCESS in 1h 11m 05s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/086608cfbcc24826a02812866f6244e5 : SUCCESS in 1h 23m 12s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c6922969304446c6975842c632951d4a : SUCCESS in 1h 57m 06s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7c44b0ce99c7432dba52d75137d0b915 : SUCCESS in 1h 47m 07s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4bfa5f132aa14dad8a31f738b8ba3493 : SUCCESS in 1h 10m 35s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/803a5ddbab6a4f9091482787ef5804cd : SUCCESS in 1h 22m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e47d2a11c74541e7ae2e9e80ff318e7c : SUCCESS in 1h 34m 10s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/09bafddcd2714dd7973bd9bfea4fa105 : SUCCESS in 1h 11m 27s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":7},{"id":"11d9203ed7701f1d723f76a79a4dc227f3e62d49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-05-07 02:23:42.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"0da9ee75dd98490265b7dd284ae7e1787ebadb5a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 04:31:06.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c33bec3be19145a986f7816664c77e71\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/20f6d2ddb3fb4f06b8a42bf7971ab1eb : SUCCESS in 25m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8ce7d97f182d402ebf549cef49ab03f0 : SUCCESS in 36m 24s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8fd37c5965364c22a63827374ab3ea7c : SUCCESS in 31m 12s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"ecdc99d5860804ceb121d48c14a2afeb61eb7803","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 05:25:13.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/90d11f8637d24ccb9d39a323900f65f6\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0090b3340fdb4f53a6a2f51a86958efb : SUCCESS in 3m 12s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/48a54340634f4243b637014084913e55 : SUCCESS in 1h 05m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/25a5425ad16e4c408c53e30b2869959a : SUCCESS in 1h 06m 02s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3214b5a8f6244e8db2d87d193fae5a51 : SUCCESS in 1h 21m 32s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bcc1c3587cf84508b8be942970a687b1 : SUCCESS in 1h 57m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ad29e1be3eec4ad0858227f22f8fcf69 : SUCCESS in 35m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9d3625d4e774af488d2d08a2e497ffc : SUCCESS in 59m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcc4bc14dc9f4831912289bf2c3f8cde : FAILURE in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2e92dbaf87fa46b2a57748fa0a097acf : SUCCESS in 35m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7e72f369aa0d4fadb3500a94c92f9f81 : SUCCESS in 27m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9702ecf946954b9593f03a10d52df049 : SUCCESS in 19m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f4c80468b2a64abaae9be7fb2baa8cf9 : SUCCESS in 30m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a15430e0e22a49f0af71d64b48339aeb : SUCCESS in 10m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/22888bc9cbf34d6390b32cf8548a41f0 : SUCCESS in 3m 09s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b50d210f73db473e9a1af7b3359892cd : SUCCESS in 1h 05m 36s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/fe99b3354306469dbf69a53df21eb1c4 : SUCCESS in 1h 07m 04s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/90dcdf53f192462a8396a2e28f1111e4 : SUCCESS in 1h 14m 35s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2bb7a3d27ba54965a3247426872bfc5c : SUCCESS in 1h 54m 08s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a5c997ed96114eaf84c8b093273465b6 : SUCCESS in 1h 47m 33s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d6f6b5adae054bfeac3fc1a3782897ec : SUCCESS in 1h 05m 38s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/fa7a043cb4ea47a4a9c099dc44fab53a : SUCCESS in 1h 14m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/871de893ab6a45f3b8f4c5c563d59afe : SUCCESS in 1h 20m 26s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/24345328e16949d3a5d9d68f47e492dc : SUCCESS in 1h 25m 34s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"1505a4027509b4bcb776a1cc3286ef9a223643d5","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-05-07 14:44:40.000000000","message":"Patch Set 8:\n\n(9 comments)","accounts_in_message":[],"_revision_number":8},{"id":"8cfb2ab5dc0a3b4fdeb16bbc4b13046d9d5d7382","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-05-07 14:44:57.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"fa3f9d10e65f8c51488758022474ec91653e3408","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-05-07 16:51:17.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"6254ff061f1db656db774c26cc446349fe1976ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 17:44:14.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c42d0bb72ce44269a31d34311cea69da\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/47abe709977c467290e2672300df47ff : SUCCESS in 28m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3ec00deff5bc46c9934045931dd7db30 : SUCCESS in 20m 42s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0f53dbe3a2c7481fb1b7fad6a45110eb : SUCCESS in 27m 34s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"de086cf72b443a7215b9c486113be43478529814","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 19:00:45.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6466b5cd850f409a97a0bbfab8aa1fab\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/63ad431fbd9541a2a02cded2e184921d : SUCCESS in 2m 38s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/40672bb85361484aa62a78ce4f89ede8 : SUCCESS in 1h 19m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a7d866348d78487da367493e087a54ba : SUCCESS in 46m 21s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a72687f2f306481fb624c7e40d2df298 : SUCCESS in 1h 29m 59s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/aae3d744117f4c46a3810e2b48127fa0 : SUCCESS in 1h 55m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/af395551ba1947258ececae3df365c1e : SUCCESS in 31m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7cb524f184e41aeb957c1954d5433f3 : SUCCESS in 38m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/464b7f39b2bf4f6c99b60dc84fed5958 : SUCCESS in 11m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7bca4f5e09ae4ad79ee3fc59454a852a : SUCCESS in 36m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2936fb7e259f470bb6f2047b91cad2ea : SUCCESS in 22m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/355ea14c5aef4e499cdc455640cb46c8 : SUCCESS in 18m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c4d49957d98344cdbb43603812a85a83 : SUCCESS in 31m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe8d98b87039476da2ecd8e84d7df7ae : SUCCESS in 6m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a883dc9117eb4fcda7ff6944474be84f : SUCCESS in 2m 46s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/738222f854a346bc854f49751fb782a0 : SUCCESS in 1h 27m 26s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/01d2c0b9c82a475589605d30c3cbb558 : SUCCESS in 57m 43s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/a92a15eac19847b6bb818a44e6331022 : SUCCESS in 1h 21m 38s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c32aac9efdc6491eaede0b57c618d7a8 : SUCCESS in 1h 49m 14s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/90ed22390af8499c918b062ac631ced6 : SUCCESS in 1h 25m 40s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/54619773972b4d05b0b1edf9d72c1f70 : SUCCESS in 50m 17s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/164dff412906437795bda919b9cace8e : SUCCESS in 32m 10s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ffad84a6be184aef82073625de7e602d : SUCCESS in 1h 28m 14s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/0def7c166cdc4562a2ccdc5961423b30 : SUCCESS in 2h 03m 15s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":9},{"id":"93fa966b5cca13a0137c2dd308077ec437ae50e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-05-07 19:58:09.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"29faeebb4a7832f2cb194576760398a4e2d932b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 20:28:57.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33ed48dec35046eba4cf61596ff749c7\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1e017e950419478681dab56d7b550d0c : SUCCESS in 23m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/07a3537c5ca14f8799a0bb6c75164675 : SUCCESS in 25m 59s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/623203a2f8524bb4ab081e3b789afbe1 : SUCCESS in 29m 12s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"1f4a82e47093846a81b93188f887249cbfcbec04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 22:45:12.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8d87e837650a4d40961ff7c56fe2a899\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/29cebfb6488d4b089c0176c4cc1e458a : SUCCESS in 1m 48s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/94bd6746fef0422d9c8627d2ada43698 : SUCCESS in 1h 02m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0cfa1d2311cb4c6c9ca353e2c2c89a84 : SUCCESS in 1h 09m 27s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/fbe70f235223460e869ccae78039b814 : SUCCESS in 1h 30m 47s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c8d2a636483a41129d92a11e5c1de8e9 : SUCCESS in 1h 56m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/17256a4a04404150a09e8c2ab79dccf9 : SUCCESS in 36m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0308661236bb466eaeb69308bd1e2e35 : SUCCESS in 37m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3d007887122465e86ff5f371658308d : FAILURE in 8m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fb792429f36e4890a85c0e2d2d852774 : SUCCESS in 20m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fed80e31d12245eeb5ae55eba4bcc297 : TIMED_OUT in 40m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4764fd18e3024724879e3d8ec77f998a : SUCCESS in 23m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c1b81c9ff64a4893b0814b24f7f552d3 : SUCCESS in 21m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/063fcfeb47524b4c88458e10b15f2c38 : SUCCESS in 6m 09s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a3028368a20e4fd0be5f73d32de5bcfe : SUCCESS in 2m 18s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a79af245facf434aa6d892a377381e2f : SUCCESS in 1h 21m 41s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/08c2574970334cf98580cadffc5ea233 : SUCCESS in 1h 19m 50s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/3d0b98f5c8b6498d8be511c718d0a62b : SUCCESS in 1h 19m 58s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/228b838d0c9d4ed28b09100098121556 : SUCCESS in 1h 05m 57s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fed9a8b2eb46475ab5bb6920f7d46d97 : SUCCESS in 1h 39m 08s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/598dd86078644c3d8ae40d3e3518132c : SUCCESS in 1h 05m 09s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/9e250501246b4b0e86dae31f8de81a68 : SUCCESS in 31m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bd8157c5183044e0bce37a061a6d33d3 : SUCCESS in 1h 26m 07s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/c43be49628744738a1fc71a412ea18e1 : SUCCESS in 2h 09m 39s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"9728a3bdb2edd8193bb11409a94db08f9a421289","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-05-08 02:09:19.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"abaa0adc644e520b6cd8e5cc9d059a4deac3754c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-08 02:12:08.000000000","message":"Patch Set 11:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/09/984409/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"0f85a2736136230cb295b59f989d06c4b921f34d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 04:28:59.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31b8ef6e568a4834a2f7ab9602d0cf77\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/75ab4e0db8ca4f52a0c56737e6a6319c : SUCCESS in 24m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3c253b1511df418ab08299b2d8fb7be7 : SUCCESS in 20m 58s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b83ccccc374e4b008a1e7dba7661ea2f : SUCCESS in 24m 19s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"c231733e06aff6ca7bee98f5b0e6d6533b076285","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 05:29:27.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/025afac305b4415f88edc112a7acf5bb\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/11c4a6acc64545a3b02db9a8d49aea1f : SUCCESS in 4m 32s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a747130fa4e4411e86103c2c52f31579 : SUCCESS in 1h 29m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8b5ad4f5b5594bf6bdd2597c99555db9 : SUCCESS in 1h 01m 48s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/aa468798d57f4b3b821d33adac3a4ad3 : SUCCESS in 1h 26m 03s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/28b4b88ff1fc4aad82fe97714d0f81ef : SUCCESS in 2h 09m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6089b2daf9fa4313b1578416fc134f19 : SUCCESS in 34m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ae0a3751442453d8e834d4bdc863b1e : SUCCESS in 1h 05m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83737a3ef7bb4fc6b17dea76e73add7a : SUCCESS in 8m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2292ee5497074f59ae2c96c1f82e44e4 : SUCCESS in 39m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/296af1dec60f4bb59ceeb5aa83596361 : SUCCESS in 29m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/37966d3ef7d740bb89958a72628ddf8e : SUCCESS in 25m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cc4a8af8c75b4b9096e7696713072ba8 : SUCCESS in 32m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7caaeb72aa0f45e586a3db66fdf42bad : SUCCESS in 11m 22s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6e815f5077c0482e93f33fc9047eda80 : SUCCESS in 4m 26s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/632b153f1ea84cf7920e07944f796dbd : SUCCESS in 1h 10m 01s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f8daebc3f25c48acaaee1460722945a4 : SUCCESS in 1h 07m 27s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d23f871302d84b53947be815a5c0931b : SUCCESS in 1h 09m 24s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c0791afce71041febedf902253bb8b62 : SUCCESS in 1h 53m 46s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/706b4b0b36c14c2dad08950e3be6334a : FAILURE in 1h 39m 17s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/901dd9dfd1294840bd2c22263168fe6b : SUCCESS in 1h 19m 00s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/256356c25ec843738e452a550d0cd098 : SUCCESS in 1h 12m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4bde965bafd94c29a9a782754ac41d58 : SUCCESS in 1h 29m 30s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/4e9320929b5f463f974fbeb336933cc0 : SUCCESS in 1h 12m 18s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":11},{"id":"7f7a70ad93a89ab4bddf4b775c4247670e2f9a8d","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-05-08 13:05:53.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"85094f422700422164a9724d140b6b7fe8194c89","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-08 13:08:31.000000000","message":"Patch Set 11:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/09/984409/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"cf774dabc21f44218d002b644fdc551efa03cc37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 14:59:32.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cbf1b6b7eb745cfa98ec835a27dbe5c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/85b3561ac6264f3b8053d25b8c8ae11b : SUCCESS in 3m 13s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7f2ddb23779d4278a9c3bad64c18e20f : SUCCESS in 1h 18m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3e7dd922d1b34a1487850f5711d7cd06 : SUCCESS in 55m 57s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/07f8c6e4f61d45ffbb0a6c8e58ad9a40 : SUCCESS in 1h 36m 48s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d6e245d34ae3428391d17bb0cff16262 : SUCCESS in 1h 51m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1307c56a2f25469faffaf0a3d976156e : SUCCESS in 30m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f7def99779d04359851f320344925f98 : SUCCESS in 49m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7d41146fc4342f091cfd881ac21c9a9 : SUCCESS in 9m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fe6a4c748cbb46908cdcd03024eec433 : SUCCESS in 38m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8295435f59554c99b894f88ffe8b259c : SUCCESS in 26m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e487315eb3874418915b8d7f9b7a4fba : SUCCESS in 25m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b2896c3773a249f1a27805ca793edaab : SUCCESS in 32m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d7bbba15d174f4dac68173d08ada729 : SUCCESS in 7m 36s\n- requirements-check https://zuul.opendev.org/t/openstack/build/91b160cee2354403a448e1d6dc2a4513 : SUCCESS in 2m 01s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f574ed16545c4294bad22964a458684e : SUCCESS in 1h 06m 12s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/daf5f7002d7444119f7c20b830a274ac : SUCCESS in 1h 11m 47s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/3ace2a4b5896439cb006b21eaf81a34c : SUCCESS in 53m 47s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7a62a086b0ba4cf7b71597168b60c519 : SUCCESS in 1h 07m 09s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6bc14162353e4f61a8fc58ac92615dd6 : SUCCESS in 1h 16m 56s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ac491c78be6046228dfda946c5850b8e : SUCCESS in 1h 05m 16s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3b7b6c9c4f1c437b8c5d3415c9247f70 : SUCCESS in 1h 13m 44s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0e77c64897104a41876cc021b4d35203 : SUCCESS in 1h 30m 03s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/10fdb233655e47559b94cfade388bb0f : SUCCESS in 1h 24m 31s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":11},{"id":"5018999f3f62aa5479f92859cebcaf622f88bd1c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 15:03:06.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c7d8f24304d423a9272f03c96a53326\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/84749ab6a4c0464c8a30142b1b18fc90 : SUCCESS in 27m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/de67ff86ecdb4a3c983e8f6a47763646 : SUCCESS in 19m 25s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/94cddb3f70b34731ab738a84118f98b1 : SUCCESS in 26m 04s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"02a17dcaccf859a8923aa54f67a0e4d0383a279d","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-05-12 18:05:12.000000000","message":"Patch Set 11: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":11},{"id":"ac22c490690a98f28566e8dd4e188854e664e27b","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-05-12 18:47:18.000000000","message":"Patch Set 11:\n\n(3 comments)","accounts_in_message":[],"_revision_number":11},{"id":"32d33ad155429b7fb67045406b8310a9be5e9500","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-05-14 02:32:27.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"057e53b983221a26957d2df1530251975cdf0583","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-05-14 02:33:52.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":12},{"id":"ddcc3bb38566180ca96ca90ce020337b93dfb8a1","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-05-14 02:37:07.000000000","message":"Patch Set 12:\n\n(4 comments)","accounts_in_message":[],"_revision_number":12},{"id":"b443d45bc5ca1e16e59e96dedc8496ce94b9518d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-14 03:14:22.000000000","message":"Patch Set 12: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/09/984409/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"e39e658fbe8e889d960cf049435c2f6a3233f423","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 05:04:48.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/080e72b78b144a6fb73f599c4addec2d\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9f561f4a31024f439c522cc1a927c3c8 : SUCCESS in 20m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2d1f453f529b41b4bf8a8b3c7ed71f5d : SUCCESS in 33m 23s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/9c85692f35c240bfa881c3ebd8c60f43 : SUCCESS in 1h 00m 17s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"b42a22fd361fd0d57cc10fbb07a44bb80a68f967","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 05:15:29.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97463407e61047489baff3a1c0bb8ac8\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/18d5bc1df9c44a11bcee0a84966e5388 : SUCCESS in 4m 01s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/83fd64a099c9448f8221f1a4090ae6d8 : SUCCESS in 1h 26m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf3b5b34f2df49b18df2a3aa971a6554 : SUCCESS in 1h 07m 59s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e9d96a9138a74bea9560bb0040781604 : SUCCESS in 1h 24m 33s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/307c7ad2cbe545439ef38d3e614b7b61 : SUCCESS in 2h 11m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/27db6922197845b09064c2b8d8da094b : SUCCESS in 32m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5904498ae5154e18a1a168da991c5c92 : SUCCESS in 42m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/923e4851ad844695ae50d701c461d5aa : SUCCESS in 11m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e5de2ec6949148a3813c3b139799c9da : SUCCESS in 30m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/31ae622901f64fbf8ee3c52813bda7ca : SUCCESS in 23m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8ebb9b097b3d4abe83e63439d9acacc5 : SUCCESS in 30m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/903b27e916e54f87be22a3f99b432bc5 : SUCCESS in 13m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/41f1ef4f07c04e2789290fac2f58b029 : SUCCESS in 2m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97ba4d14fc97408481caab863027329f : SUCCESS in 7m 14s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/631dab3ed1cc4a1ca285cd00117d13a6 : SUCCESS in 1h 25m 25s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/59324e385faf4bcb923cc6dca29cd908 : SUCCESS in 1h 10m 20s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/a0980f9a10dc44a8b24b25a33d991db1 : SUCCESS in 1h 22m 19s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2d528e8bb9954712998519daea0cf588 : SUCCESS in 2h 01m 44s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fb078b0b06604d599ad7cce86bf19cf6 : SUCCESS in 1h 42m 00s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a1a2888b4d0842ef8175e5c0714c3e3a : SUCCESS in 1h 00m 36s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5db207173bef4e95977d85d20080b4fa : SUCCESS in 1h 14m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/59f8829aabb14d84af389543beaa88b5 : SUCCESS in 1h 28m 09s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/7056a0a525684890a2c1a28cb51ca90f : SUCCESS in 2h 20m 18s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":12},{"id":"9e1a97def3637ccf94f17a9d6e4dde9430693eae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-05-14 14:15:35.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"dc94673c34393494e58628257179d8e9f5d9fdcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 16:30:19.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/431f2a0325494f188207c356ef9af767\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4182111d698541aeb1b37d912bdb8c54 : SUCCESS in 3m 02s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c2dde1f4168b41d9ac290088dc6de5c3 : SUCCESS in 1h 46m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/66ef86233a2f4e06b109ffc9e1a37125 : SUCCESS in 34m 24s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/be4b8ca2ec174864a0e40fc01088f143 : SUCCESS in 1h 44m 35s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6721d3eb4f4d4b8ebeb5a3ac156028b8 : SUCCESS in 1h 56m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4564a660184440afbeb290dee9544dcd : SUCCESS in 36m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4bfcaea981f14acdb2c8f3bd97e73102 : SUCCESS in 1h 20m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88b7c9f9273a47b0b6fdc5e3cabf930d : SUCCESS in 10m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c64c36de697248dabf0635dfb2d3a967 : SUCCESS in 27m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4cdaad35879a49f680ce45db2a7471c9 : SUCCESS in 28m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/26dbef2ca2184f1185b0799d3d0cbf9c : SUCCESS in 39m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5fe1aa0e23a4bceaf9094b0b69d7aba : SUCCESS in 12m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f56bdb1cdabd4c9a80e11d535a7728c4 : SUCCESS in 3m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80612d569c044b4080d648d470643a28 : SUCCESS in 12m 32s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/20c81c703a9544e895ce3c35b799e4c6 : SUCCESS in 1h 33m 44s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7b1d369327a24572bb8bd4c3bed9027d : SUCCESS in 1h 14m 44s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/fb3a06b4515942ac99e98819954843c6 : SUCCESS in 38m 39s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a88ae5d06e144f63b3340f53e369c9b5 : SUCCESS in 1h 51m 27s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1d72b401d38d41599724458418a4fe46 : SUCCESS in 1h 43m 17s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d2de4d00c5b743b0b4f25da18429f806 : SUCCESS in 1h 11m 26s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/291e11d13a1b47a3be7b0e2508c9a75f : SUCCESS in 1h 21m 17s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1542e64a6c054eb08354d2476921e7b6 : SUCCESS in 1h 30m 59s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/f96620e81d884b7b91d7640f1904a564 : RETRY_LIMIT Host unreachable in 49m 52s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":13},{"id":"13c47697cdc93327a28673f7623dada80e2d4890","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 18:28:55.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73bec410717e4866bc64786ea3412b7e\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c5540ac97ae74487a209a631a0b4b8e2 : SUCCESS in 49m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/570bed40c7354eb1b1de109e5658fa30 : SUCCESS in 26m 54s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c5d76dd8548d4541b2e9a2cde5b9f0cf : TIMED_OUT in 1h 21m 00s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"c0bc5027a1a4785ee49604de85def4621062cbc7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-14 22:08:44.000000000","message":"Patch Set 13: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/09/984409/13 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":13},{"id":"0247fb2d1c0b22a6281dee22271c6ec67dc1b802","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-05-15 02:51:42.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"f0fcf8f2f21a72c1b4eb888123f634d356922327","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-05-15 03:20:04.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"58511d6290f70a1f91343e605d0d095e91b73bf4","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-05-15 03:24:58.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"a781f7bf50c12b333fae3ffaac57cc9f29a7a24c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-05-15 03:46:08.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"461268e46df88835d8af21731d95cd85a829a0a3","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-05-15 03:53:16.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"5b9b34e8f434e86c89015accf4993787a1c3bfda","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-15 05:10:00.000000000","message":"Patch Set 14: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/09/984409/14 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":14},{"id":"acb73714b66345b7361fa2b50173e09a154f52ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 06:08:12.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/daba8ad6c4604b4482f99cf9741a24f5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a43b6ec074e442d89f2b884d0f1fe115 : SUCCESS in 2m 05s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0d5aab0eb55f4c32a48cbdb9fe1e8c56 : SUCCESS in 1h 34m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/53ab1166ab50456abfdcdf44ac85040f : SUCCESS in 1h 04m 21s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/69cbae71fc3845baaee97e5727f57fd2 : SUCCESS in 1h 18m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/77b9ab68bdf64fabac14fb8b2aebd446 : SUCCESS in 2h 20m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/eb07c5ee1fa64d61bce5775651cc1e81 : SUCCESS in 31m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eec0ccf1571d4a4eb396dea25436dd65 : SUCCESS in 44m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f07809fd0e1f4685b36b020a23847e15 : FAILURE in 3m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/63123fb7762a46c785f0c10d23f5564e : SUCCESS in 25m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bd447417a8ce4ccb820d7f7e02562a62 : SUCCESS in 24m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cd5276126f1b432887aa564cdf0b1b4c : SUCCESS in 32m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b1dc1b4e4274f8bb7adea4afe05ae69 : SUCCESS in 12m 49s\n- requirements-check https://zuul.opendev.org/t/openstack/build/30888e98d99b4d2e93c4fefc27fb07dd : SUCCESS in 2m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a237449527f4f9496b1515cfe20e0c5 : SUCCESS in 10m 03s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c6f239a471ac46fe888634a5bfd36ce5 : SUCCESS in 1h 20m 27s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1e5494b37ddf4fd190892cce52d310a6 : SUCCESS in 1h 07m 22s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d103add1bc554088bef31afba53daa79 : SUCCESS in 1h 15m 39s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6314b49d8df34055b7fe8432155dedd3 : SUCCESS in 1h 13m 01s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d76d0fa7ad5d4d17a748c40555f4d8a4 : SUCCESS in 1h 03m 28s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/92b6e8759e4e4f2f8c10b5aaea3936f9 : SUCCESS in 1h 04m 51s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/899283d20d7c431fae7de6991fc61215 : SUCCESS in 1h 23m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d4e8cc507ddd4dee9944fdcfa7261066 : SUCCESS in 1h 29m 41s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/b71c224ff153440fa6f39272bc148ae0 : RETRY_LIMIT Host unreachable in 1h 02m 30s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"4520aec367b4d1e13a5ef2bc45052c626b64ad3f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-05-15 07:03:40.000000000","message":"Patch Set 12: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"a9bc51956f3b7fff1afcd3a7ddd91ab9ef53b33e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-05-15 07:05:23.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"8cf94354fa23b232631e06cccbeff20562da1ce8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 11:16:59.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86197baf0d474f99b8aecebd3723c03c\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ddb7120472b64be28f767a2874474ace : SUCCESS in 27m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/644d7d21fe3d40f09f65e7fae6471a4d : SUCCESS in 38m 58s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/10fd00956f6a4e83b536dc3a0e3f188e : SUCCESS in 52m 28s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"c0b0f072a20f8839e7ac2f6f31bd765761763590","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-05-15 13:24:28.000000000","message":"Patch Set 14: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":14},{"id":"bb65e6d82fc84965a09da2e5b04c6e059ec15dff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-05-15 16:54:27.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1, Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":15},{"id":"7a67a9e9ed526b914c764f9b8042edc08cb9fe87","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-05-15 17:04:04.000000000","message":"Patch Set 15:\n\n(11 comments)","accounts_in_message":[],"_revision_number":15},{"id":"003ec504e8e4f0f6e26038d0711cbe866ef04dfb","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-05-15 17:24:37.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"40e94091925891931c1325b998ed23057743bd0b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-15 17:37:13.000000000","message":"Patch Set 15: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/09/984409/15 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":15},{"id":"1f5ca9adc959a59ce33e66e2fc781061a608d110","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 18:42:15.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97b44137fe3d4da4b5715e18b70acd1b\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8a3bceb83175489291de70cea9765ed2 : SUCCESS in 42m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8b2181ec48e14d0187bc3b944cccae99 : SUCCESS in 41m 38s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/39dcc9cb294947108b2ce758263786bb : SUCCESS in 55m 52s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"0ed24838d03729c92997e1f7b5e52df2807e86b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 18:48:26.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fd8f87eed7c441888b72c500ae32131\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cfa47544a8b74923bd13a6826489860b : SUCCESS in 3m 03s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1de0acb2c43c41b787b47da36592d69f : SUCCESS in 1h 31m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c1bd0f66e8ed452496ffa157205a1dc2 : SUCCESS in 1h 08m 06s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9e0327399d634ee1a2bc6f395aca024a : SUCCESS in 1h 22m 50s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/33dafe52996d4da9bd758292427b9d67 : SUCCESS in 1h 52m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/54c22d89f988445b865a36cbc936c7a1 : SUCCESS in 31m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6858cc046be3499aa05f88655c464fab : SUCCESS in 1h 09m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bbe602732b84d41a079a2477fa95ec0 : SUCCESS in 7m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f34a1740962440dabb56d45067b7f42b : SUCCESS in 33m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b06d1a133f3b424a94288449a00d262a : SUCCESS in 22m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/194c928c970f476697f65a638574e5e6 : SUCCESS in 32m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38332ba3001141be8f035de3aeba5060 : SUCCESS in 10m 59s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d5b179cb90f64fac8478c80be36a7fc4 : SUCCESS in 2m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3f53f04a2abb4a5dbd337e0e4fa721b7 : SUCCESS in 8m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ddead0ee37094aeba355b504ee245089 : SUCCESS in 1h 09m 32s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/181fee17cb46497da75b421acdf4e152 : SUCCESS in 1h 14m 32s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/c128536a76b54b3987ec9fb6078aca44 : SUCCESS in 1h 19m 23s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4fc6baca7e16433ea823c4d4c65fc706 : SUCCESS in 1h 45m 14s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f5c30dd0ac924a7c928345a8777b1f49 : SUCCESS in 1h 19m 15s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d86ccade4288429899ac8fe84350491a : SUCCESS in 1h 09m 34s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e6c2bb48f3cf4c45b38d1f4b57ed415c : SUCCESS in 1h 12m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e77cdf31157847fcb67073f0617ffcc5 : SUCCESS in 1h 24m 01s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/347bc110e7f643e6a074becec45081b5 : SUCCESS in 1h 26m 03s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":15},{"id":"2ad66d894eb693815aff9ddb250b13bdbbab4c67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-05-18 02:10:07.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":16},{"id":"7fe3caec064ea3b21fbb5931ed6129cd25d261f1","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-05-18 02:10:25.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"f9649c194d7da56cf1d69aaac807f8172f054984","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-18 03:29:04.000000000","message":"Patch Set 16: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/09/984409/16 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":16},{"id":"677450f99689ee14c6453c12ab16c75369768f17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 05:07:45.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8a6505919a845838c32de07ced948f0\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9c15f2efca584577ba3b5471a5b3d6c9 : SUCCESS in 3m 09s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/be8e52db42da4ee2bd5f0524a2a460bb : SUCCESS in 1h 28m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8db05c0689464d62b534045e2a7d49fa : SUCCESS in 1h 07m 01s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7b0b7f71a97143b38bf1b60a25799102 : SUCCESS in 1h 32m 45s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8abff13cce624d0a9a2d789d1b2b4344 : SUCCESS in 1h 53m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/60bec761b7f34f3aa5e0b6af21e40f5e : SUCCESS in 33m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7cc27ec060aa41af81063f3a149e0219 : SUCCESS in 1h 04m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bc4176f5eaa4ef6a0485a421ae1c215 : SUCCESS in 9m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/37c45e66364a4bfb92977b9f7f5c0072 : SUCCESS in 31m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1b04895de6c94881bbaf96f77232d8c5 : SUCCESS in 24m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7926218a1ab0407295bc1ef74aac96d3 : SUCCESS in 35m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cfaf094d66641c2aca92a9e3db7659b : SUCCESS in 12m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b900e8ba011640ce93606f2492da64bc : SUCCESS in 3m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bfc9cb31c0a94114bb3d3025a307ffe1 : SUCCESS in 17m 58s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/15532871a9804b3294c68b31ec5dd517 : SUCCESS in 1h 18m 48s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9564b3fe732145e2a4f7e640e8fb68dd : SUCCESS in 54m 37s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d0849d3a4f9049dd8f8b4d9e6f388348 : SUCCESS in 56m 51s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4b9780c951644e8095fe31541e0f0c1a : SUCCESS in 2h 06m 40s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f06c23d040a84972bdeef091b3ee76cc : SUCCESS in 1h 39m 20s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/239b33b43a4844439e39ff53b18c7f12 : SUCCESS in 57m 03s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/624adf9cff0e4aa48628bcd18eddec80 : SUCCESS in 1h 06m 46s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c4cb93c8e1bc44d4baffb4cf358e844e : SUCCESS in 1h 29m 15s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/568860d7fdcf4fcbb7557c7d01a63056 : SUCCESS in 2h 20m 56s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":16},{"id":"4ee46ec04b8a11360fb8979159108ae19a6412ee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 05:42:29.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fdba017e9474b7aac431410239110dd\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d94c7f67ca374a2385b66743b8d76e65 : SUCCESS in 57m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/45d36222e7a94949a603cf336b72e622 : SUCCESS in 22m 21s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d8fe56ad26764a4e9e032394bc794fa0 : SUCCESS in 50m 42s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"9e018025d70b8d05c88eae432ed96b7f7d9b8933","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-05-18 07:06:43.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"2e29e02c25235e926dab34e4b2edfae2936e82ef","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-05-18 08:17:41.000000000","message":"Patch Set 16:\n\n(2 comments)","accounts_in_message":[],"_revision_number":16},{"id":"a78c53cacfce023f7702d68a4480bbb7d1fa06e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-05-18 13:01:30.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":17},{"id":"8eabdbe1c0460b3cefc8455f1fe961625320623d","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-05-18 13:01:42.000000000","message":"Patch Set 17:\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"3820c482fd580dba0f5ebcc337d01f9532cf2fb7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 14:25:54.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef95a6a80b484ab88d11700128e09a4b\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/498723adf24b4d729466e07f420e6833 : SUCCESS in 1h 16m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/db4a466bc99440cda1c48765f947e65a : SUCCESS in 1h 05m 26s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/367cbdab938c4196a87feb68100bc242 : TIMED_OUT in 1h 21m 13s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"cadd46d6a35b4d5f133ee368e340af2693dc81f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-05-18 14:32:34.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.","accounts_in_message":[],"_revision_number":18},{"id":"138dce0ace40c99acecec9cd008d851fa00f8e3f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-05-18 14:36:39.000000000","message":"Patch Set 18: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"e5fbd3b2d60f7d68212315c685b89953091377c4","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-05-18 14:58:36.000000000","message":"Patch Set 18: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":18},{"id":"09a78dbac6b8e956a2776f5ce301552fb990fb6d","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-05-18 14:59:02.000000000","message":"Patch Set 18: -Code-Review","accounts_in_message":[],"_revision_number":18},{"id":"e86866a9a830c3669de737730a132e367361900e","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-05-18 14:59:50.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"c8ddd81617f2f5620ac6f7d343c16e4db62c4086","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 15:31:04.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77fa3b99ca83471eb16211d930e05271\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9a43352f79f84c05980daf70c308bdaa : SUCCESS in 55m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9287a1503e6d4a4e9a1450d30a7758a0 : SUCCESS in 48m 32s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0b34b6c6c09a4bee8a19b9a94afc4e98 : SUCCESS in 33m 53s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"3cfbac7df3aaf78d04f02e4455cff3f510fb5dcd","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-05-18 16:09:45.000000000","message":"Patch Set 18:\n\n(3 comments)","accounts_in_message":[],"_revision_number":18},{"id":"a2044703dc7d448ef1fd503e3dc51f89e94d157b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-18 16:37:06.000000000","message":"Patch Set 17: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/09/984409/17 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":17},{"id":"65f48250b83f7b6e73051933c5e942293030368d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 16:49:29.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c75eaed4e4d7483189c6e1868137b10c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/bedc2bd502a34365b55126aae55284cb : SUCCESS in 2m 27s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1c94afb0ae2b43b5b378158477756543 : SUCCESS in 1h 29m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d5cc98df3f4244d4b67047990650d147 : SUCCESS in 35m 34s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/329ea1c6e8b04de0a238ae16516a64cd : SUCCESS in 1h 35m 01s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5497e1d2440546cbae881ddbda318ed8 : SUCCESS in 1h 44m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fda715593c04436a80c3b08cf0b699e8 : SUCCESS in 32m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38148faa5edc4bd2b324af5ead721391 : SUCCESS in 43m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52f55deacedb46199c266e556a5eb171 : SUCCESS in 10m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/58e3f9d4f87e412b886aac76c8b21292 : SUCCESS in 35m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/49c02d04a9b74869bd83988d59844a22 : SUCCESS in 31m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1bd3478654274daf9afc0b0127c4bf97 : SUCCESS in 37m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff20b499ad04466f92132ccd388c28c1 : SUCCESS in 9m 19s\n- requirements-check https://zuul.opendev.org/t/openstack/build/530f023a49754c099dc6743388331b37 : SUCCESS in 2m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df686557627e4a578233341fc853feab : SUCCESS in 19m 33s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c7c2bfb46fa44a32b93744c1ce2acebf : SUCCESS in 1h 11m 27s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9509cbd5253e49f694c2bf7d7fae9679 : SUCCESS in 39m 54s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/2bf0ea217cd04d45b65063a8c44f576f : SUCCESS in 1h 16m 25s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/25df011b25424afcbb46f967d66c330b : SUCCESS in 2h 02m 38s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/05ea9300f0cd4d6e80b69715e10974c2 : SUCCESS in 57m 47s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b855db493dd0421396def5e74c9ae06f : SUCCESS in 1h 00m 21s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/323cd92a113d44b38cbae7ffddff06a9 : SUCCESS in 1h 07m 31s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/79485e5285834cedb47f37d08572be08 : SUCCESS in 1h 03m 30s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/1ec838c5045d4318babb3b5259ec8bab : SUCCESS in 2h 10m 31s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":18},{"id":"0325ec3bbe96b59418fa9762ee157fbc046ee37d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-18 18:03:03.000000000","message":"Patch Set 18: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/09/984409/18 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":18},{"id":"719608344e0c1f979141b3c0b039ae2c0f5f4796","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-05-19 09:20:04.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"8143ea1580182c890d81c574543e1ddb49d4877c","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-05-19 13:50:30.000000000","message":"Patch Set 18:\n\n(3 comments)","accounts_in_message":[],"_revision_number":18},{"id":"bc1f4ee30298baad2e9fa561b8823d2f99fa5231","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-05-19 15:53:09.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":19},{"id":"b4e57c6197a3505c1f076801f1cef921bb92a15a","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-05-19 15:54:03.000000000","message":"Patch Set 19:\n\n(3 comments)","accounts_in_message":[],"_revision_number":19},{"id":"60f854b75f1e311dd63e914a723bc6afb7ce2ea4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-05-19 16:27:38.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"1051fe3fa9bb5eff947c2343e2571388e364ac24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-05-19 16:52:15.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"4c74c1fd2297e53cf76ae3c0d8b771d3dfdd5304","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-05-19 16:54:57.000000000","message":"Patch Set 21: Code-Review+2","accounts_in_message":[],"_revision_number":21},{"id":"6a8017bd4c2820a45778715162ca847f61da27e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 19:05:12.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/322a9e050d1b429bab5cc865fc2fe0a0\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/77a518e6b40847a09d5125951c4d2af9 : SUCCESS in 1m 44s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d563cdf68ca145178a389666827d95e5 : SUCCESS in 1h 40m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/226ee875d0234d9c9fc9c774276d45ea : SUCCESS in 45m 28s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d0f75fc4ef3346c78f6dc6897a9206c4 : SUCCESS in 1h 33m 44s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8e37c24460cb4533aabf570b041b9df0 : SUCCESS in 1h 59m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d52cd86dca42470790709f5bd931ee15 : SUCCESS in 32m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ec47edac54046799c6c0f89b1752f8b : SUCCESS in 57m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c80a8677428453ba8120e8895868a6e : FAILURE in 8m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/08e8f5de3b4045fdb432c6aed471ab9f : SUCCESS in 30m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d9f14c1a8234424eadff914632249588 : SUCCESS in 26m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b7c07aef531d4363b6eace566b1dbfaf : SUCCESS in 34m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4887322248cb475e8d3178bc0a6606c3 : SUCCESS in 11m 56s\n- requirements-check https://zuul.opendev.org/t/openstack/build/90efb8a1003142219d565c33dcc4c974 : SUCCESS in 2m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80b63ff371d14f708d520a282e820ff4 : SUCCESS in 11m 43s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4e375530a93c45f7bde5d96635fa19ac : FAILURE in 47m 42s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/473822a3687f4840ae3b0015f60226e1 : SUCCESS in 51m 37s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/958b3b297f34429280244e8046b4dfd9 : SUCCESS in 53m 29s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6afd7072dc9149a787ea709635264f20 : SUCCESS in 2h 05m 30s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/789444a1af66423ca0df5fa242983659 : SUCCESS in 1h 29m 58s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/777fbf03dce146f9803fc7292c5ea4c6 : SUCCESS in 52m 53s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/643d604d1e0e4c2f856caebc9208795b : SUCCESS in 1h 28m 37s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4402988022904df7a4558d64c1f294ad : SUCCESS in 1h 11m 52s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/9d03b0b08d174e11acf755d1aa228c6a : SUCCESS in 1h 28m 14s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py\n  Comments left for invalid file neutron/conf/agent/database/agents_db.py","accounts_in_message":[],"_revision_number":21},{"id":"259038b74cefae78f853aacc9b5cd054915f6b8b","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-05-19 19:58:21.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"4c9c2373801c8884ba79913cd65ce1082e39177f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-19 21:04:40.000000000","message":"Patch Set 19: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/09/984409/19 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":19},{"id":"cf0dd65f31d31681c52fd673ea0f61ac91b6fd8b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-19 21:46:50.000000000","message":"Patch Set 20: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/09/984409/20 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":20},{"id":"9db44f828ca6cdc8794f4fcc0b9784f05c69800e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-19 23:12:20.000000000","message":"Patch Set 21: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/09/984409/21 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":21},{"id":"5188e32cdf960628d3f5669cbe71325903f0eba5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 23:32:06.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/414907fd9b4340bb980140be9056dbb2\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6c65e5f82df6413999b01d157de01eef : SUCCESS in 30m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3ffd0d566b6d43039a724ac8f7ec09b0 : SUCCESS in 47m 46s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2c6834b893be4dac936f6fbe1d72b70f : SUCCESS in 1h 10m 51s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"4d0dc8a9d8013a153e7b2ae973f7e6069c57d868","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-05-20 06:14:37.000000000","message":"Patch Set 21: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"18cdc2994fb0c469d2345c014e41a06d69038cde","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 08:21:33.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e1bee4b44654bfdbb6ef975536bc6a0\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/11bd8b1e359a4b5b8bb11b4d0aabb8e7 : SUCCESS in 20m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d5e78d51bb2245508295840e86847d19 : SUCCESS in 42m 42s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e0d722c95d614b5ab6e09a0d1ec181ca : SUCCESS in 32m 43s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"7548931d2c5b5d7813b19de56ab3a7b0b678947f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 08:24:05.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec45a41c8d544aaab06349cca8d0639c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6e411380adb547df8ad84bd19b44ba70 : SUCCESS in 2m 42s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e8e7ce6bffff4a1aa20325dc79ddb90d : SUCCESS in 1h 34m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/929dd70cc1214988be762b364c6a536b : SUCCESS in 45m 48s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a8305fefb9464b4f8139a7cdbd3ddab4 : SUCCESS in 1h 27m 39s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/af4b76e1ad754333b4ee4b1bb50f993e : SUCCESS in 1h 46m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/401896110ec34af19479993101fa1a67 : SUCCESS in 31m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/23c4424829904449a9b3650317747019 : SUCCESS in 1h 20m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c73e4997fca481cad7b329582327939 : SUCCESS in 12m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/065b8901abad47d082bb198b393fa9bc : SUCCESS in 29m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1cdba33710c04ab987ffe8c4530096c4 : SUCCESS in 20m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1e6af1998d6242d2ac38f29ef2f5bed3 : SUCCESS in 35m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4f41634dd7b46cabec158d3afc9e062 : SUCCESS in 13m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d55ad020061d4d08b9c66c09c738d4c8 : SUCCESS in 1m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6400a59a0294443cb349f13b08ad5c77 : SUCCESS in 18m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7746e8d16e6e4ce49b75fd2e6aa6cb02 : SUCCESS in 1h 21m 01s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5f208cb73f61449aa3fccce770489ead : SUCCESS in 40m 11s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/3174053fa0614a109f67d6b0c98bd90c : SUCCESS in 1h 18m 03s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1929623decba4cef80e0e9aab1f698f7 : SUCCESS in 2h 03m 06s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bafb4466e7a44d48b24df11f7190c641 : SUCCESS in 1h 44m 52s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/81f6ea179cc748adbc2e2e0b04033996 : SUCCESS in 1h 10m 17s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/6b47b25858bd4eccb9f7be8417d69dfe : SUCCESS in 31m 02s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/35dae8ee786243d7819ec1542926a117 : SUCCESS in 34m 41s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/d03c37ede2144913b1c8a42839e0ab3f : SUCCESS in 1h 23m 48s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":21},{"id":"95bad175a386094d820ae25e8f98bd6ea8a236ac","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-20 12:48:27.000000000","message":"Patch Set 21:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/09/984409/21 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":21},{"id":"49d541b6039217a84b1943a2bf923ecb317b4fcd","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-05-20 12:58:23.000000000","message":"Patch Set 21: Workflow+1","accounts_in_message":[],"_revision_number":21},{"id":"11451605d775dfe087b26a7ed8cf657619ee4749","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 12:59:18.000000000","message":"Patch Set 21: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":21},{"id":"479934b55896355af408267400c8ed43a908e0fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 15:19:43.000000000","message":"Patch Set 21: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c678a0bb92df4b6db717965176005fa0\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5e87f2e90aed48af907ef9dbab992212 : SUCCESS in 3m 00s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f4c4ffff1404476fa85106889ee91d27 : SUCCESS in 1h 32m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2bbf5d454d134a3db76343493280dcf4 : SUCCESS in 1h 07m 38s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2a96a86580014d5b9732666ed86da9d4 : SUCCESS in 1h 20m 17s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/379e90ef921b4e3bad752dd19fc7b8bd : SUCCESS in 2h 01m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d787182e7b44c0d9898ee395a3394a3 : SUCCESS in 10m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4687b11229694934be0ab0d6f51bc554 : SUCCESS in 29m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/09beca4eb7f544c4b6a3516ad98dcc0a : SUCCESS in 20m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/740e501e7505443187dc838bbaecca75 : SUCCESS in 13m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0e1f57e5c5f34487813d4d956675db0f : SUCCESS in 2m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/184a62d5104142ccb418d7d3874cb328 : SUCCESS in 22m 12s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/008b6c1ffe934b4a89b7a7bc43ccba1c : SUCCESS in 1h 37m 42s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/80a6ac4264a6495bbfe81814119e5ff0 : SUCCESS in 2h 14m 45s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/5dc713f7845c4894b5af0acfe8ef93d4 : SUCCESS in 43m 07s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/22f5c6b5ca5f400eb0dc3f19453d16fe : SUCCESS in 1h 24m 54s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e057d71ed88c41348211c2dce4decca8 : SUCCESS in 1h 11m 39s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":21},{"id":"e67a1dac50713d1aaa1a73db5e1b1a895f81d1a8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 15:19:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":21},{"id":"ad6785ce0a5bc07453eaf7872de8414c77358fd9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 15:22:58.000000000","message":"Patch Set 21:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a92d9084a0194a8eaedc670a68b6a0e9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d74b38e8a3a4d3f97672b5d5fe897dd : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":21}],"current_revision_number":21,"current_revision":"0b12b3a845287fa052e4b63823e98aa442e40547","revisions":{"1fbf69b0b8af095c589e31d03515cb84c6c467b2":{"kind":"REWORK","_number":1,"created":"2026-04-13 15:13:56.000000000","uploader":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"ref":"refs/changes/09/984409/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/984409/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/984409/1"}}},"commit":{"parents":[{"commit":"54befe6c096275cfe0049a2f1ab1abbee3140c8c","subject":"Use unittest.skip decorator","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/54befe6c096275cfe0049a2f1ab1abbee3140c8c"}]}],"author":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-04-13 02:15:09.000000000","tz":-240},"committer":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-04-13 15:12:32.000000000","tz":-240},"subject":"Added in OVN Agent to monitor netlink msg","message":"Added in OVN Agent to monitor netlink msg\n\nAdded a thread to OVN Agent BGP Extension to monitor netlink messages.\nOnly RTNLGRP_LINK messages are currently monitored, specifically for VRF\ncreation and deletion.\n\nAfter the monitoring thread subscribes to RTNLGRP_LINK messages, the\nmonitoring thread receives messages related to a link regardless of the\nlink type (bridge, vrf, etc.) and regardless of the event (creation,\ndeletion, link up, link down).  Thus, it is up to the message handling\ncode to filter for the VRF link type and for the relevant events\n(creation, deletion).\n\nClaude generated the initial code, given the EVPN Type-5 route spec and\nFRR\u0027s code to monitor VRFs.  Claude was also explicitly instructed to\nrefactor the initial code, including moving netlink start-up code\ntogether with thread start-up, splitting the message handling code\nbetween generic loop and specific messages, and refining the vrf\nmessage handling code to not break out of the middle of the function.\n\nRelated-bug: 2144617\nAssisted-by: Claude\nChange-Id: I76a9f0fa27043d374be01b034af0ed1009c23ef8\nSigned-off-by: Helen Chen \u003cichen@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1fbf69b0b8af095c589e31d03515cb84c6c467b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1fbf69b0b8af095c589e31d03515cb84c6c467b2"}]},"branch":"refs/heads/master"},"df21c80008b059c97e59f6806006ba9b2a4484d8":{"kind":"REWORK","_number":2,"created":"2026-04-14 16:00:33.000000000","uploader":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"ref":"refs/changes/09/984409/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/984409/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/984409/2"}}},"commit":{"parents":[{"commit":"54befe6c096275cfe0049a2f1ab1abbee3140c8c","subject":"Use unittest.skip decorator","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/54befe6c096275cfe0049a2f1ab1abbee3140c8c"}]}],"author":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-04-13 02:15:09.000000000","tz":-240},"committer":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-04-14 15:59:46.000000000","tz":-240},"subject":"Added in OVN Agent to monitor netlink msg","message":"Added in OVN Agent to monitor netlink msg\n\nAdded a thread to OVN Agent BGP Extension to monitor netlink messages.\nOnly RTNLGRP_LINK messages are currently monitored, specifically for VRF\ncreation and deletion.\n\nAfter the monitoring thread subscribes to RTNLGRP_LINK messages, the\nmonitoring thread receives messages related to a link regardless of the\nlink type (bridge, vrf, etc.) and regardless of the event (creation,\ndeletion, link up, link down).  Thus, it is up to the message handling\ncode to filter for the VRF link type and for the relevant events\n(creation, deletion).\n\nClaude generated the initial code, given the EVPN Type-5 route spec and\nFRR\u0027s code to monitor VRFs.  Claude was also explicitly instructed to\nrefactor the initial code, including moving netlink start-up code\ntogether with thread start-up, splitting the message handling code\nbetween generic loop and specific messages, and refining the vrf\nmessage handling code to not break out of the middle of the function.\n\nRelated-bug: 2144617\nAssisted-by: Claude\nChange-Id: I76a9f0fa27043d374be01b034af0ed1009c23ef8\nSigned-off-by: Helen Chen \u003cichen@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/df21c80008b059c97e59f6806006ba9b2a4484d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/df21c80008b059c97e59f6806006ba9b2a4484d8"}]},"branch":"refs/heads/master"},"944aa7f457f8382f5b991c282e9fb24146ae6a4a":{"kind":"REWORK","_number":3,"created":"2026-04-29 14:44:54.000000000","uploader":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"ref":"refs/changes/09/984409/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/984409/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/984409/3"}}},"commit":{"parents":[{"commit":"3e9e57acc4161a9684d7af4fc75853f35dedb742","subject":"Merge \"Use HA_Chassis_Group in the OVN L3 scheduler\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3e9e57acc4161a9684d7af4fc75853f35dedb742"}]}],"author":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-04-13 02:15:09.000000000","tz":-240},"committer":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-04-29 14:44:20.000000000","tz":-240},"subject":"Created OVN Agent EVPN extension \u0026 netlink monitor","message":"Created OVN Agent EVPN extension \u0026 netlink monitor\n\nCreated a new OVN Agent EVPN Extension.  This extension has a thread to\nmonitor netlink messages.  Only RTMGRP_LINK messages are currently\nmonitored, specifically for VRF creation and deletion.\n\nAfter the monitoring thread subscribes to RTMGRP_LINK messages, the\nmonitoring thread receives messages related to a link regardless of the\nlink type (bridge, vrf, etc.) and regardless of the event (creation,\ndeletion, link up, link down).  The extension defines two message\nhandler functions, one for RTM_NEWLINK and one for RTM_DELLINK.  Each\nhandler function also filters to only process messages related to VRF\nlink type.\n\nClaude generated the initial code, given the EVPN Type-5 route spec and\nFRR\u0027s code to monitor VRFs.  Claude was also explicitly instructed to\nrefactor the initial code, including moving netlink start-up code\ntogether with thread start-up, splitting the message handling code\nbetween generic loop and specific messages, and refining the vrf\nmessage handling code to not break out of the middle of the function.\n\nRelated-bug: 2144617\nAssisted-by: Claude\nChange-Id: I76a9f0fa27043d374be01b034af0ed1009c23ef8\nSigned-off-by: Helen Chen \u003cichen@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/944aa7f457f8382f5b991c282e9fb24146ae6a4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/944aa7f457f8382f5b991c282e9fb24146ae6a4a"}]},"branch":"refs/heads/master"},"1b0ed938c790a5ab4fc358024109f675f79fa43c":{"kind":"REWORK","_number":4,"created":"2026-04-30 16:11:51.000000000","uploader":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"ref":"refs/changes/09/984409/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/984409/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/984409/4"}}},"commit":{"parents":[{"commit":"b85e35d6a6f89d454525962b3023350afa2da3e9","subject":"Merge \"Enable \"filter_validation\" extension by default\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b85e35d6a6f89d454525962b3023350afa2da3e9"}]}],"author":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-04-29 15:15:25.000000000","tz":-240},"committer":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-04-29 15:15:25.000000000","tz":-240},"subject":"Created OVN Agent EVPN extension \u0026 netlink monitor","message":"Created OVN Agent EVPN extension \u0026 netlink monitor\n\nCreated a new OVN Agent EVPN Extension.  This extension has a thread to\nmonitor netlink messages.  Only RTMGRP_LINK messages are currently\nmonitored, specifically for VRF creation and deletion.\n\nAfter the monitoring thread subscribes to RTMGRP_LINK messages, the\nmonitoring thread receives messages related to a link regardless of the\nlink type (bridge, vrf, etc.) and regardless of the event (creation,\ndeletion, link up, link down).  The extension defines two message\nhandler functions, one for RTM_NEWLINK and one for RTM_DELLINK.  Each\nhandler function also filters to only process messages related to VRF\nlink type.\n\nClaude generated the initial code, given the EVPN Type-5 route spec and\nFRR\u0027s code to monitor VRFs.  Claude was also explicitly instructed to\nrefactor the initial code, including moving netlink start-up code\ntogether with thread start-up, splitting the message handling code\nbetween generic loop and specific messages, and refining the vrf\nmessage handling code to not break out of the middle of the function.\n\nRelated-bug: 2144617\nAssisted-by: Claude\nChange-Id: I76a9f0fa27043d374be01b034af0ed1009c23ef8\nSigned-off-by: Helen Chen \u003cichen@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1b0ed938c790a5ab4fc358024109f675f79fa43c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1b0ed938c790a5ab4fc358024109f675f79fa43c"}]},"branch":"refs/heads/master"},"b085787fff6df794b69b4c138355c6b503b8b00e":{"kind":"REWORK","_number":5,"created":"2026-05-01 20:54:38.000000000","uploader":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"ref":"refs/changes/09/984409/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/984409/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/984409/5"}}},"commit":{"parents":[{"commit":"b85e35d6a6f89d454525962b3023350afa2da3e9","subject":"Merge \"Enable \"filter_validation\" extension by default\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b85e35d6a6f89d454525962b3023350afa2da3e9"}]}],"author":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-04-13 02:15:09.000000000","tz":-240},"committer":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-05-01 20:53:44.000000000","tz":-240},"subject":"Created OVN Agent EVPN extension \u0026 netlink monitor","message":"Created OVN Agent EVPN extension \u0026 netlink monitor\n\nCreated a new OVN Agent EVPN Extension.  This extension has a thread to\nmonitor netlink messages.  Only RTMGRP_LINK messages are currently\nmonitored, specifically for VRF creation and deletion.\n\nAfter the monitoring thread subscribes to RTMGRP_LINK messages, the\nmonitoring thread receives messages related to a link regardless of the\nlink type (bridge, vrf, etc.) and regardless of the event (creation,\ndeletion, link up, link down).  The extension defines two message\nhandler functions, one for RTM_NEWLINK and one for RTM_DELLINK.  Each\nhandler function also filters to only process messages related to VRF\nlink type.\n\nClaude generated the initial code, given the EVPN Type-5 route spec and\nFRR\u0027s code to monitor VRFs.  Claude was also explicitly instructed to\nrefactor the initial code, including moving netlink start-up code\ntogether with thread start-up, splitting the message handling code\nbetween generic loop and specific messages, and refining the vrf\nmessage handling code to not break out of the middle of the function.\nClaude generated most of the unit test code.\n\nRelated-bug: 2144617\nAssisted-by: Claude\nChange-Id: I76a9f0fa27043d374be01b034af0ed1009c23ef8\nSigned-off-by: Helen Chen \u003cichen@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b085787fff6df794b69b4c138355c6b503b8b00e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b085787fff6df794b69b4c138355c6b503b8b00e"}]},"branch":"refs/heads/master"},"dc86f6e374d947769a8c0713f6c77536e1a13ac9":{"kind":"REWORK","_number":6,"created":"2026-05-04 14:06:10.000000000","uploader":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"ref":"refs/changes/09/984409/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/984409/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/984409/6"}}},"commit":{"parents":[{"commit":"5461533bbeeeb31c8192660f2db37692f65042ca","subject":"Merge \"Remove fip64 extension support from floating IP pools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5461533bbeeeb31c8192660f2db37692f65042ca"}]}],"author":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-04-13 02:15:09.000000000","tz":-240},"committer":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-05-04 14:05:57.000000000","tz":-240},"subject":"Created OVN Agent EVPN extension \u0026 netlink monitor","message":"Created OVN Agent EVPN extension \u0026 netlink monitor\n\nCreated a new OVN Agent EVPN Extension.  This extension has a thread to\nmonitor netlink messages.  Only RTMGRP_LINK messages are currently\nmonitored, specifically for VRF creation and deletion.\n\nAfter the monitoring thread subscribes to RTMGRP_LINK messages, the\nmonitoring thread receives messages related to a link regardless of the\nlink type (bridge, vrf, etc.) and regardless of the event (creation,\ndeletion, link up, link down).  The extension defines two message\nhandler functions, one for RTM_NEWLINK and one for RTM_DELLINK.  Each\nhandler function also filters to only process messages related to VRF\nlink type.\n\nClaude generated the initial code, given the EVPN Type-5 route spec and\nFRR\u0027s code to monitor VRFs.  Claude was also explicitly instructed to\nrefactor the initial code, including moving netlink start-up code\ntogether with thread start-up, splitting the message handling code\nbetween generic loop and specific messages, and refining the vrf\nmessage handling code to not break out of the middle of the function.\nClaude generated most of the unit test code.\n\nRelated-bug: 2144617\nAssisted-by: Claude\nChange-Id: I76a9f0fa27043d374be01b034af0ed1009c23ef8\nSigned-off-by: Helen Chen \u003cichen@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dc86f6e374d947769a8c0713f6c77536e1a13ac9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dc86f6e374d947769a8c0713f6c77536e1a13ac9"}]},"branch":"refs/heads/master"},"2af3d5e6961e12c75a5a9b78ea9efd61f3fb5ce1":{"kind":"REWORK","_number":7,"created":"2026-05-06 19:44:19.000000000","uploader":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"ref":"refs/changes/09/984409/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/984409/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/984409/7"}}},"commit":{"parents":[{"commit":"37c326f0b677917f6d946faacddc04d54a7d143e","subject":"Merge \"Replace OVN Metadata agent with OVN agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/37c326f0b677917f6d946faacddc04d54a7d143e"}]}],"author":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-04-13 02:15:09.000000000","tz":-240},"committer":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-05-06 19:43:40.000000000","tz":-240},"subject":"Created OVN Agent EVPN extension \u0026 netlink monitor","message":"Created OVN Agent EVPN extension \u0026 netlink monitor\n\nCreated a new OVN Agent EVPN Extension.  This extension has a thread to\nmonitor netlink messages.  Only RTMGRP_LINK messages are currently\nmonitored, specifically for VRF creation and deletion.\n\nAfter the monitoring thread subscribes to RTMGRP_LINK messages, the\nmonitoring thread receives messages related to a link regardless of the\nlink type (bridge, vrf, etc.) and regardless of the event (creation,\ndeletion, link up, link down).  The extension defines two message\nhandler functions, one for RTM_NEWLINK and one for RTM_DELLINK.  Each\nhandler function also filters to only process messages related to VRF\nlink type.\n\nClaude generated the initial code, given the EVPN Type-5 route spec and\nFRR\u0027s code to monitor VRFs.  Claude was also explicitly instructed to\nrefactor the initial code, including moving netlink start-up code\ntogether with thread start-up, splitting the message handling code\nbetween generic loop and specific messages, and refining the vrf\nmessage handling code to not break out of the middle of the function.\nClaude generated most of the unit test code.\n\nRelated-bug: 2144617\nAssisted-by: Claude\nChange-Id: I76a9f0fa27043d374be01b034af0ed1009c23ef8\nSigned-off-by: Helen Chen \u003cichen@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2af3d5e6961e12c75a5a9b78ea9efd61f3fb5ce1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2af3d5e6961e12c75a5a9b78ea9efd61f3fb5ce1"}]},"branch":"refs/heads/master"},"71b7bc080ab70c7144ac8b0f0dcdd6f01bf5d80e":{"kind":"REWORK","_number":8,"created":"2026-05-07 02:23:42.000000000","uploader":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"ref":"refs/changes/09/984409/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/984409/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/984409/8"}}},"commit":{"parents":[{"commit":"37c326f0b677917f6d946faacddc04d54a7d143e","subject":"Merge \"Replace OVN Metadata agent with OVN agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/37c326f0b677917f6d946faacddc04d54a7d143e"}]}],"author":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-04-13 02:15:09.000000000","tz":-240},"committer":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-05-07 01:40:32.000000000","tz":-240},"subject":"Created OVN Agent EVPN extension \u0026 netlink monitor","message":"Created OVN Agent EVPN extension \u0026 netlink monitor\n\nCreated a new OVN Agent EVPN Extension.  This extension has a thread to\nmonitor netlink messages.  Only RTMGRP_LINK messages are currently\nmonitored, specifically for VRF creation and deletion.\n\nAfter the monitoring thread subscribes to RTMGRP_LINK messages, the\nmonitoring thread receives messages related to a link regardless of the\nlink type (bridge, vrf, etc.) and regardless of the event (creation,\ndeletion, link up, link down).  The extension defines two message\nhandler functions, one for RTM_NEWLINK and one for RTM_DELLINK.  Each\nhandler function also filters to only process messages related to VRF\nlink type.\n\nClaude generated the initial code, given the EVPN Type-5 route spec and\nFRR\u0027s code to monitor VRFs.  Claude was also explicitly instructed to\nrefactor the initial code, including moving netlink start-up code\ntogether with thread start-up, splitting the message handling code\nbetween generic loop and specific messages, and refining the vrf\nmessage handling code to not break out of the middle of the function.\nClaude generated most of the unit test code.\n\nRelated-bug: 2144617\nAssisted-by: Claude\nChange-Id: I76a9f0fa27043d374be01b034af0ed1009c23ef8\nSigned-off-by: Helen Chen \u003cichen@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/71b7bc080ab70c7144ac8b0f0dcdd6f01bf5d80e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/71b7bc080ab70c7144ac8b0f0dcdd6f01bf5d80e"}]},"branch":"refs/heads/master"},"816e6c2ee2fcfe18ddde6e28940c84e2c173622b":{"kind":"REWORK","_number":9,"created":"2026-05-07 16:51:17.000000000","uploader":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"ref":"refs/changes/09/984409/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/984409/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/984409/9"}}},"commit":{"parents":[{"commit":"37c326f0b677917f6d946faacddc04d54a7d143e","subject":"Merge \"Replace OVN Metadata agent with OVN agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/37c326f0b677917f6d946faacddc04d54a7d143e"}]}],"author":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-04-13 02:15:09.000000000","tz":-240},"committer":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-05-07 16:51:08.000000000","tz":-240},"subject":"Created OVN Agent EVPN extension \u0026 netlink monitor","message":"Created OVN Agent EVPN extension \u0026 netlink monitor\n\nCreated a new OVN Agent EVPN Extension.  This extension has a thread to\nmonitor netlink messages.  Only RTMGRP_LINK messages are currently\nmonitored, specifically for VRF creation and deletion.\n\nAfter the monitoring thread subscribes to RTMGRP_LINK messages, the\nmonitoring thread receives messages related to a link regardless of the\nlink type (bridge, vrf, etc.) and regardless of the event (creation,\ndeletion, link up, link down).  The extension defines two message\nhandler functions, one for RTM_NEWLINK and one for RTM_DELLINK.  Each\nhandler function also filters to only process messages related to VRF\nlink type.\n\nClaude generated the initial code, given the EVPN Type-5 route spec and\nFRR\u0027s code to monitor VRFs.  Claude was also explicitly instructed to\nrefactor the initial code, including moving netlink start-up code\ntogether with thread start-up, splitting the message handling code\nbetween generic loop and specific messages, and refining the vrf\nmessage handling code to not break out of the middle of the function.\nClaude generated most of the unit test code.\n\nRelated-bug: 2144617\nAssisted-by: Claude\nChange-Id: I76a9f0fa27043d374be01b034af0ed1009c23ef8\nSigned-off-by: Helen Chen \u003cichen@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/816e6c2ee2fcfe18ddde6e28940c84e2c173622b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/816e6c2ee2fcfe18ddde6e28940c84e2c173622b"}]},"branch":"refs/heads/master"},"5879bc1155b49ce0d593b336004c9a70036468e9":{"kind":"REWORK","_number":10,"created":"2026-05-07 19:58:09.000000000","uploader":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"ref":"refs/changes/09/984409/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/984409/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/984409/10"}}},"commit":{"parents":[{"commit":"be6db3977def39fad33a1d3da824f9d027fd5ea9","subject":"Merge \"[OVN] Only set NAT gateway_port when distributed FIP is enabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/be6db3977def39fad33a1d3da824f9d027fd5ea9"}]}],"author":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-04-13 02:15:09.000000000","tz":-240},"committer":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-05-07 19:57:39.000000000","tz":-240},"subject":"Created OVN Agent EVPN extension \u0026 netlink monitor","message":"Created OVN Agent EVPN extension \u0026 netlink monitor\n\nCreated a new OVN Agent EVPN Extension.  This extension has a thread to\nmonitor netlink messages.  Only RTMGRP_LINK messages are currently\nmonitored, specifically for VRF creation and deletion.\n\nAfter the monitoring thread subscribes to RTMGRP_LINK messages, the\nmonitoring thread receives messages related to a link regardless of the\nlink type (bridge, vrf, etc.) and regardless of the event (creation,\ndeletion, link up, link down).  The extension defines two message\nhandler functions, one for RTM_NEWLINK and one for RTM_DELLINK.  Each\nhandler function also filters to only process messages related to VRF\nlink type.\n\nClaude generated the initial code, given the EVPN Type-5 route spec and\nFRR\u0027s code to monitor VRFs.  Claude was also explicitly instructed to\nrefactor the initial code, including moving netlink start-up code\ntogether with thread start-up, splitting the message handling code\nbetween generic loop and specific messages, and refining the vrf\nmessage handling code to not break out of the middle of the function.\nClaude generated most of the unit test code.\n\nRelated-bug: 2144617\nAssisted-by: Claude\nChange-Id: I76a9f0fa27043d374be01b034af0ed1009c23ef8\nSigned-off-by: Helen Chen \u003cichen@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5879bc1155b49ce0d593b336004c9a70036468e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5879bc1155b49ce0d593b336004c9a70036468e9"}]},"branch":"refs/heads/master"},"78f5e39812e2596121cea62cf5672cf4f33664ab":{"kind":"REWORK","_number":11,"created":"2026-05-08 02:09:19.000000000","uploader":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"ref":"refs/changes/09/984409/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/984409/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/984409/11"}}},"commit":{"parents":[{"commit":"be6db3977def39fad33a1d3da824f9d027fd5ea9","subject":"Merge \"[OVN] Only set NAT gateway_port when distributed FIP is enabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/be6db3977def39fad33a1d3da824f9d027fd5ea9"}]}],"author":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-04-13 02:15:09.000000000","tz":-240},"committer":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-05-07 16:51:59.000000000","tz":-240},"subject":"Created OVN Agent EVPN extension \u0026 netlink monitor","message":"Created OVN Agent EVPN extension \u0026 netlink monitor\n\nCreated a new OVN Agent EVPN Extension.  This extension has a thread to\nmonitor netlink messages.  Only RTMGRP_LINK messages are currently\nmonitored, specifically for VRF creation and deletion.\n\nAfter the monitoring thread subscribes to RTMGRP_LINK messages, the\nmonitoring thread receives messages related to a link regardless of the\nlink type (bridge, vrf, etc.) and regardless of the event (creation,\ndeletion, link up, link down).  The extension defines two message\nhandler functions, one for RTM_NEWLINK and one for RTM_DELLINK.  Each\nhandler function also filters to only process messages related to VRF\nlink type.\n\nClaude generated the initial code, given the EVPN Type-5 route spec and\nFRR\u0027s code to monitor VRFs.  Claude was also explicitly instructed to\nrefactor the initial code, including moving netlink start-up code\ntogether with thread start-up, splitting the message handling code\nbetween generic loop and specific messages, and refining the vrf\nmessage handling code to not break out of the middle of the function.\nClaude generated most of the unit test code.\n\nRelated-bug: 2144617\nAssisted-by: Claude\nChange-Id: I76a9f0fa27043d374be01b034af0ed1009c23ef8\nSigned-off-by: Helen Chen \u003cichen@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/78f5e39812e2596121cea62cf5672cf4f33664ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/78f5e39812e2596121cea62cf5672cf4f33664ab"}]},"branch":"refs/heads/master"},"8a6b36ec501e60506379c11c594290e9548516b6":{"kind":"REWORK","_number":12,"created":"2026-05-14 02:32:27.000000000","uploader":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"ref":"refs/changes/09/984409/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/984409/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/984409/12"}}},"commit":{"parents":[{"commit":"4f13e89b5c5ef848e0fb8ec5519fa23bb49d9abf","subject":"Merge \"Add missing address_group CRUD and action policy rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4f13e89b5c5ef848e0fb8ec5519fa23bb49d9abf"}]}],"author":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-04-13 02:15:09.000000000","tz":-240},"committer":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-05-14 02:31:40.000000000","tz":-240},"subject":"Created OVN Agent EVPN extension \u0026 netlink monitor","message":"Created OVN Agent EVPN extension \u0026 netlink monitor\n\nCreated a new OVN Agent EVPN Extension.  This extension has a thread to\nmonitor netlink messages.  Only RTMGRP_LINK messages are currently\nmonitored, specifically for VRF creation and deletion.\n\nAfter the monitoring thread subscribes to RTMGRP_LINK messages, the\nmonitoring thread receives messages related to a link regardless of the\nlink type (bridge, vrf, etc.) and regardless of the event (creation,\ndeletion, link up, link down).  The extension defines two message\nhandler functions, one for RTM_NEWLINK and one for RTM_DELLINK.  Each\nhandler function also filters to only process messages related to VRF\nlink type.\n\nClaude generated the initial code, given the EVPN Type-5 route spec and\nFRR\u0027s code to monitor VRFs.  Claude was also explicitly instructed to\nrefactor the initial code, including moving netlink start-up code\ntogether with thread start-up, splitting the message handling code\nbetween generic loop and specific messages, and refining the vrf\nmessage handling code to not break out of the middle of the function.\nClaude generated most of the unit test and functional test code.\n\nRelated-bug: 2144617\nAssisted-by: Claude\nChange-Id: I76a9f0fa27043d374be01b034af0ed1009c23ef8\nSigned-off-by: Helen Chen \u003cichen@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8a6b36ec501e60506379c11c594290e9548516b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8a6b36ec501e60506379c11c594290e9548516b6"}]},"branch":"refs/heads/master"},"a18cf48d3348aa6aa2140215c42db32aec63003c":{"kind":"REWORK","_number":13,"created":"2026-05-14 14:15:35.000000000","uploader":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"ref":"refs/changes/09/984409/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/984409/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/984409/13"}}},"commit":{"parents":[{"commit":"4f13e89b5c5ef848e0fb8ec5519fa23bb49d9abf","subject":"Merge \"Add missing address_group CRUD and action policy rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4f13e89b5c5ef848e0fb8ec5519fa23bb49d9abf"}]}],"author":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-04-13 02:15:09.000000000","tz":-240},"committer":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-05-14 14:15:22.000000000","tz":-240},"subject":"Created OVN Agent EVPN extension \u0026 netlink monitor","message":"Created OVN Agent EVPN extension \u0026 netlink monitor\n\nCreated a new OVN Agent EVPN Extension.  This extension has a thread to\nmonitor netlink messages.  Only RTMGRP_LINK messages are currently\nmonitored, specifically for VRF creation and deletion.\n\nAfter the monitoring thread subscribes to RTMGRP_LINK messages, the\nmonitoring thread receives messages related to a link regardless of the\nlink type (bridge, vrf, etc.) and regardless of the event (creation,\ndeletion, link up, link down).  The extension defines two message\nhandler functions, one for RTM_NEWLINK and one for RTM_DELLINK.  Each\nhandler function also filters to only process messages related to VRF\nlink type.\n\nClaude generated the initial code, given the EVPN Type-5 route spec and\nFRR\u0027s code to monitor VRFs.  Claude was also explicitly instructed to\nrefactor the initial code, including moving netlink start-up code\ntogether with thread start-up, splitting the message handling code\nbetween generic loop and specific messages, and refining the vrf\nmessage handling code to not break out of the middle of the function.\nClaude generated most of the unit test and functional test code.\n\nRelated-bug: 2144617\nAssisted-by: Claude\nChange-Id: I76a9f0fa27043d374be01b034af0ed1009c23ef8\nSigned-off-by: Helen Chen \u003cichen@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a18cf48d3348aa6aa2140215c42db32aec63003c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a18cf48d3348aa6aa2140215c42db32aec63003c"}]},"branch":"refs/heads/master"},"cc6a462f82224216e3674529479edacec490d09a":{"kind":"REWORK","_number":14,"created":"2026-05-15 03:46:08.000000000","uploader":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"ref":"refs/changes/09/984409/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/984409/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/984409/14"}}},"commit":{"parents":[{"commit":"f0641d7faadca4a4ca4b9de4970a159a0395ce9a","subject":"Merge \"pyproject: Update packaging configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f0641d7faadca4a4ca4b9de4970a159a0395ce9a"}]}],"author":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-04-13 02:15:09.000000000","tz":-240},"committer":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-05-14 14:29:12.000000000","tz":-240},"subject":"Created OVN Agent EVPN extension \u0026 netlink monitor","message":"Created OVN Agent EVPN extension \u0026 netlink monitor\n\nCreated a new OVN Agent EVPN Extension.  This extension has a thread to\nmonitor netlink messages.  Only RTMGRP_LINK messages are currently\nmonitored, specifically for VRF creation and deletion.\n\nAfter the monitoring thread subscribes to RTMGRP_LINK messages, the\nmonitoring thread receives messages related to a link regardless of the\nlink type (bridge, vrf, etc.) and regardless of the event (creation,\ndeletion, link up, link down).  The extension defines two message\nhandler functions, one for RTM_NEWLINK and one for RTM_DELLINK.  Each\nhandler function also filters to only process messages related to VRF\nlink type.\n\nClaude generated the initial code, given the EVPN Type-5 route spec and\nFRR\u0027s code to monitor VRFs.  Claude was also explicitly instructed to\nrefactor the initial code, including moving netlink start-up code\ntogether with thread start-up, splitting the message handling code\nbetween generic loop and specific messages, and refining the vrf\nmessage handling code to not break out of the middle of the function.\nClaude generated most of the unit test and functional test code.\n\nRelated-bug: 2144617\nAssisted-by: Claude\nChange-Id: I76a9f0fa27043d374be01b034af0ed1009c23ef8\nSigned-off-by: Helen Chen \u003cichen@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cc6a462f82224216e3674529479edacec490d09a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cc6a462f82224216e3674529479edacec490d09a"}]},"branch":"refs/heads/master"},"85a26af9c97051f5775dce881351dca44a22b8ad":{"kind":"REWORK","_number":15,"created":"2026-05-15 16:54:27.000000000","uploader":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"ref":"refs/changes/09/984409/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/984409/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/984409/15"}}},"commit":{"parents":[{"commit":"17ae01276c23a3bced83fc95ceebae418d2bc47e","subject":"Merge \"remove pbr as runtime dependency\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/17ae01276c23a3bced83fc95ceebae418d2bc47e"}]}],"author":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-04-13 02:15:09.000000000","tz":-240},"committer":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-05-15 16:53:27.000000000","tz":-240},"subject":"Created OVN Agent EVPN extension \u0026 netlink monitor","message":"Created OVN Agent EVPN extension \u0026 netlink monitor\n\nCreated a new OVN Agent EVPN Extension.  This extension has a thread to\nmonitor netlink messages.  Only RTMGRP_LINK messages are currently\nmonitored, specifically for VRF creation and deletion.\n\nAfter the monitoring thread subscribes to RTMGRP_LINK messages, the\nmonitoring thread receives messages related to a link regardless of the\nlink type (bridge, vrf, etc.) and regardless of the event (creation,\ndeletion, link up, link down).  The extension defines two message\nhandler functions, one for RTM_NEWLINK and one for RTM_DELLINK.  Each\nhandler function also filters to only process messages related to VRF\nlink type.\n\nClaude generated the initial code, given the EVPN Type-5 route spec and\nFRR\u0027s code to monitor VRFs.  Claude was also explicitly instructed to\nrefactor the initial code, including moving netlink start-up code\ntogether with thread start-up, splitting the message handling code\nbetween generic loop and specific messages, and refining the vrf\nmessage handling code to not break out of the middle of the function.\nClaude generated most of the unit test and functional test code.\n\nRelated-bug: 2144617\nAssisted-by: Claude\nChange-Id: I76a9f0fa27043d374be01b034af0ed1009c23ef8\nSigned-off-by: Helen Chen \u003cichen@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/85a26af9c97051f5775dce881351dca44a22b8ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/85a26af9c97051f5775dce881351dca44a22b8ad"}]},"branch":"refs/heads/master"},"3f03cfd012d2361eefce430f9ed43a048204aee3":{"kind":"REWORK","_number":16,"created":"2026-05-18 02:10:07.000000000","uploader":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"ref":"refs/changes/09/984409/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/984409/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/984409/16"}}},"commit":{"parents":[{"commit":"17ae01276c23a3bced83fc95ceebae418d2bc47e","subject":"Merge \"remove pbr as runtime dependency\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/17ae01276c23a3bced83fc95ceebae418d2bc47e"}]}],"author":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-04-13 02:15:09.000000000","tz":-240},"committer":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-05-17 03:08:07.000000000","tz":-240},"subject":"Created OVN Agent EVPN extension \u0026 netlink monitor","message":"Created OVN Agent EVPN extension \u0026 netlink monitor\n\nCreated a new OVN Agent EVPN Extension.  This extension has a thread to\nmonitor netlink messages.  Only RTMGRP_LINK messages are currently\nmonitored, specifically for VRF creation and deletion.\n\nAfter the monitoring thread subscribes to RTMGRP_LINK messages, the\nmonitoring thread receives messages related to a link regardless of the\nlink type (bridge, vrf, etc.) and regardless of the event (creation,\ndeletion, link up, link down).  The extension defines two message\nhandler functions, one for RTM_NEWLINK and one for RTM_DELLINK.  Each\nhandler function also filters to only process messages related to VRF\nlink type.\n\nClaude generated the initial code, given the EVPN Type-5 route spec and\nFRR\u0027s code to monitor VRFs.  Claude was also explicitly instructed to\nrefactor the initial code, including moving netlink start-up code\ntogether with thread start-up, splitting the message handling code\nbetween generic loop and specific messages, and refining the vrf\nmessage handling code to not break out of the middle of the function.\nClaude generated most of the unit test and functional test code.\n\nRelated-bug: 2144617\nAssisted-by: Claude\nChange-Id: I76a9f0fa27043d374be01b034af0ed1009c23ef8\nSigned-off-by: Helen Chen \u003cichen@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3f03cfd012d2361eefce430f9ed43a048204aee3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3f03cfd012d2361eefce430f9ed43a048204aee3"}]},"branch":"refs/heads/master"},"498a2f215cff521ad5154de3d0291f373e7fd6de":{"kind":"REWORK","_number":17,"created":"2026-05-18 13:01:30.000000000","uploader":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"ref":"refs/changes/09/984409/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/984409/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/984409/17"}}},"commit":{"parents":[{"commit":"17ae01276c23a3bced83fc95ceebae418d2bc47e","subject":"Merge \"remove pbr as runtime dependency\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/17ae01276c23a3bced83fc95ceebae418d2bc47e"}]}],"author":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-04-13 02:15:09.000000000","tz":-240},"committer":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-05-18 13:01:21.000000000","tz":-240},"subject":"Created OVN Agent EVPN extension \u0026 netlink monitor","message":"Created OVN Agent EVPN extension \u0026 netlink monitor\n\nCreated a new OVN Agent EVPN Extension.  This extension has a thread to\nmonitor netlink messages.  Only RTMGRP_LINK messages are currently\nmonitored, specifically for VRF creation and deletion.\n\nAfter the monitoring thread subscribes to RTMGRP_LINK messages, the\nmonitoring thread receives messages related to a link regardless of the\nlink type (bridge, vrf, etc.) and regardless of the event (creation,\ndeletion, link up, link down).  The extension defines two message\nhandler functions, one for RTM_NEWLINK and one for RTM_DELLINK.  Each\nhandler function also filters to only process messages related to VRF\nlink type.\n\nClaude generated the initial code, given the EVPN Type-5 route spec and\nFRR\u0027s code to monitor VRFs.  Claude was also explicitly instructed to\nrefactor the initial code, including moving netlink start-up code\ntogether with thread start-up, splitting the message handling code\nbetween generic loop and specific messages, and refining the vrf\nmessage handling code to not break out of the middle of the function.\nClaude generated most of the unit test and functional test code.\n\nRelated-bug: 2144617\nAssisted-by: Claude\nChange-Id: I76a9f0fa27043d374be01b034af0ed1009c23ef8\nSigned-off-by: Helen Chen \u003cichen@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/498a2f215cff521ad5154de3d0291f373e7fd6de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/498a2f215cff521ad5154de3d0291f373e7fd6de"}]},"branch":"refs/heads/master"},"3d6fda56d365ae1382cfe4a8074c6cab90866276":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2026-05-18 14:32:34.000000000","uploader":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"ref":"refs/changes/09/984409/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/984409/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/984409/18"}}},"commit":{"parents":[{"commit":"9c365e62d1be31f1cf37a023316ac379b97d2407","subject":"Merge \"Guard BGPWorker against missing _reconciler\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9c365e62d1be31f1cf37a023316ac379b97d2407"}]}],"author":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-04-13 02:15:09.000000000","tz":-240},"committer":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-05-18 13:02:51.000000000","tz":-240},"subject":"Created OVN Agent EVPN extension \u0026 netlink monitor","message":"Created OVN Agent EVPN extension \u0026 netlink monitor\n\nCreated a new OVN Agent EVPN Extension.  This extension has a thread to\nmonitor netlink messages.  Only RTMGRP_LINK messages are currently\nmonitored, specifically for VRF creation and deletion.\n\nAfter the monitoring thread subscribes to RTMGRP_LINK messages, the\nmonitoring thread receives messages related to a link regardless of the\nlink type (bridge, vrf, etc.) and regardless of the event (creation,\ndeletion, link up, link down).  The extension defines two message\nhandler functions, one for RTM_NEWLINK and one for RTM_DELLINK.  Each\nhandler function also filters to only process messages related to VRF\nlink type.\n\nClaude generated the initial code, given the EVPN Type-5 route spec and\nFRR\u0027s code to monitor VRFs.  Claude was also explicitly instructed to\nrefactor the initial code, including moving netlink start-up code\ntogether with thread start-up, splitting the message handling code\nbetween generic loop and specific messages, and refining the vrf\nmessage handling code to not break out of the middle of the function.\nClaude generated most of the unit test and functional test code.\n\nRelated-bug: 2144617\nAssisted-by: Claude\nChange-Id: I76a9f0fa27043d374be01b034af0ed1009c23ef8\nSigned-off-by: Helen Chen \u003cichen@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3d6fda56d365ae1382cfe4a8074c6cab90866276"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3d6fda56d365ae1382cfe4a8074c6cab90866276"}]},"branch":"refs/heads/master"},"4c8b081a1ebc50e132f7c6d47a83855eb6fd8ea1":{"kind":"REWORK","_number":19,"created":"2026-05-19 15:53:09.000000000","uploader":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"ref":"refs/changes/09/984409/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/984409/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/984409/19"}}},"commit":{"parents":[{"commit":"30374afe1ee08f3f6d3d7507310a6ffd3551e215","subject":"Merge \"OVNL3RouterPlugin register events in Neutron API workers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/30374afe1ee08f3f6d3d7507310a6ffd3551e215"}]}],"author":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-04-13 02:15:09.000000000","tz":-240},"committer":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-05-19 15:48:53.000000000","tz":-240},"subject":"Created OVN Agent EVPN extension \u0026 netlink monitor","message":"Created OVN Agent EVPN extension \u0026 netlink monitor\n\nCreated a new OVN Agent EVPN Extension.  This extension has a thread to\nmonitor netlink messages.  Only RTMGRP_LINK messages are currently\nmonitored, specifically for VRF creation and deletion.\n\nAfter the monitoring thread subscribes to RTMGRP_LINK messages, the\nmonitoring thread receives messages related to a link regardless of the\nlink type (bridge, vrf, etc.) and regardless of the event (creation,\ndeletion, link up, link down).  The extension defines two message\nhandler functions, one for RTM_NEWLINK and one for RTM_DELLINK.  Each\nhandler function also filters to only process messages related to VRF\nlink type.\n\nClaude generated the initial code, given the EVPN Type-5 route spec and\nFRR\u0027s code to monitor VRFs.  Claude was also explicitly instructed to\nrefactor the initial code, including moving netlink start-up code\ntogether with thread start-up, splitting the message handling code\nbetween generic loop and specific messages, and refining the vrf\nmessage handling code to not break out of the middle of the function.\nClaude generated most of the unit test and functional test code.\n\nRelated-bug: 2144617\nAssisted-by: Claude\nChange-Id: I76a9f0fa27043d374be01b034af0ed1009c23ef8\nSigned-off-by: Helen Chen \u003cichen@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4c8b081a1ebc50e132f7c6d47a83855eb6fd8ea1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4c8b081a1ebc50e132f7c6d47a83855eb6fd8ea1"}]},"branch":"refs/heads/master"},"807d0f4985c8d732c446ab65f748f4949583d5c3":{"kind":"REWORK","_number":20,"created":"2026-05-19 16:27:38.000000000","uploader":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"ref":"refs/changes/09/984409/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/984409/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/984409/20"}}},"commit":{"parents":[{"commit":"30374afe1ee08f3f6d3d7507310a6ffd3551e215","subject":"Merge \"OVNL3RouterPlugin register events in Neutron API workers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/30374afe1ee08f3f6d3d7507310a6ffd3551e215"}]}],"author":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-04-13 02:15:09.000000000","tz":-240},"committer":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-05-19 16:08:36.000000000","tz":-240},"subject":"Created OVN Agent EVPN extension \u0026 netlink monitor","message":"Created OVN Agent EVPN extension \u0026 netlink monitor\n\nCreated a new OVN Agent EVPN Extension.  This extension has a thread to\nmonitor netlink messages.  Only RTMGRP_LINK messages are currently\nmonitored, specifically for VRF creation and deletion.\n\nAfter the monitoring thread subscribes to RTMGRP_LINK messages, the\nmonitoring thread receives messages related to a link regardless of the\nlink type (bridge, vrf, etc.) and regardless of the event (creation,\ndeletion, link up, link down).  The extension defines two message\nhandler functions, one for RTM_NEWLINK and one for RTM_DELLINK.  Each\nhandler function also filters to only process messages related to VRF\nlink type.\n\nClaude generated the initial code, given the EVPN Type-5 route spec and\nFRR\u0027s code to monitor VRFs.  Claude was also explicitly instructed to\nrefactor the initial code, including moving netlink start-up code\ntogether with thread start-up, splitting the message handling code\nbetween generic loop and specific messages, and refining the vrf\nmessage handling code to not break out of the middle of the function.\nClaude generated most of the unit test and functional test code.\n\nRelated-bug: 2144617\nAssisted-by: Claude\nChange-Id: I76a9f0fa27043d374be01b034af0ed1009c23ef8\nSigned-off-by: Helen Chen \u003cichen@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/807d0f4985c8d732c446ab65f748f4949583d5c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/807d0f4985c8d732c446ab65f748f4949583d5c3"}]},"branch":"refs/heads/master"},"0b12b3a845287fa052e4b63823e98aa442e40547":{"kind":"REWORK","_number":21,"created":"2026-05-19 16:52:15.000000000","uploader":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"ref":"refs/changes/09/984409/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/984409/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/984409/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/984409/21"}}},"commit":{"parents":[{"commit":"30374afe1ee08f3f6d3d7507310a6ffd3551e215","subject":"Merge \"OVNL3RouterPlugin register events in Neutron API workers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/30374afe1ee08f3f6d3d7507310a6ffd3551e215"}]}],"author":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-04-13 02:15:09.000000000","tz":-240},"committer":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-05-19 16:52:05.000000000","tz":-240},"subject":"Created OVN Agent EVPN extension \u0026 netlink monitor","message":"Created OVN Agent EVPN extension \u0026 netlink monitor\n\nCreated a new OVN Agent EVPN Extension.  This extension has a thread to\nmonitor netlink messages.  Only RTMGRP_LINK messages are currently\nmonitored, specifically for VRF creation and deletion.\n\nAfter the monitoring thread subscribes to RTMGRP_LINK messages, the\nmonitoring thread receives messages related to a link regardless of the\nlink type (bridge, vrf, etc.) and regardless of the event (creation,\ndeletion, link up, link down).  The extension defines two message\nhandler functions, one for RTM_NEWLINK and one for RTM_DELLINK.  Each\nhandler function also filters to only process messages related to VRF\nlink type.\n\nClaude generated the initial code, given the EVPN Type-5 route spec and\nFRR\u0027s code to monitor VRFs.  Claude was also explicitly instructed to\nrefactor the initial code, including moving netlink start-up code\ntogether with thread start-up, splitting the message handling code\nbetween generic loop and specific messages, and refining the vrf\nmessage handling code to not break out of the middle of the function.\nClaude generated most of the unit test and functional test code.\n\nRelated-bug: 2144617\nAssisted-by: Claude\nChange-Id: I76a9f0fa27043d374be01b034af0ed1009c23ef8\nSigned-off-by: Helen Chen \u003cichen@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0b12b3a845287fa052e4b63823e98aa442e40547"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0b12b3a845287fa052e4b63823e98aa442e40547"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
