)]}'
{"id":"openstack%2Fneutron~984482","triplet_id":"openstack%2Fneutron~master~I43c0aefa6447872c84b054cfe2e0ab9eb9addcf5","project":"openstack/neutron","branch":"master","topic":"bug/2144979","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-04-14 06:44:26.000000000","reason":"Change was abandoned"},"36760":{"account":{"_account_id":36760,"name":"Jimin Shin","display_name":"Jimin Shin","email":"jimin3.shin@samsung.com","username":"Jimin"},"last_update":"2026-04-14 06:44:16.000000000","reason":"\u003cGERRIT_ACCOUNT_36760\u003e replied on the change","reason_account":{"_account_id":36760,"name":"Jimin Shin","display_name":"Jimin Shin","email":"jimin3.shin@samsung.com","username":"Jimin"}}},"hashtags":[],"change_id":"I43c0aefa6447872c84b054cfe2e0ab9eb9addcf5","subject":"Sanitize MAC address on port create/update","status":"ABANDONED","created":"2026-04-14 01:49:30.000000000","updated":"2026-04-14 06:44:26.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"09760bf794f033e98c2137a9beb674c1495b89f3","_number":984482,"virtual_id_number":984482,"owner":{"_account_id":36760,"name":"Jimin Shin","display_name":"Jimin Shin","email":"jimin3.shin@samsung.com","username":"Jimin"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-04-14 06:28:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":-1,"date":"2026-04-14 06:17:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-14 03:49:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-14 06:17:32.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"7d7870dac078fba63862fb708d3c33f4441f1695","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36760,"name":"Jimin Shin","display_name":"Jimin Shin","email":"jimin3.shin@samsung.com","username":"Jimin"},"date":"2026-04-14 01:49:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c14ce22d6cd24513f92fe2bc0ff9ab15fbee9a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 03:49:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/85555cb902f445e796512eeb1d79e53a\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e177b8608fcc449d859b83040e0dead4 : SUCCESS in 1h 39m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/04d36c48156f47c58c0043db76f48f13 : SUCCESS in 54m 11s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f45c689fd10c4c658be36bb0a80a6c8e : SUCCESS in 1h 27m 14s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f57e63030794428c984b22ff46caca31 : SUCCESS in 1h 55m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e610b191305e4ff4b6ca3f82509e93f5 : SUCCESS in 33m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6bc64def062f49599df42a1b435de941 : TIMED_OUT in 1h 21m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dadf3a75791d4c95af948d318f2c36dd : SUCCESS in 12m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e65754ed4f7145be9adb22ca247efee3 : SUCCESS in 38m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3d695d5003f243c6a018e8e0ba896052 : SUCCESS in 21m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6f16e5806dce465a953d5634de409f26 : SUCCESS in 21m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0292c45b2de740e1b454ead4c74bf9af : SUCCESS in 6m 46s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/78cef80c69804ed680cb84847dc0c05b : SUCCESS in 1h 23m 12s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1d445ca820a441b5bddbdfec4d56e713 : SUCCESS in 1h 09m 11s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/2d029b590b7a4207848e0e43d30812da : SUCCESS in 45m 12s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a1ab6380776646c9afce7b6b0bf0f169 : SUCCESS in 1h 56m 37s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0324613935014416836c5b706736969f : SUCCESS in 1h 37m 00s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d09c49c87412482d9f0b444a21973825 : SUCCESS in 48m 15s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ae0d150e6cc84d8294ad36352d667c25 : SUCCESS in 1h 03m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b2ab4949adaa4e0ab1ec3c742224b618 : SUCCESS in 1h 08m 43s","accounts_in_message":[],"_revision_number":1},{"id":"33132161fb73bac70e7a6a59c4f395386c546c70","author":{"_account_id":36760,"name":"Jimin Shin","display_name":"Jimin Shin","email":"jimin3.shin@samsung.com","username":"Jimin"},"date":"2026-04-14 04:10:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a562771d20dbdf7f22fee43e4f3aef4f785f1906","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-14 06:17:32.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"10d56a4fafe6af682e74017bde9f1deff86d7749","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 06:28:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7bbc8c62670c40cea501e287bc3f8d3a\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/69180edab9f94955a458e5db28aab3ca : SUCCESS in 1h 07m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bc630516971e48998f5dfac2647303c9 : SUCCESS in 1h 08m 09s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c6913877aa3c423d8389a127d995adf3 : SUCCESS in 1h 19m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/652095aa6e2a46b29aa3132a72b59de1 : SUCCESS in 2h 15m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2d1d1a6aa2f9409396c4cb2c7ffc9fed : SUCCESS in 31m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/905af8267bbd41b2aec5e361641edf0b : SUCCESS in 44m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f46cbd8616449faab8794868ada018e : SUCCESS in 28m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fd13c00070a54c0e9817afcbef2089cf : SUCCESS in 20m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/81f7c8d438fa4d2cb03b54efa798f050 : SUCCESS in 17m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ca23cf8c36d3409a81622aca82b733cf : SUCCESS in 20m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15090300440346ad87edca49eb0a83d5 : SUCCESS in 11m 46s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1098b0493bcd4eca98162b33953eeb4e : SUCCESS in 1h 11m 37s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c9881424c08f49e8a5f3f55d6d4f8ef9 : SUCCESS in 1h 18m 27s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/418b804d698a40b7b137152a26a161cf : SUCCESS in 43m 03s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ae2a45efd5044a1cba9926a1e3bd725c : SUCCESS in 1h 58m 02s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1aca3025e9484e1199e021b4c7882fcb : SUCCESS in 1h 32m 29s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a24c19a5c5394fd0b147ab9726b6e9e4 : SUCCESS in 1h 04m 03s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c4e3448d742f404d82fbe33ffde4dd12 : SUCCESS in 31m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fda725c0550b465b954453132ee841b0 : SUCCESS in 36m 51s","accounts_in_message":[],"_revision_number":1},{"id":"57d514ec1b60282f9b1dbb57ae6bf2141af228bc","author":{"_account_id":36760,"name":"Jimin Shin","display_name":"Jimin Shin","email":"jimin3.shin@samsung.com","username":"Jimin"},"date":"2026-04-14 06:44:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"09760bf794f033e98c2137a9beb674c1495b89f3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":36760,"name":"Jimin Shin","display_name":"Jimin Shin","email":"jimin3.shin@samsung.com","username":"Jimin"},"date":"2026-04-14 06:44:26.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7faac0ab9f0fbd5f67f3efca66163374f497123c","revisions":{"7faac0ab9f0fbd5f67f3efca66163374f497123c":{"kind":"REWORK","_number":1,"created":"2026-04-14 01:49:30.000000000","uploader":{"_account_id":36760,"name":"Jimin Shin","display_name":"Jimin Shin","email":"jimin3.shin@samsung.com","username":"Jimin"},"ref":"refs/changes/82/984482/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/984482/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/984482/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/984482/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/984482/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/984482/1"}}},"commit":{"parents":[{"commit":"720802c8038d4ba0fe85824f437ed630eb184404","subject":"Merge \"Use project_id in policy and miscellaneous plugin code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/720802c8038d4ba0fe85824f437ed630eb184404"}]}],"author":{"name":"Jimin Shin","email":"jimin3.shin@samsung.com","date":"2026-04-09 12:45:39.000000000","tz":540},"committer":{"name":"Jimin Shin","email":"jimin3.shin@samsung.com","date":"2026-04-14 00:28:37.000000000","tz":540},"subject":"Sanitize MAC address on port create/update","message":"Sanitize MAC address on port create/update\n\nCurrently, MAC‑address sanitizing is done only for bulk‑port creation.\nThis change adds logic to sanitize the provided MAC address during both\nindividual port creation and port update.\n\nCloses-Bug: #2144979\nChange-Id: I43c0aefa6447872c84b054cfe2e0ab9eb9addcf5\nAssisted-By: glm-4.7\nSigned-off-by: Jimin Shin \u003cjimin3.shin@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7faac0ab9f0fbd5f67f3efca66163374f497123c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7faac0ab9f0fbd5f67f3efca66163374f497123c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
