)]}'
{"id":"openstack%2Fneutron~985363","triplet_id":"openstack%2Fneutron~master~I40cf808374c4e29609d6dbf480ff2a557cf88343","project":"openstack/neutron","branch":"master","topic":"bug/2148804","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-04-29 12:14:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I40cf808374c4e29609d6dbf480ff2a557cf88343","subject":"Add missing Router ID and AZ hits in the HCG.external_ids","status":"MERGED","created":"2026-04-20 11:34:16.000000000","updated":"2026-05-04 15:10:40.000000000","submitted":"2026-04-29 12:14:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"985363-bug/2148804","meta_rev_id":"a5f9383235f0e552a2be387499a5f561315a43ef","_number":985363,"virtual_id_number":985363,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-29 12:14:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-04-29 09:52:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2026-04-22 04:07:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-04-29 09:52:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-04-28 09:47:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-20 14:34:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-04-20 19:34:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-21 00:33:55.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-04-21 06:07:46.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"CC"},{"updated":"2026-04-22 04:07:00.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-04-29 06:57:45.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2026-04-29 06:57:45.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2026-04-29 09:52:50.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"bbf600db3304451a09a579c396f18437e75588cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-20 11:34:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f5aa675b22643ee6475ca1fb2a07e9628af933d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 14:34:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15b8ec8bc31f4a89ab3d30df9857de44\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fab1dbaf3eff4a5690b6800f355d0028 : SUCCESS in 28m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/998d888f3f5c4ca8aae4bb782809b5c8 : SUCCESS in 18m 29s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/bf8b3ab1ead9469e84aca15b3109bc1d : SUCCESS in 27m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5b76a7538fad179ef634fc6f203148f32f6080f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-20 14:48:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0112b5c65d2d7576be3cc4c478cb7027cb38d76a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 17:46:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4bced3373d294fbfa6971e1d5260058a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/12bb8e38fcf346fb8fba106c830b8c2a : SUCCESS in 24m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3ba6bf83e48649b7811de5993f122ac1 : SUCCESS in 21m 58s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/01c81558e7824eebba3e04c3f9c2fc13 : SUCCESS in 25m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bb83526ae71620d777e621d3dd07759ed49473e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 19:34:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4180a6f3b4614a109a468031cca9a4da\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/03402e3a345a42d0b265bebba29ceff9 : SUCCESS in 1h 14m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1728e51b08934805a2585a3f1ee58b54 : SUCCESS in 1h 06m 48s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9f67368832474d1ea4b854ec0570a203 : SUCCESS in 2h 18m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/53f8d1086be540189f4f48a3b6ff36df : SUCCESS in 47m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9dbc47e171d9438294f8b8cd96a30b0f : SUCCESS in 11m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8bfdcddc28b14c619cc6d792d8c3f673 : SUCCESS in 22m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/204529c686c243d39e3ac38071c5ac43 : SUCCESS in 23m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/40c8f9338b104a5cb3adbf5b20d37691 : SUCCESS in 25m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9428d7b1bd8c4c07b0ea9764fdffeb0f : SUCCESS in 37m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d0de7c3fa2a40299d1cc83118da4444 : SUCCESS in 15m 19s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e6858de8e04f428ea16e00dd6c6ac670 : FAILURE in 43m 52s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/496e3c535d48458d90621f67aa616f37 : SUCCESS in 1h 09m 07s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7f3fd8fff5fb47958b3b0fb919377939 : SUCCESS in 1h 10m 27s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/20851a33bc4d4971ab7315a40ed283c0 : SUCCESS in 1h 34m 49s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/f02a9947ef454dfba497d57ba8bf6712 : SUCCESS in 2h 13m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7fe0ff7f40624dccda1b2700ef59de318da77e10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-21 06:05:55.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"12e36c32758305d84ba009ca192327038beff17e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-21 06:07:29.000000000","message":"Patch Set 3: Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"de4058a4886048e83f0f0d2441f5c1fffca8f138","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 08:09:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c1a1c7372564716ae9fd4ae3c4f20c6\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4ef1f106c4a248e38666ed154245dbc3 : SUCCESS in 1h 10m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8ccf848b647641e68c5002aa66f42936 : SUCCESS in 36m 55s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/00c21d548fa245919cd44e1382313e8a : SUCCESS in 2h 00m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5310640aa324d418ac6d1a6962eba80 : SUCCESS in 37m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ed5e78b806748c19b06797b2fdd0d2a : SUCCESS in 20m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6529b284b00349b698655f7674aadf2e : SUCCESS in 22m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e4b5fec4c8c44bdc884be9b444554a74 : SUCCESS in 24m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d3691f0ac80042d3a67f86deb4944c81 : SUCCESS in 18m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/aeaf8fb477f645ae83c4919cf99ffb30 : SUCCESS in 20m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae65201f7e1d4836aec058827c0b855f : SUCCESS in 11m 50s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1998fdc83d6c472981ee3592668b3853 : SUCCESS in 1h 22m 31s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ca29c81e75ec40a4b380617c71ab5762 : SUCCESS in 1h 00m 41s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/53b6cde8033c481d9597c23aa56f9ad0 : SUCCESS in 1h 07m 07s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/080dca747c9b454da283b3cc566dcd31 : SUCCESS in 1h 19m 45s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/64bc2107a5574a5797f79d89d1392ca7 : SUCCESS in 1h 41m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"54e6a39247baa397fefe975e1433a16205eb1032","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 08:13:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c143393220fb40b8808fad66e3f228b9\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3d8ff4bed25d4925b3193df7738a0d1f : SUCCESS in 36m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b69afa48c133446ba0123f870f59286b : SUCCESS in 27m 26s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5762cae7bbdc421b90b7b0e3b530a405 : SUCCESS in 25m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"427ed59172f66167c248c0c94402889f1dd3444d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-22 04:07:00.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"0d5a3ee04acdf900874c3e4ceee57c484fa29d80","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-28 09:47:19.000000000","message":"Patch Set 3: Review-Priority+2","accounts_in_message":[],"_revision_number":3},{"id":"c189064504f6d58391a310195a66a4badb52926d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-04-29 09:52:50.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"3b332145dbcf71884e14c93784c608aafc9ec725","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 09:55:20.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c07782f3ef7a7590bec4a2802ec3c344ed13d211","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 12:14:58.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4bd1ef6764464e6aa023a4bf348664a5\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/33f294fcecfe4f8d814c2bb41880ddd9 : SUCCESS in 1h 36m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8cd14bbe063a46859ee9ad78a943bf7d : SUCCESS in 1h 00m 49s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/114d572625284e2ca76d585db0b5d088 : SUCCESS in 2h 18m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7ce6928510f42ebabe4dd209ded1fed : SUCCESS in 12m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9f49c104698d4da6aa4ccbb27473cd0d : SUCCESS in 38m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/01a62650a86e4b338670a837cbcf4309 : SUCCESS in 22m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a83fa5f919ec4aeabb18a80806b21bfd : SUCCESS in 23m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5f7773620d14fc5971ea6eb458e7bb4 : SUCCESS in 11m 13s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/de5e845c59ba496096bf069b23b4e914 : SUCCESS in 1h 19m 50s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/eebb9264746d47b48b290167e0661488 : SUCCESS in 1h 00m 03s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/90177f05445b4ed8a85d4d764f276d0e : SUCCESS in 49m 28s","accounts_in_message":[],"_revision_number":3},{"id":"bacad91e8d0dcbe2e965e9a7ac5935205bb60988","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 12:14:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"84220c6605f6cb30bb80eb455c7d9ab2fe08e6c2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 12:16:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/340fbbe4b8164311ab74ad31073a93da\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bccbca8ef48e423b8860ee5e667bdea9 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":3},{"id":"a5f9383235f0e552a2be387499a5f561315a43ef","tag":"autogenerated:gerrit:revert","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-05-04 15:10:40.000000000","message":"Created a revert of this change as I33bf4657498759565b819d10a98cdf8788710b56","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9cc303d11a2053e5436282a36925e0fca7118ba5","revisions":{"3dad01b06e8b7e69e32f7a574b44d4b2da43fd83":{"kind":"REWORK","_number":1,"created":"2026-04-20 11:34:16.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/63/985363/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/63/985363/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/985363/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/985363/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/985363/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/63/985363/1"}}},"commit":{"parents":[{"commit":"7e423793733cea025e662fbad4b6369431bf4419","subject":"Merge \"bgp: Redistribute local only addresses\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e423793733cea025e662fbad4b6369431bf4419"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-04-20 10:43:36.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-04-20 10:43:36.000000000","tz":120},"subject":"WIP \u003d\u003d Add missing Router ID and AZ hits in the HCG.external_ids","message":"WIP \u003d\u003d Add missing Router ID and AZ hits in the HCG.external_ids\n\nCloses-Bug: #2148804\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I40cf808374c4e29609d6dbf480ff2a557cf88343\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3dad01b06e8b7e69e32f7a574b44d4b2da43fd83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3dad01b06e8b7e69e32f7a574b44d4b2da43fd83"}]},"branch":"refs/heads/master"},"9ba9ca94d3faecdcfcb2bd2b9bfd4b3f84746812":{"kind":"REWORK","_number":2,"created":"2026-04-20 14:48:14.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/63/985363/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/63/985363/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/985363/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/985363/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/985363/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/63/985363/2"}}},"commit":{"parents":[{"commit":"7e423793733cea025e662fbad4b6369431bf4419","subject":"Merge \"bgp: Redistribute local only addresses\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e423793733cea025e662fbad4b6369431bf4419"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-04-20 10:43:36.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-04-20 14:43:49.000000000","tz":120},"subject":"Add missing Router ID and AZ hits in the HCG.external_ids","message":"Add missing Router ID and AZ hits in the HCG.external_ids\n\nSince [1], present in master (2026.2 only), the ``Logical_Router``\nuse ``HA_Chassis_Group`` to provide HA. However, the implementation\nof the underlying method were not adding the \u0027neutron:router_id\u0027\nand \u0027neutron:availability_zone_hints\u0027 to the ``external_ids`` column.\n\nIn particular the \u0027neutron:router_id\u0027 key is needed to identify that\nthis ``HA_Chassis_Group`` belongs to a router.\n\n[1]https://review.opendev.org/c/openstack/neutron/+/947317\n\nCloses-Bug: #2148804\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I40cf808374c4e29609d6dbf480ff2a557cf88343\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9ba9ca94d3faecdcfcb2bd2b9bfd4b3f84746812"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9ba9ca94d3faecdcfcb2bd2b9bfd4b3f84746812"}]},"branch":"refs/heads/master"},"9cc303d11a2053e5436282a36925e0fca7118ba5":{"kind":"REWORK","_number":3,"created":"2026-04-21 06:05:55.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/63/985363/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/63/985363/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/985363/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/985363/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/985363/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/63/985363/3"}}},"commit":{"parents":[{"commit":"7e423793733cea025e662fbad4b6369431bf4419","subject":"Merge \"bgp: Redistribute local only addresses\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e423793733cea025e662fbad4b6369431bf4419"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-04-20 10:43:36.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-04-21 06:05:42.000000000","tz":120},"subject":"Add missing Router ID and AZ hits in the HCG.external_ids","message":"Add missing Router ID and AZ hits in the HCG.external_ids\n\nSince [1], present in master (2026.2 only), the ``Logical_Router``\nuse ``HA_Chassis_Group`` to provide HA. However, the implementation\nof the underlying method were not adding the \u0027neutron:router_id\u0027\nand \u0027neutron:availability_zone_hints\u0027 to the ``external_ids`` column.\n\nIn particular the \u0027neutron:router_id\u0027 key is needed to identify that\nthis ``HA_Chassis_Group`` belongs to a router.\n\n[1]https://review.opendev.org/c/openstack/neutron/+/947317\n\nCloses-Bug: #2148804\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I40cf808374c4e29609d6dbf480ff2a557cf88343\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9cc303d11a2053e5436282a36925e0fca7118ba5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9cc303d11a2053e5436282a36925e0fca7118ba5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
