)]}'
{"id":"openstack%2Fneutron~985783","triplet_id":"openstack%2Fneutron~stable%2F2026.1~I990caa72db794c6812e3eb979155a67108385379","project":"openstack/neutron","branch":"stable/2026.1","attention_set":{},"removed_from_attention_set":{"37089":{"account":{"_account_id":37089,"name":"Chris Buggy","display_name":"cbuggy","email":"cbuggy@redhat.com","username":"cbuggy"},"last_update":"2026-04-23 09:05:02.000000000","reason":"Change was submitted"},"38061":{"account":{"_account_id":38061,"name":"Martin Morgenstern","email":"martin.morgenstern@cloudandheat.com","username":"martinmo"},"last_update":"2026-04-23 09:05:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I990caa72db794c6812e3eb979155a67108385379","subject":"Add logging decorator for OVN maintenance methods","status":"MERGED","created":"2026-04-22 09:22:10.000000000","updated":"2026-04-23 09:06:03.000000000","submitted":"2026-04-23 09:05:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"985783","meta_rev_id":"4f990f9794a7f15411fc7c9cea1e9f9bf0490e70","_number":985783,"virtual_id_number":985783,"owner":{"_account_id":38061,"name":"Martin Morgenstern","email":"martin.morgenstern@cloudandheat.com","username":"martinmo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-23 09:05:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-04-23 07:07:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-04-22 18:50:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-04-23 07:07:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":37089,"name":"Chris Buggy","display_name":"cbuggy","email":"cbuggy@redhat.com","username":"cbuggy"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37089,"name":"Chris Buggy","display_name":"cbuggy","email":"cbuggy@redhat.com","username":"cbuggy"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-22 09:22:10.000000000","updated_by":{"_account_id":38061,"name":"Martin Morgenstern","email":"martin.morgenstern@cloudandheat.com","username":"martinmo"},"reviewer":{"_account_id":37089,"name":"Chris Buggy","display_name":"cbuggy","email":"cbuggy@redhat.com","username":"cbuggy"},"state":"REVIEWER"},{"updated":"2026-04-22 10:00:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-04-22 11:58:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-22 18:50:03.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-04-23 07:07:19.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"87a5200267cb834a0ec45e596f4b3e4ebf909e0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38061,"name":"Martin Morgenstern","email":"martin.morgenstern@cloudandheat.com","username":"martinmo"},"date":"2026-04-22 09:22:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9f591a8938eb3517928b84a83f79c892c78c6205","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 10:00:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8f4f967e7c44fd09fc4b1f1f125fa0f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/69b2d4e77ef14db4aacacca051f797ea : SUCCESS in 26m 07s\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/810ff518fac640dda967413d60def245 : SUCCESS in 21m 56s\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ffb1982ded5f4937bd88fc33c94a335f : SUCCESS in 36m 07s","accounts_in_message":[],"_revision_number":1},{"id":"cefc94eca462369fe7def2e5778aa9eea0925fba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 11:58:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/623841147f804011b88542c9d7a26215\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/55b513cd3c4e49a69256f8989a0f976e : SUCCESS in 1h 23m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/acfadb6e5ef349c18ec050a2226e278b : SUCCESS in 1h 02m 46s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/ca6ccaaee65c4da0a131d56ff2c8a76d : SUCCESS in 2h 35m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2da879c9a76e4f389ed7819c78510fc3 : SUCCESS in 51m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f042f772a586496fbaa104449a9ae62a : SUCCESS in 11m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/052e8a5a83bd408a9cbd95c4bfa3606e : SUCCESS in 35m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f9289861a5a74bd8a3b12062f2fd7eb6 : SUCCESS in 16m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/af118261d3074e5288a500728d0d3714 : SUCCESS in 32m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ac8aeb4f04b4e189468989b9ce443ae : SUCCESS in 11m 53s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/88af3ce9fe154b789cdfae20d9e4c6bc : SUCCESS in 1h 20m 23s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9da90c080cb94dfd8f239ab33908625d : SUCCESS in 1h 19m 21s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/bdbac15422ab423198919010e5da7bf7 : SUCCESS in 1h 20m 11s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/eeeeaedd5c0f49958b4097e3934a6fd3 : SUCCESS in 1h 10m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a703f4e7d18d499c8a062500b7747d89 : SUCCESS in 1h 31m 58s","accounts_in_message":[],"_revision_number":1},{"id":"89855a8af851b3de4052de7b9c42dc72974d218f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-22 18:50:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fa60aa9c90ece7a505cf59eb302412cf58214d6f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-23 07:07:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f2ac22d70273d94a16e25f72d544a3734616f7e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 07:07:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"532342055118ec21349163a0cb5ccaa4ef57fd20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 09:05:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd9b8ec9a17941ab9d82b385d40d1f69\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1720ccce5dcd49daaac24b20d3c1e51e : SUCCESS in 1h 45m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fb7937fbe3d84cbf92641e061da60a95 : SUCCESS in 1h 03m 06s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/1944d053a5744e91b79f028bed31476e : SUCCESS in 1h 56m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4470503b5a847edbf551e0e254c9ca5 : SUCCESS in 20m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d243ac730ff24fb49e09817e6aef1eb0 : SUCCESS in 26m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7a43c49bd47244c5b11fc1631347d853 : SUCCESS in 18m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fc2e27aca8449d6a795ed05b8316c71 : SUCCESS in 11m 41s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5f3dd0e0d41d4d538712eb422dbbe45d : SUCCESS in 1h 23m 40s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/bdf0db9025334d6ca79ca19e36c668c0 : SUCCESS in 1h 15m 13s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/96a95d1ec1e84340840c1c4e1067cc1e : SUCCESS in 1h 02m 59s","accounts_in_message":[],"_revision_number":1},{"id":"da982257ed8412ce30bb386c2af89b4bc20afef6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 09:05:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4f990f9794a7f15411fc7c9cea1e9f9bf0490e70","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 09:06:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34bf48c9162541dc9da3e0693a4b00bf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c106a9c430774c2496c36cb62d72ab9d : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"61f100d4b987a493b9bbf61301caf2ed41e544d2","revisions":{"61f100d4b987a493b9bbf61301caf2ed41e544d2":{"kind":"REWORK","_number":1,"created":"2026-04-22 09:22:10.000000000","uploader":{"_account_id":38061,"name":"Martin Morgenstern","email":"martin.morgenstern@cloudandheat.com","username":"martinmo"},"ref":"refs/changes/83/985783/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/83/985783/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/985783/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/985783/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/985783/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/83/985783/1"}}},"commit":{"parents":[{"commit":"6f49b3bbef1c94f4b6b9d380e9155a7836c4bd14","subject":"Fix AttributeError on DeletedChassis missing other_config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6f49b3bbef1c94f4b6b9d380e9155a7836c4bd14"}]}],"author":{"name":"Chris Buggy","email":"cbuggy@redhat.com","date":"2026-03-20 17:15:37.000000000","tz":0},"committer":{"name":"Martin Morgenstern","email":"martin.morgenstern@cloudandheat.com","date":"2026-04-22 09:22:08.000000000","tz":120},"subject":"Add logging decorator for OVN maintenance methods","message":"Add logging decorator for OVN maintenance methods\n\nThe OVN maintenance methods had an inconsistent log strategy.\nThis patch introduces a log_maintenance_task decorator using\noslo_utils.timeutils to ensure all maintenance methods\nlog their start and end times consistently without code duplication\n\nCloses-Bug: #2142833\nAssisted-By: Claude composer-1.5\nChange-Id: I990caa72db794c6812e3eb979155a67108385379\nSigned-off-by: Chris Buggy \u003ccbuggy@redhat.com\u003e\n(cherry picked from commit e37278b32ae5464d84b1295e7e3c0c9ab8b5053f)\nSigned-off-by: Martin Morgenstern \u003cmartin.morgenstern@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/61f100d4b987a493b9bbf61301caf2ed41e544d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/61f100d4b987a493b9bbf61301caf2ed41e544d2"}]},"branch":"refs/heads/stable/2026.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
