)]}'
{"id":"openstack%2Fneutron~986613","triplet_id":"openstack%2Fneutron~stable%2F2025.1~Idc7d789ffe0289bf28fb289456196c307dd4a706","project":"openstack/neutron","branch":"stable/2025.1","topic":"bug/2150543","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-05-04 14:12:38.000000000","reason":"Change was submitted"},"31291":{"account":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"last_update":"2026-05-04 14:12:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idc7d789ffe0289bf28fb289456196c307dd4a706","subject":"Fix return value of empty-MAC guard in metadata agent","status":"MERGED","created":"2026-04-29 07:05:47.000000000","updated":"2026-05-04 14:14:33.000000000","submitted":"2026-05-04 14:12:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"986613-bug/2150543","meta_rev_id":"e4a5925e92e63dc40f3afd40b2c6566615fdad92","_number":986613,"virtual_id_number":986613,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-04 14:12:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2026-05-04 11:35:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2026-04-29 16:17:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":1,"date":"2026-05-04 11:35:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"}],"reviewers":{"CC":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}],"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-29 07:05:47.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"state":"REVIEWER"},{"updated":"2026-04-29 07:06:28.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"CC"},{"updated":"2026-04-29 07:06:28.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-04-29 09:22:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-29 16:17:56.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-05-04 11:35:53.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"6e07fe28adbf806c2d09cfe795d5cd4ae8fa69ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-29 07:05:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f4ae1bc89977e4279ffc59cd04504abd1768d96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 09:22:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/084ad2ccde7b4134aaa9b0baefa2e6fe\n\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/407eb7f8503f4923a9c16cd982c75d58 : SUCCESS in 2h 15m 56s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/1cffc9dac5bb4586abac087850698cb1 : SUCCESS in 1h 09m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0b948900a104b7aaf6d6711278b6fe7 : SUCCESS in 44m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7609f9611f9d4c8bb72509f36cd34468 : SUCCESS in 11m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0a79821e9a714e5e8bc574a574372002 : SUCCESS in 22m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2a53fc8803804299ab38dbfc4c956598 : SUCCESS in 21m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc6164063e544d13a04fe1ef1cb233d5 : SUCCESS in 6m 16s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/000ac5302d634a69b265bd9b7220484f : SUCCESS in 1h 18m 13s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2b2f1c7fa73a444c8e33ca6f16c44199 : SUCCESS in 1h 03m 00s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/a6c7226878fb49a0867967398c8a21dc : SUCCESS in 1h 11m 17s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/fc8a5cb3e42049739660264f15d5b70e : SUCCESS in 1h 02m 12s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/308a2f8205ab4a29a1fb5f4d44452d98 : SUCCESS in 35m 08s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/cb8550a05fc743ba9c42f0fb63229550 : SUCCESS in 1h 06m 37s","accounts_in_message":[],"_revision_number":1},{"id":"dc5b55e86faf417a10e3bdbfffe3651651102b57","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-29 16:17:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c5c447b161e7acf3b131905bab7cac8504c87ec8","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-05-04 11:35:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c37360a47a2fb6d16f9b6753308c2e3b413c42e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 11:36:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"35775fee195bb911119a689c52f9fdbdf05376e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 14:12:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b25572c1df84d679fd417b887df7fe1\n\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/7cd14b85e94847d087059e27987e7ae4 : SUCCESS in 2h 35m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2a8c92ae40948bfb7e63311cdda9b4f : SUCCESS in 11m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fac3a92a4d4843c9a73cff681ac91abc : SUCCESS in 23m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/77a4697091054594bf950ad39342c082 : SUCCESS in 20m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0734ce503a72456e8f0928eff69d7576 : SUCCESS in 13m 44s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/49085dde7c4c4912a81b7e631e6adb0e : SUCCESS in 1h 18m 14s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/545ce996b7bd45f0bc36ae912f2b88c6 : SUCCESS in 1h 11m 34s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0714fcc07abf41e0b086ce28dbdc2347 : SUCCESS in 51m 27s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e9f7485884a141e7a0adff0e3467d81d : SUCCESS in 1h 29m 21s","accounts_in_message":[],"_revision_number":1},{"id":"9487e55eb69d13ba6c201756b8dcfdaf08d3f997","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 14:12:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e4a5925e92e63dc40f3afd40b2c6566615fdad92","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 14:14:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b19285454b74c7fb21059c7fc033df7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69310c813f9c45a1a551bcc9fe2c1852 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d57e150e6dd928952b721d7b0b62b227d62ad30b","revisions":{"d57e150e6dd928952b721d7b0b62b227d62ad30b":{"kind":"REWORK","_number":1,"created":"2026-04-29 07:05:47.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/986613/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/986613/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/986613/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/986613/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/986613/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/986613/1"}}},"commit":{"parents":[{"commit":"af550868732b1a1807a18e89b0d6808d29677186","subject":"Merge \"Ignore designate NotFound errors during PTR delete\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/af550868732b1a1807a18e89b0d6808d29677186"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-04-28 13:01:44.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-04-29 07:05:40.000000000","tz":120},"subject":"Fix return value of empty-MAC guard in metadata agent","message":"Fix return value of empty-MAC guard in metadata agent\n\nThe empty-MAC guard in _get_port_ip4_ips_and_ip6_flag returns []\ninstead of the expected (list, bool) tuple. This would cause a\ncrash if a port with an empty MAC column were encountered during\ndatapath provisioning. Fix the return value to ([], False).\n\nConflicts:\n  neutron/agent/ovn/metadata/agent.py\n\nCloses-Bug: #2150543\n\nAssisted-By: Claude Opus 4.6\nChange-Id: Idc7d789ffe0289bf28fb289456196c307dd4a706\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n(cherry picked from commit 6f529d902f7deb08187a09ac8a71d2c45d23f92b)\n(cherry picked from commit bd4fa33be86a2e02a58d067c46c5431ca56fcc14)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d57e150e6dd928952b721d7b0b62b227d62ad30b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d57e150e6dd928952b721d7b0b62b227d62ad30b"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
