)]}'
{"id":"openstack%2Fneutron~986962","triplet_id":"openstack%2Fneutron~master~I1a010db313a9796b7fa92faa2075ca752b875883","project":"openstack/neutron","branch":"master","topic":"neutron-project-id-migration","attention_set":{},"removed_from_attention_set":{"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2026-08-10 08:36:13.000000000","reason":"Change was abandoned"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-08-21 19:48:02.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I1a010db313a9796b7fa92faa2075ca752b875883","subject":"WIP: Add project_id attribute for neutron-lib change","status":"ABANDONED","created":"2026-05-01 21:43:54.000000000","updated":"2026-10-08 13:54:40.000000000","total_comment_count":29,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6cb35015051174e1b4e3b691b9509f2d1881c6d2","_number":986962,"virtual_id_number":986962,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2026-06-22 16:37:14.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-06-22 16:58:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-01 22:20:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-01 22:20:28.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-05-07 21:55:58.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-10 13:18:32.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-30 13:37:01.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2026-08-10 08:36:12.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"64b30b09da718be0b71b2f702b0e5e80f8c4e2a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-01 21:43:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ae7e8205379ae37c7546b23d32cce8f0ede99db7","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-01 21:52:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5db010efc8e4085eb557eebc0659e4ec101787e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-01 22:20:12.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"8b61883dca9be053a2f3059621a713d00ac123dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-01 22:20:23.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/neutron-lib/+/986960 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"39c8e870208f436cc68e5f7ef9a1428c2a0a3877","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-01 22:20:25.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/neutron-lib/+/986960 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"6e32847d02ba54b4f6a721d16066707fefe9c409","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-05-01 22:20:28.000000000","message":"Patch Set 2: Code-Review-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/neutron-lib/+/986960 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"be293276e1de1253b2c619ddca4040ed05220f9c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-01 22:21:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5645ead42a288de3ebdd74494c11a832f23f9215","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-01 22:21:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6e22d86cd36f5d1836d7b56ca6e596f58b6a9071","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-01 22:56:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8224241c96af77a6dcff224cb82ddd74c3cb793f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-01 23:24:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac6d46cd005d483a8ab703d90b15b41c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1eda38815e0f4f76b99d0d943d49c2e9 : SUCCESS in 1h 01m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d7c5c73cf41c488ea95718340125f487 : SUCCESS in 49m 06s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d133ea0d535f440081769b58079dea95 : SUCCESS in 1h 00m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"02712be3806e838d2ce364d8e390a7103143a7d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 01:02:29.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d01eb8bfdad841f68f1b8fd98a542aa9\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/11769850a23c424fbc7cb88fd25c1774 : SUCCESS in 1h 26m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/003b30e8a2cc475fb4f2ace6eaf68394 : SUCCESS in 59m 46s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b9f7f6bed82940ddb6832b7f35f27846 : FAILURE in 1h 27m 39s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e4fec376aa7f43f48b7c3eed24922a13 : FAILURE in 2h 38m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/26266205ea9b486aab2e94e6f2e9fb4d : SUCCESS in 34m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81bbe72a7bf744c9b1cfecf1dc6f0639 : SUCCESS in 1h 03m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1b0fc767e28400582dedc6ddbf91b04 : SUCCESS in 11m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01c37708146045bba039304e9d5476c8 : SUCCESS in 20m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/36c068963ee946c0a1261081757084d7 : SUCCESS in 31m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a7f0be484eef4b5880b465616466823f : SUCCESS in 26m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a62d9015677d4290adc90c7a0e72126b : SUCCESS in 40m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8f701d14de84154ae47aa679834a68b : SUCCESS in 12m 35s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2a1307c335c94d2ea55a690d4dcbb806 : SUCCESS in 1h 04m 26s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c69301ac18934c8f95a3004534f4c4a7 : SUCCESS in 1h 13m 04s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/6018c56e62174a12bea430bafc00ceaa : SUCCESS in 1h 12m 24s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1a3b22cb7c1948628ccff02f160c1bcd : SUCCESS in 1h 32m 25s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a19e207634c14f30ab8330b5e1da9b85 : SUCCESS in 1h 41m 13s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5941e33d67e74a50b97e38e97282ba37 : SUCCESS in 1h 03m 12s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b193ace2014849b8940031b159f2275f : SUCCESS in 50m 25s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f9fc12dbba9e49fd808237650b639593 : SUCCESS in 1h 28m 21s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/98c97fe918a0455bb5950e5a938b2384 : SUCCESS in 2h 01m 18s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":2},{"id":"77de0e9ed3c4520e66d19a944cfaaa24edce4ab2","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 01:33:35.000000000","message":"Patch Set 2:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8376308e3e3144aab987bc45f5117779\n\n- openstack-tox-py313-with-oslo-master https://zuul.opendev.org/t/openstack/build/69eb603571e745fb8c474a03b3fe7447 : SUCCESS in 18m 30s\n- openstack-tox-py313-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/6264632807834a3fad67136afe9b00b3 : FAILURE in 24m 29s\n- openstack-tox-py313-with-ovsdbapp-master https://zuul.opendev.org/t/openstack/build/afa319efaa664cc3a5a232ea7278709b : SUCCESS in 24m 37s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/c1bcd6c3482742cf95e2fdd2bbe0f8e3 : TIMED_OUT in 3h 05m 21s (non-voting)\n- neutron-tempest-plugin-openvswitch-distributed-dhcp https://zuul.opendev.org/t/openstack/build/7223eea9ab504a60a7352dfbbd9d3b03 : FAILURE in 1h 20m 38s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-distributed-dhcp https://zuul.opendev.org/t/openstack/build/164591f18e924dcc9adec524a8e11ab9 : FAILURE in 1h 46m 29s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/4a2c0dbdbb784ad5b98e8dbec8affb06 : FAILURE in 1h 45m 23s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/78af37a3132343bbba05d8de41ceaddf : FAILURE in 1h 49m 18s\n- neutron-functional-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/18430ce1b4c4458cb235c8c9ae922d9a : FAILURE in 27m 15s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ad1ba9fbdc8642e980927a21fb4ed52c : SUCCESS in 1h 29m 28s\n- neutron-ovs-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/d5ecfb8971fa4592b365a7260688ad56 : FAILURE in 14m 44s\n- neutron-ovn-tempest-loki https://zuul.opendev.org/t/openstack/build/a2b2a88fb99e4a5c8ebef526028dc161 : SUCCESS in 1h 31m 12s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e963c1c4af014ca79d168492b908a7d5 : SUCCESS in 1h 25m 37s\n- neutron-functional-fips https://zuul.opendev.org/t/openstack/build/0ab219f33189454ab3952b9169255ead : SUCCESS in 1h 09m 57s\n- neutron-functional-with-pyroute2-master https://zuul.opendev.org/t/openstack/build/4b64235250a74b969cc41092f9acad93 : SUCCESS in 1h 24m 07s\n- neutron-functional-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/e97feeaeee4a4543b2685ef2f434813b : FAILURE in 43m 06s\n- neutron-functional-with-ovsdbapp-master https://zuul.opendev.org/t/openstack/build/5c31442a1e9e41b6b2db1d5e2f04632d : SUCCESS in 44m 14s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/331fe9fb668b4dd8b5aaa7332df3dd5c : SUCCESS in 59m 47s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/f8384754af8548c98205a763a3eb8663 : SUCCESS in 1h 59m 04s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/b1ede6c44848487e950da1922757f925 : SUCCESS in 2h 18m 13s\n- neutron-ovs-tempest-with-os-ken-master https://zuul.opendev.org/t/openstack/build/27804c248d674f4bae2cc471dfec6241 : SUCCESS in 1h 00m 08s\n- neutron-ovn-tempest-mariadb-full https://zuul.opendev.org/t/openstack/build/06401f4808904fd9b69689b8606e4517 : SUCCESS in 56m 21s\n- neutron-ovn-tempest-ipv6-only-ovs-master https://zuul.opendev.org/t/openstack/build/d1d88060ad584270a3bbb0fff0d1d89a : SUCCESS in 1h 15m 08s\n- neutron-ovn-tempest-ovs-master-centos-9-stream https://zuul.opendev.org/t/openstack/build/9f5ff3d2fde044b0bb6b64c1852ed457 : SUCCESS in 45m 21s\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/34b854250ce547a7bae6e83a3b091070 : SUCCESS in 1h 36m 07s\n- neutron-ovn-tempest-full-multinode-ovs-lts https://zuul.opendev.org/t/openstack/build/eeaac43f282f4bdb98f873a322fb615d : SUCCESS in 1h 29m 58s\n- neutron-ovn-tempest-ovs-master https://zuul.opendev.org/t/openstack/build/0571e9bfe31c45119a5024d95fc10782 : SUCCESS in 1h 37m 12s\n- neutron-ovn-tempest-ovs-lts https://zuul.opendev.org/t/openstack/build/4ede71e39dba4f5cb2e6cb65fb3b37b5 : SUCCESS in 1h 29m 47s\n- neutron-ovn-tempest-plugin-ovs-master https://zuul.opendev.org/t/openstack/build/0955843edb4846ef9d0f8fa093d485ec : FAILURE in 2h 21m 17s\n- neutron-ovn-tempest-plugin-ovs-lts https://zuul.opendev.org/t/openstack/build/b8c802c3f61846a082a6cdb4e83ed46c : FAILURE in 2h 11m 43s\n- neutron-ovn-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/900a67fcf36a41b48adffd443bd23c1e : FAILURE in 23m 50s\n- neutron-ovn-tempest-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/7d615325e53d4e6c98cf43e6e84c9415 : FAILURE in 26m 34s\n- neutron-ovs-tempest-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/742ae7b377be4ff49f42fc7235019aad : FAILURE in 15m 00s\n- neutron-ovs-tempest-fips https://zuul.opendev.org/t/openstack/build/abac298608964315b9bab9f41b1da8e0 : SUCCESS in 48m 57s\n- neutron-ovn-tempest-ovs-release-fips https://zuul.opendev.org/t/openstack/build/dbe0244ab7c44bee86c226b1bf47bbfe : SUCCESS in 45m 20s\n- devstack-tobiko https://zuul.opendev.org/t/openstack/build/b5d28b1153d64de8a3702691583a3d3b : TIMED_OUT in 3h 03m 43s\n- devstack-tobiko-ovs https://zuul.opendev.org/t/openstack/build/d9aaf67a3ef740dfa560d036c597cf92 : SUCCESS in 1h 02m 37s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/8f55ceb8802141b69f9b6bbcf98bfe71 : SUCCESS in 1h 12m 50s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/34419da751ba4331ad6daaa2999e5a9f : SUCCESS in 26m 59s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/c452751efeac40dab97bee81fa0617b7 : FAILURE in 25m 14s\n- neutron-functional-with-oslo-master https://zuul.opendev.org/t/openstack/build/de34f57f56d241278da73952b5ddfd45 : SUCCESS in 1h 06m 36s\n- neutron-ovs-tempest-with-oslo-master https://zuul.opendev.org/t/openstack/build/37f7c212deea475ab7e5b9d285b56048 : SUCCESS in 1h 23m 11s\n- neutron-ovn-tempest-ovs-release-with-oslo-master https://zuul.opendev.org/t/openstack/build/a2d363ac658b4ae1b23f9a711de34ecc : SUCCESS in 1h 31m 04s\n- neutron-tempest-plugin-ovn-with-ovn-metadata-agent https://zuul.opendev.org/t/openstack/build/0a1c58fd51144dd38f1b8cebafc71809 : FAILURE in 2h 30m 09s\n- neutron-ovn-grenade-multinode-ovn-metadata-agent https://zuul.opendev.org/t/openstack/build/817c71f21b2641f3bab6faad85ece189 : SUCCESS in 1h 19m 40s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/5ba4c7ced86b47b489fea5527f3b7719 : POST_FAILURE in 18m 42s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/6d7e44bfa1b242a4a733e341ea3f4631 : SUCCESS in 1h 16m 59s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/07101742d8604744a089f0a4ab0ff518 : SUCCESS in 1h 15m 14s","accounts_in_message":[],"_revision_number":2},{"id":"1088085dacd8c132a96cbe9da090906c2bcdecfd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-02 01:37:34.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"5483259c2a6115366763ab585c7234c059df7dd1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-02 01:46:35.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6801a368f947ce0f6b79ae18b48418368532526d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 02:13:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/863df00cef4041ba8d807d2aadc47b5c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/26460c3904c94ee4849dad330bca79ef : SUCCESS in 23m 13s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1fbb62eb753c45488949e9aa0188deef : SUCCESS in 16m 54s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ec175877b329441d848c68a7f5828808 : SUCCESS in 33m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bbe63d8918b9bca92aec8165d7155a0eef64ee44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 02:40:56.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/neutron-lib/+/986960 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"b179ca4eff65943a076b85712b2db45d4d0b85da","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 02:41:03.000000000","message":"Patch Set 3:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/neutron-lib/+/986960 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"cab19a54590ea5646e63ebb9412246743cad7f00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-02 02:41:21.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"3c6c1269132350d85ad60ed9304e522bdf754092","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 04:28:52.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1772c13adc38477a9ac5af9478cd3db9\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/dddffb5b358b4716b08199aa6f8f7b28 : FAILURE in 30m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a90c321437fe4063b801fad48157f758 : FAILURE in 1h 24m 03s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5173b82d4cd94258bf471c868b6490af : FAILURE in 27m 43s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c4bf629f222c40f4af08d2d84ebcf073 : FAILURE in 28m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6b23f38e35b948438f140013ebfe1d4c : FAILURE in 30m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b4289a0914c413081dfe2032531b84c : SUCCESS in 1h 12m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f026a4e4a9f4f5aac991672ac863d3a : FAILURE in 5m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e3a8e7b5666f4319a329bc58ff4c6b4d : SUCCESS in 36m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8882f2b4c6a94e44adeee0975acef862 : SUCCESS in 22m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/35eb3015d0a14c25840b342c5333ea1f : SUCCESS in 25m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/00807a7161aa428ea881d42f51d8e86c : SUCCESS in 27m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b960a76d9db346538bd4840fa80ad1c6 : SUCCESS in 12m 38s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6bc8f521bd93476c8754487914481fa8 : SUCCESS in 1h 01m 40s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b14a90f4675a4b95b8d878c0f0023d3e : FAILURE in 48m 46s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/539ef3d8bd09497a9b2c684e5a4c3430 : FAILURE in 36m 58s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/00108d1245ad4891bb59f6c0c6b75afa : FAILURE in 28m 28s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/88efdc813254437ca95236fe9eca71a7 : FAILURE in 29m 20s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5341e62db9c74f35ad8b80819c43a44a : FAILURE in 32m 58s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c7132518a0844902872b6a3a4d85a40c : SUCCESS in 1h 09m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/50de7897d68a43f9af711cc30346d40c : SUCCESS in 1h 24m 22s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/d0323d013cd74ee1a63a1b17d6016ceb : FAILURE in 1h 02m 23s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/dvr_local_router.py","accounts_in_message":[],"_revision_number":4},{"id":"5df4661272963156e428890506323db6969bd8ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-02 04:44:25.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"7dfb9c10dcc9092a16daf30c8efd47c53bce4646","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 04:57:59.000000000","message":"Patch Set 5:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/tempest/+/986967 is needed.","accounts_in_message":[],"_revision_number":5},{"id":"9315f8fc7bb4be6160623c8dceac4b3206925fa0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 04:58:06.000000000","message":"Patch Set 5: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/tempest/+/986967 is needed.","accounts_in_message":[],"_revision_number":5},{"id":"fedd5271b32b41684cf754957ff11bfce054dbe1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-02 04:58:28.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d8692182705266cf1e353604f536d6ea523a16f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 05:34:31.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb3fa0fc603b48ab9c6adffca34f35ae\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/453caf53f1494020b894b0c813519633 : SUCCESS in 31m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/28086e54afe341ada9cbe285a48c19cf : SUCCESS in 17m 46s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3118089518274191a284a83c71f2471b : SUCCESS in 23m 48s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"078b50039117d9957d0acc582e07cfda85360b37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 06:45:17.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/13cfe7665a564f5b85fdd8cb03c08088\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/23c305c1431b4041adb75ad1b5e4ad7a : FAILURE in 57m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0ba15721c054494fbb44cb42079cf4b3 : FAILURE in 1h 00m 28s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/69d720b8abb248128c350ccf4811b1a8 : FAILURE in 1h 02m 02s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6b09ab670fed4e4cba3467764e390671 : FAILURE in 1h 45m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b29483e6f29946b1b295d1dc1e66d56a : SUCCESS in 31m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc53e8409e644e628f1338e363f5e019 : SUCCESS in 1h 07m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0459c80c2d654fe3a36e705d2e78393d : SUCCESS in 7m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7464869eff684ab391ee2d22d0f6c36a : SUCCESS in 41m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f5f6eaa8d17f4510a219f9dce21ae827 : SUCCESS in 27m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c2190e295c3949a1a7cbed4b211b769f : SUCCESS in 23m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/06e226d8a86a4f229ede0d88ffcd5975 : SUCCESS in 32m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c87bd738136847a0bb569c0b43231263 : SUCCESS in 7m 18s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/bed2316da55147be83446bac4f77de6c : SUCCESS in 1h 13m 02s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a0df2ade77824b8fa61c939c2b852217 : FAILURE in 41m 11s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/bf78f902b72647a2ba2873dd28beff7c : FAILURE in 1h 21m 31s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e8bae978b8a5482681507dfdff8d4a3e : FAILURE in 40m 00s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/292677d4ebce4ac2886337dbd9267c72 : FAILURE in 1h 02m 15s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3f892d6317e04fbcb89daba299fa6fbf : POST_FAILURE in 23m 45s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3476f09ca02749f988e780c094c8892d : SUCCESS in 1h 13m 54s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/675caddd1f1645a7b0ae37facef41a1c : SUCCESS in 45m 27s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/a407d9cfb85f4bfba55cbdd06f293899 : FAILURE in 1h 37m 21s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":5},{"id":"3ec0d89afbe078882d09e39db4b3b2d2bb8a7886","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-05 14:54:32.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"78ef881deefe76b69d4c4c271b04c9d966be6fcc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 16:24:19.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/699c54d822a54e4abeb43aa5bf0097e7\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f02d0e9d15d4423080ef2b2e80662e59 : SUCCESS in 25m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9423dd63f2064b61872c2aeaa3616427 : SUCCESS in 27m 04s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/964f7188645c479eab71368d53892279 : SUCCESS in 26m 44s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e3d0482ccf75e251a24c007ce1efe26a900302ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 17:40:53.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/95163c7ce94b4b4da414576d98de79c3\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9102aebd8e52499bb1ca7ba6ae3bc7b0 : FAILURE in 54m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ec3bb6a9f0254eb8a59804a8b4f87f9a : FAILURE in 49m 33s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/14f7ed14244b419986d861249d846b23 : FAILURE in 59m 43s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/45fb49d5cfb948f18ac4ae95d4ccf2b8 : FAILURE in 1h 24m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/aba0a9d37c194e8c829cf78a5c4fb6bb : SUCCESS in 33m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d537f78c718f43cb8aefeece716a3fa9 : SUCCESS in 1h 05m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af1f3a81f9ff463d87f4dcb4348da832 : SUCCESS in 11m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ad4c573a5be241ddb45903f49dc1a14e : SUCCESS in 36m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2d03d14c78d14f37a5fe5fee3293c938 : SUCCESS in 23m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b6978f09f40e4c439f36d047714fce02 : SUCCESS in 27m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8bb7c8ad3e6a44a484a5a441c6e14ee5 : SUCCESS in 23m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cb720a0a71d41f8953bed727055d571 : SUCCESS in 12m 31s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/58714178f2d746e0b6a58524194d178f : SUCCESS in 1h 30m 55s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/cf970d4439c348ddbabd5c6d8b1ba11c : FAILURE in 1h 09m 57s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/cbe54be7312340f88010540a4645c704 : FAILURE in 1h 02m 36s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3b8ec75c8ab84375b9804b283c090918 : FAILURE in 1h 05m 30s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c7b2ac01665f40d88db90b1d3db11c4d : FAILURE in 1h 05m 50s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7e6b5d36321746d6b5b4e99760b6ba79 : FAILURE in 45m 39s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b6e79a29ebe74f64a3cb62103850126f : SUCCESS in 1h 11m 03s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/70fced913b59405d8d1d19cc65a472a9 : SUCCESS in 59m 45s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/ca74a67f18d74a7189690bc2a89df27d : FAILURE in 1h 31m 54s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":5},{"id":"389099c7b43ea6736d6c263ee734986058c9a402","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-05 20:38:26.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c2e9e40697f6f2b233bc4e94584268c9ae8d5c69","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 21:05:29.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa0440a62757459488b912cace0c769c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dec782216ce048e486441b0387c8d246 : SUCCESS in 25m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8726a707a0e14dc7b8b5627bcb4dc166 : SUCCESS in 22m 42s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/91106be4b0d049f19992bbb2acd7a344 : SUCCESS in 25m 00s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4862a16621efe83fc86683083aff42539743e468","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 23:06:54.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb17a14f79354395bb2ae8338764c8e5\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4e6e33686c1f43c69064ae69304b6659 : FAILURE in 43m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3411ab025faf4eefa722ae75c4bfd52f : FAILURE in 51m 41s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/da9deccb0fee47df9f2728274ebc490a : FAILURE in 1h 03m 15s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/15675c12d7e64962b5ae56c80913fb0e : FAILURE in 1h 26m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5fe0abccdf824054b0d5b92470a34429 : SUCCESS in 30m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60c925d5d4cf4662bc6f657a2a6149ef : SUCCESS in 1h 03m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fee9b6268383454f98d848792f1c29bf : SUCCESS in 15m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/91a9e24c711f4c0d8b07b28d96489007 : SUCCESS in 29m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1880abeadfd5425bbe175628eef90c13 : SUCCESS in 19m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0c6e3248084c4e18879888aae033d23d : SUCCESS in 18m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bc138416ef3e4c679fa22cb12bf3b5c6 : SUCCESS in 28m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f88bd80dafbd4263bc9e4df9ff7cdb3a : SUCCESS in 11m 41s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b2899a0703c447d28a347003768be748 : SUCCESS in 1h 06m 54s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4f07c6fbe8674443a0645ef557189667 : FAILURE in 1h 11m 13s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/adbc16b1e59c446688f8cd9deedc1be6 : FAILURE in 1h 17m 30s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d24b301165b149f9bd9d8aa78fda3564 : FAILURE in 1h 02m 30s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ae67449715d2422e9f9d141d3151b406 : FAILURE in 1h 05m 45s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e8c3595c88d641b289c996008e5cc406 : FAILURE in 59m 05s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7abba75a894a430997c932d7807d9e72 : SUCCESS in 38m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3d23a946a021433686e04631d12cb0e0 : SUCCESS in 1h 20m 23s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/74264a2c935044a9adf631d287db5eb8 : FAILURE in 2h 22m 16s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":5},{"id":"68d215cd972fbb9f39d948cae3ff49d1fd6d3bbc","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-06 19:16:45.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3a818830583c22efccd37ff1530440fccaaa5d2f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 20:23:12.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c613512757a4f45a8ff08cc13172d95\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fa2bcd44d72742ed87087b33c38a4089 : SUCCESS in 24m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/652ebeba188440b4b0dad61649e995bc : SUCCESS in 33m 24s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/557be230ea804fdc8ed5fca68f5d67c9 : SUCCESS in 23m 19s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9345095e9e1aad2c016092f5c4b627245087bfa3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 20:56:24.000000000","message":"Patch Set 5:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/neutron-lib/+/986960 is needed.","accounts_in_message":[],"_revision_number":5},{"id":"b1560465de4474b25e7fac598c863357b9bf0314","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-06 21:41:55.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"2b457d14881445620baf01659f97d1e8b1ac4cef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 22:13:57.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5074f12f92748e2a83cc6af59977e6f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5871bd4775684d3b883445a918d5a902 : SUCCESS in 30m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6340ccc2c214494c8d61ab99c0331175 : SUCCESS in 23m 35s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/815b64aee16a4fd39e870870c7b84b66 : SUCCESS in 29m 33s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"bd95d3f858cda691114313b01f6f13873ddf7357","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-06 22:57:36.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"f1c65c30e5380b20bc62dd9f94d6bafebf43e55d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-06 23:11:16.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"022841fa9c6d69a5c7736ab24bcffa4bb183f8e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 23:26:54.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b42a427f2f7e49ae9a1291f9f951d300\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3d15512be71e4e4a8e4107bef1bdb1a3 : SUCCESS in 27m 34s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2eccc1f9723b4dc4bd57fe0049677c5e : SUCCESS in 18m 23s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/da7125b0d7554fd9942169da993e1008 : SUCCESS in 24m 36s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"120bae2f20cd9b721420f417c77e705438ecae73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 00:45:47.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f87f07b6687145e6a2d5a81cadd22702\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/506b25b2464849aeb07c14a74cd1f336 : FAILURE in 53m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7d0f5226d84f4c2c8bc67f2d809ecf47 : FAILURE in 47m 38s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/915ddb5240a4438a98e930b18d412ff7 : FAILURE in 1h 13m 40s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d3468ef2af574cb698d8be6d471bdb8c : FAILURE in 1h 18m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/107d3d73f8a543c3bcd1a0f741ad7adc : SUCCESS in 31m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f385b5bca7249e0bddd7bb11b0301af : SUCCESS in 1h 03m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/819de4b78cf043618d0877d420ac5c3b : SUCCESS in 10m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f296efad6acb4208886d566f4aaf9bee : SUCCESS in 40m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/93fbc1177da547cb802d1c89866a959a : SUCCESS in 31m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0255b12e2c1f45e5ad8bf41378afc735 : SUCCESS in 26m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5447430918fb4345a47c8b3becc33117 : SUCCESS in 36m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f1604475811478daa553868431fb922 : SUCCESS in 7m 59s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2a734483ca944ea68ae85ce7b2fd8407 : SUCCESS in 1h 06m 12s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f1f6c4742d9e44b2830d50349a3c0558 : FAILURE in 49m 42s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/800a88b8b7fd4ef8a75e60dbd17b9d52 : FAILURE in 1h 17m 51s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/cb4a8af6088d450d813877c40d1235c4 : FAILURE in 1h 07m 22s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b8b6d7fe54494ae886ef201d83290cfa : FAILURE in 45m 36s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/55777c179a794b7db9547a73a24f51af : FAILURE in 47m 41s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0d924e96ba2d433ab75c674ab093c94b : SUCCESS in 1h 08m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ae15c6d2defb4bd3bb37cd5a11d96737 : SUCCESS in 1h 23m 30s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/5f5eac0fcf3a4f1da59c53f8138a1e94 : FAILURE in 52m 20s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":7},{"id":"20c0ab3df4d3da440822dd29742e90e594bc1f32","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 02:23:08.000000000","message":"Patch Set 7:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a010fb39f4564b3a8dd8989a709153c2\n\n- openstack-tox-py313-with-oslo-master https://zuul.opendev.org/t/openstack/build/0c85f7b873334b5e8ed4209383ac21cf : SUCCESS in 31m 07s\n- openstack-tox-py313-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/e362f4c89313415497adfbb266e628ee : FAILURE in 21m 30s\n- openstack-tox-py313-with-ovsdbapp-master https://zuul.opendev.org/t/openstack/build/927f3924f29f4553803f0b3e0fbb32f3 : SUCCESS in 30m 29s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/94efa90de08643e882d11e9476586545 : FAILURE in 2h 25m 13s (non-voting)\n- neutron-tempest-plugin-openvswitch-distributed-dhcp https://zuul.opendev.org/t/openstack/build/79ad62621e3f4e2a89a34610be33cdec : FAILURE in 1h 07m 25s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-distributed-dhcp https://zuul.opendev.org/t/openstack/build/4a2d84f625334ae08bc7723434cbeff9 : FAILURE in 1h 17m 12s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/6fc42d24f7094f4f8127cd94688fdaeb : FAILURE in 1h 17m 01s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8f70f06c8b95403ca14f6ce6cf38634d : FAILURE in 1h 05m 55s\n- neutron-functional-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/c7dc4d370a0d4d83aba3158a70bd9743 : FAILURE in 24m 52s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/add7aee7609c4c3e9a7b3b650c9db2d9 : FAILURE in 57m 35s\n- neutron-ovs-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/5bde695ea27a40afa7d43f4f2bb79348 : FAILURE in 20m 08s\n- neutron-ovn-tempest-loki https://zuul.opendev.org/t/openstack/build/e63483a146324599a8f4e68387ff8911 : FAILURE in 1h 00m 04s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ba34696a732e486889c3fbe8467166e4 : SUCCESS in 1h 22m 13s\n- neutron-functional-fips https://zuul.opendev.org/t/openstack/build/950200cbeaa94ae6ba84401c067e25ac : SUCCESS in 1h 00m 12s\n- neutron-functional-with-pyroute2-master https://zuul.opendev.org/t/openstack/build/11b49618c65546cea4029617cd6671c9 : SUCCESS in 1h 20m 45s\n- neutron-functional-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/6e8a05c7a4d44b25a76ab1ca4166cfa0 : FAILURE in 29m 05s\n- neutron-functional-with-ovsdbapp-master https://zuul.opendev.org/t/openstack/build/98c1d5a59c684200ad23d42c9e792b1e : SUCCESS in 1h 20m 27s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2bc178da2b0445bc83f66505ab77d521 : FAILURE in 51m 49s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/01c7e3af7e344544a08e684984dbbe8c : FAILURE in 19m 22s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5886b293880f4c03b489e77badb8295d : FAILURE in 43m 43s\n- neutron-ovs-tempest-with-os-ken-master https://zuul.opendev.org/t/openstack/build/5620e853bbd54ce2a444c7a9556b9d59 : FAILURE in 53m 51s\n- neutron-ovn-tempest-mariadb-full https://zuul.opendev.org/t/openstack/build/be996c7aac5a4cc6b95708090b623789 : FAILURE in 38m 12s\n- neutron-ovn-tempest-ipv6-only-ovs-master https://zuul.opendev.org/t/openstack/build/8aee1d881aa14a01acaf058ceb8dc175 : FAILURE in 40m 33s\n- neutron-ovn-tempest-ovs-master-centos-9-stream https://zuul.opendev.org/t/openstack/build/8435e46c75e540ce960bc97ca92e03e2 : FAILURE in 1h 09m 23s\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/97893ebd8ae147c7a798d0812549360c : FAILURE in 1h 25m 46s\n- neutron-ovn-tempest-full-multinode-ovs-lts https://zuul.opendev.org/t/openstack/build/97787729c4744f418c89d7aafc195fd5 : FAILURE in 43m 16s\n- neutron-ovn-tempest-ovs-master https://zuul.opendev.org/t/openstack/build/3548be7fe66a400d934b683d99c3b458 : FAILURE in 49m 46s\n- neutron-ovn-tempest-ovs-lts https://zuul.opendev.org/t/openstack/build/1e4ca4456d1a474db630bbcfa5005edd : FAILURE in 57m 21s\n- neutron-ovn-tempest-plugin-ovs-master https://zuul.opendev.org/t/openstack/build/1a6cb4976651491187f974d3a54f06d4 : FAILURE in 1h 27m 14s\n- neutron-ovn-tempest-plugin-ovs-lts https://zuul.opendev.org/t/openstack/build/2f22fb4d56f34a26b00d1d41e203d779 : FAILURE in 1h 32m 26s\n- neutron-ovn-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/d7406666a4074e43a18f1259e31b28c7 : FAILURE in 20m 59s\n- neutron-ovn-tempest-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/ae474b3584614c098ec32154758138f0 : FAILURE in 26m 04s\n- neutron-ovs-tempest-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/a3a3945d868a46d597a223058767abdc : FAILURE in 27m 27s\n- neutron-ovs-tempest-fips https://zuul.opendev.org/t/openstack/build/15340f2328d94e1e9c8075fc90788722 : FAILURE in 56m 20s\n- neutron-ovn-tempest-ovs-release-fips https://zuul.opendev.org/t/openstack/build/1b6653ae4afd48b8831243e1b26cfdf7 : FAILURE in 1h 09m 44s\n- devstack-tobiko https://zuul.opendev.org/t/openstack/build/2d079f46fb924b2f88713b45a6b5a59e : TIMED_OUT in 3h 04m 29s\n- devstack-tobiko-ovs https://zuul.opendev.org/t/openstack/build/32d87a39f1354fcc9812af90417a0d97 : SUCCESS in 1h 01m 35s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/437aaf6533934743b9d96f2997c7264c : FAILURE in 1h 10m 26s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/8e28f641304a4b16963e0248ad2bcb89 : SUCCESS in 20m 43s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/e2d5634de8da4facb16aa13dd7086bf5 : FAILURE in 17m 22s\n- neutron-functional-with-oslo-master https://zuul.opendev.org/t/openstack/build/13f76afc1a3444089b3e111e49635c18 : SUCCESS in 1h 21m 34s\n- neutron-ovs-tempest-with-oslo-master https://zuul.opendev.org/t/openstack/build/9f0e607684a34e6298c342dc3a61443f : FAILURE in 47m 25s\n- neutron-ovn-tempest-ovs-release-with-oslo-master https://zuul.opendev.org/t/openstack/build/8a92cc8aae3c4ff5b79b04745a534127 : FAILURE in 55m 21s\n- neutron-tempest-plugin-ovn-with-ovn-metadata-agent https://zuul.opendev.org/t/openstack/build/3dbc9f3ba122498b84385e33914d503e : FAILURE in 1h 34m 40s\n- neutron-ovn-grenade-multinode-ovn-metadata-agent https://zuul.opendev.org/t/openstack/build/3755c16223d74df0879c4c3cd3bfdda0 : FAILURE in 53m 22s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/2b5fd36057dd4cd2876f70331aa74cc1 : SUCCESS in 37m 50s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d63d9c8f8b4f463898a6d48119daa01c : FAILURE in 1h 12m 37s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f75c92a881d04e8cb0b223384ccfb16e : FAILURE in 1h 13m 03s","accounts_in_message":[],"_revision_number":7},{"id":"5a02bcb55c0f1ffd61bfdfa1392bff436e13f387","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-07 13:14:57.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"b825898153f39400e978976c96921ba9234f46b7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 14:01:42.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4c3cb6073594e268a30673f9addbe37\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4756057ffeff468b9dc1039f698432c5 : SUCCESS in 24m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/91c2303782b5459d938814a880fe9e77 : SUCCESS in 24m 32s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/fed249e6876749e096e936863528afae : SUCCESS in 23m 10s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"d9784c083f50285b7f61eed632f010cf5097fd1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 16:02:51.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f03e7e5ddbc14b90bda6a36078874ad4\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7b40fae3495f4235af3a1024a5ff8f38 : SUCCESS in 1h 14m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5526cb8c29b741e6adcf03ae386fd4fb : SUCCESS in 54m 53s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6c2547fed37746da9717e5d3177a3597 : FAILURE in 1h 22m 18s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/51aee624043a478f83aac4f9264b0543 : FAILURE in 1h 40m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b83341f22f6047c081745409c91355ea : SUCCESS in 33m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc813b01a8b8427fa90e0121c41319f4 : SUCCESS in 1h 00m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/805b71bbc877489c9fa7c2ff1cd39e42 : SUCCESS in 11m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7762ce5ccb6642818fb4b671b0371a51 : SUCCESS in 28m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ddeb54871ef540c9b3a6ceee3bb28a55 : SUCCESS in 35m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b15adda5e5194bc88ccd89cf9627494c : SUCCESS in 29m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1d21bc231069487585d52438540470f4 : SUCCESS in 39m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee3b94752c3147ecad4e5b955a90fd79 : SUCCESS in 11m 54s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e059c14fb98346d486edb61fce283cfd : SUCCESS in 42m 29s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7de2be2c606847e6aa6d45c406e1f510 : SUCCESS in 1h 08m 53s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/03e06a0c6d084394bb2a1fd7db385114 : SUCCESS in 55m 34s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3152db66efbf4e5c9e3071c5d34e76ee : SUCCESS in 1h 30m 18s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1ea4aa2da7064c0abb86f512ae9ece86 : SUCCESS in 57m 10s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/fb7c2286da8c4978b7c85819f9871552 : SUCCESS in 1h 03m 55s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/116b885707984b9cbfcb04de1c444642 : SUCCESS in 1h 20m 00s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0893073f529245a196470d2639e93aa5 : SUCCESS in 1h 07m 03s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/37acd8f1388d450aacdb27d6dc126492 : SUCCESS in 2h 36m 53s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":8},{"id":"fe17873cd10dfab18314afd2ac13ce8d6daafb62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-07 17:59:08.000000000","message":"Patch Set 9: Patch Set 8 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"8db70821150f6560c6f6e7f6594ee70a1bf431b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 18:29:26.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa5a23108e094cef8d16fc442f7d9c06\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a051a15234104ec786d71935149b8e0d : SUCCESS in 23m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cb22cab0da8c4da0a304be90eab7c8a8 : SUCCESS in 17m 22s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d578652a48e54fac9fb146c99e8d1c9f : SUCCESS in 28m 13s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"9202cb8ab973bf5d472963845be2ad6e88b558fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-07 19:29:01.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"3be1312dddb88d57f3c5d6cce41d86b0a2a8030f","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-05-07 19:29:07.000000000","message":"Patch Set 10: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/c06a6dad10a04560a8f720873575ca56\n\n\nWarning:\n  Dependency cycle detected and project openstack/neutron-lib doesn\u0027t allow circular dependencies","accounts_in_message":[],"_revision_number":10},{"id":"add2f761cdd63b4237dfa4dd93abb167fbb3aecd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 19:29:14.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a8d35a9664a41428d5746099908f236\n\n\nWarning:\n  Dependency cycle detected and project openstack/neutron-lib doesn\u0027t allow circular dependencies","accounts_in_message":[],"_revision_number":10},{"id":"7ce96f5c31552815ef49e0179c959ecd21a22a60","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 19:29:30.000000000","message":"Patch Set 10:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dce6f947ac434248b57b45444e860c76\n\n\nWarning:\n  Dependency cycle detected and project openstack/neutron-lib doesn\u0027t allow circular dependencies","accounts_in_message":[],"_revision_number":10},{"id":"000f16d44b03828749af2c736a33b538c57ad5e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-07 19:31:38.000000000","message":"Uploaded patch set 11: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"eee9f74ad6ec72454da8f86aa22b4cb61369cfc1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 20:04:05.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f165d2bad504d3b8dcd983e7fece338\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5733442be8be400da6db3ca6d45d26c4 : SUCCESS in 26m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/254d088fb4634fbba9ffc68548aa5628 : SUCCESS in 18m 38s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8be82479c07745e3aa79783944c757a8 : SUCCESS in 30m 35s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"15c0e57913f8e96ae074a2bc98ec07be84b278ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-07 21:40:14.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"b50b541e13997230044fb342af9685dd8f5d7c42","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-07 21:55:58.000000000","message":"Patch Set 12:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/62/986962/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"193f2001585e4aa996f65b6fe780e83d63cd0f51","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 22:13:16.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f606375ea3c04425bd65dff66bb9ad13\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d1e6075503a0435f9a5b585cd9a5c290 : SUCCESS in 23m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0868a705c9c4465885dc500100e411aa : SUCCESS in 29m 18s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/064b787e3b734569a2020547e9476db4 : SUCCESS in 31m 14s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"6c4b8505d294b9649603094c9a44ab9ea64c58d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-07 22:45:34.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"62a8bd5f922650852e721c09f63cdc50e4a3a4f6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-07 22:48:00.000000000","message":"Patch Set 13:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/62/986962/13 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":13},{"id":"6d4568abfba1db811f4bef013e604608070aa7f2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 23:22:49.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5c56541843b49efa69bf8f940e0400f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/311a71c3ca7f421b957f4dc315d2603c : SUCCESS in 26m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ec9ba424d0414175b31289dd77727bf5 : SUCCESS in 18m 49s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/03a5f1c65c084a4480634d421a2866de : SUCCESS in 35m 01s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"4f390cb4673ae3187c73772a21dff476a2c452b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 00:58:08.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b934c9daaf3b4ead86611a04eceaf38c\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/422c16f77688435b94b771a565b8afe6 : SUCCESS in 1h 31m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d34ee6b15dd408981b24cfc2a41afdc : SUCCESS in 1h 03m 04s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9cc7bc7fa6134b7383a162c71ffb15fc : FAILURE in 1h 15m 20s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fe66a5b8b42440a7a9fcc7d6ea957d06 : FAILURE in 2h 11m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fcf7378d87fd4836b55afef1ebcdfbe9 : SUCCESS in 31m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c5e94cbd2f542af99a457815ffa24da : SUCCESS in 46m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7393a48845942f48792eb1c521851f6 : SUCCESS in 6m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/58851e77d4984b66963af20fc04c1641 : SUCCESS in 36m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/26a11c23d00c41b2a5be743cd61ac6c5 : SUCCESS in 24m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d49eb82c476348db92a9bb3249ead48d : SUCCESS in 23m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3bbcc60e1527485ebd9f4e0060309241 : SUCCESS in 21m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13253467bbc94ac1ac3ce1a737fcbd4e : SUCCESS in 12m 39s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/108a777627e94142b1578ddbb3f5c44b : SUCCESS in 1h 27m 02s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6141fb25e5424527aae9844cc3e6eb1f : SUCCESS in 1h 18m 03s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/34bd29d0d8304d778c5a44388457ea9e : SUCCESS in 1h 20m 41s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/58a052ae0e8640bc91794d12a98df0d5 : SUCCESS in 1h 09m 28s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e8c70bf72a2c471e9930a2fd30de82ba : SUCCESS in 1h 03m 09s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e3b888fffcb147b8b6d03b9deb92745c : SUCCESS in 1h 08m 08s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/6a934d74d17046868d202358a2dbd67a : SUCCESS in 1h 13m 29s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bbf72cb861db4d7196aa6eafb9b7700c : SUCCESS in 1h 30m 22s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/7514e3ed61464d27baef35e3d61f0a69 : SUCCESS in 1h 12m 35s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":13},{"id":"6224c557ebdd10fcfc0ea19d89ac496f82cf6972","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-08 02:22:10.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"48b264abd0372322952ee3591400e89d0a19979a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-08 02:24:31.000000000","message":"Patch Set 14:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/62/986962/14 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":14},{"id":"d94bad6449d12c55988c39f69b184250f67ee127","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-08 04:29:10.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"b92a109556217d0d54d5d7dea9bdcc8200114bea","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-08 04:31:52.000000000","message":"Patch Set 15:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/62/986962/15 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":15},{"id":"70faa0e9a70cb67482a1c296569425aee2560dde","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 05:20:22.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0b4f2d6e2b5475f8550af18ed0174bf\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/aa5b2eda25bb4ef086c289f10df90a13 : SUCCESS in 26m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/189261b4faa34d78916be82ed8514712 : SUCCESS in 18m 13s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f4ac0f2710fa4294a4e410e154247b22 : SUCCESS in 49m 11s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"037c4b3c85dc5f0f19e6bd12b9ebeeba9589c1d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 06:17:41.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/81d9ecbe39164e748ebc05f44a992b63\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6aa70dbdffb9434680f4b108e152156e : FAILURE in 56m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b55a6d023a4f482c8a3bae4db6b60fc4 : FAILURE in 1h 03m 51s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c90ad4b01e9f4eee9989ea1d77b34f06 : FAILURE in 1h 06m 27s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/664fc2c69e0845f0b8fca95a97961da4 : FAILURE in 1h 24m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/76dcaff386d14346b005b30bfb35f534 : SUCCESS in 41m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee33929515044996a023f8d80895e384 : SUCCESS in 35m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5df7fb11655040f4824be244e48bf9da : SUCCESS in 6m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0b6ee26946a5435cbdab2a40eee9a8a7 : SUCCESS in 34m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/32b905bf74b74082b8eb3e27ab3d6e81 : SUCCESS in 24m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/17e2928957d54f03a20c0745d46b05e2 : SUCCESS in 17m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/435bf30fd84743ae85b40e45fbfea617 : SUCCESS in 31m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ae847b955b14a1cb119300b4aa57cb5 : SUCCESS in 9m 21s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9a73dc2476c14216af98daaf21daa958 : SUCCESS in 1h 46m 40s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4f1972750e804acaa39a7ac243dd3571 : FAILURE in 34m 03s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/951bfc22748641c6b0b2728a063ee12d : FAILURE in 56m 15s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b279f4940bb94076ae474bfc70f4b45f : FAILURE in 32m 30s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/39af84984f3949c2ba3c517e786583c1 : FAILURE in 1h 13m 13s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/39d49626b56d4054a982d76382bd4de1 : FAILURE in 37m 23s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/88aed1ddebc54c42b73ca57ed7da090b : SUCCESS in 47m 25s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5f67bd8be7604b6b956fab2d4e5d6b0d : SUCCESS in 35m 09s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/ebe292b64f044f7db9efe7b1aaaacd41 : FAILURE in 1h 00m 23s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":15},{"id":"e67d3f4b28acf82eba6d688b643ffbac5645001d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-08 06:19:07.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":16},{"id":"6cbafcc2c9cb20f94b324111e01f397dc2eaabe0","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-08 06:19:53.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"895a0baaaa9d27678add05caddfe7f2b25040d3f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-08 06:21:45.000000000","message":"Patch Set 16:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/62/986962/16 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":16},{"id":"ce399344bad6e120e40e9573a618a64a57ed1321","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 07:17:19.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5333039496643648b5841cd784a32b9\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a3dd3f0bb34041a19a19caf3a7107e5d : SUCCESS in 26m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a382a873f3d94c0b876d02bf53c9e0ea : SUCCESS in 19m 53s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e2cfeeb092624fc6a88f72a9cf5decaa : SUCCESS in 42m 12s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"4f912ad8b1276ca7e593bff7d4220a0a598f44a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 09:35:45.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/04db4e5869754459a055084bc0eb28df\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/11150bb5b23a4062b2b074312f96e0ce : SUCCESS in 1h 34m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3b173853705f451fb0d34e78006ca526 : SUCCESS in 48m 12s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d184ffce12294036a9cb14acd3f425ff : FAILURE in 1h 26m 55s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0b7e25f97a5041d58797309bb387b647 : FAILURE in 1h 59m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d84aca5f3d774db39022059af8a3e1c1 : SUCCESS in 31m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7cadcc6595cf4a61bb288566ead0e1fd : SUCCESS in 1h 00m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b79811d9fe6b403484fddb09d145207c : SUCCESS in 8m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3f2a2d8cc739495d8e7d01a7aa22b688 : SUCCESS in 25m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f3c0053227164dacaecb25346a5dfb84 : SUCCESS in 32m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a409f22872f347bf965cb78f5ab1499a : SUCCESS in 25m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bcb2e5dcf43f47d0b5c661e31b0f8c0e : SUCCESS in 33m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c43b41df3984323b4324616ad5ee243 : SUCCESS in 12m 13s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4b30fb8174214c63ae41370cc6ab1626 : SUCCESS in 1h 06m 30s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9b7accd78afb42cd8ebbd670ed9aa1ed : SUCCESS in 59m 07s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/b5e765032d1e44e9b4fec3bbdbf4a607 : SUCCESS in 1h 23m 53s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e34f903efc5c48009e4ad5089c7334c9 : SUCCESS in 1h 05m 10s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a14606714ea646b9903d24c8bc4deba9 : SUCCESS in 1h 28m 08s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b1e0da938b24451fb2037adb52aa21fa : SUCCESS in 52m 46s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/45c4e53de4f94e6d8c26ae9f11e65aee : SUCCESS in 1h 10m 48s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/efdc34cccef84a22a76066c739668d73 : SUCCESS in 1h 04m 27s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/237007a9ec2749d89e3af9a1b302127a : TIMED_OUT in 3h 04m 37s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":16},{"id":"fbe67515df822cea695f977acb09256deb496a66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-08 21:13:55.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":17},{"id":"9eb5a16a6e7d1bbd6667fac6a2deaeb13e7ca382","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-08 21:17:30.000000000","message":"Patch Set 17:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/62/986962/17 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":17},{"id":"c375c993a50ce039d5d08a979eb75c2d686e4ea2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 21:52:45.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fb30911655c486d99846e54c13661ab\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/09b96db5f522401cbdd548877db15d9f : FAILURE in 37m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3b774122dcf446d7983a17289972b7be : FAILURE in 17m 06s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a8c8f1fb786348d181ca8d1cf64315b3 : FAILURE in 30m 54s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"bde4318d52ab59fb699c1f3e0df481c9f149d81b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-08 22:40:11.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"b7acd0b6d3f9cd2b0ccea2e9732bb85049fa64c6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-08 22:43:52.000000000","message":"Patch Set 18:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/62/986962/18 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":18},{"id":"c8afe98724486da84d4da6e9c95a1e5a761ad436","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 23:46:11.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10ea7fd750354a338207ffa2512d6cc8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ebad76b91ede4e98aea9f28bc855262a : SUCCESS in 38m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c1473200db994acba349ae6bc28d4847 : SUCCESS in 27m 34s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/dcf9e09bd2904451bda57712f5dd79a9 : SUCCESS in 47m 40s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"6a45526d64df018129089fa323c0dd6b29ff2159","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-09 01:11:33.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fc760ccb471840509534a4926fac789a\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d6c8131128a54a16b1e59bc945dc7c53 : SUCCESS in 1h 21m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/50e96830133745c68c94a6094f35de04 : SUCCESS in 1h 16m 06s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/10c9adb7d01a4512a6003ecff033e5cf : FAILURE in 1h 26m 49s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/06695ca731784889933dcf05897b4d8b : SUCCESS in 1h 49m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5fa0237754b44af1ad22be981e8f72c8 : SUCCESS in 32m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd530705cebc4a81b6ed739c12ca02ac : SUCCESS in 45m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b2cc748764e40688008aef408e6e5c5 : SUCCESS in 11m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f9f2e5e5cf074516a41bec244404ff28 : SUCCESS in 20m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/91975be88007499b997d2566b614be9a : SUCCESS in 19m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b2d8cf375eba4c95b2a74e8eba75ac26 : SUCCESS in 25m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1ce86ffe44a4403998e1fac3dbe62708 : SUCCESS in 20m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8bdb1b0e2144df5ac50980eae1d7a27 : SUCCESS in 13m 26s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/075e7b6ccc9d41a4b2eae5e40a62f15f : SUCCESS in 45m 16s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0c47d5eae9ee43dd8407e7fff2d4e3e8 : SUCCESS in 1h 14m 34s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/0259b1f6b16142359841e779d98665fb : SUCCESS in 1h 31m 09s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9e882a9162004986836175b527fa9600 : SUCCESS in 1h 05m 14s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e3002bcde325408aa02036a4d10224b5 : SUCCESS in 2h 07m 54s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d288180d42aa49c1aa12dd3750310c0c : SUCCESS in 1h 10m 45s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e00f3b307a17403bb91edb4952872282 : SUCCESS in 30m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/749c7da4b7aa41788b404cd6b251acf7 : SUCCESS in 44m 19s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/67e0f8d881d0445f95137d98def4c86c : SUCCESS in 2h 09m 43s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":18},{"id":"f033ffbac3e6a1f48335b761afca2b66fc1e6553","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-10 12:36:47.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"006c8715bc7f20faea8006c0dcc20865ab2aa338","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-10 13:18:32.000000000","message":"Patch Set 18: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/62/986962/18 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":18},{"id":"1ba21dd211d30334bb683c73fdc2e079fa51a011","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-10 13:28:09.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89c43cbe906842e2b212c3b3514b6eb9\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d5f8fd6e3c3c4bc6a07c4d03495ca9a4 : SUCCESS in 39m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4653bb1398834b679d9879afefaffd7d : SUCCESS in 17m 02s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/434ec959194449ea9e49ba8bd64248ea : SUCCESS in 49m 16s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"a9328136eef02c3b76f657b3d871c6aa9b511d00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-10 14:29:57.000000000","message":"Patch Set 18:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f5e50fbe788b4f2a99ae0c0fdea0c10c\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/34c73b3ed11646a38d211d24516504dd : SUCCESS in 1h 45m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d919176042e8472c854620d9aed7e036 : SUCCESS in 1h 02m 48s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/872d1cdeaf534b14ac5254ca5d9b5e89 : SUCCESS in 34m 11s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2506d0d346dd4184a5c8735cd92fa6e3 : SUCCESS in 35m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/88b234d18f8a4afb8fa71e846bafe294 : FAILURE in 31m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d4071ff08eb7424cb7496136bb814e87 : SUCCESS in 54m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73e2929d5a7b4fb9805923c25f5cbf56 : SUCCESS in 10m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/602ae42920f7440393b8037d9e405a95 : SUCCESS in 35m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/514903781e6944e1826ce1d1349775ef : SUCCESS in 32m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/549a0ba2307041d18f01c5ba8bd0bf66 : SUCCESS in 23m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4c8b56edd523419fbf033221b3ea88c4 : SUCCESS in 20m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c92899744e7342f7a890aaf82fb3e7e1 : SUCCESS in 10m 21s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/91fa80d1b7c24347a2cc8505d5309c13 : SUCCESS in 41m 47s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/852cdbc3bec849329993ddb666fdda96 : SUCCESS in 50m 14s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ddd69246b49b4a00b3fad4d53976fa88 : SUCCESS in 37m 06s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a2cca2b5a47f468b8474c3e8bf97e03a : SUCCESS in 1h 19m 30s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/487b30868b3b4dfca4acd6273e20f50d : SUCCESS in 1h 11m 31s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/25b3065a64b94f3badb231e1bd84bd5c : SUCCESS in 46m 16s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3cf7a52727e44918b669b0da0cafdf00 : SUCCESS in 1h 29m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/05513c2dc8934ef99f07efcb627556b5 : SUCCESS in 1h 25m 52s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/efac7d440c3c48b1aa407638a1171128 : SUCCESS in 48m 22s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":18},{"id":"723189a1a863ba7494f2ab3b1307bd64f1b51593","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-11 10:10:55.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"b7dc3e73489f6d73af8286ff6629934b1db8959f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 11:04:18.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ac1569d720d45af87880e943126f618\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1725e97ad5e04df08f9a60576ce1ca4f : SUCCESS in 25m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bd7e468d84b14e818b89f8d543022195 : SUCCESS in 18m 06s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8f5194732a3742488fa873de12c1b808 : SUCCESS in 33m 26s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"4f012cb9c50f4df3795ede5b7b50766028a2e180","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 12:18:51.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81a541dc35e04060b17ef4dadbe32746\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/982e406f9f834a86a825c83d4fe78514 : SUCCESS in 1h 02m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f00beb6face748ee875698cf94df1245 : SUCCESS in 34m 16s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8a338228f42e467da74c98ef8b87f442 : SUCCESS in 1h 26m 54s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fcae04c391f4416083039119a74ed5a4 : SUCCESS in 2h 04m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3004ffd003924ac28a70e977971fda00 : SUCCESS in 31m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60e35adaea434dbd93922774511ac3d4 : SUCCESS in 36m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d0a7116c6e84a968c9822ef46727034 : SUCCESS in 6m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8d7f6de57bba46c3a01a920fa28342cf : SUCCESS in 20m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/60b03bc88df44d98925c205625ce8d72 : SUCCESS in 23m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/568f4eee18e04fa28c812169a5ee5639 : SUCCESS in 25m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/02e77c5ccfe54682b131d07be2293980 : SUCCESS in 33m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9f41821817248999b0459b63f9a418c : SUCCESS in 12m 42s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/870cc1a6ddcd4f98a198d7be0e60579b : SUCCESS in 1h 06m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/fdb89c2a46bb46a4b141465c24d2c84b : SUCCESS in 1h 12m 19s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/549ec59bc4bb4a1b8a9176501118a2bc : SUCCESS in 1h 24m 16s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/898de4d1f82e466eaa44157c62b55aa4 : SUCCESS in 1h 53m 48s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d4b8c018ca8e4e97b0f9621da651ddba : SUCCESS in 57m 49s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9ab26cc83a6f4c61b210ffd182871634 : SUCCESS in 1h 10m 58s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2ea2931f23494cb4a1cece529839f158 : SUCCESS in 36m 27s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e1d533a35d20499ab9f9f85736e77c18 : SUCCESS in 1h 09m 44s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/aac5be2651034eb5b25363358da6adb6 : SUCCESS in 1h 35m 21s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":18},{"id":"12b94a9be95e26b944cd911bf1f0055db4220cf7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-11 12:28:58.000000000","message":"Patch Set 18:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/62/986962/18 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":18},{"id":"02b0c03f33fc3d3dcf0b041aa1cef8d93dc97467","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-02 02:02:26.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":19},{"id":"044844c6caba3655433f98e3914b4f4c1e4a11a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 04:22:10.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5be5f87d63ee4fa1a7b67ec878e74039\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5c182b1cf15f449abd3231b2e7c3d69e : FAILURE in 1h 03m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5f3b3159fa0c446884392bae946178e6 : FAILURE in 1h 06m 05s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/150dd5bf95ea4181aea94da3ecb84d1f : FAILURE in 47m 04s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3bc8d6c67c5642619594ec1c07807ab0 : FAILURE in 47m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d3e0a055fcd34daf85c2aabe0889ef7f : SUCCESS in 34m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32208dd3978d4fa6bbc786f7611af448 : FAILURE in 1h 07m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f4f361cedc042a8b7465f9c41a9328c : SUCCESS in 13m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ef20e1d7922c4d3d9bd7ef5f18cba58d : FAILURE in 27m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/992f63347ea5400da70b996c817dea21 : FAILURE in 25m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c934186137f7456a97027fb87804600f : FAILURE in 29m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6955ca7e466a482eb65880071a61e1a5 : SUCCESS in 13m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a6197e56a2c401b86dc0c87982ed5f2 : SUCCESS in 14m 18s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/41c9d95b790d4ed2ba4404800bcb462a : FAILURE in 48m 21s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f23001cc155d4068b2b0e268e40dbfb4 : FAILURE in 1h 09m 20s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/f5d3b8e4373e46a28409c6540e7ce27c : FAILURE in 1h 14m 42s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/00f25e0e966e469b9dc1c76bf9a966d6 : FAILURE in 48m 04s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c50113715c3546339c202ca381f3bccc : FAILURE in 1h 04m 50s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/09325e8d330d46a782b00b3d7065c27e : FAILURE in 41m 30s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/02f1de6309bd46598147dda560d2e7de : FAILURE in 1h 05m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f9a7a8b4e9e140c2a50db19c8f69cd39 : FAILURE in 1h 29m 58s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/75e2a24cc4514142b1950dc18d4761b9 : FAILURE in 1h 10m 48s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":19},{"id":"3a788063fe382a15f28f90cae09199947270c708","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 04:54:24.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13007874c8834bf6a71750d752865e8e\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4978aa8b56a54ae4a1d8f787e84d4a34 : FAILURE in 56m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2e8ce545d0b0441d8b8c75b9ccb5a412 : FAILURE in 43m 23s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ac30f211600541cbb202a731f4df031e : FAILURE in 58m 07s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"146aafa936d0e11f4d26872fe782beabae0f4192","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-02 05:00:54.000000000","message":"Patch Set 19:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/62/986962/19 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":19},{"id":"ec2568216aab61d4ffb7f5f7034c9196d3f445e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-03 01:25:20.000000000","message":"Patch Set 20: Patch Set 19 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":20},{"id":"d60612567d7b660bc3fa1ca0bdbe06e4f68b9459","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-03 01:35:41.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"1486ea50e3b52c0256d6ff4fa4f6d1ad4ff2f16a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 03:15:14.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b984491ee8fb49da99472e9ce9557601\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/052eb568487e4bcf950f46a4de7cb22d : FAILURE in 55m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5bf48dbb1ed04680862df24bf9d00313 : FAILURE in 1h 08m 11s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8d51af7a9b7c4f1490b9435c47bcc435 : FAILURE in 1h 20m 45s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/29904bd2871e4ae0aba626702e91e708 : FAILURE in 1h 36m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/759fe7aa621d49fa86adda4264707b09 : SUCCESS in 35m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51b4466c1f1743598838ba0d508edc33 : SUCCESS in 1h 04m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40741440130042dd821d4c7bf1469511 : SUCCESS in 8m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cf9731831e8748faad1256ab257b9497 : SUCCESS in 30m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/be0934cf812d412b8254f1cd120ae89f : SUCCESS in 26m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fa8858375a404438ba2a904f98f9de41 : SUCCESS in 33m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fd24965efc3451a8fb5bf5d76dbe614 : SUCCESS in 12m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1cf38636da8b481995b02560673a1d8e : SUCCESS in 20m 37s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/86d56b6253b14f86a47770ef106e5671 : SUCCESS in 1h 28m 09s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8a09cda2af444618924462c6973236ca : FAILURE in 1h 09m 32s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/6031f1905086470ab060e90af5e0b44c : FAILURE in 1h 13m 32s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2a14b6f7e1394acd88fe9ad6a4aa1fdb : FAILURE in 47m 49s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2588598aab3a4444967ec014e050b4b1 : FAILURE in 1h 07m 28s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/75fd3e76ec8649dfbef409e52995097a : FAILURE in 54m 16s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8f12fcd4523c41c9a9d62fb2e65e4913 : SUCCESS in 1h 02m 18s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/488f134a353e401183839aba361f6606 : SUCCESS in 56m 30s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/f0225d09f6234f239e78960b9c8697a3 : POST_FAILURE Host unreachable in 58m 45s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":21},{"id":"8f30fec70a90de54074578fb8e17153ab327f4b3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 08:08:27.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb6875ac5c6a4e32b353cf12bdc0a4a5\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ad47c586d69045ff8b25f124bc8154eb : SUCCESS in 51m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a89bf685a8924c2b9b2ea08cc069e3b0 : SUCCESS in 37m 58s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/fdddf648ce9143e5bc1134b6e7f93000 : SUCCESS in 1h 06m 14s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"f8b57fc27d2859b6f1dadb3a4b53a8db56e8eb8d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-03 11:36:12.000000000","message":"Patch Set 20:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/62/986962/20 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":20},{"id":"37c02928aa7d7c7d7c67367831baa67367d74f26","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-03 12:19:19.000000000","message":"Patch Set 21: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/62/986962/21 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":21},{"id":"0b30d722efc3c98f95e45d022ea3d9600b10face","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-05 03:37:34.000000000","message":"Uploaded patch set 22: Commit message was updated.\n\nOutdated Votes:\n* Verified+1, Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":22},{"id":"09d80a609f369111bc2d417f4350e54191077024","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 05:38:18.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0bac1f2562024de8b333b0abd95c1885\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/267cc095946e4211b08da1c0ffd8d2a9 : FAILURE in 52m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6d6c9f7b35a5451c849937fa926e0630 : FAILURE in 46m 52s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/de1b689713dd42b0aae7125fa00bce4f : FAILURE in 1h 13m 40s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ec9bf91cef2d4587944664dd14e7324b : FAILURE in 1h 29m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a093c8487f194b95b9711cbe2f9af22d : SUCCESS in 32m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46994762abb04ee4b2e2365e308422cf : SUCCESS in 1h 05m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7795673c91b34f45a0da79cf03efe154 : SUCCESS in 11m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5c6f27d202864fad87dd3d1a90e10e34 : SUCCESS in 28m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9adcc2f49ddd4b6d8ece33673fa58f56 : SUCCESS in 29m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e9a1681653154bc5819afc305cc79c6b : SUCCESS in 36m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/115aba840dbb4a2b9a09dcb28c6cdee0 : SUCCESS in 12m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b69011f142a547fba726e6c07c802fa4 : SUCCESS in 6m 20s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/625c52c6311e477db3f796f9e19003a8 : SUCCESS in 1h 27m 15s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b4b8b46dafe9443f96e18285813b1054 : FAILURE in 1h 08m 58s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/b7d1d878bdce41b2832d762fcebe9ec6 : FAILURE in 38m 21s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/800f20f9d80742c4b13e8ac4aa0636f7 : FAILURE in 1h 03m 25s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/99fdce8045b341d7982308b4655ad2b6 : FAILURE in 35m 44s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/fb527b104d07445c94c3d0416e11bce8 : FAILURE in 1h 04m 08s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/61638a90ac7043918c177e8daf7db8fa : SUCCESS in 1h 17m 19s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a83f1f13e7904809b92c559e6cb73b40 : SUCCESS in 36m 53s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/24a8f7f46840439eb042c2665e31b22a : FAILURE in 1h 54m 32s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":22},{"id":"58d475a572fc46209a08335941cbdcdbcc9c05a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 05:44:04.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb916d110f3e4e638210d338a8d3d2f6\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ea52428c9a834a0f8f0203d3dc041a7a : SUCCESS in 27m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1e706b730c684221b7ad26f4a1a25eb0 : SUCCESS in 52m 20s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/28b6677e1def452aa7e06527d1ef32bb : SUCCESS in 37m 21s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"6633cec0825edda67cefca700103409034221021","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-05 06:31:47.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"72f027cd0d10e018b7b935550774843b38b14f05","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 07:47:50.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f6d124bbd1c48c78c9e123f3bc90651\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/71f78a3c4f6d43468c1d25fed8ee1404 : SUCCESS in 29m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f1c69bc86839472597446ac5e68f18f9 : SUCCESS in 54m 35s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/84033fae72054b40800b44707acbb75d : SUCCESS in 1h 13m 12s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"ec0dbe9d856f2e6d2d3d26fa5a7be6da4b214fd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 08:25:29.000000000","message":"Patch Set 22:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f01ae6c1817244e48d99f7712304ba32\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/fa384806b4534cb280f0643812032ab1 : FAILURE in 37m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/58d5f036e56643e48233fbde005fa7de : FAILURE in 40m 06s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4e28196987e94ea7bbf6350478862697 : FAILURE in 1h 18m 04s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/cf62356f33a64399a89cca5292817ed2 : FAILURE in 1h 27m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d21fc48006cd45559b470cfd91ff4fe5 : SUCCESS in 33m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0fc1b1b78e7747ee9bebc7f825122d36 : SUCCESS in 47m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcc500a6dab248e5819800dda16b6593 : SUCCESS in 10m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4608c7f96e8244aead1c4e50e62f3649 : SUCCESS in 23m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0067ba00ed684087ba5ac004e74c21aa : SUCCESS in 21m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6c4f4976c26b4848bbb50a95e1766f95 : SUCCESS in 42m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d418c3aa24e74d9ab54fdaba255871ba : SUCCESS in 8m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36c387240f6047a5ae2d380016af3967 : SUCCESS in 12m 55s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a9bf9798466747d099bc371f6f42d815 : SUCCESS in 46m 17s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1c1ffcd5f3624beb8683f88cd795e098 : FAILURE in 1h 08m 06s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/62c8f65bf9224ad581702d7c8418f146 : FAILURE in 1h 14m 17s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/427ad31bb7f14c66bc3a38249e3b53b4 : FAILURE in 1h 04m 50s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0f0bd6ed56824c67a1a9e54f203e2a61 : FAILURE in 47m 52s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/074d5892f6064b51b7c1d257856c360a : FAILURE in 45m 46s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c6f59a1d7b04493eaf23afa8b1c69c61 : SUCCESS in 1h 11m 46s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5a3373446310478ab4b67f0072551680 : SUCCESS in 1h 32m 55s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/6e6483fcce004321b3f292e95dbf42e5 : FAILURE in 1h 24m 48s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":22},{"id":"be5f07c93396c7bbd453c0adf465e222f06f7419","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-05 15:16:58.000000000","message":"Patch Set 22: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/62/986962/22 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":22},{"id":"1fa43185a6a5ecbccb169de17ce978a17f461793","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-05 16:02:18.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"8c7f0d24b4cb65c9d1ea0175870148d3ce62b3d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 17:37:26.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09aa0314d78d46428eb33c1fe58bffd0\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f98169eaa4b34687ab54b69bf1c2d6f0 : SUCCESS in 54m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5156bea42dfe461bae7518e0d0016134 : SUCCESS in 20m 12s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/715637b2b5a74456bb7c30319379ea40 : SUCCESS in 55m 01s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"94c91f6f24f47af5fe59a116a2d26000f56db87b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 18:24:29.000000000","message":"Patch Set 22:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/312f03a33aa44ef6ba999550383907c2\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/260ccabaea7840d5aa84f928dea61329 : FAILURE in 55m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba50f1d3f8f043adbbff26c830e26e9e : FAILURE in 1h 11m 56s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c705e33b6e974ca7a40050fef728980e : FAILURE in 1h 16m 10s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/27d7b474dbd54955bb19bdd22d4e9de4 : FAILURE in 1h 37m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6fe04d9379114692a45a8a0bb0265f40 : SUCCESS in 33m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee6c04f7abfc40f09f9b4bd1c9c7bf71 : SUCCESS in 1h 18m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06d56bfccbae4c02b129fb79124b1bf7 : SUCCESS in 8m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/244454a9d18b44b4ae37f78db3276ca6 : SUCCESS in 37m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6922b19f4d5b4780b08696aaf2ea25d7 : SUCCESS in 27m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0ec71722982045348488e9fbdf3432e0 : SUCCESS in 33m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/278326a5fb8d4aba800a85e3643a4443 : SUCCESS in 8m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e892e1c6c6f40c28c3f10b12c8381c7 : SUCCESS in 20m 58s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f73d73c8d92b417cae51ccd9efd51bd1 : SUCCESS in 1h 32m 26s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0fcdee76544e4bf8ba68db086cc3b446 : FAILURE in 1h 08m 07s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/08101b2c371248a0b3783a710efbf8e7 : FAILURE in 1h 18m 10s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f1bb121130714e6686673d6e8600424e : FAILURE in 1h 07m 58s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0d760526955943a899c60f8cb6da5fa9 : FAILURE in 1h 10m 25s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/bd53878c18b54636b7d2555306663cc3 : FAILURE in 57m 21s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/897c26db178e4dac90f614d7e081a018 : SUCCESS in 1h 18m 37s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b2a6465351644018b2b19c9f6b8eb14a : SUCCESS in 1h 27m 08s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/bdd37ffe2a754928b5ce260f0a3a5ab3 : FAILURE in 1h 34m 46s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":22},{"id":"ebc7d27e7ab274e5e9a6268ed5a4a350315f7bc6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-05 21:19:45.000000000","message":"Uploaded patch set 23: Commit message was updated.\n\nOutdated Votes:\n* Verified+1, Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":23},{"id":"f180edbf886671e2854b6c969e4c1c67ab0c90fa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 22:09:37.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f645492a940b4a05bde37dff58ca2130\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/254d8c6ce7114d708beb14520729dc9c : SUCCESS in 47m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ccccef798e2b45ad8da7db4d4b33df87 : SUCCESS in 43m 46s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/6ad30bbe5af142deb163c3c74a3dcd1d : SUCCESS in 26m 19s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"6eedaed5daea54d4cd6041dfe8090709dcd2eca1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-05 22:29:37.000000000","message":"Uploaded patch set 24: Commit message was updated.","accounts_in_message":[],"_revision_number":24},{"id":"c70fb13d1cba0631d2c60d912c0c22be6b520c97","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 23:35:38.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3cadd6669b44547a4c5e7c506b0d613\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7c5c2de24fbb415fbc1f141a3e04b77b : SUCCESS in 49m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5580fa03ae064c939fa71e624ada316f : SUCCESS in 46m 47s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/255ad0120d2d42908565b2333dc156e8 : SUCCESS in 1h 02m 57s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"15cac6d73034a64a224e286743e77cf1558856dc","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-05 23:46:07.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"15ac3809462d37e3695986b6bcd036332a99e18c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-06 00:25:14.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5b2c0a33472c4dfc94321b4a8d0bebb3\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e006d5ccd5cd409ab573db33d65c0c43 : FAILURE in 1h 12m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b2be14181fcf493b972fff95fb14bbd1 : FAILURE in 38m 43s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e046ddfd43eb4cb9947c471574e1fda8 : POST_FAILURE in 1h 29m 17s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ece0e295ee3e4e2096672913487169c3 : FAILURE in 1h 54m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3825013c73204966a1f00d83c8c06fc3 : SUCCESS in 33m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79adde9e436d46a9a02e223b74d270dd : SUCCESS in 1h 08m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6143184b47d4f95a5c9cd7ea25f5d63 : SUCCESS in 5m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/69ef49aa27de4f6c954f3c9272dbf8c9 : SUCCESS in 37m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/71c37215e8b04201905405b025a6bf57 : SUCCESS in 16m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/70f7ad1cadb246ff9de625855beaeebb : SUCCESS in 33m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0585438afb04c5d8189ba8a7c149b2c : SUCCESS in 6m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6b610b6a2664ca7a4768ab1bd19635f : SUCCESS in 6m 36s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f1f95e540cc14f2eaffd66075216df88 : FAILURE in 32m 00s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/814ac047a383400688b33e3e9a0b6a15 : SUCCESS in 39m 57s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/f21f91caf74143248cca51e0d5c1641b : SUCCESS in 1h 15m 55s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/636bf839fdb546878951faa628b14015 : FAILURE in 44m 21s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fa9cc09b23c143c9b5fa5da42e692bd5 : FAILURE in 43m 38s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/08ba13cd80b141358b0552ddec16677f : FAILURE in 56m 41s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/883cbd59016047be9997cf211455301f : SUCCESS in 1h 11m 44s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/70db3a5c89084a2cb44c0906e853be33 : SUCCESS in 36m 45s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/cc56276258c84ea38b52e55f88212ac9 : SUCCESS in 1h 08m 10s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":24},{"id":"afbfd85d582a457746fa1c48a9953a46e1087f42","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-06 02:54:58.000000000","message":"Patch Set 22: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/62/986962/22 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":22},{"id":"88778bbb12a0f37089942a0cd4f15b836709834e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-06 09:08:30.000000000","message":"Patch Set 23: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/62/986962/23 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":23},{"id":"736a01c5f1ddca7fea637b12d12699b81c1ada3f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-06 09:49:55.000000000","message":"Patch Set 24: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/62/986962/24 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":24},{"id":"37741f7524df5347ae2203dd7d326315ba63945a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-09 00:11:43.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"25a461990c4426e198e2075a18314d7051fba0f0","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-09 00:13:19.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"96bc2fff2561f8f45cc4cf0db4c7fff315ed42c9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 01:36:51.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa5f66bf53ad4b48b92eeee25f54dde9\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/de725c7a9a2e4395a5ead82b6a4c69cc : SUCCESS in 26m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e229df0b24d5424d95db0415f59c3e26 : SUCCESS in 25m 19s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/6f27c0be7d81412baa6ad6979145ff97 : SUCCESS in 52m 39s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"1debebc444692336d0a1d2cc178f68974c8f9f49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 02:27:30.000000000","message":"Patch Set 24:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4674c605e43e42c5995cb644ca3c28f7\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b2ffb23238df4885ae34b6b6b2d4ce93 : FAILURE in 56m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b1a3c84c0ce14b138b1c7a4e90791df8 : FAILURE in 1h 01m 52s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/da3af4d2b72043609b82cc2609ae4afc : FAILURE in 1h 44m 06s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6af1cb8467bf4033912f2f27ee60302b : FAILURE in 2h 12m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b5f0fa1b5fa94f89847ebcdda5f10bd0 : SUCCESS in 34m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/334efd88261c45f98d4591b6533253fa : SUCCESS in 40m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e44c84098267433185e385aa31efd2f0 : SUCCESS in 7m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/82c379e3252645f3b8e54eb4cdabe165 : SUCCESS in 24m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2c3769c422d144a1b9d024b4e6a3451e : SUCCESS in 24m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ab80ff5b29e245bba3396f8e5ed568b8 : SUCCESS in 24m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/198154c7dedf4c07aa6e1eb859ed2863 : SUCCESS in 12m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c22d2bdcd9a40cda9b75453ccf4d87c : SUCCESS in 16m 33s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/07caa4b9bff34ad2ad36d51500699387 : SUCCESS in 55m 09s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2ab3629812d7461480266c0daf6e8d41 : SUCCESS in 44m 02s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/2a9d31d5aacb49d6b390c274e40d53ad : SUCCESS in 58m 08s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/830cb682226140f993b3dae29db5eccc : FAILURE in 1h 17m 25s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c757be5b36f144ac93eab75239e822f4 : FAILURE in 1h 13m 47s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f8b38de50a99489bb757ae20c6709d7f : FAILURE in 1h 04m 21s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/315b97e0d6fa47b7bbbd6adad00d36bc : SUCCESS in 36m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f35c65d694154105975c693646bd08ed : SUCCESS in 1h 24m 59s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/f83e36f313a24809950b877c9d18c484 : RETRY_LIMIT Host unreachable in 1h 05m 51s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":24},{"id":"661190c2245a7e75fab9cc3eff2cf48e0a5034e4","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-09 02:29:35.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"95f6a591dbd15b2a690b93ec248d51eb495fa5af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 05:04:44.000000000","message":"Patch Set 24:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1967d2d6569249edbaa4e41e79d9ad47\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b1fe367951ba47a99c40f74ffc8c92f5 : FAILURE in 1h 18m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/823492e4162040cf89dbde107df3f401 : FAILURE in 1h 05m 39s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e224f2c040e24ca580aca9c2f07ba4fe : FAILURE in 1h 22m 18s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/acdd607ca23e40b9973b1b3c42bc360b : FAILURE in 2h 03m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bdff6be529014f4ab6e91495a814d973 : SUCCESS in 36m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9318b8f34d3d48218f9549c864b52493 : SUCCESS in 1h 20m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/863a53f435da443db6dd7f26cff12daa : SUCCESS in 12m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8c7d662677204a22a11d7f18396b7143 : SUCCESS in 32m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0a90673e35284ddf9bc5affef2a7f011 : SUCCESS in 26m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4cfc4e1e93cc4a628d8d2239112a9524 : SUCCESS in 31m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec05c24df155438084960a2e2daeb89d : SUCCESS in 13m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71fa44dc88914badafd6ccabe78ef8c8 : SUCCESS in 7m 43s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cc0508b742ba424f85d80ca93dfa4249 : SUCCESS in 1h 06m 03s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a91c3333b0434a0f8b04e7b311b21715 : SUCCESS in 1h 10m 12s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/3777875be4ea46dd8cf42b93741bcd39 : SUCCESS in 1h 01m 39s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7012d6c2b3ac4310a6162f1dabd8d3d7 : FAILURE in 1h 02m 47s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5371d3582be04347a3f6b77a90d806c3 : FAILURE in 1h 21m 39s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ae38aad39d3b4a9aa5da40d231356d6b : FAILURE in 47m 27s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a325021e1f80436eae223b4ffd1fecaf : SUCCESS in 1h 09m 13s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9c9d262bafcf473385f5e347c550f0e1 : SUCCESS in 1h 25m 04s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/9f5233235bad463083c3d3e8daf4b327 : SUCCESS in 1h 53m 18s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":24},{"id":"28c94f0f6ca84e28fbdaefb66efa728b521986da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 07:28:31.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ddac6d6468c640a997f71c784063a8be\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a3c8690a57db49cfb3eea4c6fad8ec3e : SUCCESS in 56m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fc994a7cb37346dba09bd6908fd9e78a : SUCCESS in 49m 45s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7e6907810f554b178f23436941749f98 : SUCCESS in 27m 07s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"ce25de442760e19fe56bf337e8275f3fb110bb53","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-09 08:25:27.000000000","message":"Patch Set 24:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/62/986962/24 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":24},{"id":"b3615c1215e7daad0895db7c9f2bdcd2593618fd","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-09 16:19:05.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"b1fe08222dbcefe0cda58fd1f280c3a9c5cea985","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 17:19:25.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8646ce2497f4ac48528427a212e0693\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8b73067584de43c19605ea5a39582ef9 : SUCCESS in 44m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/122f49eb7c10442f83b7bfcb52832e65 : SUCCESS in 47m 35s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/6a4b56403b794aa6a8677373bd515ab3 : SUCCESS in 28m 02s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"5cc4f6545581b6ef8ab6764d67982b171d175d31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 17:29:17.000000000","message":"Patch Set 24:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/tempest/+/991754 is needed.","accounts_in_message":[],"_revision_number":24},{"id":"1a3dace7c4885e73546d2b0e7806fe297a917a92","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-09 17:30:26.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"12f2c5f4a6cce4f111a3ad84c29e7e2f8d6517b4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 20:00:28.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74e89fae11a44e44b6a3f10d04b99166\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4ab86c72956542ef9cdfd92ffe0eb681 : SUCCESS in 50m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/40413c59557843aa825cf313710a52a6 : SUCCESS in 40m 45s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0d252f911c6b40c69a02d66cc7c7cb8b : SUCCESS in 35m 46s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"1f3158d18981a11ca523fb40b2c69b5e0f3dbe81","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-09 20:23:09.000000000","message":"Topic set to neutron-project-id-migration","accounts_in_message":[],"_revision_number":24},{"id":"9fee9530f84a25531d628f50d8b007e39bbca3fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 20:31:07.000000000","message":"Patch Set 24:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cbb32aa853b94200ba79d3371a114b10\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e48042e319324e578b69223b3489fd5b : FAILURE in 1h 16m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/012c3c70a52b4295b82c8cbef05b01f1 : FAILURE in 1h 12m 01s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/db0f1df0684248f5a22e91d4802243bc : SUCCESS in 1h 35m 50s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ea440e3df8e4416b8b2556f799b93f4c : SUCCESS in 1h 59m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e004a2572ee6446a811adb326226393b : SUCCESS in 38m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef285811da924d24af7256dca2e172b3 : SUCCESS in 46m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51b22cf657a94de4a951243177e5f602 : SUCCESS in 11m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/06e8e8ddae8d4f08a750569e7b66288b : SUCCESS in 27m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/25871a158cc54f95a6bf8d819877d818 : SUCCESS in 24m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f632cd4cb859430fb74666be5f9471fb : SUCCESS in 34m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58f9df201e964869b02d24df33feb6fd : SUCCESS in 7m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bddcab138683492ca605e0e07cefa9a7 : SUCCESS in 18m 06s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/207d30fe74a74ec6bbf7376a85849e27 : SUCCESS in 1h 17m 07s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7964a1b838b64b78b068ff53df9467e9 : SUCCESS in 40m 59s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/35b01292e7794a308c611645284f4163 : SUCCESS in 1h 05m 50s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/381e42e10dcb42f19a9b08e05efdb682 : FAILURE in 1h 21m 32s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b3be0eb1b01b4e10ad729adb8a4b7a8c : FAILURE in 1h 32m 35s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6f9eb9235adf4236870f236540bfd3e5 : FAILURE in 1h 11m 46s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/164dfe238d514ac997faa0a2aeb20f06 : SUCCESS in 1h 16m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ef74b7c933bb4d60bbdad6f98cbcbf90 : SUCCESS in 1h 26m 47s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/324a2a4f604244a19deaf43dc1464ed9 : SUCCESS in 2h 53m 25s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":24},{"id":"31b8ebecaf36f1fed186f325447706952b538a69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-09 22:25:35.000000000","message":"Uploaded patch set 25: Commit message was updated.\n\nOutdated Votes:\n* Verified+1, Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":25},{"id":"4cf6232e3959b4e2345bd639e12821531e313b08","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 23:43:53.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96a5a34363184299ae75e07399b7221a\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/63d5584f0e464d04b65d0ab971313dee : SUCCESS in 57m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/75f09c977d8f4a39b8f36bf93a500782 : SUCCESS in 16m 48s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5f6e7484670a438b8cdc7de131f72bd9 : SUCCESS in 1h 12m 56s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"ab5ddfe8806a0aac17f89d18bba1952775eca487","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 00:33:01.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ce416a88e074fea949c33a2f4fbed3e\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/95d751a8e06f4a81ad54c48a93d76be0 : FAILURE in 1h 12m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/761642d368704d779f201c57008e24b7 : SUCCESS in 54m 29s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/aea473d4e4354c6a980a1adf7c976a35 : SUCCESS in 1h 32m 37s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a9846644a3dd40deb989758035a08124 : SUCCESS in 2h 05m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/132d5a3b74e5418ca852e9d8ac9aede4 : SUCCESS in 34m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d49093bb2a743de971353ac8afafcaf : SUCCESS in 1h 02m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/430e64428b0646529f5a2ed34e6bb778 : SUCCESS in 7m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a5bd9d328c3a4f6e9b4b1f21d88da45c : SUCCESS in 31m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f47bc33fde6c49d2bd4ff334fa68f7ae : SUCCESS in 17m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7134e875a72b4b578bc6be5617503ce7 : TIMED_OUT in 1h 01m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9670806aeeb042cf867a45c59d91c020 : SUCCESS in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a2b82cfb9a274cab9cc3105e8b30a164 : SUCCESS in 20m 09s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/58e18c7d858b448dbb18aa4ce69ff368 : SUCCESS in 58m 29s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f6aabf27f31342d38f5c845bbb168506 : SUCCESS in 1h 00m 48s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/88449739e4774ff5adfc68df227f6ceb : SUCCESS in 1h 13m 47s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a82249a760ae442ca4e2a2ddfec62873 : FAILURE in 1h 20m 40s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d37b9b0bbb904d68a379ed41278f03ca : FAILURE in 1h 32m 48s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c47fe1dea0c2439ba5738a5e4cf59619 : FAILURE in 57m 51s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c9ce014631594541a66f34217a1320c4 : SUCCESS in 38m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6ce6cb72ef334908b6bedaab8605aafb : SUCCESS in 1h 11m 30s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":25},{"id":"080f1f3ff3c396c83b323d916bd391c0468773f8","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-10 05:56:32.000000000","message":"Patch Set 24: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/62/986962/24 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":24},{"id":"91efbeadfb8e844b3b42bf46e811812d94d02f06","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-10 10:10:28.000000000","message":"Patch Set 25: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/62/986962/25 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":25},{"id":"a7f9097af656635a8e65c2cac9d9377dbcf30b07","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-12 03:24:19.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"fd1f06a5c33af68a95c379763271fbade77ca278","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-12 03:37:07.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"593ed099c013dfc84af80fbf2e20aa9dbbf49a53","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-12 04:18:54.000000000","message":"Patch Set 25:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/62/986962/25 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":25},{"id":"9d762f298fdaa306b46c60094cc04ab523856ea2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-12 05:10:02.000000000","message":"Patch Set 25:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/62/986962/25 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":25},{"id":"d05d7dc2123e03ef39d7e226b6130cea0758496e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-12 13:35:37.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"175241920bf96e3e6785ef752f0a38aa5e01b5ef","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-06-12 13:36:56.000000000","message":"Patch Set 25: Code-Review-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/nova for 962604,16","accounts_in_message":[],"_revision_number":25},{"id":"976f5295de5509e529d71d2c57c518821758c6b3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-12 13:38:49.000000000","message":"Patch Set 25:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/62/986962/25 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":25},{"id":"bae0802c6a10602542dc05317b06a04de9ff3992","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-12 20:53:54.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"a5c635202c6a93465a16eae2c672d3e4d51c7cee","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-12 21:33:14.000000000","message":"Patch Set 25:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/62/986962/25 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":25},{"id":"9b4e5363155e954329646d5343f7f3a9f37c3d98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-12 22:02:33.000000000","message":"Patch Set 25:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/nova/+/991948 is needed.","accounts_in_message":[],"_revision_number":25},{"id":"177a96d4ddb67ee10eeda28a7d65c3ae8cf761e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-12 22:02:33.000000000","message":"Patch Set 25:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/nova/+/991948 is needed.","accounts_in_message":[],"_revision_number":25},{"id":"56415020131849db3445ae54d4d0a12a873fe64c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-12 22:02:56.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"4a166bdc9a694183439b156419d1d98492948687","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-12 22:51:34.000000000","message":"Patch Set 25:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/62/986962/25 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":25},{"id":"e92dbd15df5a28d7d2e2dd86ce1a40a7dd6684c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-12 23:56:43.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b93d1384f67449b820d69c946fb083d\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/93c51728d97945d294de49c8a7e99d89 : SUCCESS in 1h 44m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0b544044afec4d938c1c6393fcb15d5a : SUCCESS in 1h 14m 18s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2ee5a2b9482f4f1da390e5acb28bdb59 : SUCCESS in 1h 30m 24s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e3073053b7c54de0a0cc1576cd65dbe3 : SUCCESS in 1h 44m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f05c15b517d04a10a103f2ceede5e73d : SUCCESS in 31m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/88226dd8726c4966a996d0c0ca2c3227 : SUCCESS in 1h 05m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/349b06e8aa634043be4481ba2b6e6238 : SUCCESS in 11m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b69a88144a284884a77769f8adf7765a : SUCCESS in 23m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/09d0d4a2d3ff4545b87bbddc9d6d49b2 : SUCCESS in 29m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f154ab33c4e1435abf32a13609cdb018 : SUCCESS in 27m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11bf6b25d47b4ab0977d843cb7f73729 : SUCCESS in 12m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/394c8fab43b743d586b8deebe0f4fb01 : SUCCESS in 21m 19s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2def39c13ab74e8183fe1b616f57bf87 : SUCCESS in 1h 32m 15s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7f481e9d70d8485a8d92417391c87eb5 : SUCCESS in 1h 12m 56s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/6d57264e1cf64807b6f88c4152c0f054 : SUCCESS in 1h 20m 09s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8b73131a7cde45fe93457ba77af60aee : SUCCESS in 1h 13m 28s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ad6e859c73574d9f810d1ab77b0cc32a : SUCCESS in 1h 33m 23s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/16a6c3ba2987489789d92bbc465c96cb : SUCCESS in 51m 58s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/6b9753285e264e5b9f7ada5c6532655e : SUCCESS in 38m 09s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b349c2b3013147f2afbf12f1cf36d2f5 : SUCCESS in 1h 01m 49s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":25},{"id":"da15df1aa620f4234fbe5eb2bd63814fc3f51f07","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-13 02:06:39.000000000","message":"Patch Set 25:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/nova/+/991948 is needed.","accounts_in_message":[],"_revision_number":25},{"id":"0495e4101bb44949ba0eed79c635d4ebe5a91bfd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-22 15:45:00.000000000","message":"Uploaded patch set 26: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":26},{"id":"d5637e2164367a91c2868f9c7f94119d5818b8d3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-22 16:37:14.000000000","message":"Patch Set 26: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/62/986962/26 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":26},{"id":"37b1d5a89ba93f0b7cc6f06e4b20f68f4e3cdad8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 16:58:52.000000000","message":"Patch Set 26:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/nova/+/994321 is needed.","accounts_in_message":[],"_revision_number":26},{"id":"9f1c894204e682fd16d15f52488f6349f0de6a7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 16:58:54.000000000","message":"Patch Set 26: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/nova/+/994321 is needed.","accounts_in_message":[],"_revision_number":26},{"id":"9c89d8044720a099a236be1e0dc6d64c5f54f5d8","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-22 17:02:20.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"ca076bb838991bbf9407aa7e4dbe36fc54ea4052","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-22 17:58:59.000000000","message":"Patch Set 26:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/62/986962/26 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":26},{"id":"aec89d7d25eefeaff4b8df2d8c93a56f27f32a6b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 18:34:33.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62de67a0faa74b769bf746dad76112e8\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f602f4e3db7c44868fef4d9c27e6782b : SUCCESS in 1h 00m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4936d6cdf8f84aa1a2d38278ba0e5e03 : SUCCESS in 49m 15s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/073c1acc369f4e4abe9848c35d8d956c : SUCCESS in 1h 13m 52s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"3ae280d4dd47ecd5afe64307f2e2dbaaf5febb54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 18:43:16.000000000","message":"Patch Set 26:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/425ace710e284f69a5970324c6766523\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e86e462434d247909e5de7ca4cc2b1dc : FAILURE in 58m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e0812dd0bb9e4a179599943ed7561df2 : SUCCESS in 38m 31s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/536548d711a241afb1496f7b5624bb8c : FAILURE in 1h 11m 49s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2da3ed49c95948958051df1a44daa4a0 : FAILURE in 1h 39m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/566851ef01f248d4afc8cf01935f7028 : SUCCESS in 31m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4925f8e22291410a927f5e5441a99945 : SUCCESS in 1h 09m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e85777bb082c45c4a9b72c413e11d3df : SUCCESS in 11m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9ba5e02a48c3439fa9c1fdd59eaf61d4 : SUCCESS in 32m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7de2896b55964b7aa2586633e316de29 : SUCCESS in 24m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b112d7939d114dadb00d2a4b9a5e08b0 : SUCCESS in 26m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1660007d3554dc6b8ce2eda23519066 : SUCCESS in 12m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a57b3a678616401c8e3a1c295ecb1c5a : SUCCESS in 22m 47s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a35c66da6a714c7da548f59c57169b56 : SUCCESS in 1h 27m 44s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c45666a0e93e4ef2a86efe5bc8525bf7 : FAILURE in 1h 03m 25s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d348cb784fb249dca8f7cd105f998c7f : FAILURE in 1h 01m 34s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1570ada329c94d7ca9a749a838440bb9 : FAILURE in 1h 04m 02s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/38b0ae410e9f4282890039f46fa0e981 : FAILURE in 1h 02m 55s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/021c66a03e914518bd2f121b14dacb91 : FAILURE in 49m 49s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/301558d97af444888af2f1cb650f21fc : SUCCESS in 30m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a56ab6e902f142f6aa0af21faa1fdf72 : SUCCESS in 42m 12s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":26},{"id":"bc8f52b430679bc5f39c1691b1e316deccdbc80f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-08-10 08:36:13.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":26},{"id":"d60f49f6dd704444164fa30ea4a9dbb0041ca79d","tag":"autogenerated:gerrit:restore","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-08-10 21:23:30.000000000","message":"Restored","accounts_in_message":[],"_revision_number":26},{"id":"e538eb9b82cdc570892b999a7c2bd992845a4066","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-10 22:52:45.000000000","message":"Patch Set 26:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9c709511b5884b79a5a95dc30c383d70\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6f138285d7994b70816175b06d7a11e4 : FAILURE in 49m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/83c6c5b95f5a4212a033c73d47a44cbf : SUCCESS in 56m 40s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/bd0e6cb636e34721a90bbb07cd2d2a05 : FAILURE in 1h 01m 32s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/60c73aaa5ddc4334abb2a01f61143aee : FAILURE in 1h 26m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2003ad7f452b476b8b3e95ad376aa4bb : SUCCESS in 33m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f8423bdb8944015a8bd86da19d737b1 : SUCCESS in 51m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f55bb244ca34d2abd333b7d3ff7fb35 : SUCCESS in 11m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1623161b99d2459c827cec73b706fbc8 : SUCCESS in 32m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7994140964fd48d5b391dce5078270ac : SUCCESS in 19m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9bac584171b84fcd803d4e53ea6ff01b : SUCCESS in 28m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9136dd6677aa4c4d965d244fa28e655f : SUCCESS in 11m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca95c537a9cc4e79977ba0ad18c9029a : SUCCESS in 20m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/53bb9a27513146b09fa931687d332720 : SUCCESS in 1h 04m 37s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a5144cd1077e41a4b1cd1f04ec238aeb : FAILURE in 1h 06m 46s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/87b05104e52e446d9239270145328364 : FAILURE in 55m 37s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b72ecf6bbf8343d68fd3ee6ba523cf97 : FAILURE in 53m 43s (non-voting)\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bd3d2918ca7c4d63b04cde180c8c58c7 : FAILURE in 54m 37s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1a368f85314547b481d320476df74045 : FAILURE in 34m 28s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4fbb9597b289480e8045a93fdf2e4361 : SUCCESS in 1h 05m 48s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6074052fda3a439fb70246f902600f71 : SUCCESS in 1h 12m 01s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":26},{"id":"d0703225141f2ac43797fe710dcec41295bb35c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-10 23:10:32.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2df3de4d1dd34e448d6ecfc27a9b8dc4\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e8355179bfd344c5bbc148db73d515c1 : SUCCESS in 46m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4a4a9c00bcd74898a1310ed2fa567794 : SUCCESS in 1h 10m 44s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0c34caa2e6f645448602613d406bd808 : TIMED_OUT in 1h 21m 27s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"50a8688d115a4d2ae42c6f05b6e4d80aeeba7473","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-08-21 16:50:30.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"88fc8f4a5c9ce77566fc19d1b9a957b06a443825","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 17:29:36.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4826f3ea647f4a10b5ac7b3730967967\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b129030acc414905a6aabf081ba366d1 : SUCCESS in 35m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4e0aa86189a04f6aa7b25dc36c50cecc : SUCCESS in 17m 50s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2196e8c36a374f808fe7ef03ad543ce4 : SUCCESS in 25m 47s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"ca552fa301530b0afaa379b49baabc071dea312d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-08-21 17:40:48.000000000","message":"Patch Set 26:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2026/Arista_Tempest_Testrefs/changes/62/986962/26 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":26},{"id":"53565f0734e3aa7938e83cf0388a9fab2fd84b7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 18:07:39.000000000","message":"Patch Set 26:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/01c9a4009d554dd9b6a4d83a029dec91\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ffea4d393e254244aeaad3b44a0107ab : FAILURE in 39m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0165ccc7b8d74637aae677f1a38d0e22 : SUCCESS in 30m 00s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c6eb79fa230c465b96c7398d272130c0 : FAILURE in 58m 37s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/be65253f43a7413f8f93b015fafe912c : FAILURE in 1h 15m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0bafff09e9184b3ea033fb1a54e747a0 : SUCCESS in 33m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/adbbdff6a23045ed9a58a7c2fb12b47a : SUCCESS in 45m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/973005f1c12a455e901ad7dec8e005b4 : SUCCESS in 10m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5bb16e399621448dadd150ef55355fab : SUCCESS in 20m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e24e8fa196c14eee85c2807caee7c748 : SUCCESS in 17m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8aa15e5e55fd4331a62bc8030588f3a0 : SUCCESS in 35m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efb97cca4ba448bebfff944c719a3777 : SUCCESS in 6m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/28f8f514307240de811cff575aac8e69 : SUCCESS in 9m 31s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/26f505b319ea4615b08a6c596097ffa1 : SUCCESS in 50m 42s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f47cd3f81af543bd92831b0ab11fa775 : FAILURE in 37m 28s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/90ce853f5ae04f18b6d2fd4d195c2590 : FAILURE in 38m 51s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a2fec8a1f0714cd49a7d11790c79fa10 : FAILURE in 29m 42s (non-voting)\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c1381787888f4ff59ed97b058426072a : FAILURE in 1h 02m 09s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8e2652aa2d8f46b08cbd6552879c997c : FAILURE in 24m 31s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ffbc9bc7de294f80b106f37f47c21ab2 : SUCCESS in 30m 03s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/791a8f11f98d449493266a525e418399 : SUCCESS in 1h 06m 36s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":26},{"id":"3dd029b3f9e44a29403305ea88fc55b710604fa9","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-08-21 19:48:02.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"6cb35015051174e1b4e3b691b9509f2d1881c6d2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-10-08 13:54:40.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":26}],"current_revision_number":26,"current_revision":"1ed58982876593077a782b033f5e2f5e98c3eb60","revisions":{"9848606922689f113c2a31c5d626ee86bfab9f9e":{"kind":"REWORK","_number":1,"created":"2026-05-01 21:43:54.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/62/986962/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/986962/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/986962/1"}}},"commit":{"parents":[{"commit":"06d17d4ebb8dbdbe16850bf6e54b60151eb5de6d","subject":"WIP: Stop using neutron-lib project info code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/06d17d4ebb8dbdbe16850bf6e54b60151eb5de6d"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-01 21:38:09.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-01 21:43:28.000000000","tz":-240},"subject":"WIP: Test neutron-lib attribute rename change","message":"WIP: Test neutron-lib attribute rename change\n\nThere is more work required here, just wanted to do\nan initial test using a neutron-lib patch that is\nrenaming all attribute names to project_id from\ntenant_id.\n\nTODO: re-enable neutron-lib attribute calls changed\nin previous patch.\n\nDepends-on: https://review.opendev.org/c/openstack/neutron-lib/+/986960\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\n\nChange-Id: I1a010db313a9796b7fa92faa2075ca752b875883\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9848606922689f113c2a31c5d626ee86bfab9f9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9848606922689f113c2a31c5d626ee86bfab9f9e"}]},"branch":"refs/heads/master"},"984cb113f2b5cd6cf3ed0ec21fdb8927e2f15c71":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-05-01 22:20:12.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/62/986962/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/986962/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/986962/2"}}},"commit":{"parents":[{"commit":"96856a79850478f2c1227502f0c25077a36b548b","subject":"WIP: Stop using neutron-lib project info code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/96856a79850478f2c1227502f0c25077a36b548b"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-01 21:38:09.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-01 22:06:25.000000000","tz":-240},"subject":"WIP: Test neutron-lib attribute rename change","message":"WIP: Test neutron-lib attribute rename change\n\nThere is more work required here, just wanted to do\nan initial test using a neutron-lib patch that is\nrenaming all attribute names to project_id from\ntenant_id.\n\nTODO: re-enable neutron-lib attribute calls changed\nin previous patch.\n\nDepends-on: https://review.opendev.org/c/openstack/neutron-lib/+/986960\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\n\nChange-Id: I1a010db313a9796b7fa92faa2075ca752b875883\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/984cb113f2b5cd6cf3ed0ec21fdb8927e2f15c71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/984cb113f2b5cd6cf3ed0ec21fdb8927e2f15c71"}]},"branch":"refs/heads/master"},"c3fefdc4b85bd5ec03281eb25d447088424d7e57":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-05-02 01:37:34.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/62/986962/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/986962/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/986962/3"}}},"commit":{"parents":[{"commit":"9925d26bc22397cb1e6e87d7417b7f753ca2245a","subject":"Merge \"Revert \"[eventlet-removal] Remove logger mechanism from functional tests\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9925d26bc22397cb1e6e87d7417b7f753ca2245a"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-01 21:38:09.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-02 01:37:34.000000000","tz":0},"subject":"WIP: Test neutron-lib attribute rename change","message":"WIP: Test neutron-lib attribute rename change\n\nThere is more work required here, just wanted to do\nan initial test using a neutron-lib patch that is\nrenaming all attribute names to project_id from\ntenant_id.\n\nTODO: re-enable neutron-lib attribute calls changed\nin previous patch.\n\nDepends-on: https://review.opendev.org/c/openstack/neutron-lib/+/986960\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\n\nChange-Id: I1a010db313a9796b7fa92faa2075ca752b875883\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c3fefdc4b85bd5ec03281eb25d447088424d7e57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c3fefdc4b85bd5ec03281eb25d447088424d7e57"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"984cb113f2b5cd6cf3ed0ec21fdb8927e2f15c71","theirs":"9925d26bc22397cb1e6e87d7417b7f753ca2245a","contains_conflicts":false}},"8225c090d278675ebbfbc3e8bc3976e052486d4b":{"kind":"REWORK","_number":4,"created":"2026-05-02 02:41:21.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/62/986962/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/986962/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/986962/4"}}},"commit":{"parents":[{"commit":"9925d26bc22397cb1e6e87d7417b7f753ca2245a","subject":"Merge \"Revert \"[eventlet-removal] Remove logger mechanism from functional tests\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9925d26bc22397cb1e6e87d7417b7f753ca2245a"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-01 21:38:09.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-02 02:41:03.000000000","tz":-240},"subject":"WIP: Test neutron-lib attribute rename change","message":"WIP: Test neutron-lib attribute rename change\n\nThere is more work required here, just wanted to do\nan initial test using a neutron-lib patch that is\nrenaming all attribute names to project_id from\ntenant_id.\n\nTODO: re-enable neutron-lib attribute calls changed\nin previous patch.\n\nDepends-on: https://review.opendev.org/c/openstack/neutron-lib/+/986960\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\n\nChange-Id: I1a010db313a9796b7fa92faa2075ca752b875883\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8225c090d278675ebbfbc3e8bc3976e052486d4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8225c090d278675ebbfbc3e8bc3976e052486d4b"}]},"branch":"refs/heads/master"},"d7cf320749bdf37094b5c43252afe0d8a156ef85":{"kind":"REWORK","_number":5,"created":"2026-05-02 04:44:25.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/62/986962/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/986962/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/986962/5"}}},"commit":{"parents":[{"commit":"9925d26bc22397cb1e6e87d7417b7f753ca2245a","subject":"Merge \"Revert \"[eventlet-removal] Remove logger mechanism from functional tests\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9925d26bc22397cb1e6e87d7417b7f753ca2245a"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-01 21:38:09.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-02 04:43:38.000000000","tz":-240},"subject":"WIP: Test neutron-lib attribute rename change","message":"WIP: Test neutron-lib attribute rename change\n\nThere is more work required here, just wanted to do\nan initial test using a neutron-lib patch that is\nrenaming all attribute names to project_id from\ntenant_id.\n\nTODO: re-enable neutron-lib attribute calls changed\nin previous patch.\n\nDepends-on: https://review.opendev.org/c/openstack/neutron-lib/+/986960\nDepends-on: https://review.opendev.org/c/openstack/tempest/+/986967\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\n\nChange-Id: I1a010db313a9796b7fa92faa2075ca752b875883\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d7cf320749bdf37094b5c43252afe0d8a156ef85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d7cf320749bdf37094b5c43252afe0d8a156ef85"}]},"branch":"refs/heads/master"},"ca88fe8a9b3f02da05259cee3a73f0764244fb1d":{"kind":"REWORK","_number":6,"created":"2026-05-06 21:41:55.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/62/986962/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/986962/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/986962/6"}}},"commit":{"parents":[{"commit":"9925d26bc22397cb1e6e87d7417b7f753ca2245a","subject":"Merge \"Revert \"[eventlet-removal] Remove logger mechanism from functional tests\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9925d26bc22397cb1e6e87d7417b7f753ca2245a"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-01 21:38:09.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-06 21:41:47.000000000","tz":-240},"subject":"WIP: Test neutron-lib attribute rename change","message":"WIP: Test neutron-lib attribute rename change\n\nThere is more work required here, just wanted to do\nan initial test using a neutron-lib patch that is\nrenaming all attribute names to project_id from\ntenant_id.\n\nTODO: re-enable neutron-lib attribute calls changed\nin previous patch.\n\nDepends-on: https://review.opendev.org/c/openstack/neutron-lib/+/986960\nDepends-on: https://review.opendev.org/c/openstack/tempest/+/986967\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\n\nChange-Id: I1a010db313a9796b7fa92faa2075ca752b875883\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ca88fe8a9b3f02da05259cee3a73f0764244fb1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ca88fe8a9b3f02da05259cee3a73f0764244fb1d"}]},"branch":"refs/heads/master"},"efef488a52556974293d53efea091bd03385ef88":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-05-06 22:57:36.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/62/986962/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/986962/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/986962/7"}}},"commit":{"parents":[{"commit":"563bb8adf22e54ea9c2046e823607ec5915469ee","subject":"Stop using neutron-lib project info code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/563bb8adf22e54ea9c2046e823607ec5915469ee"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-01 21:38:09.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-06 22:57:36.000000000","tz":0},"subject":"WIP: Test neutron-lib attribute rename change","message":"WIP: Test neutron-lib attribute rename change\n\nThere is more work required here, just wanted to do\nan initial test using a neutron-lib patch that is\nrenaming all attribute names to project_id from\ntenant_id.\n\nTODO: re-enable neutron-lib attribute calls changed\nin previous patch.\n\nDepends-on: https://review.opendev.org/c/openstack/neutron-lib/+/986960\nDepends-on: https://review.opendev.org/c/openstack/tempest/+/986967\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\n\nChange-Id: I1a010db313a9796b7fa92faa2075ca752b875883\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/efef488a52556974293d53efea091bd03385ef88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/efef488a52556974293d53efea091bd03385ef88"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"ca88fe8a9b3f02da05259cee3a73f0764244fb1d","theirs":"563bb8adf22e54ea9c2046e823607ec5915469ee","contains_conflicts":false}},"9fac0702888ca57646ddab97d20fe2ec0e7b6be2":{"kind":"REWORK","_number":8,"created":"2026-05-07 13:14:57.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/62/986962/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/986962/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/986962/8"}}},"commit":{"parents":[{"commit":"9925d26bc22397cb1e6e87d7417b7f753ca2245a","subject":"Merge \"Revert \"[eventlet-removal] Remove logger mechanism from functional tests\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9925d26bc22397cb1e6e87d7417b7f753ca2245a"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-01 21:38:09.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-07 13:14:26.000000000","tz":-240},"subject":"WIP: Test neutron-lib attribute rename change","message":"WIP: Test neutron-lib attribute rename change\n\nThere is more work required here, just wanted to do\nan initial test using a neutron-lib patch that is\nrenaming all attribute names to project_id from\ntenant_id.\n\nTODO: re-enable neutron-lib attribute calls changed\nin previous patch.\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\n\nChange-Id: I1a010db313a9796b7fa92faa2075ca752b875883\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9fac0702888ca57646ddab97d20fe2ec0e7b6be2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9fac0702888ca57646ddab97d20fe2ec0e7b6be2"}]},"branch":"refs/heads/master"},"37803cbd3693383e092639fac7fa1af9a462f9f3":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2026-05-07 17:59:08.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/62/986962/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/986962/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/986962/9"}}},"commit":{"parents":[{"commit":"563bb8adf22e54ea9c2046e823607ec5915469ee","subject":"Stop using neutron-lib project info code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/563bb8adf22e54ea9c2046e823607ec5915469ee"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-01 21:38:09.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-07 17:59:08.000000000","tz":0},"subject":"WIP: Test neutron-lib attribute rename change","message":"WIP: Test neutron-lib attribute rename change\n\nThere is more work required here, just wanted to do\nan initial test using a neutron-lib patch that is\nrenaming all attribute names to project_id from\ntenant_id.\n\nTODO: re-enable neutron-lib attribute calls changed\nin previous patch.\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\n\nChange-Id: I1a010db313a9796b7fa92faa2075ca752b875883\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/37803cbd3693383e092639fac7fa1af9a462f9f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/37803cbd3693383e092639fac7fa1af9a462f9f3"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"9fac0702888ca57646ddab97d20fe2ec0e7b6be2","theirs":"563bb8adf22e54ea9c2046e823607ec5915469ee","contains_conflicts":false}},"80e2d139dadf842fc85d539c07f6cf18b690dc62":{"kind":"REWORK","_number":10,"created":"2026-05-07 19:29:01.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/62/986962/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/986962/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/986962/10"}}},"commit":{"parents":[{"commit":"563bb8adf22e54ea9c2046e823607ec5915469ee","subject":"Stop using neutron-lib project info code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/563bb8adf22e54ea9c2046e823607ec5915469ee"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-01 21:38:09.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-07 19:27:44.000000000","tz":-240},"subject":"WIP: Test neutron-lib attribute rename change","message":"WIP: Test neutron-lib attribute rename change\n\nThere is more work required here, just wanted to do\nan initial test using a neutron-lib patch that is\nrenaming all attribute names to project_id from\ntenant_id.\n\nTODO: re-enable neutron-lib attribute calls changed\nin previous patch.\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\n\nDepends-on: https://review.opendev.org/c/openstack/neutron-lib/+/986960\nDepends-on: https://review.opendev.org/c/openstack/tempest/+/986967\n\nChange-Id: I1a010db313a9796b7fa92faa2075ca752b875883\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/80e2d139dadf842fc85d539c07f6cf18b690dc62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/80e2d139dadf842fc85d539c07f6cf18b690dc62"}]},"branch":"refs/heads/master"},"d79cc6b7357475b20946694b01d87c18e4db887d":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2026-05-07 19:31:38.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/62/986962/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/986962/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/986962/11"}}},"commit":{"parents":[{"commit":"563bb8adf22e54ea9c2046e823607ec5915469ee","subject":"Stop using neutron-lib project info code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/563bb8adf22e54ea9c2046e823607ec5915469ee"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-01 21:38:09.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-07 19:31:22.000000000","tz":-240},"subject":"WIP: Test neutron-lib attribute rename change","message":"WIP: Test neutron-lib attribute rename change\n\nThere is more work required here, just wanted to do\nan initial test using a neutron-lib patch that is\nrenaming all attribute names to project_id from\ntenant_id.\n\nTODO: re-enable neutron-lib attribute calls changed\nin previous patch.\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\n\nDepends-on: https://review.opendev.org/c/openstack/tempest/+/986967\n\nChange-Id: I1a010db313a9796b7fa92faa2075ca752b875883\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d79cc6b7357475b20946694b01d87c18e4db887d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d79cc6b7357475b20946694b01d87c18e4db887d"}]},"branch":"refs/heads/master"},"0cd4be39a98739894cd4e4731d1ccddff6b8d414":{"kind":"REWORK","_number":12,"created":"2026-05-07 21:40:14.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/62/986962/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/986962/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/986962/12"}}},"commit":{"parents":[{"commit":"563bb8adf22e54ea9c2046e823607ec5915469ee","subject":"Stop using neutron-lib project info code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/563bb8adf22e54ea9c2046e823607ec5915469ee"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-01 21:38:09.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-07 21:15:16.000000000","tz":-240},"subject":"WIP: Add project_id attributes for neutron-lib change","message":"WIP: Add project_id attributes for neutron-lib change\n\nAdd project_id attributes to RBAC and security group\ncode to help with transition to only using project_id\nattributes in neutron-lib. Should be able to remove\nafter that completes.\n\nFound and fixed a usage of \u0027tenant_id\u0027 in the SG code\nwhile debugging.\n\nTODO: re-enable neutron-lib attribute calls changed\nin previous patch.\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\n\nDepends-on: https://review.opendev.org/c/openstack/tempest/+/986967\n\nChange-Id: I1a010db313a9796b7fa92faa2075ca752b875883\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0cd4be39a98739894cd4e4731d1ccddff6b8d414"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0cd4be39a98739894cd4e4731d1ccddff6b8d414"}]},"branch":"refs/heads/master"},"322ee3de12498653d2f50c659f9bf594f4755d1c":{"kind":"REWORK","_number":13,"created":"2026-05-07 22:45:34.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/62/986962/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/986962/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/986962/13"}}},"commit":{"parents":[{"commit":"563bb8adf22e54ea9c2046e823607ec5915469ee","subject":"Stop using neutron-lib project info code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/563bb8adf22e54ea9c2046e823607ec5915469ee"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-01 21:38:09.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-07 22:44:07.000000000","tz":-240},"subject":"WIP: Add project_id attributes for neutron-lib change","message":"WIP: Add project_id attributes for neutron-lib change\n\nAdd project_id attributes to RBAC and security group\ncode to help with transition to only using project_id\nattributes in neutron-lib. Changed tenant_id to not be\nrequired, should be able to remove completely after\nthis work completes.\n\nFound and fixed a usage of \u0027tenant_id\u0027 in the SG code\nwhile debugging.\n\nTODO: re-enable neutron-lib attribute calls changed\nin previous patch.\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\n\nDepends-on: https://review.opendev.org/c/openstack/tempest/+/986967\n\nChange-Id: I1a010db313a9796b7fa92faa2075ca752b875883\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/322ee3de12498653d2f50c659f9bf594f4755d1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/322ee3de12498653d2f50c659f9bf594f4755d1c"}]},"branch":"refs/heads/master"},"0d7dc8ec48730aa0f44f22a67f3489ec59b90179":{"kind":"REWORK","_number":14,"created":"2026-05-08 02:22:10.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/62/986962/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/986962/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/986962/14"}}},"commit":{"parents":[{"commit":"563bb8adf22e54ea9c2046e823607ec5915469ee","subject":"Stop using neutron-lib project info code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/563bb8adf22e54ea9c2046e823607ec5915469ee"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-01 21:38:09.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-08 02:21:48.000000000","tz":-240},"subject":"WIP: Add project_id attributes for neutron-lib change","message":"WIP: Add project_id attributes for neutron-lib change\n\nAdd project_id attributes to RBAC and security group\ncode to help with transition to only using project_id\nattributes in neutron-lib. Changed tenant_id to not be\nrequired, should be able to remove completely after\nthis work completes.\n\nFound and fixed a usage of \u0027tenant_id\u0027 in the SG code\nwhile debugging.\n\nTODO: re-enable neutron-lib attribute calls changed\nin previous patch.\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\n\nDepends-on: https://review.opendev.org/c/openstack/tempest/+/986967\n\nChange-Id: I1a010db313a9796b7fa92faa2075ca752b875883\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0d7dc8ec48730aa0f44f22a67f3489ec59b90179"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0d7dc8ec48730aa0f44f22a67f3489ec59b90179"}]},"branch":"refs/heads/master"},"3882b52edc70595eb3bf6c8cdbe0321453ed7e5c":{"kind":"REWORK","_number":15,"created":"2026-05-08 04:29:10.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/62/986962/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/986962/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/986962/15"}}},"commit":{"parents":[{"commit":"563bb8adf22e54ea9c2046e823607ec5915469ee","subject":"Stop using neutron-lib project info code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/563bb8adf22e54ea9c2046e823607ec5915469ee"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-01 21:38:09.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-08 04:28:45.000000000","tz":-240},"subject":"WIP: Add project_id attributes for neutron-lib change","message":"WIP: Add project_id attributes for neutron-lib change\n\nAdd project_id attributes to RBAC and security group\ncode to help with transition to only using project_id\nattributes in neutron-lib. Changed tenant_id to not be\nrequired, should be able to remove completely after\nthis work completes.\n\nFound and fixed a usage of \u0027tenant_id\u0027 in the SG code\nwhile debugging.\n\nTODO: re-enable neutron-lib attribute calls changed\nin previous patch.\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\n\nDepends-on: https://review.opendev.org/c/openstack/tempest/+/986967\n\nChange-Id: I1a010db313a9796b7fa92faa2075ca752b875883\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3882b52edc70595eb3bf6c8cdbe0321453ed7e5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3882b52edc70595eb3bf6c8cdbe0321453ed7e5c"}]},"branch":"refs/heads/master"},"d34c59553ffe39409ce5a14c66d4a06386e65204":{"kind":"REWORK","_number":16,"created":"2026-05-08 06:19:07.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/62/986962/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/986962/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/986962/16"}}},"commit":{"parents":[{"commit":"563bb8adf22e54ea9c2046e823607ec5915469ee","subject":"Stop using neutron-lib project info code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/563bb8adf22e54ea9c2046e823607ec5915469ee"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-01 21:38:09.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-08 06:17:10.000000000","tz":-240},"subject":"WIP: Add project_id attributes for neutron-lib change","message":"WIP: Add project_id attributes for neutron-lib change\n\nAdd project_id attributes to RBAC and security group\ncode to help with transition to only using project_id\nattributes in neutron-lib. Changed tenant_id to not be\nrequired, should be able to remove completely after\nthis work completes.\n\nFound and fixed a usage of \u0027tenant_id\u0027 in the SG code\nwhile debugging.\n\nTODO: re-enable neutron-lib attribute calls changed\nin previous patch.\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\n\nDepends-on: https://review.opendev.org/c/openstack/tempest/+/986967\n\nChange-Id: I1a010db313a9796b7fa92faa2075ca752b875883\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d34c59553ffe39409ce5a14c66d4a06386e65204"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d34c59553ffe39409ce5a14c66d4a06386e65204"}]},"branch":"refs/heads/master"},"fef5d81ddbd70471965ea474cc6f20af31faad2c":{"kind":"REWORK","_number":17,"created":"2026-05-08 21:13:55.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/62/986962/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/986962/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/986962/17"}}},"commit":{"parents":[{"commit":"563bb8adf22e54ea9c2046e823607ec5915469ee","subject":"Stop using neutron-lib project info code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/563bb8adf22e54ea9c2046e823607ec5915469ee"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-01 21:38:09.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-08 21:11:58.000000000","tz":-240},"subject":"WIP: Add project_id attributes for neutron-lib change","message":"WIP: Add project_id attributes for neutron-lib change\n\nAdd project_id attributes to RBAC and security group\ncode to help with transition to only using project_id\nattributes in neutron-lib. Changed tenant_id to not be\nrequired, should be able to remove completely after\nthis work completes.\n\nFound and fixed a usage of \u0027tenant_id\u0027 in the SG code\nwhile debugging.\n\nTODO: re-enable neutron-lib attribute calls changed\nin previous patch.\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\n\nDepends-on: https://review.opendev.org/c/openstack/tempest/+/986967\n\nChange-Id: I1a010db313a9796b7fa92faa2075ca752b875883\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fef5d81ddbd70471965ea474cc6f20af31faad2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fef5d81ddbd70471965ea474cc6f20af31faad2c"}]},"branch":"refs/heads/master"},"240b5a1a55f71eefbd86060f28ebb6d7fe47c0fe":{"kind":"REWORK","_number":18,"created":"2026-05-08 22:40:11.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/62/986962/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/986962/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/986962/18"}}},"commit":{"parents":[{"commit":"563bb8adf22e54ea9c2046e823607ec5915469ee","subject":"Stop using neutron-lib project info code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/563bb8adf22e54ea9c2046e823607ec5915469ee"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-01 21:38:09.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-08 22:38:11.000000000","tz":-240},"subject":"WIP: Add project_id attributes for neutron-lib change","message":"WIP: Add project_id attributes for neutron-lib change\n\nAdd project_id attributes to RBAC and security group\ncode to help with transition to only using project_id\nattributes in neutron-lib. Changed tenant_id to not be\nrequired, should be able to remove completely after\nthis work completes.\n\nFound and fixed a usage of \u0027tenant_id\u0027 in the SG code\nwhile debugging.\n\nTODO: re-enable neutron-lib attribute calls changed\nin previous patch.\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\n\nDepends-on: https://review.opendev.org/c/openstack/tempest/+/986967\nDepends-on: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/987959\n\nChange-Id: I1a010db313a9796b7fa92faa2075ca752b875883\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/240b5a1a55f71eefbd86060f28ebb6d7fe47c0fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/240b5a1a55f71eefbd86060f28ebb6d7fe47c0fe"}]},"branch":"refs/heads/master"},"646c3cd6884a438ce669200250c9888bb73110e4":{"kind":"REWORK","_number":19,"created":"2026-06-02 02:02:26.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/62/986962/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/986962/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/986962/19"}}},"commit":{"parents":[{"commit":"563bb8adf22e54ea9c2046e823607ec5915469ee","subject":"Stop using neutron-lib project info code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/563bb8adf22e54ea9c2046e823607ec5915469ee"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-01 21:38:09.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-06-02 02:01:35.000000000","tz":-240},"subject":"WIP: Add project_id attributes for neutron-lib change","message":"WIP: Add project_id attributes for neutron-lib change\n\nAdd project_id attributes to RBAC and security group\ncode to help with transition to only using project_id\nattributes in neutron-lib. Changed tenant_id to not be\nrequired, should be able to remove completely after\nthis work completes.\n\nFound and fixed a usage of \u0027tenant_id\u0027 in the SG code\nwhile debugging.\n\nTODO: re-enable neutron-lib attribute calls changed\nin previous patch.\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\n\nDepends-on: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/987959\n\nChange-Id: I1a010db313a9796b7fa92faa2075ca752b875883\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/646c3cd6884a438ce669200250c9888bb73110e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/646c3cd6884a438ce669200250c9888bb73110e4"}]},"branch":"refs/heads/master"},"c39aa323ff2d0dd4139f35d44f350a89f382229c":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2026-06-03 01:25:20.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/62/986962/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/986962/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/986962/20"}}},"commit":{"parents":[{"commit":"7e735f1db6ceabf17e0f82815cebd0e6b0f5e42a","subject":"Stop using neutron-lib project info code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e735f1db6ceabf17e0f82815cebd0e6b0f5e42a"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-01 21:38:09.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-06-03 01:25:20.000000000","tz":0},"subject":"WIP: Add project_id attributes for neutron-lib change","message":"WIP: Add project_id attributes for neutron-lib change\n\nAdd project_id attributes to RBAC and security group\ncode to help with transition to only using project_id\nattributes in neutron-lib. Changed tenant_id to not be\nrequired, should be able to remove completely after\nthis work completes.\n\nFound and fixed a usage of \u0027tenant_id\u0027 in the SG code\nwhile debugging.\n\nTODO: re-enable neutron-lib attribute calls changed\nin previous patch.\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\n\nDepends-on: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/987959\n\nChange-Id: I1a010db313a9796b7fa92faa2075ca752b875883\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c39aa323ff2d0dd4139f35d44f350a89f382229c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c39aa323ff2d0dd4139f35d44f350a89f382229c"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"646c3cd6884a438ce669200250c9888bb73110e4","theirs":"7e735f1db6ceabf17e0f82815cebd0e6b0f5e42a","contains_conflicts":false}},"29255db1965f58b38bfc688e53ee946999e2ab35":{"kind":"REWORK","_number":21,"created":"2026-06-03 01:35:41.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/62/986962/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/986962/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/986962/21"}}},"commit":{"parents":[{"commit":"7e735f1db6ceabf17e0f82815cebd0e6b0f5e42a","subject":"Stop using neutron-lib project info code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e735f1db6ceabf17e0f82815cebd0e6b0f5e42a"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-01 21:38:09.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-06-03 01:30:40.000000000","tz":-240},"subject":"WIP: Add project_id attribute for neutron-lib change","message":"WIP: Add project_id attribute for neutron-lib change\n\nAdd project_id attribute to RBAC, security group and\nsecurity group default rule code, removing tenant_id,\nto help with transition to only using project_id\nattributes in neutron-lib.\n\nFound and fixed a usage of \u0027tenant_id\u0027 in the SG code\nwhile debugging.\n\nTODO: re-enable neutron-lib attribute calls changed\nin previous patch.\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\n\nDepends-on: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/987959\n\nChange-Id: I1a010db313a9796b7fa92faa2075ca752b875883\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/29255db1965f58b38bfc688e53ee946999e2ab35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/29255db1965f58b38bfc688e53ee946999e2ab35"}]},"branch":"refs/heads/master"},"77f18c34bdc9f4f6e8ad9af5d96dea16f35d7a3b":{"kind":"NO_CODE_CHANGE","_number":22,"created":"2026-06-05 03:37:34.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/62/986962/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/986962/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/986962/22"}}},"commit":{"parents":[{"commit":"7e735f1db6ceabf17e0f82815cebd0e6b0f5e42a","subject":"Stop using neutron-lib project info code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e735f1db6ceabf17e0f82815cebd0e6b0f5e42a"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-01 21:38:09.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-06-05 03:36:55.000000000","tz":-240},"subject":"WIP: Add project_id attribute for neutron-lib change","message":"WIP: Add project_id attribute for neutron-lib change\n\nAdd project_id attribute to RBAC, security group and\nsecurity group default rule code, removing tenant_id,\nto help with transition to only using project_id\nattributes in neutron-lib.\n\nFound and fixed a usage of \u0027tenant_id\u0027 in the SG code\nwhile debugging.\n\nTODO: re-enable neutron-lib attribute calls changed\nin previous patch.\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\n\nDepends-on: https://review.opendev.org/c/openstack/tempest/+/991754\nDepends-on: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/987959\n\nChange-Id: I1a010db313a9796b7fa92faa2075ca752b875883\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/77f18c34bdc9f4f6e8ad9af5d96dea16f35d7a3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/77f18c34bdc9f4f6e8ad9af5d96dea16f35d7a3b"}]},"branch":"refs/heads/master"},"3a8af90edb5fd1979487d6f443c175afe67ae8b1":{"kind":"NO_CODE_CHANGE","_number":23,"created":"2026-06-05 21:19:45.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/62/986962/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/986962/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/986962/23"}}},"commit":{"parents":[{"commit":"7e735f1db6ceabf17e0f82815cebd0e6b0f5e42a","subject":"Stop using neutron-lib project info code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e735f1db6ceabf17e0f82815cebd0e6b0f5e42a"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-01 21:38:09.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-06-05 21:13:50.000000000","tz":-240},"subject":"WIP: Add project_id attribute for neutron-lib change","message":"WIP: Add project_id attribute for neutron-lib change\n\nAdd project_id attribute to RBAC, security group and\nsecurity group default rule code, removing tenant_id,\nto help with transition to only using project_id\nattributes in neutron-lib.\n\nFound and fixed a usage of \u0027tenant_id\u0027 in the SG code\nwhile debugging.\n\nTODO: re-enable neutron-lib attribute calls changed\nin previous patch.\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\n\nDepends-on: https://review.opendev.org/c/openstack/tempest/+/991754\nDepends-on: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/987959\nDepends-on: https://review.opendev.org/c/openstack/nova/+/928022\n\nChange-Id: I1a010db313a9796b7fa92faa2075ca752b875883\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3a8af90edb5fd1979487d6f443c175afe67ae8b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3a8af90edb5fd1979487d6f443c175afe67ae8b1"}]},"branch":"refs/heads/master"},"b44a7fbc277d00269e8c2292d71775c399ce0d3f":{"kind":"NO_CODE_CHANGE","_number":24,"created":"2026-06-05 22:29:37.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/62/986962/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/986962/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/986962/24"}}},"commit":{"parents":[{"commit":"7e735f1db6ceabf17e0f82815cebd0e6b0f5e42a","subject":"Stop using neutron-lib project info code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e735f1db6ceabf17e0f82815cebd0e6b0f5e42a"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-01 21:38:09.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-06-05 22:28:57.000000000","tz":-240},"subject":"WIP: Add project_id attribute for neutron-lib change","message":"WIP: Add project_id attribute for neutron-lib change\n\nAdd project_id attribute to RBAC, security group and\nsecurity group default rule code, removing tenant_id,\nto help with transition to only using project_id\nattributes in neutron-lib.\n\nFound and fixed a usage of \u0027tenant_id\u0027 in the SG code\nwhile debugging.\n\nTODO: re-enable neutron-lib attribute calls changed\nin previous patch.\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\n\nDepends-on: https://review.opendev.org/c/openstack/tempest/+/991754\nDepends-on: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/987959\nDepends-on: https://review.opendev.org/c/openstack/nova/+/991948\n\nChange-Id: I1a010db313a9796b7fa92faa2075ca752b875883\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b44a7fbc277d00269e8c2292d71775c399ce0d3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b44a7fbc277d00269e8c2292d71775c399ce0d3f"}]},"branch":"refs/heads/master"},"bd5bf0da4837e21f9b1ab765e665cd87298ab098":{"kind":"NO_CODE_CHANGE","_number":25,"created":"2026-06-09 22:25:35.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/62/986962/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/986962/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/986962/25"}}},"commit":{"parents":[{"commit":"7e735f1db6ceabf17e0f82815cebd0e6b0f5e42a","subject":"Stop using neutron-lib project info code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e735f1db6ceabf17e0f82815cebd0e6b0f5e42a"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-01 21:38:09.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-06-09 20:45:58.000000000","tz":-240},"subject":"WIP: Add project_id attribute for neutron-lib change","message":"WIP: Add project_id attribute for neutron-lib change\n\nAdd project_id attribute to RBAC, security group and\nsecurity group default rule code, removing tenant_id,\nto help with transition to only using project_id\nattributes in neutron-lib.\n\nFound and fixed a usage of \u0027tenant_id\u0027 in the SG code\nwhile debugging.\n\nTODO: re-enable neutron-lib attribute calls changed\nin previous patch.\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\n\nDepends-on: https://review.opendev.org/c/openstack/tempest/+/991754\nDepends-on: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/987959\nDepends-on: https://review.opendev.org/c/openstack/nova/+/991948\nDepends-on: https://review.opendev.org/c/openstack/openstacksdk/+/992536\n\nChange-Id: I1a010db313a9796b7fa92faa2075ca752b875883\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bd5bf0da4837e21f9b1ab765e665cd87298ab098"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bd5bf0da4837e21f9b1ab765e665cd87298ab098"}]},"branch":"refs/heads/master"},"1ed58982876593077a782b033f5e2f5e98c3eb60":{"kind":"NO_CODE_CHANGE","_number":26,"created":"2026-06-22 15:45:00.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/62/986962/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/986962/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/986962/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/986962/26"}}},"commit":{"parents":[{"commit":"7e735f1db6ceabf17e0f82815cebd0e6b0f5e42a","subject":"Stop using neutron-lib project info code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e735f1db6ceabf17e0f82815cebd0e6b0f5e42a"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-01 21:38:09.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-06-22 15:43:18.000000000","tz":-240},"subject":"WIP: Add project_id attribute for neutron-lib change","message":"WIP: Add project_id attribute for neutron-lib change\n\nAdd project_id attribute to RBAC, security group and\nsecurity group default rule code, removing tenant_id,\nto help with transition to only using project_id\nattributes in neutron-lib.\n\nFound and fixed a usage of \u0027tenant_id\u0027 in the SG code\nwhile debugging.\n\nTODO: re-enable neutron-lib attribute calls changed\nin previous patch.\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\n\nDepends-on: https://review.opendev.org/c/openstack/tempest/+/991754\nDepends-on: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/987959\nDepends-on: https://review.opendev.org/c/openstack/nova/+/994321\nDepends-on: https://review.opendev.org/c/openstack/openstacksdk/+/992536\n\nChange-Id: I1a010db313a9796b7fa92faa2075ca752b875883\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1ed58982876593077a782b033f5e2f5e98c3eb60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1ed58982876593077a782b033f5e2f5e98c3eb60"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
