)]}'
{"id":"openstack%2Fneutron~987007","triplet_id":"openstack%2Fneutron~master~I25cd93d2382ada6bc8728ea7fd1bd8f0fcc6e1fb","project":"openstack/neutron","branch":"master","topic":"neutronclient_test-requirements","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-07-07 21:14:34.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-07-07 21:14:34.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-07-07 19:05:10.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I25cd93d2382ada6bc8728ea7fd1bd8f0fcc6e1fb","subject":"Move neutronclient to test requirements","status":"MERGED","created":"2026-05-02 15:24:41.000000000","updated":"2026-07-07 21:17:10.000000000","submitted":"2026-07-07 21:14:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"987007-neutronclient_test-requirements","meta_rev_id":"fe136f7ea4c191751acc9dead5abdf7f79a03a7e","_number":987007,"virtual_id_number":987007,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-07 21:14:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-07-01 10:35:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-07-07 19:05:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-07-07 19:05:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-07-01 10:35:28.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-02 22:27:20.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-05-03 04:20:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-03 04:53:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-01 10:35:28.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"e655aeab3d5920501d86972c58448fffaa81ace0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-02 15:24:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ce5fd46c57898a5d837335e4bc02f5ae71c10b9","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-02 22:27:20.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"519b73090a07076be2eed1a4d77f9bcc5dee8734","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-03 02:09:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e86e2330d322bf51b1af2437b0d629c9df270794","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-03 04:20:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0da9b5f3ca074ff3823b3364345aa101\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c1b4a08ff7ec44d4a9e6d425c682011b : SUCCESS in 43m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cf8676f80d404489819ea870e67a4317 : SUCCESS in 21m 07s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d3d58543b4554208b9afd3dc39599d35 : SUCCESS in 24m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2155f2d59375d6a8e796d7f00fd7b309f6856e1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-03 04:53:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/698efef774004ca9bdbfba94446893c2\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3848fff9751842059895baa868d04ccb : SUCCESS in 1h 37m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/470e73b4071b4eee86d4af82cb652325 : SUCCESS in 33m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb230147323a40c8b907d7a350772ccb : SUCCESS in 1h 09m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a40d3f2fd2b4d34a8468945ded0419a : FAILURE in 8m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/28c8cc5d4bec4c4ba9430ebc62cec81e : SUCCESS in 27m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b6f1136c25194dacbaf5a1c2b1bc54b3 : SUCCESS in 27m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4c0b8ba6d6784c148f8cde42af0261cd : SUCCESS in 21m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8d08d0d3fa30486da2aedb25f6bd8b32 : SUCCESS in 39m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6222a0f38473468eac0c6e7489cec350 : SUCCESS in 12m 31s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0e2cccebaea4455f89a2badd7e7d1c09 : SUCCESS in 2m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c60f729af9694b528be1038dafa4f777 : SUCCESS in 10m 08s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/dce3449adabb4504b6a9833dc8af4538 : SUCCESS in 1h 30m 29s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c67f0c7f51cc427e8f93cf198f1c3a3e : SUCCESS in 1h 02m 56s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/7412b5ee20ed41d9871853b759b82daa : SUCCESS in 1h 01m 45s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/54efc259fcce4ab5952ffe47f11b288c : SUCCESS in 2h 02m 09s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/26fec9f4b4b24720b3c2e912b48592e3 : SUCCESS in 1h 38m 14s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e90f29f22e6a4c6083eeabd06c059cdd : SUCCESS in 1h 03m 26s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/77f4a46e3a554263b1d84f3e537abde8 : SUCCESS in 1h 00m 03s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/699a4d0ca28d4cd58e9542be193a0c0b : SUCCESS in 1h 05m 18s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/dvr_local_router.py","accounts_in_message":[],"_revision_number":1},{"id":"d0aa276c09c34a63744e2547d396381a8d2da2bd","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-03 17:52:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3dff5be4fa66c00cf05326d6c2c6304b92793f15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-04 16:34:59.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"b206ff6a684c7363d6f045aeb2dbf94157dd93b8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 17:14:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8a0f41804994116b90373461bbccb05\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e0f60e3cfd9f4b15897eed4524190efb : SUCCESS in 24m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/09a3fb13fa6347eb9310510ee10ba3e6 : SUCCESS in 22m 28s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4d5034e1dd5040bb86781852c1608caf : SUCCESS in 30m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fafaffb45bddb6d1d237c408de73fa65323c3ef8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 18:44:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0875f187f9334fd891bf5ddcbba092f6\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/56b5e604b5494ca8b770d1034991f992 : SUCCESS in 1h 07m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a49d6530096e47efaae90719a368818d : SUCCESS in 55m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00e5ab5c1d8e44db9cbbf3840f437a69 : SUCCESS in 1h 13m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e863932e15d24382907c6f40cb560b00 : SUCCESS in 11m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1f09b103f1864dfb97bf8f20ece0dd88 : SUCCESS in 43m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ba3df207813348ed95ca699f652a3214 : SUCCESS in 32m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/295cc6b6a931411daa536870a35e70a6 : SUCCESS in 30m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f53ce2e9e6ff465abd1df9330e25c69c : SUCCESS in 23m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6897456237eb412488a6af824adb6753 : SUCCESS in 13m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9856449f33e242f385576169c343940a : SUCCESS in 3m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/551358e349ca4089848a2d795e98f0f0 : SUCCESS in 17m 50s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5998f874e00647f1932290fc3b246f71 : FAILURE in 53m 53s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6d7ed3204dde42d68169438271dd7ee3 : SUCCESS in 1h 00m 15s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/28104aa92866439da4dd00b1ac301fce : SUCCESS in 1h 20m 27s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4e1cb72940f3491f85f63ebf44844d3b : SUCCESS in 1h 58m 37s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fb8be7ae05e9473bbf7a493dae6c7e8c : SUCCESS in 1h 46m 47s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/63b1629a0a7c4367ac4cc6b5609844a8 : SUCCESS in 1h 06m 10s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/9c78d48c20e242bc88f56698790e443b : SUCCESS in 1h 01m 16s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/610a19ac8d4b4a55bb2d3a38f67269d0 : SUCCESS in 1h 10m 24s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":2},{"id":"e2df8c7e9e1c732d046a4a7de9423ff4a335959b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-04 20:52:20.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dc6fa404285f38f0335d3d4fbe29e390e46aebe6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 21:22:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94a13637fad1436abf899233c8acba75\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4860d316aa25441a948abe20b6495ea9 : SUCCESS in 27m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5b915661f05a4968aca7833638e5b395 : SUCCESS in 19m 46s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f4a6ba4f770447c79dd8ac2fa78af140 : SUCCESS in 26m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d53fb2e9193693eacef138bdebe6ac49faed35e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 23:05:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e165c37e988d4d89b466a03623cb64da\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0a242eb874f54c1ab545f42d38bcc2f5 : SUCCESS in 1h 04m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/503f83f539224fb6a8976146de2f5248 : SUCCESS in 33m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aad0e2ef9e964712a6da14afa76802fb : SUCCESS in 42m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2073220fe4cc482ba28d212beb6ff707 : SUCCESS in 10m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fba9a3fb861041bebd87c24c63a2b537 : SUCCESS in 35m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/95b52831d7f04ce5ad1991cae6182cf2 : SUCCESS in 18m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a1aba2ee7b8a4c9fb580d53be76672bf : SUCCESS in 23m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/62a6eea7840342fc91d1dbc0ab527b7f : SUCCESS in 32m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e9b98845d064dd9a8c6c01cb47cdfbb : SUCCESS in 6m 09s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c4d7351747e64fac9687d1f59ed94449 : SUCCESS in 1m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9cf91c6fe59e46bcb95e12376a17ce18 : SUCCESS in 21m 19s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/16363dfd201642ce874350e33f87cb3c : SUCCESS in 44m 18s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e42bb0b49c074b4a9f0e9b069a3d16f6 : SUCCESS in 41m 03s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/0f25e8759df14b1bb348c7679b0a6fe2 : SUCCESS in 39m 21s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9c247b22bb7b426ea7c4768567155d22 : SUCCESS in 2h 06m 43s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9ac42065e20c40708c503fc3238ec60e : SUCCESS in 1h 07m 31s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4c8d6ed2b701424093dcc5307357acf2 : SUCCESS in 57m 47s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7cf80612681d4b24851757e9db938a67 : SUCCESS in 1h 14m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c8939261e76b4c098631ab7f5b40f132 : SUCCESS in 38m 02s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":2},{"id":"74710fd2c2f5287190f3d722d51a5a3f110a8897","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-07-01 10:35:16.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"bf21b900df67110c538051ed40313c182aca25e2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-07-01 10:35:28.000000000","message":"Patch Set 3: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"ae31195afc526bd2e0f97713c562f41cc9b3ca79","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-01 11:33:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df96d96d39794081904a4d8b33d57278\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f093b693a3a74cee8e613b7453077ebb : SUCCESS in 21m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9e4565b12943460ba141aa76586bfdb0 : SUCCESS in 39m 52s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ca7fbfc5d8b549f3a0b9d87a260a8044 : SUCCESS in 56m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"33d11aeba54356a74eb658934ec7d929f195d5aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-01 12:33:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c72de090d94f4a56afcee6be68017f6c\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/aced5b9e9acf42c7a55e887b71a78d01 : SUCCESS in 1h 33m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/47eb3b86f01d449fbaf4181db5c0cad6 : SUCCESS in 48m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c71e5f085db4a078c18bd4724846626 : SUCCESS in 39m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cc32171aeaf4a4e87b736dfdb20e6bb : SUCCESS in 10m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0c744dae8637414595856bee9f505a92 : SUCCESS in 29m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0b1fc76148a544b5a910829995280d44 : SUCCESS in 19m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/51a1205510ad4224bbddcf2c467f840f : SUCCESS in 36m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9d0aaec57a346e3a159524abf46ceca : SUCCESS in 13m 09s\n- requirements-check https://zuul.opendev.org/t/openstack/build/90ec0017c5094c2cb65e84252f9dad14 : SUCCESS in 3m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1f6188b07bc4a4f90a13f2eb0fb96ad : SUCCESS in 15m 42s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1b2c3236357e4fd48ca725fe50a7b860 : SUCCESS in 47m 21s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4ba52b611f214f69bdfd8dc1c5c32f95 : SUCCESS in 1h 09m 07s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/99344352b7754bcb9f575e1bf6dd25ee : SUCCESS in 1h 15m 32s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9c6ef08d88644d2484ca2e354cd1795f : SUCCESS in 1h 50m 45s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bc787012e79748deb8c0767b4e5d088b : SUCCESS in 1h 51m 42s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c7b0fdadd662442fb0ba4599ecd73511 : SUCCESS in 55m 30s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d9d9ecf4b4354c62bf3f5e80718f794b : SUCCESS in 1h 15m 54s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1f6ddb9a5e8d436383cb7014c167d310 : SUCCESS in 1h 21m 17s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":3},{"id":"7ff5bda44762b36e0b063b10d4a8748f15fa3fac","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-07-07 19:05:10.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"694e352530525d2c06622eb5e6c4df4744ccfee2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 19:05:58.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e8fd761c679325f0e51a0a91f554a1d5be4bf7b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 21:14:33.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d4410476a4d4f05b8057fdeb8d9791e\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/26940ac76e4044319c90d6b79c033659 : SUCCESS in 1h 38m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/24d4cdd692d84b91b28167ce405a8fd7 : SUCCESS in 37m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/067b850627894637b6e0143e392cdf5f : SUCCESS in 7m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/32c4ec8cc7fd4f5db3a8b6b95a8aa2c4 : SUCCESS in 25m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/24eddf9d21a54b068964cf0c70f0950c : SUCCESS in 26m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5eff6ff11b554d248ba73e219b35cf12 : SUCCESS in 11m 58s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2ae4c3e3135a46e981eee55c722fa785 : SUCCESS in 3m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5908effa57a0465bb4cedc23d7fd43f0 : SUCCESS in 21m 28s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5ad258c2b9c14d6d8706a63bfa377bbc : SUCCESS in 58m 14s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9abd2fd199e94d388e547651df7a33ad : SUCCESS in 1h 10m 09s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/c0f66cae609e4c6d84fd865176643555 : SUCCESS in 1h 13m 00s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9285a806bb8a4d37be289ef3cb5849a5 : SUCCESS in 1h 10m 19s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/793682dfde84424aa216c9b85b6bb81e : SUCCESS in 1h 02m 34s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":3},{"id":"2c73bd999e846805ebf3727aa2eda76d84cf325a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 21:14:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"fe136f7ea4c191751acc9dead5abdf7f79a03a7e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 21:17:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cce2ab8bc9404f6eb30c7d93055bf132\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8850a8eb34a24e1193c8743f7810a8d1 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b7826ecebb448a89fbd18024467dbf4 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"957e4f382b8bdc75242b1e856562df54fa4569bf","revisions":{"3d480d3664e90bd2cffd0c9c7b9dcb50d0d567df":{"kind":"REWORK","_number":1,"created":"2026-05-02 15:24:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/07/987007/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/07/987007/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/07/987007/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/07/987007/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/07/987007/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/07/987007/1"}}},"commit":{"parents":[{"commit":"9925d26bc22397cb1e6e87d7417b7f753ca2245a","subject":"Merge \"Revert \"[eventlet-removal] Remove logger mechanism from functional tests\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9925d26bc22397cb1e6e87d7417b7f753ca2245a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-02 15:23:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-02 15:24:29.000000000","tz":540},"subject":"Move neutronclient to test requirements","message":"Move neutronclient to test requirements\n\nIt\u0027s now used only in fullstack tests.\n\nChange-Id: I25cd93d2382ada6bc8728ea7fd1bd8f0fcc6e1fb\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3d480d3664e90bd2cffd0c9c7b9dcb50d0d567df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3d480d3664e90bd2cffd0c9c7b9dcb50d0d567df"}]},"branch":"refs/heads/master"},"3ac2186a6b8bc2dc613f3ddea7bdc9ed9c77a9b5":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-05-04 16:34:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/07/987007/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/07/987007/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/07/987007/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/07/987007/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/07/987007/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/07/987007/2"}}},"commit":{"parents":[{"commit":"6149ed656db8e8b4c181bd734c47537e9e180b92","subject":"Merge \"Fix pylint invalid-sequence-index false positive for internal_ports (2)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6149ed656db8e8b4c181bd734c47537e9e180b92"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-02 15:23:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-04 16:34:59.000000000","tz":0},"subject":"Move neutronclient to test requirements","message":"Move neutronclient to test requirements\n\nIt\u0027s now used only in fullstack tests.\n\nChange-Id: I25cd93d2382ada6bc8728ea7fd1bd8f0fcc6e1fb\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3ac2186a6b8bc2dc613f3ddea7bdc9ed9c77a9b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3ac2186a6b8bc2dc613f3ddea7bdc9ed9c77a9b5"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"3d480d3664e90bd2cffd0c9c7b9dcb50d0d567df","theirs":"6149ed656db8e8b4c181bd734c47537e9e180b92","contains_conflicts":false}},"957e4f382b8bdc75242b1e856562df54fa4569bf":{"kind":"REWORK","_number":3,"created":"2026-07-01 10:35:16.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/07/987007/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/07/987007/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/07/987007/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/07/987007/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/07/987007/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/07/987007/3"}}},"commit":{"parents":[{"commit":"ac4c89bf4339bf92431a7039a803856317aa0558","subject":"Merge \"Fix error handling in PVLAN\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ac4c89bf4339bf92431a7039a803856317aa0558"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-02 15:23:40.000000000","tz":540},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-07-01 10:34:51.000000000","tz":120},"subject":"Move neutronclient to test requirements","message":"Move neutronclient to test requirements\n\nIt\u0027s now used only in fullstack tests.\n\nChange-Id: I25cd93d2382ada6bc8728ea7fd1bd8f0fcc6e1fb\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/957e4f382b8bdc75242b1e856562df54fa4569bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/957e4f382b8bdc75242b1e856562df54fa4569bf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
