)]}'
{"id":"openstack%2Fneutron~987738","triplet_id":"openstack%2Fneutron~master~I343b9773c949ece56f0fd8ec06c490d54f536dc8","project":"openstack/neutron","branch":"master","hashtags":[],"change_id":"I343b9773c949ece56f0fd8ec06c490d54f536dc8","subject":"DNM: Testing n-t-p","status":"NEW","created":"2026-05-07 18:03:51.000000000","updated":"2026-05-07 20:06:42.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0be25882e1ca5baf3510b87fdb138c6fc7c8d249","_number":987738,"virtual_id_number":987738,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-05-07 20:06:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-07 18:43:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-07 20:06:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"17ad0cd968384e41bbceb65ed4f11f0e6c865ab2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-07 18:03:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d820014bea8760f734375624294980090fd378e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 18:43:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/220c9bf09afe41dc8d425a4f7dec0671\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5f44e27954ba4c8986c217783351c4dd : SUCCESS in 28m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7a0d82e296a14f65a04da0cf84051312 : SUCCESS in 19m 43s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/75d54d8d8970461ba5f7b466d1aa9181 : SUCCESS in 37m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0be25882e1ca5baf3510b87fdb138c6fc7c8d249","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 20:06:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5bcf6e49b974b629a01c4ebff195f50\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3c480e6378244d6598e5dbbfa3d66958 : SUCCESS in 1h 24m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7341b31a3ada4a5a824171ec581a1f5c : SUCCESS in 1h 25m 33s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7bbd378c81da428ba77dc46a149ce7df : SUCCESS in 1h 28m 06s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/56839414559c4f98a12fc9da0cea95ed : SUCCESS in 1h 39m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a4a235fe941a4ff0aca0faf9134fc27e : SUCCESS in 33m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba9b986527ad49f190d15b03aee4110d : SUCCESS in 44m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/383385792a42433a89597892884c0363 : SUCCESS in 8m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dc9ce520c237414188b85c6d29aa3874 : SUCCESS in 38m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b0c7622912ea42b1beae173e6b098f6e : SUCCESS in 26m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0e15840d310b4cd2b5f424244e030efe : SUCCESS in 26m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/79c35994c7e441c09e1d545ffd728a24 : SUCCESS in 40m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec42ebe5d601495d8dcd6a57051715ea : SUCCESS in 11m 19s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d7dbbee8b204424aa58b262a78239549 : SUCCESS in 1h 21m 25s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b02be88f7ae74261a3629c6298e16522 : SUCCESS in 1h 10m 56s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/0f42ea1fe4e541fe9d430f9293ec3b98 : SUCCESS in 1h 16m 19s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1d731110288942b2a70fe43803b25732 : SUCCESS in 1h 26m 06s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/865029ffbedb4fa885c97be4f207840f : SUCCESS in 58m 03s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/189a204789b14e0c8ba0d4b97629abeb : SUCCESS in 1h 02m 39s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8f95b75b958449d7abe4d49ab4057454 : SUCCESS in 1h 14m 04s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/39fd61c2e6264936860e821c1624a708 : SUCCESS in 1h 32m 26s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/53456df7aa62448a81c934af63c39418 : SUCCESS in 1h 59m 11s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"39329f645bda3b13f116bcee11ce362afa584b3b","revisions":{"39329f645bda3b13f116bcee11ce362afa584b3b":{"kind":"REWORK","_number":1,"created":"2026-05-07 18:03:51.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/38/987738/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/38/987738/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/38/987738/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/38/987738/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/38/987738/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/38/987738/1"}}},"commit":{"parents":[{"commit":"387f7a3fe64c17797500d44ccfd36111a132ef42","subject":"Merge \"Fix deletion of floating IPs with managed DNS records\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/387f7a3fe64c17797500d44ccfd36111a132ef42"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-07 18:02:25.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-07 18:02:25.000000000","tz":-240},"subject":"DNM: Testing n-t-p","message":"DNM: Testing n-t-p\n\nEven though I have another patch depending on it, try\none with no other changes to make sure not to break\nmaster branch.\n\nDepends-on: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/982736\n\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nChange-Id: I343b9773c949ece56f0fd8ec06c490d54f536dc8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/39329f645bda3b13f116bcee11ce362afa584b3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/39329f645bda3b13f116bcee11ce362afa584b3b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
