)]}'
{"id":"openstack%2Fneutron~988096","triplet_id":"openstack%2Fneutron~master~Ic9e0927d33c933e47cff99321d5eb5157075827a","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-05-14 02:44:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic9e0927d33c933e47cff99321d5eb5157075827a","subject":"Skip functional tests for git config update","status":"MERGED","created":"2026-05-11 11:49:56.000000000","updated":"2026-05-14 02:48:24.000000000","submitted":"2026-05-14 02:44:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"988096","meta_rev_id":"6466adbbcc813194e409e0f21fd877c4bb661ab9","_number":988096,"virtual_id_number":988096,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-14 02:44:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2026-05-12 07:37:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2026-05-14 00:47:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2026-05-14 00:47:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-11 12:24:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-11 12:49:20.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2026-05-11 14:04:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-12 07:37:33.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-05-14 00:47:03.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"e9bc825c80eeffc24ba1eafa43651d33da11598d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-11 11:49:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"073e7169059df06c061d670166919b29decfa696","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 12:24:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46c24bcc4469495ba384bd8106f9fbe1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/110177f1dec447c6af911c4c85a37931 : SUCCESS in 22m 46s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/893dc7eb1fc246ef97c80df6a638388b : SUCCESS in 16m 57s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1338e004e2614e079a3531fd7fa587d7 : SUCCESS in 32m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dc0921fd6584530685391c0a1e97502ab7d57494","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 14:04:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f6b2bbfd44741a181e32dcce47c2692\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/65d0e775656248a5ae656244b52839eb : SUCCESS in 33m 14s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a759c41490724d469f23dd80057d936e : SUCCESS in 1h 29m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/21183991057549db8fb5a1add4cd266a : SUCCESS in 2h 04m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1e18c97ea569435f99b1aa9450f227c4 : SUCCESS in 32m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/409aa31fdf8f4389849fe8b7dfc2adae : SUCCESS in 1h 09m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37fb86c304464e8482d917096450fa85 : SUCCESS in 6m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/71b2b9d14aed4ad89f1a53638065b042 : SUCCESS in 30m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/52181c9cc25b45329627f93d4a74d43c : SUCCESS in 25m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d31dfca4c925429694ff2a5589fd61cc : SUCCESS in 16m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/38dbcb7ce7b24b1f97ab69c319d141e8 : SUCCESS in 20m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e50f2a93c14451dba35e3b977ce201b : SUCCESS in 12m 10s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/52bcb0619d4544a094f6797da732f587 : SUCCESS in 2h 08m 55s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":1},{"id":"50762ff88f76c4e3c7a9464fab1fb151829dc6d3","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-05-12 07:37:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b0cfab5d30077cde82bcb50675460f0cf964791a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-14 00:47:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"447d54380e7d828336e605059600c046b67c4000","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 00:47:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a55c942ccd48fb4b3dbb80a7d1cfb3faf6aa0652","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 02:44:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5eee766a9a12415196ce7d4d0b52caff\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/17936f7a8e2143eabb30f40816439c74 : SUCCESS in 44m 59s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/00763807760c43b9a753fd977890509e : SUCCESS in 1h 31m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4bd880a172ec4c88a2e9c3414f176776 : SUCCESS in 1h 56m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45db1d60d02c4954ba466e822b46f0ec : SUCCESS in 11m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/02826f38d2f648ccb71bdaf40312e49f : SUCCESS in 22m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/63df6d86f1db4d21b836d0666315d839 : SUCCESS in 16m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c448a368e05486d93d507a5f5e03c8c : SUCCESS in 11m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/49ad2374f7984f0e9186cf0226fc6aa1 : SUCCESS in 8m 48s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":1},{"id":"7a0234b5deea8d3220bc98d80d4b470a191425db","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 02:44:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6466adbbcc813194e409e0f21fd877c4bb661ab9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 02:48:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e779d47689b4cc9a956514708c8a113\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cbc084f4d7244c8b7dfc6de56b74125 : SUCCESS in 1m 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4bb139eb2457479b9bc4b9690a1d5e8d4549c56e","revisions":{"4bb139eb2457479b9bc4b9690a1d5e8d4549c56e":{"kind":"REWORK","_number":1,"created":"2026-05-11 11:49:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/988096/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/96/988096/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/988096/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/988096/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/988096/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/96/988096/1"}}},"commit":{"parents":[{"commit":"8c322ca73ea593e6e8eb75bf88a3b15ba13f60ee","subject":"Merge \"Add --port option to remove-duplicated-port-bindings script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c322ca73ea593e6e8eb75bf88a3b15ba13f60ee"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-11 11:49:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-11 11:49:28.000000000","tz":540},"subject":"Skip functional tests for git config update","message":"Skip functional tests for git config update\n\n.gitignore and .gitreview have no functional impact by themselves.\n\nChange-Id: Ic9e0927d33c933e47cff99321d5eb5157075827a\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4bb139eb2457479b9bc4b9690a1d5e8d4549c56e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4bb139eb2457479b9bc4b9690a1d5e8d4549c56e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
