)]}'
{"id":"openstack%2Fneutron~990002","triplet_id":"openstack%2Fneutron~master~Ia9d319fbbf8047e756da555db3f52a3f9a704392","project":"openstack/neutron","branch":"master","topic":"bug/2154192","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-05-28 18:29:57.000000000","reason":"Change was submitted"},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2026-05-28 18:29:57.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-05-28 16:28:04.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"Ia9d319fbbf8047e756da555db3f52a3f9a704392","subject":"Group periodic thread workers with selective wrapping","status":"MERGED","created":"2026-05-25 18:17:02.000000000","updated":"2026-05-28 18:31:31.000000000","submitted":"2026-05-28 18:29:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"990002-bug/2154192","meta_rev_id":"3460ab1563dff35abec368a78c05eeff7f352b50","_number":990002,"virtual_id_number":990002,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-05-28 08:07:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-28 18:29:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-28 09:54:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2026-05-28 16:28:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2026-05-28 16:28:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-05-28 09:05:19.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-25 18:29:22.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"CC"},{"updated":"2026-05-25 19:11:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-25 19:40:59.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-25 21:32:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-26 13:34:58.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-05-26 15:34:31.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2026-05-28 02:09:06.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-05-28 09:54:55.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"bc16aaa9212d8afc3f4b0ef60d2d562ec43bc7c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-25 18:17:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"323b96d04763333bbd64ed298c013ff082e15969","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-05-25 18:29:22.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3d94b4fd34b8a4a912ca02893cbe84680d50fb62","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 19:11:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/678589f97cdc479aa7ac2e01577651a1\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/94322f6146754db5aa9f12dc837fa32d : SUCCESS in 47m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c3345d29e166441494098501250799e7 : SUCCESS in 19m 03s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/aca2970c1aae4154a135b0a870e3629b : SUCCESS in 45m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5f8053a7fa654592f76bacfd00869912b745636a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-25 19:40:59.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/02/990002/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"be43bdba604bec0310437cb205b713830068802d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 21:32:27.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7806e8a50ee548d7bc097412070c995e\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/977e108446fc48a7bf53b173490069a7 : SUCCESS in 1h 08m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b022d402e11e42529a57dc2c93daa6b3 : SUCCESS in 1h 00m 59s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b5ca6ff98d1644049a99a799972949da : SUCCESS in 1h 15m 35s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/348d5d0292e841aca3813f7e5ec07a3b : SUCCESS in 2h 10m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2cfb9980c54a4f0e9f9f94f65300ecc7 : SUCCESS in 31m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54fe36d0050a48588e22bb2152613c56 : SUCCESS in 41m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b8f034d5e6840eaa2291c4902464072 : FAILURE in 7m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b7b9d865951a47afa40b8a066de9b6bb : SUCCESS in 29m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b4bca17fb7c745868b28555c1de6138c : SUCCESS in 19m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/57d99e90b11c406488dd82b7f2d68060 : SUCCESS in 28m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adc5173a185240fda8e3f97711f447cf : SUCCESS in 10m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee47a41ef4d248168df7375014e61bc3 : SUCCESS in 21m 02s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/115b010872994250a5a4388c1203e9bb : SUCCESS in 48m 05s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/291c73ea4ce04f8fb16ea59b2d411aad : SUCCESS in 42m 52s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/47c0652fb25642e398fd61dbd7da463b : SUCCESS in 36m 43s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/657edcf0bfe5409e86bdf5e34a93e4a6 : SUCCESS in 1h 44m 19s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4154364c39284cb6bdc5944f085ef083 : SUCCESS in 1h 01m 58s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/09843bc034b24381bf0c32784b6002c8 : SUCCESS in 52m 43s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4793264dba66428cb610eb5a05e5258a : SUCCESS in 1h 28m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2be6fd1226f74aee951f6929e755e0fe : SUCCESS in 1h 31m 47s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/0aad0be88ef4484e874c486ec1718ec8 : TIMED_OUT in 3h 07m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f9cc2a3e9248e5db5d8eb412368965190992cb09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-26 07:17:50.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1, Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"f45808fe83823a8ca6ca2a1e9ab9832154b99f21","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-26 07:18:12.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"062dd1360cb6c6afdfc49813411cf69f9bb01eb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-26 07:58:10.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0aa17aba83098212dbd65c72e9f05bc2e4952837","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-26 08:06:36.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/02/990002/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"1a3fee42a152c4175fecad0ebe2ec0e9faeb87c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 09:39:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2c64939c1984dfa836da63266285b6e\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5c38ec60913849a5ab18b72627e0cbb2 : SUCCESS in 22m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/73f90b42f6344e4eb94bafde11a7f2e8 : SUCCESS in 17m 31s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5cb1a6d59d034594aa8abb2d8d79d738 : SUCCESS in 1h 02m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"60953f0f0edb357f8eae34a193d50116495fc7f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 10:04:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1b2d6a7dec64c769a44799ddc5f9936\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/db0c0b532f97411286da463d17e57111 : SUCCESS in 1h 24m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/13178022329246fd94881691a8731307 : SUCCESS in 1h 13m 25s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/fc796c9a359e407bb8887ad847bacda3 : SUCCESS in 1h 35m 28s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/47bde97ad59f49e58a33bf941c1d2d01 : SUCCESS in 2h 03m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e6e86d32c2704a8aaa568423a7d3f1b7 : SUCCESS in 33m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ec500559e8c4da5b63111777bcd857e : SUCCESS in 1h 08m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f52ab8fd2e84b3d88a761b53a3dfe55 : SUCCESS in 12m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/309423c31fe64389ba8b4b378608550c : SUCCESS in 20m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f7b53140f45f4081b59df31f297e3107 : SUCCESS in 22m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/18a63972c5474be0ac29be2a86ce53b0 : SUCCESS in 30m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4694b54a35364503bae6f846cc37013a : SUCCESS in 9m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4cfc2b6b504a4b3db61204185d49a335 : SUCCESS in 12m 45s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e26d8bc1b83c455cb0be71f0bfea61c5 : SUCCESS in 1h 20m 27s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c58b8a9accda43c093b313da677953ae : SUCCESS in 1h 06m 21s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/2a7b398b3b654c309d62bb71d81e6576 : SUCCESS in 1h 19m 40s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/bb66dd5b3e1a426a8fb87c7a24449e84 : SUCCESS in 1h 48m 41s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/84c8a4d7beeb4f9c8f42e4294dfd0e43 : SUCCESS in 1h 44m 43s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/97f541bcb1974674b028c613ab2f7997 : SUCCESS in 1h 07m 51s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/fc9e6be154de4f05b0ab38a2138a07e4 : SUCCESS in 1h 14m 09s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f89acc2b29ff4942bcf0433fecb70937 : SUCCESS in 1h 05m 24s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/305095c640cc448bbaeb53948354e363 : POST_FAILURE Host unreachable in 1h 10m 24s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":3},{"id":"d5ade1ac2742db4a4372bf7e089dcf04a105689a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-26 10:30:42.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/02/990002/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"472279277f098cdb485b46404bb7f6fd7be03a48","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-26 10:39:39.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3f7ad0001ddd85f315c211f6976da465820a05af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-26 14:02:13.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"6d22ad0236f27d43714b137a8ef32ab46e9ea54a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 16:06:30.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a61f5e4617e42f2863c82eb5c9a2509\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/372e7ce496c842a4ad5597f6160e2cfa : SUCCESS in 19m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e9a4155ef044418f83f62cc25db7b3d8 : SUCCESS in 43m 25s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/bb35c673267c4629921613e2a8a8cbd2 : SUCCESS in 1h 01m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"545cc359cbf78c0d05e5c27854b155b7e137d0e3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-26 16:16:56.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/02/990002/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"b3b5425f51087972af89b7cd50be669175b351de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 16:32:42.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb43f02579964510ab56ff34f52897e5\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/655f97fc35de4e7bad086cfba2f32831 : SUCCESS in 1h 30m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8acc24df4d4b4b9194d12bac983fed96 : SUCCESS in 38m 16s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6039bf0321064f9ba7c5d23a0d6df7e9 : SUCCESS in 1h 43m 28s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/af967ec9519642f9b339b54a791e9982 : SUCCESS in 2h 24m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/40291e9f863b4e379ac25a763b7f7a76 : SUCCESS in 35m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de9f9b9290654969b76a122f53e02a75 : FAILURE in 52m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b33ed53e2c52400b8e26f90e5223710e : FAILURE in 7m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d4b976e9f38348b0841a8af11a933795 : FAILURE in 29m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6b95692e38854ea787806107f184e1a2 : SUCCESS in 17m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/02c9040c160848b6a0e88d2836207f24 : TIMED_OUT in 1h 01m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a694ff438db14f58a2c904bd8f526d49 : SUCCESS in 8m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8a21550e88b4a6d9ece0bb77c14fe9a : SUCCESS in 13m 21s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/78a852e376164eb4a7e4b781c3e786fd : SUCCESS in 47m 34s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/730b59a21af0474aa573c01dccf6cfa9 : SUCCESS in 56m 35s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/b88948b7ced344eb8040cad384956ef7 : SUCCESS in 1h 16m 50s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a760f84d20684a249b325852664be863 : SUCCESS in 1h 54m 13s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8701544329dc43a98112fb04ffb6824f : SUCCESS in 1h 46m 51s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6ec67e9c93aa4881a03ed8b377177795 : SUCCESS in 1h 11m 19s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/cae5b759be094f9b850c2de172e417e1 : SUCCESS in 36m 47s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/30a47b654ea04ea69dc34744034fe55b : SUCCESS in 1h 55m 54s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/08eb57a42a294761a9a84053853d1953 : SUCCESS in 1h 36m 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f312fce1063ca506bbc2dff1c39479c74beabbe6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-27 07:48:48.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1, Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"9c53b9d0b4154a227492602f594c9eafc5699587","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-27 09:02:24.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/02/990002/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"2abd3f5840db15558f1b94ef682ed1771d918a64","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 09:30:29.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6dc9bb5e24754a6fb7d1bee1a46a6517\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9638bb1f5bf8411ca058f47b3dd23782 : SUCCESS in 34m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d43b9e26173c4f93b53de1038e3d28b8 : SUCCESS in 48m 03s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d33bfb5ca2e54e70b8d8ea7f8741c01b : SUCCESS in 1h 05m 45s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"16dd4a9389ca54422f9e94d2870a2ce687aed16d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 10:24:13.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b32fb6c404724cc6a0097fc9c6284486\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4857a78426fc4ddab79a7effb3079152 : SUCCESS in 1h 10m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/740117b529604bf18f8bca4b9c12a159 : SUCCESS in 1h 06m 15s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/fc2a9e9d517a48608695c22cc8cbcf42 : SUCCESS in 1h 26m 07s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1cd3e4db00164549a4ba8e1234781f05 : SUCCESS in 2h 04m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f8d244a088854b2197a07e91c6302c58 : SUCCESS in 32m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2835c21fa2244dac8bb6152b8247a2d5 : SUCCESS in 57m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cd9a71923074a6990ed0c6b1734f057 : SUCCESS in 11m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5da2819a0b734ff0b422857fd408936f : SUCCESS in 27m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2b5f676780b247ca80be2ce28b23c5ae : SUCCESS in 33m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1a581aaab870483791ddd823c58e907a : SUCCESS in 21m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da385d782774444984bb217861c83ab0 : SUCCESS in 6m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/493f4b460b4341128744733668d390d8 : SUCCESS in 19m 48s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9f4f664302cd49baaab17678eafb7069 : SUCCESS in 1h 10m 47s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7268a6314f0a4f8797facaa2e0083719 : SUCCESS in 1h 04m 10s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/71b3f039f9124084bcbd271bfad29f21 : SUCCESS in 1h 19m 30s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/455ef53698b94807aa5f2051cddbf700 : SUCCESS in 1h 58m 11s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f4876d939e43417cb257ed8aedeed9df : SUCCESS in 1h 46m 13s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9819fa35e4f0454db9b25ca541059829 : SUCCESS in 1h 01m 39s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/08658405a32e4949a29f473feee60fbf : SUCCESS in 1h 09m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7ac5b6f844c047c197cb04e5a62b047f : SUCCESS in 1h 23m 23s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/716fe6e76f274b83a3f78c886ac400f2 : SUCCESS in 2h 28m 28s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":5},{"id":"a567a09664231efcd83d86ad1c26c9d6572f64c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-27 17:49:20.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"f335d03ab20fe86bc2913d2cc4877b86c739128e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 19:32:30.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/59d0db8089dd4bc7a30ae3eecbcea7c6\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3e7deba6c5cd4d04ae676d620c701655 : SUCCESS in 1h 12m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/525c7bd9e3174efabf14e7044bbc1767 : SUCCESS in 1h 09m 16s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/84e00b18cb964bb19d4fb1ba2e69cec8 : SUCCESS in 1h 22m 07s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fe0ae9bda7d441378a135122acc94799 : SUCCESS in 1h 41m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b712fdc6b19240d288b9a606150f8c4b : SUCCESS in 33m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a58681d327b4afe882792688c997351 : SUCCESS in 39m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5cddb28f606471c845ba82ca3b8cc5a : FAILURE in 4m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9f82204fcfd74e6a8dc0ca8500bccf3b : SUCCESS in 24m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fb33e8c08a1d44d886cde96e435e93f0 : SUCCESS in 29m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/81f27373bebc429bb66652213ed9bd38 : SUCCESS in 26m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ff51a38d5314cdd97724c70e67d3915 : SUCCESS in 11m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf6d4debd98545c1b614e0f2e89e6096 : SUCCESS in 9m 44s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/50ec86f5c6824cf4a03d40c5d2a0009f : SUCCESS in 50m 34s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4a51b8b089a54dd5bc62f95ff20f60e4 : SUCCESS in 53m 34s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/0d0c57e83a05421999d2e2d29d39d818 : SUCCESS in 1h 01m 44s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/06fe75911290464fb3f418a8d72c5b3a : SUCCESS in 1h 33m 31s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/12cdefce850b445ab719c46304938e18 : SUCCESS in 1h 25m 14s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/02f0771ef195459fa29e9aaefabd6f4c : SUCCESS in 1h 01m 49s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3852a0a9e01d42fb9c4a3248485eaf6c : SUCCESS in 38m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/26c42baee6694a3d8aa2deae8edddd76 : SUCCESS in 47m 56s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/0d3d4a69f1e44b9790884f9a8ff94c16 : SUCCESS in 1h 03m 19s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a30d49dd61b11f6310199ed6cf4108b5b7bbd90d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-27 20:43:49.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"98cbb53ba6f7bf3c2eac8adecd577afe14eb2769","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 22:56:06.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6166fe0d9b048bf97956cbb6d0d701e\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2db9ef02b258416db062029801896613 : SUCCESS in 1h 04m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ebb894d151ed4ffca726fe5cdae702f0 : SUCCESS in 1h 06m 53s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/53d3a85b7c3545dbbba04522c0775c90 : SUCCESS in 1h 23m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/04417f6a49e04b2b969bd8138b03ef0f : SUCCESS in 2h 10m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6138a3da9665433b9d4e6105eeb3bd71 : SUCCESS in 33m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f7b9cce45c294618a890f335758ae76a : SUCCESS in 1h 04m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b27e9e6e9d2849c89ebed37ce01beabc : SUCCESS in 10m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bf01c1336c0942739abe4ab7557f0055 : SUCCESS in 30m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8428283d58744d63a0d1b75bcd70f608 : SUCCESS in 22m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/66bfafd6fb4544db8a379de6ffe7a4a8 : SUCCESS in 32m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fa8c1223bcd45a79d12b8bff92de96c : SUCCESS in 12m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13e3389910af4d6ba645102aed56e587 : SUCCESS in 16m 50s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e0643095c5de411e8dcfb4a456f234a7 : SUCCESS in 52m 03s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c2003de8c2cd400d88dc8f65d4853e49 : SUCCESS in 1h 10m 10s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ec5820fade404e35b7133fda1911dafb : SUCCESS in 39m 07s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/991adff07d014de882eee5c7878faeda : SUCCESS in 1h 55m 47s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6547e0aa371645df9870050bb62bd17d : SUCCESS in 57m 26s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6b4c2af162204e9a802b493115b4ac0d : SUCCESS in 1h 12m 45s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/00728a0beba24b4ba7354d45b6e4e1c2 : SUCCESS in 1h 09m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/47a2fe6eecb344458d74bb4c75148031 : SUCCESS in 1h 20m 08s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/57d39f4cdd8c425297b5de2e36487e51 : RETRY_LIMIT in 3m 40s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":7},{"id":"219209e6017d570e44745673ce549ee071d41c40","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 23:54:58.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05f8d50e4c7446438861bcadb664291d\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/add03c4ae2474e699f17e9931b9801f7 : SUCCESS in 54m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/415bb86e1e974e498f3f01ec6072fa0a : SUCCESS in 36m 51s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/cd450fbd819e4c12a19bab8ce6123f54 : SUCCESS in 24m 37s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"03c401fe06c1880e91a24486b54a7083b9b47029","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-28 02:06:06.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/02/990002/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"8c90ffebda138ea75f731b785a84bb97cdb5606b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-28 02:09:06.000000000","message":"Patch Set 7: Code-Review+2\n\n(5 comments)","accounts_in_message":[],"_revision_number":7},{"id":"90ab66e3f6cf324a2256db980d33f1e006d16106","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-28 03:31:48.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/02/990002/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"0876c4dd144994538def666b045305a5aead3d41","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-28 06:30:24.000000000","message":"Patch Set 7:\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"d6dff744823952197114e0df6771a89c5f0b84bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-28 06:30:48.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"5c1d7a682ff24131f8450a6ade38f91ebfd9acde","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-28 06:46:42.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"c1a4210bcdc5c008d7ac0b16ba431b607dd0fcbc","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-28 08:07:06.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/02/990002/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"1f708856d06d801db735b28f7b9fc572d41062b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 08:51:48.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99436b3e2a8840f5af6a77abc661942a\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/fa89159aed914d0997d490e778acc0ab : SUCCESS in 1h 18m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5c9b051c17074afe82181e427d5c2d4f : SUCCESS in 38m 19s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3ceffc0e6ab543f7a7ff4ee575b75431 : SUCCESS in 1h 32m 28s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e199acd1360c4a0496185c9352615807 : SUCCESS in 2h 04m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/59c4bae623fc476981f3f4fbe3b0a8fa : SUCCESS in 35m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ffee56a62a374fbcbcdcbc9dcc3ba359 : SUCCESS in 37m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d60c786f70846539ac1c1e81ce8e02d : SUCCESS in 6m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/023b00c031b94cad853b148c639c9e59 : SUCCESS in 22m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a826b6a4e95e49ce860e1e1d14fb8e6b : SUCCESS in 19m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fa13b68a71404cb183d66d958111986a : SUCCESS in 27m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2d6bcc78e4f4f2f949bb4c8522bd46e : SUCCESS in 5m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/684541338d8e48549197fa0aef52aee6 : SUCCESS in 15m 44s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b8f17d22fc3a45b58a98c9fbdddb1bb0 : SUCCESS in 1h 05m 11s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5edd4e873f27449693309d0875f7cb36 : SUCCESS in 1h 11m 25s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/7b8892638aea4b6c910cc1615e6c6a43 : SUCCESS in 1h 19m 20s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/644afae492404ef983840af489a07814 : SUCCESS in 1h 05m 35s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ae5f546e13604aaa8bdd425b276278f2 : SUCCESS in 1h 40m 36s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/053bfad1470449e9900e3135ee0f8fc3 : SUCCESS in 1h 03m 56s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ee451c63d388438b9ad57fc41418f411 : SUCCESS in 1h 07m 54s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e8f0a805ea6d4d99b798a043d7ffc4da : SUCCESS in 36m 41s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/027b607a01864bd4aeb1d227495d38f3 : SUCCESS in 2h 14m 19s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":8},{"id":"73e1f9b42ad1eb8ea3c7c38fd8b81ec94f67cd98","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-28 09:05:19.000000000","message":"Patch Set 8: Review-Priority+1","accounts_in_message":[],"_revision_number":8},{"id":"bd7e4254e17376b36f72299bc47827c1ef8b1cce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 09:19:12.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21aa32c2964840a88618b0b0572918f2\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/238ced62402143a08688b2d75bb532ac : SUCCESS in 25m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/af7a533eb2f04fbe8563a6beb1d93629 : SUCCESS in 40m 44s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d10599d0b5db4a2298e3c5ca9ed53264 : SUCCESS in 59m 38s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"9fd63c897c5f77d904cbaca97da2f6bea98a3b71","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-05-28 09:54:55.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"60a589c112d60c10ddf8c8f460b01bf210e1aa25","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-28 16:28:04.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a98db0b39b952caf4b020caeaa1436b7aba9cf8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 16:28:44.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"f1cc4955f456fef202318d07e84ed838513ec254","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 18:29:57.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19735f7adcca4521b76f5d42ace820b6\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0e03b044bd584b6cb4a84171ddc039b4 : SUCCESS in 1h 32m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9babe3a304144236a6e29b90e69a75cc : SUCCESS in 1h 02m 47s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5238f101a9954716995a8efb42c42584 : SUCCESS in 1h 27m 04s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f353f384bbca45588a186cd77ece1f46 : SUCCESS in 1h 55m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56dc24dc391f47d79379f1579c77d18d : SUCCESS in 5m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/67b7ed38f06941be8775b91207f37c23 : SUCCESS in 28m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1d753d776cd6418c95ac7cc4a4ed74d0 : SUCCESS in 26m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff8b5a63f50d42ad9eb7080864af0f89 : SUCCESS in 9m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a107b45ca1348d991bdd168e03c035b : SUCCESS in 16m 07s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e40b7885abf34bfda8fe634f1e5bb45b : SUCCESS in 49m 34s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/81ee3bf916c94a76bdd5deb18a3922c0 : SUCCESS in 1h 55m 33s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/38ccaae0275d4024ab1b6f156fd34622 : SUCCESS in 43m 16s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/fb3f6ad6737441218a086facce53de8e : SUCCESS in 1h 02m 47s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2fb5c9f6e4cf4df591dc2725534359b7 : SUCCESS in 55m 59s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":8},{"id":"acffcf415d0d1762983d4bdf87fa97cce2eeb406","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 18:29:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"3460ab1563dff35abec368a78c05eeff7f352b50","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 18:31:31.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f7f3ea1ab934515b057985943eb184c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47dbc143820a49318c11d89c74b5fbc7 : SUCCESS in 1m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/292bdc7a47e742e4832918117811522c : SUCCESS in 44s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"4bc5833176bc70ec67aa25e3ebedf42f7f15e373","revisions":{"4700fdc5ec84aa30ca58681e4fb29093922e6d28":{"kind":"REWORK","_number":1,"created":"2026-05-25 18:17:02.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/02/990002/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/990002/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/990002/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/990002/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/990002/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/990002/1"}}},"commit":{"parents":[{"commit":"9769135199fcfc59f2f64692b395a372e93ab4f5","subject":"Revert \"Group periodic thread workers together\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9769135199fcfc59f2f64692b395a372e93ab4f5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-05-25 15:17:06.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-05-25 18:16:44.000000000","tz":120},"subject":"WIP \u003d\u003d Group periodic thread workers with selective wrapping","message":"WIP \u003d\u003d Group periodic thread workers with selective wrapping\n\nThe neutron-periodic-workers service should run ML2 plugin periodic\nworkers (PeriodicWorker instances with worker_process_count\u003d0) in a\nsingle process to reduce memory use on constrained nodes. A previous\nattempt wrapped all thread workers unconditionally, but that broke\nRpcWorker instances added by service plugins such as L3 and metering\n(LP#2154192).\n\nIntroduce _prepare_periodic_workers() to classify plugin workers before\nstarting them:\n- Skip MaintenanceWorker and AllServicesNeutronWorker entirely.\n- Start RpcWorker instances individually, never inside\n  AllServicesNeutronWorker.\n- Wrap other workers with worker_process_count\u003d0 in a single\n  AllServicesNeutronWorker.\n- Start workers with worker_process_count\u003e0 in their own processes.\n\nCloses-Bug: #2154192\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: Ia9d319fbbf8047e756da555db3f52a3f9a704392\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4700fdc5ec84aa30ca58681e4fb29093922e6d28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4700fdc5ec84aa30ca58681e4fb29093922e6d28"}]},"branch":"refs/heads/master"},"ba85de7e06883ad487b62de804ef1bbc4bd4178f":{"kind":"REWORK","_number":2,"created":"2026-05-26 07:17:50.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/02/990002/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/990002/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/990002/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/990002/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/990002/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/990002/2"}}},"commit":{"parents":[{"commit":"9769135199fcfc59f2f64692b395a372e93ab4f5","subject":"Revert \"Group periodic thread workers together\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9769135199fcfc59f2f64692b395a372e93ab4f5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-05-25 15:17:06.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-05-26 07:17:41.000000000","tz":120},"subject":"Group periodic thread workers with selective wrapping","message":"Group periodic thread workers with selective wrapping\n\nThe neutron-periodic-workers service should run ML2 plugin periodic\nworkers (inheriting from ``NeutronBaseWorker``).\n\nThere are some of them that can be wrapped inside a\n``AllServicesNeutronWorker`` in order to reduce memory use. All the\nworkers with ``worker_process_count\u003d0`` will be spawned in a single\nprocess, all together.\n\nThis patch introduces ``_prepare_periodic_workers()`` to classify\nthe plugin workers before starting them:\n- Skip ``MaintenanceWorker`` and ``AllServicesNeutronWorker`` entirely.\n- Wrap other workers with ``worker_process_count\u003d0`` in a single\n  ``AllServicesNeutronWorker``.\n- Start workers with worker_process_count\u003e0 in their own processes.\n\nThis patch is also updating the ``BGPWorker.worker_process_count\u003d1``,\nin order to execute it in an isolated process.\n\nThis patch is also updating ``OVNL3RouterPlugin._post_fork_initialize``,\nin order to check the worker type before any check. In a periodic\nworker, the ``_nb_ovn`` or ``_sb_ovn`` IDLs are not initialized,\nblocking the execution of the ``start`` call.\n\nCloses-Bug: #2154192\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: Ia9d319fbbf8047e756da555db3f52a3f9a704392\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ba85de7e06883ad487b62de804ef1bbc4bd4178f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ba85de7e06883ad487b62de804ef1bbc4bd4178f"}]},"branch":"refs/heads/master"},"d5182de4378d1a8630bd78e91294e3087a52d0a9":{"kind":"REWORK","_number":3,"created":"2026-05-26 07:58:10.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/02/990002/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/990002/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/990002/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/990002/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/990002/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/990002/3"}}},"commit":{"parents":[{"commit":"9769135199fcfc59f2f64692b395a372e93ab4f5","subject":"Revert \"Group periodic thread workers together\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9769135199fcfc59f2f64692b395a372e93ab4f5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-05-25 15:17:06.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-05-26 07:56:24.000000000","tz":120},"subject":"Group periodic thread workers with selective wrapping","message":"Group periodic thread workers with selective wrapping\n\nThe neutron-periodic-workers service should run ML2 plugin periodic\nworkers (inheriting from ``NeutronBaseWorker``).\n\nThere are some of them that can be wrapped inside a\n``AllServicesNeutronWorker`` in order to reduce memory use. All the\nworkers with ``worker_process_count\u003d0`` will be spawned in a single\nprocess, all together.\n\nThis patch introduces ``_prepare_periodic_workers()`` to classify\nthe plugin workers before starting them:\n- Skip ``MaintenanceWorker`` and ``AllServicesNeutronWorker`` entirely.\n- Wrap other workers with ``worker_process_count\u003d0`` in a single\n  ``AllServicesNeutronWorker``.\n- Start workers with worker_process_count\u003e0 in their own processes.\n\nThis patch is also updating the ``BGPWorker.worker_process_count\u003d1``,\nin order to execute it in an isolated process.\n\nThis patch is also updating ``OVNL3RouterPlugin._post_fork_initialize``,\nin order to check the worker type before any check. In a periodic\nworker, the ``_nb_ovn`` or ``_sb_ovn`` IDLs are not initialized,\nblocking the execution of the ``start`` call.\n\nCloses-Bug: #2154192\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: Ia9d319fbbf8047e756da555db3f52a3f9a704392\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d5182de4378d1a8630bd78e91294e3087a52d0a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d5182de4378d1a8630bd78e91294e3087a52d0a9"}]},"branch":"refs/heads/master"},"62a56fa84666cfbf20400e5096c94b6b629bacf3":{"kind":"REWORK","_number":4,"created":"2026-05-26 14:02:13.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/02/990002/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/990002/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/990002/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/990002/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/990002/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/990002/4"}}},"commit":{"parents":[{"commit":"aa18156e29c7883b83f5be7de76cdd82ba886d05","subject":"Add worker lifecycle DEBUG logging decorator","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aa18156e29c7883b83f5be7de76cdd82ba886d05"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-05-25 15:17:06.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-05-26 14:01:58.000000000","tz":120},"subject":"Group periodic thread workers with selective wrapping","message":"Group periodic thread workers with selective wrapping\n\nThe neutron-periodic-workers service should run ML2 plugin periodic\nworkers (inheriting from ``NeutronBaseWorker``).\n\nThere are some of them that can be wrapped inside a\n``AllServicesNeutronWorker`` in order to reduce memory use. All the\nworkers with ``worker_process_count\u003d0`` will be spawned in a single\nprocess, all together.\n\nThis patch introduces ``_prepare_periodic_workers()`` to classify\nthe plugin workers before starting them:\n- Skip ``MaintenanceWorker`` and ``AllServicesNeutronWorker`` entirely.\n- Wrap other workers with ``worker_process_count\u003d0`` in a single\n  ``AllServicesNeutronWorker``.\n- Start workers with worker_process_count\u003e0 in their own processes.\n\nThis patch is also updating the ``BGPWorker.worker_process_count\u003d1``,\nin order to execute it in an isolated process.\n\nCloses-Bug: #2154192\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: Ia9d319fbbf8047e756da555db3f52a3f9a704392\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/62a56fa84666cfbf20400e5096c94b6b629bacf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/62a56fa84666cfbf20400e5096c94b6b629bacf3"}]},"branch":"refs/heads/master"},"7f6c2cbea6a0c90a8c2718562ab95bc977a03c24":{"kind":"REWORK","_number":5,"created":"2026-05-27 07:48:48.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/02/990002/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/990002/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/990002/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/990002/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/990002/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/990002/5"}}},"commit":{"parents":[{"commit":"9769135199fcfc59f2f64692b395a372e93ab4f5","subject":"Revert \"Group periodic thread workers together\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9769135199fcfc59f2f64692b395a372e93ab4f5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-05-25 15:17:06.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-05-27 07:31:31.000000000","tz":120},"subject":"Group periodic thread workers with selective wrapping","message":"Group periodic thread workers with selective wrapping\n\nThe neutron-periodic-workers service should run ML2 plugin periodic\nworkers (inheriting from ``NeutronBaseWorker``).\n\nThere are some of them that can be wrapped inside a\n``AllServicesNeutronWorker`` in order to reduce memory use. All the\nworkers with ``worker_process_count\u003d0`` will be spawned in a single\nprocess, all together.\n\nThis patch introduces ``_prepare_periodic_workers()`` to classify\nthe plugin workers before starting them:\n- Skip ``MaintenanceWorker`` and ``AllServicesNeutronWorker`` entirely.\n- Wrap other workers with ``worker_process_count\u003d0`` in a single\n  ``AllServicesNeutronWorker``.\n- Start workers with worker_process_count\u003e0 in their own processes.\n\nThis patch is also updating the ``BGPWorker.worker_process_count\u003d1``,\nin order to execute it in an isolated process.\n\nThis patch is also updating ``OVNL3RouterPlugin._post_fork_initialize``,\nin order to check the worker type before any check. In a periodic\nworker, the ``_nb_ovn`` or ``_sb_ovn`` IDLs are not initialized,\nblocking the execution of the ``start`` call.\n\nCloses-Bug: #2154192\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: Ia9d319fbbf8047e756da555db3f52a3f9a704392\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7f6c2cbea6a0c90a8c2718562ab95bc977a03c24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7f6c2cbea6a0c90a8c2718562ab95bc977a03c24"}]},"branch":"refs/heads/master"},"c8a50b8f5ac45b77db7720ffb1b6ad5d61b4c83a":{"kind":"REWORK","_number":6,"created":"2026-05-27 17:49:20.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/02/990002/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/990002/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/990002/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/990002/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/990002/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/990002/6"}}},"commit":{"parents":[{"commit":"a5391d674a0099043cd410112b0185ac8480df74","subject":"Merge \"doc: Add OVN BGP admin guide for spine-and-leaf fabrics\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a5391d674a0099043cd410112b0185ac8480df74"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-05-25 15:17:06.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-05-27 17:49:01.000000000","tz":120},"subject":"Group periodic thread workers with selective wrapping","message":"Group periodic thread workers with selective wrapping\n\nThe neutron-periodic-workers service should run ML2 plugin periodic\nworkers (inheriting from ``NeutronBaseWorker``).\n\nThere are some of them that can be wrapped inside a\n``AllServicesNeutronWorker`` in order to reduce memory use. All the\nworkers with ``worker_process_count\u003d0`` will be spawned in a single\nprocess, all together.\n\nThis patch introduces ``_prepare_periodic_workers()`` to classify\nthe plugin workers before starting them:\n- Skip ``MaintenanceWorker`` and ``AllServicesNeutronWorker`` entirely.\n- Wrap other workers with ``worker_process_count\u003d0`` in a single\n  ``AllServicesNeutronWorker``.\n- Start workers with worker_process_count\u003e0 in their own processes.\n\nThis patch is also updating the ``BGPWorker.worker_process_count\u003d1``,\nin order to execute it in an isolated process.\n\nCloses-Bug: #2154192\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: Ia9d319fbbf8047e756da555db3f52a3f9a704392\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c8a50b8f5ac45b77db7720ffb1b6ad5d61b4c83a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c8a50b8f5ac45b77db7720ffb1b6ad5d61b4c83a"}]},"branch":"refs/heads/master"},"fe53b5731d78bbb9231ab7d7ae3bce9714167ade":{"kind":"REWORK","_number":7,"created":"2026-05-27 20:43:49.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/02/990002/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/990002/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/990002/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/990002/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/990002/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/990002/7"}}},"commit":{"parents":[{"commit":"78aad983c676661be1a228c6262a80ac528d2bd6","subject":"Merge \"Fix port RBAC policies to require network ownership\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/78aad983c676661be1a228c6262a80ac528d2bd6"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-05-25 15:17:06.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-05-27 20:41:25.000000000","tz":120},"subject":"Group periodic thread workers with selective wrapping","message":"Group periodic thread workers with selective wrapping\n\nThe neutron-periodic-workers service should run ML2 plugin periodic\nworkers (inheriting from ``NeutronBaseWorker``).\n\nThere are some of them that can be wrapped inside a\n``AllServicesNeutronWorker`` in order to reduce memory use. All the\nworkers with ``worker_process_count\u003d0`` will be spawned in a single\nprocess, all together.\n\nThis patch introduces ``_prepare_periodic_workers()`` to classify\nthe plugin workers before starting them:\n- Skip ``MaintenanceWorker`` and ``AllServicesNeutronWorker`` entirely.\n- Wrap other workers with ``worker_process_count\u003d0`` in a single\n  ``AllServicesNeutronWorker``.\n- Start workers with worker_process_count\u003e0 in their own processes.\n\nThis patch is also updating the ``BGPWorker.worker_process_count\u003d1``,\nin order to execute it in an isolated process.\n\nCloses-Bug: #2154192\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: Ia9d319fbbf8047e756da555db3f52a3f9a704392\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fe53b5731d78bbb9231ab7d7ae3bce9714167ade"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fe53b5731d78bbb9231ab7d7ae3bce9714167ade"}]},"branch":"refs/heads/master"},"4bc5833176bc70ec67aa25e3ebedf42f7f15e373":{"kind":"REWORK","_number":8,"created":"2026-05-28 06:30:48.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/02/990002/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/990002/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/990002/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/990002/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/990002/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/990002/8"}}},"commit":{"parents":[{"commit":"e0504685dd6d5aa399f528518267fc48ad8d93c1","subject":"Merge \"ml2: notify mechanism drivers in reverse order during port delete\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e0504685dd6d5aa399f528518267fc48ad8d93c1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-05-25 15:17:06.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-05-28 06:30:33.000000000","tz":120},"subject":"Group periodic thread workers with selective wrapping","message":"Group periodic thread workers with selective wrapping\n\nThe neutron-periodic-workers service should run ML2 plugin periodic\nworkers (inheriting from ``NeutronBaseWorker``).\n\nThere are some of them that can be wrapped inside a\n``AllServicesNeutronWorker`` in order to reduce memory use. All the\nworkers with ``worker_process_count\u003d0`` will be spawned in a single\nprocess, all together.\n\nThis patch introduces ``_prepare_periodic_workers()`` to classify\nthe plugin workers before starting them:\n- Skip ``MaintenanceWorker`` and ``AllServicesNeutronWorker`` entirely.\n- Wrap other workers with ``worker_process_count\u003d0`` in a single\n  ``AllServicesNeutronWorker``.\n- Start workers with worker_process_count\u003e0 in their own processes.\n\nThis patch is also updating the ``BGPWorker.worker_process_count\u003d1``,\nin order to execute it in an isolated process.\n\nCloses-Bug: #2154192\nAssisted-By: Claude Composer 2.5\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: Ia9d319fbbf8047e756da555db3f52a3f9a704392\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4bc5833176bc70ec67aa25e3ebedf42f7f15e373"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4bc5833176bc70ec67aa25e3ebedf42f7f15e373"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
