)]}'
{"id":"openstack%2Fneutron~990125","triplet_id":"openstack%2Fneutron~stable%2F2026.1~Ie68e57da6222965e79d015a5568190d8bc29b9e8","project":"openstack/neutron","branch":"stable/2026.1","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-05-26 21:29:31.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":["OSSN-0102"],"change_id":"Ie68e57da6222965e79d015a5568190d8bc29b9e8","subject":"Fix PF GET/PUT parent floating IP validation","status":"MERGED","created":"2026-05-26 14:49:22.000000000","updated":"2026-07-07 06:37:51.000000000","submitted":"2026-05-27 01:44:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"990125","cherry_pick_of_change":989624,"cherry_pick_of_patch_set":1,"meta_rev_id":"a872749b7b6448ba2d2cd924999c7a898be80f0b","_number":990125,"virtual_id_number":990125,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-27 01:44:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-05-26 14:49:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-05-26 15:22:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-05-26 15:35:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-05-26 14:50:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-26 14:49:53.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-05-26 15:22:22.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-05-26 17:09:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1289d1b0f0487b22f0e1fbdec9671540a0b4bc50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-26 14:49:22.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f810de8b7b10040130b157d63d29f72f6d062fe4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-26 14:49:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"30e7fc2f598cac3d560b4a9ae439cc7bfb655c8b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-26 14:50:35.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"713b3944e4d5d2a2dca38eaba38f8eaa79e68e5d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-05-26 15:22:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ca8aa83d2b2925a164270f8faefc526cfa785277","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-05-26 15:23:00.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"27ac515f735d2ea38b35dc2d9f0f225b170662a9","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-05-26 15:35:36.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"02a79bcaa1477fa77b274a41e1be303168c64767","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 17:09:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/84f072459a314eacac585ade35c7b2a4\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/78fc00a501714d61ad84f50ea72a838c : SUCCESS in 1h 20m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3317ecf695b4430abb948555b27688f2 : SUCCESS in 1h 11m 03s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/5afb01dda35946b48452415186cd1860 : SUCCESS in 1h 59m 51s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/cb7468b47bb64142bfabcab6cf4a8d95 : SUCCESS in 1h 40m 59s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/616bccc44e694f6e976bd9bed9dfa708 : SUCCESS in 2h 12m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ee383e0899d452da3c7dafe80beac27 : SUCCESS in 49m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34a3018d6c1646c088e0e8c0e53c6a03 : SUCCESS in 26m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1a39cfb00af64f08beeb61f2e84ee2c4 : SUCCESS in 37m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b5a8c54200564197aed71fb86c1761b4 : SUCCESS in 24m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a728f9474a534617b0835ba20a6db44a : SUCCESS in 36m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3921d360554545248876b3dc30746b0b : SUCCESS in 12m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/618da54571e3412d81a8227c1f10804f : SUCCESS in 20m 23s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f91f2a1c97734a85b4366de14b74bd4f : SUCCESS in 41m 47s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/df2e225bb0b447f38a68277d8e48fd85 : SUCCESS in 1h 20m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4d42a3049b524ac894f12ee604dbf9b5 : SUCCESS in 1h 27m 47s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/89b65fafadc24581ac96e20232a5f056 : SUCCESS in 1h 10m 19s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/f79f55fb706f44958bb565007d8c0c9b : SUCCESS in 1h 13m 19s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d4fa1756f9fa4be78902af3622c7c524 : FAILURE in 1h 04m 36s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fcab4cefd0964b3bab0f51982cf6392b : SUCCESS in 1h 36m 19s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f00a8c4dc08f49fe96e046268e0750aa : SUCCESS in 1h 15m 58s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3cbc75e8afb74cdb9df2fc79934616a4 : SUCCESS in 1h 11m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/37302b229fe84f5eabbc08add4a1293d : SUCCESS in 1h 25m 32s","accounts_in_message":[],"_revision_number":1},{"id":"c6fa044e56d5f78647cf4332d3435438d2a2a0f3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 17:47:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39619d1576dd4a9183126169766d8023\n\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1751528787634960b559a5d0505ec265 : SUCCESS in 35m 47s\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1755ec6a45434d7da75035dc5efe08a5 : SUCCESS in 27m 10s\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a1d61b816ce6472ab3f799e251d2b523 : SUCCESS in 33m 01s","accounts_in_message":[],"_revision_number":1},{"id":"ac5aa54476f769e55d43f71cb9fe9e7a1b83c868","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-26 21:29:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"75ad19a79219aaecf7004009e30a5c5ae48e9b21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 23:37:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c23bb8167754accb7cfd88c272b9708\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/47e039f4efee495faf055c3fb6b15ae7 : SUCCESS in 1h 27m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/46d492705a80405fbf031931faba713f : SUCCESS in 1h 04m 51s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/196be64b829c438d9237a1e639ffb284 : SUCCESS in 1h 18m 10s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/af5d0b662085438bbf4ab4a6258a35cb : SUCCESS in 2h 06m 16s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/bfffedb3bbdb4bdb8732cf770e438f9f : SUCCESS in 2h 01m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2bd04741933471685834cd6949b02ba : SUCCESS in 37m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9c6eb1068274365ac9b8216fbeea6b4 : SUCCESS in 30m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5fac076842874ae592f86edad92637e2 : SUCCESS in 33m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c7a11dcb1a294449a2dbee7ceefa5f13 : SUCCESS in 20m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b86563af3db1428886e21dfa6f8994d5 : SUCCESS in 33m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d11360c7eafa4659afbb02b04a7396a5 : SUCCESS in 12m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78186fa41c0842d9ae3e4284fb53a962 : SUCCESS in 16m 50s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4136ec98c6b74cb7a2695f5b9938913a : SUCCESS in 57m 12s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d7bd17eb47f748bc98495368c4b7d2e8 : SUCCESS in 1h 16m 48s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f2e64da23f464ea1a57eb79661a2ccc2 : SUCCESS in 1h 02m 29s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/48925b6a5e314d1ca37f1b978abff490 : SUCCESS in 45m 45s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/94b384f2eaf9445d8fcbbf9873b1c353 : SUCCESS in 1h 13m 39s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4d9b2004654745ab9b334ab49b5e0063 : SUCCESS in 1h 50m 15s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a079f00cd3c348cc9ffbafb897cf229b : SUCCESS in 1h 04m 05s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1610d8893d9040a0a9a12f24cbf30c38 : SUCCESS in 1h 07m 22s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c83d0b61b7204ae8b28242faca1b2a5f : SUCCESS in 1h 04m 25s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/72c8876ce22e4a44b389c7f91bef421d : SUCCESS in 1h 28m 17s","accounts_in_message":[],"_revision_number":1},{"id":"294c5597b6a4b5a033b7e59836473a3b6e988a4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 23:37:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fb00768b0ad6c2de3d1cad07dba5e6e5357cc155","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 01:44:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/499f59971b5646818ab61f02d94e83ee\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7d76ebe62a854698bce3e93c0488f660 : SUCCESS in 1h 06m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f4c14b913b2047e88e96f08d12cff4ad : SUCCESS in 49m 38s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/008f15a8948f400c945ab7695363d45f : SUCCESS in 1h 53m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/464db1d55c634cc689e1c429c01693ad : SUCCESS in 24m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c8769a19076c48b0afb8cf7b90f9695c : SUCCESS in 26m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a68f345687924bdc87cdaf5a7cdc30c8 : SUCCESS in 19m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6381df2667114a1f8cb6b29fe38b751c : SUCCESS in 11m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fcce663175514d7f8d7ad1d22696b272 : SUCCESS in 8m 44s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/067949b4cacf4c1cba3975811dda205a : SUCCESS in 1h 21m 01s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d89ded21f8554173af6aae0b1195d860 : SUCCESS in 1h 55m 49s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/aa9eeeccb6f747d8afe726cec52a2f34 : SUCCESS in 1h 15m 45s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/538c4d339ef646cf906cb86ae290ed9d : SUCCESS in 1h 13m 31s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0a7e37fc21954727a7dd85712da19c80 : SUCCESS in 1h 06m 23s","accounts_in_message":[],"_revision_number":1},{"id":"52e705497da3c73b7e6acdaa4db8c442a061a3d4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 01:44:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a7c9d10f498fadb35b31325c3c9ff7d1b092c4a9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 01:45:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e17ffac48bbf4015b90b761050af46cf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75f526b044e943dd9a9098bfe24f0938 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1},{"id":"7f23cd6dc278455d9ad46ac061b5de73b8c96686","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 02:39:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cbbb3e2d2834910966afb99d9c1a4d1\n\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f4ab4f83f03e499ebe4cf3dcd18db4d2 : SUCCESS in 20m 05s\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f2f21cb3d85a40d6b9161611dd64264e : SUCCESS in 54m 51s\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/85dd1d07dd56410a968f6182bdf82b6a : SUCCESS in 1h 00m 38s","accounts_in_message":[],"_revision_number":1},{"id":"a872749b7b6448ba2d2cd924999c7a898be80f0b","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-07-07 06:37:51.000000000","message":"Hashtag added: OSSN-0102","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d540e9df0b3d26880bd6c5c9e2f55d8bc8b1cbe8","revisions":{"d540e9df0b3d26880bd6c5c9e2f55d8bc8b1cbe8":{"kind":"REWORK","_number":1,"created":"2026-05-26 14:49:22.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/25/990125/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/25/990125/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/25/990125/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/25/990125/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/25/990125/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/25/990125/1"}}},"commit":{"parents":[{"commit":"a5e1a3dda76a555bf6ded0c14b0e99f77b1257f9","subject":"Merge \"OVNL3RouterPlugin register events in Neutron API workers\" into stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a5e1a3dda76a555bf6ded0c14b0e99f77b1257f9"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-21 21:50:47.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-26 14:49:22.000000000","tz":0},"subject":"Fix PF GET/PUT parent floating IP validation","message":"Fix PF GET/PUT parent floating IP validation\n\nVerify the port-forwarding floating IP id matches the\nchild floating IP id so we satisfy the policy requirements\nfor GET and PUT operations.\n\nCloses-bug: #2150121\n\nChange-Id: Ie68e57da6222965e79d015a5568190d8bc29b9e8\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nAssisted-by: Claude Sonnet 4.6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d540e9df0b3d26880bd6c5c9e2f55d8bc8b1cbe8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d540e9df0b3d26880bd6c5c9e2f55d8bc8b1cbe8"}]},"branch":"refs/heads/stable/2026.1","conflicts":{"ours":"a5e1a3dda76a555bf6ded0c14b0e99f77b1257f9","theirs":"df79bcc977d1d428b6e81da8ceeab1f19de6eb81","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
