)]}'
{"id":"openstack%2Fneutron~990126","triplet_id":"openstack%2Fneutron~stable%2F2025.2~Ie68e57da6222965e79d015a5568190d8bc29b9e8","project":"openstack/neutron","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-05-27 07:58:23.000000000","reason":"Change was submitted"}},"hashtags":["OSSN-0102"],"change_id":"Ie68e57da6222965e79d015a5568190d8bc29b9e8","subject":"Fix PF GET/PUT parent floating IP validation","status":"MERGED","created":"2026-05-26 14:50:01.000000000","updated":"2026-07-07 06:37:51.000000000","submitted":"2026-05-27 07:58:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"990126","cherry_pick_of_change":989624,"cherry_pick_of_patch_set":1,"meta_rev_id":"86c6d32b251c9c7575a8a66366e84f71de7b075e","_number":990126,"virtual_id_number":990126,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-27 07:58:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-05-26 14:50:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-05-26 15:22:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-05-27 05:47:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-05-26 14:50:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-26 14:50:41.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-05-26 15:22:34.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-05-26 17:11:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e5b4f9ecd1a7ad6ad6479d2cc8bbca1559a1a31e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-26 14:50:01.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2039b86c61e1907a94699e790ad345d3dc343b2b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-26 14:50:41.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"45799cbb8a24737c97e7f0fc6c17f288d617a407","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-05-26 15:22:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ae082a2b5c00f8232cfa2c2d2b31458c239ae8e3","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-05-26 15:23:06.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"ce9b939e1af2ac644d6ef744f70ffcabee5a6729","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 17:11:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2a8d1ff2e26461ab1caad7a79080632\n\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/75228237917d439d8c98cf9def1e1c61 : SUCCESS in 1h 30m 46s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/18760031ed3c4b18a18df13936dbc102 : SUCCESS in 1h 53m 52s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/f4c6d30fb8b042b8a27f39e088a2368e : SUCCESS in 2h 13m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7cca65316d944440ad99a695c107665b : SUCCESS in 31m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97bb025146004bada0877ea9d2772d1b : SUCCESS in 6m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/defd78388a404b4b8dcfa48319dd6e0b : SUCCESS in 21m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9a8475d81fdf4aee84640ec3b7f876f1 : SUCCESS in 19m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bb6a21cac25e4201afd6907475d3230f : FAILURE in 25m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2b245f5d06e4be4a19aaca0b6b3cbdf : SUCCESS in 11m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2e76ca9f2284c1d8bc427666ba534aa : SUCCESS in 16m 55s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/76873940da8f4808a8195f2df66428c6 : SUCCESS in 1h 34m 27s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/af32a9bafd58423abaf89bed2dfde32d : SUCCESS in 1h 11m 29s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/f997701a58484024be8d72a1cb5d2d9f : SUCCESS in 42m 27s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4d2b40cae814425081c9f530a0e5388f : SUCCESS in 1h 06m 18s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c8eb177fa413453f8a154c83afbc7187 : SUCCESS in 1h 43m 53s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f485d55a7b904806b5ea32ed24af3e82 : SUCCESS in 55m 18s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/25f5e0643d9644ec97da797437d3039f : SUCCESS in 33m 52s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/00fe7ea95e0b4642a0a571e4857677cf : SUCCESS in 1h 20m 44s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/02045beaf0de43e9a97db3a280846ca4 : SUCCESS in 1h 42m 14s","accounts_in_message":[],"_revision_number":1},{"id":"5abb5cb2386c8452086f2a540bb1b74d340cdc07","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 17:56:13.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bde727e49f0488480f737fe8a68a2da\n\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/085b063f596a4fa6b563c02546656ead : FAILURE in 39m 06s\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7215b1f39c0244d5b89cfb8ea4595a26 : SUCCESS in 33m 29s\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7eb38ec4d4d34676b6c5e7947eb78062 : SUCCESS in 31m 45s","accounts_in_message":[],"_revision_number":1},{"id":"fb26cf7360bc002e41cc9fb839f1eb8dc60cf984","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-05-27 05:47:01.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2654ffaa76265c5ad971274ee56c358ed6a7c35c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 05:47:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"026e3bde58dffc35e52089a1021203e318bd6232","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 07:58:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe281f25740146689fae067ad9c4babe\n\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/e5b534a6d95a436889a03f9962a7a73e : SUCCESS in 2h 10m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a97d840404984254bb48496fbf96f1ed : SUCCESS in 9m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e6a9dbeaf59a41ea9e3dfec87d3284ff : SUCCESS in 14m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b024bf3e09e9446f893c043fa2ffff94 : SUCCESS in 17m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bf55115bc624539b1a8aa8ec081b21d : SUCCESS in 13m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27827644e96745cd99405b0d2f4cfae5 : SUCCESS in 6m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/bfbf4cdeffa947279cc281fe7b5f194f : SUCCESS in 1h 25m 03s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e983b9b60719421082165b19006aa5f4 : SUCCESS in 1h 42m 17s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/353b5a1e3de84519804051770d938cdc : SUCCESS in 1h 01m 43s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d0a78dcbfb58456fb4ab86d65dd269ea : SUCCESS in 34m 16s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1bda98fbe23a40dba809b3858b01aa78 : SUCCESS in 47m 41s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3c52d899d9d6483487aa728096647d87 : SUCCESS in 1h 03m 33s","accounts_in_message":[],"_revision_number":1},{"id":"37bdda8ea308e32e010d1a9e0b10cbf2c5f5ee7f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 07:58:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6d5d64c2a1d5f4dfd6168db0e7558aba551cb2be","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 07:59:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8327e31b0ee6464b83477c0fd3bee521\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9582a95c46dc4ec9907297b29793fab5 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1},{"id":"86c6d32b251c9c7575a8a66366e84f71de7b075e","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-07-07 06:37:51.000000000","message":"Hashtag added: OSSN-0102","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fd5fe9cd7b52bb015bbd3709ac07a87604f28df0","revisions":{"fd5fe9cd7b52bb015bbd3709ac07a87604f28df0":{"kind":"REWORK","_number":1,"created":"2026-05-26 14:50:01.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/26/990126/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/26/990126/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/26/990126/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/26/990126/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/26/990126/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/26/990126/1"}}},"commit":{"parents":[{"commit":"fa942dba497fd06912b3c03e82ff5f98918a1542","subject":"Fix plural policy names in tagging controller and floatingip policy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fa942dba497fd06912b3c03e82ff5f98918a1542"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-21 21:50:47.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-26 14:50:01.000000000","tz":0},"subject":"Fix PF GET/PUT parent floating IP validation","message":"Fix PF GET/PUT parent floating IP validation\n\nVerify the port-forwarding floating IP id matches the\nchild floating IP id so we satisfy the policy requirements\nfor GET and PUT operations.\n\nCloses-bug: #2150121\n\nChange-Id: Ie68e57da6222965e79d015a5568190d8bc29b9e8\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nAssisted-by: Claude Sonnet 4.6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fd5fe9cd7b52bb015bbd3709ac07a87604f28df0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fd5fe9cd7b52bb015bbd3709ac07a87604f28df0"}]},"branch":"refs/heads/stable/2025.2","conflicts":{"ours":"fa942dba497fd06912b3c03e82ff5f98918a1542","theirs":"df79bcc977d1d428b6e81da8ceeab1f19de6eb81","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
