)]}'
{"id":"openstack%2Fneutron~990127","triplet_id":"openstack%2Fneutron~stable%2F2025.1~Ie68e57da6222965e79d015a5568190d8bc29b9e8","project":"openstack/neutron","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-05-29 15:21:45.000000000","reason":"Change was submitted"}},"hashtags":["OSSN-0102"],"change_id":"Ie68e57da6222965e79d015a5568190d8bc29b9e8","subject":"Fix PF GET/PUT parent floating IP validation","status":"MERGED","created":"2026-05-26 14:50:57.000000000","updated":"2026-07-07 06:37:51.000000000","submitted":"2026-05-29 15:21:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"990127","cherry_pick_of_change":989624,"cherry_pick_of_patch_set":1,"meta_rev_id":"d811b0d2fbb3a98c81f66f80c1fca27cebd2e274","_number":990127,"virtual_id_number":990127,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-29 15:21:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-05-26 14:51:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-05-26 15:23:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-05-29 09:46:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-05-26 14:51:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-26 14:51:40.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-05-26 15:23:12.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-05-26 17:22:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"39b31c8468bc94607c058eb7ec4b67c0c06d314d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-26 14:50:57.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"701ec88b0aff206cfa5e4e9e57ddcc133f1b7103","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-26 14:51:40.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"0160979a4b7a6fbd168d0502b4ceb340cb954b50","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-05-26 15:23:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f0888cf7a36a123fff49c705280fa2afd812b735","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 17:22:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/95ea9506789447eba11d9f5f247b185d\n\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/6913828754cd4175bb1119e69bb37335 : SUCCESS in 1h 22m 14s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/95a043c7ef034d1eb1a60b10f5a0487d : SUCCESS in 2h 05m 25s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/b8ecec74e1a74f7baf81446153c6579f : SUCCESS in 2h 22m 19s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/fcc46d0b82b2407eaa7af8d9307596e9 : FAILURE in 24m 52s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/eb893ce66d174bbc912650f64df1e7a4 : FAILURE in 26m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3289a721096d4312bc5a6aa3947ad3f2 : SUCCESS in 46m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6288700839fb4963a6c0909c33ad53a6 : SUCCESS in 11m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a805cf5e74584759a56e70fa00c3c9b7 : SUCCESS in 23m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cc16dc750db84288b43cfe2cca167d5c : SUCCESS in 20m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef44cbf9200c485d800db0a0fc808804 : SUCCESS in 13m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8cf3ce41d21c49209400350489f7c467 : SUCCESS in 20m 43s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/db142e42e5354ba6898930efbddebe4a : FAILURE in 36m 25s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/92e85a177bd04921a4c92860c7949702 : FAILURE in 24m 59s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e69126ebe87d4c03acb5951d37a5515e : TIMED_OUT in 2h 12m 51s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/8b5a83958ea646048b85534ccb774a56 : SUCCESS in 1h 25m 44s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ac1161896c1247fb8492d8c3c1128caa : FAILURE in 17m 40s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/e039ca563c1b4b3aab1fb7d9a8e969c2 : FAILURE in 26m 18s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1fde6936551148a4864a9c642496a8ba : SUCCESS in 1h 50m 52s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/08bd1434ff1d460293628e163aebaad2 : FAILURE in 23m 45s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/92e9b6a454444e87b156f04d34ab0df9 : SUCCESS in 48m 37s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8efc05dac8ec406cabb91b81bc26edda : SUCCESS in 46m 38s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/014f92b58c964e38a4c32a9bf7c5fc88 : SUCCESS in 1h 07m 47s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8d625701aca248c481dc4bd68907d7ac : SUCCESS in 1h 26m 41s","accounts_in_message":[],"_revision_number":1},{"id":"766794cee0239d03e8491df10f2c4212f4db1a00","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-26 21:28:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a90221e52380f0ab0774b8124133b31a52acee09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 23:30:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c489a315aefb47dab19052ddec4fbb41\n\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/cb6b4daff4294c65be918cdd812501b5 : SUCCESS in 1h 19m 27s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/347353ebad7c4738b075aeccb0982b03 : SUCCESS in 2h 00m 44s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/6ee0340e78cd4dd8bba26e76b2fbefb0 : SUCCESS in 1h 40m 52s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/b2981390e93d46e2be1a7fce624d163e : FAILURE in 22m 52s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/4ee1bbf8358f44fbb9bec790abda6022 : FAILURE in 23m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/392ec81dcb014cbcaca506c8e3615cb4 : SUCCESS in 46m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eea5ff60eac146dca299aa39aa699e0c : SUCCESS in 8m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8cbef294da9c4c06b9a2ae94f25370dc : SUCCESS in 19m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4b29f2c098b64f8bbc3d359e8e0e6a72 : SUCCESS in 21m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f19d266562f49468a5b71d24f636a8c : SUCCESS in 12m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea5c7c8a5e9d4fa9a4fcc7c4c15af399 : SUCCESS in 23m 11s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a82ef4be72cf4d47bb59fbcb12be1d93 : FAILURE in 34m 02s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1f40245ddd864e91b8e206cfc3a146ec : FAILURE in 44m 12s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8f4870b9acb04638a81f3d31532f5c44 : SUCCESS in 1h 25m 58s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/c63f2c56050f4a58a05ac27d724d2f51 : SUCCESS in 1h 00m 42s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b98f973dd5d84f83b4ad9c9069494dc6 : FAILURE in 33m 17s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/b55cea10b1a144559520398a8d42c803 : FAILURE in 47m 51s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/278a0becd90546b1b5b1ffb3486e035a : SUCCESS in 1h 01m 06s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/85b5ce6a36a84aa089ef24131e529560 : FAILURE in 14m 57s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6413eb94600e4967a693258469f16a1b : SUCCESS in 1h 09m 59s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b43cf92b7d9e4281a5b57151faed65c8 : SUCCESS in 1h 02m 55s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a57c55290ab54387af40302ab88b84ce : SUCCESS in 1h 17m 36s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e3988e000b324d6b9a6583288624d127 : SUCCESS in 1h 08m 04s","accounts_in_message":[],"_revision_number":1},{"id":"86c85efcf49fe21f8e2f1a85f8fddb068e0c2a65","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-27 06:35:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5c3aa338e739c49d37881e2b0936fc6c99915787","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-05-27 07:20:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"922bc063eb3f5e604e4b192425560df81c825095","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 08:37:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/92940083714a408a951280657f44e8da\n\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/fdbc6e5c1b9d455099a4e9b607537026 : SUCCESS in 1h 33m 54s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/7dfa0fc1c52246e8b9459daf4de5bdb4 : SUCCESS in 2h 01m 12s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/00ba7f1687a247cbbcb0c8571bb23e89 : SUCCESS in 1h 51m 03s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/5d3ce100a1f145b1b860eed32d4a326b : FAILURE in 24m 01s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/3da8e26134dd49339022038d92abe008 : FAILURE in 25m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7bbbe77570314e7fa8c7a1f8d7a23519 : SUCCESS in 48m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f408b69a8f114e5cb51f6d2aeda64be0 : SUCCESS in 6m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/989c7347c5f1432197134753d713aaa2 : SUCCESS in 15m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c4d6dc920b6a431db2415aee90d0c25f : SUCCESS in 19m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad45c9d44a0246b6b3baf0a06b0ccbca : SUCCESS in 13m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4699abbb7f2944fb9d06cc1f2336ae1b : SUCCESS in 10m 29s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/56d1dff62d0f44afa8c262c13fcf041a : FAILURE in 27m 47s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8e139a49e05d4c199437577679ebf2e4 : FAILURE in 38m 22s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/080edec3bd304c3482c63297cd8b1e4c : SUCCESS in 1h 23m 32s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/d9cf6245ea3e475da3a95d1dbc95cdee : SUCCESS in 1h 15m 42s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/29181c46709940eaa828d664111c0611 : FAILURE in 30m 50s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/21198b59b4d54b7fa44970a77ffc0183 : FAILURE in 46m 24s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/05a3530cb2e5476ca831d93a8cd2e23f : SUCCESS in 1h 46m 01s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/000d52bc8ce44b2993f1488db63cb491 : FAILURE in 25m 25s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2c549f04e2124f51b6d64cf41e5ef32f : SUCCESS in 1h 08m 30s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0e8c0bf707e7406ea8c55f9a063ff7cf : SUCCESS in 1h 02m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9c46972e95dd49a6a4e0e5d24ea8a7ed : SUCCESS in 1h 04m 16s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/cf2c5f835d6746a18f67b4185d66dacd : SUCCESS in 1h 19m 52s","accounts_in_message":[],"_revision_number":1},{"id":"09e60915d38013540fe8c5290c46187d1ee2d070","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-29 09:46:51.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"39448be025d943ebc7b9f9fa308f1c99334bf372","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 12:35:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba63936b3b964c129d7604392a6af849\n\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/d3e74e033e04426d8f98368acba79018 : SUCCESS in 1h 52m 59s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/38807807a5a5483d8c035b5fd5e323d3 : SUCCESS in 1h 59m 51s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/1951d860b3284d3ab894050902d72b35 : SUCCESS in 2h 45m 27s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/a1a40660c3564695ab1c418bc596e224 : SUCCESS in 34m 58s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/79090208cf324df2ad382050deee652b : SUCCESS in 32m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2481b4818db342a8892f718ad9a7a79e : SUCCESS in 48m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45569079746d4da597667da861aecaf3 : SUCCESS in 11m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74064a206c034a9388db489376eccc3a : SUCCESS in 24m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cf8efb485a684541aa75bc51adc1efef : SUCCESS in 20m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d2880c2a2b047bda7a290782ac3eb66 : SUCCESS in 11m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e308e478c02349d8a464f5c11c752d22 : SUCCESS in 20m 59s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/128b88c7231642dbafa5759618608e96 : SUCCESS in 48m 47s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ba8697e8394e40c3bac4bb1f7b9cdc9b : SUCCESS in 59m 26s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/31b4660252c04f1b80b1b5ea29ca517a : SUCCESS in 1h 24m 28s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/45286531bd9d484eb78bd2cffd8b720c : SUCCESS in 1h 10m 09s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/02aad4aa1a8d4e0b8fc46ebfee58a5e1 : SUCCESS in 1h 03m 44s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/b8542326bd4c453b9e97cbcf5ba33c0a : SUCCESS in 1h 12m 53s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/122dd2ff8d1a49cf978bd789cc21ea99 : SUCCESS in 1h 45m 15s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/dd67d35c75994a95a28e74022d3c90d4 : SUCCESS in 1h 35m 26s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/337113e84d744089a48c4dbd8969aa11 : SUCCESS in 51m 56s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/aa395bc9e7a5400f9e16edccba83630d : SUCCESS in 27m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e9cee9eedb9e460cadeab7e926e774df : SUCCESS in 1h 14m 23s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3e4f4f5f4a5541b8b7767226518f93c5 : SUCCESS in 1h 26m 56s","accounts_in_message":[],"_revision_number":1},{"id":"2fb1123aa19b48bf5debdd2a4dfc4e4e9d902769","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 12:36:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4407733ecc5923d5758eaf575eca602bde8daa8a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 15:21:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11d3a725227e47c3a4f62ca3861cf1ed\n\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/68bc10f0590445559e25005f91a4257c : SUCCESS in 2h 14m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c11bbeb10c04ba68aa324db58068ed9 : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0a8c1fc5e92440078d37cc5e40744aa7 : SUCCESS in 14m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1242aa46fd254439bdc2a2908faf07d5 : SUCCESS in 28m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae556ccb3474450484d46c2cde62cb99 : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dde05ba3f54c4d6c907b4b3bf4304db5 : SUCCESS in 21m 11s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6c585a2897824ea4a736ed29d4b95021 : SUCCESS in 1h 31m 50s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/de9100a4d88240c58158b45ea251478f : SUCCESS in 1h 45m 22s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1e980e50a6344ce38cd4a1e4abb9cd2d : SUCCESS in 1h 08m 36s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/25b6bd8e72e44e6c99c8d5ea7127c231 : SUCCESS in 1h 19m 07s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6a2653cade884074951ecc586b138e29 : SUCCESS in 1h 05m 36s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b423b85174a643c3bb08701680e9dd59 : SUCCESS in 1h 05m 32s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a8ab3b28a2ab4ab8b7037ae0bbc17268 : SUCCESS in 1h 18m 19s","accounts_in_message":[],"_revision_number":1},{"id":"6937deefb7f69057290c5ae49190f9859db5176d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 15:21:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b4ebe8341820149d30226ecae0e397d2b74d8bc1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 15:23:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3feebd556ded43a9b73e4d777b1ebd03\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61df5f77ad6a4d6a98208ec1fc721fb3 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1},{"id":"d811b0d2fbb3a98c81f66f80c1fca27cebd2e274","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-07-07 06:37:51.000000000","message":"Hashtag added: OSSN-0102","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a40814d48bacbbf0485a98b2ce385cd357fa36d5","revisions":{"a40814d48bacbbf0485a98b2ce385cd357fa36d5":{"kind":"REWORK","_number":1,"created":"2026-05-26 14:50:57.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/27/990127/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/27/990127/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/27/990127/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/27/990127/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/27/990127/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/27/990127/1"}}},"commit":{"parents":[{"commit":"07f565b10513b90fc53cea0b1b0ca6ad67084ae3","subject":"Merge \"[Fullstack] Kill processes during cleanup phase\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/07f565b10513b90fc53cea0b1b0ca6ad67084ae3"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-21 21:50:47.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-26 14:50:57.000000000","tz":0},"subject":"Fix PF GET/PUT parent floating IP validation","message":"Fix PF GET/PUT parent floating IP validation\n\nVerify the port-forwarding floating IP id matches the\nchild floating IP id so we satisfy the policy requirements\nfor GET and PUT operations.\n\nCloses-bug: #2150121\n\nChange-Id: Ie68e57da6222965e79d015a5568190d8bc29b9e8\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nAssisted-by: Claude Sonnet 4.6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a40814d48bacbbf0485a98b2ce385cd357fa36d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a40814d48bacbbf0485a98b2ce385cd357fa36d5"}]},"branch":"refs/heads/stable/2025.1","conflicts":{"ours":"07f565b10513b90fc53cea0b1b0ca6ad67084ae3","theirs":"df79bcc977d1d428b6e81da8ceeab1f19de6eb81","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
