)]}'
{"id":"openstack%2Fneutron~990128","triplet_id":"openstack%2Fneutron~unmaintained%2F2024.1~Ie68e57da6222965e79d015a5568190d8bc29b9e8","project":"openstack/neutron","branch":"unmaintained/2024.1","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-06-10 11:15:12.000000000","reason":"Change was submitted"}},"hashtags":["OSSN-0102"],"change_id":"Ie68e57da6222965e79d015a5568190d8bc29b9e8","subject":"Fix PF GET/PUT parent floating IP validation","status":"MERGED","created":"2026-05-26 14:51:57.000000000","updated":"2026-07-10 17:46:10.000000000","submitted":"2026-06-10 11:15:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"990128","cherry_pick_of_change":989624,"cherry_pick_of_patch_set":1,"meta_rev_id":"29a843cb88f4da2cad663b65bf6ec8a57ba0742f","_number":990128,"virtual_id_number":990128,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-10 11:15:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-05-26 14:53:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-06-10 09:17:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2026-06-09 15:22:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-06-10 09:17:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-05-26 14:53:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-26 14:53:15.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-05-26 16:46:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-09 15:22:36.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-06-10 09:17:56.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"b438632e2e5223ecc16b74c9423f96c09d8bacde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-26 14:51:57.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"47fbe84c523ac8d0f29b35d619c46fc42b2137db","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-26 14:53:15.000000000","message":"Patch Set 1: Code-Review+1 Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"c1b957aac58acb97d7f40faa50ee6a96cb7000b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 16:46:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa75a8d465fa4a0aa2fa26f9871e0531\n\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/4670fabf53fd4d2abd1460180bb9fb4f : SUCCESS in 1h 00m 30s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/1e5bafa9373f48c1945c4af191257df2 : SUCCESS in 1h 24m 15s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/a5328fb5ae2a4c9fa16a3ad2673474fc : SUCCESS in 1h 46m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd7cb6390a064a01b130892c1b4d08a3 : SUCCESS in 40m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61a024d4c8f2440780957c08059dd8ab : SUCCESS in 11m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a19cb1c2beca42c09b9e40dc8dad6451 : SUCCESS in 24m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ddaca697bcc14fb2a1730d13b1e9c898 : SUCCESS in 23m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9b6bf689dfb4844a7c8e6245d98b724 : SUCCESS in 13m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f747db28c7d4f2eb8f3519f20906edd : SUCCESS in 19m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8c924871025043a098156c7dd1c59842 : SUCCESS in 1h 25m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5cdc639c8df948cdbdee3017c9c1fa78 : SUCCESS in 1h 05m 20s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4beb67cf6a91489aa3d3e06b99b8865d : SUCCESS in 1h 07m 25s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/647a1ce840e44a7ea12073d589240488 : SUCCESS in 1h 12m 12s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5b693279d8944471bcf65a50ee986965 : SUCCESS in 1h 28m 46s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/dd3c664e08034c10ba17f5468afe1010 : SUCCESS in 1h 13m 20s","accounts_in_message":[],"_revision_number":1},{"id":"50b92cc5e04b5d9f8028d44985e3f39984a79ff3","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-06-09 15:22:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b202c894ec2aa36e765728f575f3ab19fdd18d3e","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2026-06-10 09:17:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"69416998024b45c552b775a0eb0453ec054855d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 09:18:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"09bf0b8a65961daf0c22fc7e8bb9f27e3835c25c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 11:15:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51a210f148cb4274a77ff45922e4961b\n\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/6efa9cf6c05e461f8fbf64cbb83a5d3d : SUCCESS in 1h 44m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a18ed67114e8417d9de172bb4be164c6 : SUCCESS in 10m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d03958be289447b83d484cac54211b7 : SUCCESS in 27m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/54c65a9ba1fd41c9a9392141c01f9661 : SUCCESS in 18m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06e72e6c485b455980d1181dce2060b7 : SUCCESS in 5m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7d011a813744c368af00c57911a4199 : SUCCESS in 9m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b9b0f215d7054acbbf05d0931534ea97 : SUCCESS in 1h 30m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5dce83de40b542e8b7eb56d0c395c61d : SUCCESS in 1h 50m 09s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9cc2725a393a49dab85c833aa2433f29 : SUCCESS in 33m 46s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5e9045f4c0b249668287d4fc51a13fd9 : SUCCESS in 44m 20s","accounts_in_message":[],"_revision_number":1},{"id":"c39fcf589435f7d8caccde7ed4e0d21dd45190b8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 11:15:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"907ee0f7e5fb0ea06fbe924114f360f96a7f630f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 11:17:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2804c51c4d344123aa70f6457ee7b583\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfaaff73885548828ff4844e071afdc6 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1},{"id":"29a843cb88f4da2cad663b65bf6ec8a57ba0742f","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-07-10 17:46:10.000000000","message":"Hashtag added: OSSN-0102","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b0411e51be7c0dde85757d2afb8cd432b01fc77a","revisions":{"b0411e51be7c0dde85757d2afb8cd432b01fc77a":{"kind":"REWORK","_number":1,"created":"2026-05-26 14:51:57.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/28/990128/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/28/990128/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/990128/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/990128/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/990128/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/28/990128/1"}}},"commit":{"parents":[{"commit":"0150b162d94f0df0a4d26ec21ef83b7e59bf3daf","subject":"Merge \"Add logging decorator for OVN maintenance methods\" into unmaintained/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0150b162d94f0df0a4d26ec21ef83b7e59bf3daf"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-21 21:50:47.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-05-26 14:51:57.000000000","tz":0},"subject":"Fix PF GET/PUT parent floating IP validation","message":"Fix PF GET/PUT parent floating IP validation\n\nVerify the port-forwarding floating IP id matches the\nchild floating IP id so we satisfy the policy requirements\nfor GET and PUT operations.\n\nCloses-bug: #2150121\n\nChange-Id: Ie68e57da6222965e79d015a5568190d8bc29b9e8\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nAssisted-by: Claude Sonnet 4.6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b0411e51be7c0dde85757d2afb8cd432b01fc77a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b0411e51be7c0dde85757d2afb8cd432b01fc77a"}]},"branch":"refs/heads/unmaintained/2024.1","conflicts":{"ours":"0150b162d94f0df0a4d26ec21ef83b7e59bf3daf","theirs":"df79bcc977d1d428b6e81da8ceeab1f19de6eb81","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
