)]}'
{"id":"openstack%2Fneutron~990136","triplet_id":"openstack%2Fneutron~master~I953db159b9c179adc4b070eec3b0cf0b6a61e6e9","project":"openstack/neutron","branch":"master","topic":"bug/2144617","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-05-28 13:36:27.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"38298":{"account":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"last_update":"2026-05-28 18:03:21.000000000","reason":"Change was submitted"},"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2026-05-28 13:37:16.000000000","reason":"\u003cGERRIT_ACCOUNT_5756\u003e replied on the change","reason_account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2026-05-28 18:03:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I953db159b9c179adc4b070eec3b0cf0b6a61e6e9","subject":"evpn: Add common constants","status":"MERGED","created":"2026-05-26 16:05:12.000000000","updated":"2026-05-28 18:06:44.000000000","submitted":"2026-05-28 18:03:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"990136-bug/2144617","meta_rev_id":"05a66a3bfee8a5414f2ebbe7da173b3f7eb22c8a","_number":990136,"virtual_id_number":990136,"owner":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-05-28 17:12:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-28 18:03:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-05-28 13:36:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-28 13:37:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-05-28 13:37:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-26 17:47:43.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-26 18:16:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-26 18:29:30.000000000","updated_by":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"reviewer":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"state":"CC"},{"updated":"2026-05-27 13:40:07.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"CC"},{"updated":"2026-05-27 14:06:16.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2026-05-27 20:53:52.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-05-27 21:11:48.000000000","updated_by":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"reviewer":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"state":"REVIEWER"},{"updated":"2026-05-28 02:40:01.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"}],"messages":[{"id":"6f3dc3c8ba695a9216c9b3ae8d53f33ed4e81d05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-05-26 16:05:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2027d53fe8c1be660f8af6e3d03e25e9ee502e92","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-05-26 16:09:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"72870198f80506295bb02263faac7c107522d828","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-26 17:47:43.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/36/990136/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"08bb2faf4740238cfa2fbe2e9ac8c54cf207a097","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 18:16:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2146932ba4248e7bba13e5090043fac\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/950b6cacfbff4857adbc9f866e7dba20 : SUCCESS in 1h 39m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e946575e43b14ced9f86f4fa9e0308b1 : SUCCESS in 1h 06m 12s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3d98b2c6aee045e9988df1a9c5852818 : SUCCESS in 1h 37m 53s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4659a07cbeb84ea4b7d705ed5586ef98 : SUCCESS in 2h 05m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f363864bbd0444d881d88fc685a856fd : SUCCESS in 32m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa7861a89bd84206ac557790bef5af1f : SUCCESS in 39m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60a08d88e2214ef9bea8d281f6cb25b9 : SUCCESS in 12m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ff634d4a49f040a7a24f8c0378968c87 : SUCCESS in 24m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/58b22b3963f8418687292146566bc050 : SUCCESS in 26m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b17d066a11ac471a846c8c44799d3a0e : SUCCESS in 32m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38bbce2aa8c541c3896368182cf561b0 : SUCCESS in 12m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f04ec13186b4e188753ef4f0fe10043 : SUCCESS in 18m 23s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/05c35bef82c14029ae079630007292c0 : SUCCESS in 1h 33m 04s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/89afa64726544565aa86348bb97a8df7 : SUCCESS in 1h 09m 06s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/76b290c36730446cb4213028661b228b : SUCCESS in 1h 27m 32s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/76f6174c46174849abfabbb9221c8461 : SUCCESS in 1h 51m 24s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d8b0acdee02544948f78da06c6e3c8da : SUCCESS in 1h 50m 51s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a61f9abb35804628864e11637e94a19e : SUCCESS in 1h 13m 14s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1fb38d6e7d924f3c97f8bd14d04757b2 : SUCCESS in 38m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bb95e5b4bfa14d858457b9722a9416fd : SUCCESS in 1h 33m 33s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/f5bec98b78234fd6af9158f9dbe933d9 : SUCCESS in 1h 22m 07s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":2},{"id":"db466931c3eb537c79d65a709c694a047be89060","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-05-26 18:29:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2f7268a00855b436a1e5709ceeaa9ecebf455e8f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-26 18:34:34.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/36/990136/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"75e9a89bface058967466a549222891e74052b2a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 21:44:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83e18aff9bbc469ab1a317f8c5b3e6ca\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8d2029eac4ad49eb96f541cf7abba179 : SUCCESS in 47m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/30eaea2258f2468b86e758a7231f5c65 : SUCCESS in 42m 47s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/937173ca1f84408f964e5ce77d349dd0 : SUCCESS in 25m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"771e3d0304ec13c69dbd2b168c4ecc96b2559d5c","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-05-26 22:19:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"21f457fe3d609ffecac4316ec16a7fc61b4afc7f","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-05-27 13:40:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"17ea45f2f9e25020d5e204faf9565425e46b9ae2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-05-27 14:04:37.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"9263b4633406e5dd4172fcf569c2e848dc5a8b59","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-05-27 14:21:51.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"70083e440aac45e77e83ccb898f8556ae6d50b87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-05-27 14:28:57.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"172811d838344e29ed5f0045a135d3cf64fd97eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 16:36:38.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba5418e3966746b9b48b652699af126e\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/fc6b4138be0f446eaa9f5812ef70beb0 : SUCCESS in 1h 27m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9be83fb62211498e9b44f50ebd5eb9dd : SUCCESS in 59m 55s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/eed4fa7130ec4cdbab7e565996a975ca : SUCCESS in 1h 38m 30s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6a42523023854be9b2ba22b9c5ae7982 : SUCCESS in 2h 05m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cc7161dfb1aa4b5b94e8f95cb8cbf34e : SUCCESS in 33m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/929c32f3c2454f988088b4561c5c8aa9 : SUCCESS in 1h 04m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f974164f10f474dbb781e3f2d6582f0 : SUCCESS in 8m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f9e1b68fb6154046bd0916806faec05c : SUCCESS in 30m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d7ab0566cd85495dafd251867d4fe1e0 : SUCCESS in 25m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/291e55ad41244e35a7827b1b67044188 : SUCCESS in 29m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d307004ddafe424a885dccd9e3221241 : SUCCESS in 12m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6680ba898ff94defa1b8df1bc58916b0 : SUCCESS in 22m 29s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fb69c54bf1304237b12697c85a3a45f9 : SUCCESS in 1h 01m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d987a6a2019f437a9adc7335063f1084 : SUCCESS in 1h 12m 31s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/bcb51b6840ff4396b4b497588a2137e6 : SUCCESS in 1h 03m 24s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8208efc7aeee47e6b33120c3c419907d : FAILURE in 1h 11m 44s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/39e84928afb94952ae55f756b79ea2ff : SUCCESS in 1h 40m 06s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/013901e912964374b888522ddd3fcc92 : SUCCESS in 1h 15m 11s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/bb0e67a91b084092816fd7c0b51e18fd : SUCCESS in 59m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/dc0e6ea03ee9409f8777c4eb35b82830 : SUCCESS in 56m 50s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/b0c9f581adad4b298a38e398ff2bb7b9 : SUCCESS in 1h 43m 55s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":4},{"id":"575e39002a65d667c66795c8c332eb6094b2f2d2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-27 18:04:12.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/36/990136/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"a211f56d91577217ddd5c1d4187910cf856fb31d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 18:19:46.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d88038d6414429fb90a6c051457c2c7\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2c24da15782d424b8db986896fea2f5e : SUCCESS in 25m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5753275e7257424cb842259e4bf03dff : SUCCESS in 24m 40s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/00a4e14350284f5d87ec707df13b3421 : SUCCESS in 45m 15s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"38c2213183f00a7cb5b3b664403b7261208a0434","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-05-27 18:39:57.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"123d94e4208ba182cfdd67a8c455b3a22d72f786","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-05-27 20:22:50.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"8e0706397ac2feec30254a7217ba4a19c1a0286c","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-05-27 20:23:02.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"4bed429fd2e817beb24396e75b4a753c1f86420b","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-05-27 20:25:34.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"426abf2c554181b1dd828ec47508a036d800afb5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-27 20:53:52.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"9b0e7bc8e7b149eab37b96fff203c97a4a713324","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-05-27 21:11:48.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"8a526debcd677d32f853fcff5dbe77a0e20c06f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 22:32:06.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d46a6b997aed41ed8d9743aa40829d16\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2d930e5439a94e4cb40bb21e08a8fd84 : SUCCESS in 1h 15m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a95e7bee3e4a4c2e8940978cbad92963 : SUCCESS in 50m 21s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/29971b9efd7a45d09bd28436e3f36c62 : SUCCESS in 1h 32m 50s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b74e8b04058f478c9bbe61401a109e57 : SUCCESS in 2h 06m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e68d1b8b629342718c5ecab10e54c9dc : SUCCESS in 37m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e8d5f1ad18644408464220d40508ec3 : SUCCESS in 1h 21m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4102938b435e40c6bef5dcaef0fadfb4 : SUCCESS in 5m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3202a76625454193a6eeb944acf46a2e : SUCCESS in 25m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e7b8dbce685746b595ab595a3fe0033c : SUCCESS in 34m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4301ca76a6cf40fab10af86b0b043479 : SUCCESS in 25m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a739795b15bb4e1ca461cdd24a97af5e : SUCCESS in 11m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a1cc29c4347460ebda15d0e14eeaa8b : SUCCESS in 20m 38s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b9985d5aefd94c2a84454ce18ea894d1 : SUCCESS in 1h 01m 08s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3b0926929bbd4569a2a9e653c404ef62 : SUCCESS in 56m 49s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/917dabde9cf841a39bd69d2d1c779afe : SUCCESS in 39m 54s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2e5e2c959d9347edb9affac7851101d8 : SUCCESS in 2h 00m 26s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4d720a2b0e0c4956bfaef9d46368f1fd : SUCCESS in 59m 32s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3771b4618215465d9d24dc576fc764bf : SUCCESS in 49m 40s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2dfd7b12e5f24fdbb52b421a2820bb4a : SUCCESS in 1h 02m 03s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/91b5728760ed4b4a81ab545c546599f5 : SUCCESS in 38m 08s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/7d41a39d723241a9a69ba0b1e5f5366d : SUCCESS in 1h 49m 07s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":5},{"id":"ca51ac1811a558d25d6ac710ca5a92bfa4e32e02","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 22:43:40.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3444cf6324054620854a1ff3cfd1dda8\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/61b765ec80fa4f1c8784920e03b94fe8 : SUCCESS in 40m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/eb56ff8708fc495eb162eff8242f0796 : SUCCESS in 36m 22s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/17566ad6605649d4bcd63c54cc8d8758 : SUCCESS in 45m 25s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"81ee0009397b9bf8e0d1ad5fd80f35334ac41c9d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-27 23:13:24.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/36/990136/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"420932cee8f3c3f6fa3026630dfae82938277933","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-05-28 02:40:01.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"e4d1c3f09b2a0cb4f8d05b6b543591737ede61c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 02:40:39.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"f4e73ec62230f7c8e6ddb52cd4cb1868babbf3e2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-28 02:48:34.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/36/990136/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"bf27a00d5a72d1d36843e8a96ca69589baa728aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 04:57:19.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a0b2ecaa512f44409610cc6b025e2004\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c6f8cd756ce84008aa42798cd99051a8 : SUCCESS in 1h 29m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fa131fd3ccb7495ca6d742c98a507d97 : FAILURE in 1h 11m 37s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/cb2ab78cafec4e07890408551199d902 : SUCCESS in 1h 18m 13s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e7ef818c386244489e0826d3dcb79ec0 : SUCCESS in 2h 15m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d702edd2d7294c7296a68c90256ce8ea : SUCCESS in 8m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1866a9c42bb745a7aacb15e7be5ff659 : SUCCESS in 25m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/92ac9e51280740f6977b52fb2394b53b : SUCCESS in 27m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13fb61943a4a4580b68983d31e391caa : SUCCESS in 13m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b1937c52a6342b8b1e9abced1f973f8 : SUCCESS in 9m 13s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/734d5314c41a44d6a19fd72970b63dce : SUCCESS in 1h 18m 14s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/eb6a48cb2469495f80fa79fa3365b740 : SUCCESS in 2h 03m 02s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/4749789138df48f8b4918fc52b3bb822 : SUCCESS in 1h 17m 17s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c8e28f754a4741bc8fce396d92c8f336 : SUCCESS in 56m 25s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f4763a1fe19a48bb9d812e4e389ac592 : SUCCESS in 1h 05m 45s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":5},{"id":"800566bda7a5f043f2bf7bcf031dbb12af94c350","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-28 08:05:38.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4f9b08090497ba570b9f17e83321f676f479b9d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 10:33:39.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cb8f8000fa542159130ce16cc69bc8f\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1bc34c39eb67488dbfa586f0920dd282 : SUCCESS in 1h 32m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/01ee8363c00741a081541e59c6a876a5 : SUCCESS in 1h 04m 19s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e6001de068a4466db104460fe2abd06a : SUCCESS in 1h 25m 18s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f18bd3012f7342ab90b01935cd733ac8 : SUCCESS in 2h 23m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f9d6b51439d047fdb425dab0e0138585 : SUCCESS in 33m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f128d4347744b1bb1a409d1c3fc535b : SUCCESS in 1h 18m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/024d0f60f17c42b985cd209f23c23884 : SUCCESS in 9m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d3ae37c68b5d4240b8bcd1e7acf84b92 : SUCCESS in 29m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/75ce4feb84ab489eb0b6c29859a58356 : SUCCESS in 16m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/942490309244425abb635bd53f4892c3 : SUCCESS in 19m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee05e9f9b35149c3ae6058f49d3f307a : SUCCESS in 6m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d2f62ee2fa3d4c5c81c845fd51657c92 : SUCCESS in 9m 36s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/93ed5d8739f343d8ab4dc49a8639b103 : SUCCESS in 1h 22m 59s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c9ec57151e8a495a8c1a54ce05a5a98b : SUCCESS in 1h 08m 43s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/70ec1413caa54f31abeac4761f83fa24 : SUCCESS in 44m 09s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5d9956642f004ec88321c7ea68615815 : SUCCESS in 1h 55m 52s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/043c3cf658a14ba29c6a86f3ab481737 : SUCCESS in 1h 53m 56s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b8e4a900d0f34a81813b2d201945f601 : SUCCESS in 52m 26s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7f0e10c1bc2e4e26a49c2a88a4e4cf1c : SUCCESS in 1h 08m 47s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/26f0308cfc9a4ffda377872d7cdc20ce : SUCCESS in 36m 29s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/4767ef036d4543719817edc05a6a9713 : SUCCESS in 1h 08m 35s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":5},{"id":"ba23050eab2f7e82b464d2614c43c676352efdb9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 10:34:11.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"267ae651d04515ba2ab29892afea00bf80ed1e74","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 11:13:17.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59a17db2edc142a0885bfb557922b649\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/53aa778ebf11487faedb4ac95fd1813b : SUCCESS in 21m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a2f3c896aa224fba8646591f2d5cbd33 : SUCCESS in 47m 02s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/be62438095a3487b9956d2014043bf8f : SUCCESS in 1h 03m 22s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3bc45dfb523b23392ab8e11d99d24a32b57ee614","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-28 11:46:05.000000000","message":"Patch Set 5:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/36/990136/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"f7b9b975727c8c51c43f9f5535d8b783a14c9b1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 12:40:38.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0b632af4c0c048c691a706414b22e848\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e4c658e8ab05437bb119445068795760 : SUCCESS in 1h 37m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a9552394ae124178a7c37bb32af66e8d : SUCCESS in 1h 01m 53s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/581820d8d62547ecb2f8ced92484c289 : SUCCESS in 1h 28m 32s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/45bc1cbebe4c44f7b1a3e86ec58f8fd1 : SUCCESS in 1h 43m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9763dd9922947398afbb757f3f9834a : SUCCESS in 10m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/54cc856552e64575a2685900a56b5dd9 : SUCCESS in 28m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/881b4a48e77a460a82cac988dc509f21 : SUCCESS in 28m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/429636f4dd444c8bbb6192fbd12882d8 : SUCCESS in 12m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f0339de54794d8e91003eb01638661d : SUCCESS in 18m 30s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ef07ce4ea96d4cf28d8c78f83c973701 : SUCCESS in 1h 25m 34s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/10715d4bfcec42c283ded85044a49ce9 : SUCCESS in 1h 52m 25s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/051c6ca9358f4847aa1423b6a7631862 : FAILURE in 40m 27s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/017588baa75b46bfbec047b989c38fbe : FAILURE in 50m 55s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c1717b9aa695471da53bafa8bc5c1373 : SUCCESS in 55m 29s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":5},{"id":"0285d9634f9e38af930a918afc35d5427dc2d838","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-05-28 12:58:17.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9c7b5c55c89f19a4269a5a97a458ae0598b20a56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-05-28 13:31:36.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1, Verified-2 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"fa851c1884e21cdfeab4e67c7c95a001d7ef8e0d","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-05-28 13:34:53.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"1ec2b64791deff816313f360a72a141ff132eb25","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-28 13:35:28.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"d67627fdc73b35e9b14664149266f12268459ee4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-05-28 13:35:56.000000000","message":"Patch Set 7: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"442e8d5ab88017047cc4caefd755c12b80bcd9a3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-28 13:36:27.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"2261920d10def677a0ed3938572a871641cd7c27","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-05-28 13:36:34.000000000","message":"Topic evpn_portbinding_events2 removed","accounts_in_message":[],"_revision_number":7},{"id":"e6f85d2f697fc0dd07effa6bd394f81ce306922d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-05-28 13:36:36.000000000","message":"Topic set to bug/2144617","accounts_in_message":[],"_revision_number":7},{"id":"1bd32de1ffce5a309f038fa3ad8ffad718fc0554","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-05-28 13:37:16.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"f23000746e4a02368d3b893bf49255662eba9b7d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-28 14:59:01.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/36/990136/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"aacc2fefc2b6bc815429b438b9232c0aa10529a4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 15:25:44.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eab87895f37a4234a82a3c3cb7bae1e7\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c02034b56f984c82b6822768cdfb60b3 : SUCCESS in 27m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9dfffc06b5674dd8a2668f043bf3e03d : SUCCESS in 24m 52s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d22c600fc43746959bee9d74b93ff70c : SUCCESS in 1h 07m 18s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"618b89c30627fab4e22b90f602c10ea9a361cc6c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-28 15:43:20.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/36/990136/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"8768e1feb91ec881c7acf52156d770a8fb7ba523","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 15:48:50.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45d1d4cd33bb4f24b0facf88f74e9845\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6e69d1859483435192f2cce0ae3d45a5 : SUCCESS in 1h 48m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7a2830f9adb142c7ba53282427abeffa : SUCCESS in 1h 02m 03s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/136bb7887b964cb08b25dedc862e9ca1 : SUCCESS in 1h 30m 56s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3e6df37bd5ae418ea0e8d16d4be1a221 : SUCCESS in 1h 57m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/818bd737ea9c43c481096c4319cc4551 : SUCCESS in 34m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5365fe75d40f470d873f57e14cd8e53c : SUCCESS in 1h 05m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1d43cc78bc24a8887e9565b7817f17a : SUCCESS in 10m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6aeac3b0775e41e687bac53dee07dc47 : SUCCESS in 29m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5cc086b2c0c04cb7993fa9efbbded0ac : SUCCESS in 17m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c004cea6687144f1ad159f37152fd08c : SUCCESS in 33m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b800686b34e34d159b39e18bb62ff3d5 : SUCCESS in 9m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/178bb6bf0cc84e969bbceb331df6bc87 : SUCCESS in 14m 13s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1a25092621fd41a3ab831d2915d001d6 : SUCCESS in 2h 09m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/679e98d5ac6346509a75bc201444b110 : SUCCESS in 1h 10m 22s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/fc10ceb5d6164350811260c15e2e3aad : SUCCESS in 46m 16s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ee8e790a48074d69b7fd518495fa4361 : SUCCESS in 1h 41m 57s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0a12f698f5ef44d58776e8dad53e32be : SUCCESS in 1h 03m 45s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0242a8ae0e1648b080f05f85e309cb28 : SUCCESS in 54m 03s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2437f0e471ee4fbaafd70cf2c596ff41 : SUCCESS in 57m 52s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4fa41b4234bc48fba9b7982afe8748aa : SUCCESS in 40m 32s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/8be6615ada94435091be698d43f50e40 : SUCCESS in 1h 14m 29s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":7},{"id":"65992f72eebeba764f5f6743805ca1d18f49f0e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 15:49:25.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"531ed7c0f02d4281f9a358e1ac1d87b670250dda","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-28 17:12:03.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/36/990136/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"6d625b56911729eadb6c7c57fd5903bb826dfce9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 18:03:21.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f32372b0bc74312b661b686e3fd7d49\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/725497edf6044aef877d163d4a50b80e : SUCCESS in 1h 23m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/260122b74a7a418582bea3e826cbfd93 : SUCCESS in 38m 01s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/39b9008a90b940b988adaac931353a5b : SUCCESS in 1h 39m 11s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/cfd50b7635ad40438b9f1c343c119e56 : SUCCESS in 2h 09m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55ee324706614892aec834e08d04d872 : SUCCESS in 7m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/de1ee8dd485a46e3a869c7120408e0ae : SUCCESS in 31m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/880d9830598b4a90a4659d6891cdac5b : SUCCESS in 24m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b81385acb10945b08fd1bfcfe95b2501 : SUCCESS in 13m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8fa24fa3d45460a9e93cb13cb17f41d : SUCCESS in 22m 51s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/806a3c65faff4eb38efb1ff81ff2e8a0 : SUCCESS in 1h 12m 56s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/76cc14007cb74cfbb740c7c3bdbdf8e4 : SUCCESS in 2h 08m 47s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/3d7c6f0d5f9947c098c94c37df8d4905 : SUCCESS in 1h 20m 52s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/fcab6a39dfd046768b853ce2d46ef4ed : SUCCESS in 1h 12m 37s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b35cbda7d74440c9b785bb376b511107 : SUCCESS in 1h 12m 29s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":7},{"id":"89049184e8ecd8b7f616c890aa64e9b035f89cf0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 18:03:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"05a66a3bfee8a5414f2ebbe7da173b3f7eb22c8a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 18:06:44.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebfb34e236fa4832acae685e8d6bc66f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/002f8b1896e94eb3814ca4a1da711281 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"0d0ff0fda620ebfb06e5de826210e029901c341b","revisions":{"299ca8bcdb7ac9305ac605ff337a1f682432b4b1":{"kind":"REWORK","_number":1,"created":"2026-05-26 16:05:12.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/36/990136/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/990136/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/990136/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/990136/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/990136/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/990136/1"}}},"commit":{"parents":[{"commit":"cd8075583b093cf81ca97d814b0ea072d9022206","subject":"Merge \"Fix PF GET/PUT parent floating IP validation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cd8075583b093cf81ca97d814b0ea072d9022206"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-05-26 16:01:21.000000000","tz":-240},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-05-26 16:04:55.000000000","tz":-240},"subject":"evpn: Add common constants","message":"evpn: Add common constants\n\nThe patch adds constants for OVN options, OVN resource naming and\nfor the agent interfaces namings.\n\nRelated-Bug: #2144617\nChange-Id: I953db159b9c179adc4b070eec3b0cf0b6a61e6e9\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/299ca8bcdb7ac9305ac605ff337a1f682432b4b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/299ca8bcdb7ac9305ac605ff337a1f682432b4b1"}]},"branch":"refs/heads/master"},"8913ceb941f90ee4895f5a4e877288aabcea5aa9":{"kind":"REWORK","_number":2,"created":"2026-05-26 16:09:31.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/36/990136/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/990136/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/990136/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/990136/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/990136/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/990136/2"}}},"commit":{"parents":[{"commit":"cd8075583b093cf81ca97d814b0ea072d9022206","subject":"Merge \"Fix PF GET/PUT parent floating IP validation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cd8075583b093cf81ca97d814b0ea072d9022206"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-05-26 16:01:21.000000000","tz":-240},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-05-26 16:09:24.000000000","tz":-240},"subject":"evpn: Add common constants","message":"evpn: Add common constants\n\nThe patch adds constants for OVN options, OVN resource naming and\nfor the agent interfaces namings.\n\nRelated-Bug: #2144617\nChange-Id: I953db159b9c179adc4b070eec3b0cf0b6a61e6e9\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8913ceb941f90ee4895f5a4e877288aabcea5aa9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8913ceb941f90ee4895f5a4e877288aabcea5aa9"}]},"branch":"refs/heads/master"},"90b2ddfc8481e3a494b9ea76b8ddfb709330c45c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-05-27 14:04:37.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/36/990136/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/990136/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/990136/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/990136/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/990136/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/990136/3"}}},"commit":{"parents":[{"commit":"7cfd848ca2e4aacfd786918adb18a5dc5da5a591","subject":"Merge \"Add PVLAN service plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7cfd848ca2e4aacfd786918adb18a5dc5da5a591"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-05-26 16:01:21.000000000","tz":-240},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-05-27 14:02:04.000000000","tz":-240},"subject":"evpn: Add common constants","message":"evpn: Add common constants\n\nThe patch adds constants for OVN options, OVN resource naming and\nfor the agent interfaces namings.\n\nRelated-Bug: #2144617\nChange-Id: I953db159b9c179adc4b070eec3b0cf0b6a61e6e9\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/90b2ddfc8481e3a494b9ea76b8ddfb709330c45c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/90b2ddfc8481e3a494b9ea76b8ddfb709330c45c"}]},"branch":"refs/heads/master"},"77b88da66aa2f1dfbd6354b23fab359f49cdb058":{"kind":"REWORK","_number":4,"created":"2026-05-27 14:28:57.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/36/990136/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/990136/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/990136/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/990136/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/990136/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/990136/4"}}},"commit":{"parents":[{"commit":"7cfd848ca2e4aacfd786918adb18a5dc5da5a591","subject":"Merge \"Add PVLAN service plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7cfd848ca2e4aacfd786918adb18a5dc5da5a591"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-05-26 16:01:21.000000000","tz":-240},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-05-27 14:28:34.000000000","tz":-240},"subject":"evpn: Add common constants","message":"evpn: Add common constants\n\nThe patch adds constants for OVN options, OVN resource naming and\nfor the agent interfaces namings.\n\nRelated-Bug: #2144617\nChange-Id: I953db159b9c179adc4b070eec3b0cf0b6a61e6e9\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/77b88da66aa2f1dfbd6354b23fab359f49cdb058"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/77b88da66aa2f1dfbd6354b23fab359f49cdb058"}]},"branch":"refs/heads/master"},"8e9b1ee75bf94b5476e10a9981c125a422a073be":{"kind":"REWORK","_number":5,"created":"2026-05-27 20:22:50.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/36/990136/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/990136/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/990136/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/990136/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/990136/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/990136/5"}}},"commit":{"parents":[{"commit":"7cfd848ca2e4aacfd786918adb18a5dc5da5a591","subject":"Merge \"Add PVLAN service plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7cfd848ca2e4aacfd786918adb18a5dc5da5a591"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-05-26 16:01:21.000000000","tz":-240},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-05-27 20:22:50.000000000","tz":0},"subject":"evpn: Add common constants","message":"evpn: Add common constants\n\nThe patch adds constants for OVN options, OVN resource naming and\nfor the agent interfaces namings.\n\nRelated-Bug: #2144617\nChange-Id: I953db159b9c179adc4b070eec3b0cf0b6a61e6e9\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8e9b1ee75bf94b5476e10a9981c125a422a073be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8e9b1ee75bf94b5476e10a9981c125a422a073be"}]},"branch":"refs/heads/master"},"950102fb328960f1f34429f0f2be09085dc68c17":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":6,"created":"2026-05-28 13:31:36.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/36/990136/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/990136/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/990136/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/990136/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/990136/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/990136/6"}}},"commit":{"parents":[{"commit":"e0504685dd6d5aa399f528518267fc48ad8d93c1","subject":"Merge \"ml2: notify mechanism drivers in reverse order during port delete\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e0504685dd6d5aa399f528518267fc48ad8d93c1"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-05-26 16:01:21.000000000","tz":-240},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2026-05-28 13:15:53.000000000","tz":0},"subject":"evpn: Add common constants","message":"evpn: Add common constants\n\nThe patch adds constants for OVN options, OVN resource naming and\nfor the agent interfaces namings.\n\nRelated-Bug: #2144617\nChange-Id: I953db159b9c179adc4b070eec3b0cf0b6a61e6e9\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n(cherry picked from commit 8e9b1ee75bf94b5476e10a9981c125a422a073be)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/950102fb328960f1f34429f0f2be09085dc68c17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/950102fb328960f1f34429f0f2be09085dc68c17"}]},"branch":"refs/heads/master"},"0d0ff0fda620ebfb06e5de826210e029901c341b":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2026-05-28 13:35:56.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/36/990136/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/990136/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/990136/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/990136/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/990136/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/990136/7"}}},"commit":{"parents":[{"commit":"e0504685dd6d5aa399f528518267fc48ad8d93c1","subject":"Merge \"ml2: notify mechanism drivers in reverse order during port delete\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e0504685dd6d5aa399f528518267fc48ad8d93c1"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-05-26 16:01:21.000000000","tz":-240},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-05-28 13:35:56.000000000","tz":0},"subject":"evpn: Add common constants","message":"evpn: Add common constants\n\nThe patch adds constants for OVN options, OVN resource naming and\nfor the agent interfaces namings.\n\nRelated-Bug: #2144617\nChange-Id: I953db159b9c179adc4b070eec3b0cf0b6a61e6e9\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0d0ff0fda620ebfb06e5de826210e029901c341b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0d0ff0fda620ebfb06e5de826210e029901c341b"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
