)]}'
{"id":"openstack%2Fneutron~990481","triplet_id":"openstack%2Fneutron~master~I7cadbdb738bf95374661798047b25bd9ef1fe640","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2026-05-30 09:52:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7cadbdb738bf95374661798047b25bd9ef1fe640","subject":"bgp: Do not learn routes from main BGP LRP","status":"MERGED","created":"2026-05-28 14:52:57.000000000","updated":"2026-05-30 09:54:10.000000000","submitted":"2026-05-30 09:52:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"990481","meta_rev_id":"ec6c0ab07ce2bb102e96b461de81f171ba935585","_number":990481,"virtual_id_number":990481,"owner":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-05-28 20:10:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-30 09:52:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-05-28 15:15:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-29 09:46:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2026-05-29 06:32:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2026-05-29 09:46:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-28 15:15:58.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-05-28 15:16:48.000000000","updated_by":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"reviewer":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"state":"CC"},{"updated":"2026-05-28 17:10:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-28 20:10:36.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-29 06:32:42.000000000","updated_by":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"reviewer":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"state":"REVIEWER"},{"updated":"2026-05-29 09:46:36.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"82c33f6e207f1787a4de2a099804174ecbeef621","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-05-28 14:52:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8c832b79aeb48220a17e0a114e66f4549a133f36","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-28 15:15:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7bf8a829a0c9a55b7d468a129fecaa851c7d7698","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 17:10:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/803253e6807a4b428a910d34b103cdf3\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6c4f2ec433cc47748fb581b53f61c915 : FAILURE in 59m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/005702073f054c6f82f805bae177eb40 : SUCCESS in 43m 41s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/69db6656795c4473a153c3a11d2c3271 : SUCCESS in 1h 16m 07s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2523c7f284134ac8b1182e82a7202b5d : SUCCESS in 2h 16m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/35be1b92857543bd98fd32c71179d2b7 : SUCCESS in 37m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df983dcece594c7ba1e82f1595c907a0 : SUCCESS in 1h 12m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a8074ccc3524a7b90c301fbc591e638 : SUCCESS in 11m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/08486e52c6e142c9877a47d82891ec92 : SUCCESS in 19m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d92569b0292a440ba3cf70d6a042661e : SUCCESS in 22m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4a432758a4214758bcfbe6c9f845248a : SUCCESS in 34m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/feb33298ac7a4885a2f855acb337bf5e : SUCCESS in 6m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d4fd9aafc7a43829f663c82f1027128 : SUCCESS in 16m 47s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e2557f083e884fcfbc76947203055ca3 : SUCCESS in 1h 41m 43s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7075875fd8e0464abe9b027d9ecc8197 : SUCCESS in 1h 11m 45s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/298636f95f8e4536ae49cd23d978a772 : SUCCESS in 1h 20m 37s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/dff5bd27792c4d22aa699c6cc57e5365 : SUCCESS in 1h 07m 32s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8b7cd1333e1a433888401a870fb8a17d : SUCCESS in 1h 46m 06s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/29d79a5fb3064c0cb333d855a02fe3c1 : SUCCESS in 1h 18m 32s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/9c3dc26e766f499cabe043ac82350eb8 : SUCCESS in 1h 12m 37s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c81227e00f9947be80799749423de15f : SUCCESS in 1h 28m 34s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/7ea451ea9bd64a0e867ef6b0e702997b : SUCCESS in 2h 04m 37s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":1},{"id":"1584577d22f076dd6b3ccbf8f97ba73ec7cb5951","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 18:20:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7423e8e7d1140b6b9652d5a24ddad74\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/05c8d6133dac48a5b7fe64a77d789647 : SUCCESS in 43m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/45c4ed085f3b4083b36f754a6f048bb3 : SUCCESS in 25m 39s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/19bf9fbfa45142dabc1f30e8f07200a1 : SUCCESS in 26m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a4a224024e1f90ce44d1d4c987a0ed266ae6cc2d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-28 20:10:36.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/81/990481/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"9dcef3b3f17c12b013a3befbb455b644b090b13b","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-05-29 06:07:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6278be25ec8d7eceefa30b572b27eb0458c64a6d","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-05-29 06:32:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0eb01db21305cafb1c7f8c0113a4187fb2064ad7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 08:31:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43ef837a46264438b554ecf1cafe84ce\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b7e4da087c124643a30ee8af657e637d : SUCCESS in 59m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8db3a20bcf28453785bf8aea2c550e00 : SUCCESS in 48m 42s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/cb421f047ee3404b9106127c26de77b5 : SUCCESS in 51m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e6a4db24f300ab2a8405d47da7d58a068228421c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 08:45:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e364d3d3d204f5086253533b75dd61c\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/31135890db254907af95f0310d32f581 : SUCCESS in 1h 02m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/29d785b59c7e408f934439b71700c3a2 : SUCCESS in 1h 09m 32s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/af3e1c321be345b9beaacf9489d28c4a : SUCCESS in 1h 40m 11s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/553065ad3e7c4c088db89b6ac2bb3507 : SUCCESS in 2h 32m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ec1896b2c4af4af79134d537dbd606fa : SUCCESS in 35m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/995c44bb407a41dbb1ab9e5e9e0817a7 : SUCCESS in 1h 02m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b13b5d7fbb894aebb883ffb8c93e83d1 : SUCCESS in 5m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1e1adb945beb44558f5921c0243507c7 : SUCCESS in 21m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/74b9f4604bd545e884c5cf0922b60014 : SUCCESS in 16m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/aba8b96655e54ccc9b9cfc89464d094b : SUCCESS in 24m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25f088c78d31401db231bd921d811ca1 : SUCCESS in 6m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0188ef3aea6243c3a58850462fabfcc7 : SUCCESS in 18m 41s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/231b2c20e6fa4369834d7edb14b6778c : SUCCESS in 45m 04s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5a3ee75ebc084b8f9344a79783abdf78 : SUCCESS in 50m 27s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/3ea3a2f6f01745ac9d8180b07a6de79d : SUCCESS in 43m 28s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7c769bed91a444158517d4bb289f752c : SUCCESS in 1h 12m 31s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/77f1855b6b704a948454a03c8cd6a58d : SUCCESS in 1h 03m 57s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/77e5dfd9d2db41e28eaa6c7d2074ff70 : SUCCESS in 51m 05s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/03000dc074af4ee2bc5763505c01da64 : SUCCESS in 31m 31s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cad4d193a76a40ca86c97b964b48e09a : SUCCESS in 1h 31m 01s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/6f565e121e0841b381bfb8fc20355b25 : SUCCESS in 2h 22m 02s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":1},{"id":"28f1483c0da3f7b9c5c4cae6f95f2c9dd9850046","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-29 09:31:50.000000000","message":"Patch Set 1:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/81/990481/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"79df9873d861a6590bebeaf4a5f5d66273d3d5ec","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-05-29 09:46:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6c51fc3b46124b3ac51d1def1d3d536d23f3f247","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 09:47:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"95c47784c0923b951c76b814452ad679b720862a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 12:10:56.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7688d68980604d9bac65acca21ab62a0\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/811f7121745d441c881618686242428e : FAILURE in 1h 49m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b330b70107a34a3fa1f3874af8e91e95 : SUCCESS in 1h 04m 18s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6206cd4d3e7b4fa8b3afb59a758de786 : SUCCESS in 1h 34m 08s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/74bcd7601f564605af9871edb7916759 : SUCCESS in 2h 23m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6ef3cf861ee489192f726877eef336d : SUCCESS in 5m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/38acd27085bb4f7eabeae3f8dffa3a49 : SUCCESS in 20m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ee02635abe344c3aac72d41824fec76c : SUCCESS in 19m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c40b2ad6072348a7bd495d77fedec64c : SUCCESS in 10m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83810401f391425e8d93880740470bce : SUCCESS in 13m 44s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d9f5e2cb6b3b4b389b94475f85c47a21 : SUCCESS in 1h 24m 04s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/edc61c40db5645d782b94ad1195fa4d5 : SUCCESS in 1h 37m 00s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/9298a457b9074195881c12ea821992d1 : SUCCESS in 57m 15s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5392fbd0db9241cd8ce54469b38ed3b5 : SUCCESS in 1h 13m 44s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b628803015684e94926d61650fdac58e : SUCCESS in 1h 06m 18s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":1},{"id":"d42a2bff130291dceb05bf1c08d2c6c1ba2000ab","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-05-29 12:39:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0c0192a68095c314d881a1fa9400d2aaa8b6effa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 15:21:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d27b61b510546e0b47e5b739077ef0c\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3ddadda1a71a44c7a443a9223f827485 : SUCCESS in 1h 26m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b42127dafa894506a2c60007498b5b97 : SUCCESS in 35m 14s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d82fc1bbaa294f70b6da1c8b1a2732ab : SUCCESS in 1h 27m 25s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9c38afcd4cec47b38605e07967c32666 : SUCCESS in 2h 02m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/345218a04fab4f2690c33e29b3b1f201 : SUCCESS in 32m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb90d698a1a140b594a67f0bbdbc09d1 : SUCCESS in 1h 00m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1ebfe34b3f0400199f260aecb6ef574 : SUCCESS in 10m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6e861e7dcd01490bab73f5bb566dadb1 : SUCCESS in 30m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d6740f395ce44f579fa8f13debb9dfa3 : SUCCESS in 25m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1c3fca6d10de47148b2325cc5d3a2889 : SUCCESS in 37m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f046eac383a343c6b75ab91d05652c60 : SUCCESS in 9m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8410c179c52e4fc997b0fc7ba3608a88 : SUCCESS in 13m 36s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/018fd6d2ea304e1b9254f3f8a6ff9b38 : SUCCESS in 46m 00s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/721db53479284189bda7fce68e749c3a : SUCCESS in 1h 11m 13s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/e1c88ca123114fc6a25c24d837351eb0 : SUCCESS in 1h 19m 21s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f57ab993fe2348d7bb2cd3679f2df5e8 : SUCCESS in 1h 25m 16s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b1b2e87a4602446ea444e5d5bf482eea : SUCCESS in 1h 46m 14s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8e25488601534cddbd557f08da9e209b : SUCCESS in 1h 18m 29s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1074a13863134ecf98df2abff5c43676 : SUCCESS in 32m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d250f15ed4644c768651df9c46e3263c : SUCCESS in 1h 11m 59s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/71ca69842849403b96e3e3c455728a93 : SUCCESS in 2h 35m 53s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":1},{"id":"a7929b6e2907ea21b391431d1fc9a293bbb38a87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 15:24:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ccfa97b38ebdda59650a0f3b5c321fe5bf311f1c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-29 17:43:59.000000000","message":"Patch Set 1:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/81/990481/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"91cb80dbca8112bd9988a27836d495de36278fb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 18:17:56.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a977070854b943e5b2c4e2f3cd51b3c9\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8ba40cafd07b4969892338c27f1062ac : SUCCESS in 1h 34m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1d2fd9736872478683a52ce6bdddf0c5 : SUCCESS in 1h 07m 54s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/efcc5023f5fc43e2898473d3d0196609 : SUCCESS in 1h 26m 28s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/089d658381dd4b938dee263acbdca9a5 : SUCCESS in 1h 50m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1725b82d8ddb418f9b2d439437698e15 : SUCCESS in 8m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ec6ff1f2c28d4436b237acd283d8aa71 : SUCCESS in 23m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/47ec19fbd4104209911488710e5aea81 : SUCCESS in 23m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/463e2add200349f7bd578e218a293ac4 : SUCCESS in 10m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fccbb0994b444d00a9bbd4bff595494e : SUCCESS in 13m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/92b14d458537418a873521a366dd4b97 : FAILURE in 29m 27s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/922c6d9ad0e44e748751be4ec8eaa29a : SUCCESS in 2h 31m 19s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/e30f38b85b824171bef43c9cc16735b9 : SUCCESS in 1h 21m 53s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9e6b6c48e6c34652926637702788f435 : SUCCESS in 51m 50s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e570b18f817c4f6194f3c21d36ba4510 : SUCCESS in 1h 02m 58s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":1},{"id":"a845e3d2d17ff5e114f8f47825efe1be8f104ec1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 19:08:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77ef0fcb3c0c41e8989698e308a6be01\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9c1fd1fe2ee8494a9194d81b814f27d9 : SUCCESS in 44m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/27613fa90da14c179b90aacfc0cb9ab1 : SUCCESS in 25m 43s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/6a9ef87e88b04056a37fe6b9af83a53f : SUCCESS in 32m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"560afad3e4265985ad98a317a337aa98913b1a0c","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-05-29 23:17:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"28dfe6a26de236a6396cd6d2569d67814eae5a2d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 00:31:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a72118b3685043fe990da59beea9e1d1\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5564515f8d0b45f7a01b4cb095c13226 : SUCCESS in 19m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/69b060f84cf54a189efcc54bbaebe612 : SUCCESS in 37m 39s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/67947a7537734dab99f13e59c8403b5f : SUCCESS in 36m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e39c14d2b9843e41c497b2e708d55baf037fa90e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-30 01:01:11.000000000","message":"Patch Set 1:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/81/990481/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"0c48eed91cdb310b36d4ae054b94644a63fb611d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 02:09:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9ecb80a28dbc4a898388fa88a30a2718\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1c517fce9beb416d80382428d02bf600 : SUCCESS in 1h 32m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9b2f8bee89154532b383221ce4e549fa : FAILURE in 1h 11m 33s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f5ec363ab6564e2294763fc77f1b57bb : SUCCESS in 1h 39m 09s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7e43561f4c2547758fdf5ea6a48636ed : SUCCESS in 2h 31m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5cf8af9ac9b54982ae050778e39c082a : SUCCESS in 52m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a621b175ba754ecc9e4cd7385ca74636 : SUCCESS in 45m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67dd8d2f4d4049df810d7a777d8a347e : SUCCESS in 11m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ae57bf968f434665a5fc1189d4e3b762 : SUCCESS in 30m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a581d48149ce4f06a81f169ea87f674f : SUCCESS in 19m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/12bc95d2d87f4395b6af9e787e7c2a32 : SUCCESS in 26m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28946f337814421baf5bbe83e32b5851 : SUCCESS in 10m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e6b16897fa94613a7922ebb9be3ba6f : SUCCESS in 21m 54s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0ddae88b2cd5443f942536fed78136c6 : SUCCESS in 1h 09m 28s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4bdee9b86449420985379f1d8a658546 : SUCCESS in 59m 04s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ba3372d5fc7f4ff0b2ee03d5852d9542 : SUCCESS in 1h 22m 35s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/42294a59381a4d1eacf4be22e44ff776 : SUCCESS in 2h 00m 20s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1371d1b0e1484ed5a3c62cfa163ba254 : SUCCESS in 1h 47m 26s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a8663206ff0949f993afea35ee3f89b0 : SUCCESS in 1h 01m 17s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/67a80c3edca24ae1b2c4c4ae653e0579 : SUCCESS in 58m 35s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8f807c47fce6422b86d799be75cacd71 : SUCCESS in 1h 26m 45s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/1605fd77b7eb4f418cd9b870cd932878 : SUCCESS in 2h 24m 27s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":1},{"id":"27e7168c72b1318311b73ce6f3e3c1a678cf2d36","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-05-30 05:19:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e96baa04861670132543535a4a753168d1680fd9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 07:30:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e538a92054a543f4b3116addac8ea5b1\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/edfaa49123d74b41a7249fd1132918e9 : SUCCESS in 19m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/88d9402bcbd543229bfd8d9627329d6b : SUCCESS in 18m 38s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/bdb1cd9f8238434cad636921b72b2913 : SUCCESS in 34m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aabf8e00748fdc5da8dc75953392b40b0e65add1","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-30 07:39:22.000000000","message":"Patch Set 1:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/81/990481/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"c999cb44967c360411b8e07bcd9f0ddf69755822","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 07:59:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c10ded76249463c9afb0a9e1bab9218\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/16cbe64cf0db407ea0430af78ac74492 : SUCCESS in 1h 27m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f08958cdd10d4cc087f193691096193b : SUCCESS in 49m 56s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7f77db5de0844cffb87a3c0e165ca9e3 : SUCCESS in 1h 25m 15s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/aea4daba53ce4f78a71924264e7a3115 : SUCCESS in 1h 55m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5c8f2acc06bf4958b22d634292fc5585 : SUCCESS in 34m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76741c7409a84e64b34e956ff27c1b95 : SUCCESS in 1h 06m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/463b9d91256543f8b516661b50d33168 : SUCCESS in 10m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5521ec2ceac44dae8780b7e070e098aa : SUCCESS in 28m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/df91d8dfd5f24542ad5c512873adc5b5 : SUCCESS in 24m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bd8a41e2b4354e3aab667450d897735d : SUCCESS in 32m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64af40e3733444058586c1519a75664a : SUCCESS in 13m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a1b165ec1224d9e8a99727f4b4bafde : SUCCESS in 19m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/090299b06e6a47afbbedbfa607855919 : SUCCESS in 1h 26m 53s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/cefa6c2821b0478f96f37ae5e9809497 : SUCCESS in 1h 13m 35s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d104e619ae6b480691869ec16a9f156b : SUCCESS in 58m 42s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b2ab1bcb4b5941ad9d5ebb4dbe0c40d4 : SUCCESS in 1h 53m 25s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5156ff281a7242cfae3bedb9d97ed772 : SUCCESS in 1h 00m 02s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/be464ffcb9b44be19eb6b012c5ac6229 : SUCCESS in 1h 04m 42s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c9d0c58ff185470cab1535fcdf5573e1 : SUCCESS in 36m 37s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/51bcc99e7aea4d649c19a64056cb46cd : SUCCESS in 42m 59s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/18691708081c412e8cc9667d46148050 : SUCCESS in 2h 31m 39s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":1},{"id":"a21bb9b55d0d2f47eeb20f65b817d3e39261b56e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 07:59:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b3f2af1cb665efbe6f515bae57c9eefa1047fd28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 09:52:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e261a62dcfa4773ac122f0817580462\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/cdc4d00b47924bc59a8b2e0384d7c043 : SUCCESS in 1h 09m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/47544ba6a0184ec78dea497879b70be0 : SUCCESS in 49m 40s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e3847d647c15495e9f71de20292ad615 : SUCCESS in 1h 17m 25s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e6cc0047a78d410aa94e6e998a175434 : SUCCESS in 1h 52m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/252c7d208c4e4d3d994951789d728064 : SUCCESS in 11m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4066e56597d448c8bfa162bcd37c7d41 : SUCCESS in 24m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e7048103f0234ca1ad6a3e0d351da849 : SUCCESS in 21m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/921ab42f777b4fd8833fcc60285cf49f : SUCCESS in 6m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b85d065a5034c1cb03981755ef819e8 : SUCCESS in 21m 04s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/55148fb588f94516aaac1d59be2fa345 : SUCCESS in 44m 53s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3da51e970e394107906d42411f5c3e30 : SUCCESS in 1h 12m 51s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ab44e8ec412b4ff3887d10b9de2b14c3 : SUCCESS in 1h 18m 49s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ea625a17291744d589ba79b8090c779e : SUCCESS in 1h 07m 22s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/bb2f9c00759b4da095b94ce46633aea8 : SUCCESS in 58m 50s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":1},{"id":"80c4a20c93b0b6e6894074b0f51abd743a655031","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 09:52:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ec6c0ab07ce2bb102e96b461de81f171ba935585","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 09:54:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/400e13a020304e729ca257942d4f83f5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdc2ebb47fa24e44b836e22b77153b07 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"918a59106c1d82f1110278a6c034e1b5ce1cc5b6","revisions":{"918a59106c1d82f1110278a6c034e1b5ce1cc5b6":{"kind":"REWORK","_number":1,"created":"2026-05-28 14:52:57.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/81/990481/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/81/990481/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/81/990481/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/81/990481/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/81/990481/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/81/990481/1"}}},"commit":{"parents":[{"commit":"e0504685dd6d5aa399f528518267fc48ad8d93c1","subject":"Merge \"ml2: notify mechanism drivers in reverse order during port delete\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e0504685dd6d5aa399f528518267fc48ad8d93c1"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-05-28 14:49:55.000000000","tz":-240},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-05-28 14:49:55.000000000","tz":-240},"subject":"bgp: Do not learn routes from main BGP LRP","message":"bgp: Do not learn routes from main BGP LRP\n\nThe LRP that connects per-chasssis router to\nthe main BGP router should never learn routes because\nall the learned routes need to come from the fabric.\n\nCloses-Bug: #2154459\n\nChange-Id: I7cadbdb738bf95374661798047b25bd9ef1fe640\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/918a59106c1d82f1110278a6c034e1b5ce1cc5b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/918a59106c1d82f1110278a6c034e1b5ce1cc5b6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
