)]}'
{"id":"openstack%2Fneutron~990787","triplet_id":"openstack%2Fneutron~master~I46cb1aec7fcb12c69044373cc76d036ca60cd734","project":"openstack/neutron","branch":"master","topic":"handle_agent_event_instances","attention_set":{},"removed_from_attention_set":{"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2026-06-01 17:50:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I46cb1aec7fcb12c69044373cc76d036ca60cd734","subject":"Allow passing instantiated events to ovn agent","status":"MERGED","created":"2026-05-30 04:31:34.000000000","updated":"2026-06-01 17:52:10.000000000","submitted":"2026-06-01 17:50:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"990787-handle_agent_event_instances","meta_rev_id":"56c3935d1d3c750520e98b0f150ff207825b9458","_number":990787,"virtual_id_number":990787,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-05-30 05:30:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-01 17:50:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-06-01 13:36:32.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2026-06-01 08:48:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-06-01 13:36:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-06-01 13:36:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-30 05:30:45.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-30 06:51:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-01 08:48:43.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2026-06-01 13:36:32.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-06-01 13:36:34.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"}],"messages":[{"id":"fe8b62e89e7502aba1d41b6761fd214ce1d2a396","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-05-30 04:31:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f05625e7fb6d98c897318f1a62fd4fadc23c89b0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-05-30 05:30:45.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2026/Arista_Tempest_Testrefs/changes/87/990787/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"7f3d67d08e695f2257130940037daf791fbcb9b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 06:51:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4247f1f0a2914557b04363a65f3360a0\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a9c46753936b4551a1f5a92ecbc137cc : SUCCESS in 1h 10m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0c5bd6b7ba244da48ee742f9eb295d9b : SUCCESS in 1h 06m 06s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/cdd5b06b9e274c8590eaef3cb14e7d23 : SUCCESS in 2h 06m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e440c52e18f40e787063cd499f353d6 : SUCCESS in 1h 00m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b673af22bc414753b09d39b5b6c69553 : SUCCESS in 9m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9a016513c4e045ba8c704e136a1a7e91 : SUCCESS in 20m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b8eeb5d153ff4988b2456dabfe8c5c06 : SUCCESS in 19m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/532e6cfbcd7d49b1b2864457479ab06b : SUCCESS in 38m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd55412cf055412c93c02c0b54e59ccf : SUCCESS in 11m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/048c5f4359044e5498081428137797a2 : SUCCESS in 19m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3ea66caa77d24b2da06080b60d0d42f5 : SUCCESS in 49m 01s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/9ea95f7f6a144fe8866eda61de8692fb : SUCCESS in 1h 19m 29s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8af73a67d4fc4226b90d81a8535fd16e : SUCCESS in 1h 02m 17s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/dfbbf799777a426b8aee2249316cc9d3 : SUCCESS in 42m 36s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/1862bd8d48394a2fa4bf4a7768d82a4d : SUCCESS in 2h 13m 27s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":1},{"id":"290587c2911f666d3a4a3f42631ce9056c5f4387","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 07:46:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6612f1395801443baa3faa1d1dcdb4f1\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f53a3ec08df8488eb8a9fb643e07b62e : SUCCESS in 50m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4b1cca1c1f7e4b7ab502cbc6d54f6ddd : SUCCESS in 36m 20s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0ff68ebc6fb7448bbba0c988b3280e18 : SUCCESS in 1h 00m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e44b51c87d75d432bb0e83b1df3bf170fa88f296","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-06-01 08:48:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dbd0a249675d8f3e6ee7973ffa9dd92d98e5defa","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-01 13:36:32.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b65f89df16162074707c7deb5291b32c51912ced","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-06-01 13:36:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e78963091b7a7ef965e9889e121c5e3be91a0628","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 13:37:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2847d88b8c6f83c1f6f3c5dcd6c3657186bcc672","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 17:50:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/170952240a1b4a808f2d2e877445f495\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e31de07bbff843aa8c09ddf8b7b8088a : SUCCESS in 1h 35m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f85830a9dea495f9752ee46879c5e54 : SUCCESS in 38m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3f44b26647ad4719aaa7c39b46631ecf : SUCCESS in 1h 43m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77ebe52ec1384c9f976e5e0d63b23f52 : SUCCESS in 10m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d575aa0afb66407cbd67c480d39db08e : SUCCESS in 45m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/46a0220db014494ea795e6cb3de3ec15 : SUCCESS in 17m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57fc2f8cb0b0429c8cdd0ef6a68427fc : SUCCESS in 12m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b85715020954e8191449bb41bebf79c : SUCCESS in 9m 28s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4f1c26baf22f402592b865a6a106118c : SUCCESS in 1h 02m 00s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/f2da1d24c7b44f39b22e8d2c19e4910f : SUCCESS in 1h 24m 06s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d53917e88027480ea136cb8aeb7f91b4 : SUCCESS in 54m 27s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":1},{"id":"3faab59b421ab21f5a9a835a36c9a4c9ca51ce73","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 17:50:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"56c3935d1d3c750520e98b0f150ff207825b9458","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 17:52:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0e9dfcb26a14b67835f891a91832c26\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48f0214fad9e4da2b8645ababcc9dc34 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9557e8e56aefa7f0cdc1c879e0c13a062f78ec13","revisions":{"9557e8e56aefa7f0cdc1c879e0c13a062f78ec13":{"kind":"REWORK","_number":1,"created":"2026-05-30 04:31:34.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/87/990787/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/87/990787/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/990787/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/990787/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/990787/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/87/990787/1"}}},"commit":{"parents":[{"commit":"6229c4bf9bae86b664cc150e2f7864cb6e813733","subject":"Merge \"doc: Add admin guide for Neutron server-side processes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6229c4bf9bae86b664cc150e2f7864cb6e813733"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2026-05-30 04:28:26.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2026-05-30 04:28:26.000000000","tz":-300},"subject":"Allow passing instantiated events to ovn agent","message":"Allow passing instantiated events to ovn agent\n\nIt is useful for events to be able to define their own arguments\nso, this allows them to eeither be passed as classes as they were\nor instantiated by the extensions themselves.\n\nSigned-off-by: Terry Wilson \u003ctwilson@redhat.com\u003e\nChange-Id: I46cb1aec7fcb12c69044373cc76d036ca60cd734\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9557e8e56aefa7f0cdc1c879e0c13a062f78ec13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9557e8e56aefa7f0cdc1c879e0c13a062f78ec13"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
