)]}'
{"id":"openstack%2Fneutron~990865","triplet_id":"openstack%2Fneutron~master~Iac26621dc81a3b03105b67d53ea4b9e28cd9e178","project":"openstack/neutron","branch":"master","topic":"query-first","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-06-01 14:32:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iac26621dc81a3b03105b67d53ea4b9e28cd9e178","subject":"Use ``objects_exist(id\u003did)`` to check resources","status":"MERGED","created":"2026-06-01 01:35:39.000000000","updated":"2026-06-01 14:34:54.000000000","submitted":"2026-06-01 14:32:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"990865-query-first","meta_rev_id":"891fbf3800bb6d8639a1451618240450660520f5","_number":990865,"virtual_id_number":990865,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-01 14:32:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2026-06-01 08:59:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2026-06-01 12:14:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2026-06-01 12:14:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-01 02:51:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-01 05:01:14.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-01 05:05:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-01 08:59:58.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2026-06-01 12:14:54.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"bf4336a683f865e8e8e8ccc88f35fea5d504e9df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-01 01:35:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bf1c819cc14b358ac47d392507ef00268f5ac25f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-01 02:50:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6913389cb9379571db9e0fca64ee4fb49051a764","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 02:51:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08a24350601541e582657b2c7551137b\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2536ef47c51a449f847bc5e159406366 : FAILURE in 23m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c3e051baaa764dac976244008d8fef6c : FAILURE in 52m 04s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/05f19fab713d4e6ca2dcc293f755d449 : FAILURE in 1h 04m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b9b73e754ba71ae83cc339434fea3063e670feb0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-01 02:58:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"92a38b332233989199a1ebd361f996661769074d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-01 02:59:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a73a27c9425827f8eda264d33d74a533d185487e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-01 05:01:14.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/65/990865/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"c5013917d2760b757cd6c35ae29b412116d41da9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 05:05:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f44edf4249c64cefa91ca7d23553789d\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a320a0862be64a9da20719cc0b40514d : SUCCESS in 1h 11m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/927313e6a3244554995c8d939676347b : SUCCESS in 53m 05s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/735cd3c2a1764cde8de73cf67cd651e3 : SUCCESS in 1h 35m 24s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1845ab441d0d46fc95b2e533f4b87726 : SUCCESS in 1h 54m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/db60aeea6bb5412d9cd6c086a18de85e : SUCCESS in 32m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd0237a49865494a95c39afa63c1857b : SUCCESS in 47m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f0ea537d73c4ece9d38ff2ee79ad383 : SUCCESS in 6m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/496c936982734ce49d1eebdd4bd11a75 : SUCCESS in 18m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3534288b65d24b608033e41a06e7db0c : SUCCESS in 20m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/02b9b606848d4e09897d88ba47c153ed : SUCCESS in 25m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d43816ba81f47ee9af504e43f19be9d : SUCCESS in 10m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f59a14e1d6824ae5b9f604f05dede1bf : SUCCESS in 20m 06s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6525e3d15cca4d18ba99c60ec0a4a9fb : SUCCESS in 1h 07m 02s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3cf48bc347864c5d978967767e0d8374 : SUCCESS in 1h 06m 44s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/09bbdac13c334a7783ec6e8ef8f76cad : SUCCESS in 1h 16m 11s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/83cfbadbbbf948bdbffc34f282232a61 : SUCCESS in 2h 00m 58s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/af8313198c51497ea9666db4252eba89 : SUCCESS in 1h 50m 07s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/16c16e01cc6d42709ec0d7c01c03d787 : SUCCESS in 1h 04m 02s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/fa87c9496bb94db38a6c77ba81e040fa : SUCCESS in 1h 19m 00s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2976d60ab5474b2b9f471123e6091dd6 : SUCCESS in 36m 34s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/b0f0416779274c9bb9ac62836d5a4c7c : FAILURE in 39m 14s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":2},{"id":"af8c8c0b5318d826a971d4dd2703ff51cf2484c6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 06:11:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/529d013dbdc2481b878a1aa919a5717a\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/72aeba85da1c4e44ab8eb7266f47d59f : SUCCESS in 20m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9fdde1ebead94c52b11f5b244f7007d0 : SUCCESS in 19m 39s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/55b861bda4b549099e29b2f5056590b5 : SUCCESS in 42m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"890a2e20555a9130c66901278f162aa398516897","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-01 06:12:40.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/65/990865/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"3345655776460d530a174858fd052d79a3259594","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-06-01 08:59:58.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6490b19078e6b3ed3c0b7a68bc3bb539f0a4c916","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-06-01 12:14:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9b6fc48b767a079eb3c4dda9781041575ca0228b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 12:16:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f1c3a6b10d242b001e2fc0f9f08b4686a48f15c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 14:32:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ddd02b5d1ae14c108e11d1064a96d94d\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2c5af9b6364d4d1890f53775c52f3d85 : SUCCESS in 1h 42m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7a0278025b62411fabf62382442d58c5 : SUCCESS in 1h 14m 36s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ac4f31f383fa4ae2847e223d921beace : SUCCESS in 1h 33m 18s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b8bd11d372f04f66a506aaa658afaa0a : SUCCESS in 1h 57m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5afeea77760d4b28ac0f0a522b5da6c1 : SUCCESS in 10m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bc92587069734995a592885861cf14ef : SUCCESS in 31m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0eefc5a9562f4f328130b02ee8b9d22c : SUCCESS in 26m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c6a5790ab0e46729a26ac33060baa32 : SUCCESS in 12m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58a0a28e6ddf420184f5154eda666170 : SUCCESS in 21m 23s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b79f4490d205478881385f90126270f2 : SUCCESS in 1h 34m 45s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1e106dd511e748268d38372b1e3b20e5 : SUCCESS in 2h 03m 31s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/a020a412d8994c42b620eab4516f3268 : SUCCESS in 1h 24m 36s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1093994d155246d8800e0d846061fb05 : SUCCESS in 1h 13m 27s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5e4a1a1632c2498ba1d8d1748528901f : SUCCESS in 1h 17m 42s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":2},{"id":"b712ea5c773ab3c04968b0ffa8ebfeea426577ed","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 14:32:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"891fbf3800bb6d8639a1451618240450660520f5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 14:34:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b094c731881415a8c988255bca3de3e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18ac5f83a2ad4132a8eb0a8d9a96a701 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f8b833749e5d238e04542a7109ed1b4dd0a75748","revisions":{"952f06348b3c5ec0ed67e6b90e7189fee500db99":{"kind":"REWORK","_number":1,"created":"2026-06-01 01:35:39.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/65/990865/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/990865/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/990865/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/990865/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/990865/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/990865/1"}}},"commit":{"parents":[{"commit":"3a97189f2789683dd6e5ac3947716fa330811e81","subject":"Merge \"bgp: Do not learn routes from main BGP LRP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3a97189f2789683dd6e5ac3947716fa330811e81"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-06-01 01:34:05.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-06-01 01:34:05.000000000","tz":-240},"subject":"Use ``query.first() is not None`` to check resources","message":"Use ``query.first() is not None`` to check resources\n\nInstead of using ``query.count()``, it is more efficient to use\n``query.first() is not None`` as it generates a simple SELECT\n... LIMIT 1 and returns the row (or None).\n\nRelated-Bug: #2150540\nChange-Id: Iac26621dc81a3b03105b67d53ea4b9e28cd9e178\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/952f06348b3c5ec0ed67e6b90e7189fee500db99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/952f06348b3c5ec0ed67e6b90e7189fee500db99"}]},"branch":"refs/heads/master"},"f8b833749e5d238e04542a7109ed1b4dd0a75748":{"kind":"REWORK","_number":2,"created":"2026-06-01 02:58:21.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/65/990865/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/990865/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/990865/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/990865/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/990865/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/990865/2"}}},"commit":{"parents":[{"commit":"3a97189f2789683dd6e5ac3947716fa330811e81","subject":"Merge \"bgp: Do not learn routes from main BGP LRP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3a97189f2789683dd6e5ac3947716fa330811e81"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-06-01 01:34:05.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-06-01 02:55:19.000000000","tz":-240},"subject":"Use ``objects_exist(id\u003did)`` to check resources","message":"Use ``objects_exist(id\u003did)`` to check resources\n\nInstead of using ``query.count()``, it is more efficient\nto use ``objects_exist(id\u003did)`` as it generates a simple\nSELECT 1 ... LIMIT 1.\n\nRelated-Bug: #2150540\nChange-Id: Iac26621dc81a3b03105b67d53ea4b9e28cd9e178\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f8b833749e5d238e04542a7109ed1b4dd0a75748"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f8b833749e5d238e04542a7109ed1b4dd0a75748"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
