)]}'
{"id":"openstack%2Fneutron~990919","triplet_id":"openstack%2Fneutron~unmaintained%2Fzed~I66676e60c584f5073de01aafa600f1b7de369d69","project":"openstack/neutron","branch":"unmaintained/zed","topic":"pkg_resources-gate-fix-neutron-um-unmaintained/2023.1-unmaintained/zed","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2026-06-10 14:03:19.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I66676e60c584f5073de01aafa600f1b7de369d69","subject":"[CI][stable-only] Workaround for missing pkg_resources","status":"ABANDONED","created":"2026-06-01 13:34:40.000000000","updated":"2026-06-10 14:03:19.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":979637,"cherry_pick_of_patch_set":1,"meta_rev_id":"94507a6790385e2218fb0db017e295a709a11af1","_number":990919,"virtual_id_number":990919,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-06-01 14:44:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-01 13:34:40.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2026-06-01 14:44:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-04 09:19:47.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"}],"messages":[{"id":"9095a65173aaf26bd037137ef2394b8db686ecae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-06-01 13:34:40.000000000","message":"Patch Set 1: Cherry Picked from branch unmaintained/2023.1.","accounts_in_message":[],"_revision_number":1},{"id":"551785fa4e9c558627306c82616db24374989824","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 14:44:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d5c35178466f4d7b8515f49584ce7007\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3ec58b4d080346eaafdace9e31dac96f : SUCCESS in 36m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaf422f6ebf54a1e8ede0f3f1590211d : SUCCESS in 14m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fbec82ad24d74a4faa3c8eeffd858d20 : SUCCESS in 17m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d82a61bf362476e8eb862f18111f067 : SUCCESS in 27m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e95292c2fb28430b9f62b8a0c2f51fd4 : SUCCESS in 29m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9eb90dca47ac4d5bad49737e8e6fcd19 : SUCCESS in 5m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb86f98f698f47ada3757f5ae3540149 : SUCCESS in 9m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2469ef2bb6e34018ba62768764a748ec : FAILURE in 40m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9049fb7505fc4d929967304e6dcfe190 : FAILURE in 1h 07m 02s","accounts_in_message":[],"_revision_number":1},{"id":"94507a6790385e2218fb0db017e295a709a11af1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-06-10 14:03:19.000000000","message":"Abandoned\n\nFixing in https://review.opendev.org/c/openstack/neutron/+/991542","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f7624dddffe7b67a1a31d38760cea807b2409d6c","revisions":{"f7624dddffe7b67a1a31d38760cea807b2409d6c":{"kind":"REWORK","_number":1,"created":"2026-06-01 13:34:40.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/19/990919/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/19/990919/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/19/990919/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/19/990919/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/19/990919/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/19/990919/1"}}},"commit":{"parents":[{"commit":"349ccdce0cce39b0f3b280a16e8eecc540b5c774","subject":"[CI][stable-only] Remove neutron-ovn-tempest-with-uwsgi-loki","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/349ccdce0cce39b0f3b280a16e8eecc540b5c774"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-02-25 08:59:34.000000000","tz":60},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2026-06-01 13:34:40.000000000","tz":0},"subject":"[CI][stable-only] Workaround for missing pkg_resources","message":"[CI][stable-only] Workaround for missing pkg_resources\n\nWith recent virtualenv release (that bundles setuptools in it) tox jobs\nstarted to fail with \u0027missing pkg_resources module\u0027 errors.\nSince this is an old branch, where we have only python version \u003c\u003d 3.11\nin gate jobs, it is enough to cap virtualenv to fix the\n\nChange-Id: I66676e60c584f5073de01aafa600f1b7de369d69\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n(cherry picked from commit ffda937ab19632512ae8e69fc68c6b98a1e3f057)\n(cherry picked from commit 8e5c82e785171202c02e50248b361967416e0c0c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f7624dddffe7b67a1a31d38760cea807b2409d6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f7624dddffe7b67a1a31d38760cea807b2409d6c"}]},"branch":"refs/heads/unmaintained/zed","conflicts":{"ours":"349ccdce0cce39b0f3b280a16e8eecc540b5c774","theirs":"8e5c82e785171202c02e50248b361967416e0c0c","contains_conflicts":false}}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
