)]}'
{"id":"openstack%2Fneutron~990928","triplet_id":"openstack%2Fneutron~master~Ia9754f78d49f6ef925d82ed373eb2b00817dbad3","project":"openstack/neutron","branch":"master","topic":"bug/2103521","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-06-03 15:51:38.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-06-03 13:29:54.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"Ia9754f78d49f6ef925d82ed373eb2b00817dbad3","subject":"Add ``update_l3-router`` policy","status":"MERGED","created":"2026-06-01 13:52:35.000000000","updated":"2026-06-03 15:53:58.000000000","submitted":"2026-06-03 15:51:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"990928-bug/2103521","meta_rev_id":"3742eb7454dcce64ef2406e187600dcfa454a263","_number":990928,"virtual_id_number":990928,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-06-02 12:17:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-03 15:51:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2026-06-03 09:34:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2026-06-03 13:29:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2026-06-03 13:29:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-06-03 09:27:46.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-01 13:53:11.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2026-06-01 13:53:11.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-06-01 15:53:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-01 16:13:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-01 17:35:27.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-03 09:34:56.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2026-06-03 13:29:54.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"fccb0a6c016b470386e64facbec3b174c02840e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-01 13:52:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8912a36f481aa562a0e8874f563686f126cad180","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 15:53:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7247760d03334f37a23f0069ad9288f6\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2b32288c51c1460baef21b8f97e39b6f : SUCCESS in 55m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f6d8321fc3a24b038f9b2a407ae03748 : SUCCESS in 19m 56s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b1551fce00054b03aae1ca3471ac3ab5 : SUCCESS in 29m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1c0f4db95c2355fb768309053b22e0b9e2d99a03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 16:13:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/567e17d4bed2407f9ca569770a353a14\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c74e8a8b173a4464a11309a2737be04e : SUCCESS in 1h 24m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fd6b003a9562483fa479cd6b08dd9f88 : SUCCESS in 1h 09m 39s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c4837349f0f34796962f35c2ee8e8343 : SUCCESS in 1h 22m 56s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9df18d7a974c4a32ba920724e5951e1e : FAILURE in 1h 55m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4407d69a27024126b4428357e6c53299 : SUCCESS in 31m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/41d8aafa725342f886f76b9cf4f9db55 : SUCCESS in 1h 05m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfd30724cc9b4118bf47da1abd2af1a8 : SUCCESS in 6m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a3879830bc8f44aea86807fc7b70348b : SUCCESS in 27m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2cf6194961d146ecaea57490a52e2bb6 : SUCCESS in 30m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e1d62c9105c14147bc8ae238b5e94216 : SUCCESS in 35m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8bb8686216b464bbce11e64deab54c8 : SUCCESS in 9m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/809e503634ca4b6a8ad432c876259eeb : SUCCESS in 12m 16s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ebefda970a8b45fba19b3869dddf18ca : SUCCESS in 1h 42m 28s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/69dff64787ea484a8ca49d81cff37748 : SUCCESS in 1h 14m 37s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/bcecf0ed7bdd48b5a16ce43361841b04 : SUCCESS in 1h 19m 11s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0526264e882c4bb880eca994507f8146 : SUCCESS in 1h 31m 00s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c88fbf5051304566938319de097bb9da : SUCCESS in 1h 08m 17s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d04b3d92020148abad8a80e578d9024d : SUCCESS in 1h 05m 43s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a6422ab6b6ec452f8ad655140f8e964b : SUCCESS in 1h 22m 00s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/71614b10f48644148568a421f3418225 : SUCCESS in 1h 15m 38s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/f2609c2ec45940d9a6e44468dd9244e6 : SUCCESS in 2h 10m 30s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":1},{"id":"9742163bd5cf86310421ceafca80a0c16014b75a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-01 17:35:27.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/28/990928/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"c50625ada513703519e9b376f5a28a04c4c509f9","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-02 05:57:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"687bb6e3a688bfd9c5a8c62b8272e247ce4e94cb","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-02 07:59:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d1f7bcb33bfca4287f10f4489dc30f07010e1955","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-02 08:00:54.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1, Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"ea270be22232d7ada4bbdf57a1bd078d328f7a5d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 09:03:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d5decb084354967ad7b9833af89b734\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ebdbd583629f4700856632263a4c304f : SUCCESS in 19m 54s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/eabc4ab093604cd5b3718de6457f4af8 : SUCCESS in 24m 30s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f98d20bc41754426b4783ad6787245ef : SUCCESS in 25m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"99dd7b7681d3fd0521982f976c0ade42b2d8c238","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 10:38:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ded0744494874f6886005923829a690c\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/47dd54a122ab44fdbee62bd1212dc157 : SUCCESS in 1h 36m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8befb6bb6ad74210a3ca55028fe8ab46 : SUCCESS in 34m 16s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0a55d9d8ea5b4d848021d6187229438f : SUCCESS in 1h 26m 44s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bc1702d4749f4a0896f92c9b2d399faa : SUCCESS in 2h 21m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/04b4090f65784095a47baec96b15f68b : SUCCESS in 32m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9aea2927924c43a58e9c8ae23cfd63b0 : SUCCESS in 1h 01m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a740182ae304a1c95caa825a9c0d68b : SUCCESS in 12m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eceb302b7d144659bf84d375eed35833 : SUCCESS in 27m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cf044d27348b40a190db7b9f475c7aff : SUCCESS in 28m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f999b4d063454ef59b0981b173c596fc : SUCCESS in 28m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/925797268118411ca0ab79234bfd3329 : SUCCESS in 12m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7348bf1392604ca18facb6f6f57d2935 : SUCCESS in 22m 01s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f59bb528c2de4929b14b271b78c465c7 : SUCCESS in 1h 22m 14s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e6e9a0264cfa48f9a93a96fa5440b03f : SUCCESS in 1h 20m 52s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/b7f3875819e64b34a11bde0f3558922f : SUCCESS in 1h 14m 53s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b118803448534c718635a3751fbbc68c : SUCCESS in 1h 57m 35s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c9967d24cd2b41d28015618375390d69 : SUCCESS in 1h 48m 25s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8051294f6d18495f991532431c47bde4 : SUCCESS in 1h 07m 09s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/75e0a5e136f240a6ba90cb741c57ad65 : SUCCESS in 1h 33m 27s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/67b8f59430b5406b8cc548057fdeed9c : SUCCESS in 1h 25m 10s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/30d2afac61cc42708a6d07ae44069653 : SUCCESS in 2h 30m 31s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":2},{"id":"2f2098d7eb24074f119c87fd89d0da357540a63d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-02 12:17:17.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/28/990928/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"322022c5235c274907799666019927773f473803","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-03 09:27:46.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"0e642b466be614bfe0a4d5f32ad0fb65d319ec16","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-06-03 09:34:56.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4acf5be1a7b89c9814e3ea4ad9ef82fffa39464c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-03 13:29:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ef67ad5ebeb0f1b1ad808e259726e2b56af85e64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 13:30:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d4d33d502dfda93a975a8900c45bc528a415d7e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 15:51:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57fffcd210924f4b9ff5c7991059229b\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d1798e0d1ec74daf9af8412615b5e245 : SUCCESS in 1h 29m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a32ea431ed1d4e559ec5c24f0b5997ca : SUCCESS in 1h 05m 20s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/18358b285b284ead9209ee5620d9e8b0 : SUCCESS in 1h 23m 19s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2aaf88223ce74d6f8d3eb30169db050d : SUCCESS in 1h 57m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/285b4f4b6163482abc8dfaabfa12ebba : SUCCESS in 11m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f245624c80944eb9990562e000c49789 : SUCCESS in 29m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/18489a28277446e9bb6a65876571761d : SUCCESS in 28m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c89fcaeb5e764f64a355c84de73e248a : SUCCESS in 13m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0341320c790a4babb94fa8d999b5f794 : SUCCESS in 20m 21s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/895c89f972e44945b46ea2f494a33178 : SUCCESS in 1h 08m 14s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/62e1ef0f7a9a4b48a98f14c28af5270b : SUCCESS in 1h 59m 09s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/53e1b72c830243ffa1ab817eba5af094 : SUCCESS in 1h 21m 00s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c02976f0534f4b99808c7d39177fac16 : SUCCESS in 59m 12s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/04629e9493bc45eabd5bd02aa04b28c2 : SUCCESS in 1h 22m 19s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":2},{"id":"183ff60a564bb66ff03eb5ccd1c9fe74c0fab09d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 15:51:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"3742eb7454dcce64ef2406e187600dcfa454a263","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 15:53:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f874945cd6ad46c7bab48d7a4a3fd14f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fee692a9b16a45cdaf91873cf84173da : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5795318950b340636af5251e223e2bd3b3f281be","revisions":{"baaa69696d61fdd1bfaaf1f963bc1d836fb1300e":{"kind":"REWORK","_number":1,"created":"2026-06-01 13:52:35.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/28/990928/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/28/990928/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/990928/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/990928/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/990928/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/28/990928/1"}}},"commit":{"parents":[{"commit":"80abeb9ec4668d7cf64f32fe07dd3ffaac9cf2df","subject":"Merge \"Use BaseChecks from neutron-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/80abeb9ec4668d7cf64f32fe07dd3ffaac9cf2df"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-06-01 13:51:25.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-06-01 13:51:25.000000000","tz":120},"subject":"Add ``update_l3-router`` policy","message":"Add ``update_l3-router`` policy\n\nThis policy was missing from the ``agent`` policies.\n\nRelated-Bug: #2103521\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: Ia9754f78d49f6ef925d82ed373eb2b00817dbad3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/baaa69696d61fdd1bfaaf1f963bc1d836fb1300e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/baaa69696d61fdd1bfaaf1f963bc1d836fb1300e"}]},"branch":"refs/heads/master"},"5795318950b340636af5251e223e2bd3b3f281be":{"kind":"REWORK","_number":2,"created":"2026-06-02 08:00:54.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/28/990928/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/28/990928/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/990928/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/990928/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/990928/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/28/990928/2"}}},"commit":{"parents":[{"commit":"3fa43b0077c25e19613741d29679cae197365d4c","subject":"Merge \"Fix race condition in worker lifecycle probe\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3fa43b0077c25e19613741d29679cae197365d4c"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-06-01 13:51:25.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-06-02 08:00:23.000000000","tz":120},"subject":"Add ``update_l3-router`` policy","message":"Add ``update_l3-router`` policy\n\nThis policy was missing from the ``agent`` policies.\n\nRelated-Bug: #2103521\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: Ia9754f78d49f6ef925d82ed373eb2b00817dbad3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5795318950b340636af5251e223e2bd3b3f281be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5795318950b340636af5251e223e2bd3b3f281be"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
