)]}'
{"id":"openstack%2Fneutron~990992","triplet_id":"openstack%2Fneutron~master~I313132bb60b69d083a8f1f5036f568d0be4f8ead","project":"openstack/neutron","branch":"master","topic":"handle_agent_event_instances","attention_set":{},"removed_from_attention_set":{"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2026-06-02 09:19:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I313132bb60b69d083a8f1f5036f568d0be4f8ead","subject":"Add docs and typing to agent extension events","status":"MERGED","created":"2026-06-01 21:05:43.000000000","updated":"2026-06-02 09:22:17.000000000","submitted":"2026-06-02 09:19:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"990992-handle_agent_event_instances","meta_rev_id":"4ecc648e6cf898be5e092b90ddc69160e25068d7","_number":990992,"virtual_id_number":990992,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-06-02 00:45:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-02 09:19:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-06-02 06:40:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2026-06-02 05:11:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-06-02 06:40:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-01 22:36:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-01 23:45:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-02 00:45:08.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-02 05:11:46.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2026-06-02 06:40:01.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"f80ecd31c6482466b62a40bc9e74ca2db30bc0ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-06-01 21:05:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e4bfff42cf9807d42bd35ece7f1c9a32793804b5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 22:36:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4955d61016c24946a7930f51fe856f46\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fc0cc93d67024c3dafb137b9d3344e38 : SUCCESS in 43m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2e63f4a0a7b34bc3be88172409dec296 : SUCCESS in 30m 18s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/13e3f295d40042d2a5bc0dea1da4a80d : SUCCESS in 24m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"96ade991b8439af19684908e28b279aa5fce4b40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 23:45:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e458785a65e0435fbf1194608a104d9d\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c7387066819243af98d4bce950964ae4 : SUCCESS in 1h 26m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6f7639997de642d29c8527b8953d22d5 : SUCCESS in 1h 06m 14s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7b9968e0fd654300af61bbea67648499 : SUCCESS in 1h 39m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f56f9c28a3f14b8590b803b089b02386 : SUCCESS in 50m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a039149d3c6a41938f17cfa77a90a2d0 : SUCCESS in 6m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a58f95ec725c493a8c6ed872d651ee33 : SUCCESS in 24m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/385b53e821de42f39090ae4b08be7235 : SUCCESS in 24m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c25e2c66513f495896693fcabda23d52 : SUCCESS in 35m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56739c4d4c1b4efdad7724f8b91f9f01 : SUCCESS in 12m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1c999347f2045b7985b1b8bab37edb2 : SUCCESS in 20m 30s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4e8768ae43454b0681e886a7387d7a14 : SUCCESS in 1h 24m 32s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/f75bc185aba344579ea0912d48be56fe : SUCCESS in 1h 15m 09s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e7dbc4a5912d453fbc616dc22147f883 : SUCCESS in 1h 17m 11s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8f4f266ce82446ed8847afaa596b7f0e : SUCCESS in 47m 48s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/de03593f849b4519bcb56b674fb1458f : SUCCESS in 2h 32m 27s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":1},{"id":"2d26704ab0d57d6500b15c6a684a56b16eaf46ad","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-02 00:45:08.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/92/990992/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"392070d4f7e9e4fe6fad8e70c1a4116ea052b9f2","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-06-02 05:11:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"05cee5170a3580862aaee97f87f827cf703150d5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-02 06:40:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"acca70477d762ea3e0640d911b0c3f60d512b65c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-02 06:40:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"975e763fa2500c1b387cb317740c18cd201683fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 06:40:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e2591aa54dc6ef3b8c37f0134eedaa9d3cb8406b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 09:19:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b66bd4de908244ca9131cfc2a8fddba5\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9f66c65b0fcb4aec8622bea449d1c1b0 : SUCCESS in 1h 02m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ce24c7a133b4be3be81eeb07dcc09a3 : SUCCESS in 1h 04m 18s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/eaaf0cf0e9ca4b138f2ad684c5a779ae : SUCCESS in 1h 43m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f54476d3bd284640acbc4a2df6df1564 : SUCCESS in 10m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/64515fb19c8d4b94a16d6e0675482a95 : SUCCESS in 21m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8eb46f4a6dfb4c8e8dc843c4132ca2f0 : SUCCESS in 15m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e961a6e009d443d1aeb9dfc025e1ec2d : SUCCESS in 10m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f3340985eaa41c68d5e71f5c18609a7 : SUCCESS in 20m 28s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e425359f01de473e8b02b4b2e7e33829 : SUCCESS in 1h 23m 25s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/1ab3147de3ad4f21ba5abe624772ae09 : SUCCESS in 1h 12m 29s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9b3fe51eed904e3f9ff7d478c776297d : SUCCESS in 48m 47s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":1},{"id":"77eca3184db20a628c837e773e6e354888b23314","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 09:19:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4ecc648e6cf898be5e092b90ddc69160e25068d7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 09:22:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/238245d84b8641ca975160196481da43\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b4bd51610e646e187bbb0e714e7b73b : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3d421642447d791a4b097af1d5460031900bbb07","revisions":{"3d421642447d791a4b097af1d5460031900bbb07":{"kind":"REWORK","_number":1,"created":"2026-06-01 21:05:43.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/92/990992/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/92/990992/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/990992/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/990992/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/990992/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/92/990992/1"}}},"commit":{"parents":[{"commit":"9557e8e56aefa7f0cdc1c879e0c13a062f78ec13","subject":"Allow passing instantiated events to ovn agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9557e8e56aefa7f0cdc1c879e0c13a062f78ec13"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2026-06-01 21:03:45.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2026-06-01 21:05:20.000000000","tz":-300},"subject":"Add docs and typing to agent extension events","message":"Add docs and typing to agent extension events\n\nThis adds some type hints and documentation to the OVNAgentExtension\n*_idl_events properties.\n\nChange-Id: I313132bb60b69d083a8f1f5036f568d0be4f8ead\nSigned-off-by: Terry Wilson \u003ctwilson@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3d421642447d791a4b097af1d5460031900bbb07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3d421642447d791a4b097af1d5460031900bbb07"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
