)]}'
{"id":"openstack%2Fneutron~991002","triplet_id":"openstack%2Fneutron~master~I1e68bc22b8c2d334039326b9b3aff44296abcb2d","project":"openstack/neutron","branch":"master","topic":"check-router","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-06-16 11:37:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1e68bc22b8c2d334039326b9b3aff44296abcb2d","subject":"Fix issues in _check_router_interface_not_in_use()","status":"MERGED","created":"2026-06-02 02:43:48.000000000","updated":"2026-06-16 11:38:37.000000000","submitted":"2026-06-16 11:37:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"991002-check-router","meta_rev_id":"2b78220fe33a40c76fe647d342dee93922872864","_number":991002,"virtual_id_number":991002,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-06-02 05:43:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-16 11:37:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-06-16 08:32:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2026-06-04 09:24:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-06-16 08:32:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-02 04:55:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-02 05:43:35.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-04 09:24:47.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-06-16 08:32:13.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"ced3c8cea0250ac33dbdba59e37cf85d25fb60b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-02 02:43:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee09f071b383d1a1c208c42e82feef5c8b9aa522","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 04:55:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0eb060d5e10d4e98aa3b71fff87064b4\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4f188487d9a348ef84eb75c3b2880112 : FAILURE in 1h 05m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/505b557b84d247adabccd9a94f3be1d4 : SUCCESS in 1h 08m 24s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7e26ade0b7cb4e799e1f6f90a1ceb470 : SUCCESS in 1h 20m 47s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/854c4723dc8f430bb7832fa44c805bc6 : SUCCESS in 1h 56m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/038f4d258bb442e09ca6a5cb95d003c7 : SUCCESS in 32m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf29755cfcb24f8089f545980fb3d8b3 : SUCCESS in 39m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04d0bfdcd4d5441492bf032cca6f7e0d : SUCCESS in 6m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f1572647c4214fc28733d7ae0700ff3d : SUCCESS in 22m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d7380ba3961e4b8bac360a1d81b1c05d : SUCCESS in 19m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/47f04f1d6ebf4f86a24bf33e153c8771 : SUCCESS in 28m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb9a5ae8a774416ba099ae2c73c61c8a : SUCCESS in 10m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd2a8687c7814dd9a96b46d8fc0d5b86 : SUCCESS in 15m 13s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ae82b42a13a546ecb21d0a5496507434 : SUCCESS in 1h 20m 29s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/74a7f536fd134a0b866f6998d1b59158 : SUCCESS in 1h 10m 39s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/9211f77cead24f96adb96565988a52df : SUCCESS in 51m 06s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e9ea3abe828c41079d4a9ac6d2f4687b : SUCCESS in 1h 53m 23s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6391ad3e078a4d50905da8c43f1aaabf : SUCCESS in 1h 24m 03s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a7438920d566486a927d2d6e143f9d9f : SUCCESS in 58m 29s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8d2a5c89ba8a4a66b65e4c18679ea573 : SUCCESS in 47m 43s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/deb0124051f546ff95841c3251706946 : SUCCESS in 1h 45m 27s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/683282e7464d43afb51cc0d168258019 : SUCCESS in 1h 17m 39s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":1},{"id":"80d870c70c87f9e56029f757850e40d70e33639b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-02 05:43:35.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/02/991002/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"51724b7da4ef030c5f44ca6028372524a1824827","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-02 05:48:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"048a6b719eb2ed20fd7868fe8aa1aa3737e09519","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 05:49:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de52fdb983dc4b9889558a3cdceef081\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ca8de837cfc64a9ca187c47206f76aa9 : SUCCESS in 48m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/dd40f55246aa47899da5f8e1f61b48f8 : TIMED_OUT in 1h 21m 07s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/edec37c141654ecf8308f9fe1366cac3 : SUCCESS in 51m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a8fef200a6eec4a7f625b83a407cf9dd109d4e97","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-02 07:53:41.000000000","message":"Patch Set 1:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/02/991002/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"b901cda24ce226b7051e738100f2f07ba11a04f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 08:17:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d11feb42db234562b82fc6feddde03fe\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f9015f712bc5451da02a630410d2755f : SUCCESS in 1h 06m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7665d6a60c96445dacc4d8c9fd7bdb99 : SUCCESS in 36m 54s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/269bc92a728d4eb681d7190e8f8bc5d7 : SUCCESS in 2h 09m 08s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a0b9b676629642ecb9abf1dc23254126 : SUCCESS in 2h 27m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/49ca296f2dc44040a79a4af583430d6c : SUCCESS in 33m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/640a0d7b069447569a9ee10e5f310cb5 : SUCCESS in 44m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f0d855342bc44c79db4e38f99027147 : SUCCESS in 6m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9e2179f1627c4ff58a5fa0be5a049f63 : SUCCESS in 19m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a37fac2eb32b46ed9fd684cddefae4a1 : SUCCESS in 27m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9fc853955d984e458f4f653b2851f964 : SUCCESS in 34m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b17196f458b7412fbfc69a3768ddd156 : SUCCESS in 13m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a77a66f4e3d458cb2973678ec0aedc3 : SUCCESS in 10m 41s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4200d057acbb4aa3b1fd857bcbb24b27 : SUCCESS in 1h 29m 06s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/20f7c0938c474ff7a0c62fba1c3bf331 : SUCCESS in 1h 09m 01s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/8bdac647a0ca4cc684dd64671e6ab454 : SUCCESS in 1h 04m 27s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/31f6b30ed2844c0788159d95ca357071 : FAILURE in 1h 10m 42s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/61909caa1fba4937b3df334799bbfe9a : SUCCESS in 1h 17m 08s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e3d3c05cf08746799190d56446458a66 : SUCCESS in 1h 04m 39s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/bf49431344a149b9868ba4b6b7e8155b : SUCCESS in 35m 03s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9a4b07b20f16493fb8afe19e0cafdc42 : SUCCESS in 1h 33m 50s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/69b313e12eb3476a80bda620113b4f85 : POST_FAILURE Host unreachable in 57m 00s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":1},{"id":"23136cd2e7e3448ede09ba1fac5a3538f25059ae","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-03 00:47:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cdd4698bebf604b494309a3ab59c43a829383d6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 03:21:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7981442171f54faeab685ce3b99b9e62\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4478209b28104f93947d9e0fe37cc4c1 : SUCCESS in 1h 29m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8c7804c6736d4f1a8e9634bf381cc9e3 : SUCCESS in 1h 08m 16s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d80a51129c034d65b2c2f627046a7372 : SUCCESS in 1h 49m 44s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f4693db1660c405c925a4f61999f40db : SUCCESS in 2h 31m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/04222b8dbe8b4fef937b1ac5059fc256 : SUCCESS in 31m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/332f07ef8b7e40b38b0f088a40222135 : SUCCESS in 56m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76bf0b31034b4ae4ab4510f51ca96dcf : SUCCESS in 11m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6942301f52284505a84548aa864ae992 : SUCCESS in 30m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/01737f9590e44bec9855e017507c5dad : SUCCESS in 26m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fee331613c764eb183a17bca145075b5 : SUCCESS in 33m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/367be41b4352423f93178eb5e61b0a2b : SUCCESS in 12m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bdf3ef0fe45b4123ae2a30c074ea5eef : SUCCESS in 10m 31s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2673d5f2732e48f4ae489c2471d4f2b3 : SUCCESS in 1h 29m 25s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/263b7ab909e94ef58b69f11ca3065c21 : SUCCESS in 1h 08m 43s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/5a4166e873674588af178de580c1f2b7 : SUCCESS in 51m 14s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b6c603406df74e529ecc4336f96a75a6 : SUCCESS in 2h 00m 44s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2e5e4ed494734106ae214f3b855f75b7 : SUCCESS in 1h 43m 05s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ad8e0e616b5540a196fb812dbf7e621c : SUCCESS in 1h 05m 08s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d5d205d2821a4890b6bb13a2f7389bd8 : SUCCESS in 1h 13m 42s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2df71f447bd54ebb94a9a780de111bda : SUCCESS in 1h 29m 29s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/d2dd9d7b5a004fc1839811536c1cf689 : RETRY_LIMIT Host unreachable in 1h 00m 15s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":1},{"id":"11bd4d291e8acd910daf17dd1aa72fd6ef4a5d01","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 04:47:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcc50d8f502146b2ba3935cb986dc3ae\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e761fd5a00f248c4b2b189d4a9f748ac : SUCCESS in 59m 13s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/03d4e360f6a442f98098dcd6b8352ad1 : SUCCESS in 42m 08s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f9359086ddb4470a9cc351f8ca706d0c : SUCCESS in 1h 03m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c0be9d3ec2f011bb826c2838aac938b533247484","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-03 08:37:19.000000000","message":"Patch Set 1:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/02/991002/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"01ea84f878082f5c4ae5e42575ef080ed63ecb0f","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-06-04 09:24:47.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7fdb6874f1d5010beb0d0dc607937851018141ea","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-16 08:32:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ffcd3f50576c91adc3c258a6c171ea031beb2203","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 08:33:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"32c6c1527e526026e49eec8f79704edf1e67791a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 11:37:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fd93f174a514de3b1e528ae33616b95\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a3aa55c7eb4c4284bacada5cebb7e9de : SUCCESS in 1h 28m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b7033f363fa34d18b105de49ca3626ee : SUCCESS in 1h 09m 01s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/cb04c06ffa044e4eb57e7fb9d8fbe134 : SUCCESS in 1h 35m 13s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/25bb7c410c324f6e8ad6c56ae7eda7c7 : SUCCESS in 2h 18m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4e19f54918243929f4cd7a1ecfeb90b : SUCCESS in 6m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6055c2abdede4e26a15826c1123afd96 : SUCCESS in 34m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ca5fc402c0214261b9ef8675c73e55a5 : SUCCESS in 17m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa12b02a3a464a9982a5a907d4ce757a : SUCCESS in 15m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/319edf0ced334b2b9f5857546c68ec90 : SUCCESS in 18m 54s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3695cc533d664130bfa4d16650d08325 : SUCCESS in 1h 16m 11s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2891de40145b4d65be341ac878b165e3 : SUCCESS in 1h 47m 56s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/71f6284582674ff0b1da07eb45fe703b : SUCCESS in 59m 11s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f1883d52b6054efb929c743dc53b8043 : SUCCESS in 43m 54s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3efb16d0b22d4d4ebc69e2f4e6d6fe7f : SUCCESS in 59m 17s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":1},{"id":"4aefdd192af0945e22ca50ef957e8d20813ed4b5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 11:37:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2b78220fe33a40c76fe647d342dee93922872864","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 11:38:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f89417de4054b58a757a3b154a353c2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41888298aa574dcdbe716557d804de4d : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7bbadcb63a2e62840c758a620888ef1236383821","revisions":{"7bbadcb63a2e62840c758a620888ef1236383821":{"kind":"REWORK","_number":1,"created":"2026-06-02 02:43:48.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/02/991002/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/991002/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/991002/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/991002/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/991002/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/991002/1"}}},"commit":{"parents":[{"commit":"c72c84cccd778375953e82df6f39d1b280a4da25","subject":"Merge \"Change some first() occurences to one_or_none()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c72c84cccd778375953e82df6f39d1b280a4da25"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-06-02 02:35:37.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-06-02 02:35:37.000000000","tz":-240},"subject":"Fix issues in _check_router_interface_not_in_use()","message":"Fix issues in _check_router_interface_not_in_use()\n\nThere are two distinct issues here:\n\n- Uncommitted reads (TOCTOU): The method creates its own\n  get_admin_context() instead of accepting the caller’s.\n  Any changes in the caller’s transaction that have not\n  been committed yet are invisible to these queries — a\n  floating IP associated to the subnet’s port by the\n  caller would not be seen. Change to use passed context\n  object.\n\n- Unguarded in_([]) with empty list: When the router has\n  no floating IPs, fip_ids \u003d [], and\n  PortForwarding.get_objects(context, floatingip_id\u003d[])\n  generates WHERE floatingip_id IN () — invalid SQL on\n  some DB engines. Change to initialize pf_objs to [] if\n  there are no floating IPs.\n\nTrivialFix\n\nAssisted-by: Claude Sonnet 4.6\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nChange-Id: I1e68bc22b8c2d334039326b9b3aff44296abcb2d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7bbadcb63a2e62840c758a620888ef1236383821"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7bbadcb63a2e62840c758a620888ef1236383821"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
