)]}'
{"id":"openstack%2Fneutron~991433","triplet_id":"openstack%2Fneutron~master~Ib4bf9e16ae44f05d12a115139b0f4a543468931e","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"31291":{"account":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"last_update":"2026-06-04 06:18:38.000000000","reason":"\u003cGERRIT_ACCOUNT_31291\u003e replied on the change","reason_account":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"}},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2026-06-04 13:08:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib4bf9e16ae44f05d12a115139b0f4a543468931e","subject":"bgp: Make main BGP router name not configurable","status":"MERGED","created":"2026-06-03 15:21:49.000000000","updated":"2026-06-04 18:53:28.000000000","submitted":"2026-06-04 13:08:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"991433","meta_rev_id":"8bc0a7ee9b74958002f9bdc1e242a3dcf70ad1f3","_number":991433,"virtual_id_number":991433,"owner":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-06-04 11:50:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-04 13:08:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-06-04 06:28:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2026-06-04 06:18:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},{"value":2,"date":"2026-06-03 19:55:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-06-04 06:28:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-03 15:32:08.000000000","updated_by":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"reviewer":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"state":"CC"},{"updated":"2026-06-03 15:38:46.000000000","updated_by":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"reviewer":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"state":"REVIEWER"},{"updated":"2026-06-03 18:38:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-03 19:55:41.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2026-06-04 04:49:56.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-04 06:28:28.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"b81be58f6d85835e3f2f6ff4ac942b3397e1ed44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-06-03 15:21:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"51d37428730ff9c96fcb34eb072d2c6501e99ace","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-06-03 15:38:46.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8d6b211301f61db5ac2ddd6d0501f9dafbb3b8ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 18:38:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/496f88d1cc27449ca7966da66c71c192\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/aa1405c2b8604f0e85f1cc2cb9d78bba : SUCCESS in 1h 26m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/61ca9ae5ae4b400b84d9fa27ebb38ef1 : SUCCESS in 1h 05m 25s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/57cf0efa2a7342a081690f2d6a9ff26b : FAILURE in 1h 36m 05s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/89be79a642c14d11a66adb7923d4ef8b : SUCCESS in 2h 28m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3d2044dfe09942668fa650a100cb8ae5 : SUCCESS in 43m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b06100c39aac43238c78c60a7b36e0c3 : SUCCESS in 39m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/568004fe5a194a06ad3fce9ebd3bfd75 : SUCCESS in 10m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b4d74731504047dd94f26a82be2899fb : SUCCESS in 21m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dce1e0c532a34a13b4e8ffc00b605330 : SUCCESS in 23m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/27400ef2041f4c3398ceea7927807273 : SUCCESS in 26m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfaa6ec9b9e74afa8663782c64497cc7 : SUCCESS in 11m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04e93d03f58e4731bfdf1b26f7497354 : SUCCESS in 17m 17s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/20dbb79ffffe46d386f4080aa91346b4 : SUCCESS in 45m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8bf2113ce93943938db79e485981b9a5 : SUCCESS in 1h 13m 20s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/a74a7f8da2d649fe9223db6c1fd9f95f : SUCCESS in 1h 19m 48s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/121dfafdb85d4d5faf6687c686eb1338 : SUCCESS in 1h 48m 44s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bde76361e7c44de6accc65814eac9af1 : SUCCESS in 1h 50m 06s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7443169ce3c5463d9f8ff65746b27fed : SUCCESS in 55m 12s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ec267275e0e445829f2cde4ac27fff8a : SUCCESS in 1h 24m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cf9b3f672776457e8c896583923021f6 : SUCCESS in 1h 32m 50s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/c9b875a5c1c04416b7f94fe810f6dde9 : SUCCESS in 2h 38m 28s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":1},{"id":"f4239e44a0e7b987e21bcdc39cb41ba7a13219a8","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-06-03 19:08:47.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"a451eb389d887b67504344caf1ca1a79c198330b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-06-03 19:54:26.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"c2af5295370090afa8281e8d2afa47af4770a838","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-06-03 19:55:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ce660939165a853e266e0b03e5149531e523b9e9","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-06-03 19:55:41.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0ca9a265b7bcf1b0854709dd99d69fbb1380c747","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 21:48:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2b141a0349548d5ba18b69efb5d6b67\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c60d08f3952247a6b8db02af4e7dc0ba : SUCCESS in 1h 19m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1dbd0061ef134a83879a8fe5ea8dd980 : SUCCESS in 53m 42s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/94256216217b4a528b3aaf0536c81bb1 : SUCCESS in 1h 24m 34s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/82f331d0030640a28c72f52062f0c888 : SUCCESS in 1h 51m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0d03fc6ebe18408c8feaeb2c3922713a : SUCCESS in 35m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9601113d52c46ac873f44929bdbb4cb : SUCCESS in 1h 05m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b70669481e7b49679086adae6ec4b916 : SUCCESS in 8m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/66243135286b44dc9684bc3f579dc07d : SUCCESS in 29m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8f9c23359f7945eb93ee6e07734b9a01 : SUCCESS in 26m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9d411e09c1614258a8994745be6e405d : SUCCESS in 27m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b726affa665143388071944938d18f27 : SUCCESS in 12m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/efedd45b4d2743baa2448b6f1613642b : SUCCESS in 16m 47s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b4c8f145b5ef482c8d7d271dc028e355 : SUCCESS in 1h 39m 00s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e783d00543634e6898e91e14fec0097b : SUCCESS in 1h 07m 27s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/1a30334d165945ad85a8fc10e60af7f3 : SUCCESS in 1h 20m 47s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/271c550989474b5dbda94ea6c1db3cfc : SUCCESS in 1h 12m 15s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/72f123eea9e74cca8c1f3d8c35d2f858 : SUCCESS in 1h 03m 40s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/026f8a8771274acc93e30008bb7cbbf0 : SUCCESS in 1h 12m 06s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4fc690d6c89043fa96e8d1823b1c6c1f : SUCCESS in 47m 54s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3abe717a4cc04ee3966d115f088c9ee6 : SUCCESS in 1h 09m 45s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/64b6c2d9f3684d6396b1f8ad2322cfcf : RETRY_LIMIT Ansible setup timeout in 3m 07s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":2},{"id":"c9ccc42938968794646a8916538628d961101911","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 04:35:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fad44ab9c68f485983e63b1d8b2cc8ce\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/56c38eb5f10d4379868c21d3026b5f7d : SUCCESS in 23m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9cec5a78e1d3452fb30c7ad0731c3977 : SUCCESS in 27m 25s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/da097b8ef2554a20a10430f55439d049 : SUCCESS in 22m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0e1b03c130956ffdd289c5dda383c3885ac031bb","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-04 04:49:56.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/33/991433/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"a0e674ec928627314d5b7f7b9352f4513c9a51d2","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-06-04 06:18:38.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8b2b6e2a61c6e78ba5bf7594276894d8c348d201","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-04 06:28:28.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3e6aaa352f1fb45e51f7d0cdd22d4599ee33aca9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 06:29:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a13360d8c69e2f3d92402ac9c7cc42b32bb90b59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 08:16:50.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/262f888b334e46f5a9ca618ee65f3179\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/253e75bad0e4492dbf782c5c9149f492 : SUCCESS in 1h 05m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7931b0c08c4d40edb35fa53584563cc9 : SUCCESS in 1h 05m 29s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4c8ab160cb0747b894155bd9eebcf4de : SUCCESS in 1h 39m 51s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a6e4593c47c94b06949a534971421f67 : SUCCESS in 1h 47m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c5087fa336e4ef99193aec7b9593fb6 : SUCCESS in 10m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/609d585f2d01404599d1eb1e7d111aee : SUCCESS in 30m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5b49382af02a43beafb549425e98bda3 : SUCCESS in 25m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ebe6705f2a248f1a335440d631a9bb1 : SUCCESS in 11m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f3a31ebbda740d5a055116725645f73 : SUCCESS in 6m 56s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/430678e6731e4be684508b1b980b0804 : FAILURE in 53m 17s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5d9dcf2d06de416fb8bed54a2bb98561 : FAILURE in 49m 26s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/c84553eb4c4b4533bd168e1ced88c7b7 : SUCCESS in 41m 55s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b3ed1119b7a248029a2a1123b4977a59 : SUCCESS in 1h 11m 36s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/25b496ab40d7461b989f8300ae8fadea : SUCCESS in 1h 02m 18s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":2},{"id":"8280fbbe6cc0486763a05bef248c1bfbd188f67b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-04 08:21:32.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/33/991433/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"264b9873d21501dd465be6dc761246bcf7df9ce6","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-06-04 08:23:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"10efff4511f6300cb0abdeb20f82e1b4a7a1e26f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 09:20:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/278a7c6e1dd749b0a8f9bee2529c058d\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bc27f771c35440eda4345d9dbaf3f155 : SUCCESS in 24m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/13d92511797c44099c55721a12e34e4d : SUCCESS in 50m 20s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c3cb98784759413699ba365685bdb36e : SUCCESS in 51m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6d14c86de28fe546c29e75a6b63580b1613d507a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 10:38:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/005a42ebe638434781f4386321f9a2e9\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4fdd4efaaaf84652ae425cf1890119f5 : SUCCESS in 1h 33m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d3c57f9545bb4a7f8c8d3d9872924e17 : SUCCESS in 34m 01s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/79c94970bdcb4ac0920a2dc0fe393f73 : SUCCESS in 1h 50m 39s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a7dbf659010146d5a37bb95aea0912f7 : SUCCESS in 2h 12m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cb849b12b5b645d186f0d09d1e67341b : SUCCESS in 36m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee439ffe1b1a473dac2d2dbdfd897979 : SUCCESS in 46m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f1f4da7584c4b0d94d3315097ca1bbd : SUCCESS in 6m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a4f24897d9874cb7a97bed0443c102db : SUCCESS in 33m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9a6ad7be030c4d50b33f1159c5f29858 : SUCCESS in 19m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/72c5f850531c462e9c79b7ee5af425e7 : SUCCESS in 35m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0074da8d9bc84294add14d88d4305059 : SUCCESS in 9m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c3ecd7315126428180d7679c1f45848a : SUCCESS in 6m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/066e9cc1bfb0498b9a1060c28db92588 : SUCCESS in 45m 14s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/019d5719fb5c459899c0a0fed03709b2 : SUCCESS in 48m 28s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ecc8cfeda9e6447a9898b8485147cb9c : SUCCESS in 47m 54s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e0fe8d9bb62b44038a96ae507e746d4d : SUCCESS in 1h 04m 22s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/90d2ff0d39b14c49bdcf92a830f16cf3 : SUCCESS in 1h 48m 17s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/14663c0101eb46af843d86d4cad9fd16 : SUCCESS in 47m 56s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b683da8120cc4cb2b37eab2f8e4c18cd : SUCCESS in 32m 00s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6600b8dca2a04efdbbc68b2d03d58d0d : SUCCESS in 37m 10s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/32c3750fe8fb40f4a37745b9fd7f0928 : SUCCESS in 1h 16m 54s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":2},{"id":"98d9c4cec982ca2bac27edb0de09a1c020964f0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 10:38:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"29e2fcb083c73e4c013bd2f578ed3dfe1c34daa8","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-04 11:50:52.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/33/991433/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"2890a753e3f8f2740e766505fa9f7e2904da4f50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 13:08:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8cdb3f48b07d4ab58d0b396f822599a2\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b2557c834e2b479bb92161083439029b : SUCCESS in 1h 17m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c0fe84616b0d44348913c3638cab152b : SUCCESS in 43m 10s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/62451a11eea24590b85196a7ddaee9ad : SUCCESS in 1h 22m 12s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f22d53ec379440dfb6ea6bc755db3f22 : SUCCESS in 2h 29m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26d25ed49a1e4f98a03f87c1b6f936d9 : SUCCESS in 9m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d9ff0c21567b4456af8225b47422fc9b : SUCCESS in 30m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/be1c819f02034cc19fab11535f1346bf : SUCCESS in 24m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9929ce574a2940d0bdf88f8983f3a65b : SUCCESS in 8m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af6f12ecbbe94d2593f3ee24b1218565 : SUCCESS in 19m 30s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1d743b3b5a364aaea79da2aa5a88e74f : SUCCESS in 1h 30m 59s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3748d33b825142a6ba5c3d3e69ba0481 : SUCCESS in 2h 01m 55s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ea4aca85d8834996825d3f76fef9be06 : SUCCESS in 1h 18m 12s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d6c769f9d25c4e5fa92a3e4d01edf142 : SUCCESS in 1h 10m 50s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3f088bd887fa414aaf3ea83625b76fad : SUCCESS in 58m 13s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":2},{"id":"cb3b04fc88432e0dca6265d2004f36d861700d3e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 13:08:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"722da7e96ba07634614b9faf8146a8135ff7121d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 13:09:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce8b619980a140b4a62dcc89c5534941\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e769a87b8f74369ae659cc93cd04ea5 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":2},{"id":"92e2887d0cb1e0db7a1f818debc0ef40951ae66f","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-06-04 13:34:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8bc0a7ee9b74958002f9bdc1e242a3dcf70ad1f3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-04 18:53:28.000000000","message":"Patch Set 2:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/33/991433/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"386781c13b1d2063933325261c57b4833550e16d","revisions":{"917a48f6479fdc492a9f0e46d2e32f8d7b166cff":{"kind":"REWORK","_number":1,"created":"2026-06-03 15:21:49.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/33/991433/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/991433/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/991433/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/991433/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/991433/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/991433/1"}}},"commit":{"parents":[{"commit":"ceb9b0f1cad403457f8ce80b5a03b33223b55b6e","subject":"Merge \"Fix duplicate ``AgentSchedulerDbMixin`` periodic worker descriptions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ceb9b0f1cad403457f8ce80b5a03b33223b55b6e"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-06-03 14:57:11.000000000","tz":-240},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-06-03 15:19:31.000000000","tz":-240},"subject":"bgp: Make main BGP router name not configurable","message":"bgp: Make main BGP router name not configurable\n\nThe patch removes the main BGP router name config option as there\nis no real use case for it. We don\u0027t have any mechanism in case\nthe config option is changed and it makes the BGP service plugin\nmore error prone to misconfigurations.\n\nAlso with the agent requirement for finding the LRP on the main\nrouter we would need to create the config option also on the agent side.\nWith that, it means all nodes in the openstack cluster would need\nto have the config option set to the same value, which creates\nthe cluster configuration more error prone too.\n\nThe benefits of hardcoded value are greater thaa the config option.\n\nRelated-Bug: #2151777\nAssisted-By: Claude Opus 4.6\n\nChange-Id: Ib4bf9e16ae44f05d12a115139b0f4a543468931e\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/917a48f6479fdc492a9f0e46d2e32f8d7b166cff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/917a48f6479fdc492a9f0e46d2e32f8d7b166cff"}]},"branch":"refs/heads/master"},"386781c13b1d2063933325261c57b4833550e16d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-06-03 19:54:26.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/33/991433/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/991433/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/991433/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/991433/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/991433/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/991433/2"}}},"commit":{"parents":[{"commit":"ceb9b0f1cad403457f8ce80b5a03b33223b55b6e","subject":"Merge \"Fix duplicate ``AgentSchedulerDbMixin`` periodic worker descriptions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ceb9b0f1cad403457f8ce80b5a03b33223b55b6e"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-06-03 14:57:11.000000000","tz":-240},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-06-03 19:54:26.000000000","tz":0},"subject":"bgp: Make main BGP router name not configurable","message":"bgp: Make main BGP router name not configurable\n\nThe patch removes the main BGP router name config option as there\nis no real use case for it. We don\u0027t have any mechanism in case\nthe config option is changed and it makes the BGP service plugin\nmore error prone to misconfigurations.\n\nAlso with the agent requirement for finding the LRP on the main\nrouter we would need to create the config option also on the agent side.\nWith that, it means all nodes in the openstack cluster would need\nto have the config option set to the same value, which creates\nthe cluster configuration more error prone too.\n\nThe benefits of hardcoded value are greater than the config option.\n\nRelated-Bug: #2151777\nAssisted-By: Claude Opus 4.6\n\nChange-Id: Ib4bf9e16ae44f05d12a115139b0f4a543468931e\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/386781c13b1d2063933325261c57b4833550e16d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/386781c13b1d2063933325261c57b4833550e16d"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
