)]}'
{"id":"openstack%2Fneutron~991552","triplet_id":"openstack%2Fneutron~stable%2F2025.1~I089999cece698af1a3b54d1341d9004d4108ae44","project":"openstack/neutron","branch":"stable/2025.1","topic":"bug/2105502","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-06-05 14:45:47.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2026-06-05 06:37:01.000000000","reason":"\u003cGERRIT_ACCOUNT_16137\u003e replied on the change","reason_account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}},"hashtags":[],"change_id":"I089999cece698af1a3b54d1341d9004d4108ae44","subject":"Allow service role more RBAC access for Octavia","status":"MERGED","created":"2026-06-04 08:29:47.000000000","updated":"2026-06-05 18:37:08.000000000","submitted":"2026-06-05 18:35:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"991552-bug/2105502","meta_rev_id":"c6cf4044d47c7b85dce3b5501404a410046c6470","_number":991552,"virtual_id_number":991552,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-05 18:35:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2026-06-04 15:06:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2026-06-05 06:37:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2026-06-04 20:03:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2026-06-04 20:03:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-04 08:29:47.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2026-06-04 09:18:12.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-06-04 11:30:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-04 12:36:41.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"CC"},{"updated":"2026-06-04 20:03:00.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"e5b7c4d2a008a496e16136e5ad5b378bee265fa5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-04 08:29:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"10ed05e419a4b53747668a6c4cfab96265407209","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-04 08:58:52.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"b4329da63c60ab5b6b9cc8d14530e21c2cf98dc8","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-06-04 09:18:12.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8036c7fe5d5b4673e327637c27c39f4182abc062","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-06-04 09:18:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b65edfe83e2abc0588fb2296cd700e57eebad287","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-06-04 09:50:46.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6821c7b85312e63e22c198c7620e51c6d2f95430","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 11:30:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d0c61f12cb841038bcc68d6c9c27e22\n\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/37e8d9501498487e8f8f8930356bf4f5 : SUCCESS in 1h 26m 20s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/c77885f789e1438fb63c2726e1640eb2 : SUCCESS in 1h 37m 21s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/da15ef4945904cd585f30aa2b3d8e714 : SUCCESS in 2h 29m 10s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/18820a6c43f44b40b18f11197fba05e7 : SUCCESS in 34m 05s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/d57ccf82f4ff423998771823cfeb8232 : SUCCESS in 1h 40m 12s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/f99aafad239c4060a0bfead3d39b5f2f : SUCCESS in 1h 17m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/895c907e06994ab3abfc7ba3cff1ad2a : SUCCESS in 34m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/334139ecc242478682fd7cbfd11ff9b6 : SUCCESS in 12m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2057d72da7384dfeb587346542beaf2e : SUCCESS in 15m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/73d12628b5af4302be77ce234b36599c : SUCCESS in 22m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f65eaaf483741d39b2ecbf9dc777936 : SUCCESS in 12m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df95bf594f4a426aa540511970256b13 : SUCCESS in 7m 21s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4158d71bc5b045d59266cfcba66c413e : SUCCESS in 1h 31m 28s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f20c5c7b7b5640928601ac0689158753 : SUCCESS in 1h 35m 45s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/17433028737c4eb4927460e1fc176357 : SUCCESS in 1h 53m 41s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7aa19cb452f2436ebd03b17f1699ba92 : SUCCESS in 1h 29m 50s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/cfb080fd8f644ef09992e98a6c9f258f : SUCCESS in 1h 05m 08s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5d9c7e78b2164507a9f578fd75688b7f : SUCCESS in 27m 38s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f05ff672f104490f96b6fd8e45497ced : SUCCESS in 1h 20m 24s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ab9e892698264b03b82a5b7db0b9b102 : SUCCESS in 1h 24m 13s","accounts_in_message":[],"_revision_number":2},{"id":"6bb94b6b83e0c89a5df386a0665fc24e237b8d8d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-04 11:41:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b9b9d92a239306869001c49b0c52008c3560c00b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-04 12:36:41.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"eb325b0cc0ee468f918f4f09371b7bd652b769e6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-04 12:37:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9a2783fda7f00afba4b9e769c71e21481c6283d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 14:57:54.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6b02913aa63147c581a69dbb8ef21d64\n\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/fc4d8fd688f14f189c5c6e23c0a4023d : SUCCESS in 1h 28m 33s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/a92f48ec707049909a28fe3c0a69a171 : SUCCESS in 1h 55m 19s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/68457a043c2244d0aba00f48aa74fc00 : SUCCESS in 2h 17m 58s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/d11a237deebf44a8b4a7a7710a83ac50 : SUCCESS in 36m 04s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/12606736a7b04586beac3f971403198a : SUCCESS in 1h 17m 46s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/66603aeea8524cb5bade0f2f6c75e9db : SUCCESS in 1h 08m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5834abb860eb4acd801d3f099157f8d7 : FAILURE in 35m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c4988afe4dd4a0bb91451cb61840319 : SUCCESS in 12m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/57373bb481b9423bb58166a6597e285a : FAILURE in 23m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ca6061014629409988b490ca9486eb0e : FAILURE in 19m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7301a3f10bab4ac1b078c4c6ad51e2d9 : SUCCESS in 12m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2232dc7388c463aba2c67afc2c74db5 : SUCCESS in 11m 11s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2dd3f3f1a36341a682c2ceb286572111 : SUCCESS in 1h 30m 12s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/07a4c7143c364c08b0a67726d9842600 : SUCCESS in 1h 29m 27s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2b27061488d64c61b4fe5b9c3cdd8622 : SUCCESS in 1h 50m 40s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/82fe0141294047cf9d950a306bb4e990 : SUCCESS in 1h 29m 22s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c02839ceb29e4f078a2c53a40fc2d5ac : SUCCESS in 1h 05m 22s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4bcf58ab7e2540a49e078f089a0f9da3 : SUCCESS in 1h 06m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/07bc8ae11b474ba2be650d7cb70c9498 : SUCCESS in 1h 15m 42s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0a91d9c80c2c41a693af28efe567f30b : SUCCESS in 1h 13m 28s","accounts_in_message":[],"_revision_number":3},{"id":"1cba27663d4ba5349593490595c284c14148964a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-04 15:06:38.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"406f3cea5cbfb5897893e7aa1cc859f2e03fe898","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 17:50:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c76723d738194f58b14cf97f16a147b9\n\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/7fe0b8ad662d468e86471c13dd49c142 : SUCCESS in 1h 27m 01s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/b0bb969ff40649339473cc990783cbd7 : SUCCESS in 1h 29m 57s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/11a2031fdd754f8f931d490a45e8c620 : SUCCESS in 2h 41m 40s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/ef92140913ed404a8e98b77269a2ff41 : SUCCESS in 32m 25s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/a7c3aed04a6f45a9911848c89ec4ecd0 : SUCCESS in 1h 34m 03s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/b7c3df4144c842cba48b2d135dfa11bb : SUCCESS in 1h 22m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6076cd8fe3224fcbbc3c1acc2fd9dbc4 : SUCCESS in 48m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/894b8ef9d1f74ae6a6f95ab9c06580bd : SUCCESS in 8m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/68f8f0f1a2d54a68ae8197352eebc7af : SUCCESS in 23m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/960c142713ed43c1bdedfb3de3e3bf6b : SUCCESS in 22m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57bb8874d08646599585eaafc911bdf6 : SUCCESS in 12m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58760fc06f2e49239ae541b9837c9efd : SUCCESS in 20m 15s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c168483f06fd45d2a4548d45f7d16f7b : SUCCESS in 1h 29m 30s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/608f90490a7a4cd9a01d4f6f6726a00b : SUCCESS in 57m 21s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7cf71d6000564b50b899f057655435c9 : SUCCESS in 1h 30m 35s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3e95609374464bfa840f798bdf63ee75 : SUCCESS in 1h 15m 49s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/997387ac500d4ef1ac0c1c53ec51a7fd : SUCCESS in 1h 00m 28s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3c6bd7a8e47e49c284088314d18a286e : SUCCESS in 1h 01m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/702e2c6f4b3e40ce87cc64a757345489 : SUCCESS in 1h 14m 43s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d5d564ac818c4dd58d34d81ed0376922 : SUCCESS in 1h 23m 49s","accounts_in_message":[],"_revision_number":4},{"id":"797aef2cd3f9106b3767765ce08dee208388963e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-04 20:03:00.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"5d6856832593d163eb6ccd465d9d02942d69258d","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-06-05 06:37:01.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"944ddd771ba4b5cb5225f52c2487e9b4fdfedef5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 07:58:23.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"6ee3cbafedb09ab8d54681aa4f5636dfea5203be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 10:20:47.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0cc3cf7eca37425e89b077c3903ef36a\n\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/06aefe9fe0124d97b8689b93b3d73440 : SUCCESS in 1h 46m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e5bf6c6ffb841ebba4c25cb30eff13c : SUCCESS in 11m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/728ca030e7a24a358bdb6de9931ec28b : SUCCESS in 22m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/93c65c61ca47401db6a24b10734cc8be : SUCCESS in 18m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66eb396dac2e4e78a637d275c042ae31 : SUCCESS in 7m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fceb3f0b19bc416c942a2ffacc025aed : SUCCESS in 6m 44s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/61312b239efc477fb65b054fc6e04cd4 : FAILURE in 26m 21s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/960fd7933e384201b3c04bc15af07c10 : SUCCESS in 1h 12m 24s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5d11a31495d348a486fe0fd622809aba : SUCCESS in 1h 52m 12s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d94280186e8d4ab4b23294eb8b3f3998 : SUCCESS in 53m 37s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/07c0ad9b90f54b8fb579701fab02daf0 : SUCCESS in 1h 08m 31s","accounts_in_message":[],"_revision_number":4},{"id":"c9afb3a1ecd506b4454c62cea50ca8a1cf303f72","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-05 11:05:43.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ea5b684aaa99dcdaab01f3d95aece6790cbcd89f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 13:24:07.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6afca81462454dbfad563d39023deea9\n\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/61439290f0254d98a2e71cf863a6a1cc : SUCCESS in 1h 45m 23s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/c43a5d4c6bff49388fe531401d3e70fb : SUCCESS in 2h 11m 31s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/87e019c32cf34aa48107a95c9b7b0638 : SUCCESS in 2h 16m 24s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/f7649b6fd4954493ac92dbe8f63d4a96 : SUCCESS in 35m 57s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/bdcd72242d624bdf8a9cbc8027e9eaf2 : SUCCESS in 1h 42m 10s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/cdaa9c088c7f452d9a5be95189a63694 : SUCCESS in 1h 33m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3dff2841e9246caa9f8cb7d351418f9 : SUCCESS in 49m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d32e90945b6484ba7179fd508bb5259 : SUCCESS in 6m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/010a358694cc4617b85b62393352aa69 : SUCCESS in 22m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3bc0601a81c94b9282ad1be44a39db3a : SUCCESS in 22m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a62163105264b4aa2b6964aca61da94 : SUCCESS in 6m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/abff6708f9d84ef68a1bce1c43955ddc : SUCCESS in 21m 18s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ec47e77e2be046ca82321d0549864776 : SUCCESS in 1h 21m 09s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/ae139788abfb45cd817d183902ae4e64 : SUCCESS in 1h 43m 02s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/516d5e718bf94fbbb0452b8f01f329e2 : SUCCESS in 1h 04m 05s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/14b1c746087c46e89e7c3179f793cfe0 : SUCCESS in 1h 35m 02s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c656008632b740fcaf1332173c327f8e : SUCCESS in 54m 10s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4753db0008984a8786fba1a61d9fb431 : FAILURE in 1h 06m 52s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cd26788148094a06bd4a7d5efc88eda3 : SUCCESS in 1h 21m 04s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/41414cac17674c29b7bf2199a2f4f8d0 : SUCCESS in 1h 13m 20s","accounts_in_message":[],"_revision_number":4},{"id":"7183d81e218ad1c182e854f3a0cdfbcb7ddb2c4e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-05 14:45:47.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4f409517d14a48ab137e1445c6439bf9a5cae330","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 16:39:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80ff9b2eda634ebb9791336408fcbbf3\n\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/b93dc4e348b54f5ea824ae550f130d4a : SUCCESS in 1h 28m 19s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/842ae7f17f2141e083d8e3edb8c9b4bb : SUCCESS in 1h 44m 23s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/fee1527057cd4683b815fbf2d7aa747e : SUCCESS in 1h 47m 06s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/a9e5d5cf88b7402fb92c958dab84919f : SUCCESS in 32m 33s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/5d16d7760a6b46ac8b0e08d5f862fe2b : SUCCESS in 1h 30m 26s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/160a6ed18c62480f906c50e129bd9f72 : SUCCESS in 1h 19m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3be020305774a6697637eb57667e31f : SUCCESS in 38m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0375ff9aec9843bea2b58e8c00b8507c : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55c4a902411643a989293db42a2dd0c3 : SUCCESS in 22m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0e02c6118226434491f977cd31141e85 : SUCCESS in 23m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52bff99479e24d18a2e38a0e054ca495 : SUCCESS in 15m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9cc1be1ba0244e9d8340cbfa5f814abb : SUCCESS in 8m 14s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/64583de87bce42f1a7e2c64d4aebc340 : SUCCESS in 1h 11m 24s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/c3a39546bccc45e881b09b7a9527fdcb : SUCCESS in 1h 34m 33s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5b97a4d00d61486d9a5c103b754bc099 : SUCCESS in 1h 39m 27s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/638eb86b63684492ac9a976769cafc6b : SUCCESS in 1h 40m 14s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/075e19bba55a4fdfa4ab666a5b27b547 : SUCCESS in 1h 06m 28s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/9c99985e8c77465e890dfd16dc83e86d : SUCCESS in 1h 01m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6b76aa1ee20046db9f295621904225df : SUCCESS in 1h 15m 52s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ff13eb86085e4de18f1cff98a3ae7163 : SUCCESS in 1h 13m 46s","accounts_in_message":[],"_revision_number":4},{"id":"33c75975b1ac9ea0f01960ba49273df6b3c5c545","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 16:40:45.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"aa1159008b780602cd98e0824943d1393113f5e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 18:35:44.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d09a6fd18f8415dbc3517d636a96511\n\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/e3dcd2c801894565a9c2fcb39bb8d3a1 : SUCCESS in 1h 54m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bee565e138ae45eeb67da553735ab5e9 : SUCCESS in 12m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/41f5c355f0b546f2aeaa3188dfb1de00 : SUCCESS in 21m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2f7e43ef404742759688d7a0b45e8dcd : SUCCESS in 14m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ac6ee8143e8426d8d79d059cbfe75f2 : SUCCESS in 12m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bbd6242d94854bfea817d7f6b29c9296 : SUCCESS in 22m 58s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5ea058823fbd4d34a4301012bd5c2e9b : SUCCESS in 1h 25m 51s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/21898fd5c626432c808557744d6d3ebf : SUCCESS in 1h 04m 28s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/59b9c8fcf9f84c41bdc6c1f78fa09245 : SUCCESS in 1h 45m 31s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3b853a8fb67a40e7993881ef27e2cbd8 : SUCCESS in 1h 00m 29s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6b3b197ae10443a9b7143dd864919a9b : SUCCESS in 1h 28m 13s","accounts_in_message":[],"_revision_number":4},{"id":"d02cfd930fb313860fb06914cb18950710909222","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 18:35:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"c6cf4044d47c7b85dce3b5501404a410046c6470","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 18:37:08.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6df9d013eb6c4d60bdd53ab34c9e35f7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3960fc7dbc224c75b7280be17a8fc8ed : SUCCESS in 1m 09s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/883bbca887b240edb3cb4ba4e33dbf80 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"67f9d4dc2632ab55ecb265cc42051e2aa64496a3","revisions":{"f9fe4e421657a9eefdb9b1de40002700fece6045":{"kind":"REWORK","_number":1,"created":"2026-06-04 08:29:47.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/52/991552/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/991552/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/991552/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/991552/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/991552/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/991552/1"}}},"commit":{"parents":[{"commit":"65489595eab8a70e5bc8092f2bbed420d0336733","subject":"Merge \"Fix PF GET/PUT parent floating IP validation\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/65489595eab8a70e5bc8092f2bbed420d0336733"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-03-24 15:16:25.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-06-04 08:29:15.000000000","tz":120},"subject":"Allow service role more RBAC access for Octavia","message":"Allow service role more RBAC access for Octavia\n\nThis updates the default RBAC rules for multiple\nresources to allow for a seamless integration with\nOctavia without having to give Octavia system scope\nadmin in the entire cloud.\n\nThe current use of the service role in the RBAC\nrules allows for pretty much all of the permissions\nthat Octavia needs today except for a few.\n\nIt needs get_subnet to be able to retrieve a subnet\nand check the details, this is low impact as we\nalready allow get_network.\n\nIt also needs get_network_ip_availability because it\nsupports to automatically select a subnet (if none\nis given) on a network based on the amount of\navailable IP addresses.\n\nThe default Amphora compute driver for Octavia uses\na keepalived and HAProxy implementation that uses\nunicast VRRP for the VIP address, this VIP address\nis added as an allowed address pair on the ports\nfor the amphora compute instances so the VIP port\nitself is not bound.\n\nOctavia also depends on being able to populate the\n``device_id`` field on a port which means it also\nneeds this patch [1] together with this one.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/947003\n\nCloses-Bug: #2105502\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\nChange-Id: I089999cece698af1a3b54d1341d9004d4108ae44\n(cherry picked from commit 65b9dc622f00e4aeac0e4f9a71db690a377cd558)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f9fe4e421657a9eefdb9b1de40002700fece6045"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f9fe4e421657a9eefdb9b1de40002700fece6045"}]},"branch":"refs/heads/stable/2025.1"},"827d663aad42db4bf06427deb8f457cc0756c018":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-06-04 08:58:52.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/52/991552/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/991552/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/991552/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/991552/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/991552/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/991552/2"}}},"commit":{"parents":[{"commit":"c6c4e1473f7c9f2777f4341adac90a40591fe4cf","subject":"[stable 2025.1] Delete all the grenade jobs from the CI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c6c4e1473f7c9f2777f4341adac90a40591fe4cf"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-03-24 15:16:25.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2026-06-04 08:58:52.000000000","tz":0},"subject":"Allow service role more RBAC access for Octavia","message":"Allow service role more RBAC access for Octavia\n\nThis updates the default RBAC rules for multiple\nresources to allow for a seamless integration with\nOctavia without having to give Octavia system scope\nadmin in the entire cloud.\n\nThe current use of the service role in the RBAC\nrules allows for pretty much all of the permissions\nthat Octavia needs today except for a few.\n\nIt needs get_subnet to be able to retrieve a subnet\nand check the details, this is low impact as we\nalready allow get_network.\n\nIt also needs get_network_ip_availability because it\nsupports to automatically select a subnet (if none\nis given) on a network based on the amount of\navailable IP addresses.\n\nThe default Amphora compute driver for Octavia uses\na keepalived and HAProxy implementation that uses\nunicast VRRP for the VIP address, this VIP address\nis added as an allowed address pair on the ports\nfor the amphora compute instances so the VIP port\nitself is not bound.\n\nOctavia also depends on being able to populate the\n``device_id`` field on a port which means it also\nneeds this patch [1] together with this one.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/947003\n\nCloses-Bug: #2105502\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\nChange-Id: I089999cece698af1a3b54d1341d9004d4108ae44\n(cherry picked from commit 65b9dc622f00e4aeac0e4f9a71db690a377cd558)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/827d663aad42db4bf06427deb8f457cc0756c018"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/827d663aad42db4bf06427deb8f457cc0756c018"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase","conflicts":{"ours":"f9fe4e421657a9eefdb9b1de40002700fece6045","theirs":"c6c4e1473f7c9f2777f4341adac90a40591fe4cf","contains_conflicts":false}},"a039e2f1ff45b15bb03b02656be1023fff306dbe":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-06-04 12:36:41.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/52/991552/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/991552/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/991552/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/991552/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/991552/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/991552/3"}}},"commit":{"parents":[{"commit":"a9318482367c637d603d072acf121c086fc1f961","subject":"Allow service role to create/update port device_id","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a9318482367c637d603d072acf121c086fc1f961"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-03-24 15:16:25.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-06-04 12:36:20.000000000","tz":120},"subject":"Allow service role more RBAC access for Octavia","message":"Allow service role more RBAC access for Octavia\n\nThis updates the default RBAC rules for multiple\nresources to allow for a seamless integration with\nOctavia without having to give Octavia system scope\nadmin in the entire cloud.\n\nThe current use of the service role in the RBAC\nrules allows for pretty much all of the permissions\nthat Octavia needs today except for a few.\n\nIt needs get_subnet to be able to retrieve a subnet\nand check the details, this is low impact as we\nalready allow get_network.\n\nIt also needs get_network_ip_availability because it\nsupports to automatically select a subnet (if none\nis given) on a network based on the amount of\navailable IP addresses.\n\nThe default Amphora compute driver for Octavia uses\na keepalived and HAProxy implementation that uses\nunicast VRRP for the VIP address, this VIP address\nis added as an allowed address pair on the ports\nfor the amphora compute instances so the VIP port\nitself is not bound.\n\nOctavia also depends on being able to populate the\n``device_id`` field on a port which means it also\nneeds this patch [1] together with this one.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/947003\n\nCloses-Bug: #2105502\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\nChange-Id: I089999cece698af1a3b54d1341d9004d4108ae44\n(cherry picked from commit 65b9dc622f00e4aeac0e4f9a71db690a377cd558)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a039e2f1ff45b15bb03b02656be1023fff306dbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a039e2f1ff45b15bb03b02656be1023fff306dbe"}]},"branch":"refs/heads/stable/2025.1"},"67f9d4dc2632ab55ecb265cc42051e2aa64496a3":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-06-04 15:06:38.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/52/991552/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/991552/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/991552/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/991552/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/991552/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/991552/4"}}},"commit":{"parents":[{"commit":"d78d001630621d27f65e60c5e4489ad46bf6737e","subject":"Allow service role to create/update port device_id","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d78d001630621d27f65e60c5e4489ad46bf6737e"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-03-24 15:16:25.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2026-06-04 15:06:38.000000000","tz":0},"subject":"Allow service role more RBAC access for Octavia","message":"Allow service role more RBAC access for Octavia\n\nThis updates the default RBAC rules for multiple\nresources to allow for a seamless integration with\nOctavia without having to give Octavia system scope\nadmin in the entire cloud.\n\nThe current use of the service role in the RBAC\nrules allows for pretty much all of the permissions\nthat Octavia needs today except for a few.\n\nIt needs get_subnet to be able to retrieve a subnet\nand check the details, this is low impact as we\nalready allow get_network.\n\nIt also needs get_network_ip_availability because it\nsupports to automatically select a subnet (if none\nis given) on a network based on the amount of\navailable IP addresses.\n\nThe default Amphora compute driver for Octavia uses\na keepalived and HAProxy implementation that uses\nunicast VRRP for the VIP address, this VIP address\nis added as an allowed address pair on the ports\nfor the amphora compute instances so the VIP port\nitself is not bound.\n\nOctavia also depends on being able to populate the\n``device_id`` field on a port which means it also\nneeds this patch [1] together with this one.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/947003\n\nCloses-Bug: #2105502\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\nChange-Id: I089999cece698af1a3b54d1341d9004d4108ae44\n(cherry picked from commit 65b9dc622f00e4aeac0e4f9a71db690a377cd558)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/67f9d4dc2632ab55ecb265cc42051e2aa64496a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/67f9d4dc2632ab55ecb265cc42051e2aa64496a3"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase","conflicts":{"ours":"a039e2f1ff45b15bb03b02656be1023fff306dbe","theirs":"d78d001630621d27f65e60c5e4489ad46bf6737e","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
