)]}'
{"id":"openstack%2Fneutron~991561","triplet_id":"openstack%2Fneutron~master~I0209096695bf00b29f0b51693e5d8b6b1ec67c34","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-06-04 23:59:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0209096695bf00b29f0b51693e5d8b6b1ec67c34","subject":"Remove remaining bind options","status":"MERGED","created":"2026-06-04 10:22:20.000000000","updated":"2026-06-05 00:01:31.000000000","submitted":"2026-06-04 23:59:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"991561","meta_rev_id":"93badf12502a92cea83d49dcfad76a2cc669ea15","_number":991561,"virtual_id_number":991561,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-06-04 20:16:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-04 23:59:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-06-04 10:41:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2026-06-04 20:39:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2026-06-04 10:29:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2026-06-04 20:39:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-06-04 10:41:16.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-04 10:29:03.000000000","updated_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"reviewer":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"state":"REVIEWER"},{"updated":"2026-06-04 10:41:16.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-06-04 11:38:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-04 13:03:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-04 20:16:34.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-04 20:39:28.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"8e53c1bc7f175876987c5b11824c20129529cfb0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-04 10:22:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d857444818c19329e2041059ee9c156c422ab998","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2026-06-04 10:29:03.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"82351df8557d77b26db1659a4405b35705b1a076","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-04 10:41:16.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"3955d016f51a26656e258e10e9c3ef01de2edf60","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 11:38:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c655dcb4955e4fae8334576882739850\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9bd9519e1ea14818a480acb8942c7dd9 : SUCCESS in 1h 00m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e00d65cdac9944fc9964cd18672d89ef : SUCCESS in 18m 23s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/96227f269ed0468bab8f7bf1a876d9fb : SUCCESS in 25m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"42261522cfed99913fb475e0acc240c9b98a1fe9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 13:03:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d415383c655a4d5985dabc700a6b8e86\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e7150c9966c44bab8071304d89fb7929 : SUCCESS in 1h 03m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ca8a17430764d9798788de377dce28d : SUCCESS in 57m 51s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/cf82e75ca3344c86a6d68b43351af39f : SUCCESS in 1h 39m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1037456e2c884a069b680c9da88b6279 : SUCCESS in 2h 36m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5b6c5754dceb44a69bc8bd5e0dd1ebdc : SUCCESS in 36m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96c3170761144a5f8decec5e15434696 : SUCCESS in 48m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65ff4cbc38eb4aafa0286bb743583a50 : SUCCESS in 11m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4937e7ccf0074e9e92cfc531ad162207 : SUCCESS in 29m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3f562fdd5133403da2bc1ae2398e77b9 : SUCCESS in 25m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5470374e73d842b1a9872afa526de39b : SUCCESS in 26m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76939d73232d4f7ba35cb1349da4d189 : SUCCESS in 9m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0291a1769335463f8ef759ce373545eb : SUCCESS in 16m 48s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7d927be1a0034d03bfa6aba574081076 : SUCCESS in 1h 08m 02s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f9a226d35e674ab6b03de36885ca3fd0 : SUCCESS in 1h 00m 18s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/b8a2f8d604734137ab55cb16e264eb9b : SUCCESS in 38m 57s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/bedfde03727f45c39453c84958baf24f : SUCCESS in 1h 53m 53s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e5d013c12a244fe1b9be56ab5d92b791 : SUCCESS in 59m 44s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/763c92087d0e461cb4676ff9c1461e18 : SUCCESS in 1h 06m 40s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/39ecf2d7d7714204ae99d2dbb3c401a6 : SUCCESS in 1h 20m 53s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/dc50d05b5f5240638332fa4ad16493ba : SUCCESS in 2h 08m 15s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/3a01c6e47bc24f129255d8c88ab8aacd : RETRY_LIMIT Host unreachable in 1h 03m 02s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":1},{"id":"3b754afd67023da7c54f58c5f4ab3f59ed74e487","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-04 20:16:34.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/61/991561/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"8caa11e0e4ad82290d60d847e651337fc466abc7","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-04 20:39:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"403ff5ce798cce82fd08bb7c231f2b8df832a46a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 20:40:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7536048764aff5295e65546a46f1451849f467e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 23:59:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc3ca83dbbf848e9b1541710caf88e53\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/73afb38a2bd647ec8310121f0d62a1b5 : SUCCESS in 1h 42m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b8b37996153a4154a4e844e0d378ef36 : SUCCESS in 1h 07m 08s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2d5124a65b934f6d891be7ce55cb3315 : SUCCESS in 1h 20m 30s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b0bb5c62b55d4ceeb2abe565c7d5b287 : SUCCESS in 1h 50m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34385cbbcc0c49ca99c94cc6f2d291e9 : SUCCESS in 11m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/87fc4b1fc6e1411bb7360c1826bd45bf : SUCCESS in 34m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9c3dee331f174726a0d94da27e3374dc : SUCCESS in 25m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10a0d7eebec9482e8a1ebde74f34928a : SUCCESS in 11m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/259611404f2944f0a6eb9479a76d14ec : SUCCESS in 23m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/96dcab8b3f8e4703a6ad5f671e2a1eda : SUCCESS in 45m 30s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a9f3492471394ff0b00ab6cb7fe40476 : SUCCESS in 2h 05m 50s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/a38fc095f559478e9dcfa975327a4edd : SUCCESS in 37m 40s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5b99c4554ad64801a9aad816f644f00c : SUCCESS in 1h 09m 36s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/55768712e19844ef971398e960da75c4 : SUCCESS in 48m 46s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":1},{"id":"351105e887afa202e3e4f45303e835df33a2ba44","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 23:59:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"93badf12502a92cea83d49dcfad76a2cc669ea15","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 00:01:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15dcd5f2ecc34cb6a9d9d35f13209011\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8699bf9e024f428bb0e145bb3540984c : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/553599ffb1094ff0bc97200edbe0efbe : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4cc0db8917c3ebc643fe092bff571f7ed3b54aa4","revisions":{"4cc0db8917c3ebc643fe092bff571f7ed3b54aa4":{"kind":"REWORK","_number":1,"created":"2026-06-04 10:22:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/61/991561/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/61/991561/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/991561/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/991561/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/991561/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/61/991561/1"}}},"commit":{"parents":[{"commit":"592609d08b266750b4382b041522b6286e0049a8","subject":"Merge \"doc: Update OVN features.rst with all supported extensions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/592609d08b266750b4382b041522b6286e0049a8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-04 10:21:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-04 10:22:07.000000000","tz":540},"subject":"Remove remaining bind options","message":"Remove remaining bind options\n\nThese were used by neutron-server which was removed.\n\nChange-Id: I0209096695bf00b29f0b51693e5d8b6b1ec67c34\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4cc0db8917c3ebc643fe092bff571f7ed3b54aa4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4cc0db8917c3ebc643fe092bff571f7ed3b54aa4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
