)]}'
{"id":"openstack%2Fneutron~991771","triplet_id":"openstack%2Fneutron~master~I67d56081c78cbd6bf15f65f0eae83fb2c06130c1","project":"openstack/neutron","branch":"master","topic":"fix/snat-maintenance","attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-06-24 10:00:52.000000000","reason":"\u003cGERRIT_ACCOUNT_39125\u003e replied on the change","reason_account":{"_account_id":39125,"name":"Felix Moebius","email":"felix.moebius@digits.schwarz","username":"felixmoebius"}}},"removed_from_attention_set":{"39125":{"account":{"_account_id":39125,"name":"Felix Moebius","email":"felix.moebius@digits.schwarz","username":"felixmoebius"},"last_update":"2026-06-24 10:00:52.000000000","reason":"\u003cGERRIT_ACCOUNT_39125\u003e replied on the change","reason_account":{"_account_id":39125,"name":"Felix Moebius","email":"felix.moebius@digits.schwarz","username":"felixmoebius"}}},"hashtags":[],"change_id":"I67d56081c78cbd6bf15f65f0eae83fb2c06130c1","subject":"[OVN] Add migration routine for ovn_router_indirect_snat","status":"NEW","created":"2026-06-05 09:01:14.000000000","updated":"2026-06-24 10:00:52.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"42bac106716852cb74299810d8415b7fcfa0cd83","_number":991771,"virtual_id_number":991771,"owner":{"_account_id":39125,"name":"Felix Moebius","email":"felix.moebius@digits.schwarz","username":"felixmoebius"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2026-06-05 19:04:23.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-06-05 11:16:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":-1,"date":"2026-06-08 18:58:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-05 10:16:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-05 11:16:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-05 19:04:23.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-08 18:58:40.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"87e3ca85037803dfcc98f4183b8f10b49b1f7aaf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39125,"name":"Felix Moebius","email":"felix.moebius@digits.schwarz","username":"felixmoebius"},"date":"2026-06-05 09:01:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0cdd8227d2f874cede1916e879235996f6ecab3f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 10:16:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a8be4a7ffd44109966b279b27f35e58\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8437bdab13154615b7a3df89a00afd10 : SUCCESS in 32m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9a3eb8b8c38547869c37f5555c3ccb0e : SUCCESS in 35m 58s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c4509a9e99f44444881ea92adeff5e11 : SUCCESS in 1h 11m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bde3a5c83a661e66a293fd1d57cf7cfe0b6e32cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 11:16:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e61093406ffc487292b485329217756f\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6d4bcaafdda448759d0caeb6970bfd70 : SUCCESS in 1h 22m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9ba983b3960745fa98ffdd02d24b9c6f : SUCCESS in 1h 02m 33s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/cdb057c0ec94453fb31a101818ba9017 : SUCCESS in 2h 14m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c33f24130a9b444f91bb35e2b7c1241e : SUCCESS in 1h 03m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34923b1e8dbd43a79accb9a7c55afd36 : SUCCESS in 10m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/89bea051c14f43f489091e54ff995a72 : SUCCESS in 30m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a79ee6156c034edf82ad1552de845a60 : SUCCESS in 25m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ff0f3e7ee73e4c1ab079d2a7436bb9b1 : SUCCESS in 24m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f4766e255ce4d7a87cfff79226afb4d : SUCCESS in 11m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6b3b5232f0e4d01ad1640d22813cf70 : SUCCESS in 19m 26s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/648b5026af754757b4feb1099f0be789 : SUCCESS in 55m 56s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/4c8b67094cf5495ea0d69dfaebe7fe93 : SUCCESS in 37m 32s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/376dd0d2c97646afaa235830a1ac9511 : SUCCESS in 1h 05m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c3e1bbd62b3240feac99226ab3c23eeb : SUCCESS in 36m 55s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/0c115c568e3145bd9bffeeefa27043f6 : RETRY_LIMIT Host unreachable in 1h 05m 36s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":1},{"id":"b851d74df316577b8b83db066c188640b7561f09","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-05 19:04:23.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/71/991771/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"e106decfbc0fef9fc045f411a855629059729d29","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-08 18:58:40.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"42bac106716852cb74299810d8415b7fcfa0cd83","author":{"_account_id":39125,"name":"Felix Moebius","email":"felix.moebius@digits.schwarz","username":"felixmoebius"},"date":"2026-06-24 10:00:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"94202bf31c7d784c3221a7a754decec91f213dfb","revisions":{"94202bf31c7d784c3221a7a754decec91f213dfb":{"kind":"REWORK","_number":1,"created":"2026-06-05 09:01:14.000000000","uploader":{"_account_id":39125,"name":"Felix Moebius","email":"felix.moebius@digits.schwarz","username":"felixmoebius"},"ref":"refs/changes/71/991771/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/71/991771/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/991771/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/991771/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/991771/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/71/991771/1"}}},"commit":{"parents":[{"commit":"bf817a8b70b834a81472c0a06059ba8a70aa9920","subject":"Merge \"Remove remaining bind options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bf817a8b70b834a81472c0a06059ba8a70aa9920"}]}],"author":{"name":"Felix Moebius","email":"felix.moebius@digits.schwarz","date":"2026-05-29 12:52:50.000000000","tz":0},"committer":{"name":"Felix Moebius","email":"felix.moebius@digits.schwarz","date":"2026-06-05 08:42:20.000000000","tz":120},"subject":"[OVN] Add migration routine for ovn_router_indirect_snat","message":"[OVN] Add migration routine for ovn_router_indirect_snat\n\nToggling the ovn_router_indirect_snat config option currently does not\nchange snat entries retroactively for existing routers. Additionally,\nwhen enabling the setting and adding a subnet to a previously created\nrouter, the new subnet will neither receive a subnet specific snat entry\nnor will the router receive the 0.0.0.0/0 snat entry thereby leaving the\nnew subnet in a broken state.\n\nAdd a migration routine to the maintenance task that will update all\nsnat-enabled routers to be populated with the correct snat entries based\non the current value of ovn_router_indirect_snat.\n\nCo-authored-by: Luca Czesla \u003cluca.czesla@digits.schwarz\u003e\nSigned-off-by: Felix Moebius \u003cfelix.moebius@digits.schwarz\u003e\nSigned-off-by: Luca Czesla \u003cluca.czesla@digits.schwarz\u003e\nChange-Id: I67d56081c78cbd6bf15f65f0eae83fb2c06130c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/94202bf31c7d784c3221a7a754decec91f213dfb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/94202bf31c7d784c3221a7a754decec91f213dfb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
