)]}'
{"id":"openstack%2Fneutron~991901","triplet_id":"openstack%2Fneutron~master~I76eb412bbb0b24652a4f64ddfda15760f1c3a36d","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-09-02 08:22:17.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-09-02 12:03:45.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I76eb412bbb0b24652a4f64ddfda15760f1c3a36d","subject":"Fix typo","status":"MERGED","created":"2026-06-05 16:53:42.000000000","updated":"2026-09-02 20:16:33.000000000","submitted":"2026-09-02 20:14:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"991901","meta_rev_id":"ccfc470a31fbfd00990f8315db87084379d596df","_number":991901,"virtual_id_number":991901,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-02 20:14:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"recommended":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"all":[{"value":1,"date":"2026-09-01 19:42:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":2,"date":"2026-09-02 08:22:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2026-09-01 20:45:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":1,"date":"2026-09-02 08:22:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}],"CC":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-01 19:31:29.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-09-01 19:42:09.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-09-01 19:47:58.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-01 20:31:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-01 20:45:48.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-09-01 21:44:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"19fb7bc9e078fbb164a8b2f0091f481c7d2a8a36","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-05 16:53:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"14da12143d8f090bd1636ab01658c29ce06597a1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-05 16:55:47.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"cf474d8f965ff1e35cc77fa79f3d4e7560a8f0fe","tag":"autogenerated:gerrit:restore","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 19:30:16.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"6dfb66492c999dde9b2be2bf4476fa15d53f7dbb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 19:30:28.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"83bcf3404b11aa3b0a2baf44ed77fdf5fbc1c22e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 19:30:32.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"154c3f8c4644e1373d8310c1d52d42f9b2e936e6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 19:31:22.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"807b801b3c98a0dea287b31202adc67036a194f3","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 19:31:29.000000000","message":"Patch Set 4:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"8dff490aeacb471fd4020d17aac1064cd9c70248","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-01 19:42:09.000000000","message":"Patch Set 4: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/65965802c9c44a6dbbad404c54f92bd0\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/76bf973163ec48c2894aaf83dcda762d : SUCCESS in 10m 00s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f26f22a5eb6adce24c408d77c584f5980599f3c3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-09-01 19:47:58.000000000","message":"Patch Set 4:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/September-2026/Arista_Tempest_Testrefs/changes/01/991901/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"f321d04c14b56e6ea2acac9092cb99c5f16fae6b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 20:31:57.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd14f67fc0f74448b6064d8d3ad024a2\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b6874a55502943ac83fb0ddae26fc5f2 : SUCCESS in 23m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4bfbca880f4d47f099f33a6f3329a939 : SUCCESS in 20m 08s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b93d97f6d5eb472cbd2ac63e048eaa75 : SUCCESS in 52m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3be9cc915801d5e902bfb9ae7b5d7f09e0a250aa","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-09-01 20:45:48.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"784a3d1e98a29131f001367cfbe1483d9f7b7d0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 21:44:21.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b08d4f5f0c2c4e0bbe5d5c263c6b30cd\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/84a79e2ca8fe4788b4ff30db352f9092 : SUCCESS in 49m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8647adc3bd8b426996c348ce84312c76 : SUCCESS in 32m 35s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/957b7618de2d46aca256173ccaf60045 : SUCCESS in 1h 00m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b6257befa12641b9bf5fc2b070f72d71 : SUCCESS in 2h 10m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/974bc78cea3e47c397de5a4bd03a4691 : SUCCESS in 1h 26m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7d5553f98a041f9894899ecd5df9995 : SUCCESS in 10m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/be3169bbcbc74e38a9a1a67c91bbb361 : SUCCESS in 32m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/41e1f39486c9428e8ecd2c496e02db8d : SUCCESS in 26m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9c9dde4ae9e540798873bea78229a7ac : SUCCESS in 38m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f68c8cc7e30465786cf745e44329cc1 : SUCCESS in 13m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fabf947be0944f27af7ead8771cad42e : SUCCESS in 9m 36s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/031fbf5aa444494e8ac7cec1c556f992 : SUCCESS in 1h 46m 07s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4960a31e79194d4a833c04054332a86b : SUCCESS in 1h 13m 25s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ff6b0de64b974e76aa22d5a0c0c255c7 : SUCCESS in 1h 21m 25s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b7accf45b08b4607a5c7f7c0bdc1a7bc : SUCCESS in 46m 07s (non-voting)\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/aea2b52bfa76407690dab073968de23b : SUCCESS in 44m 30s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b8be585562a74a538f4243625597e89a : SUCCESS in 39m 09s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c2de3edb33b249d696a08ce643e4f14b : SUCCESS in 1h 07m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7775df3608e146f18191c7793530bc10 : SUCCESS in 1h 28m 27s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":4},{"id":"4167562e37a708b5b6f0214e6a48139a28a1308c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-09-02 08:22:17.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"ed63fa2a7a2f21211db53ca97ef0bf793c4500a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 08:22:43.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"415e53c2244dd2ebc3e006df4ceb7e7af62ed1db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 20:14:38.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d67480ee1be54d6799bb10713a4a593d\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/445b7cd5e32d4d14ba831036a0ecebba : SUCCESS in 1h 28m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/84f45f199e7b452b843d955f3eef11ac : SUCCESS in 1h 03m 51s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/34f82410020040ad9d938e38ec4539bc : SUCCESS in 1h 03m 17s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3575594bb577484c83e9a585f5ca6fe2 : SUCCESS in 2h 00m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e94edf796bb4653bc84fa37093e419d : SUCCESS in 13m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f5be79a321c74ad3b6713693de18d91d : SUCCESS in 31m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dcbf105ec89345f59aaf7756fd20def0 : SUCCESS in 29m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f586734053014f1b88373260e80da48c : SUCCESS in 6m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27dc507d8dba4c7b8ed8a7f01303c889 : SUCCESS in 14m 31s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/768f372a1c8e4fd49f923e473314222d : SUCCESS in 52m 34s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4bf961a7e871424db582965143658c99 : SUCCESS in 1h 28m 42s (non-voting)\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/41f6095ba7e4410b97ef92c216682c51 : SUCCESS in 50m 04s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b278af94a44d464aa0e8bd9723ac3a93 : SUCCESS in 54m 05s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8454173e78db4212bd7b5e92941a1097 : SUCCESS in 1h 12m 12s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":4},{"id":"b6e44a581d61acd15195b17b812e340b77f30262","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 20:14:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"ccfc470a31fbfd00990f8315db87084379d596df","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 20:16:33.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e887e68a77c54a92b691585bde37967a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91724d0e77f146db9beda929b853252c : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"91fe6b27c27c6cc0dbd6a8092384bd626b079c86","revisions":{"8f0386f726db9caef0ce1bd9908752891edac842":{"kind":"REWORK","_number":1,"created":"2026-06-05 16:53:42.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/01/991901/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/991901/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/991901/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/991901/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/991901/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/991901/1"}}},"commit":{"parents":[{"commit":"8effa8b064d5593e0aad482ec922b41f6d954857","subject":"Merge \"evpn: Implement DB model for evpn_vni attribute\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8effa8b064d5593e0aad482ec922b41f6d954857"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-05 16:53:42.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-05 16:53:42.000000000","tz":0},"subject":"fix small typo","message":"fix small typo\n\nChange-Id: I76eb412bbb0b24652a4f64ddfda15760f1c3a36d\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8f0386f726db9caef0ce1bd9908752891edac842"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8f0386f726db9caef0ce1bd9908752891edac842"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"b25fbd63d1a54061d925aa0feff507dcf2af7f2a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-09-01 19:30:28.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/01/991901/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/991901/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/991901/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/991901/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/991901/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/991901/2"}}},"commit":{"parents":[{"commit":"8effa8b064d5593e0aad482ec922b41f6d954857","subject":"Merge \"evpn: Implement DB model for evpn_vni attribute\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8effa8b064d5593e0aad482ec922b41f6d954857"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-05 16:53:42.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 19:30:28.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: I76eb412bbb0b24652a4f64ddfda15760f1c3a36d\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b25fbd63d1a54061d925aa0feff507dcf2af7f2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b25fbd63d1a54061d925aa0feff507dcf2af7f2a"}]},"branch":"refs/heads/master","description":"Edit commit message","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"82f7923b6a80614af6378a9bf442148bf52b83a3":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-09-01 19:30:32.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/01/991901/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/991901/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/991901/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/991901/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/991901/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/991901/3"}}},"commit":{"parents":[{"commit":"64314906989b85c7fd472a242bb891fd484774d7","subject":"Merge \"pvlan: defer pg_drop init subscription until OVN driver loads\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/64314906989b85c7fd472a242bb891fd484774d7"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-05 16:53:42.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 19:30:32.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: I76eb412bbb0b24652a4f64ddfda15760f1c3a36d\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/82f7923b6a80614af6378a9bf442148bf52b83a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/82f7923b6a80614af6378a9bf442148bf52b83a3"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"8effa8b064d5593e0aad482ec922b41f6d954857","ours":"b25fbd63d1a54061d925aa0feff507dcf2af7f2a","theirs":"64314906989b85c7fd472a242bb891fd484774d7","merge_strategy":"recursive","contains_conflicts":false}},"91fe6b27c27c6cc0dbd6a8092384bd626b079c86":{"kind":"REWORK","_number":4,"created":"2026-09-01 19:31:22.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/01/991901/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/991901/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/991901/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/991901/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/991901/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/991901/4"}}},"commit":{"parents":[{"commit":"64314906989b85c7fd472a242bb891fd484774d7","subject":"Merge \"pvlan: defer pg_drop init subscription until OVN driver loads\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/64314906989b85c7fd472a242bb891fd484774d7"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-05 16:53:42.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 19:31:22.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: I76eb412bbb0b24652a4f64ddfda15760f1c3a36d\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/91fe6b27c27c6cc0dbd6a8092384bd626b079c86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/91fe6b27c27c6cc0dbd6a8092384bd626b079c86"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
