)]}'
{"id":"openstack%2Fneutron~992014","triplet_id":"openstack%2Fneutron~master~I3bf0c45a233e692f5b6905d72d943556dd9e61e1","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-06-10 12:13:08.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2026-06-10 14:16:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3bf0c45a233e692f5b6905d72d943556dd9e61e1","subject":"Switch from setting LSP.tag to LSP.tag_request","status":"MERGED","created":"2026-06-07 20:16:48.000000000","updated":"2026-06-10 14:18:47.000000000","submitted":"2026-06-10 14:16:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"992014","meta_rev_id":"c17e63909840ad1f37a92c73d142eb2c19e61503","_number":992014,"virtual_id_number":992014,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2026-06-10 12:19:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-10 14:16:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2026-06-10 13:37:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2026-06-10 09:36:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2026-06-10 12:13:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2026-06-10 12:13:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2026-06-10 00:47:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-07 20:59:19.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-07 21:34:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-07 22:31:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-07 23:36:48.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-06-08 08:49:28.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-06-10 09:36:17.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2026-06-10 13:37:51.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"e0e85b5050f041f6ee0ec1629183ccec59e01d10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-06-07 20:16:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"87031bca3dc749a052ec173c94504b297df9f73d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-06-07 20:34:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7c7df00613f3fc9cf007edc33eb3ab5e1df0ac77","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-07 20:59:19.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/14/992014/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"4fccd7c6f51640df3d67127dfbeccce503a56a87","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-07 21:34:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aaa9eb0bf6ec460384b47b18d27e6e2d\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8db67387abc64bea8fbbacf621b347a9 : SUCCESS in 39m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/eeac06c3f9a14672ab6443cd348bb28d : SUCCESS in 19m 07s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5cbf20cf424c47d59b36e37906ed6492 : SUCCESS in 54m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"844488f71efd2d5c8f4bbd482feb62412ddd8b73","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-07 21:40:39.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/14/992014/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"9490b55e5fb418b3162943bde83e541a24608394","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-07 22:31:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0ed73884972645b187da5d3d8eb494fa\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9225c414fd294649a9f9e5c30e3acdae : SUCCESS in 1h 12m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/421290adfb7d4f3493a15fbe1199c7b7 : SUCCESS in 1h 05m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/187ab8f8d5c847e8905d0547a32da4fb : SUCCESS in 1h 55m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/37c683bfc8a448328909998406150d0c : SUCCESS in 52m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/153ae4d1d1e848a8aad78bb317c53b82 : SUCCESS in 11m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/34ed60a9f87d48008fc1f72d079d6431 : SUCCESS in 25m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/571cf13e0a0d45caad7b3e18aa90759b : SUCCESS in 26m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/09410d38569745c6915b89ab20074e32 : SUCCESS in 32m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fadb9064b7314774a729d4ba50cd7df0 : SUCCESS in 12m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a623b6e7d518464f928e6a0c37cc7dc1 : SUCCESS in 22m 04s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4cef9d5e8b8f475a9f65cab8085e76a2 : FAILURE in 49m 53s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/7e46f905d53f4cd6ace4447075473717 : SUCCESS in 1h 08m 28s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/bb80c1783b8e45de845ed07eb5cdfe1d : SUCCESS in 1h 05m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0fe8b0f56c6b4500862772c9d4a99a75 : SUCCESS in 1h 24m 29s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/1aae50b8c0c348169439e72f63dbaabb : RETRY_LIMIT Host unreachable in 52m 12s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":2},{"id":"e331bd873b0f5fc1065d60db11dc01864e685402","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-07 23:36:48.000000000","message":"Patch Set 2: Review-Priority+2","accounts_in_message":[],"_revision_number":2},{"id":"e1b8680d0b54ab322723ca487120e43b870dca9c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-06-08 08:49:28.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c9a85c60b852a7cb1aec2cd6ea4a4d143a0061c6","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-06-08 13:57:44.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9069d3d901cd30792bf55c47f9d63af4a1d6dbc4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-06-10 00:47:45.000000000","message":"Uploaded patch set 3.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1, Verified-1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"3208173f091d1cbd86e4934324da8394d99d3e55","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 01:51:52.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b203f350ccf04682a20d91a1752da048\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/133505f36f464999a69195618a5dfa5d : SUCCESS in 53m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7d6fab02e59f4668a148ec8d0859ef41 : SUCCESS in 40m 27s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f469c10cdb0143b391dfd5bf7f10cfca : SUCCESS in 36m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"24ec6840300bfa2743f687171d0c3231121be900","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 03:08:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44fa638f0d4f4a3d9fba4fee611e115a\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9aaff94ce7764fce860f52cb03efd90a : SUCCESS in 1h 10m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4f3c41200e5c4e78b0c6be704e2f7133 : SUCCESS in 36m 54s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/beb70cb011054f3e9b835f672948eba5 : SUCCESS in 2h 18m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d18fe99fd2244f29a9be0f191c23516 : SUCCESS in 37m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad01bd3413994c13ab5c2d5093dc7700 : SUCCESS in 11m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8bd0804676214092a378c2f51304b4bc : SUCCESS in 23m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e306b2e510fd422e8f96fb775211b25b : SUCCESS in 18m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/32fc07eed7954ec7a1958f819fe2ccf4 : SUCCESS in 32m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/617fdcb836f44ab6ad6e632a61aaf0ae : SUCCESS in 9m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31c09ffd96fc46ac87bc67d169dc1047 : SUCCESS in 7m 28s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/866637093f3544c1bd4dbbfc402603c7 : SUCCESS in 1h 30m 00s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/1ba1cbd0adcc47a2b9751b8c2ae5bfa6 : SUCCESS in 1h 15m 45s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3c68aced621c4e4f9b2461b26f35e57d : SUCCESS in 43m 47s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/97c51b00c47144049022ef426d33f055 : SUCCESS in 1h 24m 45s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":3},{"id":"35417199da04adff67449541e4170e29583348a6","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-06-10 09:36:17.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"f7d43fdca1e502969d3617382fe301bb3ea6231c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-06-10 12:13:08.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b9e03d46532c53424a916c1f5ce8767117eea921","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 12:15:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ffd5452ded48058a12441c299f8614eeecf844f5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-10 12:19:14.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/14/992014/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"fa2e4b3ec5ecb4d52cf04210599e6edf5a4fb342","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-06-10 13:37:51.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5b414bc89ab81a992bf14957060a888b040c3944","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 14:16:23.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43f77379c6864d62a69ad54d78d30916\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5a0b08c615464c11b23f1c89b16e0af3 : SUCCESS in 1h 06m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3cf872d1c5004b269610747792b518e4 : SUCCESS in 37m 01s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/275e6f23a3fe4044bab08e08a527d4a3 : SUCCESS in 2h 00m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d127b78e251b42cba1091cb269521794 : SUCCESS in 6m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cebfbcbe1a9f4c5cb9027c5a39afeda0 : SUCCESS in 40m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2c5fe92c8a054728b84c9b178f73289b : SUCCESS in 16m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57d3deecb1b04cb28c66d24200fbee8d : SUCCESS in 12m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1dc29196da214eb386d6baba3eab762d : SUCCESS in 15m 38s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/78454c7e43224fad84d6654f1358fa64 : SUCCESS in 58m 51s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/4cc15408f5e84151a0725dea7c745a52 : SUCCESS in 1h 21m 37s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b8b6cc84e2bf44b894829a5d67518356 : SUCCESS in 1h 20m 50s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":3},{"id":"83eed55d5e47e638705c49a1fa4fbf4ee8a4f237","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 14:16:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"c17e63909840ad1f37a92c73d142eb2c19e61503","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 14:18:47.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56902905cd474a65b759aa1975346abb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/213cdd2240584cf78c56340abaadc49a : SUCCESS in 1m 59s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ae7e7b5118eee278d0b9be60ef64a5df2964b09c","revisions":{"a9972bed296692457af1af072c39a8c3dabcab92":{"kind":"REWORK","_number":1,"created":"2026-06-07 20:16:48.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/14/992014/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/14/992014/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/992014/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/992014/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/992014/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/14/992014/1"}}},"commit":{"parents":[{"commit":"02c48204331524ea0ffdcc6e53f4f1c7391f5f68","subject":"Fix functional test failure due to OVN commit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/02c48204331524ea0ffdcc6e53f4f1c7391f5f68"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2026-06-07 20:11:11.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2026-06-07 20:11:11.000000000","tz":0},"subject":"Switch from setting LSP.tag to LSP.tag_request","message":"Switch from setting LSP.tag to LSP.tag_request\n\nSetting LSP.tag has been documented as something only northd is\nsupposed to do. The CMS is supposed to set tag_request. A recent\npatch to OVN will clear tags that do not have a tag_request. A\nprevious patch to unbreak the gate sets both tag and tag_request in\na case that was breaking the funcitonal tests, but this patch\nchanges another place that was setting tag directly and adapts the\nAdd/Set commands for LSPs to convert any passed \u0027tag\u0027 to a\n\u0027tag_request\u0027. Only test code read tag after setting it, and in\nthat case we now wait for northd to set the tag before reading.\n\nRelated-Bug: #2155789\n\nChange-Id: I3bf0c45a233e692f5b6905d72d943556dd9e61e1\nSigned-off-by: Terry Wilson \u003ctwilson@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a9972bed296692457af1af072c39a8c3dabcab92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a9972bed296692457af1af072c39a8c3dabcab92"}]},"branch":"refs/heads/master"},"980983003c9a032b1eff96e0f5313f53bcea9c88":{"kind":"REWORK","_number":2,"created":"2026-06-07 20:34:56.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/14/992014/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/14/992014/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/992014/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/992014/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/992014/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/14/992014/2"}}},"commit":{"parents":[{"commit":"02c48204331524ea0ffdcc6e53f4f1c7391f5f68","subject":"Fix functional test failure due to OVN commit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/02c48204331524ea0ffdcc6e53f4f1c7391f5f68"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2026-06-07 20:11:11.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2026-06-07 20:34:50.000000000","tz":0},"subject":"Switch from setting LSP.tag to LSP.tag_request","message":"Switch from setting LSP.tag to LSP.tag_request\n\nSetting LSP.tag has been documented as something only northd is\nsupposed to do. The CMS is supposed to set tag_request. A recent\npatch to OVN will clear tags that do not have a tag_request. A\nprevious patch to unbreak the gate sets both tag and tag_request in\na case that was breaking the funcitonal tests, but this patch\nchanges another place that was setting tag directly and adapts the\nAdd/Set commands for LSPs to convert any passed \u0027tag\u0027 to a\n\u0027tag_request\u0027. Only test code read tag after setting it, and in\nthat case we now wait for northd to set the tag before reading.\n\nRelated-Bug: #2155789\n\nChange-Id: I3bf0c45a233e692f5b6905d72d943556dd9e61e1\nSigned-off-by: Terry Wilson \u003ctwilson@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/980983003c9a032b1eff96e0f5313f53bcea9c88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/980983003c9a032b1eff96e0f5313f53bcea9c88"}]},"branch":"refs/heads/master"},"ae7e7b5118eee278d0b9be60ef64a5df2964b09c":{"kind":"REWORK","_number":3,"created":"2026-06-10 00:47:45.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/14/992014/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/14/992014/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/992014/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/992014/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/992014/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/14/992014/3"}}},"commit":{"parents":[{"commit":"02c48204331524ea0ffdcc6e53f4f1c7391f5f68","subject":"Fix functional test failure due to OVN commit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/02c48204331524ea0ffdcc6e53f4f1c7391f5f68"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2026-06-07 20:11:11.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2026-06-10 00:47:38.000000000","tz":0},"subject":"Switch from setting LSP.tag to LSP.tag_request","message":"Switch from setting LSP.tag to LSP.tag_request\n\nSetting LSP.tag has been documented as something only northd is\nsupposed to do. The CMS is supposed to set tag_request. A recent\npatch to OVN will clear tags that do not have a tag_request. A\nprevious patch to unbreak the gate sets both tag and tag_request in\na case that was breaking the funcitonal tests, but this patch\nchanges another place that was setting tag directly and adapts the\nAdd/Set commands for LSPs to convert any passed \u0027tag\u0027 to a\n\u0027tag_request\u0027. Only test code read tag after setting it, and in\nthat case we now wait for northd to set the tag before reading.\n\nRelated-Bug: #2155789\n\nChange-Id: I3bf0c45a233e692f5b6905d72d943556dd9e61e1\nSigned-off-by: Terry Wilson \u003ctwilson@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ae7e7b5118eee278d0b9be60ef64a5df2964b09c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ae7e7b5118eee278d0b9be60ef64a5df2964b09c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
