)]}'
{"id":"openstack%2Fneutron~992018","triplet_id":"openstack%2Fneutron~master~I037110ce0bc9b2569147cab7c44a02063b108e5c","project":"openstack/neutron","branch":"master","topic":"bug/2144617","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2026-06-10 07:15:20.000000000","reason":"\u003cGERRIT_ACCOUNT_13861\u003e replied on the change","reason_account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2026-06-10 11:16:07.000000000","reason":"Change was submitted"},"38298":{"account":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"last_update":"2026-06-08 16:19:00.000000000","reason":"\u003cGERRIT_ACCOUNT_38298\u003e replied on the change","reason_account":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"}},"34271":{"account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"last_update":"2026-06-10 11:16:07.000000000","reason":"Change was submitted"},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2026-06-10 11:16:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I037110ce0bc9b2569147cab7c44a02063b108e5c","subject":"Integrate FrrDriver into OVN Agent EVPN Extension","status":"MERGED","created":"2026-06-08 01:19:32.000000000","updated":"2026-06-10 17:35:23.000000000","submitted":"2026-06-10 11:16:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"992018-bug/2144617","meta_rev_id":"e793c1b2f018de565a234b742592bd1844f98416","_number":992018,"virtual_id_number":992018,"owner":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-06-10 17:35:23.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-10 11:16:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-06-10 07:09:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2026-06-10 07:15:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2026-06-10 07:08:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-06-10 07:09:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-06-10 07:09:05.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"}],"reviewers":{"REVIEWER":[{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-08 01:19:50.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2026-06-08 01:19:50.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"state":"REVIEWER"},{"updated":"2026-06-08 03:15:32.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-08 21:49:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-08 21:50:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-09 08:25:33.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2026-06-10 05:20:15.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2026-06-10 07:08:44.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2026-06-10 07:09:05.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"ea14d3eec77d736e664d867bf1b846ec96547ff8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-06-08 01:19:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5cd39941e32d3294df6ee25cfe1ce7af2de5b9a3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-08 03:15:32.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/18/992018/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"a53e2c467433874b7613aa79b833e35d68e69baf","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-06-08 15:17:51.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"d825647ba9e0425f436a92edcf6d175b89482d23","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-06-08 15:18:02.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"aae5690e64bbe71f02fe3b33e2a4bd1f5e310b24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-06-08 15:23:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b863a6815554de7489b4995f76094fd751e14034","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-06-08 15:31:18.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0484b5f545dd4ee3ba5b9c186fdcbbdb5299bfc3","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-06-08 15:31:54.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"dc99ea642d99fa38e7b3f0246bdb38844660304b","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-06-08 16:19:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4be47777d3b3afe5db648e6b52f9b216d65e6124","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-06-08 18:26:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"89d5e4bfe3a4b24fa73e3d2ddf3fe2e54fe702a9","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-06-08 18:44:26.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c18336de104ec4a11cc11903852b4098d2e42495","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-06-08 21:48:27.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":4},{"id":"af38170d0ef24e49cb3513e13a74d04cde72962a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 21:49:10.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/neutron for 991099,13","accounts_in_message":[],"_revision_number":4},{"id":"f52520f79f4f8a4b330a5345a16fca68891649ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 21:50:05.000000000","message":"Patch Set 4: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/neutron for 991099,13","accounts_in_message":[],"_revision_number":4},{"id":"4755da1cbf648f42e8c07d8e1cbe31fd2d072b9a","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-06-08 21:51:17.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"034f7584666fd41eb669a36911dd7bda9ab1a4a9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-09 00:17:31.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/18/992018/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"a2bd15e5b3e153deef86f02e8fd11f9a82ca75aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-06-09 00:59:38.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"9c80027bc0fd25f7e66a3aeb47382873fa5c42f4","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-09 01:00:10.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/18/992018/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"2fd131d1cc615a8a69ccad0577b315d301909558","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 02:43:12.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/787ae9db09aa4c7fbc338cb7e05d9112\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4b1fa18993364de89b372d200f361a07 : SUCCESS in 21m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bd90fd5243234c3ea6c64221d7d24e75 : SUCCESS in 17m 12s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e55cf8d3182e4550a0399b81b9a131ce : SUCCESS in 1h 09m 40s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"cd160bc601ef1674250b11a77f26fd006a7f8641","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 03:12:36.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5e4de1049d434c78a4d3e90e657e76ce\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/07ffda5d9b1247eb9ece995a5e391f0f : SUCCESS in 1h 06m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dcbf07c6886145f193a92fe6a5ba4c50 : SUCCESS in 34m 35s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b8f78b70af694c62afa82160020a37d5 : SUCCESS in 1h 45m 40s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/cf5ab366e55b4095945927adab228345 : SUCCESS in 2h 04m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a9c6e5169bd341a9a73eee119ebb6cb3 : SUCCESS in 33m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8fb0f29fbd7f4235b37c0fd6f557c5aa : SUCCESS in 40m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d08aef15f7964f139d7e839886878ae3 : SUCCESS in 12m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a1bc2061eb1c40f7acb218e986388202 : SUCCESS in 18m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9606b670ed234241a4c941dfc87c2b5a : SUCCESS in 16m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e9d72f2ea3564b50aa4cb62092858cc1 : SUCCESS in 27m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe950b05f0f14f419a7ebe2de03765a4 : SUCCESS in 6m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c142c549bc0e40d095d9b4b533e88236 : SUCCESS in 16m 19s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/453adb700c524ee78d447316abb8b62f : FAILURE in 31m 21s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d516793512414769ad84866b6b1b8d09 : SUCCESS in 54m 33s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/976c045d74ac4dad8f4a26b42aa6f1fe : SUCCESS in 1h 18m 45s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d47fdafe787b4242ab875db145c29f70 : SUCCESS in 1h 34m 31s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4dfcddef1549406b99cffff5937213b8 : SUCCESS in 1h 42m 16s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/04fcf3077c9a4b75b88b91be0de97634 : SUCCESS in 54m 17s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ce50f366067b48b3b51ff5dfd89f89fe : SUCCESS in 1h 13m 09s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a1177feeb3f8450e80a5d42603fd73ee : SUCCESS in 1h 07m 03s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/fba63f59b740431d9b7466707d252da0 : RETRY_LIMIT Host unreachable in 1h 11m 22s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":5},{"id":"79c7c2990f8f75901b273bcd2b8f9337ddc4e1c7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-09 07:01:02.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/18/992018/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"8bef11edfce4692ddbae508fa7319654972292af","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-06-09 08:25:33.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"e6ca4b026dd438cb7863118643e8cc0ff7164f44","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-09 09:52:26.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/18/992018/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"ace29ab62cf61e19e9c5284fb0ad12359fcbfd03","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-06-09 12:55:47.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a3d6bd72b85b2c8012366936921a3a7de83fe8e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-06-10 01:38:47.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1, Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"f79393406afd0b051319e1fcce85ddcc2b8dcc34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-06-10 02:00:40.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"51b6ccb7fa5f22a9a495787569ccadc5b17e031f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-06-10 02:21:22.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"9b15f5b85ea5b46c0150107ffb8ffb3a5462a522","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-06-10 02:30:07.000000000","message":"Patch Set 8:\n\n(4 comments)","accounts_in_message":[],"_revision_number":8},{"id":"97e273962d0937f4ff46b825e06c60322df8a7e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-06-10 04:26:33.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"f0e9e3d7c8dc1f1f53092a71ac824688620e40af","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-06-10 05:20:15.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"cc4d74cb151c2f76d0ddae84128fdd50c2ac67cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 05:40:46.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c81b33ee2ffb4a97b9ddf8879a8391fe\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a4c3dc9754df4d30b89776a451cb5f63 : SUCCESS in 54m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e019515a47224974b5c396a8fe58c289 : SUCCESS in 48m 21s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c540f956965c4cff8085d89ae6e706e1 : SUCCESS in 25m 45s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"90dc3ca2abcb427fcc9e23fb5a3969b8d2ae15be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 06:58:35.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a9591fc3b0045a3a483a9b0035186e8\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/53949814ac5f4d37b9a1d52734a612ce : SUCCESS in 1h 05m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b09068be3d8141e48c9cc72452d76cdf : SUCCESS in 1h 05m 19s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/fb6484c3946c4f81aa792f65b034160b : SUCCESS in 1h 30m 02s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8d8ccd415ae5426eaa534a0b9cc199ac : SUCCESS in 2h 29m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/16f41452757d44d59307293c752eac33 : SUCCESS in 34m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/257dd56aaf824038897d8f037998b6d4 : SUCCESS in 38m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49e1d96c8bad4426be1d378197caae9c : SUCCESS in 6m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5d77c6497fdb42029ae81546e428b91b : SUCCESS in 21m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/865ed36d3412482bb2a0b42478e74dc0 : SUCCESS in 24m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0452f073f00f4ce2afbc36a79cb90b6e : SUCCESS in 25m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ad97c9738704bc3abe53e7020a384e9 : SUCCESS in 5m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cdd35d50ff364b9d90432fef151d8fcb : SUCCESS in 17m 55s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a31736d005dd4f5ea930804ae22cb980 : SUCCESS in 1h 34m 56s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4fdbe07821744222b64fba4c08540634 : SUCCESS in 58m 37s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/066d5e0c425a499cb9dc8674079b395b : SUCCESS in 41m 11s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2564474895034c709a10146f44a420a2 : SUCCESS in 1h 07m 47s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7b5557ba4e5344a19a95715005530967 : FAILURE in 50m 55s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4f4e43158e1e439dbe0335aa60a6fd9c : SUCCESS in 55m 48s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7b5353f176fd4764b990d9f5d530e3d0 : SUCCESS in 36m 48s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/484214d91c2e424c9a269a64055d6acb : SUCCESS in 1h 33m 35s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":9},{"id":"f7ac3a1ddd606a1eff263bbed79785ddab4ac1a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-06-10 07:07:47.000000000","message":"Patch Set 10: Published edit on patch set 9.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"59deb77606844ec9d456e0f28f718fa42780283c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-06-10 07:08:44.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"8753e16edda40fb6b7c7ba3ce0d70b146fba4a8f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-10 07:09:05.000000000","message":"Patch Set 10: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":10},{"id":"df32f2b172ec58c92b8ec138f1dfcf6089c4fa96","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-10 07:09:15.000000000","message":"Patch Set 10: Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"a96d41d7730f87d1510a3f39abb0171a996d29eb","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-06-10 07:15:20.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"b83b35b9f0550fd171ba8d084ce49eee8e41dfce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 08:26:20.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b9663fdb36f483caa7797ac44c7faab\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5ca0d519e7f541ba9178204204d49d16 : SUCCESS in 59m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/41e1d7da137546d4a3e359d82f48a2c6 : SUCCESS in 20m 41s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/415cd16e5c944ad5a20c3e558d978f96 : SUCCESS in 1h 12m 00s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"b3ea21bce08ed19362f85eb92308a6e42a2159ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 09:18:40.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b41f6a35eed451f93513087abbafcd8\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e4e8405386ca4e4b9a2fe14cce3e3174 : SUCCESS in 1h 06m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cc6b7465c5cd4638be6f619ffbff16df : SUCCESS in 37m 07s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/353fd478b686432d9d58849629fee428 : SUCCESS in 1h 36m 23s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/39470cb065ea4a118f2f04a6d45795e7 : SUCCESS in 2h 09m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/60814c3a9cc44f53813f0c1156019bfc : SUCCESS in 37m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9e1a6c14d7e454680cafcf08a8be46c : SUCCESS in 1h 16m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/278aff79c2f14f38abd6d730dc1b7428 : SUCCESS in 17m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0ca954a17b4a4315b04d1895931b26a7 : SUCCESS in 22m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4bb8e45353de4918b1eb97816c35d27c : SUCCESS in 16m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e8a330b1e93c496a8fc2d0ec5d45addd : SUCCESS in 26m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ea5bb7c545b499caa82d8ae67f2d17e : SUCCESS in 12m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5557ede4973a4cec9c18c6b7b6a54fca : SUCCESS in 19m 18s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6a3392186af048df9c2680199f1bb8e8 : SUCCESS in 1h 08m 39s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2ba51e7f93644a05bcecfd8e5cb4f133 : SUCCESS in 1h 06m 51s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/6168a2806df840cdb20941fcd7e06540 : SUCCESS in 1h 17m 17s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d95cb69e9e4c4a26b2a9f8408b682cb8 : SUCCESS in 1h 45m 03s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7ec263acaff54c878ecabd62fda71ca4 : SUCCESS in 58m 45s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/756ce3cd807a429a95ffd65a8bc5e2aa : SUCCESS in 51m 15s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/6eb80e4d4dfc46c4a65bc04990d79e66 : SUCCESS in 1h 03m 55s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c6b9f79dc9cc427891613abbcae27332 : SUCCESS in 43m 55s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":10},{"id":"4c9495569a03f0043a2a9d9be0810d451264de15","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 09:19:12.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"6d05d9d360ee3531087c9f865e3b59978ef1cd38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 11:16:07.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/781860029eee4715a2eab300008d8977\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f88a1f5adb454cda8a9b5ede2a92845c : SUCCESS in 1h 37m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d10bd4b3660e472d85de88cd1382852a : SUCCESS in 48m 19s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5c196f7fdaf84c1db24bbdc62f13b155 : SUCCESS in 1h 39m 15s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6db1e0d21140453c9dc02bd7dc32c352 : SUCCESS in 1h 54m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d08ec1c5ddfd4ad6b067ba28e6f508ee : SUCCESS in 11m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6c320b8cc2c148f6870f83a7d332e351 : SUCCESS in 29m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6ab90a97cf434ce6baa530437dcd039e : SUCCESS in 22m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6435df2d0f6b4843a305c2dea0a1ae83 : SUCCESS in 6m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a6fbf1d2afb45d0b922353a4ae9bfa1 : SUCCESS in 21m 56s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/61441206e0f24f95a7da7503b95714cc : SUCCESS in 1h 07m 32s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8ce2c3f13e06483f858edbc86446c61f : SUCCESS in 1h 55m 23s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/c0deadd1a4e440fe83e54ec0980ee7ec : SUCCESS in 1h 27m 02s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/048e898ecde140bdbe096472192865d5 : SUCCESS in 1h 08m 26s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2125e6de8e204f51908698f50d6f90c9 : SUCCESS in 1h 19m 10s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":10},{"id":"2724727071afd1417aadf8ccc358b361be4b325f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 11:16:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"5837ff754bdf0fd2728bf6ecf9339075e48ed9bb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 11:17:52.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be0f82d209e2469995725750108289e7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a8441ff5af54c35b9b25b330bbc926a : SUCCESS in 54s","accounts_in_message":[],"_revision_number":10},{"id":"4a39cdc885dfa87a3e04c9a093ce7f5b546d1787","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-10 13:01:21.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/18/992018/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"0694aea5de0a3ee5c593bc75de60d1b757f24b4a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-10 14:29:25.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/18/992018/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"db831be2b25a81f0ec24bb5269ee75ff321ca097","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-10 15:17:23.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/18/992018/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"ea25aa30c7da5375ceb55a79e1cd6316dc4d1b93","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-10 16:03:03.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/18/992018/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"e793c1b2f018de565a234b742592bd1844f98416","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-10 17:35:23.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/18/992018/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"7f986bb19940921c3b8ca48be0eebad0a20004a9","revisions":{"d9c1359c15332d9434dd815c997632270a016e7a":{"kind":"REWORK","_number":1,"created":"2026-06-08 01:19:32.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/18/992018/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/992018/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/992018/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/992018/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/992018/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/992018/1"}}},"commit":{"parents":[{"commit":"4d695d80e3f81addd10e9be01a24e685d376c229","subject":"Frr should write running config to memory to persist reboots","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4d695d80e3f81addd10e9be01a24e685d376c229"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-08 01:00:32.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-08 01:18:56.000000000","tz":-240},"subject":"[WIP] Integrated FrrDriver into OVN Agent EVPN Extension","message":"[WIP] Integrated FrrDriver into OVN Agent EVPN Extension\n\nThe FrrVtyshDriver is now called from within thew EVPN State machine\nwhen the state machine call _advertise and _unadvertise.\nThis patch created oslo config for evpn extension to read local device\ninterface from it.\nThe ovn-evpn-local-ip is used as the BGP router-id by the FRR driver.\n\nRelated-Bug: #2144617\nAssiste-By: Claude Opus 4.6\nChange-Id: I037110ce0bc9b2569147cab7c44a02063b108e5c\nSigned-off-by: Miro Tomaska \u003cmtomaska@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d9c1359c15332d9434dd815c997632270a016e7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d9c1359c15332d9434dd815c997632270a016e7a"}]},"branch":"refs/heads/master"},"8cf359f039369f5b152f589d0d5727d1c41cc9b5":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2026-06-08 15:17:51.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/18/992018/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/992018/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/992018/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/992018/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/992018/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/992018/2"}}},"commit":{"parents":[{"commit":"792933be78d7806016855a29c116af5e5578e8f1","subject":"Integrated SVD into OVN Agent EVPN Extension","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/792933be78d7806016855a29c116af5e5578e8f1"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-08 01:00:32.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-08 15:17:38.000000000","tz":-240},"subject":"Integrate FrrDriver into OVN Agent EVPN Extension","message":"Integrate FrrDriver into OVN Agent EVPN Extension\n\nThe FrrVtyshDriver is now called from within thew EVPN State machine\nwhen the state machine calls _advertise and _unadvertise.\nThis patch creates config.py for evpn extension in order to keep\nconfigurations for evpn separate from the OVN agent configs.\nIt then adds bgp_local_interface configurationw which will hold the\nname of the interface used for BGP peer connection\nThe OVN ovn-evpn-local-ip option is used as the BGP router-id by\nthe FRR driver.\n\nRelated-Bug: #2144617\nAssiste-By: Claude Opus 4.6\nChange-Id: I037110ce0bc9b2569147cab7c44a02063b108e5c\nSigned-off-by: Miro Tomaska \u003cmtomaska@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8cf359f039369f5b152f589d0d5727d1c41cc9b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8cf359f039369f5b152f589d0d5727d1c41cc9b5"}]},"branch":"refs/heads/master"},"20858fff9f2008b862f2692302bcd4fb306090e8":{"kind":"REWORK","_number":3,"created":"2026-06-08 15:23:59.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/18/992018/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/992018/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/992018/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/992018/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/992018/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/992018/3"}}},"commit":{"parents":[{"commit":"792933be78d7806016855a29c116af5e5578e8f1","subject":"Integrated SVD into OVN Agent EVPN Extension","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/792933be78d7806016855a29c116af5e5578e8f1"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-08 01:00:32.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-08 15:23:45.000000000","tz":-240},"subject":"Integrate FrrDriver into OVN Agent EVPN Extension","message":"Integrate FrrDriver into OVN Agent EVPN Extension\n\nThe FrrVtyshDriver is now called from within thew EVPN State machine\nwhen the state machine calls _advertise and _unadvertise.\nThis patch creates config.py for evpn extension in order to keep\nconfigurations for evpn separate from the OVN agent configs.\nIt then adds bgp_local_interface configurationw which will hold the\nname of the interface used for BGP peer connection\nThe OVN ovn-evpn-local-ip option is used as the BGP router-id by\nthe FRR driver.\n\nRelated-Bug: #2144617\nAssiste-By: Claude Opus 4.6\nChange-Id: I037110ce0bc9b2569147cab7c44a02063b108e5c\nSigned-off-by: Miro Tomaska \u003cmtomaska@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/20858fff9f2008b862f2692302bcd4fb306090e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/20858fff9f2008b862f2692302bcd4fb306090e8"}]},"branch":"refs/heads/master"},"ca5933c179c7c81d51af1b42b8d41f1e8582e62b":{"kind":"REWORK","_number":4,"created":"2026-06-08 21:48:27.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/18/992018/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/992018/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/992018/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/992018/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/992018/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/992018/4"}}},"commit":{"parents":[{"commit":"792933be78d7806016855a29c116af5e5578e8f1","subject":"Integrated SVD into OVN Agent EVPN Extension","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/792933be78d7806016855a29c116af5e5578e8f1"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-08 01:00:32.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-08 21:48:14.000000000","tz":-240},"subject":"Integrate FrrDriver into OVN Agent EVPN Extension","message":"Integrate FrrDriver into OVN Agent EVPN Extension\n\nThis patch creates new FrrvtyshDriver adapter (FsmFrrVtyshDriver) which\nis adopted to operated with EVPM finite state machine(fsm), where vrf\nis handled by the fsm. It also adopts to fact that FSM only needs static\nbgp-router-id and asn for each instance.\n\nThis patch also creates config.py for evpn extension in order to keep\nconfigurations for evpn separate from the OVN agent configs.\nThe bgp router-id will the ovn-evpn-local-ip from Openv_switch table.\n\nRelated-Bug: #2144617\nAssiste-By: Claude Opus 4.6\nChange-Id: I037110ce0bc9b2569147cab7c44a02063b108e5c\nSigned-off-by: Miro Tomaska \u003cmtomaska@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ca5933c179c7c81d51af1b42b8d41f1e8582e62b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ca5933c179c7c81d51af1b42b8d41f1e8582e62b"}]},"branch":"refs/heads/master"},"f662183493258665ba089f2349a088db783f9705":{"kind":"REWORK","_number":5,"created":"2026-06-09 00:59:38.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/18/992018/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/992018/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/992018/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/992018/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/992018/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/992018/5"}}},"commit":{"parents":[{"commit":"40c377f477bf6e1f89746a94c60ee0a790fc134a","subject":"Frr should write running config to memory to persist reboots","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/40c377f477bf6e1f89746a94c60ee0a790fc134a"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-08 01:00:32.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-09 00:59:26.000000000","tz":-240},"subject":"Integrate FrrDriver into OVN Agent EVPN Extension","message":"Integrate FrrDriver into OVN Agent EVPN Extension\n\nThis patch creates new FrrvtyshDriver adapter (FsmFrrVtyshDriver) which\nis adopted to operated with EVPM finite state machine(fsm), where vrf\nis handled by the fsm. It also adopts to fact that FSM only needs static\nbgp-router-id and asn for each instance.\n\nThis patch also creates config.py for evpn extension in order to keep\nconfigurations for evpn separate from the OVN agent configs.\nThe bgp router-id will the ovn-evpn-local-ip from Openv_switch table.\n\nRelated-Bug: #2144617\nAssiste-By: Claude Opus 4.6\nChange-Id: I037110ce0bc9b2569147cab7c44a02063b108e5c\nSigned-off-by: Miro Tomaska \u003cmtomaska@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f662183493258665ba089f2349a088db783f9705"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f662183493258665ba089f2349a088db783f9705"}]},"branch":"refs/heads/master"},"b3c77505cdc3b42a92f4ad710f63e9a1f85044e7":{"kind":"REWORK","_number":6,"created":"2026-06-10 01:38:47.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/18/992018/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/992018/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/992018/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/992018/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/992018/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/992018/6"}}},"commit":{"parents":[{"commit":"e5d06f7f669afb45b1c7ba6669be5e915bdadc20","subject":"Merge \"quota: Fix quota details API error with unloaded service plugins\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e5d06f7f669afb45b1c7ba6669be5e915bdadc20"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-08 01:00:32.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-10 01:34:58.000000000","tz":-240},"subject":"Integrate FrrDriver into OVN Agent EVPN Extension","message":"Integrate FrrDriver into OVN Agent EVPN Extension\n\nThis patch creates new FrrvtyshDriver adapter (FsmFrrVtyshDriver) which\nis adopted to operated with EVPN finite state machine(fsm), where vrf\nis handled by the fsm. It also adopts to fact that FSM only needs static\nbgp-router-id and asn for each instance.\n\nThis patch also creates config.py for evpn extension in order to keep\nconfigurations for evpn separate from the OVN agent configs.\nThe bgp router-id will the ovn-evpn-local-ip from Openv_switch table.\n\nRelated-Bug: #2144617\nAssiste-By: Claude Opus 4.6\nChange-Id: I037110ce0bc9b2569147cab7c44a02063b108e5c\nSigned-off-by: Miro Tomaska \u003cmtomaska@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b3c77505cdc3b42a92f4ad710f63e9a1f85044e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b3c77505cdc3b42a92f4ad710f63e9a1f85044e7"}]},"branch":"refs/heads/master"},"118617bb11af90863130659c2ce407b4d0de25c6":{"kind":"REWORK","_number":7,"created":"2026-06-10 02:00:40.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/18/992018/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/992018/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/992018/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/992018/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/992018/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/992018/7"}}},"commit":{"parents":[{"commit":"e5d06f7f669afb45b1c7ba6669be5e915bdadc20","subject":"Merge \"quota: Fix quota details API error with unloaded service plugins\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e5d06f7f669afb45b1c7ba6669be5e915bdadc20"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-08 01:00:32.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-10 01:58:05.000000000","tz":-240},"subject":"Integrate FrrDriver into OVN Agent EVPN Extension","message":"Integrate FrrDriver into OVN Agent EVPN Extension\n\nThis patch creates new FrrvtyshDriver adapter (FsmFrrVtyshDriver) which\nis adopted to operated with EVPN finite state machine(fsm), where vrf\nis handled by the fsm. It also adopts to fact that FSM only needs static\nbgp-router-id and asn for each instance.\n\nThis patch also creates config.py for evpn extension in order to keep\nconfigurations for evpn separate from the OVN agent configs.\nThe bgp router-id will the ovn-evpn-local-ip from Openv_switch table.\n\nRelated-Bug: #2144617\nAssiste-By: Claude Opus 4.6\nChange-Id: I037110ce0bc9b2569147cab7c44a02063b108e5c\nSigned-off-by: Miro Tomaska \u003cmtomaska@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/118617bb11af90863130659c2ce407b4d0de25c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/118617bb11af90863130659c2ce407b4d0de25c6"}]},"branch":"refs/heads/master"},"c5b5d0f8208089046551e81faa6767e06f85bf59":{"kind":"REWORK","_number":8,"created":"2026-06-10 02:21:22.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/18/992018/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/992018/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/992018/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/992018/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/992018/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/992018/8"}}},"commit":{"parents":[{"commit":"e5d06f7f669afb45b1c7ba6669be5e915bdadc20","subject":"Merge \"quota: Fix quota details API error with unloaded service plugins\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e5d06f7f669afb45b1c7ba6669be5e915bdadc20"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-08 01:00:32.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-10 02:19:33.000000000","tz":-240},"subject":"Integrate FrrDriver into OVN Agent EVPN Extension","message":"Integrate FrrDriver into OVN Agent EVPN Extension\n\nThis patch creates new FrrvtyshDriver adapter (FsmFrrVtyshDriver) which\nis adopted to operated with EVPN finite state machine(fsm), where vrf\nis handled by the fsm. It also adopts to fact that FSM only needs static\nbgp-router-id and asn for each instance.\n\nThis patch also creates config.py for evpn extension in order to keep\nconfigurations for the evpn extension separate from the OVN agent\nconfigs.\novn-evpn-local-ip from Open_vswitch table is used as the bgp router-id\nthat will be configured by the frr driver.\n\nRelated-Bug: #2144617\nAssiste-By: Claude Opus 4.6\nChange-Id: I037110ce0bc9b2569147cab7c44a02063b108e5c\nSigned-off-by: Miro Tomaska \u003cmtomaska@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c5b5d0f8208089046551e81faa6767e06f85bf59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c5b5d0f8208089046551e81faa6767e06f85bf59"}]},"branch":"refs/heads/master"},"4f99d03be676e77616b8658c1dfdd1285d49c22b":{"kind":"REWORK","_number":9,"created":"2026-06-10 04:26:33.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/18/992018/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/992018/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/992018/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/992018/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/992018/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/992018/9"}}},"commit":{"parents":[{"commit":"8f1f4b872ebd7d3971948a04378c9441df2e75e2","subject":"Merge \"evpn: Remove EVPN_VRF_NAME_LEN constant\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8f1f4b872ebd7d3971948a04378c9441df2e75e2"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-08 01:00:32.000000000","tz":-240},"committer":{"name":"Yatin Karel","email":"ykarel@redhat.com","date":"2026-06-10 04:22:01.000000000","tz":330},"subject":"Integrate FrrDriver into OVN Agent EVPN Extension","message":"Integrate FrrDriver into OVN Agent EVPN Extension\n\nThis patch creates new FrrvtyshDriver adapter (FsmFrrVtyshDriver) which\nis adopted to operated with EVPN finite state machine(fsm), where vrf\nis handled by the fsm. It also adopts to fact that FSM only needs static\nbgp-router-id and asn for each instance.\n\nThis patch also creates config.py for evpn extension in order to keep\nconfigurations for the evpn extension separate from the OVN agent\nconfigs.\novn-evpn-local-ip from Open_vswitch table is used as the bgp router-id\nthat will be configured by the frr driver.\n\nRelated-Bug: #2144617\nAssiste-By: Claude Opus 4.6\nChange-Id: I037110ce0bc9b2569147cab7c44a02063b108e5c\nSigned-off-by: Miro Tomaska \u003cmtomaska@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4f99d03be676e77616b8658c1dfdd1285d49c22b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4f99d03be676e77616b8658c1dfdd1285d49c22b"}]},"branch":"refs/heads/master"},"7f986bb19940921c3b8ca48be0eebad0a20004a9":{"kind":"REWORK","_number":10,"created":"2026-06-10 07:07:47.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/18/992018/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/992018/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/992018/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/992018/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/992018/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/992018/10"}}},"commit":{"parents":[{"commit":"8f1f4b872ebd7d3971948a04378c9441df2e75e2","subject":"Merge \"evpn: Remove EVPN_VRF_NAME_LEN constant\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8f1f4b872ebd7d3971948a04378c9441df2e75e2"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-08 01:00:32.000000000","tz":-240},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-06-10 07:07:09.000000000","tz":0},"subject":"Integrate FrrDriver into OVN Agent EVPN Extension","message":"Integrate FrrDriver into OVN Agent EVPN Extension\n\nThis patch creates new FrrvtyshDriver adapter (FsmFrrVtyshDriver) which\nis adopted to operated with EVPN finite state machine(fsm), where vrf\nis handled by the fsm. It also adopts to fact that FSM only needs static\nbgp-router-id and asn for each instance.\n\nThis patch also creates config.py for evpn extension in order to keep\nconfigurations for the evpn extension separate from the OVN agent\nconfigs.\novn-evpn-local-ip from Open_vswitch table is used as the bgp router-id\nthat will be configured by the frr driver.\n\nRelated-Bug: #2144617\nAssiste-By: Claude Opus 4.6\nChange-Id: I037110ce0bc9b2569147cab7c44a02063b108e5c\nSigned-off-by: Miro Tomaska \u003cmtomaska@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7f986bb19940921c3b8ca48be0eebad0a20004a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7f986bb19940921c3b8ca48be0eebad0a20004a9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
