)]}'
{"id":"openstack%2Fneutron~992655","triplet_id":"openstack%2Fneutron~master~I5b1c8064c448c88358ccfdc63c1b3f17499a84d4","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2026-08-10 08:36:46.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I5b1c8064c448c88358ccfdc63c1b3f17499a84d4","subject":"[WIP] evpn: Use scheduling for EVPN LRPs","status":"ABANDONED","created":"2026-06-10 13:55:06.000000000","updated":"2026-08-10 08:36:46.000000000","total_comment_count":3,"unresolved_comment_count":2,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"3ca0ab097d40cddb239cc38bdd368842eeac5271","_number":992655,"virtual_id_number":992655,"owner":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2026-06-10 22:02:07.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-06-10 16:12:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}]},"reviewer_updates":[{"updated":"2026-06-10 14:24:13.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2026-06-10 16:00:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-10 16:12:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-10 22:02:07.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-24 08:56:41.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"4cb5846fe7aff2e33faff633da9b35d100c8819d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-06-10 13:55:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"031e2e53de7472c91032eada24f8dc544957bc3e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 16:00:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ae201d03e56460fb4e873265b3efbbb\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/71985386941a47e6b8009db715188614 : FAILURE in 53m 45s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3bffb16381e943faa2f7ba9dd2db2e45 : FAILURE in 50m 51s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b4a3eac8afa04a0da1f273b727de342e : FAILURE in 38m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2b0665ff2f4715ff2a0b3482905e4b88affa336e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 16:12:51.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8272e2318ae34faeb6d076aae520f2fa\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b317c828e1cd4ecbbd045bdfc0d593bd : FAILURE in 1h 05m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dfa25761335744a39a7bbee16b393525 : SUCCESS in 44m 15s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e3fe66789a6f4646bf6e00c9ffbd3e32 : SUCCESS in 1h 49m 38s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/987d6771eebd4db2bff44457856ea564 : FAILURE in 2h 02m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b7a1846fbee74f6084494e457c5959be : SUCCESS in 33m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3f529f1a614477a95ae769c40681799 : FAILURE in 45m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0a7ea9c83e64f34a47b6c7216a0debf : FAILURE in 5m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e241e0d3ffad40c0bc51bdeda24238a2 : FAILURE in 26m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f0e4d7cdfb194747803bc28c97e81bf7 : FAILURE in 26m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/df00851db0784ed5bf7f23c0d204297a : FAILURE in 25m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84f873e16b13487688dca5c3d5ca07dc : SUCCESS in 7m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/344986d455f04666bcf7fe1ae9803c08 : SUCCESS in 17m 31s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a0c95077fdbe49afa4f263c6eb8a2aef : FAILURE in 38m 03s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b2b4b78a25e44c0baa7998db58270ed1 : SUCCESS in 39m 05s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/e0e9ea1a032c4d90aab605808ffee2c3 : SUCCESS in 51m 39s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e3d7051e95db48bcb75096d448bab36e : SUCCESS in 2h 10m 50s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/61ce00ddf76c46f9aa4e9da2f861c477 : SUCCESS in 1h 21m 41s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/92d524b0720841778fbb74647159f03d : FAILURE in 51m 01s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f403968125a04dadad5c2e97cc412de8 : SUCCESS in 47m 07s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f356382b45944692871e9acceb3d7cd1 : SUCCESS in 1h 22m 27s","accounts_in_message":[],"_revision_number":1},{"id":"03317b2c4795b7344c455f90acf48af197342509","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-10 22:02:07.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/55/992655/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"d1dad58734c0676fba5f79641624e212efc86c53","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-06-24 08:56:41.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3ca0ab097d40cddb239cc38bdd368842eeac5271","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-08-10 08:36:46.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4dc4f736bf02bece2717d24c7683c07dd11d3bc3","revisions":{"4dc4f736bf02bece2717d24c7683c07dd11d3bc3":{"kind":"REWORK","_number":1,"created":"2026-06-10 13:55:06.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/55/992655/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/992655/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/992655/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/992655/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/992655/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/992655/1"}}},"commit":{"parents":[{"commit":"1427e59c689898fc3140c3271426050c709aa0b8","subject":"Merge \"doc: Document runtime ``uwsgi`` Python module in WSGI guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1427e59c689898fc3140c3271426050c709aa0b8"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-06-10 13:54:14.000000000","tz":-240},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-06-10 13:54:57.000000000","tz":-240},"subject":"[WIP] evpn: Use scheduling for EVPN LRPs","message":"[WIP] evpn: Use scheduling for EVPN LRPs\n\nChange-Id: I5b1c8064c448c88358ccfdc63c1b3f17499a84d4\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4dc4f736bf02bece2717d24c7683c07dd11d3bc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4dc4f736bf02bece2717d24c7683c07dd11d3bc3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
