)]}'
{"id":"openstack%2Fneutron~993148","triplet_id":"openstack%2Fneutron~master~I578835796a600f8d0527c7d44a1da1dbea19f7c8","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-06-24 08:07:07.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2026-06-24 07:07:14.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2026-06-24 10:29:48.000000000","reason":"Change was submitted"},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2026-06-24 10:29:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I578835796a600f8d0527c7d44a1da1dbea19f7c8","subject":"ovn_client: Reset only defined LRP.options","status":"MERGED","created":"2026-06-12 16:08:02.000000000","updated":"2026-06-24 10:34:50.000000000","submitted":"2026-06-24 10:29:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"993148","meta_rev_id":"f046a2c2b1211e4778dca9093bbced88f4d517b9","_number":993148,"virtual_id_number":993148,"owner":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-06-23 17:06:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-24 10:29:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-06-24 08:07:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2026-06-24 07:07:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-06-24 08:07:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-12 16:38:17.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-12 18:18:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-15 09:37:55.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2026-06-16 06:13:42.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-06-17 01:11:13.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"}],"messages":[{"id":"a513487bab6b65cc6b607780dac8fa42cee69910","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-06-12 16:08:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9d2fc3d5dec51bced33b7098a51e7d3fb9c1950e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-12 16:38:17.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/48/993148/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"d3831d79eb4074568451129582f421fd8b4799cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-12 18:18:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d4389f64e2f343829e66d2b6d46f229f\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ac9db80a744749b0ab5e7f255e3641f3 : SUCCESS in 1h 41m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2a919998d7e647b29cef8b65dff9ab15 : SUCCESS in 1h 02m 29s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a9a4d2ba138f4b178fbbb660555c647c : SUCCESS in 2h 07m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd3f1360bfba4452960939c4842500d6 : SUCCESS in 44m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4824ed231e6b48398be2a266eb85fc25 : SUCCESS in 7m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3d2827e96de14948af48f30db3ccab0d : SUCCESS in 34m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/61cf8952cffb4776ab56b4bf4a43008b : SUCCESS in 27m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/66d7b106630242288e6d799f55fe6e35 : SUCCESS in 35m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec67e1eb6ba748ecbfd6dc1a87c4386b : SUCCESS in 12m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7bdca6519ab24620a8fca6d7a9bce61c : SUCCESS in 13m 48s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1b313ceb97f543588c930729bab67ce5 : FAILURE in 59m 01s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/2d1aff2029ea4c749399a34e648e41ae : SUCCESS in 1h 16m 10s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2b927436ba534df9b0039bbf6fe576c5 : SUCCESS in 1h 14m 56s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2106dc9faca84a2288fa078d23e246df : SUCCESS in 1h 31m 48s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":1},{"id":"daaa1e6af474cc3fbefe24b7e475434453502c18","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-06-13 14:55:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f6369519f811df0a74585d7c6a9f1b474647dc47","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-13 16:08:24.000000000","message":"Patch Set 1:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/48/993148/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"20c06c82ed94becf08a2daf3d8a7bea67bdf8800","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-13 16:56:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/614d01501f9b4b2894cdd73ce8a8170e\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/90fd1fb4c1d74a51925954fc5c0136cc : SUCCESS in 1h 01m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4811fdc6960c40b99d8b049b88ea1b16 : SUCCESS in 1h 11m 02s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/562b512ce92d4df888ba41dcfe539705 : SUCCESS in 2h 00m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4dca975dfb364596809c6c74fbf5d3ee : SUCCESS in 45m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afb458ba557a485f83a299821262f8fd : SUCCESS in 10m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fe9f2190eaad460a8c1b1635f16575fd : SUCCESS in 29m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cc4bae92de56463cb4038d2de6b8b299 : SUCCESS in 24m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4661c38badaf4a028e20f8022e28ff55 : SUCCESS in 28m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca1bfb46c06843b3918f01c91c8ccded : SUCCESS in 6m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad60aac3c4af45a2b57b6fc3c433e570 : SUCCESS in 23m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4f29206309c34609be56fc60d9e11add : FAILURE in 1h 14m 50s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/aecda1c307b24131b45b71bf61dec577 : SUCCESS in 1h 03m 08s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2e406ed4534a47179ea438e89ed3033f : SUCCESS in 54m 56s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/06e920d5891e43b39672f8c0baed9cf7 : SUCCESS in 1h 37m 23s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":1},{"id":"cde5eef1e398c87ea3d0a3f5f5aff41d9d599d1a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-06-15 09:37:55.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c1fb3ea8ab589075218e54608f2971d6bb3ae012","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-15 10:19:59.000000000","message":"Patch Set 1:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/48/993148/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"286cf25b07af5537bf8ae2e700125d4edd40d955","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 11:41:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ad28658f2fc14df4a47eb3c1e2d4aca8\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7e7119ea953a4e66a8ec2499ce29eeb6 : POST_FAILURE in 1h 28m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f4cf171966d646ecaa2d378c3ad09e5c : POST_FAILURE in 33m 42s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8c8f4866c0744968b00ac4c59fc8eb83 : POST_FAILURE in 2h 01m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f3afb74673b94f688d77f446fb010deb : POST_FAILURE in 43m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63bb0bd39b9049d2b1bafdfb126fff7c : POST_FAILURE in 5m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d3f5220a7a0d43c58346136aac8b8f0f : POST_FAILURE in 23m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3eebc4e4376c49bebbb336bdcaacc5f8 : SUCCESS in 22m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2964d7a4f2194cc696c0a03bdc92e412 : POST_FAILURE in 26m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aaa95a52350146bf971d214013d52825 : SUCCESS in 12m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb82cab8054141c689552c60b4c50bf7 : SUCCESS in 22m 28s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3a3e0048ac534423ad86037a70c64e8d : POST_FAILURE in 52m 43s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/021bd72a523a4a1f84cb60c91441f488 : SUCCESS in 38m 30s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9181aa8be87e4b96b63856c358a121df : SUCCESS in 1h 22m 41s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f8c223a0b9e0460ea9c37139b2f795d1 : POST_FAILURE in 44m 14s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":1},{"id":"92195163dabcfe9d7f03aa1deabce0d512f361e4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-16 06:13:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eabc1313a37340c69c868e3de57a865bb8a798b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 08:02:36.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1fc5280aec8542c399d8a3f933aeac7c\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c258665ecd1c475590be811d17004300 : SUCCESS in 1h 36m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9a352c509f5449a3b64a93da6a351967 : SUCCESS in 1h 14m 06s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/76fa49916f9646049e0ce9bf21f6cceb : SUCCESS in 1h 46m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b7245b9e920043b2b96c240d5b4ba465 : SUCCESS in 1h 03m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a76a76dcd487410fa74261ca1c5fff5b : SUCCESS in 9m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/10518f8aa2ca4b229e71393a5458c701 : SUCCESS in 28m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0cf95ac1ad7549a3bde4e9a023677456 : SUCCESS in 24m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0ce82bede9194f8397d245b48cde29ae : SUCCESS in 24m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc48c2d7fe0a42f1a92fc8fce91644a3 : SUCCESS in 6m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/887c7f870f2045deb4c158c439d4b2a9 : SUCCESS in 10m 12s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/96e5b0c2c0ae400f91c090eeb77cf963 : FAILURE in 51m 42s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d41a50accc684603865bb5adf5ba18c8 : SUCCESS in 52m 10s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/dc9427a5631048d5b33f56d36afe6926 : SUCCESS in 49m 05s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b9ad8d2c59d2480fa9f31490d5e38742 : SUCCESS in 44m 03s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":1},{"id":"28da75cd3c6011804c1a597f6e5180c76f89e280","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-16 08:07:46.000000000","message":"Patch Set 1:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/48/993148/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"b5a14fa185cea8cefbefb5b97e96c99bbf8a47af","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-16 08:26:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cd531289a5e01039a31f390be14c12d2beab365a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-16 09:28:51.000000000","message":"Patch Set 1:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/48/993148/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"62e39ada9b9d68329c37f4b384815e2658025b36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 10:30:14.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/84c1d1835f8845ffb2ee721cc4f521b6\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/698ce485304343a1bdbcc415b0655157 : SUCCESS in 1h 26m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/140e41bebd34471d9525875913d4451a : SUCCESS in 1h 03m 51s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1887f067eb1d44c9b6cc6c0768b7a42d : SUCCESS in 2h 01m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5892d786275c4bf2984915053709fba0 : SUCCESS in 1h 07m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b3f09cf5a214d18b001a7025507b17a : SUCCESS in 10m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d500f677d91148f393695108c085493a : SUCCESS in 26m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fa09c4a2b038499db2cb3e9cdcc616a6 : SUCCESS in 27m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d59e6f49a9034810b13b63d620fdd1c7 : SUCCESS in 32m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3a169c87ad648f4aa9573d6589b8b46 : SUCCESS in 14m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bbec388d1b0c4f22a72b0a0e6e03207a : SUCCESS in 8m 25s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d77c146afb08425291b6a8dca7758188 : FAILURE in 1h 27m 06s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/a12516c03fb34ac6801b0feed8aa5c3f : SUCCESS in 1h 24m 18s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9d33ea0872a447c08f988626ba56abbd : SUCCESS in 1h 05m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e332481af06f49b49eb9011f50c58718 : SUCCESS in 1h 07m 39s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":1},{"id":"6a8781be7e2a991f0b0875948c74ca9137d8774f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 21:53:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1525f8645ebd4b9583558d91024935b9\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a3fd7f03faf24dc4b8d2be4d0bd83540 : SUCCESS in 51m 23s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b8de7ceefdad42e99c0f27b43045be1d : POST_FAILURE in 40m 49s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/57796e3649f64c5da688633584d239a6 : SUCCESS in 1h 01m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b683c8cac813974c6bac189590090c5e1ed35512","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-06-16 21:56:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c420ca9c506c6de8dea239db76fc56dce7437637","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-16 22:51:47.000000000","message":"Patch Set 1:\n\nArista third party testing PASSED [ Failed to upload log files ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"e0eee2b9d8db05e50245f2b875fc29d04906551f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 00:35:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/29b8d700ad2a4205964ec010aa85c380\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b178e8ad2b1b42f3ad410a846b01c002 : SUCCESS in 1h 27m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0bc898700d33404cbce0e3223efa4899 : SUCCESS in 1h 10m 01s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b11a1c74efcb4bf7be4f49003504a5e4 : SUCCESS in 2h 38m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/216f56866a874f8482c0dec147ed7546 : SUCCESS in 1h 06m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7372bb7ded454bb5a81d809c0233085f : SUCCESS in 8m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/11bc0319a13345f5ad880a68a57e7cb7 : SUCCESS in 22m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/10961d0211a349bca3c5d937d374b1f7 : SUCCESS in 19m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8476b908b0ee417ba2b583b906ee8878 : SUCCESS in 25m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/710fe5a922194cd59a9a5cf4fb403ded : SUCCESS in 11m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9329998f896548f88da1660f876327c8 : FAILURE in 18m 33s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3269556dbb8f4987a7b8ce07d55ccc35 : FAILURE in 51m 51s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/f0558c10358c435c90b001824a48b925 : SUCCESS in 1h 00m 00s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b9fda7838fe540a18be473ebed2f8ad2 : SUCCESS in 1h 07m 42s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5fa82c55e2b14783867c2c575c61ae29 : SUCCESS in 1h 24m 52s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":1},{"id":"e39c99612a34662fea9e655d720ff5a31632dfe3","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-06-17 01:11:13.000000000","message":"Patch Set 1: Code-Review-1 Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cde80e3fb808ccc42c37cef35dd72aaf67dcee6c","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-06-17 04:28:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a335361527f830418e009a130922fefd8a94b912","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-06-17 12:28:46.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"243ebda753b99ece6f156ba073489911fd42a925","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-17 12:32:43.000000000","message":"Removed Workflow+1 by \u003cGERRIT_ACCOUNT_16688\u003e\n","accounts_in_message":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"_revision_number":1},{"id":"5d5611e108fe0d19ea391206976758e58613ffb1","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-17 12:32:45.000000000","message":"Removed Code-Review+2 by \u003cGERRIT_ACCOUNT_16688\u003e\n","accounts_in_message":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"_revision_number":1},{"id":"c8c3546da9e59367b3ab61ac5a7f5b60148c0f7e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-17 12:33:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"df5054ef0c5421a8d81bd42676231c2a26922e74","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-06-17 12:36:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6d6a2ac244ebbe1f1517d1d393d24e0ce5ef27c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 19:15:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00fb4f165d5f4218987bba5e3ac2b499\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1597cb8532e048fdac46fa5717777486 : SUCCESS in 20m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3ca969bd7a824dbabc5eff059dd2d39c : SUCCESS in 25m 39s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/db0eba5de0e849bd93c71d11f7e2f720 : SUCCESS in 26m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4ae3c31d82d5db6e11608b69f0932a7bb1377bea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-06-18 17:23:41.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1, Verified-1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"bd08587386a0bfc6220dc483eff6a18552b974cc","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-06-18 17:24:39.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"1e5149a1dad7de0b2a04aefb3a905707b805ddf8","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-06-18 17:24:44.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"5e766d5541d5eecd1462c77c967627dc61575539","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-18 18:04:52.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/48/993148/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"625abeaf35fea83d90334113f8108842494d2b09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 19:27:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0ee86aa0e2b24ec082433616e5dcdfbd\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1ddce59fe49147fe9eb9930442ade5fa : SUCCESS in 1h 35m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/712b682d6c1c450da89135420361c6f8 : SUCCESS in 33m 27s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b2d09e26c64c4ebb937dae3a78580241 : SUCCESS in 2h 02m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f8d635bed384db693182fae00a5b3a7 : SUCCESS in 1h 09m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adc487f550234798aa178a208a719ef7 : SUCCESS in 11m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0d6ad625bb5340b59a7a55f6e210522e : SUCCESS in 27m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1a5109ac7c974fc3bbb08e9d05b08bf0 : SUCCESS in 25m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/efc3f30898ba4307ae1012af0929403f : SUCCESS in 38m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/974913c8fff94fd18a26c525f2306dce : SUCCESS in 6m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/efe0efc00949453b9ce9e4ae4ebf47bd : SUCCESS in 23m 35s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/730e1e429f3848258255b9689311c0f7 : FAILURE in 53m 14s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/bacffbf9aed54fa88f00807c2fba1a8e : SUCCESS in 37m 47s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6ce47ad683e2426f898317beab234481 : SUCCESS in 1h 08m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d15998a0abe04e02ae45c24b02e613f8 : SUCCESS in 1h 09m 43s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":2},{"id":"ec98c4899f5280a6d952eda517e8e4de01c98ce4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 19:57:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25d4334194bb43c3b62e6d0e72a775be\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7a5016d0e99a40c4bfebc8c16de15f7a : SUCCESS in 54m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/76e21bd5b26743b2bbebc223f990af4e : SUCCESS in 44m 46s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/80eaa3453f7044d7a3681c04b63d1072 : SUCCESS in 57m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"29b5b50ac607ca81c72d8908691ba99b4dc1aa32","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-06-22 20:44:51.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1, Verified-1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"5a9b7b5dc7a5d0ecbed84f055c51b7cd25bd05d3","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-06-22 20:45:02.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"a4011f0482669e7e5f6249935b40328d572ca8ee","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-22 21:30:59.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/48/993148/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"e5c868592b2f969ba9d354b13af908932f818a39","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 22:09:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7cb90b70b5d14dd4b734f99e875c86e8\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/29ffbe4bb3b94ac4bd5097ba262ff994 : FAILURE in 53m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/62acfea5e2b041dd8ad5a418291fe627 : FAILURE in 50m 12s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b132c1313b7e4d44b4bfb2033ddbcad5 : FAILURE in 57m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b65872ddddc4a563eb51f439f8ceea27a7b96578","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 22:34:09.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8973d0fe35124956a49ef0b6f047bc5e\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5c41fe068c16427ba743b129e3040aa4 : SUCCESS in 1h 05m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6551cd6b57174de3bfd4d4d52c1105c4 : SUCCESS in 36m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5317eb39fb854124b77c69e24b9727e2 : SUCCESS in 1h 48m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3308d4d814d445e8a697226a7ad8ef72 : FAILURE in 37m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3095e990646a45cb88228b7a5654cac1 : SUCCESS in 5m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3f6795338ec446e6afcb5d68c1c4b6a3 : FAILURE in 19m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/be04bc7dc91d48c597e2c568a7e4d5a0 : FAILURE in 16m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2516750ebeba4efb845ab119e9d78de1 : FAILURE in 21m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a03515618ea43babfe588fcb2d7a197 : SUCCESS in 12m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10e4beec5a234b4385524d621e11023f : SUCCESS in 20m 50s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6e59d264019e4756940993f22cc834d8 : SUCCESS in 1h 32m 23s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/68f278731d014dda945a0ba0c5abf433 : SUCCESS in 38m 54s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ecbb5f42939e4b58babafcbcbc84bed8 : SUCCESS in 1h 11m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9857696bd3a645b18c1100d2a97959f5 : SUCCESS in 1h 21m 44s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":3},{"id":"18be8d097f12c05d47db4eb703c7386c76ef0a80","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-06-23 08:11:19.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"86b2603597f80aa3ca42c257c68fcafff6fc9dcc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-06-23 16:21:40.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1, Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"be61ffcd52e7e1dfe3360c35f6ced0a3e082f048","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-23 17:06:46.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/48/993148/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"3d7c1bfa0868339a648925fa3a6c357e24302915","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 17:21:47.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91822cd06fe24884a82159a52d4370c9\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9140da65c88d4b71bccee097ac1f6928 : SUCCESS in 55m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a4da6af81350424793746099c18e05bd : SUCCESS in 38m 18s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1b09b888f8844df6975d87e23541e60a : SUCCESS in 55m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6e6582c5714490ffda461ab1b460f5877bd533b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 18:26:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8479c5ee0a804f2baf728042a61254f7\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c3c4cb2d9d04494d87599e1ccbbdeb00 : SUCCESS in 1h 16m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8ccdad006d0f49f0b037cf8a13ac19d6 : SUCCESS in 1h 07m 49s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0edb06f202164256b98eb481e7a6a5aa : SUCCESS in 2h 03m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/26d3e6707f244054bd785604b518e47c : SUCCESS in 44m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bec05045476e4560b51142496516975d : SUCCESS in 11m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/13da740c7d3444609e40fc3167c1dd50 : SUCCESS in 32m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e1bc42c7544c496f9872c713f264d989 : SUCCESS in 19m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d93743421a4747d4b37fe6393afe88ad : SUCCESS in 26m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c6e47db45b34312b27dfd45bbb34d7a : SUCCESS in 13m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f105f1b1f7c4ed1afeb356a339a0e5a : SUCCESS in 20m 47s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/335d5d578d32477397305d3651b7fd48 : SUCCESS in 1h 42m 29s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/a649fad3645c4b66acf80120d006335a : SUCCESS in 1h 20m 12s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d2a1c4e0761740fcb19e3c6ea1161c37 : SUCCESS in 1h 14m 38s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/81fbe92c3c5c442295262c76b123498d : SUCCESS in 1h 42m 13s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":4},{"id":"0426a2da18b89088ab4e9d89b41f7dc39793f7fa","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-06-24 07:07:14.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"7230eb7da0dd3bdbe0e94dc3d83782410836593f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-24 08:07:07.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3f79bac350513a89e5f763822f0e3365daa0289c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 08:07:45.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"32fdc502b529395401fc2da8c80bb04d7c4a27e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 10:29:48.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba16af3079bd4e3d87fb76752fb0e8c1\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ff7f9f2085d54ce38bd09f9d61d08635 : SUCCESS in 1h 32m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5955e2ec49024a33ac0b0feb3f560aa3 : SUCCESS in 1h 07m 21s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8026dc8a120c41e3b484d1841b5e9d97 : SUCCESS in 2h 20m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37541acfd8b547158d175bba9821e62b : SUCCESS in 6m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/72d39a01ef4546e7b6cbce25ebcf38e4 : SUCCESS in 44m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cc03fe17d9c1411bb70da1a54d58b040 : SUCCESS in 26m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92c14543fb314adca1f165c3b23ef5ba : SUCCESS in 13m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fdeff67811b341bd9f113950c7d854d5 : SUCCESS in 15m 26s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c2880730791f45d9a96ffb340077a908 : SUCCESS in 1h 30m 23s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/0c89aa97d2864714bbdd10b43fd690b0 : SUCCESS in 1h 14m 52s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7b07bca20f22467aa1ec71f3c561baf7 : SUCCESS in 53m 02s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":4},{"id":"a2fe2ab02ed464ddc50210b76fd3d8b3628c63b8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 10:29:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"f046a2c2b1211e4778dca9093bbced88f4d517b9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 10:34:50.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf96d0cfd0db4af2a0c658cb1902dbf9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b75eba421894f8884f7fe81954f8787 : SUCCESS in 1m 41s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"534c649a05899b464d7c325cd85f5b5d73e0250e","revisions":{"b3b9e96cb05dc93b2e3468cd1144da6240921ffc":{"kind":"REWORK","_number":1,"created":"2026-06-12 16:08:02.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/48/993148/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/48/993148/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/993148/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/993148/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/993148/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/48/993148/1"}}},"commit":{"parents":[{"commit":"de087e8f12a916d46565a360b9a350356d8ed994","subject":"Merge \"ovn: pass gateway LRPs to _check_external_ips_changed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/de087e8f12a916d46565a360b9a350356d8ed994"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-06-12 16:01:36.000000000","tz":-240},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-06-12 16:01:36.000000000","tz":-240},"subject":"Update dict columns with UpdateLRouterPortCommand","message":"Update dict columns with UpdateLRouterPortCommand\n\nThe command used to overwrite the columns such as external_ids or\noptions if called. The patch adds the new key-value pairs or updates\nthe existing ones.\n\nThe networks column, as a map type, is kept as is to not change the\nbehavior when an IP is changed.\n\nAssisted-By: Claude Opus 4.6\nCloses-Bug: #2156617\n\nChange-Id: I578835796a600f8d0527c7d44a1da1dbea19f7c8\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b3b9e96cb05dc93b2e3468cd1144da6240921ffc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b3b9e96cb05dc93b2e3468cd1144da6240921ffc"}]},"branch":"refs/heads/master"},"7a505499b439f3af2829d655b32d0f2c06e7a9bc":{"kind":"REWORK","_number":2,"created":"2026-06-18 17:23:41.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/48/993148/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/48/993148/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/993148/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/993148/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/993148/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/48/993148/2"}}},"commit":{"parents":[{"commit":"de087e8f12a916d46565a360b9a350356d8ed994","subject":"Merge \"ovn: pass gateway LRPs to _check_external_ips_changed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/de087e8f12a916d46565a360b9a350356d8ed994"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-06-12 16:01:36.000000000","tz":-240},"committer":{"name":"Jakub Libosvar","email":"jlibosva@redhat.com","date":"2026-06-18 17:14:15.000000000","tz":-240},"subject":"Update dict columns with UpdateLRouterPortCommand","message":"Update dict columns with UpdateLRouterPortCommand\n\nThe command used to overwrite the columns such as external_ids or\noptions if called. The patch adds the new key-value pairs or updates\nthe existing ones.\n\nThe networks column, as a map type, is kept as is to not change the\nbehavior when an IP is changed.\n\nAssisted-By: Claude Opus 4.6\nCloses-Bug: #2156617\n\nChange-Id: I578835796a600f8d0527c7d44a1da1dbea19f7c8\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7a505499b439f3af2829d655b32d0f2c06e7a9bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7a505499b439f3af2829d655b32d0f2c06e7a9bc"}]},"branch":"refs/heads/master"},"2717f560f90283b0b1e210e798218d8fa4db07b2":{"kind":"REWORK","_number":3,"created":"2026-06-22 20:44:51.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/48/993148/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/48/993148/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/993148/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/993148/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/993148/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/48/993148/3"}}},"commit":{"parents":[{"commit":"869fb25568fe0967cb9915b41f89a10383a0ee4f","subject":"Merge \"[CI] Move CentOS jobs to 10-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/869fb25568fe0967cb9915b41f89a10383a0ee4f"}]}],"author":{"name":"Jakub Libosvar","email":"jlibosva@redhat.com","date":"2026-06-22 20:33:23.000000000","tz":-240},"committer":{"name":"Jakub Libosvar","email":"jlibosva@redhat.com","date":"2026-06-22 20:44:45.000000000","tz":-240},"subject":"ovn_client: Reset only defined LRP.options","message":"ovn_client: Reset only defined LRP.options\n\nThe patch enumerates LRP.options keys that ovn_client manages.\nIn order to avoid overwriting key-value pairs that are added by\nother service plugins, such as EVPN, we replace only those that are\nknown to the ovn_client.\n\nAssisted-By: Claude Opus 4.6\nCloses-Bug: #2156617\n\nChange-Id: I578835796a600f8d0527c7d44a1da1dbea19f7c8\nSigned-off-by: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2717f560f90283b0b1e210e798218d8fa4db07b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2717f560f90283b0b1e210e798218d8fa4db07b2"}]},"branch":"refs/heads/master"},"534c649a05899b464d7c325cd85f5b5d73e0250e":{"kind":"REWORK","_number":4,"created":"2026-06-23 16:21:40.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/48/993148/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/48/993148/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/993148/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/993148/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/993148/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/48/993148/4"}}},"commit":{"parents":[{"commit":"869fb25568fe0967cb9915b41f89a10383a0ee4f","subject":"Merge \"[CI] Move CentOS jobs to 10-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/869fb25568fe0967cb9915b41f89a10383a0ee4f"}]}],"author":{"name":"Jakub Libosvar","email":"jlibosva@redhat.com","date":"2026-06-22 20:33:23.000000000","tz":-240},"committer":{"name":"Jakub Libosvar","email":"jlibosva@redhat.com","date":"2026-06-23 16:21:33.000000000","tz":-240},"subject":"ovn_client: Reset only defined LRP.options","message":"ovn_client: Reset only defined LRP.options\n\nThe patch enumerates LRP.options keys that ovn_client manages.\nIn order to avoid overwriting key-value pairs that are added by\nother service plugins, such as EVPN, we replace only those that are\nknown to the ovn_client.\n\nAssisted-By: Claude Opus 4.6\nCloses-Bug: #2156617\n\nChange-Id: I578835796a600f8d0527c7d44a1da1dbea19f7c8\nSigned-off-by: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/534c649a05899b464d7c325cd85f5b5d73e0250e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/534c649a05899b464d7c325cd85f5b5d73e0250e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
